)]}'
{"id":"openstack%2Fkolla~1006000","triplet_id":"openstack%2Fkolla~master~I953a3ef93efe128644a2addfc4e9710e333ba480","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I953a3ef93efe128644a2addfc4e9710e333ba480","subject":"Switch to Hibiscus sources","status":"NEW","created":"2026-09-17 08:33:43.000000000","updated":"2026-09-22 23:53:37.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"e2ba9735b587d95038bdb24c0449262f6034a88d","_number":1006000,"virtual_id_number":1006000,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-21 13:08:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2026-09-17 08:34:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-17 08:34:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-17 08:36:55.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"}],"messages":[{"id":"22707ec87f93a122e4aedd0d1dc3b2cce6864d18","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-17 08:33:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c64794f1e86b9344d00bedc5bee9357b0d8161ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 08:34:02.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch master.  The\nproblem was:\n\n  extra keys not allowed @ data[\u0027required-projects\u0027][1][\u0027openstack/aodh\u0027]\n\nThe problem appears in the \"kolla-base\" job stanza:\n\n  job:\n      name: kolla-base\n      parent: base\n      timeout: 10800\n      post-timeout: 10800\n      pre-run: tests/playbooks/pre.yml\n      run: tests/playbooks/run.yml\n      post-run: tests/playbooks/post.yml\n      attempts: 5\n  ...\n\n  in \"openstack/kolla/zuul.d/base.yaml@master\", line 2","accounts_in_message":[],"_revision_number":1},{"id":"846d4237fe79ebca7c2828ab6efcbcdd4c6e0e32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 08:34:15.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch master.  The\nproblem was:\n\n  extra keys not allowed @ data[\u0027required-projects\u0027][1][\u0027openstack/aodh\u0027]\n\nThe problem appears in the \"kolla-base\" job stanza:\n\n  job:\n      name: kolla-base\n      parent: base\n      timeout: 10800\n      post-timeout: 10800\n      pre-run: tests/playbooks/pre.yml\n      run: tests/playbooks/run.yml\n      post-run: tests/playbooks/post.yml\n      attempts: 5\n  ...\n\n  in \"openstack/kolla/zuul.d/base.yaml@master\", line 2","accounts_in_message":[],"_revision_number":1},{"id":"d73a7b7664448709bb92bf5f0f3c7dbebe62c0fa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-17 08:44:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"acec35b0015bf56260227d2fa44b9986cdb642dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 10:26:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ba9819386704c7caa93acbc1bd2194c\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/c362c4ceaaff4791828a47f1992621c3 : SUCCESS in 49m 10s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/1a36309414cb4fa19823e4a9445cbcc5 : SUCCESS in 37m 00s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/87551347b9364bc4adb2f467346a4e88 : SUCCESS in 45m 21s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/c06fafbc7e4e429f88ac3379ecfe2bc2 : FAILURE in 32m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d0c911f52f5e401a8379b295dc600ed9 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9c3860076abf40fba90148299571965c : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bf4537e4ab474cd8a224e472c2f3bc78 : SUCCESS in 6m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fe359a46e086b151f02e88cd4045a8ad77b3a2ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 11:32:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/25fa090ab6b84469985a56bf222578ff\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/ca02bd56894c4857b2e8bfd4bb47656b : FAILURE in 30m 47s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/a3bf779a7d4e4253a06e2156ae8f7ffe : FAILURE in 38m 36s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/bdc4c8fa9b77481fa386510278a40082 : FAILURE in 31m 30s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/91f8245562064d8b96c38d627e36d29e : FAILURE in 1h 13m 36s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef008090b1dc41fdb9037d16c83e702e : FAILURE in 24m 29s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7ae537afbd724d4b9f0d981fdc9c1124 : FAILURE in 38m 17s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/2feca5bc213c4b40887d03495b50d52a : FAILURE in 41m 25s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/b8befe88a17442c9a92c5afe1aff8558 : FAILURE in 30m 47s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/8a284bbdca5241b3a388776ffbfec9c8 : FAILURE in 18m 14s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/06514f28875e462d864820bd585073a5 : FAILURE in 15m 17s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c76a8a46e764387a7d9878de6b48ffd : FAILURE in 18m 21s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/dd4268d4c7e34ee6b9a6c96558a589aa : FAILURE in 23m 34s\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/694b9cad405d4bff933b5a3f9288b270 : FAILURE in 25m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/e0bc5e9997bf4837a8624fd01f2902a9 : FAILURE in 23m 06s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/050932e5eb4e44b2a9de6ec90c931fff : SUCCESS in 36m 31s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/afb90a4a04514bba8bdac1f0e5e5e4be : SUCCESS in 26m 32s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/c9714af766e340be8cb29c94209fa2c3 : FAILURE in 12m 13s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8185e4e8647e4f0bb897407a8c739a91 : SUCCESS in 42m 39s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/ba0e15f91f574c48ad8697d75398d0c7 : FAILURE in 17m 11s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5448a5a44e9e47a39f2e3c9625db7dea : SUCCESS in 31m 56s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/5d9b4d506809434d911b68dbd87f097a : FAILURE in 11m 46s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a85ba42412c44a429a037360e6b6de37 : FAILURE in 9m 32s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/62e7bcaa8c6d41f4a163a0f720dba1c3 : FAILURE in 11m 13s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8a9eabd37ad04ac899955d63bc30e18b : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc5a78eee28c45c2ae346eed1465d1e4 : FAILURE in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ce9be6efb6b6458787d811e66cab4434 : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ae54214f0d944ad97741a2c3ff346cb : SUCCESS in 3m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e5ebb0bbd28c4ecfa703c94313234848 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af1c7afd59e6470eae27b076da7f3b69 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcd7d2f154214852b9059da924c3bb38 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cf198f5237a4c4093b97b2ac7d9d6eb : SUCCESS in 10m 16s","accounts_in_message":[],"_revision_number":2},{"id":"6d728bbcc76af045ac52272a20d93ff9e31ea81b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-21 12:05:21.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"8961916440563c6b2a003878b6f8856f0b6f9487","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-21 12:22:18.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"83cbcb3b32eb867b2f90fa9d6a22d0da46e30a2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:08:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/55ca544191b7487f93881d257b1aa7c6\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/f87e37773a90476bb3d3a409ab851036 : FAILURE in 23m 44s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/4997b70df24047c2bac76be4cbe0ca12 : FAILURE in 27m 31s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/84c00e77bade4f30917c54a45cc09c19 : SUCCESS in 26m 54s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/bbb473859aaf411d984af4a1c2ebe946 : SUCCESS in 43m 21s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7b76a7803c74bd3aec22774cdb5f17c : FAILURE in 23m 38s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1b6aefa942904fa08ab6deb2b0d3f7b3 : FAILURE in 39m 16s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/62aaa6c0923f47b0b7c479f229323c78 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c892184d07a4b2c841dc35f2c791a1d : FAILURE in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/93a45560bf18434b9b84ad741c2fcbc3 : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/17d805bf75af4bf6bf1687ec4b8e7e72 : SUCCESS in 5m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/02e01bd632a74d939ffbd7f705fa6b82 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a322331f8dc415b8330dd029c08d89f : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cd6ddd5e67a4655af571214866c3ff1 : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2f72cf878d540f5a3f9ac7ea11b5fd0 : SUCCESS in 10m 26s","accounts_in_message":[],"_revision_number":4},{"id":"69b05f654510cda2565c0d9b578930bd1a5270fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:25:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7983a6f9b9294c268d0c409fe6483405\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/61a2787ec52d4235a4eed2e3662b532a : FAILURE in 49m 48s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/bec1e9ff0ee241198f36d206bb7c0a85 : SUCCESS in 59m 20s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/ba6287fb80e048b3bae803c64dac1950 : FAILURE in 50m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f72f3940a30043298962c21752cf4265 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/256c8fceeb2e4739917d38331abc9846 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f4913962cff042a58298a96be318a709 : SUCCESS in 9m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9c00bccdd3da8c676076da575bbfc9caf855b0ff","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-22 21:24:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4e4301f5652feed5a730ff6ddd7d7dcae96abfab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 22:21:18.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cb4cd749c2f4bc4b745572f1cf8e1fc\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/10571dafd1b143f395460f7075ae46d4 : SUCCESS in 30m 31s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/2698134874884dbdacadaa30349af801 : SUCCESS in 35m 35s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/998c572047e24b2189da0a0f7e247a48 : SUCCESS in 27m 29s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/997c9365e4c94c3198ff4ab53bebb4cc : SUCCESS in 53m 13s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4a65cb5e19640d1b875ef3fc0a33f37 : SUCCESS in 28m 04s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c7564bbbec8545d79824c9a2a18b3c46 : SUCCESS in 34m 51s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f255a26761094a278080e50e0041108c : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6191f9586c6f42ecb98695aaaafb2a4e : FAILURE in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e74ba0aade9e4f10877bfd741f1fed3b : SUCCESS in 4m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9ec6214c9b404d90bdbafaf5208da58d : SUCCESS in 4m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7233e7f489644801a374941e280a5bc3 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/310f836097e34b80bcfac8201888a900 : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f201214fd21e4f8cb3e36d1a5c216f92 : SUCCESS in 12m 41s","accounts_in_message":[],"_revision_number":4},{"id":"e2ba9735b587d95038bdb24c0449262f6034a88d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 23:53:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3174acc331574627b3bd2c1b8c7d0cff\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/73da013f82d5409fbcdcdbd7e552a7e0 : SUCCESS in 56m 51s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/edc6da0d68d3469e959c3d7613cdc92f : SUCCESS in 54m 26s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/4de4f8f568934c2b86847891eaa63219 : SUCCESS in 43m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f7028539d4bd492fba8ad6f04e0a61e1 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b772ca9173094ed5be9bc84d5cfd06f0 : SUCCESS in 8m 33s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ed50817f1c2a209abae3c8f91d2df2fa768363ee","revisions":{"1ccf9a3dc170d161434dd426e8d92c2eb4d5f449":{"kind":"REWORK","_number":1,"created":"2026-09-17 08:33:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/1006000/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/1006000/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/1006000/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/1006000/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/1006000/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/1006000/1"}}},"commit":{"parents":[{"commit":"0cbc63835b7d24c726da1bb44a46fbdd3e09680e","subject":"Updated from check-sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cbc63835b7d24c726da1bb44a46fbdd3e09680e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-09-17 08:33:30.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-09-17 08:33:40.000000000","tz":120},"subject":"WIP: Try out Hibscus RC1","message":"WIP: Try out Hibscus RC1\n\nChange-Id: I953a3ef93efe128644a2addfc4e9710e333ba480\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ccf9a3dc170d161434dd426e8d92c2eb4d5f449"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ccf9a3dc170d161434dd426e8d92c2eb4d5f449"}]},"branch":"refs/heads/master"},"d2a7741ebac6b18b42f5284b1b2ca01c6390bd22":{"kind":"REWORK","_number":2,"created":"2026-09-17 08:44:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/1006000/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/1006000/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/1006000/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/1006000/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/1006000/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/1006000/2"}}},"commit":{"parents":[{"commit":"0cbc63835b7d24c726da1bb44a46fbdd3e09680e","subject":"Updated from check-sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cbc63835b7d24c726da1bb44a46fbdd3e09680e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-09-17 08:33:30.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-09-17 08:44:15.000000000","tz":120},"subject":"WIP: Try out Hibscus RC1","message":"WIP: Try out Hibscus RC1\n\nChange-Id: I953a3ef93efe128644a2addfc4e9710e333ba480\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2a7741ebac6b18b42f5284b1b2ca01c6390bd22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2a7741ebac6b18b42f5284b1b2ca01c6390bd22"}]},"branch":"refs/heads/master"},"abfd9daf16a16ff8c1a24fc66870411d8cc1d1ce":{"kind":"REWORK","_number":3,"created":"2026-09-21 12:05:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/1006000/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/1006000/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/1006000/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/1006000/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/1006000/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/1006000/3"}}},"commit":{"parents":[{"commit":"0cbc63835b7d24c726da1bb44a46fbdd3e09680e","subject":"Updated from check-sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cbc63835b7d24c726da1bb44a46fbdd3e09680e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-09-17 08:33:30.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-09-21 12:05:21.000000000","tz":0},"subject":"Switch to Hibiscus sources","message":"Switch to Hibiscus sources\n\nChange-Id: I953a3ef93efe128644a2addfc4e9710e333ba480\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abfd9daf16a16ff8c1a24fc66870411d8cc1d1ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abfd9daf16a16ff8c1a24fc66870411d8cc1d1ce"}]},"branch":"refs/heads/master"},"ed50817f1c2a209abae3c8f91d2df2fa768363ee":{"kind":"REWORK","_number":4,"created":"2026-09-21 12:22:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/1006000/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/1006000/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/1006000/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/1006000/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/1006000/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/1006000/4"}}},"commit":{"parents":[{"commit":"0cbc63835b7d24c726da1bb44a46fbdd3e09680e","subject":"Updated from check-sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cbc63835b7d24c726da1bb44a46fbdd3e09680e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-09-17 08:33:30.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-09-21 12:22:17.000000000","tz":0},"subject":"Switch to Hibiscus sources","message":"Switch to Hibiscus sources\n\nChange-Id: I953a3ef93efe128644a2addfc4e9710e333ba480\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed50817f1c2a209abae3c8f91d2df2fa768363ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed50817f1c2a209abae3c8f91d2df2fa768363ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
