)]}'
{"id":"openstack%2Fkolla~1006024","triplet_id":"openstack%2Fkolla~master~Id345a419f714b54b343a6b9e90f01ef902408abf","project":"openstack/kolla","branch":"master","attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-09-17 14:11:28.000000000","reason":"\u003cGERRIT_ACCOUNT_37306\u003e replied on the change","reason_account":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"}},"37306":{"account":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"last_update":"2026-09-17 20:11:52.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-09-18 10:55:22.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"Id345a419f714b54b343a6b9e90f01ef902408abf","subject":"Fix permissions of kolla_extend_start scripts","status":"NEW","created":"2026-09-17 12:28:50.000000000","updated":"2026-09-18 11:42:26.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"75e69f83ffd19fe732b627247a740d90f7d6a09e","_number":1006024,"virtual_id_number":1006024,"owner":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-18 11:42:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2026-09-18 10:55:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-17 13:44:49.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2026-09-17 16:23:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-17 17:57:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-17 20:00:58.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"aeee9b6c31ed32b322b801bf9be0ffaa7e592435","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-09-17 12:28:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5642d2a33b432402674d9df4dcf3a2667a7cfbaf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-17 13:44:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"308471267c4c3df769904127f8301279d293229d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-09-17 14:11:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9a4f364580ddfc32bf52ae2d71523f096e778d6a","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-09-17 14:11:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"379bd7d5735702eae3af254e8ce7099b7b4650c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 16:23:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c3a940934974c688523c064bd71fcea\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/edb9f663d4094719ac79c9ce238efc38 : SUCCESS in 41m 12s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/e5f80fb020fb4bc1b1a5b0bf08e3a1e5 : SUCCESS in 1h 04m 40s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/75768bbb7bc44536b12b58e4c31b6ef2 : SUCCESS in 45m 28s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/52baa6e8fa9f43da87d8a4601178eb7a : FAILURE in 1h 20m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/239c90ab88a04a478905532be2c08b2a : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/68ce4e1c8063410187984bdc94745825 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2ea861595129413f808fa2e1b6e8f219 : SUCCESS in 5m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c7a163fed160ef9d5925e1303a99cc15717b1a78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 17:57:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4ae940bd2434521a56145e1585a88b8\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/53ebad40705841f7aa5083fa452246bf : SUCCESS in 34m 25s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/441597ed4b274d5ca76e31e341f73214 : SUCCESS in 31m 55s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/4f6c13e22ec04bfdab05491e6710d27f : SUCCESS in 32m 08s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/22cb7eb48433424888e44c71abf6414c : SUCCESS in 1h 11m 07s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/270a4c1011124acf971dff350bd60090 : FAILURE in 33m 42s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c5681f3bdd454c139569cab5fb3a2be6 : FAILURE in 33m 12s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/62522070f96a4284b1ff0366e8e2fc00 : FAILURE in 13m 24s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/ccce2cbdc4a64c1abcdd9afed0216481 : SUCCESS in 2h 08m 29s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/51a59c72614d4bd3a72ff081733665ef : SUCCESS in 1h 01m 48s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/9b94ff3ce67e4f258f583b9123e2555b : SUCCESS in 1h 25m 03s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a022a24dee754174af60ff8a5abc79c2 : SUCCESS in 1h 16m 17s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/2340d2846dee48e5ae1c554110f09739 : SUCCESS in 1h 56m 24s\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/87fee48b19944603bc3bfc1a2904684d : SUCCESS in 1h 16m 53s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/fe90a3e04cbf451bb86fbc0c167b004f : SUCCESS in 1h 18m 32s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/a858f815d9ae4c888d6adc51c4db44d1 : FAILURE in 13m 58s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/7201d11b6d8c46fd8fec5fe89ab342bd : SUCCESS in 1h 44m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/2e39ae6241f2454f80032a001ae2832f : SUCCESS in 2h 00m 00s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/cf7589790b454c418144a5bb172542b3 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52161d6b06904486a01f57c1b12ad7e9 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e6516b7a0dfc4822bd5dc383af46d25e : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/857df35a19d34d93b45737fbff6aef05 : SUCCESS in 3m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/70e7e9f126314f3c9aa5d53dbbfe7e6b : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a50f840cc83245ceb39e00d80a5b5c7d : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/260f2ffbf8b64614922659c201c9135b : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74385e4619f0438cadfc473873d53f7f : SUCCESS in 10m 28s","accounts_in_message":[],"_revision_number":2},{"id":"25198f1de098260fd1c4852fa1d1dbdc2abcec8b","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-09-17 18:01:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e13cce0dc362cc2e315e91207eda9e31c027ac06","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 19:38:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0e0f36e2b394ed896c76373d9fc644d\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/15f2b9602c60451ebe887737c67f5df0 : SUCCESS in 53m 17s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/6b9f39df10e94a3797be3d74a5f682c0 : SUCCESS in 54m 56s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/be1a8d18401f4f58be447ff35f8e5c78 : SUCCESS in 52m 14s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/82ba0c1664e24f0a8c8543a898a7aac4 : FAILURE in 1h 29m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1187d0c0ee0a4cbe9c76621e1e69d169 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/12785cce577447cd8a24b856b068e6e7 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ca45309806b343b199c8e0152650a02d : SUCCESS in 6m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a6efef24a40ee1f18faf796d5e260864957ab5d3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-17 20:00:58.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"84b4f3b16e9c490d5faf94caaa3ec4b2a539829e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 20:11:52.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e8561ddc1e740d5b32dc5b97e587ff1\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/84f57bc359074c2c8aa5646ddd912317 : SUCCESS in 20m 59s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/dd124a6e48244938b96433bca235cc6e : SUCCESS in 31m 35s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/a278af7558e6490b848c63a32f66ad3b : SUCCESS in 30m 17s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/b09019ee3b39418e82a064c7e2e18f30 : SUCCESS in 41m 02s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9cba48cf877406abe13d8943c757473 : FAILURE in 28m 37s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/27747b7d534947d99aa63fb2783fbf46 : FAILURE in 32m 27s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/4a2b9c9fd7db42eab28fa88c8f141c54 : SUCCESS in 1h 11m 58s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/9a15ec5b36b24ee5b90948b63164a081 : SUCCESS in 1h 46m 01s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/fbe32d6cf8f84e24bfc3a7977f6e00c0 : SUCCESS in 1h 02m 37s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/bab08f2a657c407da6f603f1fe921fc3 : SUCCESS in 1h 33m 18s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ee6c8c7c7604de984212ff0e0962887 : SUCCESS in 1h 19m 47s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/4b53928dbbc7434ea8e62f30a9445a27 : SUCCESS in 1h 41m 39s\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/a42af3696a164a7fadaaf606964af133 : SUCCESS in 1h 28m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/94e19d39a64c46dd9a6124c4f23f1a5d : SUCCESS in 1h 17m 43s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/064080880be24b95b4fbf8f59ce01fcc : SUCCESS in 2h 07m 03s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/5540e638aad045088e02dcf444e5cbbc : SUCCESS in 1h 38m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/16f751501d8343fb835f53a83a886136 : SUCCESS in 1h 48m 21s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/021988e337aa486aaaf114018875c028 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/383c3a0491f74c549c02b0da0bd22222 : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9cac0eca4ac24e48ac4e796172cfa6b6 : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f13e1139f2324e468973155c3f7d8854 : SUCCESS in 3m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d77ec616151a4d7fab3d9e8ff073bf8d : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35bbc508c2304494a24f5e0f4861929c : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f861385912364f43a5650ef37f4ae805 : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16d531fbb02d4b549ef3f5416a1c3279 : SUCCESS in 9m 41s","accounts_in_message":[],"_revision_number":2},{"id":"74824a3ce2a61c8c1e23a10ea08d00dbbbd70634","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-09-18 07:17:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"3012825c2b732457c26e80c010927eb513794752","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-18 08:51:36.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e3f01dc55991160777d38c6e50dd8e615633f63f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 09:16:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd3e7fcc78254ca4b231c437f5e505bf\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/7a50f1f0b333458e9c70f0079f16769c : SUCCESS in 16m 20s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/5d20045d5d194eaca5789d69f00df53b : SUCCESS in 19m 15s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/5b1ed12a96e24d66809677327057ff62 : SUCCESS in 20m 23s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/d9545942a0d34765834078ec7416faf8 : SUCCESS in 45m 47s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5d53aef61424d6bb9dc71c758490e0a : FAILURE in 25m 23s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8849a1d7484c48169bc0473f0ed07619 : FAILURE in 24m 55s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/f0d572547139405d8f6f4d02635c589f : SUCCESS in 1h 17m 03s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/732094aa778e4397bf9706cf6ae023bf : SUCCESS in 1h 55m 09s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/0ef14965929c4b30a55860b0007fa895 : SUCCESS in 1h 01m 35s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/5e6e566bddce4fbcaeb178ba54bb65dc : SUCCESS in 1h 26m 52s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df0ffe5fdaca438f98ce18a8b21f5c04 : SUCCESS in 1h 12m 17s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/879919dc6bdf4509b4795680de1cdf5b : SUCCESS in 1h 43m 17s\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/c99c6efcba544b49845282e85e2e70ec : SUCCESS in 1h 20m 18s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/278912743c4c4d668f7dd6824113758d : SUCCESS in 1h 25m 53s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/107b7230e3bb46e7b03f3392a7b51f3f : SUCCESS in 1h 54m 09s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/15ddb7d0e6614d0d93e57bc566760654 : SUCCESS in 1h 40m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/33daf27c34f0428c8cd2ddc5e9466899 : SUCCESS in 1h 45m 51s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/cfc4f22f40d446f2bcb21c30eabdc04f : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc954e248a174bf9a7cefbfcc6a3dba0 : SUCCESS in 2m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/14be1bc24fb74298a2c870452707a54a : SUCCESS in 2m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0064d27293c54ff1a9279363afee6473 : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3507c6b8a6c84cc9b4d99a1773af6154 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e377ce6ef4364eb9beed9bb1725af0df : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9607e50c938479db771c5ece0ee0545 : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fcd8cc7be014337a9310c6d8cc06425 : SUCCESS in 9m 53s","accounts_in_message":[],"_revision_number":3},{"id":"eb6d2a9f4375c3cb913ae3b9fd809247c288078e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 09:17:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d301fb55de6476d9593614419e8a86e\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/381da6e0aef847d386e5fa0d1180c61c : SUCCESS in 39m 15s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/89c5377703e74b6bb2457c060cf18827 : SUCCESS in 55m 52s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/1d95d1be7ef34174ac85e858b4deb387 : SUCCESS in 38m 38s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/9b9d484b066943b99ab00bceff6759f7 : FAILURE in 1h 18m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/202fa13ff83e40b1ada2b410f2b68fa1 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3fd4dbf9d6da4f2596e72a60603eb2f6 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/87918d3c906c409a993f843cf670b0df : SUCCESS in 8m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4be9a86ae00a0b5e54fff31923aa495e0db57de2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-09-18 09:37:48.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"5929ed8a8f3bc5961d93fdb682cecdd318c60c1d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-18 10:55:22.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6cab21358867f4905586c05a9e0dcc3c3680b8ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 10:58:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2318bbca26c449998f80d908e8ac4b7\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/01c1891963884b94a7f477d93bad3a97 : SUCCESS in 38m 25s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/820b12b5ce0c48a2a676ee4335d10537 : SUCCESS in 56m 55s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/cecda3913d804110aa060f9b1dbc35f7 : SUCCESS in 35m 19s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/9dd355eb6dcb4675b7f13183ff093f19 : FAILURE in 1h 15m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3987b37991984b42be3b95a1ef8a6f50 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fbfbff32b22b49108f2924ab930d84fd : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/88ab0fc63cd04f1b91745610579320c6 : SUCCESS in 9m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"75e69f83ffd19fe732b627247a740d90f7d6a09e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 11:42:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/972ce2b799034cc99bf2a3dcc893d8d1\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/5d40e3cc69be43c88ef7d7fa576ce8f7 : SUCCESS in 31m 50s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/c8d34b4681464cceb1d12eb78bbf38c8 : SUCCESS in 37m 31s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/9c8a322167a24c769fbd058480133591 : SUCCESS in 33m 42s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/0b2ab31efe364331b3415e1bbc621407 : SUCCESS in 1h 11m 54s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15f965fc6d984c6f9f0782bac31aee13 : SUCCESS in 36m 56s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5b613d9713804b5c94e0a998dcfd2e36 : SUCCESS in 35m 36s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/1ad4a8bde0f3456d892da07b0ab9c190 : SUCCESS in 1h 16m 47s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/5ff7e34888854fbe9191718588ff3dab : SUCCESS in 1h 58m 17s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/282c381450834092b6d0f06002280bea : SUCCESS in 1h 03m 33s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/88d3b8b579ae426992362d1a82c7a3fd : SUCCESS in 1h 26m 37s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e606fc3c8c24b96a44d7c7dc0084ba8 : SUCCESS in 1h 23m 36s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/2a6516c03c26431183c73f1238ea4425 : SUCCESS in 1h 54m 05s\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/7e46b4d242094a8a980c4500ce35b9e7 : SUCCESS in 1h 28m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/233e774252964e36885f882f6255e0c8 : SUCCESS in 1h 33m 41s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/6fd947ef41104290a210f16ba9405c66 : SUCCESS in 1h 05m 53s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/9b67397b55a04196a2806eba69878d2a : SUCCESS in 1h 17m 45s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/e17f63fbd2354d1f84107801b9872375 : SUCCESS in 1h 59m 09s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/4afe61a8017545149af48d4138a5ac43 : SUCCESS in 1h 42m 18s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/b8c9b3ea34864699b51b143b8a0b5f7d : SUCCESS in 1h 52m 19s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9bf4b9a7f9f6497d8a878177f023cd6a : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dac19bb442824592a98e1786a0c6bc8c : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f94561d5b60b4f25af8938c90013f610 : SUCCESS in 4m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/65fcbfe6883a4b8ca01cbdb24e29d1e4 : SUCCESS in 3m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d6e63e839dc24736a2bdf25f06e5f5ca : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d448667d4b3f4e0484c7cc9a32550867 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/130aca24be3a4770a1ccdc505e2376cc : SUCCESS in 8m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a5690aba22942b28ba4bc96cd0278bf : SUCCESS in 12m 07s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9ef712a3d05150d01264a7968ebe7cabe2fc105e","revisions":{"67fece1b6fd83e2a4f8d7b2c1ab3c76fec584aee":{"kind":"REWORK","_number":1,"created":"2026-09-17 12:28:50.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/24/1006024/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/1006024/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/1006024/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/1006024/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/1006024/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/1006024/1"}}},"commit":{"parents":[{"commit":"0cbc63835b7d24c726da1bb44a46fbdd3e09680e","subject":"Updated from check-sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cbc63835b7d24c726da1bb44a46fbdd3e09680e"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-09-17 12:27:44.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-09-17 12:28:40.000000000","tz":120},"subject":"Fix permissions of kolla_extend_start after pip install","message":"Fix permissions of kolla_extend_start after pip install\n\nThe kolla_extend_start file is tracked with mode 0644, but pip\ninstallation can install it with mode 0640. This prevents service\nusers from reading the file and causes container startup failures.\n\nCloses-Bug: #2167588\nChange-Id: Id345a419f714b54b343a6b9e90f01ef902408abf\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67fece1b6fd83e2a4f8d7b2c1ab3c76fec584aee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67fece1b6fd83e2a4f8d7b2c1ab3c76fec584aee"}]},"branch":"refs/heads/master"},"d5edbb52c98fd7fec9e29059b7f1c67865bb81b4":{"kind":"REWORK","_number":2,"created":"2026-09-17 14:11:07.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/24/1006024/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/1006024/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/1006024/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/1006024/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/1006024/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/1006024/2"}}},"commit":{"parents":[{"commit":"0cbc63835b7d24c726da1bb44a46fbdd3e09680e","subject":"Updated from check-sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cbc63835b7d24c726da1bb44a46fbdd3e09680e"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-09-17 12:27:44.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-09-17 14:10:41.000000000","tz":120},"subject":"Fix permissions of kolla_extend_start in metadata agent","message":"Fix permissions of kolla_extend_start in metadata agent\n\nThe neutron-metadata-agent Dockerfile does not explicitly set the\npermissions of kolla_extend_start. After pip installation, the file\ncan have mode 0640, preventing the service user from reading it.\n\nCloses-Bug: #2167588\nChange-Id: Id345a419f714b54b343a6b9e90f01ef902408abf\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5edbb52c98fd7fec9e29059b7f1c67865bb81b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5edbb52c98fd7fec9e29059b7f1c67865bb81b4"}]},"branch":"refs/heads/master"},"9c83656388cc1aeab05e877aa7bcf336d9fe547b":{"kind":"REWORK","_number":3,"created":"2026-09-18 07:17:28.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/24/1006024/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/1006024/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/1006024/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/1006024/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/1006024/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/1006024/3"}}},"commit":{"parents":[{"commit":"0cbc63835b7d24c726da1bb44a46fbdd3e09680e","subject":"Updated from check-sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cbc63835b7d24c726da1bb44a46fbdd3e09680e"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-09-17 12:27:44.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-09-18 07:16:59.000000000","tz":120},"subject":"Fix permissions of kolla_extend_start scripts","message":"Fix permissions of kolla_extend_start scripts\n\nThe Neutron metadata, DHCP and L3 agent Dockerfiles do not\nexplicitly set the permissions of kolla_extend_start. When the\nfile is installed by pip, it can get mode 0640, preventing the\nservice user from reading it.\n\nCloses-Bug: #2167588\nChange-Id: Id345a419f714b54b343a6b9e90f01ef902408abf\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c83656388cc1aeab05e877aa7bcf336d9fe547b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c83656388cc1aeab05e877aa7bcf336d9fe547b"}]},"branch":"refs/heads/master"},"9ef712a3d05150d01264a7968ebe7cabe2fc105e":{"kind":"REWORK","_number":4,"created":"2026-09-18 09:37:48.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/24/1006024/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/1006024/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/1006024/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/1006024/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/1006024/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/1006024/4"}}},"commit":{"parents":[{"commit":"0cbc63835b7d24c726da1bb44a46fbdd3e09680e","subject":"Updated from check-sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cbc63835b7d24c726da1bb44a46fbdd3e09680e"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-09-17 12:27:44.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-09-18 09:37:30.000000000","tz":120},"subject":"Fix permissions of kolla_extend_start scripts","message":"Fix permissions of kolla_extend_start scripts\n\nThe Neutron metadata, DHCP, L3 agent and letsencrypt Dockerfiles do\nnot explicitly set the permissions of kolla_extend_start. When the\nfile is installed by pip, it can get mode 0640, preventing the\nservice user from reading it.\n\nCloses-Bug: #2167588\nChange-Id: Id345a419f714b54b343a6b9e90f01ef902408abf\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ef712a3d05150d01264a7968ebe7cabe2fc105e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ef712a3d05150d01264a7968ebe7cabe2fc105e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
