)]}'
{"id":"openstack%2Fkolla~1006144","triplet_id":"openstack%2Fkolla~master~I11fb14bd470e78eb5fe6f75a6494ef61483e67d8","project":"openstack/kolla","branch":"master","topic":"openstack/kolla/check-sources","attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-09-22 02:33:25.000000000","reason":"Vote got outdated and was removed: Code-Review+1"}},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-09-22 20:35:41.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I11fb14bd470e78eb5fe6f75a6494ef61483e67d8","subject":"Updated from check-sources","status":"NEW","created":"2026-09-18 02:55:40.000000000","updated":"2026-09-25 02:38:18.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7a8a8ec7c039641465c8b936b551ddf6d5a778d4","_number":1006144,"virtual_id_number":1006144,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-09-25 02:38:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 04:44:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-18 05:05:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-20 14:43:22.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-09-21 04:46:33.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"87d6e294a92ead15523cabe3f0af3134a62180f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-09-18 02:55:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a616b7a4791b5c2b32477f35840baf916ab44f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 04:44:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60c88a17c9c742a98e960134f36f62f8\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/b4879f024374473c97457181e401eeb2 : SUCCESS in 45m 36s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/6e24c3fa6913427582b8a0693d6596f8 : SUCCESS in 49m 22s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/91bdbc35f53c424e92b2a55f3d60f7eb : SUCCESS in 46m 11s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/2cc7c62f90f14d55995d1154b9ed6d10 : FAILURE in 1h 21m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dfa7efd0b58f4b749c25e7766e022007 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dc5a76506582464fabd034c8c65d3d0e : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/020bb073e8f748378ee7513d6fa7684e : SUCCESS in 8m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"07c80521123c62c5a22878f224587dc3edbf4e1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 05:05:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/54f3b418727b4e158d516f67e8acc272\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/20f8016b9f25471c95e10dab89b1cf33 : SUCCESS in 30m 11s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/5f1358de4ac74f0092c4d65bfa7a2cba : SUCCESS in 30m 04s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/5e21e8a0670a4f98af62024dd8cb7747 : SUCCESS in 31m 11s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/07d33aa70fe0491e8bec9868307d863b : SUCCESS in 41m 50s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2218887e69d48cab8aa2c19051fd698 : FAILURE in 18m 01s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a78e6e403ee84206bbf47c1d7071ef39 : FAILURE in 18m 47s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/42871130c64340d5925a933dc59bde52 : SUCCESS in 1h 16m 18s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/17ee8d18e2054a93ac6faa7632a139d4 : SUCCESS in 1h 55m 36s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/a6e097c5387d4c94bd575bf6b86ea180 : SUCCESS in 1h 05m 38s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/a97feffe64314db7a45deb01c66f15ca : SUCCESS in 1h 26m 02s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/454cc60d06604338a683a3ab31439b02 : SUCCESS in 1h 22m 33s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/abb7f19697604d378134a88063946542 : SUCCESS in 1h 48m 02s\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/1ade305d1b0749a68502897666c2ffe8 : FAILURE in 20m 12s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/003e6d6778a34096aadf0301631513f3 : SUCCESS in 1h 27m 53s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/5eceaf472f60422d81507c5b5dc76064 : SUCCESS in 44m 56s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/abe9be73c8fb49aea9eb6f9a88221648 : SUCCESS in 25m 16s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/11a326339b5045e68b6978f0f19a5ff7 : SUCCESS in 15m 58s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a40d4a7f35a14f2bbefa1d050e16f506 : SUCCESS in 38m 14s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/5fd42c5a2f8c4357adc0dd5e3b79bb7f : SUCCESS in 33m 03s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ed21141e68db43cf89ce4eb19ae2b4b9 : SUCCESS in 24m 05s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/2a5d4653a39d4443b9f81245ce6d5a36 : SUCCESS in 36m 30s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8cbe33edef64409289546d00bb0c6aff : SUCCESS in 42m 54s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/0b5fc26f35724c51bcb9fa3d7ae1996e : SUCCESS in 57m 53s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0f7d944ceb3f4fe5b843524de3d1aa38 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/101887bd775e4d63b5d7fd3e9c2961b2 : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eaff138db53a48959205129faba34e30 : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2b53577e017b4d71ad94707c46154260 : SUCCESS in 4m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fe91a14f454a4653b5ffbf9cc80df5f0 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7c7e3010fc84c74b01b51e031d40fba : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4b84b0c38e54a9a9277e5fb16cd2c56 : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71c6acfcbe80484b8774c5b4977de018 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ff9e9d765642dc88508784fe0aee0ca227a9f9d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-09-19 02:38:37.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7a6ff50fede1c6c8632b3f01b726f7cf77f49c2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 03:15:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7d9e761f1014456bd2e988afd145cbe\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/2871e252ab2e4ad3ac31587c2b19be8e : RETRY_LIMIT (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/3d99cf0ee5d84c5ca47c9c380bd3d299 : RETRY_LIMIT (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/4334ceecae274fd88f59476630454967 : RETRY_LIMIT (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/86b7c34b226e434981dee50acadaa796 : RETRY_LIMIT (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5316850d769a468a84bb3bf93d9fde97 : RETRY_LIMIT (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8b00957bb6f84ac7a8658453df20bced : RETRY_LIMIT (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4e2dcf1006b94d6b967b39223d4c2818 : SUCCESS in 9m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d7eaedc204f703eb78f144b043a055a23bb072b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 03:18:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/47d4927daedd4088b918c1ed2e45f6ec\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/bc6f3441ab424a6ea2de7293a9fa1754 : SUCCESS in 28m 22s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/b04be0e848ce4e87b854ad88f2b72992 : RETRY_LIMIT\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/15faa468ec3c40d3b37a84a2a614c47d : RETRY_LIMIT\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/be7bd0bf0f174643a08c17d94e9389c7 : RETRY_LIMIT\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/333d4e8ab3a9489ab7342490a743dff9 : RETRY_LIMIT\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7d9361c61c46435c8c79b5dd9e263a63 : SUCCESS in 33m 52s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/e5b60ae52f2646aa80379ba8ffa5a9d9 : RETRY_LIMIT (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/635352591af04b069ba27e504df99ddd : RETRY_LIMIT (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/73c229a758c54e30b9334d942e5f0270 : RETRY_LIMIT\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/34e734f250c7481e873f39d66a1ef5fe : RETRY_LIMIT\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71063f9d86954854802239ae71599eb8 : RETRY_LIMIT\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/bfdf6af2509e47468336a206cf2a03e1 : RETRY_LIMIT\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/cf69e4438da54065a5ac9f6581ae39de : RETRY_LIMIT (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/9c54196bc48a4dd8924491fbf16c8d64 : RETRY_LIMIT (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/fd29bee05df24d73a6c840f31adebfb6 : RETRY_LIMIT\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/d949f10db1f747f0a2c1fe128a1dc619 : RETRY_LIMIT\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/b18cac210ba34aaab888b70f8bdc9129 : RETRY_LIMIT\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e71482cc1c88445b86a7fb0561da5dbf : RETRY_LIMIT\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/a5bc7439d86f45da97d858a3662a628b : RETRY_LIMIT\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e8331974b01b48958629aff95a09397d : RETRY_LIMIT\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/276064e3142e4dc2bbe64253b6c073c4 : RETRY_LIMIT\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0a00ac077d22452685e0edd673ce4bf9 : RETRY_LIMIT\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/740c5c42f0ad4f9cbcd66131b041ccdf : RETRY_LIMIT\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/2cfb6742687e46f0affd1ffb05586bcf : RETRY_LIMIT\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04c5a2a9da1c44fba0db9575a5b82a77 : RETRY_LIMIT\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84012b2a2fb142aeb4a2324567622948 : RETRY_LIMIT\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9f8f65d4c90d43fe84e7532fe3a2666a : RETRY_LIMIT\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ab5113daebad43caad11b67596fa4c66 : RETRY_LIMIT (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86d2855d1a7c46739f9ac47171f5a3b5 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09445af0839d4536ae99213e55a1c5e4 : RETRY_LIMIT\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d3e4aa94d9b4a5ea16de81216e11e5f : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":2},{"id":"bc618e8f1c4433ed2275d78fe47ed0fdfce7f474","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-09-20 03:07:23.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"e00b57177807f7089b6cbf173a5e4b5fc21267ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 04:32:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/667326ff0df64dd8b9588b61fcca5968\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/54063b990fdc472188cacf056fe149af : SUCCESS in 47m 51s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/1559ad8aa1a8409696cc1f89b064a4c4 : SUCCESS in 46m 35s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/4b548af52f4843d4b4f9b504dce6adfc : SUCCESS in 47m 38s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/5cb4ba142eb44b14b3687d4a68dc20c1 : FAILURE in 1h 22m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/20976fffd1484744ad6d1660f489ef80 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/178e99d5f394450dab76421655ce8bef : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f325b0e511a043b28c4dfa2dc7b3d010 : SUCCESS in 9m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e518e20fd7906b22757aeb73c8eccda70f15cdb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 05:00:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f7883875ea04211bf6d81f4a010ae09\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/790e24f8c7fd43bb975b4fee59f9fc67 : SUCCESS in 30m 20s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/59884d8d95144702b1d5aed7ea9db7e9 : SUCCESS in 32m 56s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/271419a210a94dc599ddeb06944cf62e : SUCCESS in 23m 32s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/b08ce49746044a3397769c44a2daf9d5 : SUCCESS in 47m 54s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a751c88660846b48dd7b64513318637 : SUCCESS in 18m 19s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2e43d45f30be4056bb4a19e181119bdc : SUCCESS in 38m 24s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/68b1f480793b46a4b4e865f779941331 : SUCCESS in 1h 21m 16s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/8f035b8bf11944b8a9d118584d4cca30 : SUCCESS in 1h 49m 31s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/a5a002ccd74e425598e7e7dbe7ae933f : SUCCESS in 1h 05m 26s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/b27d2230893f4307b8e2c968b279eccf : SUCCESS in 1h 32m 47s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ed836f046cc4b11b6680d9645c4a556 : SUCCESS in 1h 16m 12s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/19aa0c0db79843859347a307e61de331 : SUCCESS in 1h 46m 41s\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/1ec6a1e703eb486ab6c26b3b2f70a40b : FAILURE in 1h 20m 40s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/f54a4d477473490f9781b76f0c8af757 : SUCCESS in 1h 28m 38s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/546e3c64aaaf4367808beb06c0dbe680 : SUCCESS in 35m 37s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/4da0c00fdedf4453b1b096e2eafce33a : SUCCESS in 26m 27s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/94742f4317994ebc97069c24842a9d26 : SUCCESS in 28m 06s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/22bd984aca45456b8dc6f1de23209ac8 : SUCCESS in 39m 05s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/d2694520259342bf98797465cfe173cb : SUCCESS in 57m 36s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/73f85c47cf814b11af807876a5276e50 : SUCCESS in 38m 10s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/9f96a0eba60545939849692a332ca062 : SUCCESS in 52m 43s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c1ca8e862c3f4296b38c6f4a3d137a0c : SUCCESS in 21m 24s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/717d78f898c94d20973e0d68a357831d : SUCCESS in 1h 01m 58s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/cdbd9fa9384640f99667eff5d00b5e50 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0920e1a07b3e41aa815126ad1833e264 : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c471ceab24a7480d9824410db391568f : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/141f32a4f47849fc84b111ccef2e42b0 : SUCCESS in 3m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0456a9b4d2f744a3ab12e44d4e43a61e : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b07b7af0bcc4449aa809dce88d9c98f : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85825a4708ff4b85849b7c721a786e1f : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2bcb46754e447de9506d6dc2f8e8eef : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":3},{"id":"44a9133826cc3d2b845f5cae29601af69e701c60","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-20 14:43:22.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3f983568d286daceee29500efd22475a0213d3c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-09-21 03:00:01.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"6d7d6c383982b64bc11c7a4ebf56a57b2ee191b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 04:38:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2522ab84a16440b49d905dd8915b7869\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/2dc66f247fee4bc6a505690cbb5cf71b : SUCCESS in 44m 10s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/6f93b59f754f427ebc6e212a516c31c5 : SUCCESS in 41m 49s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/3a2e5fbfbf0e4204bba4a5ad21169ced : SUCCESS in 50m 38s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/274c40939f4b4b2fbc2bc1510071ef79 : FAILURE in 1h 22m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0f8280e989bd4d3d8925833b15c613b8 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8d9893ca50ba40ff8b5323679a9eb151 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/07a106d439894c23819e6003f8d36532 : SUCCESS in 8m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"79a9bd25d16e2ffb50936899367b4e0311a77d5b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-21 04:46:33.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2b91e5380da4eb36abdb3d7673e2abcf9792fa7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 04:54:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f3b080924924d108569e324b5b1224d\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/43d8673115544252b7093f65c6f65695 : SUCCESS in 20m 31s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/00143eb177fe477c81c902824247fb0b : SUCCESS in 20m 17s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/3712fe1b2fe348ae9bfe6440766f6217 : SUCCESS in 25m 03s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/2dcb46f99b454a318adf7f122733e0de : SUCCESS in 47m 23s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f643b790a0454b2f8b7eb70c90106b8c : SUCCESS in 22m 48s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a33286ea781f4bd6add0f1daeb0f0b9d : SUCCESS in 21m 15s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/fdae360a201544be94d103396fc39666 : SUCCESS in 1h 17m 07s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/a44ed1cfbecd4aad93acf2f7193bf159 : SUCCESS in 1h 45m 39s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/477a8019352944b396f4055e1c54cb63 : SUCCESS in 1h 05m 59s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/3b9c9a9e573b4c638cec77ad140fd3e7 : SUCCESS in 1h 32m 02s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/756744946b3d4647ba708cb6d37c3ad2 : SUCCESS in 1h 18m 01s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/2b8f7ff194344072b6f579a63d27631f : FAILURE in 1h 13m 15s\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/2bad4d6f7b0b497ea6fb6c0e4d4f0319 : FAILURE in 9m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/504d89e3ddea434c8e4c30a878c557a1 : FAILURE in 8m 04s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/fb4c8757c86b43489a44cf5f525c06da : FAILURE in 7m 51s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/52983078b7f74dbea58353323cc2809b : SUCCESS in 24m 51s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/124cb37367bd4ec1b441b2bef57b83b0 : SUCCESS in 27m 10s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8b059b4b4487465c936aa796f5d5781e : SUCCESS in 22m 31s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/9a4dde92cc8547589f9966806b4afac1 : SUCCESS in 52m 52s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ee9304239fa94032ad815916a4eb3295 : SUCCESS in 23m 13s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/c6c1b5d71dca485c877b68a834a52383 : SUCCESS in 49m 26s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/916f2007daae4fbc92c38ef7d3e7bb94 : SUCCESS in 40m 48s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/adf4ca4fc4494192bc3b1a8855fd3536 : SUCCESS in 55m 16s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a079c027e59743819273ee599ddde7a9 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dafabfce99b4dc0a407838b0aedefc7 : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2560f6b7deff4210a232a089a1a5d0c6 : SUCCESS in 2m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/16ff51797afc463ead1ee803bdaf26a9 : SUCCESS in 2m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b364c5c940c84e36b9c06591acb94a9a : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ce2ab6bfe2643359e8c46acb1660bf2 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fccc423221346c88cad63e5473f3fb2 : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/588247ccd427433996b0f94498cccd68 : SUCCESS in 11m 06s","accounts_in_message":[],"_revision_number":4},{"id":"7dfbeb411c78774f86d07f5bccd20edce62275e5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-21 10:36:10.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"aba1909b074de17923201a54d03bc3ae9a98cf62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 12:39:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5eeb4e68cb946de93ff502aab55089e\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/626cfebd1eda435da375ac2326778793 : SUCCESS in 48m 15s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/c005ad4136cf4256bdbfb447051af265 : SUCCESS in 53m 00s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/ee12d520e6944f6c91dae06bb703d7f3 : SUCCESS in 46m 56s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/c308918fa1ba4e8b85aefb55cca1d30f : FAILURE in 1h 21m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/84aad4cd78b04e86a78cdbe2c620f0cb : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d7d4e438e7c94de7b9de79c4543b03c9 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/38c1da75685744f5a1b611eeef325f7b : SUCCESS in 10m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dc9317d93ef8a2bb64093bea4fcbcc9f4ede7982","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 12:56:30.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b9ab5f5acfb4593a4dfe6e5ac85247a\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/8ff710eb2d884ea099aa26f299cdb3ae : SUCCESS in 30m 11s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/19110b08acbc4ddd85b6a0077a257cda : SUCCESS in 26m 01s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/38b1c93f327f4f9db8a079093224b1cf : SUCCESS in 22m 26s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/b8985afad33d4d568db5839e9e4f3f45 : SUCCESS in 43m 15s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/595d8d4771c44483bfef96c4debcf358 : SUCCESS in 18m 14s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c27685733fd54e98b47470a2aa333a5f : SUCCESS in 30m 59s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/81aeac60c9484783ac6acbaf6ce8e430 : SUCCESS in 1h 25m 30s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/3fecceaac7c14e6085f2de0bba849c68 : SUCCESS in 2h 00m 07s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/854fcc3a5335410aab36975290a5cb16 : SUCCESS in 1h 13m 48s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/cd69efe27adf4eba9e9a4e416c2f0501 : SUCCESS in 1h 33m 57s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/caba73e17e404230ab2c97db93d8bc23 : FAILURE in 12m 16s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/06b60795a82241cdba030293eec3dc93 : FAILURE in 22m 23s\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/8ffe17696f2747b3aa2dba01645326e3 : SUCCESS in 1h 30m 56s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/617a0f19508f40378c1f2a67bc4e07d0 : SUCCESS in 1h 34m 28s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/c678a362c6754053b19631180847c952 : FAILURE in 10m 09s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/1fad6e831865455bbfabd6758c537e9d : SUCCESS in 30m 00s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/12e42c6dcb1440dfab0d4f2d79965287 : SUCCESS in 30m 18s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/4b0753fca3e2442d94108eb5edde0041 : SUCCESS in 33m 01s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/2ff181d86689410a8bb454979a0982d8 : SUCCESS in 53m 47s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/6c5bc13c06a241bb9d435ccc5df06b20 : SUCCESS in 40m 21s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/39aeb3d047a84851aee1b92756668cbc : SUCCESS in 54m 19s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/187cfbf9f9e94e08a613fde516e9f967 : SUCCESS in 40m 53s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/4480defd80204d2e9716fb6173596610 : SUCCESS in 54m 25s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ea54934b01a14b54aab1e7d46443ae17 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce9168f6a9af4267bdef0b2121420e35 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ff9603313244032b3f2f1127ccccba7 : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd2216044af4415289302d0d4920073b : SUCCESS in 2m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5158713e91be4c67991d9f2d48634678 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08cd6baab3dc4c7c9635218c93cd4d46 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/289d2d61d303441e8cc4a79e2a50fd2a : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54b0d87e620446ed8c8ac043707dfee5 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":4},{"id":"52693cee20f41620fd8e5beff23805f663aa26f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-09-22 02:33:25.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"aeda3a40b1a177a5c8a2bb16bad0309d86d6f019","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 05:01:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40f94d8a23a946bca5806b6eb557a671\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/1f4dae187d504b8c89bd819adf9077a1 : SUCCESS in 52m 15s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/07cbae5cf4e64b16924505fa15e91278 : SUCCESS in 52m 11s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/8f64a6a359384e71ba32c987b120e5db : SUCCESS in 46m 57s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/c75396813ac1498092e13617ef87fa19 : FAILURE in 1h 21m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/166aadf6a3044cfb87660109a69d69ce : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e2892097e43547f6b1a5d10b0efdcfae : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/90d4ef77274b4bfa88fc765ba8ec3c52 : SUCCESS in 9m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1e0673283afcd6d0c9ea446c451a47a66b2cd02a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 05:03:30.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d814c4ae76ed463ea88120c671ab9f0f\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/f4506b4033aa4c71b7316be078280582 : SUCCESS in 24m 04s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/846267e0795644f9bf64226a822fc4d9 : SUCCESS in 31m 24s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/7caadfbb1b9145c08fbd76e14236925e : SUCCESS in 24m 14s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a732427627fc462ead92fe9f36955f3d : SUCCESS in 1h 14m 49s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85d2a596a6a34662bc92bab72096ef21 : SUCCESS in 31m 51s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4fbc18b3e99044ac8e484ca9765b6e97 : SUCCESS in 25m 00s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/2cd40ffe39b14bd58a98d4e51268687a : FAILURE in 9m 54s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/00bbeca5a7c1444e81b94142fd3cd6d7 : SUCCESS in 2h 07m 02s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/09a6481b0af144d9872f1818966a95b9 : SUCCESS in 1h 14m 23s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/68941f2a636849fbb6c5235275612c89 : SUCCESS in 1h 39m 28s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a98081fee814f65a47c5bb8afa16804 : SUCCESS in 1h 15m 41s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/1c5ef70f1f0241f489ad371265d5d249 : FAILURE in 15m 46s\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/827d62c10aa7449093d3fc56c0ca72a3 : FAILURE in 1h 08m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/3f715c7ff22f44a58effcda1d931a884 : SUCCESS in 1h 39m 09s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/5f1d1c89242a4ca689bea9533694a0bc : SUCCESS in 52m 06s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/dc9600036e174d3cb02d0b4a03ba0ebf : SUCCESS in 29m 41s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/e3ab32ab93f94c16907680bb86cb4f31 : SUCCESS in 33m 15s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/cd7e3013a2b1432c82db18e56f1230a2 : SUCCESS in 26m 08s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/cfc1d36f645f49029740b0bb423b84b4 : SUCCESS in 30m 41s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/839b420b777540b38c383fb98e598db1 : SUCCESS in 33m 59s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/4ccf46fd099c48d48bf2c5efc81edef2 : SUCCESS in 52m 36s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b4c2a77463d141ffbeb5248a4d7a0f1f : SUCCESS in 40m 19s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/f026b304eccf472b8c96f36bf5daaeee : SUCCESS in 42m 58s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3ea248e9087e44a99a183b0fdef41e11 : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1442b5c7f1284ed2bec93e738b069d70 : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/85f926897f8f402c97dd5cafe4309cb2 : SUCCESS in 3m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b10de2acf41d4890912340b76283e5d2 : SUCCESS in 3m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/698c28ebc4454927bf0b947dfa947d4c : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dcb8ec49c6f4aa4a77bd4e54e07790d : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc04321dd32d4ffeb0ef4f920c913359 : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d98ccf35db1484c8a74a9c3be09ffb3 : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":5},{"id":"e47452ff2fe462a041f8e3bf113dbb8a6c29e789","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-22 20:35:41.000000000","message":"Patch Set 5: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0a7003bb62400f8adbf9b7cf2292f3e84adfaa66","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 22:15:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04d1ec79bfa14d63a6337ba5a2788611\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/ca2832ab8bc14b47894804e4d6799ea9 : SUCCESS in 53m 59s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/1a1c9bc9fa19458bb3edd0bbb07c18d4 : SUCCESS in 1h 03m 14s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/65a5b6709bd9485fbaf77ed56df9d52d : SUCCESS in 51m 45s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/07655859b6214d6da588d92bda585793 : FAILURE in 1h 32m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/095622f3fcfb40e49704273e776aeb43 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/539883ce89354d48aff32198ef5d72c2 : SUCCESS in 9m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"58441e5f5b9bd239742d0a25ad3726c1d1687678","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 22:33:11.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56f9f1f5f6e044bdb00189b5404a28f1\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/6029b63f504f4dbcabf93c518c40684c : SUCCESS in 27m 33s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/752062b357194a979259ee769abfee8d : SUCCESS in 28m 44s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b6bb2692b86b485fafa4f718b1c5a0ca : FAILURE in 21m 23s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/63d6e33f41a34383a971e53417f7251a : FAILURE in 55m 15s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f31129ef0334f41833987f4a3c6c474 : SUCCESS in 36m 07s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/055c21b8199c4e68bc0074c498096f73 : SUCCESS in 33m 23s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/ad1b815cc1b347699984f37ca5f82a5d : FAILURE in 23m 12s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/06429f6de7ed4ae9b91eb02049c10adb : FAILURE in 9m 58s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/02f69b5286c242c29b9587e1919fae81 : FAILURE in 18m 33s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/7075297b983048f5a7b5e5a2337e5765 : FAILURE in 17m 45s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/936d1ef71fd34c038b6332ef7cb2bb73 : SUCCESS in 1h 30m 19s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/0630b6a83fd2424aa890eec66a2fe5ef : SUCCESS in 1h 45m 26s\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/5a5769cb49cc440ba71c8cca905b7fe9 : FAILURE in 16m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/5526149cba4c40a0922769f26f5c8171 : SUCCESS in 1h 35m 37s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/9c8d3826e0b44275aff6925360bb2842 : FAILURE in 16m 46s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/f178457baf8a4c2086e33e687b7a29f7 : SUCCESS in 25m 36s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/218dc44584264ac98fa3adb561f80d7a : SUCCESS in 19m 37s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7d8d7089017945e9b51a1ad7179e7340 : SUCCESS in 23m 15s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/6d27bccc11af467daf011ec4d4a6b327 : SUCCESS in 1h 02m 38s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/84532225dd8d413da4855a8837538a0e : SUCCESS in 49m 46s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/12483a949da0449e8afa085ed9bdfbda : SUCCESS in 39m 26s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f62aa0aa73f94f2ab57d1899b7c20d38 : SUCCESS in 39m 27s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/aeba0f96fc344634963bc57e2cf7f84e : SUCCESS in 57m 43s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/74685c5325ed411e80727b94cf916024 : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c47b59c42e2a4a7cb4a2a0ca7c638861 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/113d172a89234b0d804e4bdc6202a1c9 : SUCCESS in 3m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/56d3ca06057a432eb95d8309efd4420d : SUCCESS in 5m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1117dffa5ee2486db686f1b2d9da03a2 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5444f65baa04500bb26b14cbcb507ae : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a8cb59939184609bf917944e320a364 : SUCCESS in 10m 31s","accounts_in_message":[],"_revision_number":5},{"id":"775d9d3906fc44a421a8ce362777d36b3f8e1519","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-09-23 02:29:58.000000000","message":"Uploaded patch set 6.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"5adb5f1ab8006d2518cfb9cfff16d896e2570bd4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 05:30:57.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87df7a3c1b144d01b845339519f39f15\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/eb3c6b51fdc7486a853f8a60755f566f : SUCCESS in 56m 21s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/4bf3f117090243f9859b58b89bb4074f : SUCCESS in 51m 32s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/8eb436bb8f7f4db5afa27908cdefe191 : SUCCESS in 53m 00s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/53cb478ae0c744c6b5ae2f2805682b71 : FAILURE in 1h 24m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c7dcf2e24ee84afc80dfa887c59f6d6a : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0d5a4896f6394070b81c4f6d8a7c20d1 : SUCCESS in 6m 54s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"06281ddaadf1547b631a17fbede9b099eca54994","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 05:35:40.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d404e0b10c4f43aca5d5284e0b7d5347\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/873a993085e84127b53f049cc319b03b : SUCCESS in 23m 57s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/f074071eae854656b189f699d393b11b : SUCCESS in 20m 18s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/c15667c2301d448c8749f98801bf3e11 : SUCCESS in 29m 25s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/0bd15d3d401b44ff89691f1e80832894 : SUCCESS in 55m 04s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ba89b0e68fe4222aedf84029dbce8ea : SUCCESS in 21m 59s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/36cdd6c2e7de47398bcf1c7e09b421ee : SUCCESS in 27m 16s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/76926911b6ce4622b45d9aefe2e6a7c6 : FAILURE in 12m 32s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/b5925ab6ece14d65b32360fc2ae02dd2 : SUCCESS in 2h 12m 04s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/728d72707db545f980b3e72fc15882bd : FAILURE in 5m 12s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/1d12c229cd7247d9ba16305f0224fb13 : SUCCESS in 1h 30m 41s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a035c18eed64ac6ac24fb1beaae0fa4 : SUCCESS in 1h 25m 41s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/555cb8fe52454953ad368b36d762692e : SUCCESS in 2h 03m 08s\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/36b45f3bb5c949a5b7f0c97d2fd2d7f9 : FAILURE in 1h 19m 11s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/b9253e73046945d6ac5fcf1e449b64fb : SUCCESS in 1h 34m 50s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/f7bc4b4d567a47d284b3b7cf5642cd3a : SUCCESS in 48m 23s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/7e78f13f7b024ee88aa3cc1bd7b1eed4 : SUCCESS in 27m 13s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/1f0b60b7f6c6440f976189c9768c4d94 : SUCCESS in 26m 21s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8b88e40701024cc2b057c26e98249d4b : SUCCESS in 37m 02s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/6654984594f9409b9e84eedd320f51e5 : RETRY_LIMIT in 1m 52s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e12df609574e4583aeac08fbbd8c50a9 : SUCCESS in 24m 08s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/069bf41b72a648138c674ea8a9972e9a : SUCCESS in 52m 12s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/2b17a8b1f7274a25affcf6a88414dd0f : SUCCESS in 21m 09s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/0969bca7e21e46198e7c9993a1bdf04d : SUCCESS in 1h 07m 04s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7e6a04f81c2942769dd5588bd3b73651 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b927e500c094b15b454bc8e16ae4dfb : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd5eeaa761bd4e4898a874342ce66110 : SUCCESS in 3m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bd2dd0a7129d45318ef8b498848a01c3 : SUCCESS in 7m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/548e60522a8a431cb86b48a6f8e0cf35 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e994797322b49de849a9e134527f082 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec011f7faf2f4cc39923ce614407cd36 : SUCCESS in 10m 43s","accounts_in_message":[],"_revision_number":6},{"id":"ccd3359d7a82b7330673e1ed187fab44ec4d85ee","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-23 05:45:46.000000000","message":"Patch Set 6: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"73991bb06767c8982a52fad2e9ae3997c8fcc79d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 07:04:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96214e40ede840b1ad07375df0b6543d\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/0206d666f6ff4e9994e1e4799e1bd13c : SUCCESS in 42m 28s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/9a2d102ab829402e83c712e7ee763269 : SUCCESS in 39m 13s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/156027387a0f4054aca7340513a7b1ae : SUCCESS in 45m 59s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/4551c94c3f5d441dac64fcb190d5617d : FAILURE in 1h 14m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b436ccca309b4910a4315cbde11cdc8c : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/29cb79a5f548464f818de44ae561d5bc : SUCCESS in 6m 17s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"68de235820097bbc8ab7ca67e7e995afcceb0dfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 07:18:25.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/93f4dc8f6797488aa9e3a1e284f1eea6\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/9443dcc667c047a6b1d57f3dbd41caa9 : SUCCESS in 29m 14s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/9c62d802d83f4540982bcc0aa0eb34ad : SUCCESS in 29m 35s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b59c14dfaf504727ad5a17a53dbb0e53 : SUCCESS in 19m 07s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/219532ae3ed64a5780b35acc06aa0346 : SUCCESS in 54m 59s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d8fc1171b8d4829949afc662684475e : SUCCESS in 27m 10s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f188cd6efad0405ea163f9ecfa825a33 : SUCCESS in 34m 40s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/401b6eadff4b4598961aeed8b7b4822c : SUCCESS in 1h 23m 35s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/ec31728627cd4f0d91b9e4011137b280 : FAILURE in 1h 08m 14s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/2ccc94342b7949b1b9b556b938780200 : SUCCESS in 1h 02m 49s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/c88bea97b323459eb780f9564b8ba133 : SUCCESS in 1h 29m 19s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f39df3de24eb4158a1b38a185726fc8b : SUCCESS in 1h 25m 32s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/57c9aa593b5b4306a0a2e02c3108c7e6 : FAILURE in 8m 56s\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/4e319c62e1954649b05deaa87efac08d : FAILURE in 27m 41s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/7601d72cb4e840e1b2f40e7e61e9ff35 : SUCCESS in 1h 30m 13s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/a76f8e6071b64da2991a8766d0e63ca6 : FAILURE in 7m 48s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/31dac3c4df8f482ead6ca640dcea0a77 : SUCCESS in 24m 09s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/51b1a4f005234d1ba77c6da7ff8b0d5d : SUCCESS in 16m 39s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b6a52a86fca046e19aed6e8bc9a1c3e8 : SUCCESS in 28m 17s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/524841c3532c4da3848d4e9b6fdf6659 : SUCCESS in 53m 24s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/eae34f6c5f494becab7362f0fbc5db09 : SUCCESS in 20m 09s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/3938656c058041fba6e45755525338fb : SUCCESS in 31m 28s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d9cb356adcc54679aca86eac11cce897 : SUCCESS in 39m 48s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/6c485a6975a849ea94fbad76dde70f24 : SUCCESS in 52m 50s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/29e6cf0c443b45df9e992246ee5edafb : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4801abfcb2d4e72aabd07b97eec2501 : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cdbf065c07dd450c988c2c80c5eb7f52 : SUCCESS in 4m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/181f055d2a924676b74503da53b4e7e3 : SUCCESS in 5m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b91f5fb3b2054963abbd1e24517289e3 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0baec0450da74153aa4678d73c453f83 : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37c6daf3b49f419b9e2463223f7b1a41 : SUCCESS in 10m 41s","accounts_in_message":[],"_revision_number":6},{"id":"44080dd9447898fe9d457586586f510572fa1e1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-09-24 02:27:29.000000000","message":"Uploaded patch set 7.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"03fed185156945ad59130fe3d7fcfb4549c7f065","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 04:25:52.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c14264e937e547669a060d5caf03b2bd\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/4ec5515be811499ca7e47a04e348338a : SUCCESS in 58m 17s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/d696051ce2fb445796978230e7772b5b : FAILURE in 53m 35s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/38145df37ec844a58079112f426518d3 : SUCCESS in 55m 53s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/eb354a719f764142ad2c3b25baaba462 : FAILURE in 1h 12m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7e3472e1e8af496190cc0259bb0cfa56 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a1ca65f2d2f5409082527a18628c5c59 : SUCCESS in 6m 19s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3a84a24c0cae82d3cc4de04d87163426926c27ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 04:49:29.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9842106b2d314d709b7085db90a068bf\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/2c08ae1d51d54e22b9a10b5ef5e3c940 : SUCCESS in 22m 58s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/d08819fcaad549c5b162623199d626ee : SUCCESS in 17m 05s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/0959f31b002e4fdfb67d1373870bed86 : SUCCESS in 28m 16s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/0da2d5c2f9c944cb90bd0ae52ed11c3d : SUCCESS in 57m 18s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b3dd6bb39d64edb84198a1840d4ae03 : SUCCESS in 24m 02s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c4a1bb86388c45e2b451416fe3b10776 : SUCCESS in 25m 09s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/008ed40730bb41d785e0086c8e047795 : FAILURE in 11m 47s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/bf3d1bfd46274b1191499d775885dc9a : FAILURE in 9m 48s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/0ab5759855a64dc28ae81d577ae0afa5 : SUCCESS in 1h 04m 16s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/2de88f5d568c44cd8b871d1d56030bbc : SUCCESS in 1h 31m 45s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dbfdab7f31ca47529305dc76ab054ec7 : FAILURE in 10m 23s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/7a27fa000f8e4138ab18b51914816d1d : FAILURE in 17m 29s\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/81fbdd5ddd274c8ea68202487a006963 : FAILURE in 15m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/00dd41d8d8574c7c9edeae91312e51b0 : SUCCESS in 1h 38m 33s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/a318cc3daf9140869510a22828881377 : FAILURE in 9m 32s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/defeac4f0c87462cb464a0c809c1f3fa : SUCCESS in 26m 44s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/01e56c9888cd462aa8340b0f86ad97e8 : SUCCESS in 26m 16s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/732733faf6bf44f58acbdffcbfa8cdba : SUCCESS in 40m 11s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/3dd991cbbb09405983083b88144f2605 : SUCCESS in 57m 07s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a45f31f1e4c34365bfa99492f665c737 : SUCCESS in 23m 38s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/6c3121054f12450dbb65abb38360964a : SUCCESS in 53m 13s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/19507b177072470797e6ffc34c3934f9 : SUCCESS in 43m 12s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/8728a952245c4d24aca1c7e43ae93336 : SUCCESS in 55m 24s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/90100d0d849340ec89770133adb5371b : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9ab7ca5f3c14aa295b83cfb92553a8f : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/235fe0a0172745ca8ec5e5ca86f1f11c : SUCCESS in 3m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/84c4388e509e4447937c17ec2706f852 : SUCCESS in 5m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71c51e9c99d44d3f80383821b1c04f14 : SUCCESS in 2m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef9f42916c6442b4900a2dded4fdd1d9 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4128594eb844ad2954f9aacb607ac78 : SUCCESS in 8m 07s","accounts_in_message":[],"_revision_number":7},{"id":"7a8a8ec7c039641465c8b936b551ddf6d5a778d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-09-25 02:38:18.000000000","message":"Uploaded patch set 8.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"90d3a8d8abce2214019c05be0080f34b17980900","revisions":{"51442a2722ab72ca759e740524f4a76dc8a81314":{"kind":"REWORK","_number":1,"created":"2026-09-18 02:55:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/1006144/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/1006144/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/1006144/1"}}},"commit":{"parents":[{"commit":"0cbc63835b7d24c726da1bb44a46fbdd3e09680e","subject":"Updated from check-sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cbc63835b7d24c726da1bb44a46fbdd3e09680e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-18 02:55:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-18 02:55:38.000000000","tz":0},"subject":"Updated from check-sources","message":"Updated from check-sources\n\nChange-Id: I11fb14bd470e78eb5fe6f75a6494ef61483e67d8\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51442a2722ab72ca759e740524f4a76dc8a81314"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51442a2722ab72ca759e740524f4a76dc8a81314"}]},"branch":"refs/heads/master"},"8bfcc0360daaad2e62dbcb94593bb164cdb70fe5":{"kind":"REWORK","_number":2,"created":"2026-09-19 02:38:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/1006144/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/1006144/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/1006144/2"}}},"commit":{"parents":[{"commit":"cd36091e3b20a33df27cae4130085b75c3f9dcab","subject":"Merge \"grafana: replace deprecated grafana-cli\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd36091e3b20a33df27cae4130085b75c3f9dcab"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-19 02:38:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-19 02:38:35.000000000","tz":0},"subject":"Updated from check-sources","message":"Updated from check-sources\n\nChange-Id: I11fb14bd470e78eb5fe6f75a6494ef61483e67d8\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8bfcc0360daaad2e62dbcb94593bb164cdb70fe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8bfcc0360daaad2e62dbcb94593bb164cdb70fe5"}]},"branch":"refs/heads/master"},"1c73e18cce08245aa4a12c23485298482cac60a0":{"kind":"REWORK","_number":3,"created":"2026-09-20 03:07:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/1006144/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/1006144/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/1006144/3"}}},"commit":{"parents":[{"commit":"cd36091e3b20a33df27cae4130085b75c3f9dcab","subject":"Merge \"grafana: replace deprecated grafana-cli\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd36091e3b20a33df27cae4130085b75c3f9dcab"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-20 03:07:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-20 03:07:21.000000000","tz":0},"subject":"Updated from check-sources","message":"Updated from check-sources\n\nChange-Id: I11fb14bd470e78eb5fe6f75a6494ef61483e67d8\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c73e18cce08245aa4a12c23485298482cac60a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c73e18cce08245aa4a12c23485298482cac60a0"}]},"branch":"refs/heads/master"},"549338d8683726c4b185156c286c1f585af2f5ad":{"kind":"REWORK","_number":4,"created":"2026-09-21 03:00:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/1006144/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/1006144/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/1006144/4"}}},"commit":{"parents":[{"commit":"cd36091e3b20a33df27cae4130085b75c3f9dcab","subject":"Merge \"grafana: replace deprecated grafana-cli\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd36091e3b20a33df27cae4130085b75c3f9dcab"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-21 02:59:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-21 02:59:53.000000000","tz":0},"subject":"Updated from check-sources","message":"Updated from check-sources\n\nChange-Id: I11fb14bd470e78eb5fe6f75a6494ef61483e67d8\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/549338d8683726c4b185156c286c1f585af2f5ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/549338d8683726c4b185156c286c1f585af2f5ad"}]},"branch":"refs/heads/master"},"fb2e62499237b224d85f8c55aa933ea53d660b09":{"kind":"REWORK","_number":5,"created":"2026-09-22 02:33:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/1006144/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/1006144/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/1006144/5"}}},"commit":{"parents":[{"commit":"cd36091e3b20a33df27cae4130085b75c3f9dcab","subject":"Merge \"grafana: replace deprecated grafana-cli\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd36091e3b20a33df27cae4130085b75c3f9dcab"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-22 02:33:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-22 02:33:23.000000000","tz":0},"subject":"Updated from check-sources","message":"Updated from check-sources\n\nChange-Id: I11fb14bd470e78eb5fe6f75a6494ef61483e67d8\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb2e62499237b224d85f8c55aa933ea53d660b09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb2e62499237b224d85f8c55aa933ea53d660b09"}]},"branch":"refs/heads/master"},"610bc7b3f4521c965d9d3e8d7921eaafd2bd506a":{"kind":"REWORK","_number":6,"created":"2026-09-23 02:29:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/1006144/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/1006144/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/1006144/6"}}},"commit":{"parents":[{"commit":"4e2c9782c4608231e39a5558d04e7b6b6715871f","subject":"Merge \"Allow lowercase from\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e2c9782c4608231e39a5558d04e7b6b6715871f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-23 02:29:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-23 02:29:53.000000000","tz":0},"subject":"Updated from check-sources","message":"Updated from check-sources\n\nChange-Id: I11fb14bd470e78eb5fe6f75a6494ef61483e67d8\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/610bc7b3f4521c965d9d3e8d7921eaafd2bd506a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/610bc7b3f4521c965d9d3e8d7921eaafd2bd506a"}]},"branch":"refs/heads/master"},"d5297dd4512a2fb091bccf43a66d5b7d548fe9cf":{"kind":"REWORK","_number":7,"created":"2026-09-24 02:27:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/1006144/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/1006144/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/1006144/7"}}},"commit":{"parents":[{"commit":"3f51ec474be2fcd493753c6b78fd26de4927f6fb","subject":"Updated ansible-lint in lint-requirements.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f51ec474be2fcd493753c6b78fd26de4927f6fb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-24 02:27:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-24 02:27:24.000000000","tz":0},"subject":"Updated from check-sources","message":"Updated from check-sources\n\nChange-Id: I11fb14bd470e78eb5fe6f75a6494ef61483e67d8\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5297dd4512a2fb091bccf43a66d5b7d548fe9cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5297dd4512a2fb091bccf43a66d5b7d548fe9cf"}]},"branch":"refs/heads/master"},"90d3a8d8abce2214019c05be0080f34b17980900":{"kind":"REWORK","_number":8,"created":"2026-09-25 02:38:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/1006144/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/1006144/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/1006144/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/1006144/8"}}},"commit":{"parents":[{"commit":"3f51ec474be2fcd493753c6b78fd26de4927f6fb","subject":"Updated ansible-lint in lint-requirements.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f51ec474be2fcd493753c6b78fd26de4927f6fb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-25 02:38:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-25 02:38:15.000000000","tz":0},"subject":"Updated from check-sources","message":"Updated from check-sources\n\nChange-Id: I11fb14bd470e78eb5fe6f75a6494ef61483e67d8\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90d3a8d8abce2214019c05be0080f34b17980900"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90d3a8d8abce2214019c05be0080f34b17980900"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY"},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
