)]}'
{"id":"openstack%2Fkolla~1006468","triplet_id":"openstack%2Fkolla~master~I117450faef298a24f975756e8601e5898484bc91","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I117450faef298a24f975756e8601e5898484bc91","subject":"proxysql: Check also match_digest and match_pattern in dedup","status":"NEW","created":"2026-09-21 08:57:09.000000000","updated":"2026-09-21 10:58:21.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"3bea67755ad1937e72dd968938318e0a19ee70cd","_number":1006468,"virtual_id_number":1006468,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-21 10:58:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"reviewer_updates":[{"updated":"2026-09-21 08:58:24.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-09-21 10:25:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-21 10:58:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b4e962a96ee3e01b46c097645970490d520c0ed7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-21 08:57:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64991971b3257737f3720d3e085e3d5659c58a6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 10:25:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c14ac288c5346a69a05d48feb59627c\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/84cf400bc6444169a352d2445f1436ee : SUCCESS in 42m 29s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/8f843195220f4eea93d5ea30e6619915 : SUCCESS in 51m 22s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/9e6d0fe3ed70466aa46779bd2835ee8d : SUCCESS in 42m 50s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/e61c7ea14d18471aae070b9647665b25 : FAILURE in 1h 24m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c5428d59dad2421999c8625ac3bd7834 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8febafb5fe0f4fc284ff0c66977f41d3 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8c967d10d37c4ac09516a55060c83ddd : SUCCESS in 9m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3bea67755ad1937e72dd968938318e0a19ee70cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 10:58:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1e8b431d09b479c9d0ad800f5f3aede\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/27ab5e00c5324c8482c126245c201e98 : SUCCESS in 26m 19s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/1a34c57d9ff04cf394d11f9da339d259 : SUCCESS in 30m 46s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/c7e45f6aadbf4c958038db79f51f64fd : FAILURE in 31m 24s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/40cfb8404fb847a085ea1044791a066d : SUCCESS in 1h 02m 01s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3ca98025e864ef098c75fd2045f7064 : SUCCESS in 28m 44s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b3af0c158d7e4d909f91a502bccb504e : SUCCESS in 30m 17s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/7597e656a73347c2a41dba20694ecc0d : FAILURE in 7m 23s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/eb0a3f1f80214b7ba9eb3e2bbf5a52a7 : FAILURE in 8m 11s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/012a055d9f69440d95664051dbb1cf56 : FAILURE in 20m 44s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/32945cc339d14b4388b080fbce6cb52f : SUCCESS in 1h 33m 06s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eacf6f9dbe594281b7a6a5e5b6ca3d5f : SUCCESS in 1h 09m 02s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/bb35ba96a2444a3eb412391b74e89f08 : SUCCESS in 1h 57m 57s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/27f4541eb77047b58e5d5bf94fc74e30 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7dddcad17454011ae74f997163ad073 : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/187aefd6c33e4efc87644add85c1e82f : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ddc5255d1d804873a99aad93e17931b1 : SUCCESS in 3m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fbfa13e782314e9fb88f9a638d87b8c3 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b502d49b0d8a4a7fa195ad59fb0257fb : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32760e56876f48eab96e2a4bc968da01 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd607c63894742af93f9ab3a9d2e711a : SUCCESS in 10m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"979e68257433738246e7615d3b8554b2d040ba4e","revisions":{"979e68257433738246e7615d3b8554b2d040ba4e":{"kind":"REWORK","_number":1,"created":"2026-09-21 08:57:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/68/1006468/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/1006468/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/1006468/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/1006468/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/1006468/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/1006468/1"}}},"commit":{"parents":[{"commit":"cd36091e3b20a33df27cae4130085b75c3f9dcab","subject":"Merge \"grafana: replace deprecated grafana-cli\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd36091e3b20a33df27cae4130085b75c3f9dcab"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-09-21 08:56:01.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-09-21 08:57:06.000000000","tz":120},"subject":"proxysql: Check also match_digest and match_pattern in dedup","message":"proxysql: Check also match_digest and match_pattern in dedup\n\nChange-Id: I117450faef298a24f975756e8601e5898484bc91\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/979e68257433738246e7615d3b8554b2d040ba4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/979e68257433738246e7615d3b8554b2d040ba4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
