)]}'
{"id":"openstack%2Fkolla~279910","triplet_id":"openstack%2Fkolla~master~I84217824817c484b6800a64cbd7767b127a3098e","project":"openstack/kolla","branch":"master","topic":"mitaka","hashtags":[],"change_id":"I84217824817c484b6800a64cbd7767b127a3098e","subject":"Use ubuntu Mitaka repo","status":"MERGED","created":"2016-02-14 02:35:06.000000000","updated":"2016-03-15 02:35:47.000000000","submitted":"2016-03-15 02:35:46.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6fc149944b3cf1a02009aac497587b912cdec7ec","_number":279910,"virtual_id_number":279910,"owner":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},{"value":2,"date":"2016-03-15 02:35:46.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15447,"name":"yongfeng","email":"dolpher.du@intel.com","username":"dolpher"},{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"_account_id":13642,"name":"Vikram Hosakote","email":"vikram.ganesh@gmail.com","username":"vhosakot"},{"value":0,"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},{"value":0,"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"recommended":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"all":[{"value":1,"date":"2016-03-14 16:01:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15447,"name":"yongfeng","email":"dolpher.du@intel.com","username":"dolpher"},{"value":-2,"date":"2016-03-15 02:35:47.000000000","post_submit":true,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":1,"date":"2016-03-14 15:53:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13642,"name":"Vikram Hosakote","email":"vikram.ganesh@gmail.com","username":"vhosakot"},{"value":2,"date":"2016-03-14 16:45:10.000000000","_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},{"value":2,"date":"2016-03-15 02:30:00.000000000","_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15447,"name":"yongfeng","email":"dolpher.du@intel.com","username":"dolpher"},{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"_account_id":13642,"name":"Vikram Hosakote","email":"vikram.ganesh@gmail.com","username":"vhosakot"},{"value":0,"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},{"value":1,"date":"2016-03-15 02:30:00.000000000","_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15447,"name":"yongfeng","email":"dolpher.du@intel.com","username":"dolpher"},{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"_account_id":13642,"name":"Vikram Hosakote","email":"vikram.ganesh@gmail.com","username":"vhosakot"},{"value":0,"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},{"value":0,"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15447,"name":"yongfeng","email":"dolpher.du@intel.com","username":"dolpher"},{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"_account_id":13642,"name":"Vikram Hosakote","email":"vikram.ganesh@gmail.com","username":"vhosakot"},{"value":0,"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},{"value":0,"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"_account_id":13642,"name":"Vikram Hosakote","email":"vikram.ganesh@gmail.com","username":"vhosakot"},{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"_account_id":15447,"name":"yongfeng","email":"dolpher.du@intel.com","username":"dolpher"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-03-14 03:05:39.000000000","updated_by":{"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},"reviewer":{"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},"state":"REVIEWER"},{"updated":"2016-03-14 05:53:44.000000000","updated_by":{"_account_id":15447,"name":"yongfeng","email":"dolpher.du@intel.com","username":"dolpher"},"reviewer":{"_account_id":15447,"name":"yongfeng","email":"dolpher.du@intel.com","username":"dolpher"},"state":"REVIEWER"},{"updated":"2016-03-14 15:23:37.000000000","updated_by":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"reviewer":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"state":"REVIEWER"},{"updated":"2016-03-14 15:53:36.000000000","updated_by":{"_account_id":13642,"name":"Vikram Hosakote","email":"vikram.ganesh@gmail.com","username":"vhosakot"},"reviewer":{"_account_id":13642,"name":"Vikram Hosakote","email":"vikram.ganesh@gmail.com","username":"vhosakot"},"state":"REVIEWER"},{"updated":"2016-03-14 16:45:10.000000000","updated_by":{"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},"reviewer":{"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},"state":"REVIEWER"},{"updated":"2016-03-15 02:30:00.000000000","updated_by":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"reviewer":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"state":"REVIEWER"},{"updated":"2016-03-15 02:35:46.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"8292b82ca65fd58b049c447e111948f4aaf4c647","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2016-02-14 02:35:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42813045bddcdfe8236b06cc3286d0353f3ebb8c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-14 03:20:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-docs http://docs-draft.openstack.org/10/279910/1/check/gate-kolla-docs/0593bc4//doc/build/html/ : SUCCESS in 51s\n- gate-kolla-linters http://logs.openstack.org/10/279910/1/check/gate-kolla-linters/90b3e85/ : SUCCESS in 37s\n- gate-kolla-python27 http://logs.openstack.org/10/279910/1/check/gate-kolla-python27/cc7d753/ : SUCCESS in 1m 16s\n- gate-kolla-python34 http://logs.openstack.org/10/279910/1/check/gate-kolla-python34/147de53/ : SUCCESS in 1m 41s (non-voting)\n- gate-kolla-dsvm-build-centos-binary http://logs.openstack.org/10/279910/1/check/gate-kolla-dsvm-build-centos-binary/5c2c51e/ : SUCCESS in 26m 30s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary http://logs.openstack.org/10/279910/1/check/gate-kolla-dsvm-deploy-centos-binary/f15b389/ : SUCCESS in 44m 55s (non-voting)\n- gate-kolla-dsvm-build-centos-source http://logs.openstack.org/10/279910/1/check/gate-kolla-dsvm-build-centos-source/9d4d2de/ : SUCCESS in 34m 39s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source http://logs.openstack.org/10/279910/1/check/gate-kolla-dsvm-deploy-centos-source/5deac48/ : FAILURE in 27m 59s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source http://logs.openstack.org/10/279910/1/check/gate-kolla-dsvm-build-ubuntu-source/19ac808/ : FAILURE in 18m 27s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source http://logs.openstack.org/10/279910/1/check/gate-kolla-dsvm-deploy-ubuntu-source/e81cbed/ : FAILURE in 11m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6bdf1d707b6cb31a6ff56c605d9902171cf62af6","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2016-02-14 04:36:31.000000000","message":"Patch Set 1: Code-Review-2\n\nWe don\u0027t switch over until its a finalized branch. Running the ppa is in liberty meant the ubuntu gate was broken constantly. We actually used the Kilo branch for stability. I don\u0027t trust cloud-archive staging stuff. Constant issues and it will affect dev for ubuntu as it did last time","accounts_in_message":[],"_revision_number":1},{"id":"389bb7afb1bcc5c5af816dcb84350947d5209912","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2016-02-14 04:48:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"547c6904972b4e0cbd9ed3b98b999f276fec326b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-14 05:37:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-docs http://docs-draft.openstack.org/10/279910/2/check/gate-kolla-docs/b161602//doc/build/html/ : SUCCESS in 54s\n- gate-kolla-linters http://logs.openstack.org/10/279910/2/check/gate-kolla-linters/3f21fe6/ : SUCCESS in 36s\n- gate-kolla-python27 http://logs.openstack.org/10/279910/2/check/gate-kolla-python27/da655ce/ : SUCCESS in 2m 30s\n- gate-kolla-python34 http://logs.openstack.org/10/279910/2/check/gate-kolla-python34/edddd22/ : SUCCESS in 1m 31s (non-voting)\n- gate-kolla-dsvm-build-centos-binary http://logs.openstack.org/10/279910/2/check/gate-kolla-dsvm-build-centos-binary/7785463/ : SUCCESS in 28m 04s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary http://logs.openstack.org/10/279910/2/check/gate-kolla-dsvm-deploy-centos-binary/d09b6a9/ : SUCCESS in 36m 22s (non-voting)\n- gate-kolla-dsvm-build-centos-source http://logs.openstack.org/10/279910/2/check/gate-kolla-dsvm-build-centos-source/0f2ee8e/ : SUCCESS in 26m 51s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source http://logs.openstack.org/10/279910/2/check/gate-kolla-dsvm-deploy-centos-source/6dbf150/ : FAILURE in 40m 28s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source http://logs.openstack.org/10/279910/2/check/gate-kolla-dsvm-build-ubuntu-source/1a94343/ : SUCCESS in 22m 53s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source http://logs.openstack.org/10/279910/2/check/gate-kolla-dsvm-deploy-ubuntu-source/09f9a5d/ : FAILURE in 14m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cfca3e0143457fde0d4bd255e0b420296e2ac3ee","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2016-02-14 18:18:48.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5821eaac3c5934cbaf9bcdb0d4491126c8454bcd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-14 19:18:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-docs http://docs-draft.openstack.org/10/279910/3/check/gate-kolla-docs/c3cd394//doc/build/html/ : SUCCESS in 58s\n- gate-kolla-linters http://logs.openstack.org/10/279910/3/check/gate-kolla-linters/c738790/ : SUCCESS in 57s\n- gate-kolla-python27 http://logs.openstack.org/10/279910/3/check/gate-kolla-python27/38abd7c/ : SUCCESS in 1m 03s\n- gate-kolla-python34 http://logs.openstack.org/10/279910/3/check/gate-kolla-python34/c2ec8a2/ : SUCCESS in 1m 35s (non-voting)\n- gate-kolla-dsvm-build-centos-binary http://logs.openstack.org/10/279910/3/check/gate-kolla-dsvm-build-centos-binary/91a1cab/ : SUCCESS in 28m 50s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary http://logs.openstack.org/10/279910/3/check/gate-kolla-dsvm-deploy-centos-binary/1a9b0f3/ : SUCCESS in 27m 30s (non-voting)\n- gate-kolla-dsvm-build-centos-source http://logs.openstack.org/10/279910/3/check/gate-kolla-dsvm-build-centos-source/4c27b19/ : SUCCESS in 51m 39s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source http://logs.openstack.org/10/279910/3/check/gate-kolla-dsvm-deploy-centos-source/ea17b5e/ : SUCCESS in 26m 43s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source http://logs.openstack.org/10/279910/3/check/gate-kolla-dsvm-build-ubuntu-source/caf1395/ : SUCCESS in 19m 07s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source http://logs.openstack.org/10/279910/3/check/gate-kolla-dsvm-deploy-ubuntu-source/d26fdb9/ : FAILURE in 16m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aceaa0f55fb32ad3088ef328f95f8a4e0e81a8a6","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2016-03-09 06:51:42.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"f647889d6aa98469d5f24dfe90839a01cbdb8a03","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-09 07:29:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-pep8 http://logs.openstack.org/10/279910/4/check/gate-kolla-pep8/9cb3352/ : SUCCESS in 3m 15s\n- gate-kolla-docs http://docs-draft.openstack.org/10/279910/4/check/gate-kolla-docs/b33efe6//doc/build/html/ : SUCCESS in 2m 17s\n- gate-kolla-python27 http://logs.openstack.org/10/279910/4/check/gate-kolla-python27/4d7a965/ : SUCCESS in 3m 36s\n- gate-kolla-python34 http://logs.openstack.org/10/279910/4/check/gate-kolla-python34/5a37c9e/ : SUCCESS in 2m 07s\n- gate-kolla-dsvm-build-centos-binary http://logs.openstack.org/10/279910/4/check/gate-kolla-dsvm-build-centos-binary/7ac9ff1/ : SUCCESS in 28m 11s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary http://logs.openstack.org/10/279910/4/check/gate-kolla-dsvm-deploy-centos-binary/ebeb827/ : SUCCESS in 17m 46s (non-voting)\n- gate-kolla-dsvm-build-centos-source http://logs.openstack.org/10/279910/4/check/gate-kolla-dsvm-build-centos-source/a682334/ : SUCCESS in 32m 45s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source http://logs.openstack.org/10/279910/4/check/gate-kolla-dsvm-deploy-centos-source/52dcbbe/ : SUCCESS in 36m 13s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source http://logs.openstack.org/10/279910/4/check/gate-kolla-dsvm-build-ubuntu-source/9c09030/ : FAILURE in 23m 11s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source http://logs.openstack.org/10/279910/4/check/gate-kolla-dsvm-deploy-ubuntu-source/7b7ccf9/ : FAILURE in 10m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8f0f43dc2bf53efa33f66e529c71ce14c004cae5","author":{"_account_id":13642,"name":"Vikram Hosakote","email":"vikram.ganesh@gmail.com","username":"vhosakot"},"date":"2016-03-09 21:03:34.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"15abf2c8fcf576d3d338e5095f06796cc2d904fa","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2016-03-09 22:19:31.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a5b92f2bb2aed3d7ad0226707d7d2071e1e0d472","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2016-03-13 16:28:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c421c5ccefdc12ce21535b9fe4da342d792b9c4e","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2016-03-13 16:48:07.000000000","message":"Patch Set 5: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"fac82edcb5b684716db1365cb7aceb8f878ae309","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-13 16:49:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-pep8 http://logs.openstack.org/10/279910/5/check/gate-kolla-pep8/d8d8bb1/ : SUCCESS in 1m 41s\n- gate-kolla-docs http://docs-draft.openstack.org/10/279910/5/check/gate-kolla-docs/68d8aa2//doc/build/html/ : SUCCESS in 3m 14s\n- gate-kolla-python27 http://logs.openstack.org/10/279910/5/check/gate-kolla-python27/b0de7c9/ : SUCCESS in 3m 34s\n- gate-kolla-python34 http://logs.openstack.org/10/279910/5/check/gate-kolla-python34/a673d0b/ : SUCCESS in 2m 05s\n- gate-kolla-dsvm-build-centos-binary http://logs.openstack.org/10/279910/5/check/gate-kolla-dsvm-build-centos-binary/83ba2ea/ : FAILURE in 13m 40s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary http://logs.openstack.org/10/279910/5/check/gate-kolla-dsvm-deploy-centos-binary/55bd294/ : FAILURE in 14m 30s (non-voting)\n- gate-kolla-dsvm-build-centos-source http://logs.openstack.org/10/279910/5/check/gate-kolla-dsvm-build-centos-source/614d8c8/ : FAILURE in 15m 51s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source http://logs.openstack.org/10/279910/5/check/gate-kolla-dsvm-deploy-centos-source/7c36f33/ : FAILURE in 14m 37s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source http://logs.openstack.org/10/279910/5/check/gate-kolla-dsvm-build-ubuntu-source/13eb8c1/ : FAILURE in 20m 29s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source http://logs.openstack.org/10/279910/5/check/gate-kolla-dsvm-deploy-ubuntu-source/4ca0675/ : FAILURE in 10m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fe5a215bb41d1489470614c6cefe5d03d0b55cf3","author":{"_account_id":13642,"name":"Vikram Hosakote","email":"vikram.ganesh@gmail.com","username":"vhosakot"},"date":"2016-03-13 16:57:58.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"18b016f1294116b6a9f5e823bf7c72b6f01843f1","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2016-03-13 22:50:48.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6f7ec73e27068d1e19c99aeb319018d888c15a85","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2016-03-13 22:52:37.000000000","message":"Patch Set 6:\n\nthis patchset i built locally and it worked. lets see what teh gate says!","accounts_in_message":[],"_revision_number":6},{"id":"728d00d628d3ee7296e58166311ab827088925bd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-13 23:10:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-pep8 http://logs.openstack.org/10/279910/6/check/gate-kolla-pep8/03aa153/ : SUCCESS in 1m 36s\n- gate-kolla-docs http://docs-draft.openstack.org/10/279910/6/check/gate-kolla-docs/265b143//doc/build/html/ : SUCCESS in 1m 38s\n- gate-kolla-python27 http://logs.openstack.org/10/279910/6/check/gate-kolla-python27/3428d86/ : SUCCESS in 3m 38s\n- gate-kolla-python34 http://logs.openstack.org/10/279910/6/check/gate-kolla-python34/b8eecf1/ : SUCCESS in 2m 00s\n- gate-kolla-dsvm-build-centos-binary http://logs.openstack.org/10/279910/6/check/gate-kolla-dsvm-build-centos-binary/0951748/ : FAILURE in 15m 00s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary http://logs.openstack.org/10/279910/6/check/gate-kolla-dsvm-deploy-centos-binary/a6631f8/ : FAILURE in 15m 03s (non-voting)\n- gate-kolla-dsvm-build-centos-source http://logs.openstack.org/10/279910/6/check/gate-kolla-dsvm-build-centos-source/10e84dc/ : FAILURE in 15m 13s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source http://logs.openstack.org/10/279910/6/check/gate-kolla-dsvm-deploy-centos-source/1c21f96/ : FAILURE in 9m 56s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source http://logs.openstack.org/10/279910/6/check/gate-kolla-dsvm-build-ubuntu-source/0a1d377/ : FAILURE in 19m 47s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source http://logs.openstack.org/10/279910/6/check/gate-kolla-dsvm-deploy-ubuntu-source/56bca92/ : FAILURE in 6m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"53a4044bc868ff839a29dd83945d5c5ce4c622f9","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2016-03-13 23:22:40.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"cd0e1936db6b840ea317e184f9af454e6be7a3c5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-13 23:43:36.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-pep8 http://logs.openstack.org/10/279910/7/check/gate-kolla-pep8/5b641f1/ : SUCCESS in 3m 10s\n- gate-kolla-docs http://docs-draft.openstack.org/10/279910/7/check/gate-kolla-docs/9240688//doc/build/html/ : SUCCESS in 2m 57s\n- gate-kolla-python27 http://logs.openstack.org/10/279910/7/check/gate-kolla-python27/79c9056/ : SUCCESS in 2m 59s\n- gate-kolla-python34 http://logs.openstack.org/10/279910/7/check/gate-kolla-python34/efbc8d8/ : SUCCESS in 4m 01s\n- gate-kolla-dsvm-build-centos-binary http://logs.openstack.org/10/279910/7/check/gate-kolla-dsvm-build-centos-binary/8aba811/ : FAILURE in 13m 07s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary http://logs.openstack.org/10/279910/7/check/gate-kolla-dsvm-deploy-centos-binary/f311ced/ : FAILURE in 15m 51s (non-voting)\n- gate-kolla-dsvm-build-centos-source http://logs.openstack.org/10/279910/7/check/gate-kolla-dsvm-build-centos-source/eb86e94/ : FAILURE in 15m 47s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source http://logs.openstack.org/10/279910/7/check/gate-kolla-dsvm-deploy-centos-source/bc2ca57/ : FAILURE in 15m 27s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source http://logs.openstack.org/10/279910/7/check/gate-kolla-dsvm-build-ubuntu-source/f9dcb52/ : FAILURE in 20m 47s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source http://logs.openstack.org/10/279910/7/check/gate-kolla-dsvm-deploy-ubuntu-source/aecfa62/ : FAILURE in 10m 50s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"66b840c892f19eebdd2118ae95d8c7117fa90f25","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2016-03-13 23:48:24.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8f83851c8f2f96b131bdff3ebefe151b91a46da4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-14 00:05:23.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-pep8 http://logs.openstack.org/10/279910/8/check/gate-kolla-pep8/379362b/ : SUCCESS in 1m 40s\n- gate-kolla-docs http://docs-draft.openstack.org/10/279910/8/check/gate-kolla-docs/363204e//doc/build/html/ : SUCCESS in 3m 17s\n- gate-kolla-python27 http://logs.openstack.org/10/279910/8/check/gate-kolla-python27/1a6a51e/ : SUCCESS in 3m 01s\n- gate-kolla-python34 http://logs.openstack.org/10/279910/8/check/gate-kolla-python34/4e3e686/ : SUCCESS in 3m 53s\n- gate-kolla-dsvm-build-centos-binary http://logs.openstack.org/10/279910/8/check/gate-kolla-dsvm-build-centos-binary/c8d4be0/ : FAILURE in 14m 50s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary http://logs.openstack.org/10/279910/8/check/gate-kolla-dsvm-deploy-centos-binary/0aff185/ : FAILURE in 14m 53s (non-voting)\n- gate-kolla-dsvm-build-centos-source http://logs.openstack.org/10/279910/8/check/gate-kolla-dsvm-build-centos-source/003aaf8/ : FAILURE in 12m 46s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source http://logs.openstack.org/10/279910/8/check/gate-kolla-dsvm-deploy-centos-source/44ef5d8/ : FAILURE in 9m 24s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source http://logs.openstack.org/10/279910/8/check/gate-kolla-dsvm-build-ubuntu-source/1481c31/ : SUCCESS in 16m 50s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source http://logs.openstack.org/10/279910/8/check/gate-kolla-dsvm-deploy-ubuntu-source/f74ea2a/ : FAILURE in 11m 58s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8a58b8f79e825a202f96aa3c29793f2b40f74fd8","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2016-03-14 00:18:03.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d20cc86f6362a08bfbb3a99bd45d94559fcca679","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2016-03-14 00:40:25.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"7474d1c602b895b98ecc9513117f2fde3cee7c21","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-14 00:57:39.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-pep8 http://logs.openstack.org/10/279910/10/check/gate-kolla-pep8/6cc0f38/ : SUCCESS in 4m 08s\n- gate-kolla-docs http://docs-draft.openstack.org/10/279910/10/check/gate-kolla-docs/91b4897//doc/build/html/ : SUCCESS in 1m 38s\n- gate-kolla-python27 http://logs.openstack.org/10/279910/10/check/gate-kolla-python27/faa2851/ : SUCCESS in 3m 52s\n- gate-kolla-python34 http://logs.openstack.org/10/279910/10/check/gate-kolla-python34/053fc50/ : SUCCESS in 3m 12s\n- gate-kolla-dsvm-build-centos-binary http://logs.openstack.org/10/279910/10/check/gate-kolla-dsvm-build-centos-binary/264a7df/ : FAILURE in 10m 18s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary http://logs.openstack.org/10/279910/10/check/gate-kolla-dsvm-deploy-centos-binary/33dc978/ : FAILURE in 10m 14s (non-voting)\n- gate-kolla-dsvm-build-centos-source http://logs.openstack.org/10/279910/10/check/gate-kolla-dsvm-build-centos-source/a710a75/ : FAILURE in 9m 17s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source http://logs.openstack.org/10/279910/10/check/gate-kolla-dsvm-deploy-centos-source/5495c3a/ : FAILURE in 15m 25s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source http://logs.openstack.org/10/279910/10/check/gate-kolla-dsvm-build-ubuntu-source/400bd25/ : SUCCESS in 14m 08s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source http://logs.openstack.org/10/279910/10/check/gate-kolla-dsvm-deploy-ubuntu-source/66449cd/ : SUCCESS in 15m 18s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"712e71acd7d45b4c450f0640038ed14efe3cd610","author":{"_account_id":13642,"name":"Vikram Hosakote","email":"vikram.ganesh@gmail.com","username":"vhosakot"},"date":"2016-03-14 01:01:50.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"f59489ad15ea8c5ba12f46bf42cc155f613da461","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2016-03-14 02:14:28.000000000","message":"Patch Set 10: Code-Review-2\n\nneed to wait for centos rdo stuff to work since this touches centos, but please review","accounts_in_message":[],"_revision_number":10},{"id":"81aeb38ef13c098d11d5ed47b7f57ee5bca6002a","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2016-03-14 02:15:55.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"ad73bdb4a5ed46424aa42b3e9cdeda9309508754","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-14 02:35:05.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-pep8 http://logs.openstack.org/10/279910/11/check/gate-kolla-pep8/67353d1/ : SUCCESS in 1m 43s\n- gate-kolla-docs http://docs-draft.openstack.org/10/279910/11/check/gate-kolla-docs/d920a33//doc/build/html/ : SUCCESS in 3m 02s\n- gate-kolla-python27 http://logs.openstack.org/10/279910/11/check/gate-kolla-python27/895447c/ : SUCCESS in 3m 21s\n- gate-kolla-python34 http://logs.openstack.org/10/279910/11/check/gate-kolla-python34/69d80d3/ : SUCCESS in 4m 00s\n- gate-kolla-dsvm-build-centos-binary http://logs.openstack.org/10/279910/11/check/gate-kolla-dsvm-build-centos-binary/f10a092/ : FAILURE in 9m 43s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary http://logs.openstack.org/10/279910/11/check/gate-kolla-dsvm-deploy-centos-binary/287f532/ : FAILURE in 17m 50s (non-voting)\n- gate-kolla-dsvm-build-centos-source http://logs.openstack.org/10/279910/11/check/gate-kolla-dsvm-build-centos-source/bf862e3/ : FAILURE in 13m 58s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source http://logs.openstack.org/10/279910/11/check/gate-kolla-dsvm-deploy-centos-source/1ef9d41/ : FAILURE in 10m 15s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source http://logs.openstack.org/10/279910/11/check/gate-kolla-dsvm-build-ubuntu-source/afbac3b/ : SUCCESS in 18m 59s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source http://logs.openstack.org/10/279910/11/check/gate-kolla-dsvm-deploy-ubuntu-source/43e397a/ : SUCCESS in 15m 49s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"598644e73c69afe446b5bfda4021a1138d328df2","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2016-03-14 02:52:38.000000000","message":"Patch Set 11: Code-Review+1\n\nLGTM, let\u0027s wait for delorean repos to be up again.","accounts_in_message":[],"_revision_number":11},{"id":"0fc2c01c19cf6be922ccc700c80c9e9ee36b6efd","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2016-03-14 02:55:37.000000000","message":"Patch Set 11:\n\nBTW, Sam you\u0027re both author and co-author on this patch. I don\u0027t know how much of Jeffrey\u0027s original work is still present in the patch, but there is something to fix here.","accounts_in_message":[],"_revision_number":11},{"id":"2c1e5cd62c331d19ee8bba4a8b1ad893df50df7b","author":{"_account_id":13642,"name":"Vikram Hosakote","email":"vikram.ganesh@gmail.com","username":"vhosakot"},"date":"2016-03-14 03:00:11.000000000","message":"Patch Set 11: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":11},{"id":"0b985fdd482969ec2c14847c9e962559a788467f","author":{"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},"date":"2016-03-14 03:05:39.000000000","message":"Patch Set 11: Code-Review+1\n\nAgree with Martin, I think this was initially proposed by Jeffrey so he\u0027s the co-author. So you might want to check the co-authors. LGTM otherwise. Ubuntu gates green.","accounts_in_message":[],"_revision_number":11},{"id":"e1611579c8fb6e6f9f36a658909a27628d156f1e","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2016-03-14 03:30:31.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"2946c4654770c71d9d79863a986827698dd18465","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-14 03:49:27.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-pep8 http://logs.openstack.org/10/279910/12/check/gate-kolla-pep8/826a447/ : SUCCESS in 3m 06s\n- gate-kolla-docs http://docs-draft.openstack.org/10/279910/12/check/gate-kolla-docs/6ba3672//doc/build/html/ : SUCCESS in 2m 54s\n- gate-kolla-python27 http://logs.openstack.org/10/279910/12/check/gate-kolla-python27/188c23a/ : SUCCESS in 3m 16s\n- gate-kolla-python34 http://logs.openstack.org/10/279910/12/check/gate-kolla-python34/8d2bc8a/ : SUCCESS in 3m 28s\n- gate-kolla-dsvm-build-centos-binary http://logs.openstack.org/10/279910/12/check/gate-kolla-dsvm-build-centos-binary/b3a5730/ : FAILURE in 9m 30s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary http://logs.openstack.org/10/279910/12/check/gate-kolla-dsvm-deploy-centos-binary/857f52d/ : FAILURE in 14m 14s (non-voting)\n- gate-kolla-dsvm-build-centos-source http://logs.openstack.org/10/279910/12/check/gate-kolla-dsvm-build-centos-source/666d96a/ : FAILURE in 15m 08s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source http://logs.openstack.org/10/279910/12/check/gate-kolla-dsvm-deploy-centos-source/3b75f8e/ : FAILURE in 9m 08s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source http://logs.openstack.org/10/279910/12/check/gate-kolla-dsvm-build-ubuntu-source/ce441cc/ : SUCCESS in 18m 48s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source http://logs.openstack.org/10/279910/12/check/gate-kolla-dsvm-deploy-ubuntu-source/3d69e34/ : SUCCESS in 15m 59s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d58febec7f3f4a2888650ec09264c52cbaca74c4","author":{"_account_id":13642,"name":"Vikram Hosakote","email":"vikram.ganesh@gmail.com","username":"vhosakot"},"date":"2016-03-14 04:30:46.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"50d7b94edbc37392a1a366b0f1bb1994e2cd07de","author":{"_account_id":15447,"name":"yongfeng","email":"dolpher.du@intel.com","username":"dolpher"},"date":"2016-03-14 05:53:44.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"d5a28f614a4dcb6b052f0da959d48e2f801abd6f","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2016-03-14 08:29:43.000000000","message":"Patch Set 12:\n\n(1 comment)\n\nSorry I am busy in few past days.\n\nthanks Sam to update this.\nOnly one question about why the stdio_handler change.\n\nbtw, martin and coolsvap are right. would you mind add me to the Co-Authored-By? :p","accounts_in_message":[],"_revision_number":12},{"id":"4d6130098e7b911b40283882942e0c28e81faf48","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2016-03-14 08:37:44.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"0a442b4bd8959e360d52fd92a4c83c1b37fac1dd","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2016-03-14 09:10:41.000000000","message":"Patch Set 12: Code-Review+1\n\ncool, roger that.\n\nthanks for the info Martin.","accounts_in_message":[],"_revision_number":12},{"id":"d38e4d5bc001af2c451290889b3c80dca760b0e9","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2016-03-14 14:59:48.000000000","message":"Patch Set 12: -Code-Review\n\nrecheck\n\njeffrey you are patch owner, not co-author. you can\u0027t change the patch owner ever","accounts_in_message":[],"_revision_number":12},{"id":"62d3bdaf45e2f9f0a81f5fc69638142b01fe6e39","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2016-03-14 15:22:35.000000000","message":"Patch Set 12:\n\n@Sam, I would argue at that point.\n\nthe Co-Author-By is relative to the commit author not patch set owner. And also, when the stackalytics count the commits sb make, it will use the *commit name* and *Co-Author-By:* syntax in the commit message, rather then the patch set owner.[0]\n\nin this commit, you are the commiter and author. You can see by using `git log --pretty\u003d\u0027commit_id:%H%nauthor_name:%an%ncommit_name::%cn%n\u0027`\n\nnormally, when changing other guy\u0027s commit, only `git commit --amend` is enough. In this way, the commiter will be changed rather than the author.\n\nI suppose that, you created a new commit and add the Change-Id to it. So the author and commiter are both you. If so, the Co-Author-By should be me not you.\n\n\n[0] https://github.com/openstack/stackalytics/blob/master/stackalytics/processor/vcs.py#L214","accounts_in_message":[],"_revision_number":12},{"id":"928d1ab1fc54f55764f822cd2eb4ead0c0d527fd","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2016-03-14 15:23:37.000000000","message":"Patch Set 12: Workflow-1\n\nohhhh i see what you mean. yes i know why that is, i never checked out your patch, i just used your change id. will fix. first want to see if this passes the gate","accounts_in_message":[],"_revision_number":12},{"id":"7d4dc02552022a115cdccc6eadb4a99a5a6e1d1a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-14 15:42:47.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-pep8 http://logs.openstack.org/10/279910/12/check/gate-kolla-pep8/43bf3b4/ : SUCCESS in 3m 02s\n- gate-kolla-docs http://docs-draft.openstack.org/10/279910/12/check/gate-kolla-docs/85d4f72//doc/build/html/ : SUCCESS in 3m 03s\n- gate-kolla-python27 http://logs.openstack.org/10/279910/12/check/gate-kolla-python27/910201e/ : SUCCESS in 1m 45s\n- gate-kolla-python34 http://logs.openstack.org/10/279910/12/check/gate-kolla-python34/f6e5064/ : SUCCESS in 4m 05s\n- gate-kolla-dsvm-build-centos-binary http://logs.openstack.org/10/279910/12/check/gate-kolla-dsvm-build-centos-binary/81fec3a/ : SUCCESS in 21m 30s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary http://logs.openstack.org/10/279910/12/check/gate-kolla-dsvm-deploy-centos-binary/a386b92/ : SUCCESS in 24m 36s (non-voting)\n- gate-kolla-dsvm-build-centos-source http://logs.openstack.org/10/279910/12/check/gate-kolla-dsvm-build-centos-source/a389978/ : SUCCESS in 41m 26s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source http://logs.openstack.org/10/279910/12/check/gate-kolla-dsvm-deploy-centos-source/f3183d0/ : SUCCESS in 29m 22s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source http://logs.openstack.org/10/279910/12/check/gate-kolla-dsvm-build-ubuntu-source/af78927/ : SUCCESS in 20m 29s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source http://logs.openstack.org/10/279910/12/check/gate-kolla-dsvm-deploy-ubuntu-source/195388e/ : SUCCESS in 16m 55s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"5eb04494b54191349397eec5ab1b4fe281af88de","author":{"_account_id":13642,"name":"Vikram Hosakote","email":"vikram.ganesh@gmail.com","username":"vhosakot"},"date":"2016-03-14 15:45:12.000000000","message":"Patch Set 12:\n\nGate all green!","accounts_in_message":[],"_revision_number":12},{"id":"e22b46fafafe372bbc9095b28ec5e6bb41a6f6f1","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2016-03-14 15:52:53.000000000","message":"Patch Set 13: Published edit on patch set 12","accounts_in_message":[],"_revision_number":13},{"id":"7a9b30acb0a3676da50e474d9456b3af60ea7589","author":{"_account_id":13642,"name":"Vikram Hosakote","email":"vikram.ganesh@gmail.com","username":"vhosakot"},"date":"2016-03-14 15:53:36.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"fd3c2f59bc40cc5ce7a74039677d968aeef52584","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2016-03-14 16:01:29.000000000","message":"Patch Set 13: Code-Review+1\n\nthanks Sam. :)","accounts_in_message":[],"_revision_number":13},{"id":"65531043cd9442b13214ab22774a1052446569a6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-14 16:36:50.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-pep8 http://logs.openstack.org/10/279910/13/check/gate-kolla-pep8/d41d484/ : SUCCESS in 3m 23s\n- gate-kolla-docs http://docs-draft.openstack.org/10/279910/13/check/gate-kolla-docs/2edac86//doc/build/html/ : SUCCESS in 2m 27s\n- gate-kolla-python27 http://logs.openstack.org/10/279910/13/check/gate-kolla-python27/9d18d63/ : SUCCESS in 3m 48s\n- gate-kolla-python34 http://logs.openstack.org/10/279910/13/check/gate-kolla-python34/df7072f/ : SUCCESS in 4m 39s\n- gate-kolla-dsvm-build-centos-binary http://logs.openstack.org/10/279910/13/check/gate-kolla-dsvm-build-centos-binary/95db9f6/ : SUCCESS in 38m 22s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary http://logs.openstack.org/10/279910/13/check/gate-kolla-dsvm-deploy-centos-binary/364f91e/ : SUCCESS in 16m 03s (non-voting)\n- gate-kolla-dsvm-build-centos-source http://logs.openstack.org/10/279910/13/check/gate-kolla-dsvm-build-centos-source/c6f6eef/ : SUCCESS in 26m 11s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source http://logs.openstack.org/10/279910/13/check/gate-kolla-dsvm-deploy-centos-source/ecc32dc/ : SUCCESS in 39m 54s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source http://logs.openstack.org/10/279910/13/check/gate-kolla-dsvm-build-ubuntu-source/a691de0/ : SUCCESS in 18m 40s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source http://logs.openstack.org/10/279910/13/check/gate-kolla-dsvm-deploy-ubuntu-source/b206356/ : SUCCESS in 16m 05s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"fb8446bac98b575cae1050f0d96d0c4673b0470c","author":{"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},"date":"2016-03-14 16:45:10.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"eb067e9b54603cfc05e849c13e46f1e7e68741ef","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2016-03-15 02:30:00.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1\n\nLGTM, and the gate confirms it works for centos.","accounts_in_message":[],"_revision_number":13},{"id":"d191c15426ed01185a2ae7e5317dbc3be4ffbab7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-15 02:30:06.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":13},{"id":"275fef74be959591d5cde46c4cadd33bb384b1ac","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-15 02:35:46.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-kolla-docs http://docs-draft.openstack.org/10/279910/13/gate/gate-kolla-docs/245df9d//doc/build/html/ : SUCCESS in 2m 27s\n- gate-kolla-pep8 http://logs.openstack.org/10/279910/13/gate/gate-kolla-pep8/a3b066f/ : SUCCESS in 1m 39s\n- gate-kolla-python27 http://logs.openstack.org/10/279910/13/gate/gate-kolla-python27/aea01ca/ : SUCCESS in 3m 01s\n- gate-kolla-python34 http://logs.openstack.org/10/279910/13/gate/gate-kolla-python34/3aa91bd/ : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":13},{"id":"fac918bde28ad6bcde6109be05acca83cafe7716","date":"2016-03-15 02:35:47.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"8b3f0d085712154d580e4a464ca7e88e612f2fb2","revisions":{"af79d76508b2eab6574b25823b2b1031b1ee3e58":{"kind":"REWORK","_number":1,"created":"2016-02-14 02:35:06.000000000","uploader":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"ref":"refs/changes/10/279910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/279910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/279910/1"}}},"commit":{"parents":[{"commit":"e92b18a8a8c0d7dd3d05ac2b48e469b5b4773215","subject":"Merge \"Do not pull rabbitmq-data image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e92b18a8a8c0d7dd3d05ac2b48e469b5b4773215"}]}],"author":{"name":"Jeffrey Zhang","email":"jeffrey.zhang@99cloud.net","date":"2016-02-14 02:33:23.000000000","tz":480},"committer":{"name":"Jeffrey Zhang","email":"jeffrey.zhang@99cloud.net","date":"2016-02-14 02:34:53.000000000","tz":480},"subject":"[WIP] Use ubuntu Mitaka staging repo","message":"[WIP] Use ubuntu Mitaka staging repo\n\nSeem that the ubuntu Mitaka staging is ok now. And replace the liberty\nrepo.\n\nTrivialFix\n\nChange-Id: I84217824817c484b6800a64cbd7767b127a3098e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af79d76508b2eab6574b25823b2b1031b1ee3e58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af79d76508b2eab6574b25823b2b1031b1ee3e58"}]},"branch":"refs/heads/master"},"b984b98ff4506fa7511749d6c188e1f0f8982fb7":{"kind":"REWORK","_number":2,"created":"2016-02-14 04:48:57.000000000","uploader":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"ref":"refs/changes/10/279910/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/279910/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/279910/2"}}},"commit":{"parents":[{"commit":"e92b18a8a8c0d7dd3d05ac2b48e469b5b4773215","subject":"Merge \"Do not pull rabbitmq-data image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e92b18a8a8c0d7dd3d05ac2b48e469b5b4773215"}]}],"author":{"name":"Jeffrey Zhang","email":"jeffrey.zhang@99cloud.net","date":"2016-02-14 02:33:23.000000000","tz":480},"committer":{"name":"Jeffrey Zhang","email":"jeffrey.zhang@99cloud.net","date":"2016-02-14 04:47:48.000000000","tz":480},"subject":"[WIP] Use ubuntu Mitaka staging repo","message":"[WIP] Use ubuntu Mitaka staging repo\n\nSeem that the ubuntu Mitaka staging is ok now. And replace the liberty\nrepo.\n\n* ceph package creates the ceph user\n* rabbitmq-server upgrade to 3.5.7\n\nTrivialFix\n\nChange-Id: I84217824817c484b6800a64cbd7767b127a3098e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b984b98ff4506fa7511749d6c188e1f0f8982fb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b984b98ff4506fa7511749d6c188e1f0f8982fb7"}]},"branch":"refs/heads/master"},"9fa665657a012861193455b97cf7a33f06aad86b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2016-02-14 18:18:48.000000000","uploader":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"ref":"refs/changes/10/279910/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/279910/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/279910/3"}}},"commit":{"parents":[{"commit":"8ef7434770846e81aa04009838bd1df69cc29ac8","subject":"Fix keystone initial auth mechanism","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ef7434770846e81aa04009838bd1df69cc29ac8"}]}],"author":{"name":"Jeffrey Zhang","email":"jeffrey.zhang@99cloud.net","date":"2016-02-14 02:33:23.000000000","tz":480},"committer":{"name":"Jeffrey Zhang","email":"jeffrey.zhang@99cloud.net","date":"2016-02-14 18:18:48.000000000","tz":0},"subject":"[WIP] Use ubuntu Mitaka staging repo","message":"[WIP] Use ubuntu Mitaka staging repo\n\nSeem that the ubuntu Mitaka staging is ok now. And replace the liberty\nrepo.\n\n* ceph package creates the ceph user\n* rabbitmq-server upgrade to 3.5.7\n\nTrivialFix\n\nChange-Id: I84217824817c484b6800a64cbd7767b127a3098e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9fa665657a012861193455b97cf7a33f06aad86b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9fa665657a012861193455b97cf7a33f06aad86b"}]},"branch":"refs/heads/master"},"8d3e6b201b8a4d834a60f0826fffce838d53f482":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2016-03-09 06:51:42.000000000","uploader":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"ref":"refs/changes/10/279910/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/279910/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/279910/4"}}},"commit":{"parents":[{"commit":"13a60622601a3fc8535c0ee8c10a1da2d0ea928b","subject":"Add documentation about critical data loss","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13a60622601a3fc8535c0ee8c10a1da2d0ea928b"}]}],"author":{"name":"Jeffrey Zhang","email":"jeffrey.zhang@99cloud.net","date":"2016-02-14 02:33:23.000000000","tz":480},"committer":{"name":"Jeffrey Zhang","email":"jeffrey.zhang@99cloud.net","date":"2016-03-09 06:51:41.000000000","tz":0},"subject":"[WIP] Use ubuntu Mitaka staging repo","message":"[WIP] Use ubuntu Mitaka staging repo\n\nSeem that the ubuntu Mitaka staging is ok now. And replace the liberty\nrepo.\n\n* ceph package creates the ceph user\n* rabbitmq-server upgrade to 3.5.7\n\nTrivialFix\n\nChange-Id: I84217824817c484b6800a64cbd7767b127a3098e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d3e6b201b8a4d834a60f0826fffce838d53f482"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d3e6b201b8a4d834a60f0826fffce838d53f482"}]},"branch":"refs/heads/master"},"475fb26b18f06a42bc5a8386622957df272b82e9":{"kind":"REWORK","_number":5,"created":"2016-03-13 16:28:24.000000000","uploader":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"ref":"refs/changes/10/279910/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/279910/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/279910/5"}}},"commit":{"parents":[{"commit":"8165fdf0bcfe80b2066d6b9288710eaeef275bad","subject":"Merge \"Remove openstack client from keystone bootstrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8165fdf0bcfe80b2066d6b9288710eaeef275bad"}]}],"author":{"name":"Jeffrey Zhang","email":"jeffrey.zhang@99cloud.net","date":"2016-02-14 02:33:23.000000000","tz":480},"committer":{"name":"Jeffrey Zhang","email":"jeffrey.zhang@99cloud.net","date":"2016-03-13 16:28:02.000000000","tz":480},"subject":"[WIP] Use ubuntu Mitaka repo","message":"[WIP] Use ubuntu Mitaka repo\n\nSeem that the ubuntu Mitaka staging is ok now. And replace the liberty\nrepo.\n\n* ceph package creates the ceph user\n* rabbitmq-server upgrade to 3.5.7\n\nTrivialFix\n\nChange-Id: I84217824817c484b6800a64cbd7767b127a3098e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/475fb26b18f06a42bc5a8386622957df272b82e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/475fb26b18f06a42bc5a8386622957df272b82e9"}]},"branch":"refs/heads/master"},"e22e9bbb8129e67ac3fe20b7f6f48e60b664489d":{"kind":"REWORK","_number":6,"created":"2016-03-13 22:50:48.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/10/279910/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/279910/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/279910/6"}}},"commit":{"parents":[{"commit":"e4829471651dd9677f2f73de4e2896902666fdd4","subject":"Merge \"Fix the nova reconfig not work issue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4829471651dd9677f2f73de4e2896902666fdd4"}]}],"author":{"name":"SamYaple","email":"sam@yaple.net","date":"2016-03-13 22:47:35.000000000","tz":0},"committer":{"name":"SamYaple","email":"sam@yaple.net","date":"2016-03-13 22:50:17.000000000","tz":0},"subject":"Use ubuntu Mitaka repo","message":"Use ubuntu Mitaka repo\n\nThe rabbitmq-server package is upgraded to 3.5.7 in cloud-archive\nso we update centos to match\n\nCo-Authored-By: Sam Yaple \u003csam@yaple.net\u003e\nTrivialFix\nChange-Id: I84217824817c484b6800a64cbd7767b127a3098e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e22e9bbb8129e67ac3fe20b7f6f48e60b664489d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e22e9bbb8129e67ac3fe20b7f6f48e60b664489d"}]},"branch":"refs/heads/master"},"72e8f9e5077a13504b5e465f9ca66e919069d0f2":{"kind":"REWORK","_number":7,"created":"2016-03-13 23:22:40.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/10/279910/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/279910/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/279910/7"}}},"commit":{"parents":[{"commit":"0bb766c12986128db9bc78251fed1d8283c44aa9","subject":"Properly pin ceph","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0bb766c12986128db9bc78251fed1d8283c44aa9"}]}],"author":{"name":"SamYaple","email":"sam@yaple.net","date":"2016-03-13 22:47:35.000000000","tz":0},"committer":{"name":"SamYaple","email":"sam@yaple.net","date":"2016-03-13 23:22:34.000000000","tz":0},"subject":"Use ubuntu Mitaka repo","message":"Use ubuntu Mitaka repo\n\nThe rabbitmq-server package is upgraded to 3.5.7 in cloud-archive\nso we update centos to match\n\nCo-Authored-By: Sam Yaple \u003csam@yaple.net\u003e\nTrivialFix\nChange-Id: I84217824817c484b6800a64cbd7767b127a3098e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e8f9e5077a13504b5e465f9ca66e919069d0f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e8f9e5077a13504b5e465f9ca66e919069d0f2"}]},"branch":"refs/heads/master"},"cced47f3eab2019e48d1fb80155c4d67b0bef7ae":{"kind":"REWORK","_number":8,"created":"2016-03-13 23:48:24.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/10/279910/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/279910/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/279910/8"}}},"commit":{"parents":[{"commit":"0bb766c12986128db9bc78251fed1d8283c44aa9","subject":"Properly pin ceph","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0bb766c12986128db9bc78251fed1d8283c44aa9"}]}],"author":{"name":"SamYaple","email":"sam@yaple.net","date":"2016-03-13 22:47:35.000000000","tz":0},"committer":{"name":"SamYaple","email":"sam@yaple.net","date":"2016-03-13 23:48:14.000000000","tz":0},"subject":"Use ubuntu Mitaka repo","message":"Use ubuntu Mitaka repo\n\nThe rabbitmq-server package is upgraded to 3.5.7 in cloud-archive\nso we update centos to match\n\nThe xen-utils package now needs an explict version, xen-utils-4.6 is\nwhat is provided by cloud-archive mitaka\n\nCo-Authored-By: Sam Yaple \u003csam@yaple.net\u003e\nTrivialFix\nChange-Id: I84217824817c484b6800a64cbd7767b127a3098e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cced47f3eab2019e48d1fb80155c4d67b0bef7ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cced47f3eab2019e48d1fb80155c4d67b0bef7ae"}]},"branch":"refs/heads/master"},"e893cd62ba595d463e01e2c49e711145992c2e62":{"kind":"REWORK","_number":9,"created":"2016-03-14 00:18:03.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/10/279910/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/279910/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/279910/9"}}},"commit":{"parents":[{"commit":"592a2753f79582037ee68e4f526420abd20da12a","subject":"haproxy formatting was incorrect","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/592a2753f79582037ee68e4f526420abd20da12a"}]}],"author":{"name":"SamYaple","email":"sam@yaple.net","date":"2016-03-13 22:47:35.000000000","tz":0},"committer":{"name":"SamYaple","email":"sam@yaple.net","date":"2016-03-14 00:17:55.000000000","tz":0},"subject":"Use ubuntu Mitaka repo","message":"Use ubuntu Mitaka repo\n\nThe rabbitmq-server package is upgraded to 3.5.7 in cloud-archive\nso we update centos to match\n\nThe xen-utils package now needs an explict version, xen-utils-4.6 is\nwhat is provided by cloud-archive mitaka\n\nCo-Authored-By: Sam Yaple \u003csam@yaple.net\u003e\nTrivialFix\nChange-Id: I84217824817c484b6800a64cbd7767b127a3098e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e893cd62ba595d463e01e2c49e711145992c2e62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e893cd62ba595d463e01e2c49e711145992c2e62"}]},"branch":"refs/heads/master"},"f0eccacdb43896478af7878ce6a0f9ac81488f0e":{"kind":"REWORK","_number":10,"created":"2016-03-14 00:40:25.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/10/279910/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/279910/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/279910/10"}}},"commit":{"parents":[{"commit":"9500bd239ac122ae41de6086112e5e5335ed1558","subject":"haproxy formatting was incorrect","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9500bd239ac122ae41de6086112e5e5335ed1558"}]}],"author":{"name":"SamYaple","email":"sam@yaple.net","date":"2016-03-13 22:47:35.000000000","tz":0},"committer":{"name":"SamYaple","email":"sam@yaple.net","date":"2016-03-14 00:40:18.000000000","tz":0},"subject":"Use ubuntu Mitaka repo","message":"Use ubuntu Mitaka repo\n\nThe rabbitmq-server package is upgraded to 3.5.7 in cloud-archive\nso we update centos to match\n\nThe xen-utils package now needs an explict version, xen-utils-4.6 is\nwhat is provided by cloud-archive mitaka\n\nCo-Authored-By: Sam Yaple \u003csam@yaple.net\u003e\nTrivialFix\nChange-Id: I84217824817c484b6800a64cbd7767b127a3098e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0eccacdb43896478af7878ce6a0f9ac81488f0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0eccacdb43896478af7878ce6a0f9ac81488f0e"}]},"branch":"refs/heads/master"},"8d944c11e6d076f6d21745fa0ac3dfac12dad015":{"kind":"REWORK","_number":11,"created":"2016-03-14 02:15:55.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/10/279910/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/279910/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/279910/11"}}},"commit":{"parents":[{"commit":"9500bd239ac122ae41de6086112e5e5335ed1558","subject":"haproxy formatting was incorrect","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9500bd239ac122ae41de6086112e5e5335ed1558"}]}],"author":{"name":"SamYaple","email":"sam@yaple.net","date":"2016-03-13 22:47:35.000000000","tz":0},"committer":{"name":"SamYaple","email":"sam@yaple.net","date":"2016-03-14 02:15:49.000000000","tz":0},"subject":"Use ubuntu Mitaka repo","message":"Use ubuntu Mitaka repo\n\nThe rabbitmq-server package is upgraded to 3.5.7 in cloud-archive\nso we update centos to match\n\nThe xen-utils package now needs an explict version, xen-utils-4.6 is\nwhat is provided by cloud-archive mitaka\n\nCo-Authored-By: Sam Yaple \u003csam@yaple.net\u003e\nTrivialFix\nChange-Id: I84217824817c484b6800a64cbd7767b127a3098e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d944c11e6d076f6d21745fa0ac3dfac12dad015"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d944c11e6d076f6d21745fa0ac3dfac12dad015"}]},"branch":"refs/heads/master"},"f9e0f911825eccd932090dc13cf8e7f43c21c980":{"kind":"REWORK","_number":12,"created":"2016-03-14 03:30:31.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/10/279910/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/279910/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/279910/12"}}},"commit":{"parents":[{"commit":"9500bd239ac122ae41de6086112e5e5335ed1558","subject":"haproxy formatting was incorrect","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9500bd239ac122ae41de6086112e5e5335ed1558"}]}],"author":{"name":"SamYaple","email":"sam@yaple.net","date":"2016-03-13 22:47:35.000000000","tz":0},"committer":{"name":"SamYaple","email":"sam@yaple.net","date":"2016-03-14 03:30:25.000000000","tz":0},"subject":"Use ubuntu Mitaka repo","message":"Use ubuntu Mitaka repo\n\nThe rabbitmq-server package is upgraded to 3.5.7 in cloud-archive\nso we update centos to match\n\nThe xen-utils package now needs an explict version, xen-utils-4.6 is\nwhat is provided by cloud-archive mitaka\n\nCo-Authored-By: Sam Yaple \u003csam@yaple.net\u003e\nTrivialFix\nChange-Id: I84217824817c484b6800a64cbd7767b127a3098e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9e0f911825eccd932090dc13cf8e7f43c21c980"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9e0f911825eccd932090dc13cf8e7f43c21c980"}]},"branch":"refs/heads/master"},"8b3f0d085712154d580e4a464ca7e88e612f2fb2":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2016-03-14 15:52:53.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/10/279910/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/279910/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/279910/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/279910/13"}}},"commit":{"parents":[{"commit":"9500bd239ac122ae41de6086112e5e5335ed1558","subject":"haproxy formatting was incorrect","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9500bd239ac122ae41de6086112e5e5335ed1558"}]}],"author":{"name":"SamYaple","email":"sam@yaple.net","date":"2016-03-13 22:47:35.000000000","tz":0},"committer":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2016-03-14 15:52:48.000000000","tz":0},"subject":"Use ubuntu Mitaka repo","message":"Use ubuntu Mitaka repo\n\nThe rabbitmq-server package is upgraded to 3.5.7 in cloud-archive\nso we update centos to match\n\nThe xen-utils package now needs an explict version, xen-utils-4.6 is\nwhat is provided by cloud-archive mitaka\n\nLibvirt 1.3 is in the ubuntu cloud-archive. This has a new daemon for\nlogging that needs further implementation in newton. For now, it has\nbeen disabled within the qemu.conf\n\nCo-Authored-By: Jeffrey Zhang \u003cjeffrey.zhang@99cloud.net\u003e\nTrivialFix\nChange-Id: I84217824817c484b6800a64cbd7767b127a3098e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b3f0d085712154d580e4a464ca7e88e612f2fb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b3f0d085712154d580e4a464ca7e88e612f2fb2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
