)]}'
{"id":"openstack%2Fkolla~349006","triplet_id":"openstack%2Fkolla~master~I0a40867ae65011fbfd53308b9d7f42d10435f5fa","project":"openstack/kolla","branch":"master","topic":"bug/1607905","hashtags":[],"change_id":"I0a40867ae65011fbfd53308b9d7f42d10435f5fa","subject":"Enable CADF events in Keystone and introduce enable_cadf_notifications","status":"MERGED","created":"2016-07-29 18:23:47.000000000","updated":"2016-09-19 14:23:14.000000000","submitted":"2016-09-19 14:23:14.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7a6864a72eb8b5051cfb2263ff02b18d75781f88","_number":349006,"virtual_id_number":349006,"owner":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},{"value":0,"_account_id":2834,"name":"Steven Dake","email":"steven.dake@gmail.com","username":"sdake"},{"value":2,"date":"2016-09-19 14:23:14.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":979,"name":"Dave Walker","email":"email@daviey.com","username":"davewalker"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14103,"name":"wangtianfa","email":"wang.tianfa@99cloud.net","username":"wangtianfa"},{"value":0,"_account_id":19384,"name":"Serguei Bezverkhi","email":"sbezverk@cisco.com","username":"sbezverk"},{"value":0,"_account_id":18009,"name":"Md Nadeem","email":"mail2nadeem92@gmail.com","username":"mail2nadeem92"},{"value":0,"_account_id":21513,"name":"lixinhu","email":"lixinhu@inspur.com","username":"lixinhu"},{"value":0,"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"all":[{"value":2,"date":"2016-09-19 07:19:46.000000000","_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},{"value":0,"_account_id":2834,"name":"Steven Dake","email":"steven.dake@gmail.com","username":"sdake"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":979,"name":"Dave Walker","email":"email@daviey.com","username":"davewalker"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14103,"name":"wangtianfa","email":"wang.tianfa@99cloud.net","username":"wangtianfa"},{"value":0,"_account_id":19384,"name":"Serguei Bezverkhi","email":"sbezverk@cisco.com","username":"sbezverk"},{"value":0,"_account_id":18009,"name":"Md Nadeem","email":"mail2nadeem92@gmail.com","username":"mail2nadeem92"},{"value":0,"_account_id":21513,"name":"lixinhu","email":"lixinhu@inspur.com","username":"lixinhu"},{"value":2,"date":"2016-09-19 14:16:26.000000000","_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},{"value":0,"_account_id":2834,"name":"Steven Dake","email":"steven.dake@gmail.com","username":"sdake"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":979,"name":"Dave Walker","email":"email@daviey.com","username":"davewalker"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14103,"name":"wangtianfa","email":"wang.tianfa@99cloud.net","username":"wangtianfa"},{"value":0,"_account_id":19384,"name":"Serguei Bezverkhi","email":"sbezverk@cisco.com","username":"sbezverk"},{"value":0,"_account_id":18009,"name":"Md Nadeem","email":"mail2nadeem92@gmail.com","username":"mail2nadeem92"},{"value":0,"_account_id":21513,"name":"lixinhu","email":"lixinhu@inspur.com","username":"lixinhu"},{"value":1,"date":"2016-09-19 14:16:26.000000000","_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},{"value":0,"_account_id":2834,"name":"Steven Dake","email":"steven.dake@gmail.com","username":"sdake"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":979,"name":"Dave Walker","email":"email@daviey.com","username":"davewalker"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14103,"name":"wangtianfa","email":"wang.tianfa@99cloud.net","username":"wangtianfa"},{"value":0,"_account_id":19384,"name":"Serguei Bezverkhi","email":"sbezverk@cisco.com","username":"sbezverk"},{"value":0,"_account_id":18009,"name":"Md Nadeem","email":"mail2nadeem92@gmail.com","username":"mail2nadeem92"},{"value":0,"_account_id":21513,"name":"lixinhu","email":"lixinhu@inspur.com","username":"lixinhu"},{"value":0,"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},{"value":0,"_account_id":2834,"name":"Steven Dake","email":"steven.dake@gmail.com","username":"sdake"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":979,"name":"Dave Walker","email":"email@daviey.com","username":"davewalker"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14103,"name":"wangtianfa","email":"wang.tianfa@99cloud.net","username":"wangtianfa"},{"value":0,"_account_id":19384,"name":"Serguei Bezverkhi","email":"sbezverk@cisco.com","username":"sbezverk"},{"value":0,"_account_id":18009,"name":"Md Nadeem","email":"mail2nadeem92@gmail.com","username":"mail2nadeem92"},{"value":0,"_account_id":21513,"name":"lixinhu","email":"lixinhu@inspur.com","username":"lixinhu"},{"value":0,"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":979,"name":"Dave Walker","email":"email@daviey.com","username":"davewalker"},{"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},{"_account_id":2834,"name":"Steven Dake","email":"steven.dake@gmail.com","username":"sdake"},{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},{"_account_id":14103,"name":"wangtianfa","email":"wang.tianfa@99cloud.net","username":"wangtianfa"},{"_account_id":18009,"name":"Md Nadeem","email":"mail2nadeem92@gmail.com","username":"mail2nadeem92"},{"_account_id":19384,"name":"Serguei Bezverkhi","email":"sbezverk@cisco.com","username":"sbezverk"},{"_account_id":21513,"name":"lixinhu","email":"lixinhu@inspur.com","username":"lixinhu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-07-29 20:42:22.000000000","updated_by":{"_account_id":19384,"name":"Serguei Bezverkhi","email":"sbezverk@cisco.com","username":"sbezverk"},"reviewer":{"_account_id":19384,"name":"Serguei Bezverkhi","email":"sbezverk@cisco.com","username":"sbezverk"},"state":"REVIEWER"},{"updated":"2016-08-19 11:03:23.000000000","updated_by":{"_account_id":979,"name":"Dave Walker","email":"email@daviey.com","username":"davewalker"},"reviewer":{"_account_id":979,"name":"Dave Walker","email":"email@daviey.com","username":"davewalker"},"state":"REVIEWER"},{"updated":"2016-09-01 09:24:42.000000000","updated_by":{"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},"reviewer":{"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},"state":"REVIEWER"},{"updated":"2016-09-07 04:17:24.000000000","updated_by":{"_account_id":18009,"name":"Md Nadeem","email":"mail2nadeem92@gmail.com","username":"mail2nadeem92"},"reviewer":{"_account_id":18009,"name":"Md Nadeem","email":"mail2nadeem92@gmail.com","username":"mail2nadeem92"},"state":"REVIEWER"},{"updated":"2016-09-07 10:56:10.000000000","updated_by":{"_account_id":14103,"name":"wangtianfa","email":"wang.tianfa@99cloud.net","username":"wangtianfa"},"reviewer":{"_account_id":14103,"name":"wangtianfa","email":"wang.tianfa@99cloud.net","username":"wangtianfa"},"state":"REVIEWER"},{"updated":"2016-09-12 07:48:03.000000000","updated_by":{"_account_id":21513,"name":"lixinhu","email":"lixinhu@inspur.com","username":"lixinhu"},"reviewer":{"_account_id":21513,"name":"lixinhu","email":"lixinhu@inspur.com","username":"lixinhu"},"state":"REVIEWER"},{"updated":"2016-09-13 05:37:59.000000000","updated_by":{"_account_id":2834,"name":"Steven Dake","email":"steven.dake@gmail.com","username":"sdake"},"reviewer":{"_account_id":2834,"name":"Steven Dake","email":"steven.dake@gmail.com","username":"sdake"},"state":"REVIEWER"},{"updated":"2016-09-19 07:19:46.000000000","updated_by":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"reviewer":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"state":"REVIEWER"},{"updated":"2016-09-19 14:16:26.000000000","updated_by":{"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},"reviewer":{"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},"state":"REVIEWER"},{"updated":"2016-09-19 14:23:14.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"caad02c104c65ecf85f69a4e62ca37b5fe2bbf98","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2016-07-29 18:23:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d452a57f253d06756c2dd7b0f67ecd8927279e61","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2016-07-29 20:18:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0005580b895356b2b4631117b9cbebd84cacbec1","author":{"_account_id":19384,"name":"Serguei Bezverkhi","email":"sbezverk@cisco.com","username":"sbezverk"},"date":"2016-07-29 20:42:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"926ad3a9f2d4f578c9fb4661d7917d609fed5d41","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-30 00:00:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-docs-ubuntu-xenial http://docs-draft.openstack.org/06/349006/2/check/gate-kolla-docs-ubuntu-xenial/fc89cc7//doc/build/html/ : SUCCESS in 5m 02s\n- gate-kolla-pep8-ubuntu-xenial http://logs.openstack.org/06/349006/2/check/gate-kolla-pep8-ubuntu-xenial/c32a1bb/ : SUCCESS in 3m 11s\n- gate-kolla-python27-ubuntu-xenial http://logs.openstack.org/06/349006/2/check/gate-kolla-python27-ubuntu-xenial/68dc9a2/ : SUCCESS in 6m 02s\n- gate-kolla-python34 http://logs.openstack.org/06/349006/2/check/gate-kolla-python34/2c00a53/ : SUCCESS in 3m 20s\n- gate-kolla-python35-nv http://logs.openstack.org/06/349006/2/check/gate-kolla-python35-nv/5784f2b/ : SUCCESS in 6m 49s (non-voting)\n- gate-kolla-tox-bandit http://logs.openstack.org/06/349006/2/check/gate-kolla-tox-bandit/5debeb1/ : FAILURE in 6m 30s (non-voting)\n- gate-kolla-dsvm-build-centos-binary http://logs.openstack.org/06/349006/2/check/gate-kolla-dsvm-build-centos-binary/150338a/ : SUCCESS in 25m 37s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary http://logs.openstack.org/06/349006/2/check/gate-kolla-dsvm-deploy-centos-binary/70517cf/ : SUCCESS in 45m 10s (non-voting)\n- gate-kolla-dsvm-build-centos-source http://logs.openstack.org/06/349006/2/check/gate-kolla-dsvm-build-centos-source/82a3d1a/ : SUCCESS in 27m 22s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source http://logs.openstack.org/06/349006/2/check/gate-kolla-dsvm-deploy-centos-source/021c8f9/ : FAILURE in 5m 28s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source http://logs.openstack.org/06/349006/2/check/gate-kolla-dsvm-build-ubuntu-source/7c87e2f/ : SUCCESS in 29m 38s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source http://logs.openstack.org/06/349006/2/check/gate-kolla-dsvm-deploy-ubuntu-source/c81c60f/ : FAILURE in 28m 17s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-binary http://logs.openstack.org/06/349006/2/check/gate-kolla-dsvm-build-oraclelinux-binary/8186896/ : FAILURE in 9m 24s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-binary http://logs.openstack.org/06/349006/2/check/gate-kolla-dsvm-deploy-oraclelinux-binary/acdb168/ : FAILURE in 19m 17s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-source http://logs.openstack.org/06/349006/2/check/gate-kolla-dsvm-build-oraclelinux-source/414ee3d/ : FAILURE in 19m 55s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-source http://logs.openstack.org/06/349006/2/check/gate-kolla-dsvm-deploy-oraclelinux-source/978b011/ : FAILURE in 19m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dbed7bd07c30ab37e7361578cc25d72b21adcc03","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2016-07-30 06:44:30.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ca6ed9f86beb6da9e338296e3c89dd52653c26d7","author":{"_account_id":979,"name":"Dave Walker","email":"email@daviey.com","username":"davewalker"},"date":"2016-07-30 09:49:19.000000000","message":"Patch Set 2:\n\nCADF is more than just enablement for ceilometer.. anyone looking to audit their deployment will likely want this.","accounts_in_message":[],"_revision_number":2},{"id":"daaebed4b80c0028895989d9b936546e42cd98d0","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2016-07-31 16:49:03.000000000","message":"Patch Set 2:\n\nI think there are environments that do not use Ceilometer but want to use CADF events. Should I introduce a new configuration parameter to enable CADF events?","accounts_in_message":[],"_revision_number":2},{"id":"3c4e018be03709038a24ac4a44d0f45dd6adf3ad","author":{"_account_id":19384,"name":"Serguei Bezverkhi","email":"sbezverk@cisco.com","username":"sbezverk"},"date":"2016-07-31 21:57:07.000000000","message":"Patch Set 2: Code-Review-1\n\nThis setting looks like a part of \"more secured\" installation. Jeffrey, what do you think if new variable gets introduced, not just to enable CADF feature but to cover all current and possible future \"more\" secured settings/features?","accounts_in_message":[],"_revision_number":2},{"id":"e5677bd68036f8648ca3a9e53f489f526e15e420","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2016-08-04 09:56:19.000000000","message":"Patch Set 2:\n\nI do not have a strong behaviour. For me it is fine to introduce a new variable to enable additional security features. What about \"enable_enhanced_security\" as variable name?","accounts_in_message":[],"_revision_number":2},{"id":"a3fe87b714b6793dfaa9121c9db9630d32cc9f10","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2016-08-05 21:31:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d98974dbfeeff2ce6bdc90761312747d426a3630","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-05 22:34:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-docs-ubuntu-xenial http://docs-draft.openstack.org/06/349006/3/check/gate-kolla-docs-ubuntu-xenial/ce889d1//doc/build/html/ : SUCCESS in 3m 36s\n- gate-kolla-pep8-ubuntu-xenial http://logs.openstack.org/06/349006/3/check/gate-kolla-pep8-ubuntu-xenial/fc3a18d/ : SUCCESS in 3m 01s\n- gate-kolla-python27-ubuntu-xenial http://logs.openstack.org/06/349006/3/check/gate-kolla-python27-ubuntu-xenial/c646dd9/ : SUCCESS in 4m 48s\n- gate-kolla-python34 http://logs.openstack.org/06/349006/3/check/gate-kolla-python34/b566796/ : SUCCESS in 3m 19s\n- gate-kolla-python35-nv http://logs.openstack.org/06/349006/3/check/gate-kolla-python35-nv/52baa18/ : SUCCESS in 6m 11s (non-voting)\n- gate-kolla-tox-bandit-ubuntu-trusty http://logs.openstack.org/06/349006/3/check/gate-kolla-tox-bandit-ubuntu-trusty/861ee90/ : FAILURE in 3m 53s (non-voting)\n- gate-kolla-tox-bandit-ubuntu-xenial http://logs.openstack.org/06/349006/3/check/gate-kolla-tox-bandit-ubuntu-xenial/fc732c0/ : FAILURE in 2m 15s (non-voting)\n- gate-kolla-dsvm-build-centos-binary http://logs.openstack.org/06/349006/3/check/gate-kolla-dsvm-build-centos-binary/381491a/ : SUCCESS in 34m 49s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary http://logs.openstack.org/06/349006/3/check/gate-kolla-dsvm-deploy-centos-binary/3cfa464/ : SUCCESS in 43m 13s (non-voting)\n- gate-kolla-dsvm-build-centos-source http://logs.openstack.org/06/349006/3/check/gate-kolla-dsvm-build-centos-source/5f04901/ : SUCCESS in 28m 21s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source http://logs.openstack.org/06/349006/3/check/gate-kolla-dsvm-deploy-centos-source/73e51a3/ : SUCCESS in 41m 11s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source http://logs.openstack.org/06/349006/3/check/gate-kolla-dsvm-build-ubuntu-source/026996e/ : SUCCESS in 23m 43s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source http://logs.openstack.org/06/349006/3/check/gate-kolla-dsvm-deploy-ubuntu-source/6bbf747/ : FAILURE in 27m 43s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-binary http://logs.openstack.org/06/349006/3/check/gate-kolla-dsvm-build-oraclelinux-binary/4cb8060/ : SUCCESS in 50m 13s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-binary http://logs.openstack.org/06/349006/3/check/gate-kolla-dsvm-deploy-oraclelinux-binary/0c1f55d/ : SUCCESS in 52m 01s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-source http://logs.openstack.org/06/349006/3/check/gate-kolla-dsvm-build-oraclelinux-source/c030379/ : SUCCESS in 33m 56s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-source http://logs.openstack.org/06/349006/3/check/gate-kolla-dsvm-deploy-oraclelinux-source/14f5246/ : SUCCESS in 50m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"00332c4920cf42ba6ae895faccbd49f81903f8ed","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2016-08-06 17:46:46.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2ca46499a54efa850b82fb9e3ca8e931594a2cb7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-12 15:45:56.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"9a8afd42c9f37407692225723b3365310f1c1727","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2016-08-17 21:27:04.000000000","message":"Patch Set 3: Workflow-1\n\nOther service will implement CADF notifications in the future. I will rebase this tomorrow and will rename the variable to enable_cadf_notifications. This way we do not have to define what we mean with \"enhanced security\".","accounts_in_message":[],"_revision_number":3},{"id":"c9ff0e42ba6e4d0beb6b036e5c0ae7df0eb6d78a","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2016-08-19 10:14:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bd20657eb94f65689ce94247c159cb79717a7c8c","author":{"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},"date":"2016-08-19 10:42:59.000000000","message":"Patch Set 4: Code-Review-1\n\nCan you add a reno for this?","accounts_in_message":[],"_revision_number":4},{"id":"5f0897e34bf3608f2e75a028d69c40ca981a2d1e","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2016-08-19 11:01:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0738a0adb6f728fe8f99cf731bc7e902163d4d74","author":{"_account_id":979,"name":"Dave Walker","email":"email@daviey.com","username":"davewalker"},"date":"2016-08-19 11:03:23.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"5f197370d595642a17e417e9d2f1172c373b1f86","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-19 12:08:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-docs-ubuntu-xenial http://docs-draft.openstack.org/06/349006/5/check/gate-kolla-docs-ubuntu-xenial/24d1416//doc/build/html/ : SUCCESS in 3m 01s\n- gate-kolla-pep8-ubuntu-xenial http://logs.openstack.org/06/349006/5/check/gate-kolla-pep8-ubuntu-xenial/7daa246/ : SUCCESS in 3m 22s\n- gate-kolla-python27-ubuntu-xenial http://logs.openstack.org/06/349006/5/check/gate-kolla-python27-ubuntu-xenial/af4bf15/ : SUCCESS in 4m 36s\n- gate-kolla-python34 http://logs.openstack.org/06/349006/5/check/gate-kolla-python34/4280346/ : SUCCESS in 2m 26s\n- gate-kolla-python35-nv http://logs.openstack.org/06/349006/5/check/gate-kolla-python35-nv/02842c4/ : SUCCESS in 3m 56s (non-voting)\n- gate-kolla-releasenotes http://docs-draft.openstack.org/06/349006/5/check/gate-kolla-releasenotes/9b1aebf//releasenotes/build/html/ : SUCCESS in 4m 29s\n- gate-kolla-tox-bandit-ubuntu-xenial http://logs.openstack.org/06/349006/5/check/gate-kolla-tox-bandit-ubuntu-xenial/b4ade10/ : FAILURE in 3m 06s (non-voting)\n- gate-kolla-dsvm-build-centos-binary-centos-7-nv http://logs.openstack.org/06/349006/5/check/gate-kolla-dsvm-build-centos-binary-centos-7-nv/8765892/ : SUCCESS in 21m 43s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary-centos-7-nv http://logs.openstack.org/06/349006/5/check/gate-kolla-dsvm-deploy-centos-binary-centos-7-nv/5fe4519/ : FAILURE in 20m 59s (non-voting)\n- gate-kolla-dsvm-build-centos-source-centos-7-nv http://logs.openstack.org/06/349006/5/check/gate-kolla-dsvm-build-centos-source-centos-7-nv/007680a/ : SUCCESS in 40m 00s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source-centos-7-nv http://logs.openstack.org/06/349006/5/check/gate-kolla-dsvm-deploy-centos-source-centos-7-nv/f236a3d/ : FAILURE in 19m 33s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source-ubuntu-xenial-nv http://logs.openstack.org/06/349006/5/check/gate-kolla-dsvm-build-ubuntu-source-ubuntu-xenial-nv/1a32a66/ : SUCCESS in 20m 49s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source-ubuntu-xenial-nv http://logs.openstack.org/06/349006/5/check/gate-kolla-dsvm-deploy-ubuntu-source-ubuntu-xenial-nv/ec07cc4/ : FAILURE in 20m 41s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-binary-centos-7-nv http://logs.openstack.org/06/349006/5/check/gate-kolla-dsvm-build-oraclelinux-binary-centos-7-nv/1ddc97e/ : SUCCESS in 52m 38s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-binary-centos-7-nv http://logs.openstack.org/06/349006/5/check/gate-kolla-dsvm-deploy-oraclelinux-binary-centos-7-nv/af915be/ : FAILURE in 27m 44s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-source-centos-7-nv http://logs.openstack.org/06/349006/5/check/gate-kolla-dsvm-build-oraclelinux-source-centos-7-nv/48d4bcf/ : SUCCESS in 34m 39s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-source-centos-7-nv http://logs.openstack.org/06/349006/5/check/gate-kolla-dsvm-deploy-oraclelinux-source-centos-7-nv/305a7c7/ : FAILURE in 16m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1f7567081b06a47694e45b282e73fabc3f0b9b5c","author":{"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},"date":"2016-08-19 13:02:47.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"155fbd76417162e3a080bd0f7e65f3b13354bd4d","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2016-08-19 13:39:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3755549ecab20187985e631a534cf4bb83265531","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2016-08-19 13:39:59.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"313a3caaadadce24a4bd3085f959f1e4640edd47","author":{"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},"date":"2016-08-19 14:01:11.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6f87e00d9060fda83c76594477ffa51a3df85d4d","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2016-08-19 14:05:28.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"81fe10077f259f82f5031207b7014253768bb32e","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2016-08-19 14:05:53.000000000","message":"Patch Set 7:\n\nI should have a break. Also fixed the commit message.","accounts_in_message":[],"_revision_number":7},{"id":"aba28e036ff3e8f288f53cf8938857e786810ee3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-19 18:51:14.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-docs-ubuntu-xenial http://docs-draft.openstack.org/06/349006/7/check/gate-kolla-docs-ubuntu-xenial/13cac58//doc/build/html/ : SUCCESS in 3m 39s\n- gate-kolla-pep8-ubuntu-xenial http://logs.openstack.org/06/349006/7/check/gate-kolla-pep8-ubuntu-xenial/4e5c491/ : SUCCESS in 3m 53s\n- gate-kolla-python27-ubuntu-xenial http://logs.openstack.org/06/349006/7/check/gate-kolla-python27-ubuntu-xenial/e8dea2f/ : SUCCESS in 4m 26s\n- gate-kolla-python34 http://logs.openstack.org/06/349006/7/check/gate-kolla-python34/e39989e/ : SUCCESS in 2m 39s\n- gate-kolla-python35-nv http://logs.openstack.org/06/349006/7/check/gate-kolla-python35-nv/73c2188/ : SUCCESS in 4m 57s (non-voting)\n- gate-kolla-releasenotes http://docs-draft.openstack.org/06/349006/7/check/gate-kolla-releasenotes/5af1512//releasenotes/build/html/ : SUCCESS in 3m 55s\n- gate-kolla-tox-bandit-ubuntu-xenial http://logs.openstack.org/06/349006/7/check/gate-kolla-tox-bandit-ubuntu-xenial/ebe2d63/ : FAILURE in 4m 13s (non-voting)\n- gate-kolla-dsvm-build-centos-binary-centos-7-nv http://logs.openstack.org/06/349006/7/check/gate-kolla-dsvm-build-centos-binary-centos-7-nv/a1b0b92/ : SUCCESS in 36m 33s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary-centos-7-nv http://logs.openstack.org/06/349006/7/check/gate-kolla-dsvm-deploy-centos-binary-centos-7-nv/9e4d8bc/ : SUCCESS in 42m 12s (non-voting)\n- gate-kolla-dsvm-build-centos-source-centos-7-nv http://logs.openstack.org/06/349006/7/check/gate-kolla-dsvm-build-centos-source-centos-7-nv/9fa5ccd/ : SUCCESS in 26m 53s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source-centos-7-nv http://logs.openstack.org/06/349006/7/check/gate-kolla-dsvm-deploy-centos-source-centos-7-nv/803a179/ : SUCCESS in 43m 05s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source-ubuntu-xenial-nv http://logs.openstack.org/06/349006/7/check/gate-kolla-dsvm-build-ubuntu-source-ubuntu-xenial-nv/be3c8bb/ : SUCCESS in 17m 21s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source-ubuntu-xenial-nv http://logs.openstack.org/06/349006/7/check/gate-kolla-dsvm-deploy-ubuntu-source-ubuntu-xenial-nv/66077ae/ : FAILURE in 20m 32s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-binary-centos-7-nv http://logs.openstack.org/06/349006/7/check/gate-kolla-dsvm-build-oraclelinux-binary-centos-7-nv/23f0d59/ : SUCCESS in 55m 38s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-binary-centos-7-nv http://logs.openstack.org/06/349006/7/check/gate-kolla-dsvm-deploy-oraclelinux-binary-centos-7-nv/d23c335/ : SUCCESS in 56m 30s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-source-centos-7-nv http://logs.openstack.org/06/349006/7/check/gate-kolla-dsvm-build-oraclelinux-source-centos-7-nv/c00441a/ : SUCCESS in 24m 07s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-source-centos-7-nv http://logs.openstack.org/06/349006/7/check/gate-kolla-dsvm-deploy-oraclelinux-source-centos-7-nv/24fdfeb/ : SUCCESS in 48m 15s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8ba3609f23f76804896e30501cbe8f290d5418a4","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2016-08-31 12:19:44.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"9719a773acaa5607a0e5f5e1139a65b0fdfb84db","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2016-08-31 12:41:09.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"f1bbd41afa9e281abc36df48f7ab090bbacd1203","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2016-08-31 12:45:04.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ea6d9a9e3078a64c1a005346cc9a714b4ee01040","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2016-08-31 12:45:53.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"3322d6c20e05a4bc7e513c78e88f57e7ecf20bcb","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2016-08-31 12:49:09.000000000","message":"Patch Set 9: Code-Review+1\n\nwait for the gate.","accounts_in_message":[],"_revision_number":9},{"id":"b8502abfdd33b21f2ff781dcec5c4e7a7bec3ae0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-31 13:45:37.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-docs-ubuntu-xenial http://docs-draft.openstack.org/06/349006/9/check/gate-kolla-docs-ubuntu-xenial/95f1d87//doc/build/html/ : SUCCESS in 3m 43s\n- gate-kolla-pep8-ubuntu-xenial http://logs.openstack.org/06/349006/9/check/gate-kolla-pep8-ubuntu-xenial/62b3c77/ : SUCCESS in 3m 46s\n- gate-kolla-python27-ubuntu-xenial http://logs.openstack.org/06/349006/9/check/gate-kolla-python27-ubuntu-xenial/7ea7e96/ : SUCCESS in 5m 17s\n- gate-kolla-python34 http://logs.openstack.org/06/349006/9/check/gate-kolla-python34/47c3989/ : SUCCESS in 2m 22s\n- gate-kolla-python35-nv http://logs.openstack.org/06/349006/9/check/gate-kolla-python35-nv/03a3bc4/ : SUCCESS in 3m 55s (non-voting)\n- gate-kolla-releasenotes http://docs-draft.openstack.org/06/349006/9/check/gate-kolla-releasenotes/de6f43e//releasenotes/build/html/ : SUCCESS in 2m 12s\n- gate-kolla-tox-bandit-ubuntu-xenial http://logs.openstack.org/06/349006/9/check/gate-kolla-tox-bandit-ubuntu-xenial/19318b4/ : SUCCESS in 2m 27s (non-voting)\n- gate-kolla-dsvm-build-centos-binary-centos-7-nv http://logs.openstack.org/06/349006/9/check/gate-kolla-dsvm-build-centos-binary-centos-7-nv/9fea712/ : SUCCESS in 37m 43s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary-centos-7-nv http://logs.openstack.org/06/349006/9/check/gate-kolla-dsvm-deploy-centos-binary-centos-7-nv/4b64ee4/ : SUCCESS in 47m 28s (non-voting)\n- gate-kolla-dsvm-build-centos-source-centos-7-nv http://logs.openstack.org/06/349006/9/check/gate-kolla-dsvm-build-centos-source-centos-7-nv/185c020/ : SUCCESS in 33m 14s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source-centos-7-nv http://logs.openstack.org/06/349006/9/check/gate-kolla-dsvm-deploy-centos-source-centos-7-nv/34bc54b/ : SUCCESS in 31m 40s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-binary-ubuntu-xenial-nv http://logs.openstack.org/06/349006/9/check/gate-kolla-dsvm-build-ubuntu-binary-ubuntu-xenial-nv/c9828b0/ : SUCCESS in 17m 15s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source-ubuntu-xenial-nv http://logs.openstack.org/06/349006/9/check/gate-kolla-dsvm-build-ubuntu-source-ubuntu-xenial-nv/637b700/ : SUCCESS in 17m 07s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-binary-ubuntu-xenial-nv http://logs.openstack.org/06/349006/9/check/gate-kolla-dsvm-deploy-ubuntu-binary-ubuntu-xenial-nv/ceb4f88/ : SUCCESS in 26m 47s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source-ubuntu-xenial-nv http://logs.openstack.org/06/349006/9/check/gate-kolla-dsvm-deploy-ubuntu-source-ubuntu-xenial-nv/2bdbfc2/ : SUCCESS in 26m 18s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-binary-centos-7-nv http://logs.openstack.org/06/349006/9/check/gate-kolla-dsvm-build-oraclelinux-binary-centos-7-nv/3721b9d/ : SUCCESS in 56m 56s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-binary-centos-7-nv http://logs.openstack.org/06/349006/9/check/gate-kolla-dsvm-deploy-oraclelinux-binary-centos-7-nv/a3d71d6/ : SUCCESS in 36m 22s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-source-centos-7-nv http://logs.openstack.org/06/349006/9/check/gate-kolla-dsvm-build-oraclelinux-source-centos-7-nv/7c67dd8/ : SUCCESS in 32m 51s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-source-centos-7-nv http://logs.openstack.org/06/349006/9/check/gate-kolla-dsvm-deploy-oraclelinux-source-centos-7-nv/19f0f75/ : SUCCESS in 36m 21s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f537310874537fadc3043aab5d4fbf5b669d1c70","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-31 14:51:50.000000000","message":"Patch Set 9: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"061e1c798d6aff94f52b91cc87f808972720e66f","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2016-08-31 18:24:10.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"31fc9403e5e26352c33f8b094f1bd3ad1d61055b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-31 19:39:03.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-docs-ubuntu-xenial http://docs-draft.openstack.org/06/349006/10/check/gate-kolla-docs-ubuntu-xenial/021f2c9//doc/build/html/ : SUCCESS in 2m 28s\n- gate-kolla-pep8-ubuntu-xenial http://logs.openstack.org/06/349006/10/check/gate-kolla-pep8-ubuntu-xenial/394a99e/ : SUCCESS in 4m 20s\n- gate-kolla-python27-ubuntu-xenial http://logs.openstack.org/06/349006/10/check/gate-kolla-python27-ubuntu-xenial/2883442/ : SUCCESS in 2m 29s\n- gate-kolla-python34 http://logs.openstack.org/06/349006/10/check/gate-kolla-python34/84180e7/ : SUCCESS in 4m 14s\n- gate-kolla-python35-nv http://logs.openstack.org/06/349006/10/check/gate-kolla-python35-nv/1472686/ : SUCCESS in 3m 09s (non-voting)\n- gate-kolla-releasenotes http://docs-draft.openstack.org/06/349006/10/check/gate-kolla-releasenotes/6d2e552//releasenotes/build/html/ : SUCCESS in 2m 21s\n- gate-kolla-tox-bandit-ubuntu-xenial http://logs.openstack.org/06/349006/10/check/gate-kolla-tox-bandit-ubuntu-xenial/9152597/ : SUCCESS in 2m 31s (non-voting)\n- gate-kolla-dsvm-build-centos-binary-centos-7-nv http://logs.openstack.org/06/349006/10/check/gate-kolla-dsvm-build-centos-binary-centos-7-nv/54019aa/ : SUCCESS in 51m 21s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary-centos-7-nv http://logs.openstack.org/06/349006/10/check/gate-kolla-dsvm-deploy-centos-binary-centos-7-nv/1fdcdeb/ : SUCCESS in 48m 06s (non-voting)\n- gate-kolla-dsvm-build-centos-source-centos-7-nv http://logs.openstack.org/06/349006/10/check/gate-kolla-dsvm-build-centos-source-centos-7-nv/50debf1/ : SUCCESS in 35m 47s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source-centos-7-nv http://logs.openstack.org/06/349006/10/check/gate-kolla-dsvm-deploy-centos-source-centos-7-nv/9098626/ : SUCCESS in 51m 21s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-binary-ubuntu-xenial-nv http://logs.openstack.org/06/349006/10/check/gate-kolla-dsvm-build-ubuntu-binary-ubuntu-xenial-nv/de5155a/ : SUCCESS in 18m 02s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source-ubuntu-xenial-nv http://logs.openstack.org/06/349006/10/check/gate-kolla-dsvm-build-ubuntu-source-ubuntu-xenial-nv/46d4724/ : SUCCESS in 19m 58s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-binary-ubuntu-xenial-nv http://logs.openstack.org/06/349006/10/check/gate-kolla-dsvm-deploy-ubuntu-binary-ubuntu-xenial-nv/a2f2851/ : SUCCESS in 26m 00s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source-ubuntu-xenial-nv http://logs.openstack.org/06/349006/10/check/gate-kolla-dsvm-deploy-ubuntu-source-ubuntu-xenial-nv/a1a5ee8/ : SUCCESS in 26m 30s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-binary-centos-7-nv http://logs.openstack.org/06/349006/10/check/gate-kolla-dsvm-build-oraclelinux-binary-centos-7-nv/fd45b82/ : SUCCESS in 39m 16s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-binary-centos-7-nv http://logs.openstack.org/06/349006/10/check/gate-kolla-dsvm-deploy-oraclelinux-binary-centos-7-nv/f7abaa7/ : FAILURE in 1h 03m 46s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-source-centos-7-nv http://logs.openstack.org/06/349006/10/check/gate-kolla-dsvm-build-oraclelinux-source-centos-7-nv/1446aab/ : SUCCESS in 46m 19s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-source-centos-7-nv http://logs.openstack.org/06/349006/10/check/gate-kolla-dsvm-deploy-oraclelinux-source-centos-7-nv/1ff36ae/ : SUCCESS in 34m 51s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"49e7504416d7cfc9e735d9063e8a89282c199997","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2016-09-01 06:21:05.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"e978ee741a3794f6132c354b54a45e29961788f9","author":{"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},"date":"2016-09-01 09:24:42.000000000","message":"Patch Set 10:\n\nWill it work with changes suggested by https://review.openstack.org/#/c/356449/","accounts_in_message":[],"_revision_number":10},{"id":"6a82e3ca57bd95b60ffc0a25e494afe409fdac9d","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2016-09-01 21:15:10.000000000","message":"Patch Set 10: Workflow-1\n\nI will check if transport_Url is available for keystone.","accounts_in_message":[],"_revision_number":10},{"id":"53064f9d68c2fda505f0ceeea73ef87eb5fef72d","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2016-09-02 00:52:54.000000000","message":"Patch Set 10:\n\nChristian Berendt wrote:\n\n\u003e I will check if transport_Url is available for keystone.\n\nya. it should work. all the OpenStack service both using oslo.messagin","accounts_in_message":[],"_revision_number":10},{"id":"55b994be14dcfbbad6ce52e85dab6458e496a127","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2016-09-06 06:58:15.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"0595689a78c6d8a65672af6fc3f6bd3291180dbe","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-06 08:01:43.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-docs-ubuntu-xenial http://docs-draft.openstack.org/06/349006/11/check/gate-kolla-docs-ubuntu-xenial/778f00f//doc/build/html/ : SUCCESS in 4m 46s\n- gate-kolla-pep8-ubuntu-xenial http://logs.openstack.org/06/349006/11/check/gate-kolla-pep8-ubuntu-xenial/61fc1a4/ : SUCCESS in 2m 04s\n- gate-kolla-python27-ubuntu-xenial http://logs.openstack.org/06/349006/11/check/gate-kolla-python27-ubuntu-xenial/51c56ef/ : SUCCESS in 3m 22s\n- gate-kolla-python34 http://logs.openstack.org/06/349006/11/check/gate-kolla-python34/d143cbb/ : SUCCESS in 2m 11s\n- gate-kolla-python35-nv http://logs.openstack.org/06/349006/11/check/gate-kolla-python35-nv/4174f59/ : SUCCESS in 2m 39s (non-voting)\n- gate-kolla-releasenotes http://docs-draft.openstack.org/06/349006/11/check/gate-kolla-releasenotes/adad82c//releasenotes/build/html/ : SUCCESS in 3m 29s\n- gate-kolla-tox-bandit-ubuntu-xenial http://logs.openstack.org/06/349006/11/check/gate-kolla-tox-bandit-ubuntu-xenial/824d375/ : SUCCESS in 4m 30s (non-voting)\n- gate-kolla-dsvm-build-centos-binary-centos-7-nv http://logs.openstack.org/06/349006/11/check/gate-kolla-dsvm-build-centos-binary-centos-7-nv/a62995e/ : SUCCESS in 34m 32s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary-centos-7-nv http://logs.openstack.org/06/349006/11/check/gate-kolla-dsvm-deploy-centos-binary-centos-7-nv/e333314/ : FAILURE in 1h 02m 22s (non-voting)\n- gate-kolla-dsvm-build-centos-source-centos-7-nv http://logs.openstack.org/06/349006/11/check/gate-kolla-dsvm-build-centos-source-centos-7-nv/75a2bac/ : SUCCESS in 25m 50s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source-centos-7-nv http://logs.openstack.org/06/349006/11/check/gate-kolla-dsvm-deploy-centos-source-centos-7-nv/b4861ae/ : SUCCESS in 47m 47s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-binary-ubuntu-xenial-nv http://logs.openstack.org/06/349006/11/check/gate-kolla-dsvm-build-ubuntu-binary-ubuntu-xenial-nv/9ac1ac4/ : SUCCESS in 22m 13s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source-ubuntu-xenial-nv http://logs.openstack.org/06/349006/11/check/gate-kolla-dsvm-build-ubuntu-source-ubuntu-xenial-nv/a352b9f/ : SUCCESS in 27m 41s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-binary-ubuntu-xenial-nv http://logs.openstack.org/06/349006/11/check/gate-kolla-dsvm-deploy-ubuntu-binary-ubuntu-xenial-nv/fe30f65/ : SUCCESS in 38m 06s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source-ubuntu-xenial-nv http://logs.openstack.org/06/349006/11/check/gate-kolla-dsvm-deploy-ubuntu-source-ubuntu-xenial-nv/ac5b5b8/ : SUCCESS in 38m 52s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-binary-centos-7-nv http://logs.openstack.org/06/349006/11/check/gate-kolla-dsvm-build-oraclelinux-binary-centos-7-nv/e3a47be/ : SUCCESS in 58m 09s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-binary-centos-7-nv http://logs.openstack.org/06/349006/11/check/gate-kolla-dsvm-deploy-oraclelinux-binary-centos-7-nv/c9dadb0/ : SUCCESS in 53m 08s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-source-centos-7-nv http://logs.openstack.org/06/349006/11/check/gate-kolla-dsvm-build-oraclelinux-source-centos-7-nv/39109fd/ : SUCCESS in 42m 20s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-source-centos-7-nv http://logs.openstack.org/06/349006/11/check/gate-kolla-dsvm-deploy-oraclelinux-source-centos-7-nv/9abd722/ : SUCCESS in 55m 44s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"b0833958b1008434921b76e049f7b4638ad429dd","author":{"_account_id":18009,"name":"Md Nadeem","email":"mail2nadeem92@gmail.com","username":"mail2nadeem92"},"date":"2016-09-07 04:17:24.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"dcb69a7e9ee264428588ad2737c2c31485df8531","author":{"_account_id":14103,"name":"wangtianfa","email":"wang.tianfa@99cloud.net","username":"wangtianfa"},"date":"2016-09-07 10:56:10.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"cfaaf7e7c00c8c773b91aac4c0a6c10551fc630c","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2016-09-12 07:41:30.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"818e8f41da5021369b3330523e9572c12de642c6","author":{"_account_id":21513,"name":"lixinhu","email":"lixinhu@inspur.com","username":"lixinhu"},"date":"2016-09-12 07:48:03.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"b3e14c3fb28959697e85e0433b8c16b3826c4295","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-12 08:48:54.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-docs-ubuntu-xenial http://docs-draft.openstack.org/06/349006/12/check/gate-kolla-docs-ubuntu-xenial/60530a3//doc/build/html/ : SUCCESS in 1m 44s\n- gate-kolla-pep8-ubuntu-xenial http://logs.openstack.org/06/349006/12/check/gate-kolla-pep8-ubuntu-xenial/5b3f946/ : SUCCESS in 2m 14s\n- gate-kolla-python27-ubuntu-xenial http://logs.openstack.org/06/349006/12/check/gate-kolla-python27-ubuntu-xenial/e88c7bc/ : SUCCESS in 2m 08s\n- gate-kolla-python34 http://logs.openstack.org/06/349006/12/check/gate-kolla-python34/dc23ef7/ : SUCCESS in 1m 44s\n- gate-kolla-python35-nv http://logs.openstack.org/06/349006/12/check/gate-kolla-python35-nv/b60214d/ : SUCCESS in 3m 02s (non-voting)\n- gate-kolla-releasenotes http://docs-draft.openstack.org/06/349006/12/check/gate-kolla-releasenotes/9f986eb//releasenotes/build/html/ : SUCCESS in 1m 02s\n- gate-kolla-tox-bandit-ubuntu-xenial http://logs.openstack.org/06/349006/12/check/gate-kolla-tox-bandit-ubuntu-xenial/cb5d9c6/ : SUCCESS in 1m 19s (non-voting)\n- gate-kolla-dsvm-build-centos-binary-centos-7-nv http://logs.openstack.org/06/349006/12/check/gate-kolla-dsvm-build-centos-binary-centos-7-nv/2b8ac1a/ : SUCCESS in 33m 13s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary-centos-7-nv http://logs.openstack.org/06/349006/12/check/gate-kolla-dsvm-deploy-centos-binary-centos-7-nv/6a77d81/ : SUCCESS in 53m 24s (non-voting)\n- gate-kolla-dsvm-build-centos-source-centos-7-nv http://logs.openstack.org/06/349006/12/check/gate-kolla-dsvm-build-centos-source-centos-7-nv/d6c4384/ : SUCCESS in 48m 13s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source-centos-7-nv http://logs.openstack.org/06/349006/12/check/gate-kolla-dsvm-deploy-centos-source-centos-7-nv/db0645a/ : SUCCESS in 30m 58s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-binary-ubuntu-xenial-nv http://logs.openstack.org/06/349006/12/check/gate-kolla-dsvm-build-ubuntu-binary-ubuntu-xenial-nv/a801d3d/ : SUCCESS in 17m 16s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source-ubuntu-xenial-nv http://logs.openstack.org/06/349006/12/check/gate-kolla-dsvm-build-ubuntu-source-ubuntu-xenial-nv/43cc6af/ : SUCCESS in 19m 23s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-binary-ubuntu-xenial-nv http://logs.openstack.org/06/349006/12/check/gate-kolla-dsvm-deploy-ubuntu-binary-ubuntu-xenial-nv/5976151/ : SUCCESS in 23m 45s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source-ubuntu-xenial-nv http://logs.openstack.org/06/349006/12/check/gate-kolla-dsvm-deploy-ubuntu-source-ubuntu-xenial-nv/f79f38e/ : SUCCESS in 24m 42s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-binary-centos-7-nv http://logs.openstack.org/06/349006/12/check/gate-kolla-dsvm-build-oraclelinux-binary-centos-7-nv/0dc631d/ : SUCCESS in 39m 23s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-binary-centos-7-nv http://logs.openstack.org/06/349006/12/check/gate-kolla-dsvm-deploy-oraclelinux-binary-centos-7-nv/9449695/ : SUCCESS in 52m 33s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-source-centos-7-nv http://logs.openstack.org/06/349006/12/check/gate-kolla-dsvm-build-oraclelinux-source-centos-7-nv/1b52172/ : SUCCESS in 43m 10s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-source-centos-7-nv http://logs.openstack.org/06/349006/12/check/gate-kolla-dsvm-deploy-oraclelinux-source-centos-7-nv/cc580d6/ : SUCCESS in 53m 39s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"597926043f3556908ab82f25215d73bf471e5d89","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-13 05:15:30.000000000","message":"Patch Set 12: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":12},{"id":"21b45a8a9b0da4162077447bbe88ac9f3aa311e1","author":{"_account_id":2834,"name":"Steven Dake","email":"steven.dake@gmail.com","username":"sdake"},"date":"2016-09-13 05:37:59.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"574f9a0c9c83f587f66f6027853338210f03b168","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-13 06:32:19.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-docs-ubuntu-xenial http://docs-draft.openstack.org/06/349006/13/check/gate-kolla-docs-ubuntu-xenial/afd79f4//doc/build/html/ : SUCCESS in 1m 54s\n- gate-kolla-pep8-ubuntu-xenial http://logs.openstack.org/06/349006/13/check/gate-kolla-pep8-ubuntu-xenial/0027720/ : SUCCESS in 2m 03s\n- gate-kolla-python27-ubuntu-xenial http://logs.openstack.org/06/349006/13/check/gate-kolla-python27-ubuntu-xenial/6bf3ea9/ : SUCCESS in 1m 17s\n- gate-kolla-python34 http://logs.openstack.org/06/349006/13/check/gate-kolla-python34/5b29288/ : SUCCESS in 3m 03s\n- gate-kolla-python35-nv http://logs.openstack.org/06/349006/13/check/gate-kolla-python35-nv/2a6e6a9/ : SUCCESS in 1m 39s (non-voting)\n- gate-kolla-releasenotes http://docs-draft.openstack.org/06/349006/13/check/gate-kolla-releasenotes/78df9bb//releasenotes/build/html/ : SUCCESS in 1m 00s\n- gate-kolla-tox-bandit-ubuntu-xenial http://logs.openstack.org/06/349006/13/check/gate-kolla-tox-bandit-ubuntu-xenial/e833ce3/ : SUCCESS in 1m 11s (non-voting)\n- gate-kolla-dsvm-build-centos-binary-centos-7-nv http://logs.openstack.org/06/349006/13/check/gate-kolla-dsvm-build-centos-binary-centos-7-nv/2de15d4/ : SUCCESS in 31m 34s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary-centos-7-nv http://logs.openstack.org/06/349006/13/check/gate-kolla-dsvm-deploy-centos-binary-centos-7-nv/49f5fea/ : SUCCESS in 52m 47s (non-voting)\n- gate-kolla-dsvm-build-centos-source-centos-7-nv http://logs.openstack.org/06/349006/13/check/gate-kolla-dsvm-build-centos-source-centos-7-nv/e3d9e2c/ : SUCCESS in 32m 15s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source-centos-7-nv http://logs.openstack.org/06/349006/13/check/gate-kolla-dsvm-deploy-centos-source-centos-7-nv/1b1f1c6/ : SUCCESS in 32m 05s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-binary-ubuntu-xenial-nv http://logs.openstack.org/06/349006/13/check/gate-kolla-dsvm-build-ubuntu-binary-ubuntu-xenial-nv/25e8d9b/ : SUCCESS in 19m 25s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source-ubuntu-xenial-nv http://logs.openstack.org/06/349006/13/check/gate-kolla-dsvm-build-ubuntu-source-ubuntu-xenial-nv/50676d4/ : SUCCESS in 18m 54s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-binary-ubuntu-xenial-nv http://logs.openstack.org/06/349006/13/check/gate-kolla-dsvm-deploy-ubuntu-binary-ubuntu-xenial-nv/dd9dde1/ : SUCCESS in 23m 44s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source-ubuntu-xenial-nv http://logs.openstack.org/06/349006/13/check/gate-kolla-dsvm-deploy-ubuntu-source-ubuntu-xenial-nv/889379e/ : SUCCESS in 24m 52s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-binary-centos-7-nv http://logs.openstack.org/06/349006/13/check/gate-kolla-dsvm-build-oraclelinux-binary-centos-7-nv/9785293/ : SUCCESS in 34m 04s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-binary-centos-7-nv http://logs.openstack.org/06/349006/13/check/gate-kolla-dsvm-deploy-oraclelinux-binary-centos-7-nv/2544eb6/ : SUCCESS in 36m 15s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-source-centos-7-nv http://logs.openstack.org/06/349006/13/check/gate-kolla-dsvm-build-oraclelinux-source-centos-7-nv/d9b64f9/ : SUCCESS in 29m 44s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-source-centos-7-nv http://logs.openstack.org/06/349006/13/check/gate-kolla-dsvm-deploy-oraclelinux-source-centos-7-nv/b833f23/ : SUCCESS in 52m 05s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"0ff1f1aedf853dfb8de9c17f2188e6f15bfef959","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2016-09-16 12:19:53.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"c09d1105800b3740cb7d7d31e684dc9e94fe193d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-16 13:37:46.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-docs-ubuntu-xenial http://docs-draft.openstack.org/06/349006/14/check/gate-kolla-docs-ubuntu-xenial/e936afc//doc/build/html/ : SUCCESS in 1m 59s\n- gate-kolla-pep8-ubuntu-xenial http://logs.openstack.org/06/349006/14/check/gate-kolla-pep8-ubuntu-xenial/b3edead/ : SUCCESS in 2m 05s\n- gate-kolla-python27-ubuntu-xenial http://logs.openstack.org/06/349006/14/check/gate-kolla-python27-ubuntu-xenial/78861f2/ : SUCCESS in 2m 04s\n- gate-kolla-python34 http://logs.openstack.org/06/349006/14/check/gate-kolla-python34/7009c69/ : SUCCESS in 1m 23s\n- gate-kolla-python35-nv http://logs.openstack.org/06/349006/14/check/gate-kolla-python35-nv/f8705cc/ : SUCCESS in 2m 46s (non-voting)\n- gate-kolla-releasenotes http://docs-draft.openstack.org/06/349006/14/check/gate-kolla-releasenotes/8ed431a//releasenotes/build/html/ : SUCCESS in 1m 40s\n- gate-kolla-tox-bandit-ubuntu-xenial http://logs.openstack.org/06/349006/14/check/gate-kolla-tox-bandit-ubuntu-xenial/b5abb3e/ : SUCCESS in 2m 13s (non-voting)\n- gate-kolla-dsvm-build-centos-binary-centos-7-nv http://logs.openstack.org/06/349006/14/check/gate-kolla-dsvm-build-centos-binary-centos-7-nv/9d96da2/ : SUCCESS in 24m 02s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary-centos-7-nv http://logs.openstack.org/06/349006/14/check/gate-kolla-dsvm-deploy-centos-binary-centos-7-nv/5ccf0a2/ : SUCCESS in 39m 50s (non-voting)\n- gate-kolla-dsvm-build-centos-source-centos-7-nv http://logs.openstack.org/06/349006/14/check/gate-kolla-dsvm-build-centos-source-centos-7-nv/9767efb/ : SUCCESS in 42m 18s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source-centos-7-nv http://logs.openstack.org/06/349006/14/check/gate-kolla-dsvm-deploy-centos-source-centos-7-nv/f55c53d/ : SUCCESS in 1h 14m 39s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-binary-ubuntu-xenial-nv http://logs.openstack.org/06/349006/14/check/gate-kolla-dsvm-build-ubuntu-binary-ubuntu-xenial-nv/614d90b/ : SUCCESS in 16m 37s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source-ubuntu-xenial-nv http://logs.openstack.org/06/349006/14/check/gate-kolla-dsvm-build-ubuntu-source-ubuntu-xenial-nv/e5431c0/ : SUCCESS in 18m 00s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-binary-ubuntu-xenial-nv http://logs.openstack.org/06/349006/14/check/gate-kolla-dsvm-deploy-ubuntu-binary-ubuntu-xenial-nv/2be1223/ : SUCCESS in 29m 50s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source-ubuntu-xenial-nv http://logs.openstack.org/06/349006/14/check/gate-kolla-dsvm-deploy-ubuntu-source-ubuntu-xenial-nv/d962be7/ : SUCCESS in 30m 51s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-binary-centos-7-nv http://logs.openstack.org/06/349006/14/check/gate-kolla-dsvm-build-oraclelinux-binary-centos-7-nv/0b8191f/ : SUCCESS in 35m 21s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-binary-centos-7-nv http://logs.openstack.org/06/349006/14/check/gate-kolla-dsvm-deploy-oraclelinux-binary-centos-7-nv/73f05b4/ : SUCCESS in 1h 05m 15s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-source-centos-7-nv http://logs.openstack.org/06/349006/14/check/gate-kolla-dsvm-build-oraclelinux-source-centos-7-nv/78cbf42/ : SUCCESS in 47m 01s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-source-centos-7-nv http://logs.openstack.org/06/349006/14/check/gate-kolla-dsvm-deploy-oraclelinux-source-centos-7-nv/42543fd/ : SUCCESS in 59m 31s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"18007954d78f346b0319822656ce64f3ec96e765","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2016-09-19 07:19:46.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"eedf82b958ae338d657c9c121ef4f68bbd8c76b7","author":{"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},"date":"2016-09-19 14:16:26.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"8933ab9dfd3ffc84b0dd13d27bdcc76431e813a4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-19 14:20:27.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":14},{"id":"dd9eb252cacea4af2dc20b19d7d0334d9979b546","date":"2016-09-19 14:23:14.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":14},{"id":"636e8e9150feb642b04227351964c924628ce893","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-19 14:23:14.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-kolla-docs-ubuntu-xenial http://docs-draft.openstack.org/06/349006/14/gate/gate-kolla-docs-ubuntu-xenial/bbfeb8f//doc/build/html/ : SUCCESS in 1m 13s\n- gate-kolla-pep8-ubuntu-xenial http://logs.openstack.org/06/349006/14/gate/gate-kolla-pep8-ubuntu-xenial/1ff5d62/ : SUCCESS in 1m 26s\n- gate-kolla-python27-ubuntu-xenial http://logs.openstack.org/06/349006/14/gate/gate-kolla-python27-ubuntu-xenial/6c0e33b/ : SUCCESS in 1m 26s\n- gate-kolla-python34 http://logs.openstack.org/06/349006/14/gate/gate-kolla-python34/8918976/ : SUCCESS in 1m 57s\n- gate-kolla-releasenotes http://docs-draft.openstack.org/06/349006/14/gate/gate-kolla-releasenotes/36c63d5//releasenotes/build/html/ : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"8f6bd4e73539e89ef556a1a8f1f33936ec24f3a4","revisions":{"1f4875f8cfee7b88fb06748f57869877e3a989d5":{"kind":"REWORK","_number":1,"created":"2016-07-29 18:23:47.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/06/349006/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/349006/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/349006/1"}}},"commit":{"parents":[{"commit":"74ea96bacfaa082db2f64aa3cc78aed59ea5822b","subject":"Merge \"Clean up Glance image after sanity check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74ea96bacfaa082db2f64aa3cc78aed59ea5822b"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-07-29 18:23:13.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-07-29 18:23:13.000000000","tz":120},"subject":"Enable CADF events in Keystone","message":"Enable CADF events in Keystone\n\nChange-Id: I0a40867ae65011fbfd53308b9d7f42d10435f5fa\nCloses-bug: #1607905\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f4875f8cfee7b88fb06748f57869877e3a989d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f4875f8cfee7b88fb06748f57869877e3a989d5"}]},"branch":"refs/heads/master"},"7ec057d5f0aa5e12d3c649a442e34c6127020abe":{"kind":"REWORK","_number":2,"created":"2016-07-29 20:18:24.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/06/349006/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/349006/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/349006/2"}}},"commit":{"parents":[{"commit":"74ea96bacfaa082db2f64aa3cc78aed59ea5822b","subject":"Merge \"Clean up Glance image after sanity check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74ea96bacfaa082db2f64aa3cc78aed59ea5822b"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-07-29 18:23:13.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-07-29 20:18:04.000000000","tz":120},"subject":"Enable CADF events in Keystone","message":"Enable CADF events in Keystone\n\nChange-Id: I0a40867ae65011fbfd53308b9d7f42d10435f5fa\nCloses-bug: #1607905\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ec057d5f0aa5e12d3c649a442e34c6127020abe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ec057d5f0aa5e12d3c649a442e34c6127020abe"}]},"branch":"refs/heads/master"},"b44e6f1338aec7b05d29d4666a1249460b1c67f9":{"kind":"REWORK","_number":3,"created":"2016-08-05 21:31:04.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/06/349006/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/349006/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/349006/3"}}},"commit":{"parents":[{"commit":"74ea96bacfaa082db2f64aa3cc78aed59ea5822b","subject":"Merge \"Clean up Glance image after sanity check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74ea96bacfaa082db2f64aa3cc78aed59ea5822b"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-07-29 18:23:13.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-08-05 21:30:41.000000000","tz":120},"subject":"Enable CADF events in Keystone and introduce kolla_enable_enhanced_security","message":"Enable CADF events in Keystone and introduce kolla_enable_enhanced_security\n\nChange-Id: I0a40867ae65011fbfd53308b9d7f42d10435f5fa\nCloses-bug: #1607905\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b44e6f1338aec7b05d29d4666a1249460b1c67f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b44e6f1338aec7b05d29d4666a1249460b1c67f9"}]},"branch":"refs/heads/master"},"233bb8e5cfee0ec6fde0e2a5fcc885ba5fc83391":{"kind":"REWORK","_number":4,"created":"2016-08-19 10:14:30.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/06/349006/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/349006/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/349006/4"}}},"commit":{"parents":[{"commit":"0b7e853784c2c0db115b0e72a3f7fc58c86c7660","subject":"Merge \"Fix the variable name mismatch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b7e853784c2c0db115b0e72a3f7fc58c86c7660"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-07-29 18:23:13.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-08-19 10:14:17.000000000","tz":120},"subject":"Enable CADF events in Keystone and introduce kolla_enable_cadf_notifications","message":"Enable CADF events in Keystone and introduce kolla_enable_cadf_notifications\n\nChange-Id: I0a40867ae65011fbfd53308b9d7f42d10435f5fa\nCloses-bug: #1607905\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/233bb8e5cfee0ec6fde0e2a5fcc885ba5fc83391"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/233bb8e5cfee0ec6fde0e2a5fcc885ba5fc83391"}]},"branch":"refs/heads/master"},"3feeb84616b6f8f55db6481ef78c8bd6e7643fa7":{"kind":"REWORK","_number":5,"created":"2016-08-19 11:01:57.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/06/349006/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/349006/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/349006/5"}}},"commit":{"parents":[{"commit":"0b7e853784c2c0db115b0e72a3f7fc58c86c7660","subject":"Merge \"Fix the variable name mismatch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b7e853784c2c0db115b0e72a3f7fc58c86c7660"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-07-29 18:23:13.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-08-19 11:01:47.000000000","tz":120},"subject":"Enable CADF events in Keystone and introduce kolla_enable_cadf_notifications","message":"Enable CADF events in Keystone and introduce kolla_enable_cadf_notifications\n\nChange-Id: I0a40867ae65011fbfd53308b9d7f42d10435f5fa\nCloses-bug: #1607905\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3feeb84616b6f8f55db6481ef78c8bd6e7643fa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3feeb84616b6f8f55db6481ef78c8bd6e7643fa7"}]},"branch":"refs/heads/master"},"10426776f71b3cc26887208012a2c3a013ad766e":{"kind":"REWORK","_number":6,"created":"2016-08-19 13:39:59.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/06/349006/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/349006/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/349006/6"}}},"commit":{"parents":[{"commit":"0b7e853784c2c0db115b0e72a3f7fc58c86c7660","subject":"Merge \"Fix the variable name mismatch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b7e853784c2c0db115b0e72a3f7fc58c86c7660"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-07-29 18:23:13.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-08-19 13:39:48.000000000","tz":120},"subject":"Enable CADF events in Keystone and introduce kolla_enable_cadf_notifications","message":"Enable CADF events in Keystone and introduce kolla_enable_cadf_notifications\n\nChange-Id: I0a40867ae65011fbfd53308b9d7f42d10435f5fa\nCloses-bug: #1607905\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10426776f71b3cc26887208012a2c3a013ad766e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10426776f71b3cc26887208012a2c3a013ad766e"}]},"branch":"refs/heads/master"},"dfaa2f2bc22a1f4407a8082a3cfdd3e2ce453bcb":{"kind":"REWORK","_number":7,"created":"2016-08-19 14:05:28.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/06/349006/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/349006/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/349006/7"}}},"commit":{"parents":[{"commit":"0b7e853784c2c0db115b0e72a3f7fc58c86c7660","subject":"Merge \"Fix the variable name mismatch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b7e853784c2c0db115b0e72a3f7fc58c86c7660"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-07-29 18:23:13.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-08-19 14:05:01.000000000","tz":120},"subject":"Enable CADF events in Keystone and introduce enable_cadf_notifications","message":"Enable CADF events in Keystone and introduce enable_cadf_notifications\n\nChange-Id: I0a40867ae65011fbfd53308b9d7f42d10435f5fa\nCloses-bug: #1607905\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dfaa2f2bc22a1f4407a8082a3cfdd3e2ce453bcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dfaa2f2bc22a1f4407a8082a3cfdd3e2ce453bcb"}]},"branch":"refs/heads/master"},"d39a4856a7be0c417d867e2f874b0cb54c7b7929":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2016-08-31 12:19:44.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/06/349006/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/349006/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/349006/8"}}},"commit":{"parents":[{"commit":"76f94f8d189d39746471a2ee599fc7e2a1344061","subject":"Merge \"Add a check before wiping Ceph disks in bootstrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76f94f8d189d39746471a2ee599fc7e2a1344061"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-07-29 18:23:13.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-08-31 12:19:44.000000000","tz":0},"subject":"Enable CADF events in Keystone and introduce enable_cadf_notifications","message":"Enable CADF events in Keystone and introduce enable_cadf_notifications\n\nChange-Id: I0a40867ae65011fbfd53308b9d7f42d10435f5fa\nCloses-bug: #1607905\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d39a4856a7be0c417d867e2f874b0cb54c7b7929"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d39a4856a7be0c417d867e2f874b0cb54c7b7929"}]},"branch":"refs/heads/master"},"e59ba6ea583b1c95ec0e7c8b7fe757493656c61c":{"kind":"REWORK","_number":9,"created":"2016-08-31 12:45:04.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/06/349006/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/349006/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/349006/9"}}},"commit":{"parents":[{"commit":"76f94f8d189d39746471a2ee599fc7e2a1344061","subject":"Merge \"Add a check before wiping Ceph disks in bootstrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76f94f8d189d39746471a2ee599fc7e2a1344061"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-07-29 18:23:13.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-08-31 12:44:53.000000000","tz":120},"subject":"Enable CADF events in Keystone and introduce enable_cadf_notifications","message":"Enable CADF events in Keystone and introduce enable_cadf_notifications\n\nChange-Id: I0a40867ae65011fbfd53308b9d7f42d10435f5fa\nCloses-bug: #1607905\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e59ba6ea583b1c95ec0e7c8b7fe757493656c61c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e59ba6ea583b1c95ec0e7c8b7fe757493656c61c"}]},"branch":"refs/heads/master"},"13fc814d0496dc700c8b6dda607f824f05147250":{"kind":"REWORK","_number":10,"created":"2016-08-31 18:24:10.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/06/349006/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/349006/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/349006/10"}}},"commit":{"parents":[{"commit":"b4c505a6c8da19be438cbf03e28a4d3f82099a11","subject":"Merge \"Allow operators to override use of get-pip.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4c505a6c8da19be438cbf03e28a4d3f82099a11"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-07-29 18:23:13.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-08-31 18:21:11.000000000","tz":120},"subject":"Enable CADF events in Keystone and introduce enable_cadf_notifications","message":"Enable CADF events in Keystone and introduce enable_cadf_notifications\n\nChange-Id: I0a40867ae65011fbfd53308b9d7f42d10435f5fa\nCloses-bug: #1607905\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13fc814d0496dc700c8b6dda607f824f05147250"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13fc814d0496dc700c8b6dda607f824f05147250"}]},"branch":"refs/heads/master"},"2ec092fabe95dc821afc4d09ba41f0adba1c8505":{"kind":"REWORK","_number":11,"created":"2016-09-06 06:58:15.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/06/349006/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/349006/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/349006/11"}}},"commit":{"parents":[{"commit":"b4c505a6c8da19be438cbf03e28a4d3f82099a11","subject":"Merge \"Allow operators to override use of get-pip.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4c505a6c8da19be438cbf03e28a4d3f82099a11"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-07-29 18:23:13.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-09-06 06:58:03.000000000","tz":120},"subject":"Enable CADF events in Keystone and introduce enable_cadf_notifications","message":"Enable CADF events in Keystone and introduce enable_cadf_notifications\n\nChange-Id: I0a40867ae65011fbfd53308b9d7f42d10435f5fa\nCloses-bug: #1607905\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ec092fabe95dc821afc4d09ba41f0adba1c8505"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ec092fabe95dc821afc4d09ba41f0adba1c8505"}]},"branch":"refs/heads/master"},"61ac8ef5ebb46ba25822402376df4c87d8cf0204":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2016-09-12 07:41:30.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/06/349006/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/349006/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/349006/12"}}},"commit":{"parents":[{"commit":"8c0f8daa22e3163041fc2bb5cd1ba5939247e600","subject":"Merge \"Add neutron-lbaas-agent to exlude image for Ubuntu binary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c0f8daa22e3163041fc2bb5cd1ba5939247e600"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-07-29 18:23:13.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-09-12 07:41:30.000000000","tz":0},"subject":"Enable CADF events in Keystone and introduce enable_cadf_notifications","message":"Enable CADF events in Keystone and introduce enable_cadf_notifications\n\nChange-Id: I0a40867ae65011fbfd53308b9d7f42d10435f5fa\nCloses-bug: #1607905\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61ac8ef5ebb46ba25822402376df4c87d8cf0204"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61ac8ef5ebb46ba25822402376df4c87d8cf0204"}]},"branch":"refs/heads/master"},"62bc7a6bad4ea9df1faf31e38e000b6492d6e4ac":{"kind":"REWORK","_number":13,"created":"2016-09-13 05:37:59.000000000","uploader":{"_account_id":2834,"name":"Steven Dake","email":"steven.dake@gmail.com","username":"sdake"},"ref":"refs/changes/06/349006/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/349006/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/349006/13"}}},"commit":{"parents":[{"commit":"6b3456bf36ad26fc5df76008196a410acb325b21","subject":"Merge \"Add Barbican ansible role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b3456bf36ad26fc5df76008196a410acb325b21"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-07-29 18:23:13.000000000","tz":120},"committer":{"name":"Steven Dake","email":"stdake@cisco.com","date":"2016-09-13 05:37:58.000000000","tz":-420},"subject":"Enable CADF events in Keystone and introduce enable_cadf_notifications","message":"Enable CADF events in Keystone and introduce enable_cadf_notifications\n\nChange-Id: I0a40867ae65011fbfd53308b9d7f42d10435f5fa\nCloses-bug: #1607905\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62bc7a6bad4ea9df1faf31e38e000b6492d6e4ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62bc7a6bad4ea9df1faf31e38e000b6492d6e4ac"}]},"branch":"refs/heads/master"},"8f6bd4e73539e89ef556a1a8f1f33936ec24f3a4":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2016-09-16 12:19:53.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/06/349006/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/349006/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/349006/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/349006/14"}}},"commit":{"parents":[{"commit":"112974b0527aff215d95813f777b554bd95807c7","subject":"Merge \"Update \"which\" command  to \"type\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/112974b0527aff215d95813f777b554bd95807c7"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-07-29 18:23:13.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2016-09-16 12:19:53.000000000","tz":0},"subject":"Enable CADF events in Keystone and introduce enable_cadf_notifications","message":"Enable CADF events in Keystone and introduce enable_cadf_notifications\n\nChange-Id: I0a40867ae65011fbfd53308b9d7f42d10435f5fa\nCloses-bug: #1607905\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bd4e73539e89ef556a1a8f1f33936ec24f3a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bd4e73539e89ef556a1a8f1f33936ec24f3a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
