)]}'
{"id":"openstack%2Fkolla~361891","triplet_id":"openstack%2Fkolla~master~I46a48e242dae6ee6c9159bc30edd8b823ffbb515","project":"openstack/kolla","branch":"master","topic":"bug/1617902","hashtags":[],"change_id":"I46a48e242dae6ee6c9159bc30edd8b823ffbb515","subject":"Fix nova_ssh container shell to allow instance resizes.","status":"MERGED","created":"2016-08-29 07:03:19.000000000","updated":"2016-08-31 12:49:17.000000000","submitted":"2016-08-31 12:49:16.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"74b2bc0b62715eac9a0ee7a296546cf77953ef65","_number":361891,"virtual_id_number":361891,"owner":{"_account_id":22827,"name":"Matthew Taylor","email":"matt@mty.net.au","username":"mtaylor22"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},{"value":0,"_account_id":2834,"name":"Steven Dake","email":"steven.dake@gmail.com","username":"sdake"},{"value":2,"date":"2016-08-31 12:49:16.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10787,"name":"Michal Jastrzebski (inc0)","email":"michal.jastrzebski@intel.com","username":"inc0"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":9927,"name":"Britt Houser","email":"bhouser@cisco.com","username":"britthouser"},{"value":0,"_account_id":14103,"name":"wangtianfa","email":"wang.tianfa@99cloud.net","username":"wangtianfa"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":16233,"name":"Mauricio Lima","email":"mauriciolimab@gmail.com","username":"limamauricio"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2834,"name":"Steven Dake","email":"steven.dake@gmail.com","username":"sdake"},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},{"value":2,"date":"2016-08-31 05:41:36.000000000","_account_id":2834,"name":"Steven Dake","email":"steven.dake@gmail.com","username":"sdake"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10787,"name":"Michal Jastrzebski (inc0)","email":"michal.jastrzebski@intel.com","username":"inc0"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":1,"date":"2016-08-31 05:00:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9927,"name":"Britt Houser","email":"bhouser@cisco.com","username":"britthouser"},{"value":1,"date":"2016-08-31 11:19:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14103,"name":"wangtianfa","email":"wang.tianfa@99cloud.net","username":"wangtianfa"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":2,"date":"2016-08-31 12:44:37.000000000","_account_id":16233,"name":"Mauricio Lima","email":"mauriciolimab@gmail.com","username":"limamauricio"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16233,"name":"Mauricio Lima","email":"mauriciolimab@gmail.com","username":"limamauricio"},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},{"value":0,"_account_id":2834,"name":"Steven Dake","email":"steven.dake@gmail.com","username":"sdake"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10787,"name":"Michal Jastrzebski (inc0)","email":"michal.jastrzebski@intel.com","username":"inc0"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":9927,"name":"Britt Houser","email":"bhouser@cisco.com","username":"britthouser"},{"value":0,"_account_id":14103,"name":"wangtianfa","email":"wang.tianfa@99cloud.net","username":"wangtianfa"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":1,"date":"2016-08-31 12:44:37.000000000","_account_id":16233,"name":"Mauricio Lima","email":"mauriciolimab@gmail.com","username":"limamauricio"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},{"value":0,"_account_id":2834,"name":"Steven Dake","email":"steven.dake@gmail.com","username":"sdake"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10787,"name":"Michal Jastrzebski (inc0)","email":"michal.jastrzebski@intel.com","username":"inc0"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":9927,"name":"Britt Houser","email":"bhouser@cisco.com","username":"britthouser"},{"value":0,"_account_id":14103,"name":"wangtianfa","email":"wang.tianfa@99cloud.net","username":"wangtianfa"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":16233,"name":"Mauricio Lima","email":"mauriciolimab@gmail.com","username":"limamauricio"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},{"value":0,"_account_id":2834,"name":"Steven Dake","email":"steven.dake@gmail.com","username":"sdake"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10787,"name":"Michal Jastrzebski (inc0)","email":"michal.jastrzebski@intel.com","username":"inc0"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":9927,"name":"Britt Houser","email":"bhouser@cisco.com","username":"britthouser"},{"value":0,"_account_id":14103,"name":"wangtianfa","email":"wang.tianfa@99cloud.net","username":"wangtianfa"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":16233,"name":"Mauricio Lima","email":"mauriciolimab@gmail.com","username":"limamauricio"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":2834,"name":"Steven Dake","email":"steven.dake@gmail.com","username":"sdake"},{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"_account_id":9927,"name":"Britt Houser","email":"bhouser@cisco.com","username":"britthouser"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":10787,"name":"Michal Jastrzebski (inc0)","email":"michal.jastrzebski@intel.com","username":"inc0"},{"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},{"_account_id":14103,"name":"wangtianfa","email":"wang.tianfa@99cloud.net","username":"wangtianfa"},{"_account_id":16233,"name":"Mauricio Lima","email":"mauriciolimab@gmail.com","username":"limamauricio"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-08-29 07:03:30.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2016-08-29 08:11:14.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2016-08-29 19:27:45.000000000","updated_by":{"_account_id":10787,"name":"Michal Jastrzebski (inc0)","email":"michal.jastrzebski@intel.com","username":"inc0"},"reviewer":{"_account_id":10787,"name":"Michal Jastrzebski (inc0)","email":"michal.jastrzebski@intel.com","username":"inc0"},"state":"REVIEWER"},{"updated":"2016-08-30 08:05:07.000000000","updated_by":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"reviewer":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"state":"REVIEWER"},{"updated":"2016-08-31 04:43:51.000000000","updated_by":{"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},"reviewer":{"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},"state":"REVIEWER"},{"updated":"2016-08-31 05:00:27.000000000","updated_by":{"_account_id":9927,"name":"Britt Houser","email":"bhouser@cisco.com","username":"britthouser"},"reviewer":{"_account_id":9927,"name":"Britt Houser","email":"bhouser@cisco.com","username":"britthouser"},"state":"REVIEWER"},{"updated":"2016-08-31 05:41:36.000000000","updated_by":{"_account_id":2834,"name":"Steven Dake","email":"steven.dake@gmail.com","username":"sdake"},"reviewer":{"_account_id":2834,"name":"Steven Dake","email":"steven.dake@gmail.com","username":"sdake"},"state":"REVIEWER"},{"updated":"2016-08-31 11:19:07.000000000","updated_by":{"_account_id":14103,"name":"wangtianfa","email":"wang.tianfa@99cloud.net","username":"wangtianfa"},"reviewer":{"_account_id":14103,"name":"wangtianfa","email":"wang.tianfa@99cloud.net","username":"wangtianfa"},"state":"REVIEWER"},{"updated":"2016-08-31 12:44:37.000000000","updated_by":{"_account_id":16233,"name":"Mauricio Lima","email":"mauriciolimab@gmail.com","username":"limamauricio"},"reviewer":{"_account_id":16233,"name":"Mauricio Lima","email":"mauriciolimab@gmail.com","username":"limamauricio"},"state":"REVIEWER"},{"updated":"2016-08-31 12:49:16.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"1e2b032d2a417be9a8d586dc4139b01ef639118c","author":{"_account_id":22827,"name":"Matthew Taylor","email":"matt@mty.net.au","username":"mtaylor22"},"date":"2016-08-29 07:03:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88d7485ca6cae66ab5ce1add4de1fb6c5b0474e3","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2016-08-29 07:03:30.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\n    Your patch will now be tested automatically by OpenStack testing frameworks\n    and once the automatic tests pass, it will be reviewed by other friendly\n    developers. They will give you feedback and may require you to refine it.\n\n    People seldom get their patch approved on the first try, so don\u0027t be\n    concerned if requested to make corrections. Feel free to modify your patch\n    and resubmit a new change-set.\n\n    Patches usually take 3 to 7 days to be reviewed so be patient and be\n    available on IRC to ask and answer questions about your work. Also it\n    takes generally at least a couple of weeks for cores to get around to\n    reviewing code. The more you participate in the community the more\n    rewarding it is for you. You may also notice that the more you get to know\n    people and get to be known, the faster your patches will be reviewed and\n    eventually approved. Get to know others and become known by doing code\n    reviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\n    Thanks again for supporting OpenStack, we look forward to working with you.\n\n    IRC: https://wiki.openstack.org/wiki/IRC\n    Workflow: http://docs.openstack.org/infra/manual/developers.html\n    Commit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"7799396922936f132cefe30f498480e19a458c45","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-29 08:00:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-docs-ubuntu-xenial http://docs-draft.openstack.org/91/361891/1/check/gate-kolla-docs-ubuntu-xenial/2ae1285//doc/build/html/ : SUCCESS in 2m 08s\n- gate-kolla-pep8-ubuntu-xenial http://logs.openstack.org/91/361891/1/check/gate-kolla-pep8-ubuntu-xenial/ea708ce/ : SUCCESS in 2m 06s\n- gate-kolla-python27-ubuntu-xenial http://logs.openstack.org/91/361891/1/check/gate-kolla-python27-ubuntu-xenial/e6f1a8e/ : SUCCESS in 4m 17s\n- gate-kolla-python34 http://logs.openstack.org/91/361891/1/check/gate-kolla-python34/7ced9de/ : SUCCESS in 4m 35s\n- gate-kolla-python35-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-python35-nv/bfa95b1/ : SUCCESS in 2m 50s (non-voting)\n- gate-kolla-tox-bandit-ubuntu-xenial http://logs.openstack.org/91/361891/1/check/gate-kolla-tox-bandit-ubuntu-xenial/3326cfe/ : SUCCESS in 2m 13s (non-voting)\n- gate-kolla-dsvm-build-centos-binary-centos-7-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-build-centos-binary-centos-7-nv/6dac6ba/ : FAILURE in 22m 35s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary-centos-7-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-deploy-centos-binary-centos-7-nv/59205cd/ : SUCCESS in 31m 31s (non-voting)\n- gate-kolla-dsvm-build-centos-source-centos-7-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-build-centos-source-centos-7-nv/4bb9a80/ : FAILURE in 22m 23s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source-centos-7-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-deploy-centos-source-centos-7-nv/b482b2e/ : SUCCESS in 32m 14s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-binary-ubuntu-xenial-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-build-ubuntu-binary-ubuntu-xenial-nv/6a6d2e5/ : FAILURE in 16m 22s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source-ubuntu-xenial-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-build-ubuntu-source-ubuntu-xenial-nv/8fc9984/ : FAILURE in 26m 44s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-binary-ubuntu-xenial-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-deploy-ubuntu-binary-ubuntu-xenial-nv/b84d804/ : SUCCESS in 35m 24s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source-ubuntu-xenial-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-deploy-ubuntu-source-ubuntu-xenial-nv/ca86152/ : SUCCESS in 39m 38s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-binary-centos-7-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-build-oraclelinux-binary-centos-7-nv/90419b4/ : FAILURE in 34m 32s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-binary-centos-7-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-deploy-oraclelinux-binary-centos-7-nv/791f303/ : SUCCESS in 53m 46s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-source-centos-7-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-build-oraclelinux-source-centos-7-nv/ed2f974/ : FAILURE in 41m 32s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-source-centos-7-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-deploy-oraclelinux-source-centos-7-nv/fffd2e4/ : SUCCESS in 50m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1375d97a4679058c2dc661e72707b41a1b048c5c","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2016-08-29 08:11:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b25419594e219336e57c9ba6b39d1d207588c63f","author":{"_account_id":10787,"name":"Michal Jastrzebski (inc0)","email":"michal.jastrzebski@intel.com","username":"inc0"},"date":"2016-08-29 19:27:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e826fce43deaf30423612599ae09cb0973a14221","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2016-08-30 08:05:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c596837613aa9b37a467c68681c3f260bb352235","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2016-08-30 08:08:22.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\na nit.\nmore over. please rebase the code and the gate will be green","accounts_in_message":[],"_revision_number":1},{"id":"b0aa520cd27bcb00ba786873838d45e29cb87ebd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-30 09:12:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-docs-ubuntu-xenial http://docs-draft.openstack.org/91/361891/1/check/gate-kolla-docs-ubuntu-xenial/fd2141b//doc/build/html/ : SUCCESS in 2m 27s\n- gate-kolla-pep8-ubuntu-xenial http://logs.openstack.org/91/361891/1/check/gate-kolla-pep8-ubuntu-xenial/93688e8/ : SUCCESS in 2m 10s\n- gate-kolla-python27-ubuntu-xenial http://logs.openstack.org/91/361891/1/check/gate-kolla-python27-ubuntu-xenial/129209e/ : SUCCESS in 2m 16s\n- gate-kolla-python34 http://logs.openstack.org/91/361891/1/check/gate-kolla-python34/3855ded/ : SUCCESS in 5m 47s\n- gate-kolla-python35-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-python35-nv/058fc70/ : SUCCESS in 2m 36s (non-voting)\n- gate-kolla-tox-bandit-ubuntu-xenial http://logs.openstack.org/91/361891/1/check/gate-kolla-tox-bandit-ubuntu-xenial/8a1ba27/ : SUCCESS in 2m 25s (non-voting)\n- gate-kolla-dsvm-build-centos-binary-centos-7-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-build-centos-binary-centos-7-nv/4284407/ : FAILURE in 1h 05m 14s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary-centos-7-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-deploy-centos-binary-centos-7-nv/5634983/ : SUCCESS in 33m 00s (non-voting)\n- gate-kolla-dsvm-build-centos-source-centos-7-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-build-centos-source-centos-7-nv/e0eb54d/ : SUCCESS in 23m 50s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source-centos-7-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-deploy-centos-source-centos-7-nv/fa7f847/ : SUCCESS in 31m 36s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-binary-ubuntu-xenial-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-build-ubuntu-binary-ubuntu-xenial-nv/19096eb/ : SUCCESS in 16m 26s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source-ubuntu-xenial-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-build-ubuntu-source-ubuntu-xenial-nv/30a16cc/ : SUCCESS in 16m 28s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-binary-ubuntu-xenial-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-deploy-ubuntu-binary-ubuntu-xenial-nv/41ed15d/ : SUCCESS in 25m 36s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source-ubuntu-xenial-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-deploy-ubuntu-source-ubuntu-xenial-nv/032b1b1/ : SUCCESS in 45m 49s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-binary-centos-7-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-build-oraclelinux-binary-centos-7-nv/aa29a05/ : SUCCESS in 35m 22s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-binary-centos-7-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-deploy-oraclelinux-binary-centos-7-nv/f00f8c3/ : SUCCESS in 54m 01s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-source-centos-7-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-build-oraclelinux-source-centos-7-nv/249e225/ : FAILURE in 5m 39s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-source-centos-7-nv http://logs.openstack.org/91/361891/1/check/gate-kolla-dsvm-deploy-oraclelinux-source-centos-7-nv/d0cb14d/ : SUCCESS in 52m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e96b1464cf07e8889f31d3cbd69ce55522f5f86","author":{"_account_id":22827,"name":"Matthew Taylor","email":"matt@mty.net.au","username":"mtaylor22"},"date":"2016-08-31 04:38:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"242f635d05e2bad761eeab05f1c8ac3f12ed2e0e","author":{"_account_id":11105,"name":"Swapnil Kulkarni","email":"coolsvap@gmail.com","username":"coolsvap"},"date":"2016-08-31 04:43:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"80d0b763174ba5ddbc6611c328c77dd615b3e461","author":{"_account_id":2834,"name":"Steven Dake","email":"steven.dake@gmail.com","username":"sdake"},"date":"2016-08-31 04:45:55.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nMatthew,\n\nThank you for th econtribution.  I think their is a tidier solution though - by modifying the dockerfile and adding  a RUN operation for the chsh.","accounts_in_message":[],"_revision_number":2},{"id":"f155d13cededc8c64aca5184e47475f6f2aa269e","author":{"_account_id":22827,"name":"Matthew Taylor","email":"matt@mty.net.au","username":"mtaylor22"},"date":"2016-08-31 04:56:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f472c4f327322d20c87e2a043fbdc34d8ac536fc","author":{"_account_id":9927,"name":"Britt Houser","email":"bhouser@cisco.com","username":"britthouser"},"date":"2016-08-31 05:00:27.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"abbf78ac00ec49a4c704d837974a65aec7dd0d16","author":{"_account_id":22827,"name":"Matthew Taylor","email":"matt@mty.net.au","username":"mtaylor22"},"date":"2016-08-31 05:12:47.000000000","message":"Patch Set 3:\n\nThanks Steven,\n\nI figured it\u0027s best to have it localised to nova-ssh container only, rather than the base Dockerfile as it\u0027s not required anywhere else (except nova-ssh).","accounts_in_message":[],"_revision_number":3},{"id":"ea27fe95746d04c45f740a575915a5658b3cb595","author":{"_account_id":2834,"name":"Steven Dake","email":"steven.dake@gmail.com","username":"sdake"},"date":"2016-08-31 05:41:36.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6aaed9d360c4750eb884e37deb0cffbe3018a461","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-31 07:54:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-kolla-docs-ubuntu-xenial http://docs-draft.openstack.org/91/361891/3/check/gate-kolla-docs-ubuntu-xenial/2ed9dfe//doc/build/html/ : SUCCESS in 2m 10s\n- gate-kolla-pep8-ubuntu-xenial http://logs.openstack.org/91/361891/3/check/gate-kolla-pep8-ubuntu-xenial/06eaa43/ : SUCCESS in 2m 19s\n- gate-kolla-python27-ubuntu-xenial http://logs.openstack.org/91/361891/3/check/gate-kolla-python27-ubuntu-xenial/d849b39/ : SUCCESS in 2m 14s\n- gate-kolla-python34 http://logs.openstack.org/91/361891/3/check/gate-kolla-python34/12a8ec1/ : SUCCESS in 2m 12s\n- gate-kolla-python35-nv http://logs.openstack.org/91/361891/3/check/gate-kolla-python35-nv/b1a6027/ : SUCCESS in 5m 11s (non-voting)\n- gate-kolla-tox-bandit-ubuntu-xenial http://logs.openstack.org/91/361891/3/check/gate-kolla-tox-bandit-ubuntu-xenial/3c0dee0/ : SUCCESS in 4m 32s (non-voting)\n- gate-kolla-dsvm-build-centos-binary-centos-7-nv http://logs.openstack.org/91/361891/3/check/gate-kolla-dsvm-build-centos-binary-centos-7-nv/c05f2c4/ : SUCCESS in 23m 36s (non-voting)\n- gate-kolla-dsvm-deploy-centos-binary-centos-7-nv http://logs.openstack.org/91/361891/3/check/gate-kolla-dsvm-deploy-centos-binary-centos-7-nv/601b373/ : SUCCESS in 32m 39s (non-voting)\n- gate-kolla-dsvm-build-centos-source-centos-7-nv http://logs.openstack.org/91/361891/3/check/gate-kolla-dsvm-build-centos-source-centos-7-nv/250c866/ : SUCCESS in 40m 51s (non-voting)\n- gate-kolla-dsvm-deploy-centos-source-centos-7-nv http://logs.openstack.org/91/361891/3/check/gate-kolla-dsvm-deploy-centos-source-centos-7-nv/6c61e78/ : SUCCESS in 44m 17s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-binary-ubuntu-xenial-nv http://logs.openstack.org/91/361891/3/check/gate-kolla-dsvm-build-ubuntu-binary-ubuntu-xenial-nv/ac73602/ : SUCCESS in 17m 09s (non-voting)\n- gate-kolla-dsvm-build-ubuntu-source-ubuntu-xenial-nv http://logs.openstack.org/91/361891/3/check/gate-kolla-dsvm-build-ubuntu-source-ubuntu-xenial-nv/eea4a8c/ : SUCCESS in 17m 33s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-binary-ubuntu-xenial-nv http://logs.openstack.org/91/361891/3/check/gate-kolla-dsvm-deploy-ubuntu-binary-ubuntu-xenial-nv/52dd017/ : SUCCESS in 37m 05s (non-voting)\n- gate-kolla-dsvm-deploy-ubuntu-source-ubuntu-xenial-nv http://logs.openstack.org/91/361891/3/check/gate-kolla-dsvm-deploy-ubuntu-source-ubuntu-xenial-nv/8fc33e8/ : SUCCESS in 39m 22s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-binary-centos-7-nv http://logs.openstack.org/91/361891/3/check/gate-kolla-dsvm-build-oraclelinux-binary-centos-7-nv/3cc1ff1/ : SUCCESS in 43m 27s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-binary-centos-7-nv http://logs.openstack.org/91/361891/3/check/gate-kolla-dsvm-deploy-oraclelinux-binary-centos-7-nv/f94d907/ : SUCCESS in 55m 20s (non-voting)\n- gate-kolla-dsvm-build-oraclelinux-source-centos-7-nv http://logs.openstack.org/91/361891/3/check/gate-kolla-dsvm-build-oraclelinux-source-centos-7-nv/31d0894/ : SUCCESS in 29m 38s (non-voting)\n- gate-kolla-dsvm-deploy-oraclelinux-source-centos-7-nv http://logs.openstack.org/91/361891/3/check/gate-kolla-dsvm-deploy-oraclelinux-source-centos-7-nv/4cb3af3/ : SUCCESS in 36m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d03401f7c8fdefea79a9e3492f7ed2090bac103b","author":{"_account_id":14103,"name":"wangtianfa","email":"wang.tianfa@99cloud.net","username":"wangtianfa"},"date":"2016-08-31 11:19:07.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"4ce5a30b6d890c7d3dc558cb96c292bf778d3839","author":{"_account_id":16233,"name":"Mauricio Lima","email":"mauriciolimab@gmail.com","username":"limamauricio"},"date":"2016-08-31 12:44:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"47c417fd1b41e88b5fc336bf8c57a8d06cd3ac00","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-31 12:44:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"1d9053157a770f11647ac1801c276b581f6e24df","date":"2016-08-31 12:49:16.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":3},{"id":"61748fc0ed39dce6162f2b9240c1a7df7d82971e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-31 12:49:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-kolla-docs-ubuntu-xenial http://docs-draft.openstack.org/91/361891/3/gate/gate-kolla-docs-ubuntu-xenial/f7e9c02//doc/build/html/ : SUCCESS in 3m 09s\n- gate-kolla-pep8-ubuntu-xenial http://logs.openstack.org/91/361891/3/gate/gate-kolla-pep8-ubuntu-xenial/df44b59/ : SUCCESS in 3m 57s\n- gate-kolla-python27-ubuntu-xenial http://logs.openstack.org/91/361891/3/gate/gate-kolla-python27-ubuntu-xenial/2972f5f/ : SUCCESS in 2m 51s\n- gate-kolla-python34 http://logs.openstack.org/91/361891/3/gate/gate-kolla-python34/c6cc786/ : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c4be56c1e54adb9f1cc6346f1d8849657b0a44ae","revisions":{"37cca1527f61b86a07958f6ea96bb1c9f7843f38":{"kind":"REWORK","_number":1,"created":"2016-08-29 07:03:19.000000000","uploader":{"_account_id":22827,"name":"Matthew Taylor","email":"matt@mty.net.au","username":"mtaylor22"},"ref":"refs/changes/91/361891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/361891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/361891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/361891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/361891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/361891/1"}}},"commit":{"parents":[{"commit":"57d3eaf70c315bd665a4533f766dbe4a2e6077f9","subject":"Merge \"Customizations for Nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57d3eaf70c315bd665a4533f766dbe4a2e6077f9"}]}],"author":{"name":"Matthew Taylor","email":"matthew.taylor@digitalpacific.com.au","date":"2016-08-29 06:53:27.000000000","tz":600},"committer":{"name":"Matthew Taylor","email":"matt@mty.net.au","date":"2016-08-29 07:03:16.000000000","tz":600},"subject":"Fix nova_ssh container shell to allow instance resizes.","message":"Fix nova_ssh container shell to allow instance resizes.\n\nAccording to the configuration reference,\nhttp://docs.openstack.org/mitaka/config-reference/compute/resize.html\nWhen using KVM, nova utilizes SSH for instance resizing.\nKolla nova_ssh container has the nova user configured with \"/sbin/nologin\" shell,\nwhich prevents nova_compute from successfully resizing instances.\nAdded fix into extend_start.sh docker file to change shell from nologin to bash.\n\nChange-Id: I46a48e242dae6ee6c9159bc30edd8b823ffbb515\nCloses-Bug: 1617902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/37cca1527f61b86a07958f6ea96bb1c9f7843f38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/37cca1527f61b86a07958f6ea96bb1c9f7843f38"}]},"branch":"refs/heads/master"},"11293cba9b641435994749e343397f2f1fd79670":{"kind":"REWORK","_number":2,"created":"2016-08-31 04:38:53.000000000","uploader":{"_account_id":22827,"name":"Matthew Taylor","email":"matt@mty.net.au","username":"mtaylor22"},"ref":"refs/changes/91/361891/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/361891/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/361891/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/361891/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/361891/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/361891/2"}}},"commit":{"parents":[{"commit":"57d3eaf70c315bd665a4533f766dbe4a2e6077f9","subject":"Merge \"Customizations for Nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57d3eaf70c315bd665a4533f766dbe4a2e6077f9"}]}],"author":{"name":"Matthew Taylor","email":"matthew.taylor@digitalpacific.com.au","date":"2016-08-29 06:53:27.000000000","tz":600},"committer":{"name":"Matthew Taylor","email":"matt@mty.net.au","date":"2016-08-31 04:38:30.000000000","tz":600},"subject":"Fix nova_ssh container shell to allow instance resizes.","message":"Fix nova_ssh container shell to allow instance resizes.\n\nAccording to the configuration reference,\nhttp://docs.openstack.org/mitaka/config-reference/compute/resize.html\nWhen using KVM, nova utilizes SSH for instance resizing.\nKolla nova_ssh container has the nova user configured with \"/sbin/nologin\" shell,\nwhich prevents nova_compute from successfully resizing instances.\nAdded fix into extend_start.sh docker file to change shell from nologin to bash.\n\nChange-Id: I46a48e242dae6ee6c9159bc30edd8b823ffbb515\nCloses-Bug: 1617902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11293cba9b641435994749e343397f2f1fd79670"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11293cba9b641435994749e343397f2f1fd79670"}]},"branch":"refs/heads/master"},"c4be56c1e54adb9f1cc6346f1d8849657b0a44ae":{"kind":"REWORK","_number":3,"created":"2016-08-31 04:56:52.000000000","uploader":{"_account_id":22827,"name":"Matthew Taylor","email":"matt@mty.net.au","username":"mtaylor22"},"ref":"refs/changes/91/361891/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/361891/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/361891/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/361891/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/361891/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/361891/3"}}},"commit":{"parents":[{"commit":"57d3eaf70c315bd665a4533f766dbe4a2e6077f9","subject":"Merge \"Customizations for Nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57d3eaf70c315bd665a4533f766dbe4a2e6077f9"}]}],"author":{"name":"Matthew Taylor","email":"matthew.taylor@digitalpacific.com.au","date":"2016-08-29 06:53:27.000000000","tz":600},"committer":{"name":"Matthew Taylor","email":"matt@mty.net.au","date":"2016-08-31 04:55:25.000000000","tz":600},"subject":"Fix nova_ssh container shell to allow instance resizes.","message":"Fix nova_ssh container shell to allow instance resizes.\n\nAccording to the configuration reference,\nhttp://docs.openstack.org/mitaka/config-reference/compute/resize.html\nWhen using KVM, nova utilizes SSH for instance resizing.\nKolla nova_ssh container has the nova user configured with \"/sbin/nologin\" shell,\nwhich prevents nova_compute from successfully resizing instances.\nAdded fix to change shell from nologin to bash.\n\nChange-Id: I46a48e242dae6ee6c9159bc30edd8b823ffbb515\nCloses-Bug: 1617902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4be56c1e54adb9f1cc6346f1d8849657b0a44ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4be56c1e54adb9f1cc6346f1d8849657b0a44ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
