)]}'
{"id":"openstack%2Fkolla~518270","triplet_id":"openstack%2Fkolla~master~I7b539490b07ff5729b0ab43f992c81770a6d026d","project":"openstack/kolla","branch":"master","topic":"bp/onos-images","hashtags":[],"change_id":"I7b539490b07ff5729b0ab43f992c81770a6d026d","subject":"wip Implement onos Docker images","status":"ABANDONED","created":"2017-11-07 07:43:22.000000000","updated":"2019-11-21 12:09:18.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"30328ec281d771f9090b434b8b69a0988be8455d","_number":518270,"virtual_id_number":518270,"owner":{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":-1,"date":"2017-11-09 10:51:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"all":[{"value":-1,"date":"2017-11-12 04:39:56.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-11-07 10:36:00.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2017-11-08 19:45:36.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2017-11-09 10:51:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-11-12 04:39:56.000000000","updated_by":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"reviewer":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"state":"REVIEWER"}],"messages":[{"id":"7be54638af23c54531d7b94589804c39dddcde9b","author":{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},"date":"2017-11-07 07:43:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c293f049b764d16480a9cf2e7e36fe98907a0b88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-07 09:22:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/70/518270/1/check/openstack-tox-pep8/918bbe0/ : SUCCESS in 3m 47s\n- openstack-tox-py27 http://logs.openstack.org/70/518270/1/check/openstack-tox-py27/fe9ec00/ : SUCCESS in 4m 01s\n- openstack-tox-py35 http://logs.openstack.org/70/518270/1/check/openstack-tox-py35/1e9a87e/ : SUCCESS in 2m 34s\n- build-openstack-sphinx-docs http://logs.openstack.org/70/518270/1/check/build-openstack-sphinx-docs/a403e9c/html/ : SUCCESS in 2m 35s\n- kolla-build-centos-binary http://logs.openstack.org/70/518270/1/check/kolla-build-centos-binary/99ec809/ : FAILURE in 33m 30s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/70/518270/1/check/kolla-build-centos-source/d6b089d/ : FAILURE in 46m 50s\n- kolla-build-ubuntu-binary http://logs.openstack.org/70/518270/1/check/kolla-build-ubuntu-binary/50b7fa3/ : SUCCESS in 21m 39s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/70/518270/1/check/kolla-build-ubuntu-source/adcf40f/ : SUCCESS in 19m 11s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/70/518270/1/check/kolla-build-oraclelinux-binary/60d8bd9/ : FAILURE in 54m 56s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/70/518270/1/check/kolla-build-oraclelinux-source/a427f99/ : FAILURE in 1h 02m 42s\n- kolla-ansible-centos-source http://logs.openstack.org/70/518270/1/check/kolla-ansible-centos-source/1e732c2/ : SUCCESS in 1h 04m 50s (non-voting)\n- kolla-ansible-centos-binary http://logs.openstack.org/70/518270/1/check/kolla-ansible-centos-binary/2602e20/ : SUCCESS in 48m 54s (non-voting)\n- kolla-ansible-ubuntu-source http://logs.openstack.org/70/518270/1/check/kolla-ansible-ubuntu-source/815bb78/ : FAILURE in 41m 19s (non-voting)\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/70/518270/1/check/kolla-ansible-ubuntu-binary/7793aae/ : FAILURE in 39m 16s (non-voting)\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/70/518270/1/check/kolla-ansible-oraclelinux-source/da8ee16/ : SUCCESS in 1h 00m 07s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/70/518270/1/check/kolla-ansible-oraclelinux-binary/2d6feed/ : SUCCESS in 1h 38m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0dfcdfc0c3f9f4d974824176a5597709a6c62bec","author":{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},"date":"2017-11-07 09:42:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fc250f94096880d97b883f23688e3df02753a809","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2017-11-07 10:36:00.000000000","message":"Patch Set 2: Code-Review+1\n\nmark","accounts_in_message":[],"_revision_number":2},{"id":"9ed71de48d163e2451c18fcf6b04c38644912de6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-07 11:16:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/70/518270/2/check/openstack-tox-pep8/df182ae/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/70/518270/2/check/openstack-tox-py27/abed323/ : SUCCESS in 2m 35s\n- openstack-tox-py35 http://logs.openstack.org/70/518270/2/check/openstack-tox-py35/ed2dc20/ : SUCCESS in 4m 58s\n- build-openstack-sphinx-docs http://logs.openstack.org/70/518270/2/check/build-openstack-sphinx-docs/128169c/html/ : SUCCESS in 2m 14s\n- kolla-build-centos-binary http://logs.openstack.org/70/518270/2/check/kolla-build-centos-binary/814b636/ : FAILURE in 37m 19s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/70/518270/2/check/kolla-build-centos-source/49b6942/ : TIMED_OUT in 1h 33m 15s\n- kolla-build-ubuntu-binary http://logs.openstack.org/70/518270/2/check/kolla-build-ubuntu-binary/47aa3e5/ : SUCCESS in 25m 37s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/70/518270/2/check/kolla-build-ubuntu-source/f1a5e4c/ : SUCCESS in 21m 46s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/70/518270/2/check/kolla-build-oraclelinux-binary/611e5c9/ : FAILURE in 56m 36s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/70/518270/2/check/kolla-build-oraclelinux-source/9051597/ : FAILURE in 1h 01m 10s\n- kolla-ansible-centos-source http://logs.openstack.org/70/518270/2/check/kolla-ansible-centos-source/707bca5/ : SUCCESS in 52m 58s (non-voting)\n- kolla-ansible-centos-binary http://logs.openstack.org/70/518270/2/check/kolla-ansible-centos-binary/6bc107d/ : SUCCESS in 1h 23m 36s (non-voting)\n- kolla-ansible-ubuntu-source http://logs.openstack.org/70/518270/2/check/kolla-ansible-ubuntu-source/a785747/ : FAILURE in 39m 47s (non-voting)\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/70/518270/2/check/kolla-ansible-ubuntu-binary/55f2419/ : FAILURE in 39m 49s (non-voting)\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/70/518270/2/check/kolla-ansible-oraclelinux-source/6e3321c/ : FAILURE in 3m 50s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/70/518270/2/check/kolla-ansible-oraclelinux-binary/64b358b/ : SUCCESS in 59m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7956d50c9e83d138ba1ed699fe0b1150ad7af6cd","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2017-11-08 19:45:36.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nmissing release note","accounts_in_message":[],"_revision_number":2},{"id":"180a7bad3b92acc332c7f8b824654d7ffa2c7b1d","author":{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},"date":"2017-11-09 06:22:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"df28755e90caabf98c5edf57f44ab55a04cea4c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-09 10:51:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/70/518270/3/check/openstack-tox-pep8/41e28e7/ : FAILURE in 2m 51s\n- openstack-tox-py27 http://logs.openstack.org/70/518270/3/check/openstack-tox-py27/cc51a20/ : SUCCESS in 2m 38s\n- openstack-tox-py35 http://logs.openstack.org/70/518270/3/check/openstack-tox-py35/96d3023/ : SUCCESS in 2m 38s\n- build-openstack-sphinx-docs http://logs.openstack.org/70/518270/3/check/build-openstack-sphinx-docs/6710825/html/ : SUCCESS in 2m 21s\n- kolla-build-centos-binary http://logs.openstack.org/70/518270/3/check/kolla-build-centos-binary/3eb8db4/ : FAILURE in 33m 00s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/70/518270/3/check/kolla-build-centos-source/fd526de/ : FAILURE in 52m 12s\n- kolla-build-ubuntu-binary http://logs.openstack.org/70/518270/3/check/kolla-build-ubuntu-binary/522de92/ : FAILURE in 34m 00s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/70/518270/3/check/kolla-build-ubuntu-source/40322c4/ : FAILURE in 22m 36s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/70/518270/3/check/kolla-build-oraclelinux-binary/f5e6696/ : TIMED_OUT in 1h 32m 41s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/70/518270/3/check/kolla-build-oraclelinux-source/3cf8a6a/ : FAILURE in 57m 59s\n- kolla-ansible-centos-source http://logs.openstack.org/70/518270/3/check/kolla-ansible-centos-source/b807e8e/ : SUCCESS in 53m 16s (non-voting)\n- kolla-ansible-centos-binary http://logs.openstack.org/70/518270/3/check/kolla-ansible-centos-binary/068489e/ : SUCCESS in 1h 31m 49s (non-voting)\n- kolla-ansible-ubuntu-source http://logs.openstack.org/70/518270/3/check/kolla-ansible-ubuntu-source/6a62969/ : FAILURE in 37m 33s (non-voting)\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/70/518270/3/check/kolla-ansible-ubuntu-binary/950447b/ : FAILURE in 41m 01s (non-voting)\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/70/518270/3/check/kolla-ansible-oraclelinux-source/40b2b1a/ : SUCCESS in 54m 20s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/70/518270/3/check/kolla-ansible-oraclelinux-binary/bbeffad/ : SUCCESS in 1h 06m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0f9ee21d9ddcd4fa0fd17b0b76665c00d3a282af","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2017-11-12 04:39:56.000000000","message":"Patch Set 3: Code-Review-1\n\nshould we introcude onos/cord images into kolla?","accounts_in_message":[],"_revision_number":3},{"id":"feeba5a8f5ed87ebcf372ba20610e46e337f5845","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-21 12:09:18.000000000","message":"Abandoned\n\nnetworking-onos is no longer maintained.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c33f4e0e3cfd859c5c6e19af0589730db3d7d91e","revisions":{"987f08b21cc059caa9d880092c984854ea3a3daf":{"kind":"REWORK","_number":1,"created":"2017-11-07 07:43:22.000000000","uploader":{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},"ref":"refs/changes/70/518270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/518270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/518270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/518270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/518270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/518270/1"}}},"commit":{"parents":[{"commit":"61495b1a8303e52f58cc4d8970c2d484b5f31520","subject":"Merge \"Dockerhub publisher job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61495b1a8303e52f58cc4d8970c2d484b5f31520"}]}],"author":{"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","date":"2017-11-07 07:38:28.000000000","tz":0},"committer":{"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","date":"2017-11-07 07:41:55.000000000","tz":0},"subject":"wip Implement onos Docker images","message":"wip Implement onos Docker images\n\nonos can integrates to openstack neutron SDN\n\nChange-Id: I7b539490b07ff5729b0ab43f992c81770a6d026d\nImplements: blueprint onos-images\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/987f08b21cc059caa9d880092c984854ea3a3daf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/987f08b21cc059caa9d880092c984854ea3a3daf"}]},"branch":"refs/heads/master"},"7cc2c6b239c2ae346cc4c0aed9bf70a1859f3557":{"kind":"REWORK","_number":2,"created":"2017-11-07 09:42:04.000000000","uploader":{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},"ref":"refs/changes/70/518270/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/518270/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/518270/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/518270/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/518270/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/518270/2"}}},"commit":{"parents":[{"commit":"61495b1a8303e52f58cc4d8970c2d484b5f31520","subject":"Merge \"Dockerhub publisher job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61495b1a8303e52f58cc4d8970c2d484b5f31520"}]}],"author":{"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","date":"2017-11-07 07:38:28.000000000","tz":0},"committer":{"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","date":"2017-11-07 09:41:46.000000000","tz":0},"subject":"wip Implement onos Docker images","message":"wip Implement onos Docker images\n\nonos can integrates to openstack neutron SDN\n\nChange-Id: I7b539490b07ff5729b0ab43f992c81770a6d026d\nImplements: blueprint onos-images\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cc2c6b239c2ae346cc4c0aed9bf70a1859f3557"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cc2c6b239c2ae346cc4c0aed9bf70a1859f3557"}]},"branch":"refs/heads/master"},"c33f4e0e3cfd859c5c6e19af0589730db3d7d91e":{"kind":"REWORK","_number":3,"created":"2017-11-09 06:22:34.000000000","uploader":{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},"ref":"refs/changes/70/518270/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/518270/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/518270/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/518270/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/518270/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/518270/3"}}},"commit":{"parents":[{"commit":"61495b1a8303e52f58cc4d8970c2d484b5f31520","subject":"Merge \"Dockerhub publisher job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61495b1a8303e52f58cc4d8970c2d484b5f31520"}]}],"author":{"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","date":"2017-11-07 07:38:28.000000000","tz":0},"committer":{"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","date":"2017-11-09 06:21:27.000000000","tz":0},"subject":"wip Implement onos Docker images","message":"wip Implement onos Docker images\n\nonos can integrates to openstack neutron SDN\n\nChange-Id: I7b539490b07ff5729b0ab43f992c81770a6d026d\nImplements: blueprint onos-images\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c33f4e0e3cfd859c5c6e19af0589730db3d7d91e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c33f4e0e3cfd859c5c6e19af0589730db3d7d91e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
