)]}'
{"id":"openstack%2Fkolla~557659","triplet_id":"openstack%2Fkolla~master~I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0","project":"openstack/kolla","branch":"master","topic":"arm64-ci","hashtags":[],"change_id":"I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0","subject":"Add Debian/source aarch64 CI job","status":"MERGED","created":"2018-03-29 09:37:47.000000000","updated":"2019-12-16 16:07:20.000000000","submitted":"2019-12-16 16:03:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"557659-1576512204042-0d4ca9dc","meta_rev_id":"c6f6d17a3a97dbd0f080f6519f109b3f480bec30","_number":557659,"virtual_id_number":557659,"owner":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28003,"name":"Kylian Mbappé","email":"lijiawang8866@gmail.com","username":"li.jiawang"},{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-16 16:03:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},{"value":0,"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":28003,"name":"Kylian Mbappé","email":"lijiawang8866@gmail.com","username":"li.jiawang"},{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":2,"date":"2019-12-16 08:25:05.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-12-16 14:18:19.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":1,"date":"2019-12-16 08:51:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},{"value":2,"date":"2019-12-16 12:20:33.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},{"value":0,"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28003,"name":"Kylian Mbappé","email":"lijiawang8866@gmail.com","username":"li.jiawang"},{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-12-16 14:18:19.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},{"value":0,"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28003,"name":"Kylian Mbappé","email":"lijiawang8866@gmail.com","username":"li.jiawang"},{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},{"value":0,"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28003,"name":"Kylian Mbappé","email":"lijiawang8866@gmail.com","username":"li.jiawang"},{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},{"value":0,"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},{"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":28003,"name":"Kylian Mbappé","email":"lijiawang8866@gmail.com","username":"li.jiawang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-04-23 13:32:46.000000000","updated_by":{"_account_id":28003,"name":"Kylian Mbappé","email":"lijiawang8866@gmail.com","username":"li.jiawang"},"reviewer":{"_account_id":28003,"name":"Kylian Mbappé","email":"lijiawang8866@gmail.com","username":"li.jiawang"},"state":"REVIEWER"},{"updated":"2018-04-29 13:24:49.000000000","updated_by":{"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},"reviewer":{"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},"state":"REVIEWER"},{"updated":"2018-05-19 08:19:35.000000000","updated_by":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"reviewer":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"state":"REVIEWER"},{"updated":"2018-06-10 09:02:39.000000000","updated_by":{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},"reviewer":{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},"state":"REVIEWER"},{"updated":"2018-07-11 07:28:05.000000000","updated_by":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"reviewer":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"state":"REVIEWER"},{"updated":"2019-03-07 08:28:54.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-03-27 05:04:25.000000000","updated_by":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"reviewer":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"state":"REVIEWER"},{"updated":"2019-06-07 09:45:37.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-13 16:08:25.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-12-13 17:25:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-12-16 08:25:05.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-12-16 12:20:33.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-12-16 14:18:19.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-12-16 16:03:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"695e1da31f1ce0cc6f10cd52965cbf80bb144c02","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-03-29 09:37:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f76cd84911e931ac5c0646e4ab59aa484f66c695","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-29 11:01:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/1/check/openstack-tox-pep8/f25f8de/ : FAILURE in 3m 18s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/1/check/openstack-tox-py27/cb5b527/ : SUCCESS in 3m 06s\n- openstack-tox-py35 http://logs.openstack.org/59/557659/1/check/openstack-tox-py35/71d7a93/ : SUCCESS in 3m 00s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/557659/1/check/build-openstack-sphinx-docs/f98af59/html/ : SUCCESS in 2m 41s\n- kolla-build-centos-binary http://logs.openstack.org/59/557659/1/check/kolla-build-centos-binary/587f446/ : SUCCESS in 43m 44s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/59/557659/1/check/kolla-build-centos-source/7fc0e16/ : SUCCESS in 44m 25s\n- kolla-ansible-centos-source http://logs.openstack.org/59/557659/1/check/kolla-ansible-centos-source/ebc6393/ : SUCCESS in 59m 02s (non-voting)\n- kolla-ansible-centos-binary http://logs.openstack.org/59/557659/1/check/kolla-ansible-centos-binary/2c049a5/ : SUCCESS in 1h 12m 44s (non-voting)\n- kolla-build-debian-source http://logs.openstack.org/59/557659/1/check/kolla-build-debian-source/8d96705/ : RETRY_LIMIT in 10m 36s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/59/557659/1/check/kolla-build-oraclelinux-binary/3b66f38/ : SUCCESS in 1h 17m 01s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/59/557659/1/check/kolla-build-oraclelinux-source/04b724d/ : SUCCESS in 1h 01m 49s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/59/557659/1/check/kolla-ansible-oraclelinux-source/e7900f3/ : SUCCESS in 1h 05m 59s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/59/557659/1/check/kolla-ansible-oraclelinux-binary/d5a8c42/ : SUCCESS in 1h 10m 14s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/1/check/openstack-tox-lower-constraints/c836541/ : SUCCESS in 2m 56s\n- kolla-build-ubuntu-binary http://logs.openstack.org/59/557659/1/check/kolla-build-ubuntu-binary/96535bb/ : FAILURE in 28m 39s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/59/557659/1/check/kolla-build-ubuntu-source/4d625cf/ : FAILURE in 27m 17s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/59/557659/1/check/kolla-ansible-ubuntu-source/d1672ba/ : SUCCESS in 52m 55s (non-voting)\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/59/557659/1/check/kolla-ansible-ubuntu-binary/1d4529e/ : FAILURE in 9m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e22cf6ef7ab45a14107b2071f41c82452422a73","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-03-30 10:25:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1781d693c990e793e0b7942630e4f98b460144d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-30 11:48:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/2/check/openstack-tox-pep8/026d113/ : FAILURE in 2m 42s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/2/check/openstack-tox-py27/760413f/ : SUCCESS in 3m 27s\n- openstack-tox-py35 http://logs.openstack.org/59/557659/2/check/openstack-tox-py35/1975ba3/ : SUCCESS in 2m 46s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/557659/2/check/build-openstack-sphinx-docs/87e652c/html/ : SUCCESS in 2m 42s\n- kolla-build-centos-binary http://logs.openstack.org/59/557659/2/check/kolla-build-centos-binary/1b8f439/ : SUCCESS in 46m 03s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/59/557659/2/check/kolla-build-centos-source/ef6a2d3/ : SUCCESS in 1h 09m 04s\n- kolla-ansible-centos-source http://logs.openstack.org/59/557659/2/check/kolla-ansible-centos-source/3ebba7d/ : SUCCESS in 1h 02m 13s (non-voting)\n- kolla-ansible-centos-binary http://logs.openstack.org/59/557659/2/check/kolla-ansible-centos-binary/d7b02af/ : SUCCESS in 58m 54s (non-voting)\n- kolla-build-debian-source http://logs.openstack.org/59/557659/2/check/kolla-build-debian-source/33eaa29/ : RETRY_LIMIT in 8m 00s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/59/557659/2/check/kolla-build-oraclelinux-binary/3d0d0e4/ : FAILURE in 1h 19m 49s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/59/557659/2/check/kolla-build-oraclelinux-source/8c0bfe6/ : SUCCESS in 1h 20m 07s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/59/557659/2/check/kolla-ansible-oraclelinux-source/3ba54f4/ : SUCCESS in 1h 03m 32s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/59/557659/2/check/kolla-ansible-oraclelinux-binary/ef552b6/ : SUCCESS in 1h 12m 32s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/2/check/openstack-tox-lower-constraints/b70ee91/ : SUCCESS in 2m 19s\n- kolla-build-ubuntu-binary http://logs.openstack.org/59/557659/2/check/kolla-build-ubuntu-binary/583e932/ : SUCCESS in 32m 48s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/59/557659/2/check/kolla-build-ubuntu-source/ccbe008/ : SUCCESS in 29m 32s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/59/557659/2/check/kolla-ansible-ubuntu-source/dbf0b1f/ : SUCCESS in 51m 37s (non-voting)\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/59/557659/2/check/kolla-ansible-ubuntu-binary/02909f1/ : SUCCESS in 44m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2e20f421c6270d414f59ff5343e536274d962aeb","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2018-04-02 03:44:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2139fef149cfcbb7a0f3be09c4ad86bb315b17d1","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2018-04-02 06:28:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b956bd528abf71062420b4849137d78a6f4f4da6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-02 06:50:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/4/check/openstack-tox-pep8/1f2e467/ : FAILURE in 2m 41s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/4/check/openstack-tox-py27/d8337e0/ : SUCCESS in 2m 55s\n- openstack-tox-py35 http://logs.openstack.org/59/557659/4/check/openstack-tox-py35/8575e99/ : SUCCESS in 2m 48s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/557659/4/check/build-openstack-sphinx-docs/4225197/html/ : SUCCESS in 3m 15s\n- kolla-build-debian-source http://logs.openstack.org/59/557659/4/check/kolla-build-debian-source/cac3fc5/ : FAILURE in 20m 55s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/4/check/openstack-tox-lower-constraints/4285116/ : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":4},{"id":"2aa451be9145f8ed258cd54d6689ad77ed9af124","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-02 08:19:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"dc832145f223f7bbe2e82f9af76c8fc71ef5ef9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-02 08:25:12.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/5/check/openstack-tox-pep8/ad33208/ : FAILURE in 2m 29s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/5/check/openstack-tox-py27/903b243/ : SUCCESS in 2m 37s\n- openstack-tox-py35 http://logs.openstack.org/59/557659/5/check/openstack-tox-py35/c4b1753/ : SUCCESS in 2m 38s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/557659/5/check/build-openstack-sphinx-docs/e34755d/html/ : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":5},{"id":"813c08c817c339e6ab8e26643ae7b5e11c16dfc8","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-02 08:26:27.000000000","message":"Patch Set 5:\n\ncheck experimental","accounts_in_message":[],"_revision_number":5},{"id":"69fd39a1276e195e65052d3b9a84e190877ac81a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-02 08:40:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/5/experimental/kolla-build-debian-source-arm64/19f2836/ : FAILURE in 13m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cdde771820aad87b791e7e6bb3f9b5716ef9f220","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-02 08:59:32.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"12dd4bf9273ed49eeb0f6c077dc503dde5e971b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-02 09:04:14.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/6/check/openstack-tox-pep8/6ba677e/ : FAILURE in 2m 46s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/6/check/openstack-tox-py27/db49400/ : SUCCESS in 3m 21s\n- openstack-tox-py35 http://logs.openstack.org/59/557659/6/check/openstack-tox-py35/fd7328a/ : SUCCESS in 2m 41s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/557659/6/check/build-openstack-sphinx-docs/4cb5798/html/ : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":6},{"id":"cafa61a1556e3ba89d6397d1f27b3e895a85cdd3","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-02 10:13:02.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8d84d4047c7d2c6a30c483d8e05d75a42e7a1a4b","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-02 10:16:39.000000000","message":"Patch Set 7:\n\ncheck experimental","accounts_in_message":[],"_revision_number":7},{"id":"0f93db9683d1a5a357948b373bbb9521338ce4be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-02 10:20:29.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/7/check/openstack-tox-pep8/7053797/ : FAILURE in 2m 34s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/7/check/openstack-tox-py27/e802833/ : SUCCESS in 2m 30s\n- openstack-tox-py35 http://logs.openstack.org/59/557659/7/check/openstack-tox-py35/aab0c36/ : SUCCESS in 2m 35s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/557659/7/check/build-openstack-sphinx-docs/3d0392a/html/ : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":7},{"id":"6073eafdc790ba2d5f9987bd104cf80e5d56bd81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-02 10:54:25.000000000","message":"Patch Set 7:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/7/experimental/kolla-build-debian-source-arm64/53712ff/ : FAILURE in 37m 01s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6985c7ce178e341a3a9329006c626bc0acd19874","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-03 01:23:02.000000000","message":"Patch Set 7:\n\ncheck experimental","accounts_in_message":[],"_revision_number":7},{"id":"c9add1ae27c1c3baa2303be81bad4b81e86573f1","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-03 02:10:09.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"aa30075e68d2e4f422981bb8ba25e48bf991a909","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-03 02:12:10.000000000","message":"Patch Set 8:\n\ncheck experimental","accounts_in_message":[],"_revision_number":8},{"id":"86c85c51b7dbe4bc533658276843e03a595fc8c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-03 02:15:00.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/8/check/openstack-tox-pep8/968f96e/ : FAILURE in 2m 40s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/8/check/openstack-tox-py27/9057114/ : SUCCESS in 2m 39s\n- openstack-tox-py35 http://logs.openstack.org/59/557659/8/check/openstack-tox-py35/77a211f/ : SUCCESS in 3m 20s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/557659/8/check/build-openstack-sphinx-docs/d8fe138/html/ : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":8},{"id":"ad24ab6ebc300208d2869f5fdc2489135cacdbfd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-03 02:54:48.000000000","message":"Patch Set 8:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/8/experimental/kolla-build-debian-source-arm64/709ff0e/ : FAILURE in 41m 41s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"77bb4b1ae24af622c041aae6d202a87b9b55f4c6","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-03 03:02:07.000000000","message":"Patch Set 8:\n\ncheck experimental","accounts_in_message":[],"_revision_number":8},{"id":"3ed931aa09e871bd36b1162e9a0b129f8a51171a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-03 03:30:36.000000000","message":"Patch Set 8:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/8/experimental/kolla-build-debian-source-arm64/0fa0aae/ : FAILURE in 27m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ddb1085d00df2f2e42d5e042a58591e60344ff22","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-03 04:30:52.000000000","message":"Patch Set 8:\n\ncheck experimental","accounts_in_message":[],"_revision_number":8},{"id":"21612834d1cf2aa3cf805faed7a8afbfae5d25b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-03 04:53:21.000000000","message":"Patch Set 8:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/8/experimental/kolla-build-debian-source-arm64/caa3ed3/ : FAILURE in 21m 42s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2fd9defb64187fbc9c8dbea25d6b9f62b11a6af4","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-03 06:22:44.000000000","message":"Patch Set 8:\n\ncheck experimental","accounts_in_message":[],"_revision_number":8},{"id":"871ed41f8f925487bae733314f6319302edcb8ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-03 06:55:22.000000000","message":"Patch Set 8:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/8/experimental/kolla-build-debian-source-arm64/9ee7288/ : FAILURE in 31m 57s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f138b1bc5c78b2029d67308424accd473be6b1ce","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-03 09:36:16.000000000","message":"Patch Set 8:\n\ncheck experimental","accounts_in_message":[],"_revision_number":8},{"id":"cb9e9eeb1dad8f52485fdcf8475703fd8308dd6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-03 10:11:06.000000000","message":"Patch Set 8:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/8/experimental/kolla-build-debian-source-arm64/07a096d/ : FAILURE in 34m 03s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"dafde5ad746b07455c39010dd943c70523870a99","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-04 05:52:55.000000000","message":"Patch Set 8:\n\ncheck experimental","accounts_in_message":[],"_revision_number":8},{"id":"604c718c08ae8faef590c7079a26800cec013b87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-04 08:45:35.000000000","message":"Patch Set 8:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/8/experimental/kolla-build-debian-source-arm64/ed14540/ : TIMED_OUT in 2h 51m 56s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"26dffdd0b0b582082d520325d7fd630f1deadbaf","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-10 03:22:43.000000000","message":"Patch Set 8:\n\ncheck experimental","accounts_in_message":[],"_revision_number":8},{"id":"9ba3d385f7881e0d7dd4678a95d92be572882e22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-10 05:09:28.000000000","message":"Patch Set 8:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/8/experimental/kolla-build-debian-source-arm64/48594e2/ : FAILURE in 1h 45m 48s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"5dea771c79549aca4da1997fc21d01b2394662fb","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-10 09:55:50.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"1d0c654a71071e3012b015217016f7f9a544ef32","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-10 10:00:28.000000000","message":"Patch Set 9:\n\ncheck experimental","accounts_in_message":[],"_revision_number":9},{"id":"8cacf32af5fac9415232c05392b9469b600d12e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-10 10:05:22.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/9/check/openstack-tox-pep8/077b075/ : FAILURE in 2m 55s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/9/check/openstack-tox-py27/c99e403/ : SUCCESS in 2m 58s\n- openstack-tox-py35 http://logs.openstack.org/59/557659/9/check/openstack-tox-py35/3caedc4/ : SUCCESS in 3m 02s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/557659/9/check/build-openstack-sphinx-docs/17ae6d4/html/ : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":9},{"id":"ada43bd5a211c0a3f5948eb5fb0e26e774ebc76a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-10 10:08:53.000000000","message":"Patch Set 9:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/9/experimental/kolla-build-debian-source-arm64/c0e00e5/ : RETRY_LIMIT in 6m 49s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"64e89995891f116fce28c14b1961d0d90036ae73","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-10 11:01:32.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"8328a056a79d2a01113b9b091a74db176913ad47","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-10 11:04:05.000000000","message":"Patch Set 10:\n\ncheck experimental","accounts_in_message":[],"_revision_number":10},{"id":"85d26d7900b43d94cf00011aa52eaf12f73a7c7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-10 11:08:49.000000000","message":"Patch Set 10:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/10/experimental/kolla-build-debian-source-arm64/0ce7753/ : RETRY_LIMIT in 3m 50s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"61c1f5792f80c1ca65199b50c8d8e8e87c8f369c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-10 11:11:11.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/10/check/openstack-tox-pep8/84a6463/ : SUCCESS in 2m 36s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/10/check/openstack-tox-py27/be0f079/ : SUCCESS in 2m 36s\n- openstack-tox-py35 http://logs.openstack.org/59/557659/10/check/openstack-tox-py35/0ae5d2a/ : SUCCESS in 2m 54s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/557659/10/check/build-openstack-sphinx-docs/9d52bdb/html/ : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":10},{"id":"a231bb031c1f30585aef6105362ac372e8f371ab","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-12 09:56:04.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"74093b92a2ad4eeecde040073df502107b6815e2","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-12 10:11:20.000000000","message":"Patch Set 11:\n\ncheck experimental","accounts_in_message":[],"_revision_number":11},{"id":"dc74280ce20a04f53e991cac06475e08508e04c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-12 10:39:49.000000000","message":"Patch Set 11:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/11/experimental/kolla-build-debian-source-arm64/2ab94f2/ : FAILURE in 27m 08s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"16388dcb93cce168efdeb6f1e88530fdd8a78353","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-12 10:46:00.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/11/check/openstack-tox-pep8/9a7b26e/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/11/check/openstack-tox-py27/7b9be7a/ : SUCCESS in 2m 42s\n- openstack-tox-py35 http://logs.openstack.org/59/557659/11/check/openstack-tox-py35/6780904/ : SUCCESS in 3m 13s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/557659/11/check/build-openstack-sphinx-docs/af78553/html/ : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":11},{"id":"dc6896b5d0d5960efaec5e10c61aa3e297907338","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-12 10:59:43.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"8111bc0c5f55b08030cf7f139d9f2bb79acbfb93","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-12 11:02:08.000000000","message":"Patch Set 12:\n\ncheck experimental","accounts_in_message":[],"_revision_number":12},{"id":"6f51266f6301b252e108b8a1592c3db33c4b2d6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-12 11:45:18.000000000","message":"Patch Set 12:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/12/experimental/kolla-build-debian-source-arm64/e698890/ : FAILURE in 41m 59s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"c0022198da26778d186885b57c384d17619c1139","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-12 12:13:01.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/12/check/openstack-tox-pep8/4b24376/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/12/check/openstack-tox-py27/bcde37f/ : SUCCESS in 3m 40s\n- openstack-tox-py35 http://logs.openstack.org/59/557659/12/check/openstack-tox-py35/beca40e/ : SUCCESS in 2m 51s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/557659/12/check/build-openstack-sphinx-docs/433b15b/html/ : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":12},{"id":"a5d61922e81472416c353e3b1a5134b90111c79b","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-13 03:20:53.000000000","message":"Patch Set 12:\n\ncheck experimental","accounts_in_message":[],"_revision_number":12},{"id":"12012684d18481847bdcdc939d7810dc6d6064b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-13 03:25:57.000000000","message":"Patch Set 12:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 kolla-build-debian-source-arm64 : NODE_FAILURE (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"a716a4325b9fe3e673d6bdb60f095a06ea4ac17a","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-13 09:13:20.000000000","message":"Patch Set 12:\n\ncheck experimental","accounts_in_message":[],"_revision_number":12},{"id":"fc62795e123d0960704a39f74f63c96549a216cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-13 11:57:47.000000000","message":"Patch Set 12:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/12/experimental/kolla-build-debian-source-arm64/e5a2f0f/ : FAILURE in 2h 43m 28s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"74121443ba8a81df1b68e45d1e4c26ac08d67266","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-17 02:07:37.000000000","message":"Patch Set 12:\n\ncheck experimental","accounts_in_message":[],"_revision_number":12},{"id":"41ce8015c8c0092d59b7984616758f05389e9679","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-17 02:50:46.000000000","message":"Patch Set 12:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 kolla-build-debian-source-arm64 : NODE_FAILURE (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"37ebdc858aded121885adbf0cc94bd284e032678","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-17 02:54:38.000000000","message":"Patch Set 12:\n\ncheck experimental","accounts_in_message":[],"_revision_number":12},{"id":"290eb883173c3a6150dc0092a1ebb129e509aef8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-17 03:36:54.000000000","message":"Patch Set 12:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 kolla-build-debian-source-arm64 : NODE_FAILURE (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"744daf554f8a8726444fcb7fe0b16f2908595f9b","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-17 08:27:43.000000000","message":"Patch Set 12:\n\ncheck experimental","accounts_in_message":[],"_revision_number":12},{"id":"ee28b3b5bae42a5b5b5d9c02b40388c0ef60591b","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-17 08:56:57.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"0ec9fdc2797b9b1856c59a3b2d012a1ba997eb36","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-17 08:57:30.000000000","message":"Patch Set 12:\n\ncheck experimental","accounts_in_message":[],"_revision_number":12},{"id":"a26e267500c1c583830a7deac4fa32ec277c6cb3","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-17 08:58:02.000000000","message":"Patch Set 13:\n\ncheck experimental","accounts_in_message":[],"_revision_number":13},{"id":"4452818a2546d9bcb6ad3e06d62caf30b21e46e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-17 08:59:48.000000000","message":"Patch Set 13:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 finger://ze10.openstack.org/d80087a77b844e3cb3706ef7fd810a72 : RETRY_LIMIT in 1m 00s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"6e1eb77beea3e2c9f77177951c718fc4786ff872","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-17 09:04:43.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/13/check/openstack-tox-pep8/c7b2350/ : SUCCESS in 2m 15s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/13/check/openstack-tox-py27/e5fd4ed/ : SUCCESS in 2m 31s\n- openstack-tox-py35 http://logs.openstack.org/59/557659/13/check/openstack-tox-py35/7125861/ : SUCCESS in 2m 54s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/557659/13/check/build-openstack-sphinx-docs/5156a12/html/ : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":13},{"id":"46c4b20ec79bdc38cf89742a16e6cd32bc2b012d","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-17 09:12:17.000000000","message":"Patch Set 13:\n\ncheck experimental","accounts_in_message":[],"_revision_number":13},{"id":"3e6597e4f36ec61a8967c169a88c423132031151","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-17 09:42:41.000000000","message":"Patch Set 12:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 kolla-build-debian-source-arm64 : NODE_FAILURE (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"98fdc0c1ac188bf4ef637070fd436cf6993c015e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-17 10:07:44.000000000","message":"Patch Set 13:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 kolla-build-debian-source-arm64 : NODE_FAILURE (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"f884949c2704fbe7432669f35b75617492ff68ba","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-18 08:38:26.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"78ff4630c2aa0e8534f54178ac006e0f42122fd8","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-18 08:40:58.000000000","message":"Patch Set 14:\n\ncheck experimental","accounts_in_message":[],"_revision_number":14},{"id":"b9fe1fed26ff766445b12c93c0ccc9d78374eb12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-18 08:46:51.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/14/check/openstack-tox-pep8/572bf00/ : SUCCESS in 2m 49s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/14/check/openstack-tox-py27/1cf7105/ : SUCCESS in 2m 54s\n- openstack-tox-py35 http://logs.openstack.org/59/557659/14/check/openstack-tox-py35/3e33235/ : SUCCESS in 3m 33s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/557659/14/check/build-openstack-sphinx-docs/b638edc/html/ : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":14},{"id":"b800e67023d368000c8b93c15c50d1fcc754938f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-18 09:47:06.000000000","message":"Patch Set 14:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/14/experimental/kolla-build-debian-source-arm64/44da485/ : FAILURE in 1h 05m 13s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"76cb7e965273598e8c09453e5b369b06b984e648","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-18 11:00:56.000000000","message":"Patch Set 14:\n\ncheck experimental","accounts_in_message":[],"_revision_number":14},{"id":"929ad361558cece871b47dfc646b0fdf9786bd4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-18 11:39:15.000000000","message":"Patch Set 14:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/14/experimental/kolla-build-debian-source-arm64/4c888fd/ : FAILURE in 37m 17s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"360e8dc4bddac55258966a2cb57ae68c10d8537f","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-19 01:33:26.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"1463859604a4f900068260147c1ba7699e02aa0a","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-19 01:41:39.000000000","message":"Patch Set 15:\n\ncheck experimental","accounts_in_message":[],"_revision_number":15},{"id":"cf882ab9d719653dfa7c46e047b2b79129a2921f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-19 01:41:47.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/15/check/openstack-tox-pep8/269f7de/ : SUCCESS in 2m 14s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/15/check/openstack-tox-py27/28497c3/ : SUCCESS in 2m 30s\n- openstack-tox-py35 http://logs.openstack.org/59/557659/15/check/openstack-tox-py35/3347048/ : SUCCESS in 2m 25s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/557659/15/check/build-openstack-sphinx-docs/303b250/html/ : SUCCESS in 2m 21s","accounts_in_message":[],"_revision_number":15},{"id":"2366537911bdbc83888ab367d8d6f3fed0a7b59d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-19 01:53:56.000000000","message":"Patch Set 15:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/15/experimental/kolla-build-debian-source-arm64/7d377d9/ : RETRY_LIMIT in 11m 27s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"0efba41babdf99936d36b2325064fc1c0cc5f902","author":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"date":"2018-04-19 07:38:58.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"a6728817dd131555441d22484c49a9b20bcecdb9","author":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"date":"2018-04-19 07:42:48.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"4ee9fda42e269a44309263c42c2fa2d56f0eeafa","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-19 08:00:44.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"031d70c8ab71da88f2b437b9c8e104b6196adcfe","author":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"date":"2018-04-19 08:28:43.000000000","message":"Patch Set 15: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"6b25e964b26f35a00b4af1c325bc2361cf169fc1","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-19 08:31:31.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"12218269819e452783ade2a4f1a54eb001912a1b","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-20 05:57:47.000000000","message":"Patch Set 15:\n\ncheck experimental","accounts_in_message":[],"_revision_number":15},{"id":"ead8af355824e50c6027010df83aa41ae5395a5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-20 06:06:25.000000000","message":"Patch Set 15:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/15/experimental/kolla-build-debian-source-arm64/83a6896/ : RETRY_LIMIT in 7m 49s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"4cb898ef9e8fbc4f1e18a998e150790826af5c8e","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-20 06:13:49.000000000","message":"Patch Set 15:\n\ncheck experimental","accounts_in_message":[],"_revision_number":15},{"id":"3487c94f3cbb3987f5680bb4a0a53415eb2129c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-20 06:21:43.000000000","message":"Patch Set 15:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/15/experimental/kolla-build-debian-source-arm64/894fcdd/ : RETRY_LIMIT in 6m 53s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"f1df90ef838eb727d2c86ac58b9043114ab7e081","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-23 06:45:18.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"8aee748ba81fca8c7fb2efe03bae0058f16d3422","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-04-23 06:45:49.000000000","message":"Patch Set 16:\n\ncheck experimental","accounts_in_message":[],"_revision_number":16},{"id":"ed95e69cb8ecd94f491af85a81b18099ab660094","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-23 06:52:51.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/16/check/openstack-tox-pep8/baf33a8/ : SUCCESS in 2m 15s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/16/check/openstack-tox-py27/b627e34/ : SUCCESS in 2m 41s\n- openstack-tox-py35 http://logs.openstack.org/59/557659/16/check/openstack-tox-py35/c115b3f/ : SUCCESS in 2m 35s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/557659/16/check/build-openstack-sphinx-docs/babcb0c/html/ : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":16},{"id":"3002be9b4b203dc4ea84bf843a2eb5e0d2670ada","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-23 07:34:00.000000000","message":"Patch Set 16:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/16/experimental/kolla-build-debian-source-arm64/7427d03/ : FAILURE in 47m 22s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"c48935a2ecfbe9eb151b387e1dc302dfc1cbd5ba","author":{"_account_id":28003,"name":"Kylian Mbappé","email":"lijiawang8866@gmail.com","username":"li.jiawang"},"date":"2018-04-23 13:32:46.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"1b51bcd3d834e35860798ce4d268bde98db74dc6","author":{"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},"date":"2018-04-29 13:24:49.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"de17929a85e847a15ea4164999b166ae3f1d78a8","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-05-16 02:25:25.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"b7a20a6e83a92e2e337b248f4ecda8c6298b810f","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-05-16 02:26:29.000000000","message":"Patch Set 17:\n\ncheck experimental","accounts_in_message":[],"_revision_number":17},{"id":"07969fb667bb2bda02ea0d708b7130210eaa5569","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-16 02:29:21.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/17/check/openstack-tox-pep8/eee954b/ : SUCCESS in 2m 37s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/17/check/openstack-tox-py27/51c4cd0/ : SUCCESS in 2m 15s\n- openstack-tox-py35 http://logs.openstack.org/59/557659/17/check/openstack-tox-py35/4af2817/ : SUCCESS in 3m 23s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/557659/17/check/build-openstack-sphinx-docs/aef6120/html/ : SUCCESS in 2m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/17/check/openstack-tox-lower-constraints/7385cc2/ : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":17},{"id":"eeb8f7e566d3695fa962638aef7aa824719f003f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-16 02:34:16.000000000","message":"Patch Set 17:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/17/experimental/kolla-build-debian-source-arm64/78d2ce5/ : RETRY_LIMIT in 7m 10s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"c118a3140fd6cc5c3e63f8734d2569cb60e9fc9e","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-05-16 02:44:17.000000000","message":"Patch Set 17:\n\ncheck experimental","accounts_in_message":[],"_revision_number":17},{"id":"253a6a1a93a89fcc3484e43742ffebff6e6b9d55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-16 02:50:04.000000000","message":"Patch Set 17:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/17/experimental/kolla-build-debian-source-arm64/14dcbd7/ : RETRY_LIMIT in 5m 11s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"a700f0a5372af998d488b2d542be65c307029958","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-05-16 06:04:09.000000000","message":"Patch Set 17:\n\ncheck experimental","accounts_in_message":[],"_revision_number":17},{"id":"9080e00621a65ca535ff196d28d7d24676de192e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-16 06:12:41.000000000","message":"Patch Set 17:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/17/experimental/kolla-build-debian-source-arm64/5924492/ : RETRY_LIMIT in 6m 23s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"535b74d75b229485572cb68d35866c9d167ca846","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-05-16 06:37:38.000000000","message":"Patch Set 17:\n\ncheck experimental","accounts_in_message":[],"_revision_number":17},{"id":"d8f8360b7fb64b76f7ee4d0619730be62a6ed9ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-16 06:47:06.000000000","message":"Patch Set 17:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/17/experimental/kolla-build-debian-source-arm64/08a5967/ : RETRY_LIMIT in 5m 16s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"ed66e334d5f28a0152fb24a856a69bebe711d4db","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-05-16 09:11:56.000000000","message":"Patch Set 17:\n\ncheck experimental","accounts_in_message":[],"_revision_number":17},{"id":"9f629e189a0dc37aca3921d5e2b646fb9e450118","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-05-16 09:12:15.000000000","message":"Patch Set 17:\n\ncheck experimental","accounts_in_message":[],"_revision_number":17},{"id":"9e224fb04bb80e6f92ed0a089f6ec6a19e2d23fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-16 09:17:07.000000000","message":"Patch Set 17:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/17/experimental/kolla-build-debian-source-arm64/f624f89/ : RETRY_LIMIT in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"ce4739572d968ae1830554f2d8d497c9bd4d57a3","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-05-16 09:23:15.000000000","message":"Patch Set 17:\n\ncheck experimental","accounts_in_message":[],"_revision_number":17},{"id":"d3c9b0846e5a701d37e37e7d2482391a793f91e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-16 09:36:08.000000000","message":"Patch Set 17:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/17/experimental/kolla-build-debian-source-arm64/9c573d1/ : RETRY_LIMIT in 5m 41s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"873196d4e328a7f6942484a357df1d43679294b0","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-05-16 10:59:34.000000000","message":"Patch Set 17:\n\ncheck experimental","accounts_in_message":[],"_revision_number":17},{"id":"2f379d6bfe1d4a21999260ba4fa175529efeceeb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-16 11:05:40.000000000","message":"Patch Set 17:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/17/experimental/kolla-build-debian-source-arm64/6179ced/ : RETRY_LIMIT in 5m 33s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"c567e494b26ceb6b9fa7adf871c453f8240c8660","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-05-17 01:00:42.000000000","message":"Patch Set 17:\n\ncheck experimental","accounts_in_message":[],"_revision_number":17},{"id":"e23aa2f563877dcea12a1fbda75463eada51c960","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-17 01:14:52.000000000","message":"Patch Set 17:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/17/experimental/kolla-build-debian-source-arm64/88ef78c/ : FAILURE in 13m 29s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"3875e7c3bc0f51d25f038573cc6cc83dc042486d","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-05-17 02:56:52.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"dcd2ad3a815b24e58f5ca3cca5c4ce70cf8e5eda","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-05-17 02:59:32.000000000","message":"Patch Set 18:\n\ncheck experimental","accounts_in_message":[],"_revision_number":18},{"id":"268635c7b0833f1cb94b6f40cb0bf1edc3780d96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-17 03:02:14.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/18/check/openstack-tox-pep8/9d1ebcb/ : SUCCESS in 3m 18s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/18/check/openstack-tox-py27/a8ab569/ : SUCCESS in 3m 23s\n- openstack-tox-py35 http://logs.openstack.org/59/557659/18/check/openstack-tox-py35/96ea401/ : SUCCESS in 2m 59s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/557659/18/check/build-openstack-sphinx-docs/84a7399/html/ : SUCCESS in 2m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/18/check/openstack-tox-lower-constraints/cd34f15/ : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":18},{"id":"9c5597dec3864636334b85c8b3449b7ec44004d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-17 03:04:04.000000000","message":"Patch Set 18:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/18/experimental/kolla-build-debian-source-arm64/145097f/ : RETRY_LIMIT in 3m 54s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"318a26dbeb5531d4097c667dfa11c41fc43a24e1","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-05-17 03:10:30.000000000","message":"Patch Set 18:\n\ncheck experimental","accounts_in_message":[],"_revision_number":18},{"id":"2abb6175e08cd3e9ff60c3fa5672bc2b1c2c4eef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-17 03:15:03.000000000","message":"Patch Set 18:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/18/experimental/kolla-build-debian-source-arm64/0849645/ : RETRY_LIMIT in 3m 52s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"807aacbde433f463a72ac382308bb2afed41436b","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-05-17 03:20:18.000000000","message":"Patch Set 18:\n\ncheck experimental","accounts_in_message":[],"_revision_number":18},{"id":"a860629a8e858ddd2a2eaafe05367f20e2f7bc74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-17 04:41:10.000000000","message":"Patch Set 18:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/18/experimental/kolla-build-debian-source-arm64/43d893a/ : FAILURE in 49m 03s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"25600cd6c5ee24e76484010aba38c7731167541e","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-05-18 08:25:53.000000000","message":"Patch Set 18:\n\ncheck experimental","accounts_in_message":[],"_revision_number":18},{"id":"961fadae6ed54392358ba04a62f0941f10cd2ad6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-18 09:38:49.000000000","message":"Patch Set 18:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/18/experimental/kolla-build-debian-source-arm64/77f05d2/ : FAILURE in 1h 12m 10s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"0de9a7fd9bdbfe6d945e90d4560b44c302f71a10","author":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"date":"2018-05-19 08:19:35.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"135b67042fb383e1d6d924a1d99e9897675af6e7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2018-05-21 11:23:09.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"653608c76b73c8e766890435883915584341650d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2018-05-21 11:23:36.000000000","message":"Patch Set 18:\n\nwill do recheck after https://review.openstack.org/#/c/569778/ gets merged","accounts_in_message":[],"_revision_number":18},{"id":"15cd3bc49e2bc0f488ded785ffd1b7075c1923d7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2018-05-21 14:47:07.000000000","message":"Patch Set 18:\n\ncheck experimental","accounts_in_message":[],"_revision_number":18},{"id":"5b831a8ccd23161a4f2f79c29ec68d951060f90f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-21 16:08:18.000000000","message":"Patch Set 18:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/18/experimental/kolla-build-debian-source-arm64/57cf48b/ : FAILURE in 1h 20m 33s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"c858a94c3d50d55a1a990a0bc24b9741f11d2cef","author":{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},"date":"2018-05-22 22:42:08.000000000","message":"Patch Set 18:\n\ncheck experimental","accounts_in_message":[],"_revision_number":18},{"id":"f8b37d8e3ac482e56b3a42e6eeaa0d5327af8c07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-22 23:51:44.000000000","message":"Patch Set 18:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/18/experimental/kolla-build-debian-source-arm64/2b83cde/ : FAILURE in 1h 09m 01s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"0297955bdf1253fd2b7f7f4dbe9930ea8fb27e1e","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2018-05-24 11:16:40.000000000","message":"Patch Set 18:\n\ncheck experimental","accounts_in_message":[],"_revision_number":18},{"id":"e54dda1b3f626289590c46485011ceccf7bcb81a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-24 12:32:12.000000000","message":"Patch Set 18:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/18/experimental/kolla-build-debian-source-arm64/fe178d5/ : FAILURE in 1h 14m 54s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"c41106bc6ec266b5606786daa57d129c847ef919","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2018-05-25 09:43:06.000000000","message":"Patch Set 19: Commit message was updated.","accounts_in_message":[],"_revision_number":19},{"id":"8e96adb1ce8d868b05e0a759c8f26904f55c6cca","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2018-05-25 09:43:25.000000000","message":"Patch Set 19:\n\ncheck experimental","accounts_in_message":[],"_revision_number":19},{"id":"372c59e6875490b6876802705e98f3f53910baaf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-25 11:10:35.000000000","message":"Patch Set 19:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/19/experimental/kolla-build-debian-source-arm64/087b826/ : FAILURE in 1h 23m 41s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"ce6716e16d01ae1c3b8c57cedac762fd07a1489f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-25 11:24:56.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/19/check/openstack-tox-pep8/7a1c9a1/ : SUCCESS in 3m 58s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/19/check/openstack-tox-py27/a35de59/ : SUCCESS in 2m 49s\n- openstack-tox-py35 http://logs.openstack.org/59/557659/19/check/openstack-tox-py35/7669912/ : SUCCESS in 2m 25s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/557659/19/check/build-openstack-sphinx-docs/52186a6/html/ : SUCCESS in 2m 54s\n- kolla-build-centos-binary http://logs.openstack.org/59/557659/19/check/kolla-build-centos-binary/bc3e85f/ : SUCCESS in 45m 43s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/59/557659/19/check/kolla-build-centos-source/780e07d/ : SUCCESS in 1h 11m 22s\n- kolla-ansible-centos-source http://logs.openstack.org/59/557659/19/check/kolla-ansible-centos-source/4959ebf/ : SUCCESS in 1h 20m 10s (non-voting)\n- kolla-ansible-centos-binary http://logs.openstack.org/59/557659/19/check/kolla-ansible-centos-binary/facbbc6/ : SUCCESS in 1h 02m 47s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/59/557659/19/check/kolla-build-oraclelinux-binary/9414050/ : SUCCESS in 1h 37m 41s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/59/557659/19/check/kolla-build-oraclelinux-source/dce0c60/ : SUCCESS in 1h 31m 11s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/59/557659/19/check/kolla-ansible-oraclelinux-source/6294c08/ : POST_FAILURE in 1m 38s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/59/557659/19/check/kolla-ansible-oraclelinux-binary/b39b6f5/ : SUCCESS in 1h 05m 17s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/19/check/openstack-tox-lower-constraints/3df12d6/ : SUCCESS in 2m 58s\n- kolla-build-ubuntu-binary http://logs.openstack.org/59/557659/19/check/kolla-build-ubuntu-binary/ece8201/ : SUCCESS in 27m 24s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/59/557659/19/check/kolla-build-ubuntu-source/ec1da13/ : SUCCESS in 23m 08s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/59/557659/19/check/kolla-ansible-ubuntu-source/7bc4eef/ : SUCCESS in 44m 00s (non-voting)\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/59/557659/19/check/kolla-ansible-ubuntu-binary/aebe0d9/ : FAILURE in 53m 17s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"497c00810a1ffd0e4d30f420be64dabc5c63d489","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2018-05-25 14:53:44.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"dc2c94835e0d799c5c1af9ddbdc21a27111f9853","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2018-05-25 14:54:14.000000000","message":"Patch Set 19:\n\ncheck experimental","accounts_in_message":[],"_revision_number":19},{"id":"68756ff5d1ea80cc711249475e73d190da852615","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-25 14:58:46.000000000","message":"Patch Set 19:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/19/experimental/kolla-build-debian-source-arm64/ad7b3bf/ : RETRY_LIMIT in 3m 55s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"17e089edef0c094d14ec5596020074c092e21e6f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-25 16:26:46.000000000","message":"Patch Set 19:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/19/check/openstack-tox-pep8/d31502e/ : SUCCESS in 2m 28s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/19/check/openstack-tox-py27/b8c938b/ : SUCCESS in 2m 31s\n- openstack-tox-py35 http://logs.openstack.org/59/557659/19/check/openstack-tox-py35/00dc7ef/ : SUCCESS in 2m 50s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/557659/19/check/build-openstack-sphinx-docs/339596a/html/ : SUCCESS in 2m 14s\n- kolla-build-centos-binary http://logs.openstack.org/59/557659/19/check/kolla-build-centos-binary/3ab8479/ : SUCCESS in 59m 16s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/59/557659/19/check/kolla-build-centos-source/a49df97/ : SUCCESS in 59m 13s\n- kolla-ansible-centos-source http://logs.openstack.org/59/557659/19/check/kolla-ansible-centos-source/d19d26d/ : SUCCESS in 1h 03m 33s (non-voting)\n- kolla-ansible-centos-binary http://logs.openstack.org/59/557659/19/check/kolla-ansible-centos-binary/52d617e/ : SUCCESS in 1h 03m 26s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/59/557659/19/check/kolla-build-oraclelinux-binary/86b287b/ : SUCCESS in 1h 31m 29s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/59/557659/19/check/kolla-build-oraclelinux-source/be101cd/ : SUCCESS in 1h 11m 07s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/59/557659/19/check/kolla-ansible-oraclelinux-source/2fc6f88/ : SUCCESS in 1h 06m 42s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/59/557659/19/check/kolla-ansible-oraclelinux-binary/e72b153/ : SUCCESS in 1h 07m 34s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/19/check/openstack-tox-lower-constraints/aee0ff0/ : SUCCESS in 2m 53s\n- kolla-build-ubuntu-binary http://logs.openstack.org/59/557659/19/check/kolla-build-ubuntu-binary/60de0cc/ : SUCCESS in 33m 02s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/59/557659/19/check/kolla-build-ubuntu-source/b0bf42e/ : SUCCESS in 22m 12s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/59/557659/19/check/kolla-ansible-ubuntu-source/c8fef83/ : SUCCESS in 57m 45s (non-voting)\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/59/557659/19/check/kolla-ansible-ubuntu-binary/f16bd74/ : FAILURE in 52m 49s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"4f406504e350c9cc9db6d8ef8e3859ec02189a75","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2018-05-28 09:20:59.000000000","message":"Patch Set 19:\n\ncheck experimental","accounts_in_message":[],"_revision_number":19},{"id":"3121d1a648bbce54dc163b2da207ba2da145f4ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-28 10:31:54.000000000","message":"Patch Set 19:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/19/experimental/kolla-build-debian-source-arm64/0540032/ : FAILURE in 1h 10m 19s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"18f1b1a6d6a201b97eddb37be7826acfa68a9c2c","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2018-05-28 14:43:00.000000000","message":"Patch Set 19:\n\ncheck experimental","accounts_in_message":[],"_revision_number":19},{"id":"559eb636ae1d1e6da7a8ab154b0e1fb4a58e6d15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-28 14:47:38.000000000","message":"Patch Set 19:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/19/experimental/kolla-build-debian-source-arm64/8b88e95/ : RETRY_LIMIT in 4m 08s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"98c01ff50a1243da7002089ede7d9ad8be581e3a","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2018-06-05 02:12:44.000000000","message":"Patch Set 19: Code-Review+1","accounts_in_message":[],"_revision_number":19},{"id":"db0b36f1b7b7a0d6fa3b6a69dea98574cdd88ba4","author":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"date":"2018-06-07 02:55:00.000000000","message":"Patch Set 19: Code-Review+1","accounts_in_message":[],"_revision_number":19},{"id":"41fdf265acfca03cab1113523e2cd5f8b655950d","author":{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},"date":"2018-06-10 09:02:39.000000000","message":"Patch Set 19:\n\ncheck experimental","accounts_in_message":[],"_revision_number":19},{"id":"4a981c6fe5d1cd224b0e661855c6ad73a124dcce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-10 09:19:06.000000000","message":"Patch Set 19:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/19/experimental/kolla-build-debian-source-arm64/ed95f46/ : FAILURE in 15m 57s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"0301335dc994ec5703852a8677cd1f197430ab98","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2018-06-11 07:45:18.000000000","message":"Patch Set 19: Code-Review-1\n\njobs are failing http://logs.openstack.org/59/557659/19/experimental/kolla-build-debian-source-arm64/ed95f46/job-output.txt.gz#_2018-06-10_09_17_50_004401","accounts_in_message":[],"_revision_number":19},{"id":"bb9e7c88bdd12414699ac5160f4814241be6a7e7","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2018-06-11 07:45:37.000000000","message":"Patch Set 19:\n\ncheck experimental","accounts_in_message":[],"_revision_number":19},{"id":"3d0f3a34a6652bc1f35b1143f1231a8e1e537c81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-11 07:51:22.000000000","message":"Patch Set 19:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/19/experimental/kolla-build-debian-source-arm64/63c7a2d/ : RETRY_LIMIT in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"ea27357ce4febd273f01190ddd1933bbc995ee35","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2018-06-12 11:21:26.000000000","message":"Patch Set 19:\n\ncheck experimental","accounts_in_message":[],"_revision_number":19},{"id":"85f0d12b734de724b3e7e152bd58bb49e7db05b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-12 11:40:54.000000000","message":"Patch Set 19:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/19/experimental/kolla-build-debian-source-arm64/2065879/ : FAILURE in 16m 02s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"78e4d86849dbc0e1e736f9d385a3f6b80e9dc70d","author":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"date":"2018-06-17 23:57:10.000000000","message":"Removed Code-Review+1 by Surya Prakash (spsurya) \u003csingh.surya64mnnit@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":19},{"id":"76770fe06e96f0e0bb934e1716daaff3ae0c136f","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-06-21 08:11:24.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"98006b6dbb28874fb9545608947e8524142ef65b","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-06-21 08:13:56.000000000","message":"Patch Set 20:\n\ncheck experimental","accounts_in_message":[],"_revision_number":20},{"id":"c14f7224f3f3a931d61f5376c341b00ba92a80bd","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-06-21 08:16:07.000000000","message":"Patch Set 20: Workflow-1","accounts_in_message":[],"_revision_number":20},{"id":"530b100d14d7962f59b6290013978ef923d9386c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-21 08:21:48.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/20/check/openstack-tox-pep8/0502fa5/ : SUCCESS in 4m 08s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/20/check/openstack-tox-py27/47780ab/ : SUCCESS in 4m 27s\n- openstack-tox-py35 http://logs.openstack.org/59/557659/20/check/openstack-tox-py35/09a31f1/ : SUCCESS in 4m 25s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/557659/20/check/build-openstack-sphinx-docs/555667b/html/ : SUCCESS in 5m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/20/check/openstack-tox-lower-constraints/cf65945/ : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":20},{"id":"f5dc9c2d2d2802e9001dd109e09060d680389f42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-21 08:22:42.000000000","message":"Patch Set 20:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/20/experimental/kolla-build-debian-source-arm64/5f2a81b/ : RETRY_LIMIT in 6m 37s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"b257fd02866d3f5d24b7ad3ab1433d45cb6e3aba","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-06-21 08:40:34.000000000","message":"Patch Set 20:\n\ncheck experimental","accounts_in_message":[],"_revision_number":20},{"id":"35098f5efdd084e12c4648c4d46775d152e97edc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-21 08:46:10.000000000","message":"Patch Set 20:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/20/experimental/kolla-build-debian-source-arm64/a0a967d/ : RETRY_LIMIT in 4m 31s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"6610b892e16db2b0e8c7ef79968acf921094f2c1","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-06-22 06:33:55.000000000","message":"Patch Set 20:\n\ncheck experimental","accounts_in_message":[],"_revision_number":20},{"id":"e76b5481c04b57400f670bc7d5ab203878551d3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-22 06:43:16.000000000","message":"Patch Set 20:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/20/experimental/kolla-build-debian-source-arm64/2679acc/ : RETRY_LIMIT in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"8b7d2bf29443c73dff25f711aef024bbcef6595e","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-06-27 12:04:42.000000000","message":"Patch Set 20:\n\ncheck experimental","accounts_in_message":[],"_revision_number":20},{"id":"113ecf4155354c6d0677132b28ccda236c83e6f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-27 12:49:15.000000000","message":"Patch Set 20:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/20/experimental/kolla-build-debian-source-arm64/27be5ab/ : FAILURE in 42m 19s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"3e6ce8e4d13e99b43d1b4ee15bd46392ff164778","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2018-06-27 23:44:28.000000000","message":"Patch Set 20:\n\ncheck experimental","accounts_in_message":[],"_revision_number":20},{"id":"cc754e6e8578c08e6a1db191d3afe29c76920b49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-28 00:18:43.000000000","message":"Patch Set 20:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/20/experimental/kolla-build-debian-source-arm64/7e54e5b/ : FAILURE in 33m 30s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"6cbe95c301b6d2aacf8333419a5b369898e4e3a3","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2018-06-28 00:35:05.000000000","message":"Patch Set 20:\n\ncheck experimental","accounts_in_message":[],"_revision_number":20},{"id":"83ee0c1804c230fc53a375466c2e1f3a2e824828","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-28 01:05:44.000000000","message":"Patch Set 20:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/20/experimental/kolla-build-debian-source-arm64/f0db2db/ : FAILURE in 29m 57s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"c694e19017770d7bf14512299a78b1f786c5d047","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2018-06-28 01:11:10.000000000","message":"Patch Set 20:\n\ncheck experimental","accounts_in_message":[],"_revision_number":20},{"id":"0ec47ecd310797f78f4b975526a8361a8a7033f0","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2018-06-28 01:23:26.000000000","message":"Patch Set 20:\n\ncheck experimental","accounts_in_message":[],"_revision_number":20},{"id":"f03089062cc109f6dc47cfc79ca0d247aba54bd1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-28 01:42:03.000000000","message":"Patch Set 20:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/20/experimental/kolla-build-debian-source-arm64/113fa7f/ : FAILURE in 30m 10s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"0ec0f26ad93520150bd77979664a25f1603f34cd","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2018-06-28 09:30:20.000000000","message":"Patch Set 21: Patch Set 20 was rebased","accounts_in_message":[],"_revision_number":21},{"id":"42b691267b920a943314d2b0d4c3cd643e01a7ba","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2018-06-28 09:30:55.000000000","message":"Patch Set 21:\n\ncheck experimental","accounts_in_message":[],"_revision_number":21},{"id":"d6de93dd08c4367c3db9ce0443127f1aa0a34ef2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-28 10:03:52.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/21/check/openstack-tox-pep8/242c84f/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/21/check/openstack-tox-py27/ad63816/ : SUCCESS in 3m 24s\n- openstack-tox-py35 http://logs.openstack.org/59/557659/21/check/openstack-tox-py35/709e418/ : SUCCESS in 3m 39s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/557659/21/check/build-openstack-sphinx-docs/12c7817/html/ : SUCCESS in 3m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/21/check/openstack-tox-lower-constraints/f93e97b/ : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":21},{"id":"12f3d339273f3cc14a09db8b817e9882a130ccca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-28 10:06:02.000000000","message":"Patch Set 21:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/21/experimental/kolla-build-debian-source-arm64/7ae964a/ : FAILURE in 33m 33s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"6bbfc3f1265a78f524eb64416bc45161f80fcb42","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-06-29 07:20:18.000000000","message":"Patch Set 21: Workflow-1\n\ncheck experimental","accounts_in_message":[],"_revision_number":21},{"id":"279dfbae9441fe88b5df590e1553695abb3b2aac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-29 08:00:59.000000000","message":"Patch Set 21:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/21/experimental/kolla-build-debian-source-arm64/0a02942/ : FAILURE in 35m 45s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"c1e4b7bd22d7115f6c22cc7ba35979a598b26dfe","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2018-06-29 08:07:18.000000000","message":"Patch Set 21:\n\nINFO:kolla.common.utils.base:Step 28/31 : RUN curl https://bootstrap.pypa.io/get-pip.py -o get-pip.py     \u0026\u0026 python get-pip.py     \u0026\u0026 rm get-pip.py     \u0026\u0026 pip --no-cache-dir install  --prefix\u003d\u0027/usr/local\u0027 dumb-init\u003d\u003d1.1.3     \u0026\u0026 chmod +x /usr/local/bin/dumb-init     \u0026\u0026 sed -i \u0027s|#!|#!/usr/local/bin/dumb-init |\u0027 /usr/local/bin/kolla_start\nINFO:kolla.common.utils.base: ---\u003e Running in d898daafbbcd\nINFO:kolla.common.utils.base:curl (https://bootstrap.pypa.io/get-pip.py): response: 200, time: 0.852176, size: 1642329\nINFO:kolla.common.utils.base:Looking in indexes: http://mirror.regionone.linaro.openstack.org:8080/pypi/simple\nINFO:kolla.common.utils.base:Collecting pip\nINFO:kolla.common.utils.base:\u001b[91m  Could not find a version that satisfies the requirement pip (from versions: )\nINFO:kolla.common.utils.base:\u001b[0m\nINFO:kolla.common.utils.base:\u001b[91mNo matching distribution found for pip\n\n\nCan pip use upstream directly?","accounts_in_message":[],"_revision_number":21},{"id":"3bf8d6db3da49b68b5401af6f78ac0f2fcbf618d","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-06-29 09:27:43.000000000","message":"Patch Set 21:\n\nUsing pypi mirror was set here:\nhttps://github.com/openstack/kolla/blob/master/tests/templates/template_overrides.j2#L7-L8","accounts_in_message":[],"_revision_number":21},{"id":"10084929f29f02b3c4836b1f3f2d624f73054518","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2018-06-29 13:59:02.000000000","message":"Patch Set 21:\n\ncheck experimental","accounts_in_message":[],"_revision_number":21},{"id":"c09c3010efd734ebbd8a9f9bd3a45a4de95d4bcc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-29 14:59:41.000000000","message":"Patch Set 21:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/21/experimental/kolla-build-debian-source-arm64/b9ee46f/ : FAILURE in 39m 25s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"370167c74687641a7d7a4de4619b638a164bdf33","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2018-06-30 03:02:01.000000000","message":"Patch Set 21:\n\ncheck experimental","accounts_in_message":[],"_revision_number":21},{"id":"252d7adb719876cb6760eb17ece1a7716cb69c7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-30 03:09:40.000000000","message":"Patch Set 21:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/21/experimental/kolla-build-debian-source-arm64/70baa81/ : RETRY_LIMIT in 7m 00s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"59f6140bcc831ba8452988bfaba5085be3d0edfd","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2018-07-11 06:23:00.000000000","message":"Patch Set 21:\n\ncheck experimental","accounts_in_message":[],"_revision_number":21},{"id":"716ccfa3c59a5aaf73c0efa823471e3e6b53a609","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-11 06:25:00.000000000","message":"Patch Set 21:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 finger://ze03.openstack.org/afbdcc332f5044b3910bbe081197b477 : RETRY_LIMIT in 22s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"127bc79ba9ad23e9d5b0bb5f3ce56328209d9b88","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2018-07-11 07:28:05.000000000","message":"Patch Set 21:\n\ncheck experimental","accounts_in_message":[],"_revision_number":21},{"id":"e5a772e2c8e12ffb0c971cb0734bd112f2281a26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-11 07:37:19.000000000","message":"Patch Set 21:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/21/experimental/kolla-build-debian-source-arm64/2f3b629/ : RETRY_LIMIT in 8m 23s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"24161662e3e846bf057bf1aac236cc3d1ed3828a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2018-07-11 09:46:14.000000000","message":"Patch Set 21:\n\ncheck experimental","accounts_in_message":[],"_revision_number":21},{"id":"679cd7c2bbd903624f58bb544189842959832cca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-11 09:54:28.000000000","message":"Patch Set 21:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/21/experimental/kolla-build-debian-source-arm64/cad2fcc/ : RETRY_LIMIT in 7m 20s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"f8ba61d3057269e835551522ad41c3cf769ccc3e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2018-07-11 10:17:48.000000000","message":"Patch Set 21:\n\nCan we do somethign to remove that RETRY_LIMIT?","accounts_in_message":[],"_revision_number":21},{"id":"990b0ff016cd4ec5caa54e782d0b558a7f98884a","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2018-07-11 10:21:02.000000000","message":"Patch Set 21:\n\nhttp://logs.openstack.org/59/557659/21/experimental/kolla-build-debian-source-arm64/cad2fcc/job-output.txt.gz#_2018-07-11_09_53_05_970737","accounts_in_message":[],"_revision_number":21},{"id":"01a8d856f41521ec311733c4d9bcfae458f46526","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-07-12 05:51:29.000000000","message":"Patch Set 21:\n\nianw: It is seems use 4.9 kernel not backports kernel, which make config-driver not work.\n\nAnyway, mirror.regionone.linaro.openstack.org has sync infra\u0027s mirror problem maybe due to firewall, which make it unusable. Still working on fixing it.","accounts_in_message":[],"_revision_number":21},{"id":"833b70cc6f0d88add32c95c1b86be787d28daf57","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2018-07-18 07:16:13.000000000","message":"Patch Set 21:\n\nlooks like collectd role from infra is not supported or at least is failing in arm64, you could try disabling the role at least to debug builds of arm64\n\nhttps://github.com/openstack/kolla/blob/master/tests/playbooks/pre.yml#L5\n\nhttps://github.com/openstack/kolla/blob/master/tests/playbooks/post.yml#L4","accounts_in_message":[],"_revision_number":21},{"id":"393842fb370a9ba6f55d5a2f904d4570ad899977","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2018-09-25 16:40:50.000000000","message":"Patch Set 21:\n\nwhats the state of this PS?","accounts_in_message":[],"_revision_number":21},{"id":"fd63e0ce11c59c4f00b7b502e69467a3a1a62cce","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-09-26 01:08:33.000000000","message":"Patch Set 21:\n\nThe previous infra node has problem with networking which blocking things. Ianw is adding new cloud infra node, so that we could restart working on this job.","accounts_in_message":[],"_revision_number":21},{"id":"46806deadb5e29e1c8aa2e68ef7e33ee753ff3cd","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-09-27 09:18:25.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"dd9f934dc36c969a573e72f6b40e4f87091eb92d","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-09-27 09:19:03.000000000","message":"Patch Set 22:\n\ncheck experimental","accounts_in_message":[],"_revision_number":22},{"id":"6f686a10eed805b2978ceb75fb94479a761e489a","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-09-27 09:19:17.000000000","message":"Patch Set 22: Workflow-1\n\ncheck experimental","accounts_in_message":[],"_revision_number":22},{"id":"6f7bbed75ae36c3ed048c7ede075f1bee0a193d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-27 09:26:38.000000000","message":"Patch Set 22:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/22/experimental/kolla-build-debian-source-arm64/ad5b4cf/ : RETRY_LIMIT in 6m 20s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"6e4d0e6d2258fb0e22b2b531797c6d2e3357e320","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-09-27 09:44:27.000000000","message":"Patch Set 22:\n\ncheck experimental","accounts_in_message":[],"_revision_number":22},{"id":"1ae6e86fec7b55c8ad7e6d518915c914a6a681f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-27 10:17:36.000000000","message":"Patch Set 22:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/22/experimental/kolla-build-debian-source-arm64/7f93ada/ : FAILURE in 32m 07s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"9c818b274a95a13a4ec37a0d45342af8ff1f4a47","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-09-27 10:43:17.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"d31df3cfa6bf8cbb50dfbc2238f2592be5fecf67","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-09-27 10:44:11.000000000","message":"Patch Set 23: Workflow-1\n\ncheck experimental","accounts_in_message":[],"_revision_number":23},{"id":"b9c7a19014d916a87b5594a9a794a8c7ebb9a0b5","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-09-27 12:11:34.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"5c0f6c43313c98ee435c466cfe6a63b074bd3718","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-09-27 12:17:49.000000000","message":"Patch Set 24:\n\ncheck experimental","accounts_in_message":[],"_revision_number":24},{"id":"ddf6e6c0ebd88025cf79a63205405f2d625b05b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-27 14:34:45.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/24/check/openstack-tox-lower-constraints/070895c/ : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":24},{"id":"a1c2c9f63ccadb837fada6f7fe12f84ae6a84788","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-27 15:16:19.000000000","message":"Patch Set 24:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/24/experimental/kolla-build-debian-source-arm64/0c15810/ : TIMED_OUT in 2h 50m 47s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"376771528dfd883b24d876b8b44bad9669b77819","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-09-28 07:31:59.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"08f73f116ce715250cec88635c0e16e6f79081a7","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-09-28 07:32:47.000000000","message":"Patch Set 25: Workflow-1\n\ncheck experimental","accounts_in_message":[],"_revision_number":25},{"id":"c0dae907cd55d687bf26123240df6cd7c3ac8332","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-28 07:58:21.000000000","message":"Patch Set 25:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/25/experimental/kolla-build-debian-source-arm64/ff00a12/ : FAILURE in 16m 12s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"003834e0197069c17f89e0b4ecab7ed0f955dc28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-28 07:58:53.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/25/check/openstack-tox-lower-constraints/59d2950/ : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":25},{"id":"67b265fb19d6bf91a382f8c53a12168aac8970bf","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-09-28 10:12:45.000000000","message":"Patch Set 25:\n\ncheck experimental","accounts_in_message":[],"_revision_number":25},{"id":"81bd23e9eb3d898e797f56a6042cb43c0eed0230","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-28 10:20:25.000000000","message":"Patch Set 25:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/25/experimental/kolla-build-debian-source-arm64/5bbd663/ : RETRY_LIMIT in 6m 56s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"0e3e3a34993146dc5421d309ee63d76d6f7d9223","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-09-28 10:24:10.000000000","message":"Patch Set 25:\n\ncheck experimental","accounts_in_message":[],"_revision_number":25},{"id":"7bcabc516e24e768cd4bb30ca6142db9901d863b","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-09-28 10:30:35.000000000","message":"Patch Set 25:\n\ncheck experimental","accounts_in_message":[],"_revision_number":25},{"id":"f411ff6bcb4dcd0a568cce16a756c8627ef115c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-28 10:30:42.000000000","message":"Patch Set 25:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/25/experimental/kolla-build-debian-source-arm64/b0d8134/ : RETRY_LIMIT in 5m 37s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"17aded099d7dfe94e1e94973ab3b732dd08a30bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-28 10:54:11.000000000","message":"Patch Set 25:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/25/experimental/kolla-build-debian-source-arm64/f1088a3/ : FAILURE in 16m 26s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"19f007522e9e04e5cf0f1c9434a499450218f86d","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-09-28 10:57:46.000000000","message":"Patch Set 25:\n\ncheck experimental","accounts_in_message":[],"_revision_number":25},{"id":"407064657fca7fb274bae88551e78b9fe94c8c7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-28 11:16:20.000000000","message":"Patch Set 25:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/25/experimental/kolla-build-debian-source-arm64/2f54150/ : FAILURE in 17m 35s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"7ab4f1a6e400aff8abd80822453d763d4cb6014f","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-09-29 00:09:37.000000000","message":"Patch Set 25:\n\ncheck experimental","accounts_in_message":[],"_revision_number":25},{"id":"83f28289bd6ddc6fe4a7dc726e72f0f2304415ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-29 00:36:06.000000000","message":"Patch Set 25:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/25/experimental/kolla-build-debian-source-arm64/b35627f/ : FAILURE in 16m 26s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"12c8632ffe4957136f319ba4fac0f72db08ff3e3","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-09-29 00:44:03.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"bb01274a64c9899c07ad1b14d069c1b8cd9af638","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-09-29 00:44:58.000000000","message":"Patch Set 26: Workflow-1\n\ncheck experimental","accounts_in_message":[],"_revision_number":26},{"id":"4eef49e0f6c39974d5c6e3ffa665327ece1d9298","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-29 01:10:23.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/26/check/openstack-tox-lower-constraints/fa11c91/ : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":26},{"id":"835b850dd229bc1c493db110d0dd0e67064e05e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-29 03:15:30.000000000","message":"Patch Set 26:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/26/experimental/kolla-build-debian-source-arm64/458886d/ : FAILURE in 2h 23m 24s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"e9c02b12313214d39baa1ef1e6dada5cb78742e2","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-09-29 07:52:35.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"ed5976ae09a3ef769781095bf6ec4a809c776ab4","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-09-29 07:53:49.000000000","message":"Patch Set 27: Workflow-1\n\ncheck experimental","accounts_in_message":[],"_revision_number":27},{"id":"233bfd0b95e414daf14ed689fd032669a1f3c80e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-29 07:56:47.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/27/check/openstack-tox-lower-constraints/800452d/ : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":27},{"id":"e94365ec221e3f02d3ec91da4f772bffe3564757","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-29 08:01:22.000000000","message":"Patch Set 27:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/27/experimental/kolla-build-debian-source-arm64/12dc73a/ : RETRY_LIMIT in 7m 04s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"f09b8b147c16eef9e7201b2a6da7624cb668dd31","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-08 08:41:33.000000000","message":"Patch Set 27:\n\ncheck experimental","accounts_in_message":[],"_revision_number":27},{"id":"51e824cb4cd8bedeecf7af4932b6dd067c3e7685","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-08 08:48:34.000000000","message":"Patch Set 27:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/27/experimental/kolla-build-debian-source-arm64/9191980/ : RETRY_LIMIT in 6m 20s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"0c2c94fdd33a0ca449b226c8f8be51bd7c74c131","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-08 09:06:22.000000000","message":"Patch Set 27:\n\ncheck experimental","accounts_in_message":[],"_revision_number":27},{"id":"25e77979446ff26ab717288d2c3ce491ec7e46a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-08 09:19:22.000000000","message":"Patch Set 27:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/27/experimental/kolla-build-debian-source-arm64/b7e7c66/ : RETRY_LIMIT in 5m 17s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"814120d741d0cb5b308aad1328d1a6bfbe8d34d4","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-08 09:25:02.000000000","message":"Patch Set 27:\n\ncheck experimental","accounts_in_message":[],"_revision_number":27},{"id":"fbf2102c45cb9b432f5de41e3a305a8a3c07e9db","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-08 09:25:39.000000000","message":"Patch Set 27:\n\ncheck experimental","accounts_in_message":[],"_revision_number":27},{"id":"137b7e627b12137623c46ddf871f2504faae9f86","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2018-10-08 09:26:23.000000000","message":"Patch Set 27:\n\nFailed on this http://logs.openstack.org/59/557659/27/experimental/kolla-build-debian-source-arm64/b7e7c66/job-output.txt.gz#_2018-10-08_09_17_32_695569","accounts_in_message":[],"_revision_number":27},{"id":"5a508ea41b78106886786255355806bf3336ef16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-08 09:26:44.000000000","message":"Patch Set 27:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 finger://ze09.openstack.org/764cadba0dcf470391961e1775a225e3 : RETRY_LIMIT in 1m 00s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"390754426abac092e7612c17120c3905dfec0cbe","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-08 09:28:19.000000000","message":"Patch Set 27:\n\ncheck experimental","accounts_in_message":[],"_revision_number":27},{"id":"20f2cf1bdadc57435382831e0f2be4e588b6eaf0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-08 09:42:23.000000000","message":"Patch Set 27:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/27/experimental/kolla-build-debian-source-arm64/8e84994/ : RETRY_LIMIT in 5m 43s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"934b277edb3031ab30a3fe6ec945745c4e6e78bb","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-08 09:53:24.000000000","message":"Patch Set 27:\n\ncheck experimental","accounts_in_message":[],"_revision_number":27},{"id":"39034dcbfe474e146d8d28e30ce82d11dbcfbe77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-08 12:55:18.000000000","message":"Patch Set 27:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/27/experimental/kolla-build-debian-source-arm64/3bd727b/ : TIMED_OUT in 2h 53m 56s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"8dd776eb05a0941b22e5b069117c064a4237098b","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-24 04:07:07.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"c561a46c0e7fedfe74c2a3983f716454af18639d","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-24 04:08:05.000000000","message":"Patch Set 28: Workflow-1\n\ncheck experimental","accounts_in_message":[],"_revision_number":28},{"id":"b44d92c13ea822a71033bc7dfc31a1d42879dfc5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-24 04:16:06.000000000","message":"Patch Set 28:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/28/experimental/kolla-build-debian-source-arm64/b5b5355/ : RETRY_LIMIT in 7m 25s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"c01a42f1c933efbcc60836d0811323e963ef2aca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-24 04:16:56.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/28/check/openstack-tox-lower-constraints/f2eca2e/ : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":28},{"id":"09523fdb7658aefa552bbfd4219cc6d46cd8a8e4","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-24 05:38:36.000000000","message":"Patch Set 28:\n\ncheck experimental","accounts_in_message":[],"_revision_number":28},{"id":"20a01758e444801515bbb7a262e5e04a5e9952cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-24 06:11:13.000000000","message":"Patch Set 28:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/28/experimental/kolla-build-debian-source-arm64/4218252/ : FAILURE in 23m 53s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"35069765a4aca638945edec10cf6f40b57d8d746","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-24 08:27:04.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"a051d2b1e7028d3031c1b2283efcb7615d91b6e0","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-24 08:27:56.000000000","message":"Patch Set 29: Workflow-1\n\ncheck experimental","accounts_in_message":[],"_revision_number":29},{"id":"0977c54e8e3cc4d05ba763cc196c5fca41c9e4ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-24 08:35:15.000000000","message":"Patch Set 29:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/29/experimental/kolla-build-debian-source-arm64/734771a/ : RETRY_LIMIT in 6m 31s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"2fdced7dfab393be7fef69ca1e69ed388ab7e2af","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-24 08:36:01.000000000","message":"Patch Set 29:\n\ncheck experimental","accounts_in_message":[],"_revision_number":29},{"id":"8809c8ffd46405251b2b4acbcae60a018ba9b6d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-24 09:11:41.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/29/check/openstack-tox-lower-constraints/2caab67/ : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":29},{"id":"f5da27d2ed13a42814186db3ca2427ac5f82a032","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-24 11:57:40.000000000","message":"Patch Set 29:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/29/experimental/kolla-build-debian-source-arm64/baa578f/ : TIMED_OUT in 2h 53m 04s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"fb4f3735074d5e151f5577b18169b2572bb6ece9","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-25 01:57:08.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"93729bcb2e8918517cfbee0ff1496221304e138b","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-25 01:58:31.000000000","message":"Patch Set 30: Workflow-1\n\ncheck experimental","accounts_in_message":[],"_revision_number":30},{"id":"c83f980ece357502471629eefeb278648a147223","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-25 02:02:14.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/30/check/openstack-tox-lower-constraints/8b6fd9c/ : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":30},{"id":"367eda4306cddd92660999039ea86324ac9f30d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-25 02:06:37.000000000","message":"Patch Set 30:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/30/experimental/kolla-build-debian-source-arm64/d9d248c/ : RETRY_LIMIT in 7m 29s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"e45a77a69db765c6479787c3652211debcddba72","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-25 02:07:42.000000000","message":"Patch Set 30:\n\ncheck experimental","accounts_in_message":[],"_revision_number":30},{"id":"a8f3efbfd8b21d19c3355cec9b64cbd015655e9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-25 05:21:07.000000000","message":"Patch Set 30:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/30/experimental/kolla-build-debian-source-arm64/d419b9b/ : TIMED_OUT in 2h 52m 54s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"efcc178d7347efad95d0297b6a1d9e2bfe22e5cf","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-25 06:21:38.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"4a8f51f23a43635746e2c30a11bb6a6c4b89198b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-25 06:54:54.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/31/check/openstack-tox-lower-constraints/c2f5e68/ : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":31},{"id":"3b4d3dd80a10e7e2e8ca5332026c3b42cc4731dc","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-25 07:18:39.000000000","message":"Patch Set 31: Workflow-1\n\ncheck experimental","accounts_in_message":[],"_revision_number":31},{"id":"db67dbeb82a14974c1d6dc4dc13966d89eb44aa2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-25 10:27:49.000000000","message":"Patch Set 31:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/31/experimental/kolla-build-debian-source-arm64/7fde424/ : TIMED_OUT in 2h 53m 25s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"c9571be7cae4fd0347d0320911b9ce187bc87f8f","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-26 06:23:12.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"a9807afd939cb53d3cf3eed004c44613bd48867e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-26 06:23:27.000000000","message":"Patch Set 32: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  The job \"kolla-base\" exceeds tenant max-job-timeout 10800.\n\nThe error appears in the following job stanza:\n\n  job:\n      name: kolla-base\n      parent: base\n      timeout: 86400\n      post-timeout: 4800\n      pre-run: tests/playbooks/pre.yml\n      run: tests/playbooks/run.yml\n      post-run: tests/playbooks/post.yml\n      attempts: 1\n      required-projects:\n        - openstack/kolla-ansible\n        - openstack/requirements\n      irrelevant-files:\n        - ^.*\\.rst$\n        - ^doc/.*\n      roles:\n        - zuul: openstack-infra/zuul-jobs\n      vars:\n        publisher: false\n\n  in \"openstack/kolla/.zuul.d/base.yaml@master\", line 46, column 3","accounts_in_message":[],"_revision_number":32},{"id":"6343e8702d1be01b574f6e133008dbe2ba0d0431","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-26 06:25:07.000000000","message":"Patch Set 32: Workflow-1\n\ncheck experimental","accounts_in_message":[],"_revision_number":32},{"id":"9199f225560d63562bf318393c56c2857d71d899","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-26 06:33:04.000000000","message":"Patch Set 32:\n\ncheck experimental","accounts_in_message":[],"_revision_number":32},{"id":"1cc2a7799b4d46dad318c4b699da8f08af17da56","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-26 06:35:08.000000000","message":"Patch Set 32:\n\ncheck experimental","accounts_in_message":[],"_revision_number":32},{"id":"7651c51f5ceb079cdcbbfe1833278c9db5e3bf83","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-26 07:02:00.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"e207d6c47786e184eb74cd63b6f8ffb0dbe5c613","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-26 08:09:19.000000000","message":"Patch Set 31:\n\ncheck experimental","accounts_in_message":[],"_revision_number":31},{"id":"a5f890ae0bee5736389bbf6d6c08d79386cad9ae","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-26 08:11:37.000000000","message":"Patch Set 33: Workflow-1\n\ncheck experimental","accounts_in_message":[],"_revision_number":33},{"id":"09915998de5d357cc9913c1f8d72d3988ba807b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-26 08:18:08.000000000","message":"Patch Set 31:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/31/experimental/kolla-build-debian-source-arm64/cc9918d/ : RETRY_LIMIT in 7m 16s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"41be50bcfa4a09aa6633aff0eac33d3dc13e5686","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-26 08:23:26.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/33/check/openstack-tox-lower-constraints/a1bec64/ : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":33},{"id":"329abebc4085906e37f42c2e81f1cebf99219330","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-26 08:40:59.000000000","message":"Patch Set 33:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/33/experimental/kolla-build-debian-source-arm64/1d50a59/ : RETRY_LIMIT in 5m 16s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"cca1bf1a77633389ca357d8d38795a7e8800173e","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-26 08:44:56.000000000","message":"Patch Set 33:\n\ncheck experimental","accounts_in_message":[],"_revision_number":33},{"id":"24cb6166d516f6169553d5cb0030cf189a06880c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-26 11:41:44.000000000","message":"Patch Set 33:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/33/experimental/kolla-build-debian-source-arm64/ce7474c/ : SUCCESS in 2h 55m 01s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"b92933611f43ad4021a356f95cb08c2e83c53997","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-27 06:22:12.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"f05b4b20f5a6682e9236b9db14206c579dc54e7a","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2018-10-27 06:25:06.000000000","message":"Patch Set 34:\n\ncheck experimental","accounts_in_message":[],"_revision_number":34},{"id":"1b0d631a0a095a8935da049a7c0a17151db7ab72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-27 08:06:29.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/59/557659/34/check/kolla-build-centos-binary/e807fa9/ : SUCCESS in 52m 01s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/59/557659/34/check/kolla-build-centos-source/6bf505f/ : SUCCESS in 54m 03s\n- kolla-ansible-centos-source http://logs.openstack.org/59/557659/34/check/kolla-ansible-centos-source/de35d1c/ : SUCCESS in 1h 13m 27s\n- kolla-ansible-centos-binary http://logs.openstack.org/59/557659/34/check/kolla-ansible-centos-binary/fb70837/ : SUCCESS in 1h 07m 48s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/59/557659/34/check/kolla-build-oraclelinux-binary/dd08cf2/ : SUCCESS in 1h 16m 09s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/59/557659/34/check/kolla-build-oraclelinux-source/7ae62e6/ : SUCCESS in 53m 24s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/59/557659/34/check/kolla-ansible-oraclelinux-source/0102080/ : SUCCESS in 1h 09m 58s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/59/557659/34/check/kolla-ansible-oraclelinux-binary/c28688d/ : SUCCESS in 1h 03m 20s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/34/check/openstack-tox-lower-constraints/a5f8486/ : SUCCESS in 3m 49s\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/34/check/openstack-tox-pep8/9910b55/ : SUCCESS in 3m 38s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/34/check/openstack-tox-py27/63a3376/ : SUCCESS in 3m 34s\n- openstack-tox-py35 http://logs.openstack.org/59/557659/34/check/openstack-tox-py35/57aef45/ : SUCCESS in 3m 40s\n- openstack-tox-py36 http://logs.openstack.org/59/557659/34/check/openstack-tox-py36/5887d90/ : SUCCESS in 3m 12s\n- openstack-tox-docs http://logs.openstack.org/59/557659/34/check/openstack-tox-docs/c3bae06/html/ : SUCCESS in 2m 52s\n- kolla-build-ubuntu-binary http://logs.openstack.org/59/557659/34/check/kolla-build-ubuntu-binary/545444b/ : SUCCESS in 1h 25m 20s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/59/557659/34/check/kolla-build-ubuntu-source/eee5a9d/ : SUCCESS in 33m 10s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/59/557659/34/check/kolla-ansible-ubuntu-source/4deb7c6/ : SUCCESS in 1h 18m 55s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/59/557659/34/check/kolla-ansible-ubuntu-binary/5eb0d4c/ : FAILURE in 1h 30m 41s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"b6da28f0c2e17353384996ed525e81452036d7a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-27 09:36:49.000000000","message":"Patch Set 34:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/34/experimental/kolla-build-debian-source-arm64/00f4a20/ : SUCCESS in 3h 02m 41s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"8c348e5435eec9e0cace935b205778f7290bd351","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2018-11-23 10:36:14.000000000","message":"Patch Set 34:\n\nis it ready for merge or not yet?","accounts_in_message":[],"_revision_number":34},{"id":"65a2ad2fd19b0c27e820ec5938f4a73b3d84daae","author":{"_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"},"date":"2018-11-27 02:47:18.000000000","message":"Patch Set 34: Code-Review+1","accounts_in_message":[],"_revision_number":34},{"id":"beee9cdd51a099ee5a26d89ed48d1811749ca969","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2018-11-27 08:39:36.000000000","message":"Patch Set 34:\n\ncheck experimental","accounts_in_message":[],"_revision_number":34},{"id":"1b06fcb15f0b4865a6d6728ea845d8c5bd323614","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2018-11-27 08:45:28.000000000","message":"Patch Set 34:\n\n(3 comments)","accounts_in_message":[],"_revision_number":34},{"id":"9011d82109fb6eebc2a337f0e37f785b478af1d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-27 09:30:29.000000000","message":"Patch Set 34:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/34/experimental/kolla-build-debian-source-arm64/110d154/ : FAILURE in 33m 39s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"2e661f34c0d905b5b70df44be0918226a46bb7ee","author":{"_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"},"date":"2019-01-05 02:17:21.000000000","message":"Removed Code-Review+1 by Yongjun Bai \u003cbaiyj666@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":34},{"id":"4b4634bf4dc530f86a31056c725c97330edc8af5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-01-14 08:54:26.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"6ad013033a25eb6e0f849d52b7d32ed1a31c44fd","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-01-14 08:54:46.000000000","message":"Patch Set 35:\n\ncheck experimental","accounts_in_message":[],"_revision_number":35},{"id":"a5a53dfdf8acb6c3a27b16dd83627c0041cd30fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-14 10:56:31.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/59/557659/35/check/kolla-build-centos-binary/87ec9d3/ : SUCCESS in 1h 00m 01s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/59/557659/35/check/kolla-build-centos-source/f3a3a31/ : SUCCESS in 1h 07m 41s\n- kolla-ansible-centos-source http://logs.openstack.org/59/557659/35/check/kolla-ansible-centos-source/7499318/ : SUCCESS in 1h 20m 22s\n- kolla-ansible-centos-binary http://logs.openstack.org/59/557659/35/check/kolla-ansible-centos-binary/7c58d5d/ : SUCCESS in 1h 04m 58s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/59/557659/35/check/kolla-build-oraclelinux-binary/adb38a7/ : SUCCESS in 56m 12s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/59/557659/35/check/kolla-build-oraclelinux-source/4dc646c/ : SUCCESS in 1h 08m 44s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/59/557659/35/check/kolla-ansible-oraclelinux-source/d73a1b2/ : SUCCESS in 1h 10m 08s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/59/557659/35/check/kolla-ansible-oraclelinux-binary/8916d8d/ : SUCCESS in 1h 05m 38s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/35/check/openstack-tox-lower-constraints/dc2bb97/ : SUCCESS in 3m 39s\n- openstack-tox-py36 http://logs.openstack.org/59/557659/35/check/openstack-tox-py36/edcee0c/ : SUCCESS in 3m 42s\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/35/check/openstack-tox-pep8/4dce9e3/ : SUCCESS in 3m 38s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/35/check/openstack-tox-py27/6362a50/ : SUCCESS in 3m 39s\n- openstack-tox-py35 http://logs.openstack.org/59/557659/35/check/openstack-tox-py35/12129a3/ : SUCCESS in 3m 38s\n- openstack-tox-docs http://logs.openstack.org/59/557659/35/check/openstack-tox-docs/2f4ac19/html/ : SUCCESS in 3m 12s\n- kolla-build-ubuntu-binary http://logs.openstack.org/59/557659/35/check/kolla-build-ubuntu-binary/68909fb/ : SUCCESS in 1h 15m 54s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/59/557659/35/check/kolla-build-ubuntu-source/32ee1ed/ : SUCCESS in 35m 13s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/59/557659/35/check/kolla-ansible-ubuntu-source/c1c9b55/ : SUCCESS in 1h 01m 59s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/59/557659/35/check/kolla-ansible-ubuntu-binary/57481fc/ : TIMED_OUT in 2h 01m 12s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"c3b06529410b06e5707d5c50b009491c12dcbd9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-14 12:16:07.000000000","message":"Patch Set 35:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/35/experimental/kolla-build-debian-source-arm64/4dc19f8/ : TIMED_OUT in 3h 03m 12s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"e104cbb2ca21016f8885337f30ce1bc44d285e96","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-02-06 15:23:46.000000000","message":"Patch Set 35:\n\ncheck experimental","accounts_in_message":[],"_revision_number":35},{"id":"973abfbdfb95b69980f5d346b184a1cad3809a26","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-02-06 15:27:57.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"6f792882dc93636b75083cbb9f74ff1de971eab6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-06 17:03:30.000000000","message":"Patch Set 35:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/35/experimental/kolla-build-debian-source-arm64/a512f37/ : FAILURE in 55m 23s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"538ee3def2a7f88ca0734ba1cc9d19e9ebe90d57","author":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"date":"2019-02-12 10:43:20.000000000","message":"Patch Set 35: Code-Review+1\n\ncan we land this, seems pretty good in shape.","accounts_in_message":[],"_revision_number":35},{"id":"e5a5af685a1ca00c3e09abb36cba36ae4f5dddf8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-02-18 15:17:22.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"16e8d4b713f95f126e398b12d3d2c298dd71de89","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-03-06 15:55:06.000000000","message":"Patch Set 35: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":35},{"id":"0a613ab4f178e5d9ff3beb4e7b5f3686b7341685","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-06 15:55:20.000000000","message":"Patch Set 35:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":35},{"id":"9c8137962e6960b0931cc7b45a457d06b1102f83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-06 15:55:21.000000000","message":"Patch Set 35: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":35},{"id":"d6f3bc6712a457e948f8d63c5cefe3501b127d93","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-03-06 15:57:57.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"b52d2f2f3681df6b51c4c4160c525201b966e31c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-06 17:34:06.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/59/557659/36/check/kolla-build-centos-binary/8703221/ : SUCCESS in 25m 33s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/59/557659/36/check/kolla-build-centos-source/f190d88/ : SUCCESS in 35m 32s\n- kolla-ansible-centos-source http://logs.openstack.org/59/557659/36/check/kolla-ansible-centos-source/69208d4/ : SUCCESS in 1h 19m 47s\n- kolla-ansible-centos-binary http://logs.openstack.org/59/557659/36/check/kolla-ansible-centos-binary/de4dd76/ : SUCCESS in 1h 10m 48s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/59/557659/36/check/kolla-build-oraclelinux-binary/c8043cf/ : SUCCESS in 55m 44s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/59/557659/36/check/kolla-build-oraclelinux-source/9a497c0/ : SUCCESS in 50m 01s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/59/557659/36/check/kolla-ansible-oraclelinux-source/9705679/ : SUCCESS in 1h 16m 28s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/59/557659/36/check/kolla-ansible-oraclelinux-binary/2691ea6/ : SUCCESS in 1h 11m 07s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/36/check/openstack-tox-lower-constraints/89d2b11/ : SUCCESS in 3m 59s\n- openstack-tox-py36 http://logs.openstack.org/59/557659/36/check/openstack-tox-py36/091426d/ : SUCCESS in 4m 01s\n- openstack-tox-py37 http://logs.openstack.org/59/557659/36/check/openstack-tox-py37/c4c848d/ : SUCCESS in 3m 31s\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/36/check/openstack-tox-pep8/6f613ec/ : SUCCESS in 3m 28s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/36/check/openstack-tox-py27/93ab92c/ : SUCCESS in 3m 57s\n- openstack-tox-docs http://logs.openstack.org/59/557659/36/check/openstack-tox-docs/0889177/html/ : SUCCESS in 3m 57s\n- kolla-build-ubuntu-binary http://logs.openstack.org/59/557659/36/check/kolla-build-ubuntu-binary/69f9b4e/ : SUCCESS in 1h 01m 27s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/59/557659/36/check/kolla-build-ubuntu-source/3758b99/ : SUCCESS in 31m 47s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/59/557659/36/check/kolla-ansible-ubuntu-source/5fadee3/ : SUCCESS in 54m 54s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/59/557659/36/check/kolla-ansible-ubuntu-binary/e6a5c32/ : SUCCESS in 1h 26m 47s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"f3519d4f2068d5e91dafcd6d611ba52d9591325e","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-03-07 08:28:54.000000000","message":"Patch Set 36: Code-Review+1\n\ncheck experimental","accounts_in_message":[],"_revision_number":36},{"id":"250ca297f579651ff49b872aa45348db487619a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-07 11:55:17.000000000","message":"Patch Set 36:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/36/experimental/kolla-build-debian-source-arm64/9251b53/ : TIMED_OUT in 3h 03m 55s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"fb1df97c860369e8d46c5938f175fb9ec7acc9d1","author":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"date":"2019-03-27 05:04:25.000000000","message":"Patch Set 36: Code-Review+1\n\nseems need to rebase, and job is getting timeout not sure if it is because of network or need to tune it from jib side.\nrest seems fine up to an extent.","accounts_in_message":[],"_revision_number":36},{"id":"0d4203b412e695b25b2ffcc948d286a6532fd1c3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-03-27 16:57:06.000000000","message":"Patch Set 36: Workflow-1 Review-Priority-1\n\nWe (Linaro) are in progress of moving from China devcloud elsewhere. Will update once it is finished.\n\nAlso not material for Stein due to feature freeze.","accounts_in_message":[],"_revision_number":36},{"id":"d8927b39db09d52598057b4d252d71d028baf2ba","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-04-24 15:29:51.000000000","message":"Patch Set 36:\n\ncheck experimental","accounts_in_message":[],"_revision_number":36},{"id":"de07cee12fc9ee34d9f69a6de3caf057f1e7dc1c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-04-24 15:49:36.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"c32d7b895ab863a9080dc1395da7bbcc17d737a5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-04-24 15:50:23.000000000","message":"Patch Set 37: -Review-Priority\n\nmachine should be in infra now","accounts_in_message":[],"_revision_number":37},{"id":"fcb5c2069b561c034c1f56f807067d201bd32cb9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-04-24 15:57:58.000000000","message":"Patch Set 37:\n\ncheck experimental","accounts_in_message":[],"_revision_number":37},{"id":"57cc30cfd9610b22247c717aec6c93d0d10c9f9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 16:38:40.000000000","message":"Patch Set 37:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/37/experimental/kolla-build-debian-source-arm64/15cf546/ : FAILURE in 35m 34s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"f8cb383bfc5430377d1369b28a0ad8fac1c4aba3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 17:34:11.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/59/557659/37/check/kolla-build-centos-binary/38324f9/ : SUCCESS in 45m 26s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/59/557659/37/check/kolla-build-centos-source/8be8e3e/ : SUCCESS in 46m 11s\n- kolla-ansible-centos-source http://logs.openstack.org/59/557659/37/check/kolla-ansible-centos-source/f27f77c/ : SUCCESS in 1h 08m 46s\n- kolla-ansible-centos-binary http://logs.openstack.org/59/557659/37/check/kolla-ansible-centos-binary/f031709/ : SUCCESS in 1h 06m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/59/557659/37/check/kolla-ansible-centos-source-upgrade/591c2aa/ : SUCCESS in 1h 16m 21s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/59/557659/37/check/kolla-build-oraclelinux-binary/ac9da8d/ : SUCCESS in 1h 03m 13s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/59/557659/37/check/kolla-build-oraclelinux-source/70f200e/ : SUCCESS in 49m 32s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/59/557659/37/check/kolla-ansible-oraclelinux-source/752c50c/ : SUCCESS in 1h 22m 25s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/59/557659/37/check/kolla-ansible-oraclelinux-binary/f99275f/ : SUCCESS in 1h 13m 35s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/37/check/openstack-tox-lower-constraints/4f102d0/ : SUCCESS in 3m 56s\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/37/check/openstack-tox-pep8/e282184/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/37/check/openstack-tox-py27/a80963b/ : SUCCESS in 3m 11s\n- openstack-tox-py36 http://logs.openstack.org/59/557659/37/check/openstack-tox-py36/a663f34/ : SUCCESS in 3m 13s\n- openstack-tox-py37 http://logs.openstack.org/59/557659/37/check/openstack-tox-py37/498ae63/ : SUCCESS in 3m 58s\n- openstack-tox-docs http://logs.openstack.org/59/557659/37/check/openstack-tox-docs/89cb8fc/html/ : SUCCESS in 3m 25s\n- kolla-build-ubuntu-binary http://logs.openstack.org/59/557659/37/check/kolla-build-ubuntu-binary/bdd99fb/ : SUCCESS in 1h 28m 56s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/59/557659/37/check/kolla-build-ubuntu-source/576c147/ : SUCCESS in 39m 48s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/59/557659/37/check/kolla-ansible-ubuntu-source/d85bb49/ : SUCCESS in 54m 51s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/59/557659/37/check/kolla-ansible-ubuntu-binary/b350d78/ : SUCCESS in 48m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/59/557659/37/check/kolla-ansible-ubuntu-source-upgrade/a829d7d/ : SUCCESS in 59m 36s","accounts_in_message":[],"_revision_number":37},{"id":"67f25abcb00cabb79aea6b83e649cb684b61c2d1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-04-29 10:25:14.000000000","message":"Patch Set 37:\n\ncheck experimental","accounts_in_message":[],"_revision_number":37},{"id":"d87fdf7aff5967894b6475f70e5009e4f2f6aabc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 10:51:44.000000000","message":"Patch Set 37:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/37/experimental/kolla-build-debian-source-arm64/784d4c7/ : FAILURE in 23m 42s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"1da5847e76e67af514402d7c4e47ee3bb73333a5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-04-29 11:42:54.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"bec63d112cd6b24cc2087e363c1dd77d45074270","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-04-29 11:52:43.000000000","message":"Patch Set 38:\n\ncheck experimental","accounts_in_message":[],"_revision_number":38},{"id":"b42516cbb5de770cf091e3fda098be909a2ea5ba","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-29 11:59:17.000000000","message":"Patch Set 38:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/59/557659/38/openstack-check/tripleo-build-containers-fedora-28-master/3f56a4d/ : RETRY_LIMIT in 2m 32s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/59/557659/38/openstack-check/tripleo-build-containers-fedora-28-stein/ec4982b/ : RETRY_LIMIT in 2m 30s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"3a4d9a57c50b6316084348165f2b56fef43c0116","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 13:11:12.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/59/557659/38/check/kolla-build-centos-binary/158b1bc/ : SUCCESS in 44m 15s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/59/557659/38/check/kolla-build-centos-source/c514fa0/ : SUCCESS in 52m 44s\n- kolla-ansible-centos-source http://logs.openstack.org/59/557659/38/check/kolla-ansible-centos-source/f4faf24/ : SUCCESS in 1h 23m 39s\n- kolla-ansible-centos-binary http://logs.openstack.org/59/557659/38/check/kolla-ansible-centos-binary/8a65066/ : SUCCESS in 1h 04m 17s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/59/557659/38/check/kolla-ansible-centos-source-upgrade/a0c5af6/ : SUCCESS in 1h 18m 54s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/59/557659/38/check/tripleo-build-containers-centos-7/3746a00/ : SUCCESS in 41m 28s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/59/557659/38/check/kolla-build-oraclelinux-binary/8750a74/ : SUCCESS in 55m 52s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/59/557659/38/check/kolla-build-oraclelinux-source/8e1ea8e/ : SUCCESS in 47m 38s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/59/557659/38/check/kolla-ansible-oraclelinux-source/d2f980b/ : SUCCESS in 1h 18m 44s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/59/557659/38/check/kolla-ansible-oraclelinux-binary/97ceefb/ : SUCCESS in 1h 03m 32s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/38/check/openstack-tox-lower-constraints/e020897/ : SUCCESS in 3m 47s\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/38/check/openstack-tox-pep8/5b3430e/ : SUCCESS in 2m 46s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/38/check/openstack-tox-py27/68caf9c/ : SUCCESS in 5m 32s\n- openstack-tox-py36 http://logs.openstack.org/59/557659/38/check/openstack-tox-py36/ee34872/ : SUCCESS in 4m 00s\n- openstack-tox-py37 http://logs.openstack.org/59/557659/38/check/openstack-tox-py37/c69df12/ : SUCCESS in 4m 38s\n- openstack-tox-docs http://logs.openstack.org/59/557659/38/check/openstack-tox-docs/0483015/html/ : SUCCESS in 2m 53s\n- kolla-build-ubuntu-binary http://logs.openstack.org/59/557659/38/check/kolla-build-ubuntu-binary/1a472df/ : SUCCESS in 28m 16s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/59/557659/38/check/kolla-build-ubuntu-source/7629130/ : FAILURE in 48m 47s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/59/557659/38/check/kolla-ansible-ubuntu-source/f5aa210/ : SUCCESS in 50m 39s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/59/557659/38/check/kolla-ansible-ubuntu-binary/69e46de/ : SUCCESS in 44m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/59/557659/38/check/kolla-ansible-ubuntu-source-upgrade/f03833e/ : SUCCESS in 1h 23m 00s","accounts_in_message":[],"_revision_number":38},{"id":"4f6038ce8049bac53cf90061c1b072677f2e3d20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 14:44:49.000000000","message":"Patch Set 38:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/38/experimental/kolla-build-debian-source-arm64/e9687f5/ : TIMED_OUT in 2h 51m 34s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"63f43c4a769336fd59d2f6c2fa86b19b434024ad","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-04-29 14:50:33.000000000","message":"Patch Set 38:\n\nbuild went fine, timeout needs to be expanded\n\nand https://review.opendev.org/642375 merged","accounts_in_message":[],"_revision_number":38},{"id":"d43f29ee6033c1abdeac266d5ed1e7b3753cb514","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-04-30 09:26:42.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"ca783bd192260e81dedfa73aaeef6ada483d36d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 09:26:58.000000000","message":"Patch Set 39: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  The job \"kolla-base\" exceeds tenant max-job-timeout 10800.\n\nThe error appears in the following job stanza:\n\n  job:\n      name: kolla-base\n      parent: base\n      timeout: 18000\n      post-timeout: 7200\n      pre-run: tests/playbooks/pre.yml\n      run: tests/playbooks/run.yml\n      post-run: tests/playbooks/post.yml\n      attempts: 1\n      required-projects:\n        - openstack/kolla-ansible\n        - openstack/requirements\n      irrelevant-files:\n        - ^.*\\.rst$\n        - ^doc/.*\n        - ^etc/.*\n        - ^lower-constraints.txt$\n        - ^releasenotes/.*$\n        - ^specs/.*$\n        - ^test-requirements.txt$\n      vars:\n        publisher: false\n\n  in \"openstack/kolla/.zuul.d/base.yaml@master\", line 46, column 3","accounts_in_message":[],"_revision_number":39},{"id":"8eeb8fb2dc990a69bce6bff35b3aa11301941551","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-04-30 09:27:13.000000000","message":"Patch Set 39:\n\ncheck experimental","accounts_in_message":[],"_revision_number":39},{"id":"e2de90d444595474184f242d72a1bb48fb859b28","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-04-30 09:40:55.000000000","message":"Uploaded patch set 40.","accounts_in_message":[],"_revision_number":40},{"id":"e636b33b397093fb345c20d40afa787d3017ea03","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-30 09:59:43.000000000","message":"Patch Set 40:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/59/557659/40/openstack-check/tripleo-build-containers-fedora-28-master/22ac2d5/ : RETRY_LIMIT in 4m 54s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/59/557659/40/openstack-check/tripleo-build-containers-fedora-28-stein/754be6f/ : RETRY_LIMIT in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"b1a830c949ab58faf9e750e54ccbcaf8f78e26dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 11:00:19.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/59/557659/40/check/kolla-build-centos-binary/6513818/ : SUCCESS in 41m 05s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/59/557659/40/check/kolla-build-centos-source/cec3639/ : SUCCESS in 51m 06s\n- kolla-ansible-centos-source http://logs.openstack.org/59/557659/40/check/kolla-ansible-centos-source/745a711/ : SUCCESS in 1h 07m 35s\n- kolla-ansible-centos-binary http://logs.openstack.org/59/557659/40/check/kolla-ansible-centos-binary/d85770d/ : SUCCESS in 1h 00m 48s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/59/557659/40/check/kolla-ansible-centos-source-upgrade/27894f5/ : SUCCESS in 1h 11m 35s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/59/557659/40/check/tripleo-build-containers-centos-7/b4ba1db/ : SUCCESS in 40m 30s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/59/557659/40/check/kolla-build-oraclelinux-binary/d3ef251/ : SUCCESS in 47m 28s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/59/557659/40/check/kolla-build-oraclelinux-source/2a04147/ : SUCCESS in 47m 39s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/59/557659/40/check/kolla-ansible-oraclelinux-source/0ba547c/ : SUCCESS in 1h 08m 54s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/59/557659/40/check/kolla-ansible-oraclelinux-binary/87d1be6/ : SUCCESS in 1h 05m 26s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/40/check/openstack-tox-lower-constraints/5d2eb1b/ : SUCCESS in 3m 51s\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/40/check/openstack-tox-pep8/938ab9c/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/40/check/openstack-tox-py27/f0bf8cc/ : SUCCESS in 2m 53s\n- openstack-tox-py36 http://logs.openstack.org/59/557659/40/check/openstack-tox-py36/9cf00ed/ : SUCCESS in 3m 11s\n- openstack-tox-py37 http://logs.openstack.org/59/557659/40/check/openstack-tox-py37/3aecc20/ : SUCCESS in 3m 11s\n- openstack-tox-docs http://logs.openstack.org/59/557659/40/check/openstack-tox-docs/c6febd9/html/ : SUCCESS in 2m 35s\n- kolla-build-ubuntu-binary http://logs.openstack.org/59/557659/40/check/kolla-build-ubuntu-binary/664a193/ : SUCCESS in 40m 06s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/59/557659/40/check/kolla-build-ubuntu-source/aea9519/ : SUCCESS in 29m 52s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/59/557659/40/check/kolla-ansible-ubuntu-source/ade458f/ : SUCCESS in 53m 10s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/59/557659/40/check/kolla-ansible-ubuntu-binary/55557c4/ : SUCCESS in 43m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/59/557659/40/check/kolla-ansible-ubuntu-source-upgrade/85ff6a5/ : SUCCESS in 1h 07m 33s","accounts_in_message":[],"_revision_number":40},{"id":"391f394b2932bfb9c114158a5d574a8871d1d49a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-06 07:56:21.000000000","message":"Uploaded patch set 41.","accounts_in_message":[],"_revision_number":41},{"id":"863bb2b623931ce721ba6d4744192394fa4e6b51","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-06 07:56:31.000000000","message":"Patch Set 41:\n\ncheck experimental","accounts_in_message":[],"_revision_number":41},{"id":"a974eeacdd434fe440ba3262f35584e6a7ded068","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-06 07:59:04.000000000","message":"Patch Set 41:\n\n(1 comment)\n\ndropped monasca as it need lot of time (pypi modules) and we are beyond 3h limit anyway","accounts_in_message":[],"_revision_number":41},{"id":"a27993ec5a4a61c83a26e1e8229bf048c8153bab","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-06 08:11:38.000000000","message":"Patch Set 41:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/59/557659/41/openstack-check/tripleo-build-containers-fedora-28-master/9d9b162/ : RETRY_LIMIT in 2m 56s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/59/557659/41/openstack-check/tripleo-build-containers-fedora-28-stein/364aa96/ : RETRY_LIMIT in 2m 29s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"936475093f05aa9fb9ef7d7cfe6cc0b0c46e9706","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 08:12:01.000000000","message":"Patch Set 41:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/41/experimental/kolla-build-debian-source-arm64/c2bdb9c/ : SUCCESS in 14m 51s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"e6b9374ad8673d95848f9b16132b1034f75b8cfc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-06 08:19:45.000000000","message":"Uploaded patch set 42.","accounts_in_message":[],"_revision_number":42},{"id":"84d3f25e43c46942d1ada67e79ae52f473179aba","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-06 08:20:04.000000000","message":"Patch Set 42:\n\ncheck experimental","accounts_in_message":[],"_revision_number":42},{"id":"4faf49ddaf17d45ea7311ddd4d5808d0849a2020","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-06 08:32:39.000000000","message":"Patch Set 42:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/59/557659/42/openstack-check/tripleo-build-containers-fedora-28-master/77ad291/ : RETRY_LIMIT in 2m 30s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/59/557659/42/openstack-check/tripleo-build-containers-fedora-28-stein/2eb809d/ : RETRY_LIMIT in 2m 43s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"5826cf3fedadf8fcdd02c509696816710e37bc74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 10:10:51.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/59/557659/42/check/kolla-build-centos-binary/e23e3f0/ : SUCCESS in 33m 12s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/59/557659/42/check/kolla-build-centos-source/2d387ad/ : SUCCESS in 48m 46s\n- kolla-ansible-centos-source http://logs.openstack.org/59/557659/42/check/kolla-ansible-centos-source/e047d4d/ : SUCCESS in 56m 36s\n- kolla-ansible-centos-binary http://logs.openstack.org/59/557659/42/check/kolla-ansible-centos-binary/290b126/ : SUCCESS in 1h 06m 21s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/59/557659/42/check/kolla-ansible-centos-source-upgrade/ce6cb21/ : SUCCESS in 1h 34m 42s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/59/557659/42/check/tripleo-build-containers-centos-7/3a47af0/ : SUCCESS in 46m 34s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/59/557659/42/check/kolla-build-oraclelinux-binary/231dd46/ : SUCCESS in 51m 09s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/59/557659/42/check/kolla-build-oraclelinux-source/d813765/ : SUCCESS in 48m 28s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/59/557659/42/check/kolla-ansible-oraclelinux-source/c444072/ : SUCCESS in 1h 08m 53s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/59/557659/42/check/kolla-ansible-oraclelinux-binary/2c1a912/ : SUCCESS in 59m 39s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/42/check/openstack-tox-lower-constraints/2b4778e/ : SUCCESS in 3m 06s\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/42/check/openstack-tox-pep8/6c0a8d8/ : SUCCESS in 2m 57s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/42/check/openstack-tox-py27/4ed8ae1/ : SUCCESS in 2m 35s\n- openstack-tox-py36 http://logs.openstack.org/59/557659/42/check/openstack-tox-py36/b52df66/ : SUCCESS in 3m 04s\n- openstack-tox-py37 http://logs.openstack.org/59/557659/42/check/openstack-tox-py37/723f2fa/ : SUCCESS in 3m 23s\n- openstack-tox-docs http://logs.openstack.org/59/557659/42/check/openstack-tox-docs/998816d/html/ : SUCCESS in 2m 49s\n- kolla-build-ubuntu-binary http://logs.openstack.org/59/557659/42/check/kolla-build-ubuntu-binary/d6289ab/ : SUCCESS in 33m 45s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/59/557659/42/check/kolla-build-ubuntu-source/0c9f8f4/ : SUCCESS in 37m 14s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/59/557659/42/check/kolla-ansible-ubuntu-source/49a90d1/ : SUCCESS in 45m 26s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/59/557659/42/check/kolla-ansible-ubuntu-binary/f41b01c/ : SUCCESS in 48m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/59/557659/42/check/kolla-ansible-ubuntu-source-upgrade/64043e1/ : SUCCESS in 55m 10s","accounts_in_message":[],"_revision_number":42},{"id":"de254055cb6617c9c784bac0e672c30799ffe7e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 10:26:31.000000000","message":"Patch Set 42:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/42/experimental/kolla-build-debian-source-arm64/4183541/ : FAILURE in 2h 01m 02s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"bb569bcde4303e51fa49d15f132d1b6552b12996","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-06 10:43:17.000000000","message":"Uploaded patch set 43.","accounts_in_message":[],"_revision_number":43},{"id":"3c3874fee1df7de105322b4942bf468f829e3cdc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-06 10:44:58.000000000","message":"Patch Set 43:\n\ncheck experimental","accounts_in_message":[],"_revision_number":43},{"id":"e34cb82a817d9d144ed3131ef3c9561ec25d4cde","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-06 11:04:19.000000000","message":"Patch Set 43:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/59/557659/43/openstack-check/tripleo-build-containers-fedora-28-master/806478b/ : RETRY_LIMIT in 2m 29s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/59/557659/43/openstack-check/tripleo-build-containers-fedora-28-stein/65fe389/ : RETRY_LIMIT in 2m 24s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"79e6b2bc30b39907707c8fe295a533c66a93e2d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 11:13:33.000000000","message":"Patch Set 43:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 kolla-build-debian-source-arm64 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"5999147b40266e38010ef15f75f6e393c5cfd753","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-06 11:15:47.000000000","message":"Patch Set 43:\n\ncheck experimental","accounts_in_message":[],"_revision_number":43},{"id":"122d4f77195b91f405a42226042fcf9ced915ae3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 12:36:31.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/59/557659/43/check/kolla-build-centos-binary/ae4c2bf/ : SUCCESS in 1h 12m 05s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/59/557659/43/check/kolla-build-centos-source/14ac344/ : SUCCESS in 1h 23m 47s\n- kolla-ansible-centos-source http://logs.openstack.org/59/557659/43/check/kolla-ansible-centos-source/cabf63f/ : SUCCESS in 1h 32m 51s\n- kolla-ansible-centos-binary http://logs.openstack.org/59/557659/43/check/kolla-ansible-centos-binary/8325ad5/ : SUCCESS in 1h 29m 12s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/59/557659/43/check/kolla-ansible-centos-source-upgrade/eac912d/ : SUCCESS in 1h 49m 23s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/59/557659/43/check/tripleo-build-containers-centos-7/5fb81be/ : SUCCESS in 1h 05m 05s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/59/557659/43/check/kolla-build-oraclelinux-binary/9665dd3/ : SUCCESS in 1h 21m 13s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/59/557659/43/check/kolla-build-oraclelinux-source/549577b/ : SUCCESS in 49m 51s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/59/557659/43/check/kolla-ansible-oraclelinux-source/423918c/ : FAILURE in 7m 31s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/59/557659/43/check/kolla-ansible-oraclelinux-binary/59854dc/ : SUCCESS in 1h 33m 34s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/43/check/openstack-tox-lower-constraints/8f9e4ff/ : SUCCESS in 5m 02s\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/43/check/openstack-tox-pep8/d101f76/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/43/check/openstack-tox-py27/218c9a6/ : SUCCESS in 4m 23s\n- openstack-tox-py36 http://logs.openstack.org/59/557659/43/check/openstack-tox-py36/55299b4/ : SUCCESS in 3m 51s\n- openstack-tox-py37 http://logs.openstack.org/59/557659/43/check/openstack-tox-py37/04bfef7/ : SUCCESS in 4m 29s\n- openstack-tox-docs http://logs.openstack.org/59/557659/43/check/openstack-tox-docs/5646f7b/html/ : SUCCESS in 3m 07s\n- kolla-build-ubuntu-binary http://logs.openstack.org/59/557659/43/check/kolla-build-ubuntu-binary/6cbe68d/ : SUCCESS in 31m 17s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/59/557659/43/check/kolla-build-ubuntu-source/0e6a0d2/ : SUCCESS in 35m 11s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/59/557659/43/check/kolla-ansible-ubuntu-source/d2676c5/ : FAILURE in 31m 10s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/59/557659/43/check/kolla-ansible-ubuntu-binary/e2be05c/ : FAILURE in 27m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/59/557659/43/check/kolla-ansible-ubuntu-source-upgrade/89f4ecc/ : FAILURE in 50m 19s","accounts_in_message":[],"_revision_number":43},{"id":"f60f1b44262b27092ec513dc6eff8b75df51c71b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 13:13:24.000000000","message":"Patch Set 43:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/43/experimental/kolla-build-debian-source-arm64/3500f84/ : SUCCESS in 1h 57m 00s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"b6127833524814a9f7b3928ec3f6d02240d2e151","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-06 14:06:41.000000000","message":"Patch Set 43:\n\nManaged to fit into 2h.\n\nShould be ready to merge once deps get into.","accounts_in_message":[],"_revision_number":43},{"id":"d7cbb663e0b47b4238385ba30250acd1ae3a02ae","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-07 08:15:58.000000000","message":"Uploaded patch set 44.","accounts_in_message":[],"_revision_number":44},{"id":"4649a4a51a7021cb2da917bfa4a2cb64066c2e2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 08:16:12.000000000","message":"Patch Set 44: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":44},{"id":"ab2594d8f2e28178f2df7019834d07d9736eab24","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-07 08:16:14.000000000","message":"Patch Set 44:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":44},{"id":"dfe190b38d096a39cdb9beb3d6a528c22d6987a3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-07 09:08:25.000000000","message":"Uploaded patch set 45: Patch Set 44 was rebased.","accounts_in_message":[],"_revision_number":45},{"id":"0b00cca44bad55542067dbb6afc1d71b2980c87b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-07 09:08:41.000000000","message":"Patch Set 45:\n\ncheck experimental","accounts_in_message":[],"_revision_number":45},{"id":"a78aeb63538b64f88893753f9e261a4dedf8ae4c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-07 09:45:37.000000000","message":"Patch Set 45:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/59/557659/45/openstack-check/tripleo-build-containers-fedora-28-master/3691356/ : SUCCESS in 36m 15s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/59/557659/45/openstack-check/tripleo-build-containers-fedora-28-stein/04248b2/ : SUCCESS in 34m 38s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"d5477cd77bfd936aa223043ec12fe0cdf9d374fb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-07 10:02:15.000000000","message":"Uploaded patch set 46.","accounts_in_message":[],"_revision_number":46},{"id":"f80590c36683c4aa68c392cc7f88e0a529ce1b97","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-07 10:03:07.000000000","message":"Patch Set 46:\n\ncheck experimental","accounts_in_message":[],"_revision_number":46},{"id":"dd7e911b35bc5d9a554af8e76374677b58d28e68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 11:42:41.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/59/557659/46/check/kolla-build-centos-binary/f8fd6c7/ : SUCCESS in 42m 58s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/59/557659/46/check/kolla-build-centos-source/1b3e46d/ : SUCCESS in 50m 25s\n- kolla-ansible-centos-source http://logs.openstack.org/59/557659/46/check/kolla-ansible-centos-source/e4f4407/ : SUCCESS in 1h 07m 18s\n- kolla-ansible-centos-binary http://logs.openstack.org/59/557659/46/check/kolla-ansible-centos-binary/604f27e/ : SUCCESS in 1h 13m 04s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/59/557659/46/check/kolla-ansible-centos-source-upgrade/c5d9fc6/ : SUCCESS in 1h 24m 16s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/59/557659/46/check/kolla-build-oraclelinux-binary/275dc01/ : SUCCESS in 1h 02m 45s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/59/557659/46/check/kolla-build-oraclelinux-source/f0298cc/ : SUCCESS in 53m 01s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/59/557659/46/check/kolla-ansible-oraclelinux-source/ebcaba3/ : SUCCESS in 1h 08m 08s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/59/557659/46/check/kolla-ansible-oraclelinux-binary/53cdf27/ : SUCCESS in 1h 10m 34s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/46/check/openstack-tox-lower-constraints/aa9fe15/ : SUCCESS in 4m 03s\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/46/check/openstack-tox-pep8/750b388/ : SUCCESS in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/46/check/openstack-tox-py27/bf95318/ : SUCCESS in 3m 27s\n- openstack-tox-py36 http://logs.openstack.org/59/557659/46/check/openstack-tox-py36/c893cc4/ : SUCCESS in 3m 38s\n- openstack-tox-py37 http://logs.openstack.org/59/557659/46/check/openstack-tox-py37/47d0b13/ : SUCCESS in 3m 41s\n- openstack-tox-docs http://logs.openstack.org/59/557659/46/check/openstack-tox-docs/e041a04/html/ : SUCCESS in 2m 56s\n- kolla-build-ubuntu-binary http://logs.openstack.org/59/557659/46/check/kolla-build-ubuntu-binary/b2e0e09/ : SUCCESS in 26m 52s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/59/557659/46/check/kolla-build-ubuntu-source/cbfb116/ : SUCCESS in 29m 35s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/59/557659/46/check/kolla-ansible-ubuntu-source/cc52650/ : SUCCESS in 46m 28s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/59/557659/46/check/kolla-ansible-ubuntu-binary/87aae38/ : SUCCESS in 45m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/59/557659/46/check/kolla-ansible-ubuntu-source-upgrade/3a11e37/ : SUCCESS in 53m 41s","accounts_in_message":[],"_revision_number":46},{"id":"6001bbaae4ee11cdb8540be9bd4b4acd0db866ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 13:16:12.000000000","message":"Patch Set 46:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/46/experimental/kolla-build-debian-source-arm64/a8c2ce2/ : TIMED_OUT in 2h 52m 06s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"5c95f6861f3a7aeb7148baed7b85499ab7b386ba","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-14 20:36:19.000000000","message":"Patch Set 46:\n\ncheck experimental","accounts_in_message":[],"_revision_number":46},{"id":"c7de22c6ed06bb85c2907c4fe8d4f945c59890f8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-14 21:14:33.000000000","message":"Uploaded patch set 47.","accounts_in_message":[],"_revision_number":47},{"id":"cf685965e8180de2f8a406ed0a75a171ab78c09c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-14 21:14:54.000000000","message":"Patch Set 47:\n\ncheck experimental","accounts_in_message":[],"_revision_number":47},{"id":"72c645f563ce8cb4dd9f7bad9d13fe16682c2ab2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 22:43:40.000000000","message":"Patch Set 47: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/59/557659/47/check/kolla-build-centos-binary/0510992/ : SUCCESS in 42m 12s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/59/557659/47/check/kolla-build-centos-source/be9a928/ : SUCCESS in 55m 49s\n- kolla-ansible-centos-source http://logs.openstack.org/59/557659/47/check/kolla-ansible-centos-source/9cb0706/ : SUCCESS in 1h 02m 45s\n- kolla-ansible-centos-binary http://logs.openstack.org/59/557659/47/check/kolla-ansible-centos-binary/cc638b0/ : SUCCESS in 1h 01m 41s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/59/557659/47/check/kolla-ansible-centos-source-upgrade/7164827/ : SUCCESS in 1h 21m 54s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/59/557659/47/check/kolla-build-oraclelinux-binary/3d88326/ : SUCCESS in 47m 34s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/59/557659/47/check/kolla-build-oraclelinux-source/71a4e26/ : SUCCESS in 45m 38s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/59/557659/47/check/kolla-ansible-oraclelinux-source/e2817bb/ : SUCCESS in 1h 19m 13s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/59/557659/47/check/kolla-ansible-oraclelinux-binary/19d2fb7/ : SUCCESS in 1h 11m 44s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/47/check/openstack-tox-lower-constraints/59948d0/ : SUCCESS in 4m 35s\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/47/check/openstack-tox-pep8/891588b/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/47/check/openstack-tox-py27/779e5b0/ : SUCCESS in 3m 26s\n- openstack-tox-py36 http://logs.openstack.org/59/557659/47/check/openstack-tox-py36/9cf124e/ : SUCCESS in 3m 44s\n- openstack-tox-py37 http://logs.openstack.org/59/557659/47/check/openstack-tox-py37/28aba40/ : SUCCESS in 3m 40s\n- openstack-tox-docs http://logs.openstack.org/59/557659/47/check/openstack-tox-docs/4726bf0/html/ : SUCCESS in 2m 58s\n- kolla-build-ubuntu-binary http://logs.openstack.org/59/557659/47/check/kolla-build-ubuntu-binary/e8ee979/ : SUCCESS in 25m 20s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/59/557659/47/check/kolla-build-ubuntu-source/589d8df/ : SUCCESS in 32m 06s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/59/557659/47/check/kolla-ansible-ubuntu-source/be47110/ : SUCCESS in 52m 21s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/59/557659/47/check/kolla-ansible-ubuntu-binary/c085936/ : SUCCESS in 45m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/59/557659/47/check/kolla-ansible-ubuntu-source-upgrade/361e426/ : SUCCESS in 55m 28s","accounts_in_message":[],"_revision_number":47},{"id":"89844b47fc20e9d46a8027ff60223fa4e8359a0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 00:07:43.000000000","message":"Patch Set 47:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/47/experimental/kolla-build-debian-source-arm64/5a48d5e/ : TIMED_OUT in 2h 51m 57s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"72068aa063de28d78e17402428281cd52ff323b2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-15 07:07:25.000000000","message":"Patch Set 47:\n\n(3 comments)\n\nYou might want to disable the other CI tasks in the meantime.\n\nAre the provisioned arm64 hosts really so slow? Maybe save some cpuinfo to have a baseline.","accounts_in_message":[],"_revision_number":47},{"id":"bd24dccc9f0fa77b2864738fda2a56ce05756a85","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-17 09:32:08.000000000","message":"Uploaded patch set 48.","accounts_in_message":[],"_revision_number":48},{"id":"275edcfff9fa25bbb0d8875127a709cf0a1dbb38","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 11:21:12.000000000","message":"Patch Set 48: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/59/557659/48/check/kolla-build-centos-binary/dfa499c/ : SUCCESS in 40m 51s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/59/557659/48/check/kolla-build-centos-source/0cdb2f9/ : SUCCESS in 48m 27s\n- kolla-ansible-centos-source http://logs.openstack.org/59/557659/48/check/kolla-ansible-centos-source/fcdc047/ : SUCCESS in 1h 21m 48s\n- kolla-ansible-centos-binary http://logs.openstack.org/59/557659/48/check/kolla-ansible-centos-binary/94b886c/ : SUCCESS in 1h 09m 13s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/59/557659/48/check/kolla-ansible-centos-source-upgrade/490e320/ : SUCCESS in 1h 20m 48s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/59/557659/48/check/kolla-build-oraclelinux-binary/97a8160/ : SUCCESS in 45m 04s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/59/557659/48/check/kolla-build-oraclelinux-source/ce9975d/ : SUCCESS in 51m 41s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/59/557659/48/check/kolla-ansible-oraclelinux-source/b11f758/ : SUCCESS in 1h 13m 13s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/59/557659/48/check/kolla-ansible-oraclelinux-binary/fef3dfa/ : SUCCESS in 1h 02m 37s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/48/check/openstack-tox-lower-constraints/fc45a97/ : SUCCESS in 3m 38s\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/48/check/openstack-tox-pep8/5ca956c/ : SUCCESS in 2m 47s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/48/check/openstack-tox-py27/335b4cd/ : SUCCESS in 3m 12s\n- openstack-tox-py36 http://logs.openstack.org/59/557659/48/check/openstack-tox-py36/dd6dda4/ : SUCCESS in 3m 35s\n- openstack-tox-py37 http://logs.openstack.org/59/557659/48/check/openstack-tox-py37/9d53fc3/ : SUCCESS in 4m 41s\n- openstack-tox-docs http://logs.openstack.org/59/557659/48/check/openstack-tox-docs/e5334a5/html/ : SUCCESS in 3m 13s\n- kolla-build-ubuntu-binary http://logs.openstack.org/59/557659/48/check/kolla-build-ubuntu-binary/7cabf8f/ : SUCCESS in 24m 09s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/59/557659/48/check/kolla-build-ubuntu-source/3adccb8/ : SUCCESS in 32m 49s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/59/557659/48/check/kolla-ansible-ubuntu-source/ac301be/ : SUCCESS in 44m 58s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/59/557659/48/check/kolla-ansible-ubuntu-binary/3ff8435/ : SUCCESS in 43m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/59/557659/48/check/kolla-ansible-ubuntu-source-upgrade/60b3bc2/ : SUCCESS in 1h 00m 52s","accounts_in_message":[],"_revision_number":48},{"id":"6029fc2c35234054b25fe22e888ae6a541e03905","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-17 11:39:43.000000000","message":"Patch Set 48:\n\ncheck experimental","accounts_in_message":[],"_revision_number":48},{"id":"64d46ac0d1e6160ccd69da71f3af1738c51fb1ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 14:32:09.000000000","message":"Patch Set 48:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/48/experimental/kolla-build-debian-source-arm64/393586f/ : TIMED_OUT in 2h 51m 48s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"655ab9874f5f353832fe017b89fde6683e26f8a8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-28 06:12:30.000000000","message":"Uploaded patch set 49.","accounts_in_message":[],"_revision_number":49},{"id":"d5b46dd798a758f204832b3970e817728ff80000","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-28 06:12:53.000000000","message":"Patch Set 49: Workflow-1\n\ncheck experimental","accounts_in_message":[],"_revision_number":49},{"id":"bfdea91f7f9e5e54609369cafde761dd7b8d88d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 06:40:06.000000000","message":"Patch Set 49:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 kolla-build-debian-source-arm64 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"db95342e26643565927cb6b0fee6dd39ea57a271","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 07:57:29.000000000","message":"Patch Set 49: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/59/557659/49/check/kolla-build-centos-binary/0a6d052/ : SUCCESS in 39m 32s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/59/557659/49/check/kolla-build-centos-source/13bff8d/ : SUCCESS in 47m 28s\n- kolla-ansible-centos-source http://logs.openstack.org/59/557659/49/check/kolla-ansible-centos-source/7ed61f8/ : SUCCESS in 1h 10m 14s\n- kolla-ansible-centos-binary http://logs.openstack.org/59/557659/49/check/kolla-ansible-centos-binary/240a7d5/ : SUCCESS in 1h 03m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/59/557659/49/check/kolla-ansible-centos-source-upgrade/ef7a289/ : SUCCESS in 1h 20m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/49/check/openstack-tox-lower-constraints/b7c728b/ : SUCCESS in 4m 26s\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/49/check/openstack-tox-pep8/a99379a/ : SUCCESS in 3m 26s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/49/check/openstack-tox-py27/e3c6672/ : SUCCESS in 3m 17s\n- openstack-tox-py36 http://logs.openstack.org/59/557659/49/check/openstack-tox-py36/50c8deb/ : SUCCESS in 3m 37s\n- openstack-tox-py37 http://logs.openstack.org/59/557659/49/check/openstack-tox-py37/5ca1e1a/ : SUCCESS in 3m 49s\n- openstack-tox-docs http://logs.openstack.org/59/557659/49/check/openstack-tox-docs/a782a23/html/ : SUCCESS in 2m 40s\n- kolla-build-ubuntu-binary http://logs.openstack.org/59/557659/49/check/kolla-build-ubuntu-binary/46bf75f/ : SUCCESS in 21m 43s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/59/557659/49/check/kolla-build-ubuntu-source/f646063/ : SUCCESS in 27m 13s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/59/557659/49/check/kolla-ansible-ubuntu-source/440f35f/ : SUCCESS in 46m 27s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/59/557659/49/check/kolla-ansible-ubuntu-binary/2e8694a/ : SUCCESS in 44m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/59/557659/49/check/kolla-ansible-ubuntu-source-upgrade/c94d1b4/ : SUCCESS in 58m 18s","accounts_in_message":[],"_revision_number":49},{"id":"49a924f65021446c093a9bbfe2f63554d7a2ec69","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-28 09:47:29.000000000","message":"Patch Set 49:\n\ncheck experimental","accounts_in_message":[],"_revision_number":49},{"id":"d0fb4f5e782c415b254a16dab1eb9c163c2a1547","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 12:24:59.000000000","message":"Patch Set 49:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-debian-source-arm64 http://logs.openstack.org/59/557659/49/experimental/kolla-build-debian-source-arm64/ba6981f/ : SUCCESS in 2h 36m 53s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"705fa453c726f42965e9339e3b6b8ca63831a5dc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-28 12:37:31.000000000","message":"Patch Set 49: -Workflow\n\nI consider it ready for another review.","accounts_in_message":[],"_revision_number":49},{"id":"980d62b3ed3ac37395e80008e6f63ba18d9005af","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-28 14:20:44.000000000","message":"Patch Set 49:\n\n(1 comment)","accounts_in_message":[],"_revision_number":49},{"id":"8fe64713e1b3a8c99c60335de2479b070a6cd24e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-28 14:35:24.000000000","message":"Patch Set 49:\n\n(2 comments)","accounts_in_message":[],"_revision_number":49},{"id":"0af8f7720be2c8b027e4d4f9b3edb45c94ec9765","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-28 14:38:42.000000000","message":"Patch Set 49:\n\n(2 comments)","accounts_in_message":[],"_revision_number":49},{"id":"b0754996c3a67e3033380615abb94c6677e3a7c8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-28 14:43:28.000000000","message":"Patch Set 49:\n\n(1 comment)","accounts_in_message":[],"_revision_number":49},{"id":"12c4dcbad0d0e1ba2bf2ec70cc9b1d98178f6a21","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-15 11:50:48.000000000","message":"Uploaded patch set 50.","accounts_in_message":[],"_revision_number":50},{"id":"0c21391cead90df7585da068f1aa1ca3f24a37c4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-15 12:38:11.000000000","message":"Uploaded patch set 51.","accounts_in_message":[],"_revision_number":51},{"id":"b8b75417d7736aa4992000c9af225e2e68bb4390","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 13:51:56.000000000","message":"Patch Set 51: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 kolla-build-centos-source-arm64 : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/51/check/openstack-tox-lower-constraints/ceec059/ : SUCCESS in 3m 27s\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/51/check/openstack-tox-pep8/2397312/ : SUCCESS in 4m 18s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/51/check/openstack-tox-py27/a587909/ : SUCCESS in 2m 49s\n- openstack-tox-py36 http://logs.openstack.org/59/557659/51/check/openstack-tox-py36/d7b0121/ : SUCCESS in 3m 02s\n- openstack-tox-py37 http://logs.openstack.org/59/557659/51/check/openstack-tox-py37/48d571a/ : SUCCESS in 3m 20s\n- openstack-tox-docs http://logs.openstack.org/59/557659/51/check/openstack-tox-docs/472db3b/html/ : SUCCESS in 3m 07s\n- kolla-build-ubuntu-binary http://logs.openstack.org/59/557659/51/check/kolla-build-ubuntu-binary/91bd36b/ : SUCCESS in 27m 44s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/59/557659/51/check/kolla-build-ubuntu-source/ee937ca/ : SUCCESS in 33m 02s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/59/557659/51/check/kolla-ansible-ubuntu-source/97b418d/ : SUCCESS in 50m 11s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/59/557659/51/check/kolla-ansible-ubuntu-binary/d47254f/ : SUCCESS in 49m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/59/557659/51/check/kolla-ansible-ubuntu-source-upgrade/18b3dce/ : SUCCESS in 59m 24s","accounts_in_message":[],"_revision_number":51},{"id":"58844e0085c8d389eeacfa5c1059799bf6e75f39","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-16 16:12:03.000000000","message":"Patch Set 51:\n\nrecheck","accounts_in_message":[],"_revision_number":51},{"id":"b2d0fe148017a6e50b19cae38e3a2c6dda895d27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 19:17:23.000000000","message":"Patch Set 51:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 http://logs.openstack.org/59/557659/51/check/kolla-build-centos-source-arm64/c334c68/ : TIMED_OUT in 3h 00m 01s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/557659/51/check/openstack-tox-lower-constraints/dd3d082/ : SUCCESS in 4m 56s\n- openstack-tox-pep8 http://logs.openstack.org/59/557659/51/check/openstack-tox-pep8/267fd31/ : SUCCESS in 5m 52s\n- openstack-tox-py27 http://logs.openstack.org/59/557659/51/check/openstack-tox-py27/de509b3/ : SUCCESS in 3m 55s\n- openstack-tox-py36 http://logs.openstack.org/59/557659/51/check/openstack-tox-py36/a9688d0/ : SUCCESS in 5m 21s\n- openstack-tox-py37 http://logs.openstack.org/59/557659/51/check/openstack-tox-py37/4e17f2a/ : SUCCESS in 5m 37s\n- openstack-tox-docs http://logs.openstack.org/59/557659/51/check/openstack-tox-docs/e2d72ea/html/ : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":51},{"id":"3485a96ab889ed222ddc0e6f10cadfd3c4e17a9e","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-08-13 16:08:25.000000000","message":"Patch Set 51: Code-Review+1","accounts_in_message":[],"_revision_number":51},{"id":"191eb56da527f37ca96018eb7013ccec734e33b3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-14 13:00:23.000000000","message":"Patch Set 51:\n\nrecheck\n\nwill fail - I need numbers","accounts_in_message":[],"_revision_number":51},{"id":"febbbbdaa2880571d95a2c99b3667bbefb9ff8ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 15:55:10.000000000","message":"Patch Set 51: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/baab785f66394d0689bfde425a914751 : SUCCESS in 2m 58s\n- kolla-build-centos-source-arm64 https://zuul.opendev.org/t/openstack/build/871027c8ddc84f6c97c36705e5d7fc24 : TIMED_OUT in 2h 52m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99ad9a2e4f774e459138f625cd83f115 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d265ed9cfbad47e6a54465094d4055ca : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d9880f29610644ddac07fc93450b772b : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e84f9ad9b7f49f2a1966d9c65413f82 : SUCCESS in 3m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2da082b11ffa4437a2228ab3910e126b : SUCCESS in 4m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0e42cec057d04f7da579899b10a560bc : FAILURE in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9b97411511c455a9136997bb2f87756 : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/765d31bc695e46e6a94460a7ee7b43da : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":51},{"id":"bc0793c9a523f2bccf0642537fbfff47d5d0fd23","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-28 10:49:49.000000000","message":"Uploaded patch set 52.","accounts_in_message":[],"_revision_number":52},{"id":"de889e8c0c51bba79dab649dc6a35b43191c560d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 11:55:27.000000000","message":"Patch Set 52: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8f3d865120cc4d81bbef2fb8e8bf65e7 : SUCCESS in 44m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edac5b26d56044e7a469b933cfc0b745 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/911d04900d9f4b419046c0e343439fba : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0546887b3eb8495c8beb2281db8e844a : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72d7352ac29243e4b6c5974f2a988273 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc4b63492fa24d7fbb6dc97985252550 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/208a7023b5434f8fbb7a97ad6d25b7e4 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":52},{"id":"4dc50c16bb5094376b8ba9e6137448816e49d362","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-20 15:54:19.000000000","message":"Uploaded patch set 53.","accounts_in_message":[],"_revision_number":53},{"id":"d51c164146693dfa9495f15b8e4c8bc2cb117751","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-20 15:56:22.000000000","message":"Uploaded patch set 54.","accounts_in_message":[],"_revision_number":54},{"id":"d97f029a5a9ea712776e8c5c1920bb0fd431b789","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-20 15:58:25.000000000","message":"Uploaded patch set 55: Patch Set 54 was rebased.","accounts_in_message":[],"_revision_number":55},{"id":"ba6d3b05c85d97e6a258ea3bb541ea50aed3fa88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 20:17:44.000000000","message":"Patch Set 55: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/fa4a62475b2d4599afd35565bb112ace : SUCCESS in 39m 18s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b7c9d252a3a4444bb42d0c621a4a1508 : SUCCESS in 49m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1ed1bf64b6a2493fab7cb338932b7001 : SUCCESS in 1h 38m 19s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8504a4bb29e34e77af5ff40851a5ee3e : SUCCESS in 1h 24m 52s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b04805a7cdd74bc795a5bcf5cf2ee521 : SUCCESS in 1h 32m 28s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/34b95ae3d0cb463fa1dc759043b9d046 : SUCCESS in 53m 49s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/a5742019383f4f13bdca673661f69252 : SUCCESS in 2h 20m 33s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/5c5af16d1042463095430a681cd4e41e : SUCCESS in 2h 22m 45s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/c828c65d2ee149cfbcad8c8c912f8d7d : SUCCESS in 1h 47m 49s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/5c89d66076c64352808af71299613c2c : TIMED_OUT in 2h 05m 27s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/3e999df506b745ed9314bb66155f0f11 : SUCCESS in 1h 41m 57s (non-voting)\n- kolla-build-centos-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/024f20dbbdff4866bd4e7e16243f8ba5 : SUCCESS in 23m 48s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df30f54fc24a499996165d136ac86ff4 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8549a56838c4a01ae3519ca94c5a466 : FAILURE in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dfa3dcc4ebc54ec1b8f877dd6c311bfc : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54c59e52b5774cb68e7d5bbb75f850f4 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb15505dcffe4a579992839e45d90d7e : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b79de57d80934ef294e6d37eee1d530d : SUCCESS in 5m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2e2d4183303244e49d34546b8d850029 : SUCCESS in 29m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2a1447e2201a41faad4f20de6f0eae1a : SUCCESS in 32m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ec3ff6f4c70f448dbe0d5bda51745c0f : SUCCESS in 52m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/22f010c74d114f89be943959c500c00a : SUCCESS in 51m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e8fdcd09c5f74d7fafef32e98a415c87 : SUCCESS in 55m 34s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/598247e376494c0cbc07133ccc95543f : SUCCESS in 1h 27m 56s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/0ad4c1514a89438d9edce0e0b37ed7e2 : SUCCESS in 1h 55m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/b0bbb3f4c72c4cdc88555b5de9c6cd2d : SUCCESS in 1h 58m 42s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/b9fa88393239427dab4db3c0de82e570 : SUCCESS in 1h 14m 35s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/b1be733b70f24c4a8a376f8aeba33eab : SUCCESS in 1h 00m 21s (non-voting)","accounts_in_message":[],"_revision_number":55},{"id":"a29937d11d6b90f90c4c019cb80ec2accbfdcdb5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-20 09:25:03.000000000","message":"Uploaded patch set 56.","accounts_in_message":[],"_revision_number":56},{"id":"320d4d3021845473ef7d85dc060d1b62a9e5187e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 11:42:49.000000000","message":"Patch Set 56: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1556a405e97d4b13ac1d6a4b196eeba7 : SUCCESS in 48m 57s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d85a805efa554ce59079056d76ff3b1d : SUCCESS in 57m 21s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4734893ae8334b8883f8ca7870c55d4e : SUCCESS in 1h 34m 20s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/abfc46a1f8d44b2db7f7e60a7c30d4de : SUCCESS in 1h 24m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8eee0e1115be4dc4ad678869912d95c4 : SUCCESS in 1h 34m 40s\n- kolla-build-centos-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/48bfcaa3ee1a4a8eaab5ed2056d5766b : SUCCESS in 24m 08s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c6e466390e2c47f8b080d84f7b1c1fa7 : SUCCESS in 18m 44s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ff631948f0214f5fb560354f994f6785 : SUCCESS in 55m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46840c4bad6a4e4981b358b60d1a7c0b : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7b72d6e53264911b028724c98a045b4 : FAILURE in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7fb531f1425450cae3c9a394acbe6d5 : SUCCESS in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/011db6f0df314da2af7031db361fa00b : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e6574d05741478f9ba790c2f31ad480 : SUCCESS in 5m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d87e3d1a92584e13badd9f74d8f609a2 : FAILURE in 24m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d36c06a34d814a03a81fdd41268a1b33 : FAILURE in 22m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/96639d243ac14bd9bf3713498650222c : FAILURE in 12m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3a4856f1e1fe4c4ba8ace6c4ef64773f : FAILURE in 19m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fab33aa8c8ed49e6b17b813821de17f2 : FAILURE in 15m 12s","accounts_in_message":[],"_revision_number":56},{"id":"1f0d8e73b513b728cad4a91745e4588406698544","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2019-11-21 09:06:12.000000000","message":"Patch Set 56: Code-Review+1\n\ncheck experimental","accounts_in_message":[],"_revision_number":56},{"id":"1b207e55cf4175d6507c720037e3ab75f3391cdd","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-11 09:36:51.000000000","message":"Uploaded patch set 57: Patch Set 56 was rebased.","accounts_in_message":[],"_revision_number":57},{"id":"0bb2ddf5cfb142a0187e62561a63c9f790672f21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 12:45:47.000000000","message":"Patch Set 57: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3094f67b24444ca98a59fc3db584607e : SUCCESS in 44m 45s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/71b3cc7d3ed84c309fe92419fc419b07 : SUCCESS in 41m 34s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/da05d72f4a4a4b9c86ab9286440f73cc : SUCCESS in 1h 20m 28s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ad7d5517369c490da69c4e31f3e76e04 : SUCCESS in 1h 12m 45s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/343531ba171a4fe4a4ed5e145897d254 : SUCCESS in 1h 36m 42s\n- kolla-build-centos-source-aarch64 https://zuul.opendev.org/t/openstack/build/21ebecec5731411787c5835d3e2ad018 : SUCCESS in 2h 35m 52s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5e1021af2d6d4efd963e3c4b9ab3e13f : SUCCESS in 31m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0b609686ae4d48ff8b64f27b2cee9531 : SUCCESS in 19m 45s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5b07cbe4709d4482a5e1c7779e171adc : SUCCESS in 2h 19m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cf6398998f554b75aac7c64c4c3a905b : SUCCESS in 55m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebdd7363e09f4a949cf76ff83dc746d4 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21589db2c0d3424c8884635851467c02 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b64b005b687f4c50b3d744ee55eb9367 : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b57de6a320a646baa6b9d7b914a15fbb : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10c2e291cc0a4635a9a09ba8691e8695 : SUCCESS in 5m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e52727e965a34e39b404544aae7c4246 : SUCCESS in 26m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/71989d3b52e1405c9527a8789dec64e3 : SUCCESS in 35m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/53719d0505b64d19b133397b8cc0bf68 : SUCCESS in 56m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3eff4799a52a4596979d18d050dd39cd : SUCCESS in 52m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/40d050f124d144f1a966d91bf1858546 : SUCCESS in 58m 31s","accounts_in_message":[],"_revision_number":57},{"id":"8055e7118f24af0880157e4ca9d3141a970a4ee9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-13 12:13:17.000000000","message":"Uploaded patch set 58.","accounts_in_message":[],"_revision_number":58},{"id":"93cf1fe73f3eb73f7203604c48e7d1b438beeb37","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-13 12:13:48.000000000","message":"Topic changed from ci/add-centos-aarch64-job to arm64-ci","accounts_in_message":[],"_revision_number":58},{"id":"4aa4535256f057d5bf104e1ca4876457ed5e640d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-13 12:14:30.000000000","message":"Patch Set 58:\n\ncheck-arm64","accounts_in_message":[],"_revision_number":58},{"id":"b78808bea14cb289f40afa6858c4431ec833df03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 14:07:41.000000000","message":"Patch Set 58: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/316c4df4b692418a8dc48b3b0519e104 : FAILURE in 34m 47s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6b144e419b704cbea5fda96beea59640 : FAILURE in 37m 17s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f512e1ae965c4aa3a4358c8fea17d1c4 : SUCCESS in 1h 48m 14s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a589affb8f7040f5864296b8239f35d7 : SUCCESS in 1h 39m 35s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/698259a7a9694287b84b95a8cb9e72df : SUCCESS in 1h 49m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ddbbf5045d0841bca423171f34fa266e : SUCCESS in 26m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/010b72a54e77412eb973fa5ff1c52e94 : SUCCESS in 20m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0270c918b5f94cdbbdad97db96bd568b : SUCCESS in 55m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83397abba56e444f9e006cd47261ccf0 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e57784b2f04e44929867d151cd89afbc : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f558b6327801421fa9b00b249e1bd60a : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d302158cc26c4e9cafca8271dccdccba : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a82c90f0c0045188ca0608323383340 : SUCCESS in 4m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8dbd82de816640408f99f3b0b5b81817 : SUCCESS in 21m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e9b20786eda849f59440c1347af61336 : SUCCESS in 27m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/470cf01ad1b341ac8b25ccad8a96fe15 : SUCCESS in 56m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/de9dd761b9e444c4bcc3022d8d1a9ac7 : SUCCESS in 56m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c6d6e17772fe4fd389f88b99b52b84f2 : SUCCESS in 59m 45s","accounts_in_message":[],"_revision_number":58},{"id":"d459550dc2212c3cab50f48593bd8b29e408b197","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-13 15:42:05.000000000","message":"Patch Set 58:\n\nrecheck","accounts_in_message":[],"_revision_number":58},{"id":"237b48e6af6efcfc507236c738ffe6a011419ced","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-12-13 15:49:53.000000000","message":"Patch Set 58: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":58},{"id":"706fce867977b770cca2da6ee6a2aa2d177cd6c7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-13 15:52:50.000000000","message":"Patch Set 59: Commit message was updated.","accounts_in_message":[],"_revision_number":59},{"id":"5fe5ea8309eb05aff5a3006f5efa6769d35d15ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 17:04:55.000000000","message":"Patch Set 59:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1918dd7b82f24ac4976489e90aef4a9c : SUCCESS in 1h 03m 27s (non-voting)","accounts_in_message":[],"_revision_number":59},{"id":"3c353de7ecaf3da52a4ea8059e5ce4da64e7f97d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-13 17:18:42.000000000","message":"Patch Set 59:\n\n(1 comment)","accounts_in_message":[],"_revision_number":59},{"id":"b10fd710c5ff69702e389b605d24c54a865395d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 17:23:38.000000000","message":"Patch Set 59: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/55655968e5d34268a4b5573362568346 : SUCCESS in 33m 57s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0c0cbceca11e4ae7a9d524836080a58d : SUCCESS in 47m 15s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/097bda9a954f4b349b7808fbc8d2153f : SUCCESS in 1h 24m 14s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ddda81ab8989442ba9c2bc71d08106f9 : SUCCESS in 1h 21m 51s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a1d13861b9474d5e945e30b9fb3582e1 : SUCCESS in 1h 27m 36s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5ac30edc2b4f4aeb84588a239b3b8cc5 : SUCCESS in 44m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/959ada9371de4c62b5255064e852b5db : SUCCESS in 42m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fbbdf6be1b5242e9a6b2d41cd0ca78c7 : SUCCESS in 56m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c3b3a3f6bb44b4f8c4ae4a0de032276 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7baff465d1434e7b8f4e84b4f8575984 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4df297606b204d75aecede29fc1ed66a : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ad75334077ec427bb9288756ab003be0 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25e989ce8563425eb1230fbde38b70f1 : SUCCESS in 5m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/318484b3271449a69f1a219777ec71ea : SUCCESS in 24m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/489a273956b947d1b647b59d6ee665a9 : SUCCESS in 31m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9b5358bfea7d464993e5711e2ca4235b : SUCCESS in 56m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/35cfe4028f74489dae09ff257589125b : SUCCESS in 56m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c9298fd1b0584705bb719764263c550b : SUCCESS in 1h 00m 33s","accounts_in_message":[],"_revision_number":59},{"id":"616a270d2871ea491a7d7e173c7c43de35eb8e32","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-16 08:18:06.000000000","message":"Patch Set 60: Published edit on patch set 59.","accounts_in_message":[],"_revision_number":60},{"id":"aff318acd990ec27eb88086b05866e71801aa67f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-16 08:25:05.000000000","message":"Patch Set 60: Code-Review+2","accounts_in_message":[],"_revision_number":60},{"id":"6a8e6d6a82abe80602bb192bf6040770e4d93237","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2019-12-16 08:51:34.000000000","message":"Patch Set 60: Code-Review+1","accounts_in_message":[],"_revision_number":60},{"id":"32f0c66e69d4ac914001d8c06a8fdf2c372a5392","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 09:22:43.000000000","message":"Patch Set 60:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5c602b43413b4d0db3df1daaa9beccc7 : SUCCESS in 1h 03m 23s (non-voting)","accounts_in_message":[],"_revision_number":60},{"id":"ba23d44ecd6c9bdf5d42ee16ac070cd744466166","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 09:45:46.000000000","message":"Patch Set 60: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7945641fbd7e490f99fb496e3a7d1473 : SUCCESS in 30m 59s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/87da8ce6b1544e7bafc0f3b74b4520c1 : SUCCESS in 37m 52s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f8d4315b77a642b090d1c62317b43963 : SUCCESS in 1h 12m 47s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/45de93ebe0f7487793d1d27a1c4f74cf : SUCCESS in 1h 22m 18s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e9734357180d465dbf410784ae852ea2 : FAILURE in 52m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9ff0d503c5604f63ac7bd7b6882f212c : SUCCESS in 25m 56s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1ffbbafaeb534b9dab173bea9ee81438 : SUCCESS in 35m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a43223efa6744ef4aa9070cb1bebbce7 : SUCCESS in 55m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fb2d290630446a0a80cafc667a01a92 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12cdff2694fc4bc7bef1f334a7d39b3f : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebb5e7cd50e64a97b499b7466c359ea8 : SUCCESS in 6m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/58553489869546cea5b64ab4437db372 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91afa16b41b64f59831572deaa0a5bcf : SUCCESS in 5m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b41d070bad644375a7762fe1676bd7d9 : SUCCESS in 43m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ef18e91d58174d87a2ab360278a42923 : SUCCESS in 25m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/be97808d092e4402805d17cbf688cd2f : SUCCESS in 52m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d79fe3d47c1a49d2a1ef45b5f447c9df : SUCCESS in 56m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c9883067fc3d45c58fe0b431f7196175 : SUCCESS in 1h 03m 22s","accounts_in_message":[],"_revision_number":60},{"id":"9e16f2914195fe4d4d7a0a492c4fd5b96c40a617","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-16 10:14:55.000000000","message":"Patch Set 60:\n\nrecheck","accounts_in_message":[],"_revision_number":60},{"id":"eadca0045dbe336c503816d18aa36b76038d6987","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 11:21:33.000000000","message":"Patch Set 60:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6d1c229100e54329973d83a6c8b59752 : SUCCESS in 1h 01m 46s (non-voting)","accounts_in_message":[],"_revision_number":60},{"id":"feeced80a8bf0c2c6b82ae4499e4784921db8e7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 11:57:31.000000000","message":"Patch Set 60: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a1932ea4d2654ab48b2e2e81769b76b6 : SUCCESS in 43m 56s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e39a536a61e6431ba8a811e20687f89f : SUCCESS in 40m 38s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/38871b4249284ca486dd01b28e7d66f3 : SUCCESS in 1h 22m 08s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7817f1558eb84110969f67558d97e33b : SUCCESS in 1h 32m 10s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c3d0c15fcc8f4d0990d907c9087344c6 : SUCCESS in 1h 27m 53s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/eb012d5ac8ea40dd80fd8c9013f7e0f7 : SUCCESS in 24m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/96019fe73b2744039fd890134e98e825 : SUCCESS in 29m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fa8269e1eb4641628dfc3c56d03ef9d6 : SUCCESS in 57m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d41dce0593344896a156e86d56ce3218 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46bca85155a140ef9eedf61dfc8729cf : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a57e4bfc7a9247999269c30bb314a234 : SUCCESS in 6m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8d674d5530ad4de2a0a1d51d6f7d25b8 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62a739bb2925474a95ee828e4cb4f26e : SUCCESS in 6m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4db2c40b49ce427c83dbc0ddf14fa7df : SUCCESS in 25m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/90d4f5a1cef64fc9a90e98418ced0e41 : SUCCESS in 29m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c6bbf1f539194cb49c36eef5132d54c8 : SUCCESS in 54m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/73b08ce2c00741a7a813d60ef564ab74 : SUCCESS in 53m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2bb4b6e228244134a0a90d8e297edd73 : SUCCESS in 1h 10m 44s","accounts_in_message":[],"_revision_number":60},{"id":"39aac4e8d6b1f6fe1fbb35bbbe3ad2b8b56d9bc4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-16 12:20:33.000000000","message":"Patch Set 60: Code-Review+2","accounts_in_message":[],"_revision_number":60},{"id":"416370ad10e3f23135f33376b8878d67c3cb1730","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-16 14:18:19.000000000","message":"Patch Set 60: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":60},{"id":"017d0614dc92703e982212bc20c1bde7fd984e4d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-16 14:18:34.000000000","message":"Patch Set 60:\n\nNice work on this.","accounts_in_message":[],"_revision_number":60},{"id":"1055e1991cb3055981216672a7cca6f588c1873d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 14:18:37.000000000","message":"Patch Set 60: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":60},{"id":"fbd5118ef9f7de5fe7b858ac135c4cf72ed30a62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 16:03:23.000000000","message":"Patch Set 60: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/dc5aefd1ad404e9aa7f7f3d1718a93e7 : SUCCESS in 36m 18s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a30b4562352545a6b58e26496e68ade8 : SUCCESS in 1h 22m 56s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/24421c54985844b2acf494af4c00af44 : SUCCESS in 1h 42m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d79d4ffdb2c9442e976267b09bdc3d5b : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cad7d46bfb644e33b3a62193373a527e : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00138d121df74e99818621df4503a18e : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eba02e0aae7a450e843c52a9a428acbc : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc5d9e00c90c4e25ab4edd8d8e5e5744 : SUCCESS in 6m 57s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5354484e906048b18140727d5a5eb2fb : SUCCESS in 34m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7aedc2ea637a4351a7581f58dd95f1d6 : SUCCESS in 52m 37s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e7576ab3a3cd41f7b8f9a7dcf2ade564 : SUCCESS in 1h 00m 20s","accounts_in_message":[],"_revision_number":60},{"id":"fb5a31eb0438f3748677f31a9471fae19fa58651","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 16:03:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":60},{"id":"c6f6d17a3a97dbd0f080f6519f109b3f480bec30","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 16:07:20.000000000","message":"Patch Set 60:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a31cc8909a8497cb9867035fbce2f80 : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":60}],"current_revision_number":60,"current_revision":"d6935edb43ab577b2895d701d6d7e7e555934d8b","revisions":{"979e515253f4697f00a1ba7f69071656e661e361":{"kind":"REWORK","_number":1,"created":"2018-03-29 09:37:47.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/1"}}},"commit":{"parents":[{"commit":"eb380579aca7e4a6aa14fe486573cd851f834228","subject":"Update the description for MongoDB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb380579aca7e4a6aa14fe486573cd851f834228"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-03-29 06:18:48.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-03-29 06:48:26.000000000","tz":480},"subject":"Add kolla-build-debian-source job","message":"Add kolla-build-debian-source job\n\nAdd kolla debian arm64 source building job.\nWhich runs on ubuntu-xenial-arm64 node.\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\nSigned-off-by: Xinliang Liu \u003cxinliang.liu@linaro.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/979e515253f4697f00a1ba7f69071656e661e361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/979e515253f4697f00a1ba7f69071656e661e361"}]},"branch":"refs/heads/master"},"6fb46aefe059c9048d063807712afe4c0dfd3c6f":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2018-03-30 10:25:54.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/2"}}},"commit":{"parents":[{"commit":"cea799bd6a1c94ff731d038a8ff2187d0d0249be","subject":"Merge \"Add ZooKeeper docs link into README.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cea799bd6a1c94ff731d038a8ff2187d0d0249be"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-03-29 06:18:48.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-03-30 10:25:51.000000000","tz":480},"subject":"Add kolla-build-debian-source job","message":"Add kolla-build-debian-source job\n\nAdd kolla debian arm64 source building job.\nWhich runs on ubuntu-xenial-arm64 node.\n\nDepends-On: https://review.openstack.org/#/c/557920/\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6fb46aefe059c9048d063807712afe4c0dfd3c6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6fb46aefe059c9048d063807712afe4c0dfd3c6f"}]},"branch":"refs/heads/master"},"11f36da78101cdb7dd4411e608b1783d7bdc4b5f":{"kind":"REWORK","_number":3,"created":"2018-04-02 03:44:38.000000000","uploader":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"ref":"refs/changes/59/557659/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/3"}}},"commit":{"parents":[{"commit":"cea799bd6a1c94ff731d038a8ff2187d0d0249be","subject":"Merge \"Add ZooKeeper docs link into README.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cea799bd6a1c94ff731d038a8ff2187d0d0249be"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-03-29 06:18:48.000000000","tz":480},"committer":{"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","date":"2018-04-02 03:44:28.000000000","tz":480},"subject":"Add kolla-build-debian-source job","message":"Add kolla-build-debian-source job\n\nAdd kolla debian arm64 source building job.\nWhich runs on ubuntu-xenial-arm64 node.\n\nDepends-On: https://review.openstack.org/#/c/557920/\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11f36da78101cdb7dd4411e608b1783d7bdc4b5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11f36da78101cdb7dd4411e608b1783d7bdc4b5f"}]},"branch":"refs/heads/master"},"5756b4a4c56b288d56163c13d14dc9c632df380d":{"kind":"REWORK","_number":4,"created":"2018-04-02 06:28:45.000000000","uploader":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"ref":"refs/changes/59/557659/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/4"}}},"commit":{"parents":[{"commit":"cea799bd6a1c94ff731d038a8ff2187d0d0249be","subject":"Merge \"Add ZooKeeper docs link into README.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cea799bd6a1c94ff731d038a8ff2187d0d0249be"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-03-29 06:18:48.000000000","tz":480},"committer":{"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","date":"2018-04-02 06:28:33.000000000","tz":480},"subject":"Add kolla-build-debian-source job","message":"Add kolla-build-debian-source job\n\nAdd kolla debian arm64 source building job.\nWhich runs on ubuntu-xenial-arm64 node.\n\nDepends-On: https://review.openstack.org/#/c/557920/\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5756b4a4c56b288d56163c13d14dc9c632df380d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5756b4a4c56b288d56163c13d14dc9c632df380d"}]},"branch":"refs/heads/master"},"17f52a7f9edf7fe0f6214e8d45eff0b902a7b514":{"kind":"REWORK","_number":5,"created":"2018-04-02 08:19:25.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/5"}}},"commit":{"parents":[{"commit":"147523d60c34ac245dd648cb09e4ace0dbda5649","subject":"Merge \"Add support for Infoblox IPAM agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/147523d60c34ac245dd648cb09e4ace0dbda5649"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-03-29 06:18:48.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-04-02 08:18:27.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\nAdd kolla debian arm64 source building job.\nWhich runs on ubuntu-xenial-arm64 node.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17f52a7f9edf7fe0f6214e8d45eff0b902a7b514"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17f52a7f9edf7fe0f6214e8d45eff0b902a7b514"}]},"branch":"refs/heads/master"},"4ba83592e281b34e90ca1a962d41eeec2cf7c26a":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2018-04-02 08:59:32.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/6"}}},"commit":{"parents":[{"commit":"147523d60c34ac245dd648cb09e4ace0dbda5649","subject":"Merge \"Add support for Infoblox IPAM agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/147523d60c34ac245dd648cb09e4ace0dbda5649"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-03-29 06:18:48.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-04-02 08:58:39.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\nWhich runs on ubuntu-xenial-arm64 node.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ba83592e281b34e90ca1a962d41eeec2cf7c26a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ba83592e281b34e90ca1a962d41eeec2cf7c26a"}]},"branch":"refs/heads/master"},"c1b2939fdb7a5fd8cc32793fd961770eba27a789":{"kind":"REWORK","_number":7,"created":"2018-04-02 10:13:02.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/7"}}},"commit":{"parents":[{"commit":"147523d60c34ac245dd648cb09e4ace0dbda5649","subject":"Merge \"Add support for Infoblox IPAM agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/147523d60c34ac245dd648cb09e4ace0dbda5649"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-03-29 06:18:48.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-04-02 10:11:49.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\nWhich runs on ubuntu-xenial-arm64 node.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1b2939fdb7a5fd8cc32793fd961770eba27a789"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1b2939fdb7a5fd8cc32793fd961770eba27a789"}]},"branch":"refs/heads/master"},"24d4fe9c8f779a81fbee25824847ca9b29acc967":{"kind":"REWORK","_number":8,"created":"2018-04-03 02:10:09.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/8"}}},"commit":{"parents":[{"commit":"147523d60c34ac245dd648cb09e4ace0dbda5649","subject":"Merge \"Add support for Infoblox IPAM agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/147523d60c34ac245dd648cb09e4ace0dbda5649"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-03-29 06:18:48.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-04-03 02:09:51.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\nWhich runs on ubuntu-xenial-arm64 node.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/24d4fe9c8f779a81fbee25824847ca9b29acc967"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/24d4fe9c8f779a81fbee25824847ca9b29acc967"}]},"branch":"refs/heads/master"},"1693938c989920dcf97cbe621c36460730d9d86a":{"kind":"REWORK","_number":9,"created":"2018-04-10 09:55:50.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/9"}}},"commit":{"parents":[{"commit":"eed81339a462c76688228d9d55c2d80fbb5e6709","subject":"Merge \"Specification for allowing Kolla Ceph to deploy bluestore OSDs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eed81339a462c76688228d9d55c2d80fbb5e6709"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-03-29 06:18:48.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-04-10 09:53:34.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1693938c989920dcf97cbe621c36460730d9d86a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1693938c989920dcf97cbe621c36460730d9d86a"}]},"branch":"refs/heads/master"},"bf43f766cea0e3a477508f119a12f865e4dca583":{"kind":"REWORK","_number":10,"created":"2018-04-10 11:01:32.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/10"}}},"commit":{"parents":[{"commit":"eed81339a462c76688228d9d55c2d80fbb5e6709","subject":"Merge \"Specification for allowing Kolla Ceph to deploy bluestore OSDs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eed81339a462c76688228d9d55c2d80fbb5e6709"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-03-29 06:18:48.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-04-10 11:00:35.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf43f766cea0e3a477508f119a12f865e4dca583"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf43f766cea0e3a477508f119a12f865e4dca583"}]},"branch":"refs/heads/master"},"aedd0e4907bc79de5dd46d99c9c846a8b955ba95":{"kind":"REWORK","_number":11,"created":"2018-04-12 09:56:04.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/11"}}},"commit":{"parents":[{"commit":"eed81339a462c76688228d9d55c2d80fbb5e6709","subject":"Merge \"Specification for allowing Kolla Ceph to deploy bluestore OSDs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eed81339a462c76688228d9d55c2d80fbb5e6709"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-03-29 06:18:48.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-04-12 09:54:46.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aedd0e4907bc79de5dd46d99c9c846a8b955ba95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aedd0e4907bc79de5dd46d99c9c846a8b955ba95"}]},"branch":"refs/heads/master"},"331953e330b5b46f526403ad2b83ff3ab041e8c1":{"kind":"REWORK","_number":12,"created":"2018-04-12 10:59:43.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/12"}}},"commit":{"parents":[{"commit":"eed81339a462c76688228d9d55c2d80fbb5e6709","subject":"Merge \"Specification for allowing Kolla Ceph to deploy bluestore OSDs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eed81339a462c76688228d9d55c2d80fbb5e6709"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-03-29 06:18:48.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-04-12 10:58:50.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/331953e330b5b46f526403ad2b83ff3ab041e8c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/331953e330b5b46f526403ad2b83ff3ab041e8c1"}]},"branch":"refs/heads/master"},"abb6cb2fd44488e807e5397651ed4a818f5c5f3f":{"kind":"REWORK","_number":13,"created":"2018-04-17 08:56:57.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/13"}}},"commit":{"parents":[{"commit":"0f2c0d54b886b979ff834be18ec716712b8ece77","subject":"Following the new PTI for document build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f2c0d54b886b979ff834be18ec716712b8ece77"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-03-29 06:18:48.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-04-17 08:53:07.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abb6cb2fd44488e807e5397651ed4a818f5c5f3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abb6cb2fd44488e807e5397651ed4a818f5c5f3f"}]},"branch":"refs/heads/master"},"13aaf79ea1482e01f510291a3d3176f8213a312f":{"kind":"REWORK","_number":14,"created":"2018-04-18 08:38:26.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/14"}}},"commit":{"parents":[{"commit":"d99fde03e1641ebba51710161c202e9c3bdb4c31","subject":"Merge \"Do not require the bootstrap password to come in via arguments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d99fde03e1641ebba51710161c202e9c3bdb4c31"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-03-29 06:18:48.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-04-18 08:37:09.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13aaf79ea1482e01f510291a3d3176f8213a312f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13aaf79ea1482e01f510291a3d3176f8213a312f"}]},"branch":"refs/heads/master"},"0d46f51dcc003ddd5c196df4912c988d12d78237":{"kind":"REWORK","_number":15,"created":"2018-04-19 01:33:26.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/15"}}},"commit":{"parents":[{"commit":"d99fde03e1641ebba51710161c202e9c3bdb4c31","subject":"Merge \"Do not require the bootstrap password to come in via arguments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d99fde03e1641ebba51710161c202e9c3bdb4c31"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-03-29 06:18:48.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-04-19 01:32:45.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d46f51dcc003ddd5c196df4912c988d12d78237"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d46f51dcc003ddd5c196df4912c988d12d78237"}]},"branch":"refs/heads/master"},"de306616e6465192711931bea4a36bd9001d4fb2":{"kind":"REWORK","_number":16,"created":"2018-04-23 06:45:18.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/16"}}},"commit":{"parents":[{"commit":"8b4b5edd76294979b22e314c5c15c64b707a4d95","subject":"Merge \"Prompt user that kolla should be installed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b4b5edd76294979b22e314c5c15c64b707a4d95"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-03-29 06:18:48.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-04-23 06:44:59.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/de306616e6465192711931bea4a36bd9001d4fb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/de306616e6465192711931bea4a36bd9001d4fb2"}]},"branch":"refs/heads/master"},"94233201520962aa6bddcb680b75ab10c487f6e7":{"kind":"REWORK","_number":17,"created":"2018-05-16 02:25:25.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/17"}}},"commit":{"parents":[{"commit":"42117a49e3b8730c106533e1c30f0fa248e01e93","subject":"Merge \"copy iPXE NBP to /tftpboot during PXE bootstrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42117a49e3b8730c106533e1c30f0fa248e01e93"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/94233201520962aa6bddcb680b75ab10c487f6e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/94233201520962aa6bddcb680b75ab10c487f6e7"}]},"branch":"refs/heads/master"},"7bffee306b03c6f96ff18afa0f3c664b586d4ad2":{"kind":"REWORK","_number":18,"created":"2018-05-17 02:56:52.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/18"}}},"commit":{"parents":[{"commit":"25f2e6b754366471aaff9faeeac0d7570bb006fc","subject":"Merge \"build: remove info that image squash support is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25f2e6b754366471aaff9faeeac0d7570bb006fc"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-17 02:53:53.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bffee306b03c6f96ff18afa0f3c664b586d4ad2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bffee306b03c6f96ff18afa0f3c664b586d4ad2"}]},"branch":"refs/heads/master"},"45655d40f940c49895d3debbcf7c120bb075070d":{"kind":"NO_CODE_CHANGE","_number":19,"created":"2018-05-25 09:43:06.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/19"}}},"commit":{"parents":[{"commit":"25f2e6b754366471aaff9faeeac0d7570bb006fc","subject":"Merge \"build: remove info that image squash support is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25f2e6b754366471aaff9faeeac0d7570bb006fc"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2018-05-25 09:42:54.000000000","tz":0},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nDepends-on: I7d85c3bfd21f6f2bed631a7817507595d4bbe521\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45655d40f940c49895d3debbcf7c120bb075070d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45655d40f940c49895d3debbcf7c120bb075070d"}]},"branch":"refs/heads/master"},"1d942fe09ecdd14641de1951bd1e8ba0e2eba80f":{"kind":"REWORK","_number":20,"created":"2018-06-21 08:11:24.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/20"}}},"commit":{"parents":[{"commit":"d8c7d1b27cf53ee0d4f0e1e0639aaea635a2d0ec","subject":"Merge \"Use RDO version of python-msgpack on RDO\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8c7d1b27cf53ee0d4f0e1e0639aaea635a2d0ec"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-06-21 08:09:26.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d942fe09ecdd14641de1951bd1e8ba0e2eba80f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d942fe09ecdd14641de1951bd1e8ba0e2eba80f"}]},"branch":"refs/heads/master"},"e0441e73e0fc51e73fa1c9efa0b7d05547654744":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2018-06-28 09:30:20.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/21"}}},"commit":{"parents":[{"commit":"eedeb3c783b2b3f28c2c857d9f8c13663cd7e6c9","subject":"Merge \"skydive: bump to version 0.18\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eedeb3c783b2b3f28c2c857d9f8c13663cd7e6c9"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2018-06-28 09:30:20.000000000","tz":0},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0441e73e0fc51e73fa1c9efa0b7d05547654744"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0441e73e0fc51e73fa1c9efa0b7d05547654744"}]},"branch":"refs/heads/master"},"c78b7d2088b49cdc7cdf36e372986e6f9e843ccb":{"kind":"REWORK","_number":22,"created":"2018-09-27 09:18:25.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/22"}}},"commit":{"parents":[{"commit":"92da071d9da8476c74ecb16e43e71b276744a863","subject":"Merge \"Downgrade Logstash for Monasca\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/92da071d9da8476c74ecb16e43e71b276744a863"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-09-27 09:18:05.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c78b7d2088b49cdc7cdf36e372986e6f9e843ccb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c78b7d2088b49cdc7cdf36e372986e6f9e843ccb"}]},"branch":"refs/heads/master"},"e48478c38fa6379885ed1ade5be99d37d2e983ac":{"kind":"REWORK","_number":23,"created":"2018-09-27 10:43:17.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/23"}}},"commit":{"parents":[{"commit":"92da071d9da8476c74ecb16e43e71b276744a863","subject":"Merge \"Downgrade Logstash for Monasca\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/92da071d9da8476c74ecb16e43e71b276744a863"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-09-27 10:43:02.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e48478c38fa6379885ed1ade5be99d37d2e983ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e48478c38fa6379885ed1ade5be99d37d2e983ac"}]},"branch":"refs/heads/master"},"5d0d0a14269298be23d2f42bbf4c577cd7fc0b32":{"kind":"REWORK","_number":24,"created":"2018-09-27 12:11:34.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/24"}}},"commit":{"parents":[{"commit":"92da071d9da8476c74ecb16e43e71b276744a863","subject":"Merge \"Downgrade Logstash for Monasca\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/92da071d9da8476c74ecb16e43e71b276744a863"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-09-27 11:31:52.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d0d0a14269298be23d2f42bbf4c577cd7fc0b32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d0d0a14269298be23d2f42bbf4c577cd7fc0b32"}]},"branch":"refs/heads/master"},"d24a2d0247d83570b494fcf64872233dd780f052":{"kind":"REWORK","_number":25,"created":"2018-09-28 07:31:59.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/25"}}},"commit":{"parents":[{"commit":"92da071d9da8476c74ecb16e43e71b276744a863","subject":"Merge \"Downgrade Logstash for Monasca\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/92da071d9da8476c74ecb16e43e71b276744a863"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-09-28 07:31:41.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d24a2d0247d83570b494fcf64872233dd780f052"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d24a2d0247d83570b494fcf64872233dd780f052"}]},"branch":"refs/heads/master"},"3c0495b59c9f30e95e0c21ce953f5a935bc5477c":{"kind":"REWORK","_number":26,"created":"2018-09-29 00:44:03.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/26"}}},"commit":{"parents":[{"commit":"92da071d9da8476c74ecb16e43e71b276744a863","subject":"Merge \"Downgrade Logstash for Monasca\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/92da071d9da8476c74ecb16e43e71b276744a863"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-09-28 10:43:14.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c0495b59c9f30e95e0c21ce953f5a935bc5477c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c0495b59c9f30e95e0c21ce953f5a935bc5477c"}]},"branch":"refs/heads/master"},"d04402ce97fcdb740270d13832efe53f2466fab8":{"kind":"REWORK","_number":27,"created":"2018-09-29 07:52:35.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/27"}}},"commit":{"parents":[{"commit":"92da071d9da8476c74ecb16e43e71b276744a863","subject":"Merge \"Downgrade Logstash for Monasca\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/92da071d9da8476c74ecb16e43e71b276744a863"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-09-29 07:51:34.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d04402ce97fcdb740270d13832efe53f2466fab8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d04402ce97fcdb740270d13832efe53f2466fab8"}]},"branch":"refs/heads/master"},"b2914f49de9ddd5b5dcd2b10732942ec87308c35":{"kind":"REWORK","_number":28,"created":"2018-10-24 04:07:07.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/28"}}},"commit":{"parents":[{"commit":"f613ff8a548accee65bd86209e4683539b66ffdb","subject":"Merge \"Improve pep8 checking along with hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f613ff8a548accee65bd86209e4683539b66ffdb"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-10-24 03:41:52.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2914f49de9ddd5b5dcd2b10732942ec87308c35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2914f49de9ddd5b5dcd2b10732942ec87308c35"}]},"branch":"refs/heads/master"},"cb91f3362d6e056f16024169db7a1142212edf59":{"kind":"REWORK","_number":29,"created":"2018-10-24 08:27:04.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/29"}}},"commit":{"parents":[{"commit":"f613ff8a548accee65bd86209e4683539b66ffdb","subject":"Merge \"Improve pep8 checking along with hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f613ff8a548accee65bd86209e4683539b66ffdb"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-10-24 08:18:48.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb91f3362d6e056f16024169db7a1142212edf59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb91f3362d6e056f16024169db7a1142212edf59"}]},"branch":"refs/heads/master"},"021421cdb9def06a2e2269d9c6ea428010b06d57":{"kind":"REWORK","_number":30,"created":"2018-10-25 01:57:08.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/30"}}},"commit":{"parents":[{"commit":"f613ff8a548accee65bd86209e4683539b66ffdb","subject":"Merge \"Improve pep8 checking along with hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f613ff8a548accee65bd86209e4683539b66ffdb"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-10-25 01:49:23.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/021421cdb9def06a2e2269d9c6ea428010b06d57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/021421cdb9def06a2e2269d9c6ea428010b06d57"}]},"branch":"refs/heads/master"},"b525026131761e489fea4eeeb06fdac530e51747":{"kind":"REWORK","_number":31,"created":"2018-10-25 06:21:38.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/31"}}},"commit":{"parents":[{"commit":"f613ff8a548accee65bd86209e4683539b66ffdb","subject":"Merge \"Improve pep8 checking along with hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f613ff8a548accee65bd86209e4683539b66ffdb"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-10-25 06:17:21.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b525026131761e489fea4eeeb06fdac530e51747"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b525026131761e489fea4eeeb06fdac530e51747"}]},"branch":"refs/heads/master"},"44bd0cca2ec905d5df969e25c6e602751f5c1447":{"kind":"REWORK","_number":32,"created":"2018-10-26 06:23:12.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/32"}}},"commit":{"parents":[{"commit":"f613ff8a548accee65bd86209e4683539b66ffdb","subject":"Merge \"Improve pep8 checking along with hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f613ff8a548accee65bd86209e4683539b66ffdb"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-10-26 06:12:57.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44bd0cca2ec905d5df969e25c6e602751f5c1447"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44bd0cca2ec905d5df969e25c6e602751f5c1447"}]},"branch":"refs/heads/master"},"7a85e67739a9ecc0fc371dfeaa5c02ae59694b3f":{"kind":"REWORK","_number":33,"created":"2018-10-26 07:02:00.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/33"}}},"commit":{"parents":[{"commit":"f613ff8a548accee65bd86209e4683539b66ffdb","subject":"Merge \"Improve pep8 checking along with hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f613ff8a548accee65bd86209e4683539b66ffdb"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-10-26 06:57:34.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a85e67739a9ecc0fc371dfeaa5c02ae59694b3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a85e67739a9ecc0fc371dfeaa5c02ae59694b3f"}]},"branch":"refs/heads/master"},"7e362fff6171d0065ebc79f13ca11445f2457f03":{"kind":"REWORK","_number":34,"created":"2018-10-27 06:22:12.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/59/557659/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/34"}}},"commit":{"parents":[{"commit":"f613ff8a548accee65bd86209e4683539b66ffdb","subject":"Merge \"Improve pep8 checking along with hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f613ff8a548accee65bd86209e4683539b66ffdb"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-10-27 06:17:56.000000000","tz":480},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e362fff6171d0065ebc79f13ca11445f2457f03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e362fff6171d0065ebc79f13ca11445f2457f03"}]},"branch":"refs/heads/master"},"c1744f8a96f0ebd6144482f14728b1b54b1acca3":{"kind":"REWORK","_number":35,"created":"2019-01-14 08:54:26.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/35"}}},"commit":{"parents":[{"commit":"c2397e62d86773c243753cab32183b43831e7ac0","subject":"Add python2-sqlalchemy-collectd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c2397e62d86773c243753cab32183b43831e7ac0"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-01-14 08:53:35.000000000","tz":60},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1744f8a96f0ebd6144482f14728b1b54b1acca3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1744f8a96f0ebd6144482f14728b1b54b1acca3"}]},"branch":"refs/heads/master"},"9e393311dcf0cf710f00a4d705a25f2f1db01e1e":{"kind":"REWORK","_number":36,"created":"2019-03-06 15:57:57.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/36"}}},"commit":{"parents":[{"commit":"ac5103ee1a5e45ef53c2187e48985977f927a3ac","subject":"Merge \"nova: Remove bridge_utils from base package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac5103ee1a5e45ef53c2187e48985977f927a3ac"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-03-06 15:57:48.000000000","tz":60},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e393311dcf0cf710f00a4d705a25f2f1db01e1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e393311dcf0cf710f00a4d705a25f2f1db01e1e"}]},"branch":"refs/heads/master"},"e8e89ae463cf60c8252351da3dc7f2d34d541b19":{"kind":"REWORK","_number":37,"created":"2019-04-24 15:49:36.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/37"}}},"commit":{"parents":[{"commit":"4e17e475eddc613f00bf84c4adc298bfd5a35bf3","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e17e475eddc613f00bf84c4adc298bfd5a35bf3"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-04-24 15:49:01.000000000","tz":120},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e8e89ae463cf60c8252351da3dc7f2d34d541b19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e8e89ae463cf60c8252351da3dc7f2d34d541b19"}]},"branch":"refs/heads/master"},"835dd50107e2769813d9bc70eb12a9fb808502ae":{"kind":"REWORK","_number":38,"created":"2019-04-29 11:42:54.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/38"}}},"commit":{"parents":[{"commit":"4e17e475eddc613f00bf84c4adc298bfd5a35bf3","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e17e475eddc613f00bf84c4adc298bfd5a35bf3"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-04-29 11:42:43.000000000","tz":120},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\n** Note: Insecure Repos enabled - will be done in Zuul instead **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/835dd50107e2769813d9bc70eb12a9fb808502ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/835dd50107e2769813d9bc70eb12a9fb808502ae"}]},"branch":"refs/heads/master"},"197c7c2eb1395dfca9db2ee0a16ec2cdcb37a9aa":{"kind":"REWORK","_number":39,"created":"2019-04-30 09:26:42.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/39"}}},"commit":{"parents":[{"commit":"4e17e475eddc613f00bf84c4adc298bfd5a35bf3","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e17e475eddc613f00bf84c4adc298bfd5a35bf3"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-04-30 09:26:11.000000000","tz":120},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\n** Note: Insecure Repos enabled - will be done in Zuul instead **\n\n** Timeout bumped to 5h to check how much is needed **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/197c7c2eb1395dfca9db2ee0a16ec2cdcb37a9aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/197c7c2eb1395dfca9db2ee0a16ec2cdcb37a9aa"}]},"branch":"refs/heads/master"},"3907f0ecb6d6ec685d1a6f5601d50594639109f0":{"kind":"REWORK","_number":40,"created":"2019-04-30 09:40:55.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/40"}}},"commit":{"parents":[{"commit":"4e17e475eddc613f00bf84c4adc298bfd5a35bf3","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e17e475eddc613f00bf84c4adc298bfd5a35bf3"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-04-30 09:40:44.000000000","tz":120},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\n** Note: Insecure Repos enabled - will be done in Zuul instead **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3907f0ecb6d6ec685d1a6f5601d50594639109f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3907f0ecb6d6ec685d1a6f5601d50594639109f0"}]},"branch":"refs/heads/master"},"3ffa18d6b811ce6efcee668db8ceaf673f6977dc":{"kind":"REWORK","_number":41,"created":"2019-05-06 07:56:21.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/41"}}},"commit":{"parents":[{"commit":"4e17e475eddc613f00bf84c4adc298bfd5a35bf3","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e17e475eddc613f00bf84c4adc298bfd5a35bf3"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-06 07:55:35.000000000","tz":120},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\n** Note: Insecure Repos enabled - will be done in Zuul instead **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ffa18d6b811ce6efcee668db8ceaf673f6977dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ffa18d6b811ce6efcee668db8ceaf673f6977dc"}]},"branch":"refs/heads/master"},"8bd28d2f8bbad41e01cab19a08f761b600be00fa":{"kind":"REWORK","_number":42,"created":"2019-05-06 08:19:45.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/42"}}},"commit":{"parents":[{"commit":"4e17e475eddc613f00bf84c4adc298bfd5a35bf3","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e17e475eddc613f00bf84c4adc298bfd5a35bf3"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-06 08:19:33.000000000","tz":120},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\n** Note: Insecure Repos enabled - will be done in Zuul instead **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8bd28d2f8bbad41e01cab19a08f761b600be00fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8bd28d2f8bbad41e01cab19a08f761b600be00fa"}]},"branch":"refs/heads/master"},"966e41af2a2d658615dcffd51633ec78a6300499":{"kind":"REWORK","_number":43,"created":"2019-05-06 10:43:17.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/43"}}},"commit":{"parents":[{"commit":"4e17e475eddc613f00bf84c4adc298bfd5a35bf3","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e17e475eddc613f00bf84c4adc298bfd5a35bf3"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-06 10:42:18.000000000","tz":120},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\n** Note: Insecure Repos enabled - will be done in Zuul instead **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nDepends-On: Ibfe0c2b8be98db56c61f74fb0247488ab3749ef4\nDepends-On: I0cc03a1ab5a79ae49fa7d13ac549e21d3ec5295d\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/966e41af2a2d658615dcffd51633ec78a6300499"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/966e41af2a2d658615dcffd51633ec78a6300499"}]},"branch":"refs/heads/master"},"6a5734fa1e8286e9bc31cb58a7ed0039f3f9cc00":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":44,"created":"2019-05-07 08:15:58.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/44"}}},"commit":{"parents":[{"commit":"98a946f9d595ec6ea230f391f75e65168ec24c56","subject":"Merge \"Drop systemd support from PAM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98a946f9d595ec6ea230f391f75e65168ec24c56"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-07 08:15:19.000000000","tz":120},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\n** Note: Insecure Repos enabled - will be done in Zuul instead **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nDepends-On: Ibfe0c2b8be98db56c61f74fb0247488ab3749ef4\nDepends-On: I0cc03a1ab5a79ae49fa7d13ac549e21d3ec5295d\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a5734fa1e8286e9bc31cb58a7ed0039f3f9cc00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a5734fa1e8286e9bc31cb58a7ed0039f3f9cc00"}]},"branch":"refs/heads/master"},"8792588ac340256e4fd6cf04d626681509b7c0a4":{"kind":"TRIVIAL_REBASE","_number":45,"created":"2019-06-07 09:08:25.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/45"}}},"commit":{"parents":[{"commit":"9cfd3a88aa1c07054b66495dbb069a0000a20738","subject":"Merge \"Ignore EM releases in version-check.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9cfd3a88aa1c07054b66495dbb069a0000a20738"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-07 09:07:36.000000000","tz":120},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\n** Note: Insecure Repos enabled - will be done in Zuul instead **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nDepends-On: Ibfe0c2b8be98db56c61f74fb0247488ab3749ef4\nDepends-On: I0cc03a1ab5a79ae49fa7d13ac549e21d3ec5295d\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8792588ac340256e4fd6cf04d626681509b7c0a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8792588ac340256e4fd6cf04d626681509b7c0a4"}]},"branch":"refs/heads/master"},"b7058fbedd986193b613bf4f3498065490cb9953":{"kind":"REWORK","_number":46,"created":"2019-06-07 10:02:15.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/46"}}},"commit":{"parents":[{"commit":"9cfd3a88aa1c07054b66495dbb069a0000a20738","subject":"Merge \"Ignore EM releases in version-check.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9cfd3a88aa1c07054b66495dbb069a0000a20738"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-07 10:00:32.000000000","tz":120},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\n** Note: Removed other jobs for saving CI resources now, after fixing\nall the errors of this job will add them back **\n\n** Note: Insecure Repos enabled - will be done in Zuul instead **\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nDepends-On: Ibfe0c2b8be98db56c61f74fb0247488ab3749ef4\nDepends-On: I0cc03a1ab5a79ae49fa7d13ac549e21d3ec5295d\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b7058fbedd986193b613bf4f3498065490cb9953"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b7058fbedd986193b613bf4f3498065490cb9953"}]},"branch":"refs/heads/master"},"319e0249022fc950e76796029001423d3dc57f9e":{"kind":"REWORK","_number":47,"created":"2019-06-14 21:14:33.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/47"}}},"commit":{"parents":[{"commit":"39a2cf066c4c79436629017155302d36243aa338","subject":"Fix lower-constraints test fail due to failing dep build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39a2cf066c4c79436629017155302d36243aa338"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-14 21:11:39.000000000","tz":120},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/319e0249022fc950e76796029001423d3dc57f9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/319e0249022fc950e76796029001423d3dc57f9e"}]},"branch":"refs/heads/master"},"1e16b811d23039f6136cecdf6cbb043a0fb831b1":{"kind":"REWORK","_number":48,"created":"2019-06-17 09:32:08.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/48"}}},"commit":{"parents":[{"commit":"45ef7176aead8d8b59135f8c168094600ea05f31","subject":"Merge \"[neutron] Do not migrate all subprojects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45ef7176aead8d8b59135f8c168094600ea05f31"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-17 09:31:38.000000000","tz":0},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\nImages list shortened to fit in 3h limit.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e16b811d23039f6136cecdf6cbb043a0fb831b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e16b811d23039f6136cecdf6cbb043a0fb831b1"}]},"branch":"refs/heads/master"},"085a61474ee3ccb9e32395e895a438d984026d8d":{"kind":"REWORK","_number":49,"created":"2019-06-28 06:12:30.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/49"}}},"commit":{"parents":[{"commit":"b27e99c2b3f83b714759f023ffc3e80842ea930e","subject":"Merge \"Move back to CentOS StorageSIG repos for Ceph and Ganesha\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b27e99c2b3f83b714759f023ffc3e80842ea930e"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2018-05-16 02:22:20.000000000","tz":480},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-06-28 06:11:44.000000000","tz":120},"subject":"Add kolla-build-debian-source-arm64 experimental job","message":"Add kolla-build-debian-source-arm64 experimental job\n\nImages list shortened to fit in 3h limit.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nDepends-on: Ie259a38aad84bd8f5392f19791edfbd1f8ec4b1a\nDepends-on: Id13545224fa0fbb729d46386a10505e0989fc3c6\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/085a61474ee3ccb9e32395e895a438d984026d8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/085a61474ee3ccb9e32395e895a438d984026d8d"}]},"branch":"refs/heads/master"},"65caff21e179d9ae7900521d12fb5ba88394e178":{"kind":"REWORK","_number":50,"created":"2019-07-15 11:50:48.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/50"}}},"commit":{"parents":[{"commit":"8ff86cfb5f521bba7ffce0e7699d425d0f11732d","subject":"Add CentOS/source aarch64 experimental job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ff86cfb5f521bba7ffce0e7699d425d0f11732d"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-15 11:49:35.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-15 11:49:35.000000000","tz":120},"subject":"Add Debian/source aarch64 experimental job","message":"Add Debian/source aarch64 experimental job\n\nAdd kolla debian arm64 source building job.\n\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65caff21e179d9ae7900521d12fb5ba88394e178"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65caff21e179d9ae7900521d12fb5ba88394e178"}]},"branch":"refs/heads/master"},"d9722a933b490b05c30ab5c76a396dc466340caa":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":51,"created":"2019-07-15 12:38:11.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/51"}}},"commit":{"parents":[{"commit":"50155fa933ba0a22b71814c52b2ee8b6e7603bee","subject":"Add CentOS/source aarch64 experimental job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50155fa933ba0a22b71814c52b2ee8b6e7603bee"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-15 11:49:35.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-15 12:37:29.000000000","tz":120},"subject":"Add Debian/source aarch64 experimental job","message":"Add Debian/source aarch64 experimental job\n\nAdd kolla debian arm64 source building job.\n\nAuthored-By: Xinliang Liu \u003cxinliang.liu@linaro.org\u003e\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9722a933b490b05c30ab5c76a396dc466340caa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9722a933b490b05c30ab5c76a396dc466340caa"}]},"branch":"refs/heads/master"},"4f073ba60037cb5b18c87274c27739804c2c1da6":{"kind":"REWORK","_number":52,"created":"2019-08-28 10:49:49.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/52"}}},"commit":{"parents":[{"commit":"6b47c82d1939dc76f5aabd094209639dcc573a5b","subject":"Add CentOS/source aarch64 experimental job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b47c82d1939dc76f5aabd094209639dcc573a5b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-15 11:49:35.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-08-28 10:48:33.000000000","tz":120},"subject":"Add Debian/source aarch64 experimental job","message":"Add Debian/source aarch64 experimental job\n\nAdd kolla debian arm64 source building job.\n\nAuthored-By: Xinliang Liu \u003cxinliang.liu@linaro.org\u003e\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4f073ba60037cb5b18c87274c27739804c2c1da6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4f073ba60037cb5b18c87274c27739804c2c1da6"}]},"branch":"refs/heads/master"},"fb5b262b214c22072e7e8ff6a5e3580e830b424d":{"kind":"REWORK","_number":53,"created":"2019-09-20 15:54:19.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/53"}}},"commit":{"parents":[{"commit":"fcc77a172b3c5dacd4577f20025027c3be84481f","subject":"Add CentOS/source aarch64 experimental job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fcc77a172b3c5dacd4577f20025027c3be84481f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 15:51:05.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 15:51:30.000000000","tz":120},"subject":"Add Debian/source aarch64 experimental job","message":"Add Debian/source aarch64 experimental job\n\nAdd kolla debian arm64 source building job.\n\nAuthored-By: Xinliang Liu \u003cxinliang.liu@linaro.org\u003e\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb5b262b214c22072e7e8ff6a5e3580e830b424d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb5b262b214c22072e7e8ff6a5e3580e830b424d"}]},"branch":"refs/heads/master"},"51597ab526287954ee37400ff2e506aba73544d7":{"kind":"REWORK","_number":54,"created":"2019-09-20 15:56:22.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/54"}}},"commit":{"parents":[{"commit":"fcc77a172b3c5dacd4577f20025027c3be84481f","subject":"Add CentOS/source aarch64 experimental job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fcc77a172b3c5dacd4577f20025027c3be84481f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 15:51:05.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 15:56:09.000000000","tz":120},"subject":"Add Debian/source aarch64 experimental job","message":"Add Debian/source aarch64 experimental job\n\nAdd kolla debian arm64 source building job.\n\nAuthored-By: Xinliang Liu \u003cxinliang.liu@linaro.org\u003e\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51597ab526287954ee37400ff2e506aba73544d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51597ab526287954ee37400ff2e506aba73544d7"}]},"branch":"refs/heads/master"},"fa73636261d7cac056bdd2f7a7d4f8aa794e97d5":{"kind":"TRIVIAL_REBASE","_number":55,"created":"2019-09-20 15:58:25.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/55"}}},"commit":{"parents":[{"commit":"40d448cb61df735948f72a81e723dcc71b7f8291","subject":"Add CentOS/source aarch64 experimental job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/40d448cb61df735948f72a81e723dcc71b7f8291"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 15:51:05.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 15:58:01.000000000","tz":120},"subject":"Add Debian/source aarch64 experimental job","message":"Add Debian/source aarch64 experimental job\n\nAdd kolla debian arm64 source building job.\n\nAuthored-By: Xinliang Liu \u003cxinliang.liu@linaro.org\u003e\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa73636261d7cac056bdd2f7a7d4f8aa794e97d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa73636261d7cac056bdd2f7a7d4f8aa794e97d5"}]},"branch":"refs/heads/master"},"9cb6f757361665e72972d8312260adc343e380e2":{"kind":"REWORK","_number":56,"created":"2019-11-20 09:25:03.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/56"}}},"commit":{"parents":[{"commit":"635f4f74c2005ef1b864be41a2d882799ce87719","subject":"Add CentOS/source aarch64 experimental job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/635f4f74c2005ef1b864be41a2d882799ce87719"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 15:51:05.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-20 09:24:37.000000000","tz":60},"subject":"Add Debian/source aarch64 experimental job","message":"Add Debian/source aarch64 experimental job\n\nAdd kolla debian arm64 source building job.\n\nAuthored-By: Xinliang Liu \u003cxinliang.liu@linaro.org\u003e\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9cb6f757361665e72972d8312260adc343e380e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9cb6f757361665e72972d8312260adc343e380e2"}]},"branch":"refs/heads/master"},"f08c952a684135793fbd71e2e1d4f580c843936d":{"kind":"TRIVIAL_REBASE","_number":57,"created":"2019-12-11 09:36:51.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/57"}}},"commit":{"parents":[{"commit":"524622b4492c526b2da8c5be3ae558ee649de7a4","subject":"Add CentOS/source aarch64 experimental job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/524622b4492c526b2da8c5be3ae558ee649de7a4"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 15:51:05.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-11 09:36:26.000000000","tz":60},"subject":"Add Debian/source aarch64 experimental job","message":"Add Debian/source aarch64 experimental job\n\nAdd kolla debian arm64 source building job.\n\nAuthored-By: Xinliang Liu \u003cxinliang.liu@linaro.org\u003e\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f08c952a684135793fbd71e2e1d4f580c843936d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f08c952a684135793fbd71e2e1d4f580c843936d"}]},"branch":"refs/heads/master"},"1e1289882e7127b47a04a6d34606432f201b8b6a":{"kind":"REWORK","_number":58,"created":"2019-12-13 12:13:17.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/58"}}},"commit":{"parents":[{"commit":"2ee796bdce78aadaeaf42796a4f4dce9e7471ce0","subject":"Merge \"Use APT pinning to set preferred version of RabbitMQ\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ee796bdce78aadaeaf42796a4f4dce9e7471ce0"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 15:44:06.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-13 12:12:07.000000000","tz":60},"subject":"Add Debian/source aarch64 experimental job","message":"Add Debian/source aarch64 experimental job\n\nAdd kolla debian arm64 source building job.\n\nAuthored-By: Xinliang Liu \u003cxinliang.liu@linaro.org\u003e\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nDepends-On: https://review.opendev.org/#/c/698606\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e1289882e7127b47a04a6d34606432f201b8b6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e1289882e7127b47a04a6d34606432f201b8b6a"}]},"branch":"refs/heads/master"},"b456952b5cc0fa59698b267e06f5a958ae1ec09d":{"kind":"NO_CODE_CHANGE","_number":59,"created":"2019-12-13 15:52:50.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/557659/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/59"}}},"commit":{"parents":[{"commit":"2ee796bdce78aadaeaf42796a4f4dce9e7471ce0","subject":"Merge \"Use APT pinning to set preferred version of RabbitMQ\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ee796bdce78aadaeaf42796a4f4dce9e7471ce0"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 15:44:06.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-13 15:52:46.000000000","tz":0},"subject":"Add Debian/source aarch64 CI job","message":"Add Debian/source aarch64 CI job\n\nAArch64 is getting popular and not all changes working on x86-64 work\nproperly on aarch64 architecture. Let check them.\n\nJob is running in separate pipeline due to low number of aarch64 nodes.\n\nAuthored-By: Xinliang Liu \u003cxinliang.liu@linaro.org\u003e\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nDepends-On: https://review.opendev.org/#/c/698606\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b456952b5cc0fa59698b267e06f5a958ae1ec09d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b456952b5cc0fa59698b267e06f5a958ae1ec09d"}]},"branch":"refs/heads/master"},"d6935edb43ab577b2895d701d6d7e7e555934d8b":{"kind":"REWORK","_number":60,"created":"2019-12-16 08:18:06.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/59/557659/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/557659/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/557659/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/557659/60"}}},"commit":{"parents":[{"commit":"2ee796bdce78aadaeaf42796a4f4dce9e7471ce0","subject":"Merge \"Use APT pinning to set preferred version of RabbitMQ\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ee796bdce78aadaeaf42796a4f4dce9e7471ce0"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 15:44:06.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-12-16 08:17:38.000000000","tz":0},"subject":"Add Debian/source aarch64 CI job","message":"Add Debian/source aarch64 CI job\n\nAArch64 is getting popular and not all changes working on x86-64 work\nproperly on aarch64 architecture. Let check them.\n\nJob is running in separate pipeline due to low number of aarch64 nodes.\n\nAuthored-By: Xinliang Liu \u003cxinliang.liu@linaro.org\u003e\nCo-Authored-By: Jeffrey Zhang \u003czhang.lei.fly@gmail.com\u003e\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nDepends-On: https://review.opendev.org/#/c/698606\n\nChange-Id: I7e7802dfa4f4d9c61d01dba4b77c01cfb24301c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6935edb43ab577b2895d701d6d7e7e555934d8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6935edb43ab577b2895d701d6d7e7e555934d8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
