)]}'
{"id":"openstack%2Fkolla~613815","triplet_id":"openstack%2Fkolla~master~I56f230d66f75dd0546325676278f91579f08c822","project":"openstack/kolla","branch":"master","topic":"bp/centos-rhel-8","hashtags":[],"change_id":"I56f230d66f75dd0546325676278f91579f08c822","subject":"Remove tgtd and scsi-target-utils support in CentOS/RHEL 8","status":"MERGED","created":"2018-10-28 22:19:55.000000000","updated":"2020-01-10 13:00:51.000000000","submitted":"2020-01-10 12:58:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"613815-1578661132032-a981c286","meta_rev_id":"73837a5e745ec830c557e0a10260dc4142897198","_number":613815,"virtual_id_number":613815,"owner":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-10 12:58:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-09 17:29:38.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},{"value":2,"date":"2020-01-10 11:17:49.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-01-10 10:52:42.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},{"value":1,"date":"2020-01-10 11:17:49.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2020-01-10 10:32:01.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-01-18 15:17:52.000000000","updated_by":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"reviewer":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"state":"REVIEWER"},{"updated":"2019-01-21 12:32:50.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-10-28 07:33:00.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2020-01-09 17:29:38.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-10 10:32:01.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-01-10 10:52:42.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-10 11:17:49.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-01-10 12:58:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"59c6437420d5392772b63974df55202814bb3878","author":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"date":"2018-10-28 22:19:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fda175b97cea029aa704e5767ace6fc4cee94eee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-28 23:20:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/15/613815/1/check/kolla-build-centos-binary/3ad2440/ : FAILURE in 51m 21s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/15/613815/1/check/kolla-build-centos-source/3a7bb98/ : FAILURE in 6m 52s\n- kolla-ansible-centos-source http://logs.openstack.org/15/613815/1/check/kolla-ansible-centos-source/8fecd9a/ : FAILURE in 5m 51s\n- kolla-ansible-centos-binary http://logs.openstack.org/15/613815/1/check/kolla-ansible-centos-binary/3849c21/ : FAILURE in 6m 06s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/15/613815/1/check/kolla-build-oraclelinux-binary/c77b0a0/ : SUCCESS in 55m 41s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/15/613815/1/check/kolla-build-oraclelinux-source/3ccb032/ : FAILURE in 4m 36s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/15/613815/1/check/kolla-ansible-oraclelinux-source/34eb479/ : FAILURE in 5m 53s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/15/613815/1/check/kolla-ansible-oraclelinux-binary/f5fb024/ : FAILURE in 6m 37s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/15/613815/1/check/openstack-tox-lower-constraints/94f1db6/ : SUCCESS in 3m 02s\n- openstack-tox-pep8 http://logs.openstack.org/15/613815/1/check/openstack-tox-pep8/e780a5f/ : SUCCESS in 3m 26s\n- openstack-tox-py27 http://logs.openstack.org/15/613815/1/check/openstack-tox-py27/e214e5b/ : SUCCESS in 2m 55s\n- openstack-tox-py35 http://logs.openstack.org/15/613815/1/check/openstack-tox-py35/8f6627d/ : SUCCESS in 2m 37s\n- openstack-tox-py36 http://logs.openstack.org/15/613815/1/check/openstack-tox-py36/57d700e/ : SUCCESS in 2m 46s\n- openstack-tox-docs http://logs.openstack.org/15/613815/1/check/openstack-tox-docs/c1cbaa1/html/ : SUCCESS in 2m 29s\n- kolla-build-ubuntu-binary http://logs.openstack.org/15/613815/1/check/kolla-build-ubuntu-binary/f4f7bb2/ : FAILURE in 2m 27s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/15/613815/1/check/kolla-build-ubuntu-source/e66ff7e/ : FAILURE in 4m 01s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/15/613815/1/check/kolla-ansible-ubuntu-source/dac6fec/ : FAILURE in 4m 29s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/15/613815/1/check/kolla-ansible-ubuntu-binary/5548b25/ : FAILURE in 4m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6539bc85a98ceb1eb24026104016430fa3d585c0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-10-31 15:03:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"173b40b87ece533a72ce579df53e5749328b5ce2","author":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"date":"2018-10-31 21:18:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0a123bbf7eb42091f979a39a704392765fd794e","author":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"date":"2018-10-31 21:22:31.000000000","message":"Patch Set 1:\n\nWhy is the CI failing? There are errors \u0027no space left on device\u0027 or \u0027Get https://registry-1.docker.io/v2/library/registry/manifests/2: received unexpected HTTP status: 503 Service Unavailable.\u0027\n\nNot sure if this change is breaking CI or something else :(","accounts_in_message":[],"_revision_number":1},{"id":"b2299f4b29c6a38690e4ef75e64ad03514d6050f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-01 09:48:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed191e9e5a9785dfd071ea686dcced94fc6f3c8b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-01 09:54:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"54decf44f786d9152fa6fd35d03a892ce5ad7e40","author":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"date":"2018-11-03 14:32:20.000000000","message":"Patch Set 1: Workflow-1\n\nCan\u0027t find anything solid that does not have paywall. There is https://groveronline.com/2012/11/tgtd-lio-in-rhel-7/\n\nRHEL KB article states: \u0027targetcli replace the scsi-target-utils in version RHEL7. Using targetcli to create and export storage SCSI.\u0027\n\nHow to skip image build for specific OS?","accounts_in_message":[],"_revision_number":1},{"id":"41e4460e81425f53bb76f9065807391ec3bd5498","author":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"date":"2018-11-03 14:32:48.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"672196e0700d4315d6a3a48ff8dd150583c03df8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-05 08:46:08.000000000","message":"Patch Set 1:\n\n\u003e Can\u0027t find anything solid that does not have paywall. There is\n \u003e https://groveronline.com/2012/11/tgtd-lio-in-rhel-7/\n \u003e \n \u003e RHEL KB article states: \u0027targetcli replace the scsi-target-utils in\n \u003e version RHEL7. Using targetcli to create and export storage SCSI.\u0027\n \u003e \n \u003e How to skip image build for specific OS?\n\nI see. So the package we\u0027re installing comes from EPEL. https://rpmfind.net/linux/RPM/epel/7/x86_64/Packages/s/scsi-target-utils-1.0.55-4.el7.x86_64.html\n\nSkipped images are defined in kolla/image/build.py, in SKIPPED_IMAGES.","accounts_in_message":[],"_revision_number":1},{"id":"f2c8987ae147b416799c13612fb8977cc354d33f","author":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"date":"2018-11-18 20:37:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"41771f4e174ed893b46ab9ec56c5c73b8cad9446","author":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"date":"2018-11-18 20:51:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"98c4824219cf22232024a83f8d8da610fd7d18eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-18 22:54:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/15/613815/3/check/kolla-build-centos-binary/f5f38e3/ : FAILURE in 1h 04m 54s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/15/613815/3/check/kolla-build-centos-source/8866b25/ : SUCCESS in 54m 08s\n- kolla-ansible-centos-source http://logs.openstack.org/15/613815/3/check/kolla-ansible-centos-source/ccd4691/ : SUCCESS in 1h 22m 17s\n- kolla-ansible-centos-binary http://logs.openstack.org/15/613815/3/check/kolla-ansible-centos-binary/16bb3b3/ : TIMED_OUT in 2h 01m 46s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/15/613815/3/check/kolla-build-oraclelinux-binary/6961a9e/ : FAILURE in 58m 55s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/15/613815/3/check/kolla-build-oraclelinux-source/a2afecf/ : FAILURE in 22m 46s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/15/613815/3/check/kolla-ansible-oraclelinux-source/b69596d/ : SUCCESS in 1h 20m 11s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/15/613815/3/check/kolla-ansible-oraclelinux-binary/d2d6a26/ : SUCCESS in 1h 38m 51s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/15/613815/3/check/openstack-tox-py36/797f472/ : SUCCESS in 2m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/15/613815/3/check/openstack-tox-lower-constraints/c7dd6e0/ : SUCCESS in 3m 01s\n- openstack-tox-pep8 http://logs.openstack.org/15/613815/3/check/openstack-tox-pep8/2cfe529/ : SUCCESS in 2m 25s\n- openstack-tox-py27 http://logs.openstack.org/15/613815/3/check/openstack-tox-py27/520edab/ : SUCCESS in 3m 34s\n- openstack-tox-py35 http://logs.openstack.org/15/613815/3/check/openstack-tox-py35/52290f6/ : SUCCESS in 2m 52s\n- openstack-tox-docs http://logs.openstack.org/15/613815/3/check/openstack-tox-docs/34ce0b0/html/ : SUCCESS in 2m 48s\n- kolla-build-ubuntu-binary http://logs.openstack.org/15/613815/3/check/kolla-build-ubuntu-binary/7d55b0e/ : SUCCESS in 1h 07m 43s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/15/613815/3/check/kolla-build-ubuntu-source/0003924/ : SUCCESS in 28m 09s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/15/613815/3/check/kolla-ansible-ubuntu-source/a94e787/ : SUCCESS in 1h 03m 15s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/15/613815/3/check/kolla-ansible-ubuntu-binary/ec3bbc5/ : TIMED_OUT in 2h 00m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7cb24d6033d0a20dc66d0317bdb37f93454ea68a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-19 10:15:02.000000000","message":"Patch Set 3: Code-Review-1\n\nThis will need a release note, if we are removing images.\n\nWe probably also need a change in kolla-ansible that allows you to use iscsid without tgtd, since IIRC the two are connected.\n\nI guess that tripleo doesn\u0027t use tgtd?","accounts_in_message":[],"_revision_number":3},{"id":"12d3cb24dc9a76578fdd9e16ec8a862bcb518121","author":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"date":"2018-11-21 00:11:27.000000000","message":"Patch Set 3: Workflow-1\n\ngonna check","accounts_in_message":[],"_revision_number":3},{"id":"14dfca4b8ad658f0028b301769a85ee4b3acded0","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2018-11-21 17:43:55.000000000","message":"Patch Set 3: Code-Review-1\n\nPlease remove scsi-target-utils from the cinder-volume image too.\n\nIn TripleO, we\u0027re building the images without EPEL already, so we\u0027re carrying the same change:\nhttps://github.com/openstack/tripleo-common/blob/c6fab39/container-images/tripleo_kolla_template_overrides.j2#L24-L25","accounts_in_message":[],"_revision_number":3},{"id":"e4ed9969e6064d748a079b6f6c045634beafd5f3","author":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"date":"2019-01-15 08:52:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8b0aed0db7701ef70b53fbbe4a8275d83bd8d93f","author":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"date":"2019-01-15 08:53:14.000000000","message":"Patch Set 3:\n\n\u003e This will need a release note, if we are removing images.\n \u003e \n \u003e We probably also need a change in kolla-ansible that allows you to\n \u003e use iscsid without tgtd, since IIRC the two are connected.\n \u003e \n \u003e I guess that tripleo doesn\u0027t use tgtd?\n\nAs far as I know there are no changes necessary in Kolla-ansible for iscsid.","accounts_in_message":[],"_revision_number":3},{"id":"193546938679cddd052ba624379db7b564ea8290","author":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"date":"2019-01-15 08:56:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3bf21ba0857abe606a87340a1ff797a58ac0cdcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-15 11:03:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/15/613815/5/check/kolla-build-centos-binary/8329d11/ : FAILURE in 53m 35s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/15/613815/5/check/kolla-build-centos-source/5c40b30/ : FAILURE in 1h 09m 36s\n- kolla-ansible-centos-source http://logs.openstack.org/15/613815/5/check/kolla-ansible-centos-source/c4cb6a8/ : SUCCESS in 1h 25m 46s\n- kolla-ansible-centos-binary http://logs.openstack.org/15/613815/5/check/kolla-ansible-centos-binary/82122fa/ : SUCCESS in 1h 08m 24s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/15/613815/5/check/tripleo-build-containers-centos-7/4dafd0d/ : SUCCESS in 37m 20s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/15/613815/5/check/kolla-build-oraclelinux-binary/adcf198/ : FAILURE in 1h 13m 44s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/15/613815/5/check/kolla-build-oraclelinux-source/185ce7a/ : FAILURE in 1h 17m 13s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/15/613815/5/check/kolla-ansible-oraclelinux-source/23e90a2/ : SUCCESS in 1h 15m 42s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/15/613815/5/check/kolla-ansible-oraclelinux-binary/fba03f0/ : SUCCESS in 1h 09m 11s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/15/613815/5/check/openstack-tox-lower-constraints/7ade72b/ : SUCCESS in 3m 38s\n- openstack-tox-py36 http://logs.openstack.org/15/613815/5/check/openstack-tox-py36/13c5b6e/ : SUCCESS in 3m 07s\n- openstack-tox-pep8 http://logs.openstack.org/15/613815/5/check/openstack-tox-pep8/71e057f/ : FAILURE in 3m 26s\n- openstack-tox-py27 http://logs.openstack.org/15/613815/5/check/openstack-tox-py27/ef20c8d/ : SUCCESS in 3m 17s\n- openstack-tox-py35 http://logs.openstack.org/15/613815/5/check/openstack-tox-py35/d88982d/ : SUCCESS in 3m 41s\n- openstack-tox-docs http://logs.openstack.org/15/613815/5/check/openstack-tox-docs/f9690ca/html/ : SUCCESS in 3m 13s\n- build-openstack-releasenotes http://logs.openstack.org/15/613815/5/check/build-openstack-releasenotes/4efd4a1/html/ : SUCCESS in 3m 36s\n- kolla-build-ubuntu-binary http://logs.openstack.org/15/613815/5/check/kolla-build-ubuntu-binary/f5108dc/ : SUCCESS in 1h 32m 05s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/15/613815/5/check/kolla-build-ubuntu-source/7b68f0b/ : SUCCESS in 33m 24s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/15/613815/5/check/kolla-ansible-ubuntu-source/54d2f31/ : SUCCESS in 1h 00m 16s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/15/613815/5/check/kolla-ansible-ubuntu-binary/5b9a677/ : TIMED_OUT in 2h 01m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8f7cbc120f99b8ce88409996b926b96fdf1f1ed0","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2019-01-16 15:53:01.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)\n\nThanks for the cleanup. There is still one more thing that needs to be removed in the cinder-volume image and then I think we\u0027re good to go.","accounts_in_message":[],"_revision_number":5},{"id":"efd1d81f8969da4e2da56ee59b0314634e62ab21","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-17 11:55:45.000000000","message":"Patch Set 5:\n\n\u003e \u003e This will need a release note, if we are removing images.\n \u003e \u003e\n \u003e \u003e We probably also need a change in kolla-ansible that allows you\n \u003e to\n \u003e \u003e use iscsid without tgtd, since IIRC the two are connected.\n \u003e \u003e\n \u003e \u003e I guess that tripleo doesn\u0027t use tgtd?\n \u003e \n \u003e As far as I know there are no changes necessary in Kolla-ansible\n \u003e for iscsid.\n\nThe kolla-ansible change required needs to disable the tgtd container on RPM-based platforms. Without that, kolla-ansible would try to deploy a tgtd container but no image will be available.\n\nThe change will probably be easiest if based upon https://review.openstack.org/#/c/569610/, but that change is a little stale :(","accounts_in_message":[],"_revision_number":5},{"id":"82a6bfa14a6019f1f3d2362befc1209928d09932","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-17 13:06:11.000000000","message":"Patch Set 5:\n\n\u003e \u003e \u003e This will need a release note, if we are removing images.\n \u003e \u003e \u003e\n \u003e \u003e \u003e We probably also need a change in kolla-ansible that allows you\n \u003e \u003e to\n \u003e \u003e \u003e use iscsid without tgtd, since IIRC the two are connected.\n \u003e \u003e \u003e\n \u003e \u003e \u003e I guess that tripleo doesn\u0027t use tgtd?\n \u003e \u003e\n \u003e \u003e As far as I know there are no changes necessary in Kolla-ansible\n \u003e \u003e for iscsid.\n \u003e \n \u003e The kolla-ansible change required needs to disable the tgtd\n \u003e container on RPM-based platforms. Without that, kolla-ansible would\n \u003e try to deploy a tgtd container but no image will be available.\n \u003e \n \u003e The change will probably be easiest if based upon https://review.openstack.org/#/c/569610/,\n \u003e but that change is a little stale :(\n\nI tried to fix up that patch, and added another to use lioadm on RPM-based distros. Can you add Depends-On: https://review.openstack.org/631479?","accounts_in_message":[],"_revision_number":5},{"id":"12d08e1d000c9f8fdd791db8559ad75ecb37dc41","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-17 13:07:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d1124ab69afb80d88fc677677241368001e23807","author":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"date":"2019-01-17 20:48:15.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e60b037352f421dea5ca5dd7107e2080bbcb2386","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-17 22:50:53.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/15/613815/6/check/kolla-build-centos-binary/a799737/ : SUCCESS in 1h 07m 44s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/15/613815/6/check/kolla-build-centos-source/debb2d7/ : SUCCESS in 1h 14m 41s\n- kolla-ansible-centos-source http://logs.openstack.org/15/613815/6/check/kolla-ansible-centos-source/7f4e183/ : SUCCESS in 1h 26m 05s\n- kolla-ansible-centos-binary http://logs.openstack.org/15/613815/6/check/kolla-ansible-centos-binary/01860b8/ : SUCCESS in 1h 36m 58s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/15/613815/6/check/tripleo-build-containers-centos-7/b9ec0f5/ : SUCCESS in 39m 04s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/15/613815/6/check/kolla-build-oraclelinux-binary/66eb737/ : SUCCESS in 1h 12m 56s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/15/613815/6/check/kolla-build-oraclelinux-source/ead9ef8/ : SUCCESS in 1h 35m 12s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/15/613815/6/check/kolla-ansible-oraclelinux-source/1adda9f/ : SUCCESS in 1h 26m 53s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/15/613815/6/check/kolla-ansible-oraclelinux-binary/d2de6de/ : SUCCESS in 1h 14m 26s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/15/613815/6/check/openstack-tox-lower-constraints/551b88a/ : SUCCESS in 3m 50s\n- openstack-tox-py36 http://logs.openstack.org/15/613815/6/check/openstack-tox-py36/b5d47de/ : SUCCESS in 7m 14s\n- openstack-tox-pep8 http://logs.openstack.org/15/613815/6/check/openstack-tox-pep8/4e74ef8/ : FAILURE in 3m 39s\n- openstack-tox-py27 http://logs.openstack.org/15/613815/6/check/openstack-tox-py27/9e22eef/ : SUCCESS in 3m 18s\n- openstack-tox-py35 http://logs.openstack.org/15/613815/6/check/openstack-tox-py35/672cf14/ : SUCCESS in 3m 33s\n- openstack-tox-docs http://logs.openstack.org/15/613815/6/check/openstack-tox-docs/7293d5f/html/ : SUCCESS in 2m 41s\n- build-openstack-releasenotes http://logs.openstack.org/15/613815/6/check/build-openstack-releasenotes/7995de9/html/ : SUCCESS in 8m 17s\n- kolla-build-ubuntu-binary http://logs.openstack.org/15/613815/6/check/kolla-build-ubuntu-binary/f3f027e/ : SUCCESS in 1h 19m 05s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/15/613815/6/check/kolla-build-ubuntu-source/378c753/ : SUCCESS in 41m 55s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/15/613815/6/check/kolla-ansible-ubuntu-source/f24d783/ : SUCCESS in 51m 03s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/15/613815/6/check/kolla-ansible-ubuntu-binary/a70e247/ : SUCCESS in 1h 52m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"63e9ac2080ea2a41fbb54e7f7088080e32e4b18a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-18 10:23:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"383e14015a11641ae1dc4b8db4805eafbb6bf622","author":{"_account_id":25588,"name":"Maciej Kucia","email":"maciej@kucia.net","username":"m.kucia"},"date":"2019-01-18 14:30:33.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"3715fb7d116c50f9a9394ca33ba392dec468eb9c","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2019-01-18 15:17:52.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"7a24cb494c1bf3d51b46811bc09e1ec5906758d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-18 16:42:29.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/15/613815/7/check/kolla-build-centos-binary/9811d7f/ : SUCCESS in 1h 02m 52s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/15/613815/7/check/kolla-build-centos-source/377664d/ : SUCCESS in 1h 05m 37s\n- kolla-ansible-centos-source http://logs.openstack.org/15/613815/7/check/kolla-ansible-centos-source/e3914f6/ : SUCCESS in 1h 25m 04s\n- kolla-ansible-centos-binary http://logs.openstack.org/15/613815/7/check/kolla-ansible-centos-binary/65f9708/ : SUCCESS in 1h 09m 17s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/15/613815/7/check/tripleo-build-containers-centos-7/f14cac1/ : SUCCESS in 38m 20s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/15/613815/7/check/kolla-build-oraclelinux-binary/cf19a22/ : SUCCESS in 1h 26m 03s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/15/613815/7/check/kolla-build-oraclelinux-source/7cd7afc/ : SUCCESS in 1h 18m 54s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/15/613815/7/check/kolla-ansible-oraclelinux-source/d9c7168/ : SUCCESS in 1h 53m 02s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/15/613815/7/check/kolla-ansible-oraclelinux-binary/55ffa73/ : TIMED_OUT in 2h 01m 33s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/15/613815/7/check/openstack-tox-lower-constraints/49fdb30/ : SUCCESS in 4m 23s\n- openstack-tox-py36 http://logs.openstack.org/15/613815/7/check/openstack-tox-py36/176aba1/ : SUCCESS in 3m 10s\n- openstack-tox-pep8 http://logs.openstack.org/15/613815/7/check/openstack-tox-pep8/6c8ccf0/ : SUCCESS in 3m 39s\n- openstack-tox-py27 http://logs.openstack.org/15/613815/7/check/openstack-tox-py27/dffdbe5/ : SUCCESS in 4m 19s\n- openstack-tox-py35 http://logs.openstack.org/15/613815/7/check/openstack-tox-py35/15b360a/ : SUCCESS in 3m 16s\n- openstack-tox-docs http://logs.openstack.org/15/613815/7/check/openstack-tox-docs/2b8faff/html/ : SUCCESS in 3m 15s\n- build-openstack-releasenotes http://logs.openstack.org/15/613815/7/check/build-openstack-releasenotes/1a483ba/html/ : SUCCESS in 4m 16s\n- kolla-build-ubuntu-binary http://logs.openstack.org/15/613815/7/check/kolla-build-ubuntu-binary/ae2b42e/ : SUCCESS in 2h 00m 51s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/15/613815/7/check/kolla-build-ubuntu-source/e9aa03d/ : SUCCESS in 28m 50s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/15/613815/7/check/kolla-ansible-ubuntu-source/5607a26/ : SUCCESS in 52m 44s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/15/613815/7/check/kolla-ansible-ubuntu-binary/aa95959/ : SUCCESS in 2h 00m 49s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"759d7b9c60c8b7ebf55e3d21a9477c81a1943399","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-18 16:43:37.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"60e90a127b50575f6f7b547ab506d1c9fd2e9b23","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-18 16:43:55.000000000","message":"Patch Set 7: -Workflow","accounts_in_message":[],"_revision_number":7},{"id":"39ba51f39a68e5d9eec01fd8865b90c7c1237b45","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-18 16:44:42.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"edf2f7445af8d90541da82035982baf29ef1f0c0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-18 16:44:59.000000000","message":"Patch Set 8: Code-Review+2\n\nAdded the dependency on the kolla-ansible patch.","accounts_in_message":[],"_revision_number":8},{"id":"493de06fb6bd23b0e22c6d285018b06b21b9ec4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-18 18:52:58.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/15/613815/8/check/kolla-build-centos-binary/700d882/ : SUCCESS in 54m 02s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/15/613815/8/check/kolla-build-centos-source/7af4df8/ : SUCCESS in 1h 14m 16s\n- kolla-ansible-centos-source http://logs.openstack.org/15/613815/8/check/kolla-ansible-centos-source/c6bb9f8/ : SUCCESS in 1h 37m 55s\n- kolla-ansible-centos-binary http://logs.openstack.org/15/613815/8/check/kolla-ansible-centos-binary/7aff869/ : SUCCESS in 1h 11m 25s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/15/613815/8/check/tripleo-build-containers-centos-7/af7f1d6/ : SUCCESS in 37m 56s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/15/613815/8/check/kolla-build-oraclelinux-binary/2fb2a21/ : SUCCESS in 1h 24m 14s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/15/613815/8/check/kolla-build-oraclelinux-source/fb5c71a/ : SUCCESS in 1h 01m 16s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/15/613815/8/check/kolla-ansible-oraclelinux-source/27c59ad/ : SUCCESS in 1h 13m 54s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/15/613815/8/check/kolla-ansible-oraclelinux-binary/f9f21c0/ : SUCCESS in 1h 05m 34s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/15/613815/8/check/openstack-tox-lower-constraints/4dbac78/ : SUCCESS in 3m 43s\n- openstack-tox-py36 http://logs.openstack.org/15/613815/8/check/openstack-tox-py36/159fd21/ : SUCCESS in 4m 11s\n- openstack-tox-pep8 http://logs.openstack.org/15/613815/8/check/openstack-tox-pep8/98a7c39/ : SUCCESS in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/15/613815/8/check/openstack-tox-py27/a46600b/ : SUCCESS in 3m 55s\n- openstack-tox-py35 http://logs.openstack.org/15/613815/8/check/openstack-tox-py35/42af0dc/ : SUCCESS in 3m 40s\n- openstack-tox-docs http://logs.openstack.org/15/613815/8/check/openstack-tox-docs/c98e525/html/ : SUCCESS in 3m 19s\n- build-openstack-releasenotes http://logs.openstack.org/15/613815/8/check/build-openstack-releasenotes/7bea395/html/ : SUCCESS in 4m 05s\n- kolla-build-ubuntu-binary http://logs.openstack.org/15/613815/8/check/kolla-build-ubuntu-binary/d224e45/ : SUCCESS in 1h 33m 17s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/15/613815/8/check/kolla-build-ubuntu-source/278cddc/ : SUCCESS in 37m 56s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/15/613815/8/check/kolla-ansible-ubuntu-source/86b81aa/ : SUCCESS in 1h 03m 40s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/15/613815/8/check/kolla-ansible-ubuntu-binary/7655154/ : TIMED_OUT in 2h 01m 13s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b794c72a1ccac9270d169479047b00e23c35017b","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-01-21 12:32:25.000000000","message":"Patch Set 8:\n\nwhat will happen with deployments already using tgtd and scsi? Is backwards compatible, will have some permission issue, will volumes already attached be unavailable during upgrade of need to be fully recreated?","accounts_in_message":[],"_revision_number":8},{"id":"123d7c6b958173ce385b80ba8f8d63c281362f65","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-01-21 12:32:50.000000000","message":"Patch Set 8: Code-Review-1\n\n-1 to avoid merge before the answers","accounts_in_message":[],"_revision_number":8},{"id":"34cc914519f0cc59158dd7301d6487d269102e49","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-21 13:12:40.000000000","message":"Patch Set 8: -Code-Review","accounts_in_message":[],"_revision_number":8},{"id":"8907e240aa1f3e9955e5e9a35ccdde6c852f360d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-21 13:13:03.000000000","message":"Patch Set 8:\n\n\u003e what will happen with deployments already using tgtd and scsi? Is\n \u003e backwards compatible, will have some permission issue, will volumes\n \u003e already attached be unavailable during upgrade of need to be fully\n \u003e recreated?\n\nGood questions, they also apply to the kolla-ansible default change https://review.openstack.org/#/c/631479/","accounts_in_message":[],"_revision_number":8},{"id":"aed9749b2f2c754989b800e16a32b6d61d3a0128","author":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"date":"2019-01-26 20:04:33.000000000","message":"Patch Set 8:\n\nI wish I could answer but I wasn\u0027t able to make tgtd work in first place and that would be a first step to consider migration.","accounts_in_message":[],"_revision_number":8},{"id":"e54d91dbdd20b302666b8673b766c19710eb43e4","author":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"date":"2019-02-09 21:04:49.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"9178b2e81d75191c5345072367ce9b77caa0e9a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-09 22:47:25.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/15/613815/9/check/kolla-build-centos-binary/de1dece/ : SUCCESS in 1h 00m 32s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/15/613815/9/check/kolla-build-centos-source/353da41/ : SUCCESS in 1h 21m 45s\n- kolla-ansible-centos-source http://logs.openstack.org/15/613815/9/check/kolla-ansible-centos-source/4b4a6e2/ : SUCCESS in 1h 03m 54s\n- kolla-ansible-centos-binary http://logs.openstack.org/15/613815/9/check/kolla-ansible-centos-binary/e9b77ed/ : FAILURE in 18m 40s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/15/613815/9/check/tripleo-build-containers-centos-7/66e1dd0/ : SUCCESS in 50m 01s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/15/613815/9/check/kolla-build-oraclelinux-binary/00b4f91/ : SUCCESS in 1h 22m 02s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/15/613815/9/check/kolla-build-oraclelinux-source/0e48a62/ : SUCCESS in 1h 23m 32s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/15/613815/9/check/kolla-ansible-oraclelinux-source/4e45672/ : SUCCESS in 1h 30m 14s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/15/613815/9/check/kolla-ansible-oraclelinux-binary/22a114a/ : FAILURE in 1h 22m 59s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/15/613815/9/check/openstack-tox-lower-constraints/259522f/ : SUCCESS in 2m 57s\n- openstack-tox-py36 http://logs.openstack.org/15/613815/9/check/openstack-tox-py36/6dd844e/ : SUCCESS in 3m 59s\n- openstack-tox-pep8 http://logs.openstack.org/15/613815/9/check/openstack-tox-pep8/d637ab5/ : FAILURE in 2m 47s\n- openstack-tox-py27 http://logs.openstack.org/15/613815/9/check/openstack-tox-py27/8b36da4/ : SUCCESS in 2m 48s\n- openstack-tox-py35 http://logs.openstack.org/15/613815/9/check/openstack-tox-py35/e2f8102/ : SUCCESS in 2m 48s\n- openstack-tox-docs http://logs.openstack.org/15/613815/9/check/openstack-tox-docs/35d7361/html/ : SUCCESS in 2m 48s\n- build-openstack-releasenotes http://logs.openstack.org/15/613815/9/check/build-openstack-releasenotes/4bf29ba/html/ : SUCCESS in 4m 11s\n- kolla-build-ubuntu-binary http://logs.openstack.org/15/613815/9/check/kolla-build-ubuntu-binary/cf8114e/ : SUCCESS in 1h 14m 42s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/15/613815/9/check/kolla-build-ubuntu-source/d328046/ : SUCCESS in 32m 48s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/15/613815/9/check/kolla-ansible-ubuntu-source/f841030/ : FAILURE in 23m 01s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/15/613815/9/check/kolla-ansible-ubuntu-binary/6695ffa/ : FAILURE in 1h 06m 43s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"085e4ae77a84f98a860c6cfc9ae1d9610010fdae","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2019-02-18 10:59:11.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"9e6f396d1a5274f224baea134af8e8be85379fb9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-18 12:36:41.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/15/613815/10/check/kolla-build-centos-binary/258aabb/ : SUCCESS in 1h 00m 39s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/15/613815/10/check/kolla-build-centos-source/2bd449d/ : SUCCESS in 1h 11m 48s\n- kolla-ansible-centos-source http://logs.openstack.org/15/613815/10/check/kolla-ansible-centos-source/e50d1f3/ : SUCCESS in 1h 12m 57s\n- kolla-ansible-centos-binary http://logs.openstack.org/15/613815/10/check/kolla-ansible-centos-binary/66b2c06/ : SUCCESS in 1h 12m 38s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/15/613815/10/check/tripleo-build-containers-centos-7/dd3a6e0/ : SUCCESS in 41m 56s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/15/613815/10/check/kolla-build-oraclelinux-binary/6e450b9/ : SUCCESS in 1h 12m 47s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/15/613815/10/check/kolla-build-oraclelinux-source/93bff46/ : SUCCESS in 1h 27m 31s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/15/613815/10/check/kolla-ansible-oraclelinux-source/ff0ed9e/ : SUCCESS in 1h 20m 47s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/15/613815/10/check/kolla-ansible-oraclelinux-binary/afb7576/ : SUCCESS in 1h 13m 55s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/15/613815/10/check/openstack-tox-lower-constraints/01ff9d9/ : SUCCESS in 4m 31s\n- openstack-tox-py36 http://logs.openstack.org/15/613815/10/check/openstack-tox-py36/089dc5c/ : SUCCESS in 4m 03s\n- openstack-tox-pep8 http://logs.openstack.org/15/613815/10/check/openstack-tox-pep8/7e54536/ : FAILURE in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/15/613815/10/check/openstack-tox-py27/4cabc76/ : SUCCESS in 4m 34s\n- openstack-tox-py35 http://logs.openstack.org/15/613815/10/check/openstack-tox-py35/b9c258e/ : SUCCESS in 4m 24s\n- openstack-tox-docs http://logs.openstack.org/15/613815/10/check/openstack-tox-docs/0cecaaa/html/ : SUCCESS in 2m 41s\n- build-openstack-releasenotes http://logs.openstack.org/15/613815/10/check/build-openstack-releasenotes/189ffd8/html/ : SUCCESS in 4m 06s\n- kolla-build-ubuntu-binary http://logs.openstack.org/15/613815/10/check/kolla-build-ubuntu-binary/623084a/ : SUCCESS in 1h 17m 39s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/15/613815/10/check/kolla-build-ubuntu-source/b0718f3/ : SUCCESS in 47m 20s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/15/613815/10/check/kolla-ansible-ubuntu-source/b1769dc/ : FAILURE in 44m 44s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/15/613815/10/check/kolla-ansible-ubuntu-binary/21b7f5f/ : SUCCESS in 1h 34m 15s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f5aed2961cce39cdbabb10174bff944791b1760b","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2019-02-21 09:02:20.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"8c74b1288f389a20f7bddaa4c3bbdcf6d801208d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-21 11:09:26.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/15/613815/11/check/kolla-build-centos-binary/a38c6a1/ : SUCCESS in 1h 26m 31s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/15/613815/11/check/kolla-build-centos-source/549f483/ : SUCCESS in 1h 04m 51s\n- kolla-ansible-centos-source http://logs.openstack.org/15/613815/11/check/kolla-ansible-centos-source/def5b28/ : SUCCESS in 1h 24m 45s\n- kolla-ansible-centos-binary http://logs.openstack.org/15/613815/11/check/kolla-ansible-centos-binary/650a60e/ : SUCCESS in 1h 24m 31s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/15/613815/11/check/tripleo-build-containers-centos-7/816a0c6/ : SUCCESS in 51m 50s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/15/613815/11/check/kolla-build-oraclelinux-binary/e24835e/ : SUCCESS in 1h 24m 07s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/15/613815/11/check/kolla-build-oraclelinux-source/7044e9d/ : SUCCESS in 1h 11m 53s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/15/613815/11/check/kolla-ansible-oraclelinux-source/948e5d3/ : SUCCESS in 1h 18m 19s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/15/613815/11/check/kolla-ansible-oraclelinux-binary/ce98b6b/ : SUCCESS in 1h 14m 46s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/15/613815/11/check/openstack-tox-lower-constraints/250f55d/ : SUCCESS in 3m 22s\n- openstack-tox-py36 http://logs.openstack.org/15/613815/11/check/openstack-tox-py36/7d8bafb/ : SUCCESS in 2m 59s\n- openstack-tox-pep8 http://logs.openstack.org/15/613815/11/check/openstack-tox-pep8/ef493d0/ : SUCCESS in 3m 31s\n- openstack-tox-py27 http://logs.openstack.org/15/613815/11/check/openstack-tox-py27/568783c/ : SUCCESS in 3m 03s\n- openstack-tox-py35 http://logs.openstack.org/15/613815/11/check/openstack-tox-py35/8093f79/ : SUCCESS in 2m 50s\n- openstack-tox-docs http://logs.openstack.org/15/613815/11/check/openstack-tox-docs/1c943f8/html/ : SUCCESS in 2m 50s\n- build-openstack-releasenotes http://logs.openstack.org/15/613815/11/check/build-openstack-releasenotes/893ff77/html/ : SUCCESS in 3m 41s\n- kolla-build-ubuntu-binary http://logs.openstack.org/15/613815/11/check/kolla-build-ubuntu-binary/286e873/ : SUCCESS in 1h 18m 17s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/15/613815/11/check/kolla-build-ubuntu-source/c13cf60/ : SUCCESS in 30m 55s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/15/613815/11/check/kolla-ansible-ubuntu-source/485cf1e/ : SUCCESS in 1h 02m 53s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/15/613815/11/check/kolla-ansible-ubuntu-binary/538f8a0/ : TIMED_OUT in 2h 01m 15s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"5c58152c0d12130dc791640519bbca4ababa1e97","author":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"date":"2019-10-27 20:01:59.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"cf017cd33db86cae72779b17f2741ee0297704ff","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-27 20:39:26.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/15/613815/12/openstack-check/tripleo-build-containers-rhel-8-master/d918409/ : FAILURE in 36m 34s","accounts_in_message":[],"_revision_number":12},{"id":"9b082b88bc363c70b42f226b58ef152de80991fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 22:17:21.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e4ccd611351f41b29e4a9272dd15c74a : FAILURE in 44m 55s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9900a6ba0df94ad9a894f0bbc4fa75d3 : FAILURE in 59m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/18ca5fa1f0994194bb9f1e4d3b568197 : SUCCESS in 1h 24m 56s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b54a64bc88ed48d4a32c1c4cda6f8cd6 : SUCCESS in 1h 26m 52s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e665d5fe2ddd4aee8341e0160970d47d : SUCCESS in 1h 29m 04s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8a7b4820ba52488491f8c90827089d1e : SUCCESS in 42m 09s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/6b5bfdc9b8504c0aa6cb5f4cf9d7167d : SUCCESS in 48m 20s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/d94afca8882b42728efeabc2716e3f88 : SUCCESS in 1h 54m 05s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/c4839b8415ee485e809202fdb996d005 : SUCCESS in 2h 12m 55s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/0b1022739b974c9c82617ba6e3e14820 : FAILURE in 34m 06s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/6283c58de36d44f989aed4e5dfa060f1 : SUCCESS in 2h 00m 31s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/3cc89860e9ad4cf7bf22948fe3936415 : SUCCESS in 1h 25m 07s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2a09fef73ea941948d87def3621a78f0 : SUCCESS in 22m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/36d75976743a421aa03c177e25dfc3da : SUCCESS in 19m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b46c2fc846c240e88226ad5382d101ff : FAILURE in 17m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f26c4133c6fb42b098ba53960530e284 : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb148f1f11194a16a81a7c9db3677b72 : SUCCESS in 6m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/11e70e211d7943448f1ef10ecb715f64 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/788b85728f584a46b1aac74c36cc7d6f : SUCCESS in 2m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/232ab5adfc374ed08376b9b85beb2a5b : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d7e5d4ab6c746ea827418b95175aaae : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20ad2c7d24694c8ab84de802ce12c2c0 : SUCCESS in 4m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3ebe71d8e0ef4281b1c0a9b8e8ecd13b : SUCCESS in 31m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ce173776bb6f4f90999ff6694800f7f7 : SUCCESS in 43m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/76f017df959e4a249031c7b314d964d1 : SUCCESS in 57m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/222e8611380b4bddb9fa0997443f609c : SUCCESS in 51m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e01bf8426cde4362bfb22a2c3f75f087 : SUCCESS in 1h 03m 28s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/546e3e3ff6f447c3905e28ff46d8ad7a : SUCCESS in 1h 27m 54s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/dbada881bab04dd38bd53cb9128429c6 : SUCCESS in 1h 28m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/8a47097fb15b4e91983a4b8fdef9b286 : FAILURE in 1h 34m 20s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/d8e83594b12e498e8682fbad75c6837a : SUCCESS in 1h 07m 11s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/cd0f302f373447b0899c1b1cf20329bb : SUCCESS in 56m 10s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f0dbc920aa5008961e243e472756c52dc78bc6c8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-28 07:32:52.000000000","message":"Patch Set 12: Review-Priority-1\n\nIt needs a deprecation cycle. To make it pass, the centos cinder-lvm scenario has to be configured to use lio and tgt added to unbuildable. Also, not sure whether tripleo does not try building tgt for centos.","accounts_in_message":[],"_revision_number":12},{"id":"184663019d192497b136cca0284933880f144162","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-31 11:36:29.000000000","message":"Patch Set 12:\n\n\u003e It needs a deprecation cycle. To make it pass, the centos\n \u003e cinder-lvm scenario has to be configured to use lio and tgt added\n \u003e to unbuildable. Also, not sure whether tripleo does not try\n \u003e building tgt for centos.\n\nI started this a while ago: https://review.opendev.org/631479","accounts_in_message":[],"_revision_number":12},{"id":"c2698aa06aa91f5ea608b69c6b6d7b9bb58c78c5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-14 11:36:48.000000000","message":"Patch Set 12: -Review-Priority","accounts_in_message":[],"_revision_number":12},{"id":"1433006f1108ec7e5bb132009033831c97bce89f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-15 07:34:37.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"292ac130ee47aba1eecf0712f2c6f428a02c7300","author":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"date":"2019-12-23 20:23:31.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":12},{"id":"37c0f653a1f4b25ed763d83c4f4a1b9462ec563c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-06 11:26:13.000000000","message":"Restored\n\nit\u0027s forced on us to be removed so let\u0027s restore Maciej\u0027s proposal","accounts_in_message":[],"_revision_number":12},{"id":"32a641d6e8f771d50ea53c874b234e6aaffc4592","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 11:26:26.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":12},{"id":"30d27f778293321cd44e300a55a01ccf24d87d3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 11:26:27.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":12},{"id":"b79035ba9911f3ef17cf205e059e88260d1dff57","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-06 11:26:28.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":12},{"id":"1d2e8560cc651d1485aba2a2c8834b84d8a49ee3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-06 11:27:22.000000000","message":"Patch Set 12:\n\nwe are aiming for full removal in c8 for ussuri","accounts_in_message":[],"_revision_number":12},{"id":"3aad1f6ecdf2676c5d7eebaa3b06b939565d9ceb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-06 16:07:07.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"4d70da86fc8ada80e56217e0c514687eab6a92d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 16:07:21.000000000","message":"Patch Set 13: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":13},{"id":"01e21939bbaaa364cb048d786e30f69e9654bfb9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 16:07:22.000000000","message":"Patch Set 13:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":13},{"id":"805a632c143bea9e26d8f034ebc07e44bc02ec6f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-06 16:07:25.000000000","message":"Patch Set 13:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":13},{"id":"2cdcc169d0aedc8d1a90c2506fb141a010a40037","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-06 16:27:39.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"e1a446e7541823a780983ff1090068ffb5afc60b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-06 16:36:33.000000000","message":"Patch Set 14:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"64921de82664cf409ebb87bb7a92ba7104b1f0b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 16:36:37.000000000","message":"Patch Set 14: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"dfeb6156605406e2cbf7b625bcfde7d3dec548e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 16:36:37.000000000","message":"Patch Set 14:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"11a3af3bd3ba75d8a23df5beeb7222457b5e550a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-06 16:36:57.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"b10962a2c531c4b859098efa8ac178ac59efbbfc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-06 17:08:47.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/8084aa7564e347a8ba95acfe46ee4e79 : SUCCESS in 29m 18s","accounts_in_message":[],"_revision_number":14},{"id":"0f432601bd230f5739f9e9dcd10464019593a4e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 17:21:58.000000000","message":"Patch Set 14:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"3945805ecc36c40111990653c3d1c7f39835106a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 17:21:59.000000000","message":"Patch Set 14:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"5e055987c252ad86120b21b82e569906e61aba51","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-09 16:27:45.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"8c67f8f3620538172ff4e8a899fa83659808452c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-09 16:55:15.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"5c64e450f06bc7d925c13e3b9d401ec0ea6d9da7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-09 17:29:38.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/11912b8a455240548c8d69c0aa0b47ea : SUCCESS in 31m 36s","accounts_in_message":[],"_revision_number":15},{"id":"b4ca59da35b1169591027ae22e81c369028bd991","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 19:52:51.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9acb330adca1408f9233a2ece589c952 : SUCCESS in 32m 21s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/881f2e85ac7d4164a5c503fd0bf8111e : SUCCESS in 43m 31s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/02e175739054460899493ad69df710ef : SUCCESS in 6m 40s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/fa695f5aea7e4dfaa38e3eb74b2860b0 : SUCCESS in 6m 18s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a2a7a10e650440afb4c4a69a6278b0ac : SUCCESS in 1h 29m 01s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5c0404078b0a40ac99faf714c09caf04 : SUCCESS in 1h 23m 44s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/5f911c74c88d4e52aafb04e71ba0a2b4 : SUCCESS in 1h 41m 51s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/bced7a7850ac4a51ad784fd53d431089 : SUCCESS in 1h 07m 39s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/b74ed23a2331499fbabb85ad22ffad58 : SUCCESS in 52m 33s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/de4e0f509d534f3eab39d548084b0d93 : SUCCESS in 2h 30m 12s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/e912e734adbf457dbb098e06955f6fed : TIMED_OUT in 2h 33m 51s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/f09dd956fb75499f869d6c6af80c8c2e : SUCCESS in 1h 47m 38s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/d8bf734244be486187358cf8fb56cf08 : SUCCESS in 2h 00m 46s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/0bbaf160610f4251a68bb0fd7c51f8af : SUCCESS in 1h 34m 45s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9ade0cc3b59043b384840d6e79fb137c : SUCCESS in 43m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/560c479eb6e64456a4ee3debb005b04f : SUCCESS in 32m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b5a3f5d947084ca790717c6cbe6653ad : SUCCESS in 1h 05m 53s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d466721f181b441b8706913fdfb42ea9 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2cb2ff7f578473394849dd147f70bc5 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a20f27461b94e8b85409d4fdb5a58b0 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d09335f6f714ae8882209082d980c2a : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2fb9549f5aa4f4998ba13116e3c3453 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5306ce52762f4a1898b02a309febe838 : SUCCESS in 5m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/71eec7bbcf86495ea1d9ed6e4b7df0fc : SUCCESS in 32m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8e75ee3ba9a04394919e76f9618e5fa3 : SUCCESS in 30m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e623ff7673e14babb77600569ceb07fb : SUCCESS in 53m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/47023261632341c4a4140e271ae44b52 : SUCCESS in 55m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7ccc42bdcc1f499db0a928a13e0c0fb9 : SUCCESS in 1h 11m 31s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/bd7304f0380143ffaed01bff40696381 : SUCCESS in 1h 23m 13s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/ec89072e08cd41c4a440e6356ae8a1cf : SUCCESS in 1h 54m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/0deb619ffe83464f994f80a6adf8f24c : TIMED_OUT in 2h 34m 14s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/ff01f22d900c4c53a1e1e3acd9a21dc3 : SUCCESS in 1h 45m 22s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/5824d27dc33447d18a8745ce53c3d1bc : SUCCESS in 58m 30s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"1916957aadabf3279de6b1098e9fc2a06cad415d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 20:09:33.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dce5d98591a24c6fb9b7b8bea918857e : TIMED_OUT in 3h 04m 27s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"78c6204d8345a895746a1ed0f01b0b2070a9c954","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-10 05:01:48.000000000","message":"Patch Set 15: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"fcd7354ae4764f77d4c3471b8340c99021fa8d18","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-10 09:21:58.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"a3ebc231155f3d3a84f2f9d7bf221e9b7df73c69","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-10 10:32:01.000000000","message":"Patch Set 15: Review-Priority+1","accounts_in_message":[],"_revision_number":15},{"id":"24d0a962040059a76d1ffc70dc89cfab8b4e8a7b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-10 10:52:42.000000000","message":"Patch Set 15: Code-Review+2\n\ncould use some comments around Ussuri/C7 but meh","accounts_in_message":[],"_revision_number":15},{"id":"d693829ac3fa85ab9ca3047a7acaefd305aaf45d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-10 11:17:49.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"8b96e4bb1713158f7a48a19d42a3fc2a54c6f9ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 11:18:05.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"98a7b8fa5aab0599e4693890ce5a411ac8b6fc3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 12:58:51.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f5539cc5e0c64dfb93c34d75a47bca2d : SUCCESS in 37m 16s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d9f052f9e63d4f25970885c3a775978e : SUCCESS in 5m 24s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ad2325e65b80481d96a18e9768e83330 : SUCCESS in 1h 09m 38s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/26a39a31c663443093bb04b83f80129f : SUCCESS in 1h 21m 12s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/25912ea2b23c47e3983aea6cf0c38680 : SUCCESS in 50m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a31627256bc8455596d353b94c860f8b : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a004a6a5e69d4c74b542dfcebb76848b : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b68b9efbd78947baba82e5b6e20a5839 : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cfe6aa182ef145caa097b622d027a6a4 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b568553f120447578f01a4aeac39c70f : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0220dbfb888649eb957b039332ce8965 : SUCCESS in 5m 21s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1cf2694a9d904a5cafb67266bc703274 : SUCCESS in 41m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/64a5194608be4c66bf07dbda07f6cc2d : SUCCESS in 1h 18m 15s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8b809d6a33cf47dc8133c42a8c519c44 : SUCCESS in 1h 38m 39s","accounts_in_message":[],"_revision_number":15},{"id":"cff7e9eb8b670f81e3e61394c3add11605a02adb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 12:58:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15},{"id":"27169295bff886b91a45aa412961e0223b5f2f53","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 13:00:51.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9511e730bea44ef9d965adb4aa3b010 : SUCCESS in 1m 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/857a33a444d84d24ab92b10cbe0542e1 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"946749312f31c36d9dac8ff3a545894e9043190d","revisions":{"15568bb10181620dd7117f4202c7285ea1f8afa5":{"kind":"REWORK","_number":1,"created":"2018-10-28 22:19:55.000000000","uploader":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"ref":"refs/changes/15/613815/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/613815/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/613815/1"}}},"commit":{"parents":[{"commit":"f613ff8a548accee65bd86209e4683539b66ffdb","subject":"Merge \"Improve pep8 checking along with hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f613ff8a548accee65bd86209e4683539b66ffdb"}]}],"author":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2018-10-28 22:14:49.000000000","tz":60},"committer":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2018-10-28 22:14:49.000000000","tz":60},"subject":"Remove tgtd and scsi-target-utils support in CentOS/RHEL","message":"Remove tgtd and scsi-target-utils support in CentOS/RHEL\n\nThe package \u0027scsi-target-utils\u0027 is no longer supported in CentOS\nand RHEL. Furthermore Cinder have switched to use lioadm so there\nis no need to keep tgtd support.\n\nChange-Id: I56f230d66f75dd0546325676278f91579f08c822\nSigned-off-by: Maciej Kucia \u003cmaciej@kucia.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15568bb10181620dd7117f4202c7285ea1f8afa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15568bb10181620dd7117f4202c7285ea1f8afa5"}]},"branch":"refs/heads/master"},"608b5c54008b800157723ab347988d2e839cef85":{"kind":"REWORK","_number":2,"created":"2018-11-18 20:37:43.000000000","uploader":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"ref":"refs/changes/15/613815/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/613815/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/613815/2"}}},"commit":{"parents":[{"commit":"1affeaa2e223705433f1d1a0e23c6af98f9e2bc3","subject":"Merge \"Replacing the HTTP protocal with HTTPS .\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1affeaa2e223705433f1d1a0e23c6af98f9e2bc3"}]}],"author":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2018-10-28 22:14:49.000000000","tz":60},"committer":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2018-11-18 20:37:28.000000000","tz":60},"subject":"Remove tgtd and scsi-target-utils support in CentOS/RHEL","message":"Remove tgtd and scsi-target-utils support in CentOS/RHEL\n\nIn RHEL 7 and beyond LIO kernel subsystem can be used\ninstead of tgtd daemon.\n\nThe package \u0027scsi-target-utils\u0027 is no longer directly\nsupported in CentOS and RHEL.\n\nWhile package is still available in EPEL, there is no\nneed to use tgt container in rpm world.\n\nChange-Id: I56f230d66f75dd0546325676278f91579f08c822\nSigned-off-by: Maciej Kucia \u003cmaciej@kucia.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/608b5c54008b800157723ab347988d2e839cef85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/608b5c54008b800157723ab347988d2e839cef85"}]},"branch":"refs/heads/master"},"a3d70efb1917b9cf240a78b38b95d1fed10df7eb":{"kind":"REWORK","_number":3,"created":"2018-11-18 20:51:40.000000000","uploader":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"ref":"refs/changes/15/613815/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/613815/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/613815/3"}}},"commit":{"parents":[{"commit":"1affeaa2e223705433f1d1a0e23c6af98f9e2bc3","subject":"Merge \"Replacing the HTTP protocal with HTTPS .\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1affeaa2e223705433f1d1a0e23c6af98f9e2bc3"}]}],"author":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2018-10-28 22:14:49.000000000","tz":60},"committer":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2018-11-18 20:51:32.000000000","tz":60},"subject":"Remove tgtd and scsi-target-utils support in CentOS/RHEL","message":"Remove tgtd and scsi-target-utils support in CentOS/RHEL\n\nIn RHEL 7 and beyond LIO kernel subsystem can be used\ninstead of tgtd daemon.\n\nThe package \u0027scsi-target-utils\u0027 is no longer directly\nsupported in CentOS and RHEL.\n\nWhile package is still available in EPEL, there is no\nneed to use tgt container in rpm world.\n\nChange-Id: I56f230d66f75dd0546325676278f91579f08c822\nSigned-off-by: Maciej Kucia \u003cmaciej@kucia.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3d70efb1917b9cf240a78b38b95d1fed10df7eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3d70efb1917b9cf240a78b38b95d1fed10df7eb"}]},"branch":"refs/heads/master"},"d748b1d92eb18362c3af0ea74376c1992b2f4cd8":{"kind":"REWORK","_number":4,"created":"2019-01-15 08:52:17.000000000","uploader":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"ref":"refs/changes/15/613815/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/613815/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/613815/4"}}},"commit":{"parents":[{"commit":"c2397e62d86773c243753cab32183b43831e7ac0","subject":"Add python2-sqlalchemy-collectd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c2397e62d86773c243753cab32183b43831e7ac0"}]}],"author":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2018-10-28 22:14:49.000000000","tz":60},"committer":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2019-01-15 08:51:46.000000000","tz":60},"subject":"Remove tgtd and scsi-target-utils support in CentOS/RHEL","message":"Remove tgtd and scsi-target-utils support in CentOS/RHEL\n\nIn RHEL 7 and beyond LIO kernel subsystem can be used\ninstead of tgtd daemon.\n\nThe package \u0027scsi-target-utils\u0027 is no longer directly\nsupported in CentOS and RHEL.\n\nWhile package is still available in EPEL, there is no\nneed to use tgt container in rpm world.\n\nChange-Id: I56f230d66f75dd0546325676278f91579f08c822\nSigned-off-by: Maciej Kucia \u003cmaciej@kucia.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d748b1d92eb18362c3af0ea74376c1992b2f4cd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d748b1d92eb18362c3af0ea74376c1992b2f4cd8"}]},"branch":"refs/heads/master"},"ebc60859260121f9b3c74fd257988ea6a7f09765":{"kind":"REWORK","_number":5,"created":"2019-01-15 08:56:31.000000000","uploader":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"ref":"refs/changes/15/613815/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/613815/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/613815/5"}}},"commit":{"parents":[{"commit":"c2397e62d86773c243753cab32183b43831e7ac0","subject":"Add python2-sqlalchemy-collectd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c2397e62d86773c243753cab32183b43831e7ac0"}]}],"author":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2018-10-28 22:14:49.000000000","tz":60},"committer":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2019-01-15 08:56:24.000000000","tz":60},"subject":"Remove tgtd and scsi-target-utils support in CentOS/RHEL","message":"Remove tgtd and scsi-target-utils support in CentOS/RHEL\n\nIn RHEL 7 and beyond LIO kernel subsystem can be used\ninstead of tgtd daemon.\n\nThe package \u0027scsi-target-utils\u0027 is no longer directly\nsupported in CentOS and RHEL.\n\nWhile package is still available in EPEL, there is no\nneed to use tgt container in rpm world.\n\nChange-Id: I56f230d66f75dd0546325676278f91579f08c822\nSigned-off-by: Maciej Kucia \u003cmaciej@kucia.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ebc60859260121f9b3c74fd257988ea6a7f09765"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ebc60859260121f9b3c74fd257988ea6a7f09765"}]},"branch":"refs/heads/master"},"5861d1360e59db9d59cb8599f1c1bd7e65cf8883":{"kind":"REWORK","_number":6,"created":"2019-01-17 20:48:15.000000000","uploader":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"ref":"refs/changes/15/613815/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/613815/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/613815/6"}}},"commit":{"parents":[{"commit":"ffcda49e343aed666ee7839b5c7af04520577af0","subject":"Merge \"Add distro_python3 option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ffcda49e343aed666ee7839b5c7af04520577af0"}]}],"author":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2018-10-28 22:14:49.000000000","tz":60},"committer":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2019-01-17 20:48:07.000000000","tz":60},"subject":"Remove tgtd and scsi-target-utils support in CentOS/RHEL","message":"Remove tgtd and scsi-target-utils support in CentOS/RHEL\n\nIn RHEL 7 and beyond LIO kernel subsystem can be used\ninstead of tgtd daemon.\n\nThe package \u0027scsi-target-utils\u0027 is no longer directly\nsupported in CentOS and RHEL.\n\nWhile package is still available in EPEL, there is no\nneed to use tgt container in rpm world.\n\nChange-Id: I56f230d66f75dd0546325676278f91579f08c822\nSigned-off-by: Maciej Kucia \u003cmaciej@kucia.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5861d1360e59db9d59cb8599f1c1bd7e65cf8883"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5861d1360e59db9d59cb8599f1c1bd7e65cf8883"}]},"branch":"refs/heads/master"},"e939303893d55702c9886ecbaa5de6ac1820d873":{"kind":"REWORK","_number":7,"created":"2019-01-18 14:30:33.000000000","uploader":{"_account_id":25588,"name":"Maciej Kucia","email":"maciej@kucia.net","username":"m.kucia"},"ref":"refs/changes/15/613815/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/613815/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/613815/7"}}},"commit":{"parents":[{"commit":"ffcda49e343aed666ee7839b5c7af04520577af0","subject":"Merge \"Add distro_python3 option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ffcda49e343aed666ee7839b5c7af04520577af0"}]}],"author":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2018-10-28 22:14:49.000000000","tz":60},"committer":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2019-01-18 14:30:20.000000000","tz":0},"subject":"Remove tgtd and scsi-target-utils support in CentOS/RHEL","message":"Remove tgtd and scsi-target-utils support in CentOS/RHEL\n\nIn RHEL 7 and beyond LIO kernel subsystem can be used\ninstead of tgtd daemon.\n\nThe package \u0027scsi-target-utils\u0027 is no longer directly\nsupported in CentOS and RHEL.\n\nWhile package is still available in EPEL, there is no\nneed to use tgt container in rpm world.\n\nChange-Id: I56f230d66f75dd0546325676278f91579f08c822\nSigned-off-by: Maciej Kucia \u003cmaciej@kucia.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e939303893d55702c9886ecbaa5de6ac1820d873"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e939303893d55702c9886ecbaa5de6ac1820d873"}]},"branch":"refs/heads/master"},"19401cbcd1f35c667251d93897a54d2a30c453f9":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2019-01-18 16:44:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/15/613815/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/613815/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/613815/8"}}},"commit":{"parents":[{"commit":"ffcda49e343aed666ee7839b5c7af04520577af0","subject":"Merge \"Add distro_python3 option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ffcda49e343aed666ee7839b5c7af04520577af0"}]}],"author":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2018-10-28 22:14:49.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-18 16:44:38.000000000","tz":0},"subject":"Remove tgtd and scsi-target-utils support in CentOS/RHEL","message":"Remove tgtd and scsi-target-utils support in CentOS/RHEL\n\nIn RHEL 7 and beyond LIO kernel subsystem can be used\ninstead of tgtd daemon.\n\nThe package \u0027scsi-target-utils\u0027 is no longer directly\nsupported in CentOS and RHEL.\n\nWhile package is still available in EPEL, there is no\nneed to use tgt container in rpm world.\n\nChange-Id: I56f230d66f75dd0546325676278f91579f08c822\nDepends-On: https://review.openstack.org/631479\nSigned-off-by: Maciej Kucia \u003cmaciej@kucia.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19401cbcd1f35c667251d93897a54d2a30c453f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19401cbcd1f35c667251d93897a54d2a30c453f9"}]},"branch":"refs/heads/master"},"050f1e820c2e4407e29ff759f9e38ca392baf145":{"kind":"REWORK","_number":9,"created":"2019-02-09 21:04:49.000000000","uploader":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"ref":"refs/changes/15/613815/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/613815/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/613815/9"}}},"commit":{"parents":[{"commit":"5922c22d8af6c34eb5c34250c3fa90dc3f853688","subject":"Merge \"remove pep8 from lower-constraints.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5922c22d8af6c34eb5c34250c3fa90dc3f853688"}]}],"author":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2018-10-28 22:14:49.000000000","tz":60},"committer":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2019-02-09 21:04:00.000000000","tz":60},"subject":"Remove tgtd and scsi-target-utils support in CentOS/RHEL","message":"Remove tgtd and scsi-target-utils support in CentOS/RHEL\n\nIn RHEL 7 and beyond LIO kernel subsystem can be used\ninstead of tgtd daemon.\n\nThe package \u0027scsi-target-utils\u0027 is no longer directly\nsupported in CentOS and RHEL.\n\nWhile package is still available in EPEL, there is no\nneed to use tgt container in rpm world.\n\nChange-Id: I56f230d66f75dd0546325676278f91579f08c822\nSigned-off-by: Maciej Kucia \u003cmaciej@kucia.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/050f1e820c2e4407e29ff759f9e38ca392baf145"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/050f1e820c2e4407e29ff759f9e38ca392baf145"}]},"branch":"refs/heads/master"},"6029af214e077da0105a56e76abcdd5834a38844":{"kind":"REWORK","_number":10,"created":"2019-02-18 10:59:11.000000000","uploader":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"ref":"refs/changes/15/613815/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/613815/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/613815/10"}}},"commit":{"parents":[{"commit":"5922c22d8af6c34eb5c34250c3fa90dc3f853688","subject":"Merge \"remove pep8 from lower-constraints.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5922c22d8af6c34eb5c34250c3fa90dc3f853688"}]}],"author":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2018-10-28 22:14:49.000000000","tz":60},"committer":{"name":"Martin André","email":"m.andre@redhat.com","date":"2019-02-18 10:58:27.000000000","tz":0},"subject":"Remove tgtd and scsi-target-utils support in CentOS/RHEL","message":"Remove tgtd and scsi-target-utils support in CentOS/RHEL\n\nIn RHEL 7 and beyond LIO kernel subsystem can be used\ninstead of tgtd daemon.\n\nThe package \u0027scsi-target-utils\u0027 is no longer directly\nsupported in CentOS and RHEL.\n\nWhile package is still available in EPEL, there is no\nneed to use tgt container in rpm world.\n\nChange-Id: I56f230d66f75dd0546325676278f91579f08c822\nSigned-off-by: Maciej Kucia \u003cmaciej@kucia.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6029af214e077da0105a56e76abcdd5834a38844"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6029af214e077da0105a56e76abcdd5834a38844"}]},"branch":"refs/heads/master"},"01eca5ae7cd7313aa0dbe76552552b0678148bb6":{"kind":"REWORK","_number":11,"created":"2019-02-21 09:02:20.000000000","uploader":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"ref":"refs/changes/15/613815/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/613815/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/613815/11"}}},"commit":{"parents":[{"commit":"5922c22d8af6c34eb5c34250c3fa90dc3f853688","subject":"Merge \"remove pep8 from lower-constraints.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5922c22d8af6c34eb5c34250c3fa90dc3f853688"}]}],"author":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2018-10-28 22:14:49.000000000","tz":60},"committer":{"name":"Martin André","email":"m.andre@redhat.com","date":"2019-02-21 09:02:04.000000000","tz":60},"subject":"Remove tgtd and scsi-target-utils support in CentOS/RHEL","message":"Remove tgtd and scsi-target-utils support in CentOS/RHEL\n\nIn RHEL 7 and beyond LIO kernel subsystem can be used\ninstead of tgtd daemon.\n\nThe package \u0027scsi-target-utils\u0027 is no longer directly\nsupported in CentOS and RHEL.\n\nWhile package is still available in EPEL, there is no\nneed to use tgt container in rpm world.\n\nChange-Id: I56f230d66f75dd0546325676278f91579f08c822\nSigned-off-by: Maciej Kucia \u003cmaciej@kucia.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01eca5ae7cd7313aa0dbe76552552b0678148bb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01eca5ae7cd7313aa0dbe76552552b0678148bb6"}]},"branch":"refs/heads/master"},"2c9693aee3aaae4677cfd999cdc3552bf9792831":{"kind":"REWORK","_number":12,"created":"2019-10-27 20:01:59.000000000","uploader":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"ref":"refs/changes/15/613815/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/613815/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/613815/12"}}},"commit":{"parents":[{"commit":"7b309f830034319e3f7db8efb424448e43ccae21","subject":"Merge \"Move to RDO Train packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b309f830034319e3f7db8efb424448e43ccae21"}]}],"author":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2018-10-28 22:14:49.000000000","tz":60},"committer":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2019-10-27 19:59:45.000000000","tz":60},"subject":"Remove tgtd and scsi-target-utils support in CentOS/RHEL","message":"Remove tgtd and scsi-target-utils support in CentOS/RHEL\n\nIn RHEL 7 and beyond LIO kernel subsystem can be used\ninstead of tgtd daemon.\n\nThe package \u0027scsi-target-utils\u0027 is no longer directly\nsupported in CentOS and RHEL.\n\nWhile package is still available in EPEL, there is no\nneed to use tgt container in rpm world.\n\nChange-Id: I56f230d66f75dd0546325676278f91579f08c822\nSigned-off-by: Maciej Kucia \u003cmaciej@kucia.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c9693aee3aaae4677cfd999cdc3552bf9792831"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c9693aee3aaae4677cfd999cdc3552bf9792831"}]},"branch":"refs/heads/master"},"85e649b418920d5c440e469f29423a1927dbb130":{"kind":"REWORK","_number":13,"created":"2020-01-06 16:07:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/15/613815/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/613815/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/613815/13"}}},"commit":{"parents":[{"commit":"7c116653e311d2a9ef769bb9227acd0ce9d3ad7a","subject":"Merge \"Add \u0027--quiet\u0027 switch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c116653e311d2a9ef769bb9227acd0ce9d3ad7a"}]}],"author":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2018-10-28 22:14:49.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-06 16:06:12.000000000","tz":0},"subject":"Remove tgtd and scsi-target-utils support in CentOS/RHEL 8","message":"Remove tgtd and scsi-target-utils support in CentOS/RHEL 8\n\nIn CentOS/RHEL 8 there is no scsi-target-utils package, nor is it\navailable in EPEL. In RHEL 7 and beyond the LIO kernel subsystem can be\nused instead of the tgtd daemon.\n\nThis change removes support for the SCSI target daemon on CentOS/RHEL 8.\nThe \u0027tgtd\u0027 image is no longer available for CentOS/RHEL 8.\n\nChange-Id: I56f230d66f75dd0546325676278f91579f08c822\nSigned-off-by: Maciej Kucia \u003cmaciej@kucia.net\u003e\nCo-Authored-By: Maciej Kucia \u003cmaciej@kucia.net\u003e\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\nRelated: blueprint centos-rhel-8\nDepends-On: https://review.opendev.org/631479/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85e649b418920d5c440e469f29423a1927dbb130"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85e649b418920d5c440e469f29423a1927dbb130"}]},"branch":"refs/heads/master"},"5a894b07763105d517e92207cc291f9b2e34fc2f":{"kind":"REWORK","_number":14,"created":"2020-01-06 16:27:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/15/613815/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/613815/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/613815/14"}}},"commit":{"parents":[{"commit":"7c116653e311d2a9ef769bb9227acd0ce9d3ad7a","subject":"Merge \"Add \u0027--quiet\u0027 switch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c116653e311d2a9ef769bb9227acd0ce9d3ad7a"}]}],"author":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2018-10-28 22:14:49.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-06 16:19:15.000000000","tz":0},"subject":"Remove tgtd and scsi-target-utils support in CentOS/RHEL 8","message":"Remove tgtd and scsi-target-utils support in CentOS/RHEL 8\n\nIn CentOS/RHEL 8 there is no scsi-target-utils package, nor is it\navailable in EPEL. In RHEL 7 and beyond the LIO kernel subsystem can be\nused instead of the tgtd daemon.\n\nThis change removes support for the SCSI target daemon on CentOS/RHEL 8.\nThe \u0027tgtd\u0027 image is no longer available for CentOS/RHEL 8.\n\nChange-Id: I56f230d66f75dd0546325676278f91579f08c822\nSigned-off-by: Maciej Kucia \u003cmaciej@kucia.net\u003e\nCo-Authored-By: Maciej Kucia \u003cmaciej@kucia.net\u003e\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\nRelated: blueprint centos-rhel-8\nDepends-On: https://review.opendev.org/631479/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a894b07763105d517e92207cc291f9b2e34fc2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a894b07763105d517e92207cc291f9b2e34fc2f"}]},"branch":"refs/heads/master"},"946749312f31c36d9dac8ff3a545894e9043190d":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2020-01-09 16:55:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/15/613815/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/613815/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/613815/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/613815/15"}}},"commit":{"parents":[{"commit":"b060623b907348e518bdf1afede0b641be56c415","subject":"CentOS 8: Add zuul build and publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b060623b907348e518bdf1afede0b641be56c415"}]}],"author":{"name":"Maciej Kucia","email":"maciej@kucia.net","date":"2018-10-28 22:14:49.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-09 16:29:42.000000000","tz":0},"subject":"Remove tgtd and scsi-target-utils support in CentOS/RHEL 8","message":"Remove tgtd and scsi-target-utils support in CentOS/RHEL 8\n\nIn CentOS/RHEL 8 there is no scsi-target-utils package, nor is it\navailable in EPEL. In RHEL 7 and beyond the LIO kernel subsystem can be\nused instead of the tgtd daemon.\n\nThis change removes support for the SCSI target daemon on CentOS/RHEL 8.\nThe \u0027tgtd\u0027 image is no longer available for CentOS/RHEL 8.\n\nChange-Id: I56f230d66f75dd0546325676278f91579f08c822\nSigned-off-by: Maciej Kucia \u003cmaciej@kucia.net\u003e\nCo-Authored-By: Maciej Kucia \u003cmaciej@kucia.net\u003e\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\nRelated: blueprint centos-rhel-8\nDepends-On: https://review.opendev.org/631479/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/946749312f31c36d9dac8ff3a545894e9043190d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/946749312f31c36d9dac8ff3a545894e9043190d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
