)]}'
{"id":"openstack%2Fkolla~615469","triplet_id":"openstack%2Fkolla~master~I279d6b43f8e4d4e30b39deb607823ceca272a4b8","project":"openstack/kolla","branch":"master","topic":"bp/masakari-support","hashtags":[],"change_id":"I279d6b43f8e4d4e30b39deb607823ceca272a4b8","subject":"Add Masakari containers","status":"MERGED","created":"2018-11-05 04:24:06.000000000","updated":"2019-07-15 03:06:10.000000000","submitted":"2019-07-15 03:06:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"615469-1563159970370-4e3e88c1","meta_rev_id":"1e49d5739f0d13207665e81e63de1c7700974e7e","_number":615469,"virtual_id_number":615469,"owner":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-07-15 03:06:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-11 23:50:07.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":23479,"name":"Murali Annamneni","email":"murali.annamneni@oracle.com","username":"mannamne"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"recommended":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"all":[{"value":0,"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":1,"date":"2019-07-13 15:55:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-07-13 18:36:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":1,"date":"2019-07-12 00:37:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":2,"date":"2019-07-15 01:58:02.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":1,"date":"2019-07-14 15:27:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":23479,"name":"Murali Annamneni","email":"murali.annamneni@oracle.com","username":"mannamne"},{"value":1,"date":"2019-07-14 23:50:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":2,"date":"2019-07-12 11:32:47.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":0,"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":1,"date":"2019-07-15 01:58:02.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":23479,"name":"Murali Annamneni","email":"murali.annamneni@oracle.com","username":"mannamne"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":23479,"name":"Murali Annamneni","email":"murali.annamneni@oracle.com","username":"mannamne"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":23479,"name":"Murali Annamneni","email":"murali.annamneni@oracle.com","username":"mannamne"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},{"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23479,"name":"Murali Annamneni","email":"murali.annamneni@oracle.com","username":"mannamne"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-11-18 15:18:13.000000000","updated_by":{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},"reviewer":{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},"state":"REVIEWER"},{"updated":"2018-12-13 12:33:20.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2018-12-14 03:34:48.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2018-12-14 14:30:37.000000000","updated_by":{"_account_id":23479,"name":"Murali Annamneni","email":"murali.annamneni@oracle.com","username":"mannamne"},"reviewer":{"_account_id":23479,"name":"Murali Annamneni","email":"murali.annamneni@oracle.com","username":"mannamne"},"state":"REVIEWER"},{"updated":"2019-03-10 15:46:45.000000000","updated_by":{"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},"reviewer":{"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},"state":"REVIEWER"},{"updated":"2019-06-27 07:10:24.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-27 22:06:09.000000000","updated_by":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"reviewer":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"state":"REVIEWER"},{"updated":"2019-06-28 06:36:53.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-07-07 05:17:50.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-07-11 23:50:07.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-12 00:37:54.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-07-12 11:32:47.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-07-13 15:55:42.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-07-13 18:36:22.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-07-14 15:27:44.000000000","updated_by":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"reviewer":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"state":"REVIEWER"},{"updated":"2019-07-14 23:50:29.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2019-07-15 01:58:02.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-07-15 03:06:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cb1679a1baf6f700bd15aa35923bbad7fe22d53b","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-05 04:24:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8327f9fea899809b5e67cf1ceb9577150952d52b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-05 06:00:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/69/615469/1/check/test-release-openstack/cda61d0/ : SUCCESS in 2m 15s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/1/check/kolla-build-centos-binary/bd9c4c5/ : SUCCESS in 57m 55s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/1/check/kolla-build-centos-source/8a09a06/ : SUCCESS in 57m 36s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/1/check/kolla-ansible-centos-source/fb48e1e/ : SUCCESS in 1h 34m 46s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/1/check/kolla-ansible-centos-binary/e8e47a9/ : SUCCESS in 1h 04m 18s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/615469/1/check/kolla-build-oraclelinux-binary/8d71071/ : FAILURE in 57m 53s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/615469/1/check/kolla-build-oraclelinux-source/dd7dd69/ : SUCCESS in 1h 04m 13s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/615469/1/check/kolla-ansible-oraclelinux-source/db016f1/ : SUCCESS in 1h 03m 51s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/615469/1/check/kolla-ansible-oraclelinux-binary/4306ea5/ : SUCCESS in 1h 21m 39s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/1/check/openstack-tox-lower-constraints/8dc9184/ : SUCCESS in 2m 42s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/1/check/openstack-tox-pep8/68b9c72/ : SUCCESS in 2m 39s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/1/check/openstack-tox-py27/d6da2b1/ : SUCCESS in 2m 48s\n- openstack-tox-py35 http://logs.openstack.org/69/615469/1/check/openstack-tox-py35/91320ee/ : SUCCESS in 3m 37s\n- openstack-tox-py36 http://logs.openstack.org/69/615469/1/check/openstack-tox-py36/543e723/ : SUCCESS in 2m 14s\n- openstack-tox-docs http://logs.openstack.org/69/615469/1/check/openstack-tox-docs/1e1b846/html/ : SUCCESS in 2m 36s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/1/check/build-openstack-releasenotes/d3f613c/html/ : SUCCESS in 3m 28s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/1/check/kolla-build-ubuntu-binary/284d8b1/ : SUCCESS in 1h 17m 40s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/1/check/kolla-build-ubuntu-source/49c63eb/ : SUCCESS in 30m 35s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/1/check/kolla-ansible-ubuntu-source/ed21b5b/ : SUCCESS in 54m 13s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/1/check/kolla-ansible-ubuntu-binary/d6e9e5e/ : FAILURE in 1h 04m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4dafb2ecf740d00a1f25db0e445b4eac13add19","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2018-11-05 07:26:13.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nmissing horizon function https://github.com/openstack/kolla/blob/master/docker/horizon/extend_start.sh#L72","accounts_in_message":[],"_revision_number":1},{"id":"1fbc83e7712491fd4b36f59b20e591f163d90b55","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-05 07:27:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fc8678161a7305af735042be775d9a44d5cc1634","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-05 07:29:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"df7d41a6397fad88ffea5d88e137dc7662cdc588","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-05 07:30:01.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4a1375287dbc77de7692c9c6f2afd455a4885ae4","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-05 08:11:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d46de81cd118f620480d3326ee78546c24d6254a","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-05 08:30:48.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f6bfb1ea2f1c5ef9830f67b327811dfba48cb615","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-05 10:33:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/69/615469/5/check/test-release-openstack/fb18bd4/ : SUCCESS in 2m 54s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/5/check/kolla-build-centos-binary/24c555a/ : FAILURE in 1h 17m 22s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/5/check/kolla-build-centos-source/129274e/ : SUCCESS in 55m 39s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/5/check/kolla-ansible-centos-source/61e9469/ : SUCCESS in 1h 45m 47s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/5/check/kolla-ansible-centos-binary/2019c14/ : FAILURE in 16m 57s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/615469/5/check/kolla-build-oraclelinux-binary/36560b8/ : FAILURE in 32m 02s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/615469/5/check/kolla-build-oraclelinux-source/e41bb40/ : SUCCESS in 1h 10m 44s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/615469/5/check/kolla-ansible-oraclelinux-source/e904c43/ : TIMED_OUT in 2h 01m 09s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/615469/5/check/kolla-ansible-oraclelinux-binary/34013c1/ : FAILURE in 1h 07m 07s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/5/check/openstack-tox-lower-constraints/e2c41b5/ : SUCCESS in 3m 59s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/5/check/openstack-tox-pep8/004641e/ : SUCCESS in 4m 16s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/5/check/openstack-tox-py27/3f4da74/ : SUCCESS in 2m 51s\n- openstack-tox-py35 http://logs.openstack.org/69/615469/5/check/openstack-tox-py35/6312660/ : SUCCESS in 3m 10s\n- openstack-tox-py36 http://logs.openstack.org/69/615469/5/check/openstack-tox-py36/c871028/ : SUCCESS in 3m 08s\n- openstack-tox-docs http://logs.openstack.org/69/615469/5/check/openstack-tox-docs/0bd704c/html/ : SUCCESS in 3m 11s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/5/check/build-openstack-releasenotes/0cf2170/html/ : SUCCESS in 3m 55s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/5/check/kolla-build-ubuntu-binary/68951e8/ : SUCCESS in 1h 30m 44s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/5/check/kolla-build-ubuntu-source/b9ffdbc/ : SUCCESS in 27m 41s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/5/check/kolla-ansible-ubuntu-source/7d00b82/ : SUCCESS in 53m 03s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/5/check/kolla-ansible-ubuntu-binary/08f259e/ : FAILURE in 57m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2bab913b51615cd0b2fbd307394a7dd84a8dd242","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-06 08:39:56.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nGenerally looks good, thanks.","accounts_in_message":[],"_revision_number":5},{"id":"95fd6e1c2297686bcf2be54f0a4e72c0f3515cbc","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-07 01:32:59.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4d50bb8398a49b76bf0448ddfffeea83e7a41fc0","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-07 01:33:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"53a6d83ea787a882acb89ddbb4b39c8ca1136f28","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-07 02:10:22.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1c758060effda465a74baf6bba2dd12237b7fd93","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-07 03:40:35.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/69/615469/7/check/test-release-openstack/dad7668/ : SUCCESS in 2m 19s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/7/check/kolla-build-centos-binary/3e11c63/ : FAILURE in 22m 38s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/7/check/kolla-build-centos-source/769f21f/ : SUCCESS in 56m 52s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/7/check/kolla-ansible-centos-source/fcb1528/ : SUCCESS in 1h 23m 36s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/7/check/kolla-ansible-centos-binary/d3b4831/ : FAILURE in 17m 56s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/615469/7/check/kolla-build-oraclelinux-binary/a5fe726/ : FAILURE in 32m 54s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/615469/7/check/kolla-build-oraclelinux-source/a41536d/ : SUCCESS in 1h 04m 36s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/615469/7/check/kolla-ansible-oraclelinux-source/a493578/ : SUCCESS in 1h 24m 45s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/615469/7/check/kolla-ansible-oraclelinux-binary/0ddb385/ : FAILURE in 1h 01m 32s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/7/check/openstack-tox-lower-constraints/cb63a51/ : SUCCESS in 3m 35s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/7/check/openstack-tox-pep8/16b9258/ : SUCCESS in 3m 35s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/7/check/openstack-tox-py27/b8f5467/ : SUCCESS in 3m 07s\n- openstack-tox-py35 http://logs.openstack.org/69/615469/7/check/openstack-tox-py35/fb73d95/ : SUCCESS in 2m 43s\n- openstack-tox-py36 http://logs.openstack.org/69/615469/7/check/openstack-tox-py36/cc6f101/ : SUCCESS in 3m 29s\n- openstack-tox-docs http://logs.openstack.org/69/615469/7/check/openstack-tox-docs/bf5aa86/html/ : SUCCESS in 3m 31s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/7/check/build-openstack-releasenotes/ac1bc8d/html/ : SUCCESS in 3m 52s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/7/check/kolla-build-ubuntu-binary/7ebfc4c/ : SUCCESS in 51m 34s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/7/check/kolla-build-ubuntu-source/77fa083/ : SUCCESS in 28m 07s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/7/check/kolla-ansible-ubuntu-source/71b0d29/ : SUCCESS in 54m 33s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/7/check/kolla-ansible-ubuntu-binary/0efd79c/ : FAILURE in 1h 25m 00s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d4927f3185a53adc91a3d85ab34682ff33c632fb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-07 10:49:31.000000000","message":"Patch Set 7: Code-Review-1\n\nCentos binary jobs failing due to missing client package, e.g. http://logs.openstack.org/69/615469/7/check/kolla-build-centos-binary/3e11c63/logs/build/000_FAILED_openstack-base.txt.gz","accounts_in_message":[],"_revision_number":7},{"id":"ca6aa0fc004268c6cace128c9dbd288c478ebc74","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-07 11:10:55.000000000","message":"Patch Set 7:\n\n\u003e Centos binary jobs failing due to missing client package, e.g.\n \u003e http://logs.openstack.org/69/615469/7/check/kolla-build-centos-binary/3e11c63/logs/build/000_FAILED_openstack-base.txt.gz\n\nDone, I\u0027ve removed it.","accounts_in_message":[],"_revision_number":7},{"id":"8d3c7ae22a07166df0d34baf307ee698904c70cd","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-07 11:11:20.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9a794701dedca2f21d3ffa9847ff27daf0d21e6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-07 13:50:55.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/69/615469/8/check/test-release-openstack/06ba47e/ : SUCCESS in 1m 54s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/8/check/kolla-build-centos-binary/8ef696d/ : FAILURE in 20m 22s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/8/check/kolla-build-centos-source/a406d8e/ : FAILURE in 2h 37m 54s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/8/check/kolla-ansible-centos-source/3e5e41d/ : SUCCESS in 1h 14m 07s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/8/check/kolla-ansible-centos-binary/cdc4605/ : SUCCESS in 1h 12m 59s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/615469/8/check/kolla-build-oraclelinux-binary/a43a4dd/ : SUCCESS in 52m 29s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/615469/8/check/kolla-build-oraclelinux-source/77475bf/ : SUCCESS in 56m 53s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/615469/8/check/kolla-ansible-oraclelinux-source/307a4c5/ : SUCCESS in 1h 21m 00s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/615469/8/check/kolla-ansible-oraclelinux-binary/8620460/ : SUCCESS in 1h 13m 43s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/8/check/openstack-tox-lower-constraints/10b537e/ : SUCCESS in 3m 40s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/8/check/openstack-tox-pep8/ec9f91d/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/8/check/openstack-tox-py27/6ff428b/ : SUCCESS in 4m 01s\n- openstack-tox-py35 http://logs.openstack.org/69/615469/8/check/openstack-tox-py35/5bb3aa7/ : SUCCESS in 3m 00s\n- openstack-tox-py36 http://logs.openstack.org/69/615469/8/check/openstack-tox-py36/ae5f4fd/ : SUCCESS in 3m 16s\n- openstack-tox-docs http://logs.openstack.org/69/615469/8/check/openstack-tox-docs/e4d5fba/html/ : SUCCESS in 2m 15s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/8/check/build-openstack-releasenotes/57d200a/html/ : SUCCESS in 3m 16s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/8/check/kolla-build-ubuntu-binary/5f4b9ab/ : SUCCESS in 1h 25m 58s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/8/check/kolla-build-ubuntu-source/52602d2/ : SUCCESS in 33m 40s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/8/check/kolla-ansible-ubuntu-source/4b4c2a0/ : SUCCESS in 1h 47m 43s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/8/check/kolla-ansible-ubuntu-binary/015783a/ : FAILURE in 1h 29m 11s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d8e95e9a2da4152bffe2f07e572034aa6ffd7227","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-09 03:52:14.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"cc6c4b2256ae3d04346968eb37052e89d72688fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-09 06:07:52.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/69/615469/8/check/test-release-openstack/530ad63/ : SUCCESS in 2m 47s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/8/check/kolla-build-centos-binary/d038395/ : SUCCESS in 1h 26m 52s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/8/check/kolla-build-centos-source/f063c84/ : FAILURE in 2h 14m 06s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/8/check/kolla-ansible-centos-source/b9af680/ : SUCCESS in 1h 20m 53s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/8/check/kolla-ansible-centos-binary/8dd1377/ : TIMED_OUT in 2h 01m 05s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/615469/8/check/kolla-build-oraclelinux-binary/a43abe5/ : FAILURE in 2h 13m 35s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/615469/8/check/kolla-build-oraclelinux-source/0158207/ : SUCCESS in 1h 11m 27s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/615469/8/check/kolla-ansible-oraclelinux-source/a00320c/ : SUCCESS in 1h 12m 42s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/615469/8/check/kolla-ansible-oraclelinux-binary/bb6e8e7/ : SUCCESS in 1h 19m 04s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/69/615469/8/check/openstack-tox-py36/1dd678d/ : SUCCESS in 3m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/8/check/openstack-tox-lower-constraints/f742aec/ : SUCCESS in 3m 29s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/8/check/openstack-tox-pep8/1fe6c36/ : SUCCESS in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/8/check/openstack-tox-py27/a080bb3/ : SUCCESS in 3m 57s\n- openstack-tox-py35 http://logs.openstack.org/69/615469/8/check/openstack-tox-py35/0f1621c/ : SUCCESS in 3m 14s\n- openstack-tox-docs http://logs.openstack.org/69/615469/8/check/openstack-tox-docs/8d5deb3/html/ : SUCCESS in 2m 42s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/8/check/build-openstack-releasenotes/0bf58fa/html/ : SUCCESS in 3m 15s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/8/check/kolla-build-ubuntu-binary/a9e8fe4/ : SUCCESS in 53m 48s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/8/check/kolla-build-ubuntu-source/9c40420/ : SUCCESS in 34m 07s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/8/check/kolla-ansible-ubuntu-source/2e036fc/ : SUCCESS in 1h 05m 44s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/8/check/kolla-ansible-ubuntu-binary/c701c9b/ : SUCCESS in 1h 52m 47s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"52df24199223a1b87024f397689ae7d0455ce6fb","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-12 02:13:09.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b2446ec5711a158a56a4a1dd2389e6ebcbb44839","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-12 04:30:22.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/69/615469/9/check/test-release-openstack/d88bc13/ : SUCCESS in 1m 50s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/9/check/kolla-build-centos-binary/b20c99a/ : FAILURE in 2h 15m 08s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/9/check/kolla-build-centos-source/08dd220/ : SUCCESS in 1h 06m 29s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/9/check/kolla-ansible-centos-source/2a63635/ : SUCCESS in 1h 23m 30s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/9/check/kolla-ansible-centos-binary/5ce3bda/ : SUCCESS in 1h 21m 44s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/615469/9/check/kolla-build-oraclelinux-binary/26b8d10/ : SUCCESS in 1h 44m 15s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/615469/9/check/kolla-build-oraclelinux-source/380393c/ : SUCCESS in 1h 19m 26s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/615469/9/check/kolla-ansible-oraclelinux-source/d01ced7/ : SUCCESS in 1h 44m 14s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/615469/9/check/kolla-ansible-oraclelinux-binary/1037e05/ : SUCCESS in 1h 21m 39s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/69/615469/9/check/openstack-tox-py36/d2494ae/ : SUCCESS in 2m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/9/check/openstack-tox-lower-constraints/dda9aec/ : SUCCESS in 2m 56s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/9/check/openstack-tox-pep8/a1e602b/ : SUCCESS in 3m 38s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/9/check/openstack-tox-py27/78ff180/ : SUCCESS in 3m 00s\n- openstack-tox-py35 http://logs.openstack.org/69/615469/9/check/openstack-tox-py35/ac1e7d9/ : SUCCESS in 3m 26s\n- openstack-tox-docs http://logs.openstack.org/69/615469/9/check/openstack-tox-docs/a969eeb/html/ : SUCCESS in 2m 51s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/9/check/build-openstack-releasenotes/ff6846b/html/ : SUCCESS in 3m 32s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/9/check/kolla-build-ubuntu-binary/a79395f/ : SUCCESS in 1h 24m 02s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/9/check/kolla-build-ubuntu-source/4466ae2/ : SUCCESS in 30m 36s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/9/check/kolla-ansible-ubuntu-source/22f85c6/ : SUCCESS in 51m 09s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/9/check/kolla-ansible-ubuntu-binary/f6791c8/ : TIMED_OUT in 2h 00m 56s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"3fdb98cd55462595b5dfecc5e4b16d847319284e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-12 09:45:13.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"9120951daacb3d072de22834ce6995536a0d39ab","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2018-11-12 13:13:47.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"3513295bdaeaf93136d85ab6e2913c1eaee1be18","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-13 01:49:05.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"2371dcc9b95e18fb8256637cd37c564c061a6c74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-13 03:50:32.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/69/615469/10/check/test-release-openstack/e7706f7/ : SUCCESS in 2m 01s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/10/check/kolla-build-centos-binary/88664c3/ : FAILURE in 23m 44s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/10/check/kolla-build-centos-source/6e62d1d/ : SUCCESS in 1h 37m 25s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/10/check/kolla-ansible-centos-source/84d73ff/ : SUCCESS in 1h 19m 27s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/10/check/kolla-ansible-centos-binary/f9f217d/ : SUCCESS in 1h 33m 11s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/615469/10/check/kolla-build-oraclelinux-binary/b2c5cab/ : SUCCESS in 1h 06m 34s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/615469/10/check/kolla-build-oraclelinux-source/9eb4ee3/ : SUCCESS in 1h 09m 17s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/615469/10/check/kolla-ansible-oraclelinux-source/3d85527/ : SUCCESS in 1h 23m 35s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/615469/10/check/kolla-ansible-oraclelinux-binary/1e4e126/ : SUCCESS in 1h 16m 05s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/69/615469/10/check/openstack-tox-py36/f2af0fd/ : SUCCESS in 2m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/10/check/openstack-tox-lower-constraints/09ecd72/ : SUCCESS in 3m 51s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/10/check/openstack-tox-pep8/a2ed6f4/ : SUCCESS in 3m 17s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/10/check/openstack-tox-py27/4178070/ : SUCCESS in 2m 38s\n- openstack-tox-py35 http://logs.openstack.org/69/615469/10/check/openstack-tox-py35/fbff9ed/ : SUCCESS in 3m 05s\n- openstack-tox-docs http://logs.openstack.org/69/615469/10/check/openstack-tox-docs/7ff8fd8/html/ : SUCCESS in 2m 29s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/10/check/build-openstack-releasenotes/f35749a/html/ : SUCCESS in 3m 14s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/10/check/kolla-build-ubuntu-binary/b2811d5/ : SUCCESS in 1h 23m 04s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/10/check/kolla-build-ubuntu-source/8841a54/ : FAILURE in 1h 05m 36s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/10/check/kolla-ansible-ubuntu-source/554f100/ : SUCCESS in 1h 03m 08s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/10/check/kolla-ansible-ubuntu-binary/d61a95f/ : SUCCESS in 2h 00m 34s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"49571b2ffdfc6af9335c9bd4a7437d3debbc2495","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-13 06:54:54.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"03e9c2420074098a325f188e5909566fa9ee1f62","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-13 07:08:14.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"9f7b9d583de48561d9c0f46e4f8660563ec7a479","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-13 08:56:46.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/69/615469/11/check/test-release-openstack/93afe74/ : SUCCESS in 2m 15s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/11/check/kolla-build-centos-binary/678f6d0/ : FAILURE in 1h 15m 26s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/11/check/kolla-build-centos-source/6a8003d/ : SUCCESS in 1h 14m 24s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/11/check/kolla-ansible-centos-source/2bb5799/ : SUCCESS in 1h 16m 31s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/11/check/kolla-ansible-centos-binary/e6b3cda/ : SUCCESS in 1h 40m 13s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/615469/11/check/kolla-build-oraclelinux-binary/82cd64d/ : SUCCESS in 1h 11m 39s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/615469/11/check/kolla-build-oraclelinux-source/e612927/ : FAILURE in 21m 06s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/615469/11/check/kolla-ansible-oraclelinux-source/69cb204/ : SUCCESS in 1h 19m 22s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/615469/11/check/kolla-ansible-oraclelinux-binary/64ab305/ : SUCCESS in 1h 15m 58s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/69/615469/11/check/openstack-tox-py36/a2c2a0f/ : SUCCESS in 3m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/11/check/openstack-tox-lower-constraints/e800111/ : SUCCESS in 3m 21s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/11/check/openstack-tox-pep8/5f0fdbc/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/11/check/openstack-tox-py27/9525100/ : SUCCESS in 4m 38s\n- openstack-tox-py35 http://logs.openstack.org/69/615469/11/check/openstack-tox-py35/4f345d1/ : SUCCESS in 3m 23s\n- openstack-tox-docs http://logs.openstack.org/69/615469/11/check/openstack-tox-docs/6945163/html/ : SUCCESS in 3m 05s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/11/check/build-openstack-releasenotes/4976b35/html/ : SUCCESS in 3m 43s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/11/check/kolla-build-ubuntu-binary/a3d9d0d/ : SUCCESS in 1h 27m 59s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/11/check/kolla-build-ubuntu-source/9894b58/ : SUCCESS in 31m 37s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/11/check/kolla-ansible-ubuntu-source/16bcb88/ : SUCCESS in 1h 01m 57s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/11/check/kolla-ansible-ubuntu-binary/859aeee/ : SUCCESS in 1h 46m 07s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"6d38da29d861d0a861e6d8f994d458bfeb144028","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-13 11:37:59.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"863665e56c49061e5620390998342c89805502b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-13 14:12:05.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/69/615469/11/check/test-release-openstack/59d8a5f/ : SUCCESS in 2m 12s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/11/check/kolla-build-centos-binary/b68f88f/ : SUCCESS in 46m 44s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/11/check/kolla-build-centos-source/f517f00/ : SUCCESS in 2h 01m 24s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/11/check/kolla-ansible-centos-source/b121584/ : SUCCESS in 1h 14m 59s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/11/check/kolla-ansible-centos-binary/02344d0/ : SUCCESS in 1h 23m 00s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/615469/11/check/kolla-build-oraclelinux-binary/26e5a8e/ : SUCCESS in 2h 10m 45s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/615469/11/check/kolla-build-oraclelinux-source/ba5716d/ : SUCCESS in 1h 11m 48s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/615469/11/check/kolla-ansible-oraclelinux-source/022e445/ : TIMED_OUT in 2h 30m 43s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/615469/11/check/kolla-ansible-oraclelinux-binary/483e424/ : SUCCESS in 1h 10m 45s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/69/615469/11/check/openstack-tox-py36/233b3bb/ : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/11/check/openstack-tox-lower-constraints/ea6ba0a/ : SUCCESS in 3m 06s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/11/check/openstack-tox-pep8/6949a80/ : SUCCESS in 3m 00s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/11/check/openstack-tox-py27/604770f/ : SUCCESS in 3m 03s\n- openstack-tox-py35 http://logs.openstack.org/69/615469/11/check/openstack-tox-py35/2c13b4d/ : SUCCESS in 3m 05s\n- openstack-tox-docs http://logs.openstack.org/69/615469/11/check/openstack-tox-docs/fcc964f/html/ : SUCCESS in 2m 54s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/11/check/build-openstack-releasenotes/1689949/html/ : SUCCESS in 4m 16s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/11/check/kolla-build-ubuntu-binary/f9aecb8/ : SUCCESS in 1h 33m 13s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/11/check/kolla-build-ubuntu-source/e1669bd/ : SUCCESS in 32m 50s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/11/check/kolla-ansible-ubuntu-source/8db631d/ : SUCCESS in 1h 04m 48s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/11/check/kolla-ansible-ubuntu-binary/133f3b4/ : TIMED_OUT in 2h 01m 56s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e83d56de4fa342fbbaac197dc0782c202fca3eff","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-17 01:29:23.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"548c6517621c9ba44448cd303ce024aae9ebd092","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-17 03:24:02.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/69/615469/12/check/test-release-openstack/d0464df/ : SUCCESS in 1m 44s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/12/check/kolla-build-centos-binary/6471fd5/ : SUCCESS in 50m 42s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/12/check/kolla-build-centos-source/8a93fb3/ : FAILURE in 1h 20m 26s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/12/check/kolla-ansible-centos-source/52bb223/ : SUCCESS in 1h 19m 56s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/12/check/kolla-ansible-centos-binary/7a61e44/ : SUCCESS in 1h 11m 58s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/615469/12/check/kolla-build-oraclelinux-binary/1248d7e/ : SUCCESS in 1h 12m 27s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/615469/12/check/kolla-build-oraclelinux-source/af15a31/ : SUCCESS in 1h 17m 35s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/615469/12/check/kolla-ansible-oraclelinux-source/46a5756/ : SUCCESS in 1h 21m 08s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/615469/12/check/kolla-ansible-oraclelinux-binary/af8b726/ : SUCCESS in 1h 13m 49s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/69/615469/12/check/openstack-tox-py36/02ac6eb/ : SUCCESS in 3m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/12/check/openstack-tox-lower-constraints/6d9bc09/ : SUCCESS in 3m 02s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/12/check/openstack-tox-pep8/8ccb550/ : SUCCESS in 3m 11s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/12/check/openstack-tox-py27/a7a3d4b/ : SUCCESS in 2m 47s\n- openstack-tox-py35 http://logs.openstack.org/69/615469/12/check/openstack-tox-py35/e949464/ : SUCCESS in 3m 19s\n- openstack-tox-docs http://logs.openstack.org/69/615469/12/check/openstack-tox-docs/ccdf263/html/ : SUCCESS in 2m 46s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/12/check/build-openstack-releasenotes/eb3a9e7/html/ : SUCCESS in 3m 28s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/12/check/kolla-build-ubuntu-binary/e456934/ : SUCCESS in 1h 17m 38s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/12/check/kolla-build-ubuntu-source/6020980/ : SUCCESS in 46m 39s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/12/check/kolla-ansible-ubuntu-source/1a8ff94/ : SUCCESS in 54m 58s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/12/check/kolla-ansible-ubuntu-binary/ecf0c79/ : SUCCESS in 1h 53m 28s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"1a01b33e01c831f288b0b5af165637c59eba99a2","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-17 04:10:33.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"ffcb36012b62f9a9b9c7f869bbdfd57b9ad97d52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-17 06:16:45.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/69/615469/12/check/test-release-openstack/885e0e9/ : SUCCESS in 2m 30s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/12/check/kolla-build-centos-binary/3885cf3/ : RETRY_LIMIT in 8m 06s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/12/check/kolla-build-centos-source/55b1c22/ : SUCCESS in 1h 11m 32s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/12/check/kolla-ansible-centos-source/262f782/ : FAILURE in 17m 17s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/12/check/kolla-ansible-centos-binary/e14b7cc/ : SUCCESS in 1h 21m 47s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/615469/12/check/kolla-build-oraclelinux-binary/a807a99/ : SUCCESS in 1h 21m 27s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/615469/12/check/kolla-build-oraclelinux-source/e018441/ : SUCCESS in 1h 23m 59s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/615469/12/check/kolla-ansible-oraclelinux-source/b503053/ : SUCCESS in 1h 08m 02s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/615469/12/check/kolla-ansible-oraclelinux-binary/949adbe/ : SUCCESS in 1h 12m 49s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/69/615469/12/check/openstack-tox-py36/26b2cce/ : SUCCESS in 2m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/12/check/openstack-tox-lower-constraints/0dc1442/ : SUCCESS in 3m 20s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/12/check/openstack-tox-pep8/f595115/ : SUCCESS in 2m 37s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/12/check/openstack-tox-py27/42e0551/ : SUCCESS in 3m 26s\n- openstack-tox-py35 http://logs.openstack.org/69/615469/12/check/openstack-tox-py35/736324c/ : SUCCESS in 3m 06s\n- openstack-tox-docs http://logs.openstack.org/69/615469/12/check/openstack-tox-docs/2999e25/html/ : SUCCESS in 3m 10s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/12/check/build-openstack-releasenotes/00924e1/html/ : SUCCESS in 3m 16s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/12/check/kolla-build-ubuntu-binary/17db497/ : SUCCESS in 1h 36m 48s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/12/check/kolla-build-ubuntu-source/4603576/ : SUCCESS in 52m 10s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/12/check/kolla-ansible-ubuntu-source/88e495a/ : SUCCESS in 1h 04m 25s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/12/check/kolla-ansible-ubuntu-binary/23ace2c/ : TIMED_OUT in 2h 01m 19s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"1980a7d0c5aeeb79bab5ea5aa98f25aa48ad861d","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-18 13:52:03.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"f7f51db7a1be7aa812e4b2b1123724f7fbef5ebd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-18 15:16:08.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/69/615469/13/check/test-release-openstack/b9ba0b1/ : SUCCESS in 1m 48s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/13/check/kolla-build-centos-binary/7a9d9c5/ : SUCCESS in 1h 18m 10s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/13/check/kolla-build-centos-source/88866b3/ : SUCCESS in 1h 14m 00s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/13/check/kolla-ansible-centos-source/9fda390/ : FAILURE in 17m 35s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/13/check/kolla-ansible-centos-binary/dd40de8/ : SUCCESS in 1h 16m 20s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/615469/13/check/kolla-build-oraclelinux-binary/ed80a9b/ : SUCCESS in 1h 09m 24s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/615469/13/check/kolla-build-oraclelinux-source/d08bf1a/ : SUCCESS in 59m 43s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/615469/13/check/kolla-ansible-oraclelinux-source/a1cf097/ : SUCCESS in 1h 07m 08s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/615469/13/check/kolla-ansible-oraclelinux-binary/c6a2dae/ : FAILURE in 16m 47s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/69/615469/13/check/openstack-tox-py36/1da2848/ : SUCCESS in 2m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/13/check/openstack-tox-lower-constraints/ec542ea/ : SUCCESS in 2m 49s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/13/check/openstack-tox-pep8/30b942a/ : SUCCESS in 2m 31s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/13/check/openstack-tox-py27/8175ebe/ : SUCCESS in 3m 11s\n- openstack-tox-py35 http://logs.openstack.org/69/615469/13/check/openstack-tox-py35/91f282c/ : SUCCESS in 3m 35s\n- openstack-tox-docs http://logs.openstack.org/69/615469/13/check/openstack-tox-docs/bf7ba89/html/ : SUCCESS in 2m 10s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/13/check/build-openstack-releasenotes/51674ae/html/ : SUCCESS in 3m 08s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/13/check/kolla-build-ubuntu-binary/5816a3f/ : SUCCESS in 1h 19m 48s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/13/check/kolla-build-ubuntu-source/f20bba7/ : SUCCESS in 32m 09s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/13/check/kolla-ansible-ubuntu-source/4a7aebc/ : SUCCESS in 49m 49s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/13/check/kolla-ansible-ubuntu-binary/e12b540/ : SUCCESS in 1h 22m 26s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"aadd618f0d5b397c7676b7fd98b3e80979619e92","author":{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},"date":"2018-11-18 15:18:13.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"5f140b050f825e967b9b271677b27a733d64ef81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-18 16:56:47.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/69/615469/13/check/test-release-openstack/1392b15/ : SUCCESS in 1m 43s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/13/check/kolla-build-centos-binary/7340532/ : FAILURE in 54m 11s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/13/check/kolla-build-centos-source/9c7877b/ : SUCCESS in 1h 08m 55s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/13/check/kolla-ansible-centos-source/a0c3ac8/ : SUCCESS in 1h 37m 30s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/13/check/kolla-ansible-centos-binary/0a8b91f/ : SUCCESS in 1h 13m 07s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/615469/13/check/kolla-build-oraclelinux-binary/80b8a5a/ : SUCCESS in 1h 07m 01s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/615469/13/check/kolla-build-oraclelinux-source/fe708e3/ : FAILURE in 17m 29s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/615469/13/check/kolla-ansible-oraclelinux-source/e8feb87/ : SUCCESS in 1h 31m 21s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/615469/13/check/kolla-ansible-oraclelinux-binary/8328bdc/ : SUCCESS in 1h 20m 01s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/69/615469/13/check/openstack-tox-py36/9f36e50/ : SUCCESS in 2m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/13/check/openstack-tox-lower-constraints/f2cb3a6/ : SUCCESS in 3m 36s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/13/check/openstack-tox-pep8/4c17170/ : SUCCESS in 2m 45s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/13/check/openstack-tox-py27/6e6b277/ : SUCCESS in 3m 11s\n- openstack-tox-py35 http://logs.openstack.org/69/615469/13/check/openstack-tox-py35/82f3bff/ : SUCCESS in 2m 29s\n- openstack-tox-docs http://logs.openstack.org/69/615469/13/check/openstack-tox-docs/dc1c3f6/html/ : SUCCESS in 2m 31s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/13/check/build-openstack-releasenotes/54af2c6/html/ : SUCCESS in 3m 36s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/13/check/kolla-build-ubuntu-binary/22755cf/ : SUCCESS in 1h 05m 46s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/13/check/kolla-build-ubuntu-source/c56d572/ : SUCCESS in 32m 55s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/13/check/kolla-ansible-ubuntu-source/fcb08f7/ : SUCCESS in 1h 01m 24s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/13/check/kolla-ansible-ubuntu-binary/6a6bb15/ : SUCCESS in 1h 27m 46s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"3341837831e341e3a55dc3a25e48ea538a603ed7","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-19 01:05:37.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"80a1be54c41984d6e1b527ffde18d90be05f1d30","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2018-11-19 01:12:05.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"01f1955716ca41957f655f25dee25ac8123db4da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-19 03:25:51.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/69/615469/13/check/test-release-openstack/9542c8d/ : SUCCESS in 2m 49s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/13/check/kolla-build-centos-binary/dd4e821/ : FAILURE in 2h 18m 50s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/13/check/kolla-build-centos-source/0cea57b/ : FAILURE in 2h 12m 59s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/13/check/kolla-ansible-centos-source/cf29ab1/ : TIMED_OUT in 2h 01m 15s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/13/check/kolla-ansible-centos-binary/52ecc13/ : SUCCESS in 1h 18m 27s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/615469/13/check/kolla-build-oraclelinux-binary/f3a1c87/ : SUCCESS in 1h 10m 24s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/615469/13/check/kolla-build-oraclelinux-source/cb0ef3c/ : SUCCESS in 1h 30m 10s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/615469/13/check/kolla-ansible-oraclelinux-source/f366561/ : TIMED_OUT in 2h 01m 24s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/615469/13/check/kolla-ansible-oraclelinux-binary/0324b9a/ : SUCCESS in 1h 18m 34s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/69/615469/13/check/openstack-tox-py36/b084640/ : SUCCESS in 2m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/13/check/openstack-tox-lower-constraints/94fb5aa/ : SUCCESS in 4m 18s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/13/check/openstack-tox-pep8/0e6702e/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/13/check/openstack-tox-py27/115de90/ : SUCCESS in 3m 08s\n- openstack-tox-py35 http://logs.openstack.org/69/615469/13/check/openstack-tox-py35/836e8d7/ : SUCCESS in 3m 56s\n- openstack-tox-docs http://logs.openstack.org/69/615469/13/check/openstack-tox-docs/680b3cb/html/ : SUCCESS in 3m 23s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/13/check/build-openstack-releasenotes/0db992d/html/ : SUCCESS in 3m 44s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/13/check/kolla-build-ubuntu-binary/0d71618/ : SUCCESS in 1h 10m 42s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/13/check/kolla-build-ubuntu-source/0e0ba33/ : SUCCESS in 29m 53s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/13/check/kolla-ansible-ubuntu-source/8acedf2/ : SUCCESS in 1h 03m 35s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/13/check/kolla-ansible-ubuntu-binary/52a58eb/ : SUCCESS in 1h 44m 21s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"4d9b4796525cffc975076860a631cf391fcf86f5","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-20 11:40:01.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"99ce5ee46a41a67629a865b890450a73c0681f81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-20 14:43:40.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/69/615469/14/check/test-release-openstack/f1e91e6/ : SUCCESS in 3m 34s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/14/check/kolla-build-centos-binary/ed6b1ba/ : SUCCESS in 1h 15m 16s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/14/check/kolla-build-centos-source/4086ea1/ : TIMED_OUT in 2h 58m 05s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/14/check/kolla-ansible-centos-source/4e9b2f2/ : SUCCESS in 1h 22m 01s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/14/check/kolla-ansible-centos-binary/bb7be2d/ : SUCCESS in 1h 25m 15s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/615469/14/check/kolla-build-oraclelinux-binary/b0b55c6/ : SUCCESS in 1h 19m 53s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/615469/14/check/kolla-build-oraclelinux-source/56f0a2b/ : SUCCESS in 1h 31m 33s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/615469/14/check/kolla-ansible-oraclelinux-source/779cef9/ : SUCCESS in 1h 18m 52s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/615469/14/check/kolla-ansible-oraclelinux-binary/e43e580/ : FAILURE in 5m 30s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/69/615469/14/check/openstack-tox-py36/11b03ba/ : SUCCESS in 3m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/14/check/openstack-tox-lower-constraints/ce09cc2/ : SUCCESS in 4m 05s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/14/check/openstack-tox-pep8/64542a6/ : SUCCESS in 3m 20s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/14/check/openstack-tox-py27/d596f17/ : SUCCESS in 3m 57s\n- openstack-tox-py35 http://logs.openstack.org/69/615469/14/check/openstack-tox-py35/eb9928e/ : SUCCESS in 4m 15s\n- openstack-tox-docs http://logs.openstack.org/69/615469/14/check/openstack-tox-docs/83a3a64/html/ : SUCCESS in 3m 25s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/14/check/build-openstack-releasenotes/07273f6/html/ : SUCCESS in 4m 27s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/14/check/kolla-build-ubuntu-binary/3ba56f5/ : SUCCESS in 1h 16m 09s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/14/check/kolla-build-ubuntu-source/5483ec7/ : SUCCESS in 41m 42s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/14/check/kolla-ansible-ubuntu-source/b8b8218/ : SUCCESS in 1h 12m 57s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/14/check/kolla-ansible-ubuntu-binary/6b15345/ : SUCCESS in 1h 54m 13s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"87e55f5ccc50c3eefce1621961fd71f5ba58db61","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-21 01:40:03.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"c5735ac8ba6de79141115f5dcc3d25abbdda6f1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-21 04:12:34.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/69/615469/14/check/test-release-openstack/757d595/ : SUCCESS in 2m 22s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/14/check/kolla-build-centos-binary/9e2fa08/ : FAILURE in 53m 04s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/14/check/kolla-build-centos-source/620aa3d/ : FAILURE in 2h 11m 33s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/14/check/kolla-ansible-centos-source/c94bd87/ : TIMED_OUT in 2h 30m 41s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/14/check/kolla-ansible-centos-binary/24dc697/ : SUCCESS in 1h 15m 38s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/615469/14/check/kolla-build-oraclelinux-binary/01dac0d/ : FAILURE in 1h 10m 28s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/615469/14/check/kolla-build-oraclelinux-source/ffb89c5/ : FAILURE in 1h 52m 18s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/615469/14/check/kolla-ansible-oraclelinux-source/6bef894/ : SUCCESS in 1h 30m 03s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/615469/14/check/kolla-ansible-oraclelinux-binary/cda7cdd/ : SUCCESS in 1h 08m 46s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/69/615469/14/check/openstack-tox-py36/66c45eb/ : SUCCESS in 2m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/14/check/openstack-tox-lower-constraints/099e066/ : SUCCESS in 3m 29s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/14/check/openstack-tox-pep8/2fd2164/ : SUCCESS in 3m 25s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/14/check/openstack-tox-py27/d58ecd5/ : SUCCESS in 2m 47s\n- openstack-tox-py35 http://logs.openstack.org/69/615469/14/check/openstack-tox-py35/a2b321b/ : SUCCESS in 4m 03s\n- openstack-tox-docs http://logs.openstack.org/69/615469/14/check/openstack-tox-docs/c6fe1b8/html/ : SUCCESS in 3m 41s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/14/check/build-openstack-releasenotes/d110c2e/html/ : SUCCESS in 3m 40s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/14/check/kolla-build-ubuntu-binary/5361fd5/ : FAILURE in 1h 09m 06s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/14/check/kolla-build-ubuntu-source/2419fb5/ : FAILURE in 30m 03s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/14/check/kolla-ansible-ubuntu-source/f65f03a/ : SUCCESS in 1h 36m 17s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/14/check/kolla-ansible-ubuntu-binary/15ab2e9/ : SUCCESS in 1h 57m 58s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"76039a7c5937c483fff540f3a20b1b3e83cc6365","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2018-11-21 14:15:20.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"974a51985cb8bec4e699c3675502148e1f856f14","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-21 14:17:11.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"87f34d35ecdee0b9fe7a0b9372bd24bb7e6f2269","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-11-22 07:49:01.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"983ea5b9ee096eebe7a418e74dd54c0817a72054","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-22 09:53:25.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/69/615469/15/check/test-release-openstack/9d01264/ : SUCCESS in 1m 58s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/15/check/kolla-build-centos-binary/f66d10d/ : FAILURE in 1h 05m 44s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/15/check/kolla-build-centos-source/c3b60f0/ : FAILURE in 1h 10m 56s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/15/check/kolla-ansible-centos-source/20dfe26/ : TIMED_OUT in 2h 03m 03s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/15/check/kolla-ansible-centos-binary/4408340/ : SUCCESS in 1h 12m 34s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/615469/15/check/kolla-build-oraclelinux-binary/c4cb6c9/ : FAILURE in 18m 00s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/615469/15/check/kolla-build-oraclelinux-source/715f56f/ : SUCCESS in 1h 25m 44s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/615469/15/check/kolla-ansible-oraclelinux-source/57daff2/ : SUCCESS in 1h 24m 26s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/615469/15/check/kolla-ansible-oraclelinux-binary/ab9d8e8/ : TIMED_OUT in 2h 01m 07s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/69/615469/15/check/openstack-tox-py36/224e4db/ : SUCCESS in 4m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/15/check/openstack-tox-lower-constraints/c6b4e48/ : SUCCESS in 3m 34s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/15/check/openstack-tox-pep8/59a5277/ : SUCCESS in 2m 49s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/15/check/openstack-tox-py27/1710623/ : SUCCESS in 2m 37s\n- openstack-tox-py35 http://logs.openstack.org/69/615469/15/check/openstack-tox-py35/7ebaddd/ : SUCCESS in 4m 17s\n- openstack-tox-docs http://logs.openstack.org/69/615469/15/check/openstack-tox-docs/fe47743/html/ : SUCCESS in 3m 01s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/15/check/build-openstack-releasenotes/8690d07/html/ : SUCCESS in 3m 37s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/15/check/kolla-build-ubuntu-binary/0504013/ : SUCCESS in 1h 11m 59s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/15/check/kolla-build-ubuntu-source/f6c317b/ : SUCCESS in 41m 56s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/15/check/kolla-ansible-ubuntu-source/e9b986d/ : SUCCESS in 1h 02m 56s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/15/check/kolla-ansible-ubuntu-binary/25dd677/ : SUCCESS in 1h 38m 07s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"7ab964febc18d0341e28e84e04e66fa86e7ff8df","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-12-11 02:17:17.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"07f7ff896b2e5a2481573115cd6c59236beec817","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-11 04:25:02.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/69/615469/16/check/test-release-openstack/cdcf19c/ : SUCCESS in 1m 54s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/16/check/kolla-build-centos-binary/46238ef/ : FAILURE in 59m 06s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/16/check/kolla-build-centos-source/18eb3b9/ : FAILURE in 32m 17s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/16/check/kolla-ansible-centos-source/e871c43/ : FAILURE in 23m 37s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/16/check/kolla-ansible-centos-binary/5992b3e/ : FAILURE in 45m 49s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/69/615469/16/check/tripleo-build-containers-centos-7/0c1e039/ : FAILURE in 40m 51s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/615469/16/check/kolla-build-oraclelinux-binary/c71667f/ : FAILURE in 59m 24s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/615469/16/check/kolla-build-oraclelinux-source/dec5c50/ : FAILURE in 30m 11s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/615469/16/check/kolla-ansible-oraclelinux-source/ae98a50/ : FAILURE in 21m 58s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/615469/16/check/kolla-ansible-oraclelinux-binary/8038a38/ : FAILURE in 33m 29s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/69/615469/16/check/openstack-tox-py36/1710745/ : SUCCESS in 3m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/16/check/openstack-tox-lower-constraints/a8a5e1a/ : SUCCESS in 3m 50s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/16/check/openstack-tox-pep8/1ef72ad/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/16/check/openstack-tox-py27/18bef94/ : SUCCESS in 3m 14s\n- openstack-tox-py35 http://logs.openstack.org/69/615469/16/check/openstack-tox-py35/178311d/ : SUCCESS in 3m 21s\n- openstack-tox-docs http://logs.openstack.org/69/615469/16/check/openstack-tox-docs/69d373c/html/ : SUCCESS in 3m 18s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/16/check/build-openstack-releasenotes/b7c9dc4/html/ : SUCCESS in 3m 38s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/16/check/kolla-build-ubuntu-binary/b40f31d/ : SUCCESS in 1h 23m 47s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/16/check/kolla-build-ubuntu-source/f84fa3b/ : SUCCESS in 34m 12s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/16/check/kolla-ansible-ubuntu-source/80ca855/ : SUCCESS in 1h 33m 19s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/16/check/kolla-ansible-ubuntu-binary/20d691d/ : TIMED_OUT in 2h 00m 59s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"b0ef1eda73f9aadff9f6ddd7f757a211b65782d9","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2018-12-11 04:25:21.000000000","message":"Patch Set 16:\n\nI noticed Zuul failed, I think you hit bug(s):\n\n- kolla-ansible-centos-source: unrecognized error\n- kolla-build-centos-source: https://bugs.launchpad.net/bugs/1708704\n- kolla-build-oraclelinux-source: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":16},{"id":"35f836cc95b736bb88f6b633accd38fc10c22b4a","author":{"_account_id":23479,"name":"Murali Annamneni","email":"murali.annamneni@oracle.com","username":"mannamne"},"date":"2018-12-13 12:30:40.000000000","message":"Patch Set 16:\n\nI\u0027m trying to build \u0026 deploy masakari containers. And I observed that  on compute node, masakari containers are failed to start with reason\nIOError: [Errno 2] No such file or directory: \u0027/var/log/kolla/masakari/masakari-hostmonitor.log\u0027\n\n\nI think, since these containers are built from openstack-base isntead of masakari-base, its missing extend-start sequence, where the log directory being created. So, if these images need to build from openstack-base, they need to have its own extend_start","accounts_in_message":[],"_revision_number":16},{"id":"5854c8116604fc998a6821ca6ff61e418fd9fa3c","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2018-12-13 12:33:20.000000000","message":"Patch Set 16:\n\nyeah, probably, good catch","accounts_in_message":[],"_revision_number":16},{"id":"5df4d2ce1af3cf737d96b232bf012ec8d95e28a7","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-12-14 01:30:14.000000000","message":"Patch Set 16:\n\n\u003e I\u0027m trying to build \u0026 deploy masakari containers. And I observed\n \u003e that  on compute node, masakari containers are failed to start with\n \u003e reason\n \u003e IOError: [Errno 2] No such file or directory: \u0027/var/log/kolla/masakari/masakari-hostmonitor.log\u0027\n \u003e \n \u003e \n \u003e I think, since these containers are built from openstack-base\n \u003e isntead of masakari-base, its missing extend-start sequence, where\n \u003e the log directory being created. So, if these images need to build\n \u003e from openstack-base, they need to have its own extend_start\n\nThank you for testing! :D","accounts_in_message":[],"_revision_number":16},{"id":"8a86bd3b27217ff5d70b74a2838276acb25e6502","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-12-14 01:31:52.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"6bd31d0ea9ca54ce3ebc5a7de11708803e98a09a","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2018-12-14 03:16:18.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"ad0894962bdca6e4dd414239c75680c8cfb5c4ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-14 03:34:31.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/69/615469/17/check/test-release-openstack/ac399a4/ : SUCCESS in 2m 18s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/17/check/kolla-build-centos-binary/94f0025/ : FAILURE in 1h 22m 04s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/17/check/kolla-build-centos-source/f6f7761/ : SUCCESS in 1h 28m 32s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/17/check/kolla-ansible-centos-source/6d68dce/ : FAILURE in 56m 22s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/17/check/kolla-ansible-centos-binary/add032b/ : SUCCESS in 1h 23m 58s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/69/615469/17/check/tripleo-build-containers-centos-7/2253085/ : SUCCESS in 32m 30s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/615469/17/check/kolla-build-oraclelinux-binary/507f85e/ : FAILURE in 57m 18s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/615469/17/check/kolla-build-oraclelinux-source/c0b0b0e/ : SUCCESS in 1h 02m 10s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/615469/17/check/kolla-ansible-oraclelinux-source/dfd8f05/ : FAILURE in 54m 21s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/615469/17/check/kolla-ansible-oraclelinux-binary/bf39c6e/ : SUCCESS in 1h 16m 38s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/69/615469/17/check/openstack-tox-py36/87739dd/ : SUCCESS in 3m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/17/check/openstack-tox-lower-constraints/daf4cf5/ : SUCCESS in 4m 04s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/17/check/openstack-tox-pep8/c54d63e/ : SUCCESS in 3m 11s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/17/check/openstack-tox-py27/5065462/ : SUCCESS in 3m 08s\n- openstack-tox-py35 http://logs.openstack.org/69/615469/17/check/openstack-tox-py35/1fe9952/ : SUCCESS in 3m 32s\n- openstack-tox-docs http://logs.openstack.org/69/615469/17/check/openstack-tox-docs/a015f8a/html/ : SUCCESS in 3m 36s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/17/check/build-openstack-releasenotes/2b8d238/html/ : SUCCESS in 3m 34s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/17/check/kolla-build-ubuntu-binary/e198107/ : SUCCESS in 1h 17m 25s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/17/check/kolla-build-ubuntu-source/29d0e1c/ : SUCCESS in 34m 18s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/17/check/kolla-ansible-ubuntu-source/f6f8d28/ : FAILURE in 46m 08s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/17/check/kolla-ansible-ubuntu-binary/395b9da/ : SUCCESS in 1h 55m 04s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"bec3faa72619ec78fd4d0152cf0913a6ffd11dc7","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2018-12-14 03:34:48.000000000","message":"Patch Set 17:\n\nI noticed Zuul failed, I think you hit bug(s):\n\n- kolla-ansible-centos-source: https://bugs.launchpad.net/bugs/1708704\n- kolla-ansible-ubuntu-source: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":17},{"id":"0693effbfd4bc9a048ba00b9da0eb140b86f7ca1","author":{"_account_id":23479,"name":"Murali Annamneni","email":"murali.annamneni@oracle.com","username":"mannamne"},"date":"2018-12-14 14:30:37.000000000","message":"Patch Set 17:\n\nI think masakari-monitors container require \"pacemaker\" packages. I see the following error in masakari-hostmonitor logs:-\nDEBUG oslo.privsep.daemon [-] privsep: reply[139747880361488]: (5, \u0027oslo_concurrency.processutils.ProcessExecutionError\u0027, (u\u0027\u0027, u\u0027Failed to get D-Bus connection: Operation not permitted\\n\u0027, 1, u\u0027systemctl status pacemaker_remote\u0027, None)) loop /var/lib/kolla/venv/lib/python2.7/site-packages/oslo_privsep/daemon.py:456\n2018-12-14 14:28:01.095 7 ERROR masakarimonitors.hostmonitor.host_handler.handle_host [-] Neither pacemaker nor pacemaker-remote is running.","accounts_in_message":[],"_revision_number":17},{"id":"f71da38ada356e2f27f881ed07543036a32b1d7e","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-12-19 10:12:11.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"29caf85ba125f921eff6a9ed41ff4446b266436c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-19 12:18:52.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/69/615469/18/check/test-release-openstack/add3f27/ : SUCCESS in 2m 04s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/18/check/kolla-build-centos-binary/da6d8d6/ : SUCCESS in 45m 01s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/18/check/kolla-build-centos-source/866ef1e/ : FAILURE in 17m 01s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/18/check/kolla-ansible-centos-source/c0565c4/ : SUCCESS in 1h 18m 36s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/18/check/kolla-ansible-centos-binary/2c3b36a/ : SUCCESS in 1h 15m 33s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/69/615469/18/check/tripleo-build-containers-centos-7/c49a386/ : SUCCESS in 32m 15s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/615469/18/check/kolla-build-oraclelinux-binary/d544383/ : SUCCESS in 1h 02m 33s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/615469/18/check/kolla-build-oraclelinux-source/7ddd8de/ : SUCCESS in 1h 08m 30s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/615469/18/check/kolla-ansible-oraclelinux-source/44c75b4/ : SUCCESS in 1h 23m 40s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/615469/18/check/kolla-ansible-oraclelinux-binary/290990f/ : SUCCESS in 1h 14m 08s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/69/615469/18/check/openstack-tox-py36/a87ee69/ : SUCCESS in 2m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/18/check/openstack-tox-lower-constraints/f4b0719/ : SUCCESS in 3m 28s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/18/check/openstack-tox-pep8/77eb0b7/ : SUCCESS in 3m 21s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/18/check/openstack-tox-py27/08b632e/ : SUCCESS in 3m 13s\n- openstack-tox-py35 http://logs.openstack.org/69/615469/18/check/openstack-tox-py35/166d641/ : SUCCESS in 3m 09s\n- openstack-tox-docs http://logs.openstack.org/69/615469/18/check/openstack-tox-docs/0317ecf/html/ : SUCCESS in 2m 16s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/18/check/build-openstack-releasenotes/5158455/html/ : SUCCESS in 3m 36s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/18/check/kolla-build-ubuntu-binary/d283132/ : SUCCESS in 1h 27m 57s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/18/check/kolla-build-ubuntu-source/99a6976/ : SUCCESS in 38m 30s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/18/check/kolla-ansible-ubuntu-source/b44be9f/ : SUCCESS in 1h 03m 12s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/18/check/kolla-ansible-ubuntu-binary/3e154bf/ : SUCCESS in 2h 00m 00s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"150074be7818ea95518f258c4d12ec361c715a49","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-12-20 01:20:16.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"4443000aeea9ee51f45cf80eb3f80d9d8d2204a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-20 01:54:11.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/69/615469/19/check/test-release-openstack/724e996/ : SUCCESS in 2m 16s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/19/check/kolla-build-centos-binary/d044900/ : FAILURE in 11m 14s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/19/check/kolla-build-centos-source/b57d3bc/ : FAILURE in 7m 49s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/19/check/kolla-ansible-centos-source/ceb43d1/ : FAILURE in 7m 55s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/19/check/kolla-ansible-centos-binary/4c9231e/ : FAILURE in 7m 40s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/69/615469/19/check/tripleo-build-containers-centos-7/7f72741/ : FAILURE in 9m 37s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/615469/19/check/kolla-build-oraclelinux-binary/ce1f97e/ : FAILURE in 10m 23s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/615469/19/check/kolla-build-oraclelinux-source/5831fc9/ : FAILURE in 11m 25s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/615469/19/check/kolla-ansible-oraclelinux-source/7816645/ : FAILURE in 11m 57s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/615469/19/check/kolla-ansible-oraclelinux-binary/d873e81/ : FAILURE in 7m 52s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/69/615469/19/check/openstack-tox-py36/768949b/ : SUCCESS in 2m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/19/check/openstack-tox-lower-constraints/61c816b/ : SUCCESS in 3m 17s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/19/check/openstack-tox-pep8/06363d1/ : SUCCESS in 3m 20s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/19/check/openstack-tox-py27/f98af2d/ : SUCCESS in 3m 11s\n- openstack-tox-py35 http://logs.openstack.org/69/615469/19/check/openstack-tox-py35/ec6a419/ : SUCCESS in 3m 13s\n- openstack-tox-docs http://logs.openstack.org/69/615469/19/check/openstack-tox-docs/277d6c0/html/ : SUCCESS in 2m 49s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/19/check/build-openstack-releasenotes/d8aa5ee/html/ : SUCCESS in 3m 23s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/19/check/kolla-build-ubuntu-binary/5b30d95/ : FAILURE in 4m 27s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/19/check/kolla-build-ubuntu-source/086942e/ : FAILURE in 3m 50s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/19/check/kolla-ansible-ubuntu-source/9032ddc/ : FAILURE in 7m 02s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/19/check/kolla-ansible-ubuntu-binary/7b0b83d/ : FAILURE in 8m 10s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"447e215ab07c99601e99e1228a8973bb977a851c","author":{"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},"date":"2019-03-04 07:14:04.000000000","message":"Patch Set 20: Patch Set 19 was rebased","accounts_in_message":[],"_revision_number":20},{"id":"cb99fec8f5fe99f788393b348f7b29e953187b31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-04 07:35:11.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/69/615469/20/check/test-release-openstack/97076f1/ : SUCCESS in 2m 14s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/20/check/kolla-build-centos-binary/09acdd5/ : FAILURE in 5m 49s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/20/check/kolla-build-centos-source/889ff3a/ : FAILURE in 5m 11s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/20/check/kolla-ansible-centos-source/2d9edd8/ : FAILURE in 7m 35s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/20/check/kolla-ansible-centos-binary/42934cc/ : FAILURE in 8m 11s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/69/615469/20/check/tripleo-build-containers-centos-7/b433a55/ : FAILURE in 12m 42s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/615469/20/check/kolla-build-oraclelinux-binary/5897bbc/ : FAILURE in 5m 20s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/615469/20/check/kolla-build-oraclelinux-source/7cc9a8d/ : FAILURE in 5m 42s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/615469/20/check/kolla-ansible-oraclelinux-source/279edd2/ : FAILURE in 9m 18s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/615469/20/check/kolla-ansible-oraclelinux-binary/23ec845/ : FAILURE in 8m 49s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/20/check/openstack-tox-lower-constraints/ffee2c5/ : SUCCESS in 3m 33s\n- openstack-tox-py36 http://logs.openstack.org/69/615469/20/check/openstack-tox-py36/5fe3efe/ : SUCCESS in 4m 24s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/20/check/openstack-tox-pep8/9170c6a/ : SUCCESS in 3m 21s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/20/check/openstack-tox-py27/62305f1/ : SUCCESS in 4m 18s\n- openstack-tox-py35 http://logs.openstack.org/69/615469/20/check/openstack-tox-py35/28c0541/ : SUCCESS in 2m 58s\n- openstack-tox-docs http://logs.openstack.org/69/615469/20/check/openstack-tox-docs/e56a510/html/ : SUCCESS in 2m 58s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/20/check/build-openstack-releasenotes/ea0b548/html/ : SUCCESS in 4m 05s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/20/check/kolla-build-ubuntu-binary/130a781/ : FAILURE in 5m 34s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/20/check/kolla-build-ubuntu-source/4e8c2cc/ : FAILURE in 4m 02s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/20/check/kolla-ansible-ubuntu-source/5fbff39/ : FAILURE in 7m 58s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/20/check/kolla-ansible-ubuntu-binary/8268048/ : FAILURE in 7m 08s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"9e5278fd0cb44874774f4a27ef73621f04046a34","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-04 07:50:30.000000000","message":"Patch Set 20:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 http://logs.rdoproject.org/69/615469/20/openstack-check/tripleo-build-containers-fedora-28/d7b4896/ : SUCCESS in 30m 01s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"e08841e3a68b1b7f1b4009676379fd8cf9413e40","author":{"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},"date":"2019-03-10 15:39:55.000000000","message":"Patch Set 21: Patch Set 20 was rebased","accounts_in_message":[],"_revision_number":21},{"id":"9e2c56232a39bc69f12d48206362bcefd17b891f","author":{"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},"date":"2019-03-10 15:46:45.000000000","message":"Patch Set 21:\n\nany progress?","accounts_in_message":[],"_revision_number":21},{"id":"206bbc027bc9da76147a19bd431de53b33d84920","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-10 16:08:00.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/69/615469/21/check/test-release-openstack/6aead5f/ : SUCCESS in 2m 18s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/21/check/kolla-build-centos-binary/9ca74cb/ : FAILURE in 6m 05s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/21/check/kolla-build-centos-source/1241014/ : FAILURE in 6m 25s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/21/check/kolla-ansible-centos-source/c6a36c7/ : FAILURE in 9m 52s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/21/check/kolla-ansible-centos-binary/752e643/ : FAILURE in 11m 36s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/69/615469/21/check/tripleo-build-containers-centos-7/35c08b4/ : FAILURE in 10m 48s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/615469/21/check/kolla-build-oraclelinux-binary/93f0c65/ : FAILURE in 8m 11s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/615469/21/check/kolla-build-oraclelinux-source/71dd8ba/ : FAILURE in 6m 23s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/615469/21/check/kolla-ansible-oraclelinux-source/1957154/ : FAILURE in 8m 12s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/615469/21/check/kolla-ansible-oraclelinux-binary/486682c/ : FAILURE in 9m 16s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/21/check/openstack-tox-lower-constraints/f60c87f/ : SUCCESS in 3m 44s\n- openstack-tox-py36 http://logs.openstack.org/69/615469/21/check/openstack-tox-py36/fe15666/ : SUCCESS in 3m 30s\n- openstack-tox-py37 http://logs.openstack.org/69/615469/21/check/openstack-tox-py37/f201b6f/ : SUCCESS in 5m 28s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/21/check/openstack-tox-pep8/f3d1720/ : SUCCESS in 3m 28s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/21/check/openstack-tox-py27/198c00d/ : SUCCESS in 4m 03s\n- openstack-tox-docs http://logs.openstack.org/69/615469/21/check/openstack-tox-docs/db2815f/html/ : SUCCESS in 3m 02s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/21/check/build-openstack-releasenotes/54b1461/html/ : SUCCESS in 4m 18s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/21/check/kolla-build-ubuntu-binary/842f474/ : FAILURE in 4m 48s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/21/check/kolla-build-ubuntu-source/3514412/ : FAILURE in 5m 00s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/21/check/kolla-ansible-ubuntu-source/44dc666/ : FAILURE in 7m 20s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/21/check/kolla-ansible-ubuntu-binary/2868291/ : FAILURE in 7m 27s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"6752bcce069802b17ff576315dadc15435ea9917","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-10 16:38:43.000000000","message":"Patch Set 21:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 http://logs.rdoproject.org/69/615469/21/openstack-check/tripleo-build-containers-fedora-28/9551f79/ : SUCCESS in 33m 59s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"f3bbc7f6f7d7ac4053350d7db97072cb6506c1f1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-12 14:13:31.000000000","message":"Patch Set 21:\n\nHi, will this be worked on during the Train cycle?","accounts_in_message":[],"_revision_number":21},{"id":"2244c2f65b6aa44313ca79bd9cb8a9909a8966d9","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-13 14:00:23.000000000","message":"Patch Set 21: Code-Review-1\n\nplease fix the conflict","accounts_in_message":[],"_revision_number":21},{"id":"9c562a9b8491410b33622731d5ef6af2e964cf91","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-06-25 18:33:03.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"d2ac7167ddd4cf0aec1c40bed83f77fcf705a8b0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-25 18:46:41.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"d7c4c0cfb5c68712a53983351dd29057a9f34b4d","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-06-25 19:11:19.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"221f03273cde99a9123958215714f2742dfbd8b5","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-06-25 19:13:43.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"cfdcf689f73cba45039948d0ce2ab5e2ec99eaf1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-25 20:18:36.000000000","message":"Patch Set 24:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/69/615469/24/openstack-check/tripleo-build-containers-fedora-28-master/a76f1bc/ : FAILURE in 5m 16s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/69/615469/24/openstack-check/tripleo-build-containers-fedora-28-stein/52acc8b/ : SUCCESS in 1h 03m 56s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"6bba39db2cde7d2da8873cd36940f9f03774798e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 22:00:57.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/69/615469/24/check/test-release-openstack/d468058/ : SUCCESS in 3m 14s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/24/check/kolla-build-centos-binary/dfed1ef/ : SUCCESS in 41m 48s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/24/check/kolla-build-centos-source/e3ace45/ : SUCCESS in 2h 10m 47s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/24/check/kolla-ansible-centos-source/abb39d2/ : TIMED_OUT in 2h 01m 03s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/24/check/kolla-ansible-centos-binary/5ad274d/ : SUCCESS in 1h 03m 41s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/69/615469/24/check/kolla-ansible-centos-source-upgrade/c8d6290/ : TIMED_OUT in 2h 01m 08s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/69/615469/24/check/tripleo-build-containers-centos-7/82653a3/ : SUCCESS in 44m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/24/check/openstack-tox-lower-constraints/1630ebc/ : SUCCESS in 3m 58s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/24/check/openstack-tox-pep8/a87bf03/ : SUCCESS in 3m 09s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/24/check/openstack-tox-py27/e8bff0c/ : SUCCESS in 3m 34s\n- openstack-tox-py36 http://logs.openstack.org/69/615469/24/check/openstack-tox-py36/1adacce/ : SUCCESS in 3m 43s\n- openstack-tox-py37 http://logs.openstack.org/69/615469/24/check/openstack-tox-py37/6e3387c/ : SUCCESS in 3m 47s\n- openstack-tox-docs http://logs.openstack.org/69/615469/24/check/openstack-tox-docs/8785c5e/html/ : SUCCESS in 2m 43s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/24/check/build-openstack-releasenotes/a6e6daf/html/ : SUCCESS in 4m 14s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/24/check/kolla-build-ubuntu-binary/1b386b2/ : SUCCESS in 23m 47s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/24/check/kolla-build-ubuntu-source/2546942/ : SUCCESS in 29m 56s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/24/check/kolla-ansible-ubuntu-source/3408a03/ : SUCCESS in 47m 46s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/24/check/kolla-ansible-ubuntu-binary/be13d53/ : SUCCESS in 53m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/69/615469/24/check/kolla-ansible-ubuntu-source-upgrade/f79750c/ : SUCCESS in 57m 56s","accounts_in_message":[],"_revision_number":24},{"id":"bfcd01ab104a1a4deaab117bbd6b820a8b815edc","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-06-26 02:10:12.000000000","message":"Patch Set 24:\n\nrecheck","accounts_in_message":[],"_revision_number":24},{"id":"d74b8570dbdbcf076ed7385c6bd3c2c78399c37b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-26 02:45:49.000000000","message":"Patch Set 24:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/69/615469/24/openstack-check/tripleo-build-containers-fedora-28-master/38be861/ : FAILURE in 5m 11s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/69/615469/24/openstack-check/tripleo-build-containers-fedora-28-stein/11bbb00/ : SUCCESS in 30m 11s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"6f574c257d36046df3ade55ca6ee38d8b0ee4e6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 04:18:24.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/69/615469/24/check/test-release-openstack/a9b73f8/ : SUCCESS in 2m 12s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/24/check/kolla-build-centos-binary/5a64bfe/ : SUCCESS in 49m 28s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/24/check/kolla-build-centos-source/c228a52/ : SUCCESS in 1h 32m 09s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/24/check/kolla-ansible-centos-source/dc02047/ : TIMED_OUT in 2h 01m 13s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/24/check/kolla-ansible-centos-binary/1b77f2f/ : SUCCESS in 1h 17m 26s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/69/615469/24/check/kolla-ansible-centos-source-upgrade/a6f9f26/ : TIMED_OUT in 2h 01m 07s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/69/615469/24/check/tripleo-build-containers-centos-7/c6ed669/ : SUCCESS in 41m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/24/check/openstack-tox-lower-constraints/3582f31/ : SUCCESS in 3m 44s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/24/check/openstack-tox-pep8/1805e75/ : SUCCESS in 3m 18s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/24/check/openstack-tox-py27/ca369a7/ : SUCCESS in 3m 22s\n- openstack-tox-py36 http://logs.openstack.org/69/615469/24/check/openstack-tox-py36/fd11403/ : SUCCESS in 3m 18s\n- openstack-tox-py37 http://logs.openstack.org/69/615469/24/check/openstack-tox-py37/56985c0/ : SUCCESS in 3m 43s\n- openstack-tox-docs http://logs.openstack.org/69/615469/24/check/openstack-tox-docs/ee5ecf7/html/ : SUCCESS in 2m 48s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/24/check/build-openstack-releasenotes/2d3acf3/html/ : SUCCESS in 4m 07s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/24/check/kolla-build-ubuntu-binary/cf79ee6/ : SUCCESS in 24m 32s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/24/check/kolla-build-ubuntu-source/56274a2/ : SUCCESS in 30m 23s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/24/check/kolla-ansible-ubuntu-source/205e858/ : SUCCESS in 49m 16s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/24/check/kolla-ansible-ubuntu-binary/e21ac62/ : SUCCESS in 42m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/69/615469/24/check/kolla-ansible-ubuntu-source-upgrade/23dab7e/ : SUCCESS in 56m 46s","accounts_in_message":[],"_revision_number":24},{"id":"f717de4b4be11cf8fb8beda49217ba455d939e4b","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-06-26 12:00:18.000000000","message":"Patch Set 24:\n\nrecheck","accounts_in_message":[],"_revision_number":24},{"id":"0dc77246595801de6e6550973b45c2f3a6f0b12d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-26 12:38:09.000000000","message":"Patch Set 24:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/69/615469/24/openstack-check/tripleo-build-containers-fedora-28-master/0092b9b/ : FAILURE in 6m 30s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/69/615469/24/openstack-check/tripleo-build-containers-fedora-28-stein/3fbc564/ : SUCCESS in 36m 08s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"89780c063e3605a14b10fe8ebbaa80e04234ce5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 14:43:07.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/69/615469/24/check/test-release-openstack/4632cdb/ : SUCCESS in 2m 00s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/24/check/kolla-build-centos-binary/f81a45b/ : SUCCESS in 40m 50s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/24/check/kolla-build-centos-source/9442773/ : SUCCESS in 2h 35m 53s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/24/check/kolla-ansible-centos-source/cbf7f22/ : TIMED_OUT in 2h 01m 04s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/24/check/kolla-ansible-centos-binary/3cc384c/ : SUCCESS in 1h 17m 13s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/69/615469/24/check/kolla-ansible-centos-source-upgrade/8789e06/ : TIMED_OUT in 2h 00m 56s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/69/615469/24/check/tripleo-build-containers-centos-7/fe2931d/ : SUCCESS in 53m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/24/check/openstack-tox-lower-constraints/3e932dd/ : SUCCESS in 3m 26s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/24/check/openstack-tox-pep8/d4a7a28/ : SUCCESS in 3m 11s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/24/check/openstack-tox-py27/8e5c054/ : SUCCESS in 3m 02s\n- openstack-tox-py36 http://logs.openstack.org/69/615469/24/check/openstack-tox-py36/ff8ed62/ : SUCCESS in 3m 22s\n- openstack-tox-py37 http://logs.openstack.org/69/615469/24/check/openstack-tox-py37/46b7d6e/ : SUCCESS in 3m 13s\n- openstack-tox-docs http://logs.openstack.org/69/615469/24/check/openstack-tox-docs/6a182ea/html/ : SUCCESS in 2m 50s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/24/check/build-openstack-releasenotes/1144276/html/ : SUCCESS in 4m 19s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/24/check/kolla-build-ubuntu-binary/301ff8c/ : SUCCESS in 24m 03s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/24/check/kolla-build-ubuntu-source/68f28a2/ : SUCCESS in 26m 23s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/24/check/kolla-ansible-ubuntu-source/7939fde/ : SUCCESS in 48m 16s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/24/check/kolla-ansible-ubuntu-binary/1bfa899/ : SUCCESS in 47m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/69/615469/24/check/kolla-ansible-ubuntu-source-upgrade/372d706/ : SUCCESS in 1h 19m 58s","accounts_in_message":[],"_revision_number":24},{"id":"9d474ed8904361fd84ee8460063f0fd2455fefe3","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-06-26 22:21:40.000000000","message":"Patch Set 24:\n\nretry","accounts_in_message":[],"_revision_number":24},{"id":"76346cd16684d806ce71b5816a846e22a8a91f51","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-06-27 01:14:10.000000000","message":"Patch Set 24:\n\nrecheck","accounts_in_message":[],"_revision_number":24},{"id":"5059064382c7aa39f5d6307f8821118f4b65987c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 01:47:26.000000000","message":"Patch Set 24:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/69/615469/24/openstack-check/tripleo-build-containers-fedora-28-master/0a28317/ : FAILURE in 5m 03s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/69/615469/24/openstack-check/tripleo-build-containers-fedora-28-stein/eeba03e/ : SUCCESS in 31m 16s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"055faba9fa365543fd537917a6dc255fb17a6f9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 02:45:04.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/69/615469/24/check/test-release-openstack/05b980d/ : SUCCESS in 2m 44s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/24/check/kolla-build-centos-binary/f8efa45/ : SUCCESS in 30m 18s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/24/check/kolla-build-centos-source/95edddc/ : SUCCESS in 50m 01s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/24/check/kolla-ansible-centos-source/d5cb9b2/ : SUCCESS in 1h 21m 55s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/24/check/kolla-ansible-centos-binary/8ff88e8/ : SUCCESS in 1h 09m 38s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/69/615469/24/check/kolla-ansible-centos-source-upgrade/7340278/ : SUCCESS in 1h 25m 06s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/69/615469/24/check/tripleo-build-containers-centos-7/f2c7531/ : SUCCESS in 46m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/24/check/openstack-tox-lower-constraints/a73840f/ : SUCCESS in 3m 27s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/24/check/openstack-tox-pep8/a7017f3/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/24/check/openstack-tox-py27/6d2dc0e/ : SUCCESS in 3m 05s\n- openstack-tox-py36 http://logs.openstack.org/69/615469/24/check/openstack-tox-py36/b6b10a8/ : SUCCESS in 3m 16s\n- openstack-tox-py37 http://logs.openstack.org/69/615469/24/check/openstack-tox-py37/7cff72a/ : SUCCESS in 3m 34s\n- openstack-tox-docs http://logs.openstack.org/69/615469/24/check/openstack-tox-docs/72e639d/html/ : SUCCESS in 2m 50s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/24/check/build-openstack-releasenotes/232996a/html/ : SUCCESS in 3m 51s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/24/check/kolla-build-ubuntu-binary/7f3ccf8/ : SUCCESS in 24m 16s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/24/check/kolla-build-ubuntu-source/eb76357/ : SUCCESS in 28m 04s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/24/check/kolla-ansible-ubuntu-source/4e523ce/ : SUCCESS in 46m 20s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/24/check/kolla-ansible-ubuntu-binary/bbf007c/ : SUCCESS in 49m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/69/615469/24/check/kolla-ansible-ubuntu-source-upgrade/b273c3c/ : SUCCESS in 53m 09s","accounts_in_message":[],"_revision_number":24},{"id":"9ce995128fad06a2e68a1e3597d0817a0f6bbb20","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-06-27 02:59:52.000000000","message":"Uploaded patch set 25: Commit message was updated.","accounts_in_message":[],"_revision_number":25},{"id":"5dfa3b3cbc0a5c6ec9708b1796987bbacc8d1b70","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 03:41:00.000000000","message":"Patch Set 25:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/69/615469/25/openstack-check/tripleo-build-containers-fedora-28-master/216688c/ : FAILURE in 5m 21s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/69/615469/25/openstack-check/tripleo-build-containers-fedora-28-stein/375c39b/ : SUCCESS in 33m 36s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"883645faadd7d44e08e54ac3fb412f464e114f32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 04:23:01.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/69/615469/25/check/test-release-openstack/c67cf0f/ : SUCCESS in 2m 37s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/25/check/kolla-build-centos-binary/1a0c72b/ : SUCCESS in 33m 35s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/25/check/kolla-build-centos-source/c6121a0/ : SUCCESS in 50m 15s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/25/check/kolla-ansible-centos-source/fd254f9/ : SUCCESS in 1h 08m 00s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/25/check/kolla-ansible-centos-binary/63d75a9/ : SUCCESS in 1h 01m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/69/615469/25/check/kolla-ansible-centos-source-upgrade/b668b3d/ : SUCCESS in 1h 16m 21s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/69/615469/25/check/tripleo-build-containers-centos-7/e475fa8/ : SUCCESS in 47m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/25/check/openstack-tox-lower-constraints/1fcc945/ : SUCCESS in 4m 26s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/25/check/openstack-tox-pep8/e84f8c0/ : SUCCESS in 4m 04s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/25/check/openstack-tox-py27/05a24e1/ : SUCCESS in 4m 33s\n- openstack-tox-py36 http://logs.openstack.org/69/615469/25/check/openstack-tox-py36/d247744/ : SUCCESS in 4m 22s\n- openstack-tox-py37 http://logs.openstack.org/69/615469/25/check/openstack-tox-py37/c9bcbdb/ : SUCCESS in 3m 55s\n- openstack-tox-docs http://logs.openstack.org/69/615469/25/check/openstack-tox-docs/8dc2dce/html/ : SUCCESS in 3m 50s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/25/check/build-openstack-releasenotes/cfc9ddd/html/ : SUCCESS in 4m 46s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/25/check/kolla-build-ubuntu-binary/4a25e87/ : SUCCESS in 27m 17s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/25/check/kolla-build-ubuntu-source/c82dca9/ : SUCCESS in 30m 36s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/25/check/kolla-ansible-ubuntu-source/9a07a82/ : SUCCESS in 44m 49s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/25/check/kolla-ansible-ubuntu-binary/ac683f5/ : SUCCESS in 45m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/69/615469/25/check/kolla-ansible-ubuntu-source-upgrade/9a0cd82/ : SUCCESS in 53m 01s","accounts_in_message":[],"_revision_number":25},{"id":"dcb85b122ee8f2bcfcbddff3747a979a86ae9b8f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-27 07:10:24.000000000","message":"Patch Set 25:\n\nWould like to see the kolla-ansible patch and CI test passing before merging this.","accounts_in_message":[],"_revision_number":25},{"id":"62bb30f39dda94a049bf6cba463c03bc8f7b1877","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-06-27 15:32:46.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"1c9da716a09614e2b2d63d783b5ec7528cf5a407","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 16:09:34.000000000","message":"Patch Set 26:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/69/615469/26/openstack-check/tripleo-build-containers-fedora-28-master/c7f0f7a/ : FAILURE in 5m 37s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/69/615469/26/openstack-check/tripleo-build-containers-fedora-28-stein/7477b19/ : SUCCESS in 35m 14s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"63ba2ec5bc82967b4d7f99952d6e2b514dccdfed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 17:38:24.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/69/615469/26/check/test-release-openstack/d45ae5f/ : SUCCESS in 3m 36s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/26/check/kolla-build-centos-binary/4eb6618/ : SUCCESS in 30m 40s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/26/check/kolla-build-centos-source/a3263eb/ : SUCCESS in 43m 03s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/26/check/kolla-ansible-centos-source/cddb83c/ : SUCCESS in 1h 13m 01s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/26/check/kolla-ansible-centos-binary/4829139/ : SUCCESS in 1h 13m 16s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/69/615469/26/check/kolla-ansible-centos-source-upgrade/e086a89/ : SUCCESS in 1h 12m 41s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/69/615469/26/check/tripleo-build-containers-centos-7/aca9b26/ : SUCCESS in 44m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/26/check/openstack-tox-lower-constraints/938093b/ : SUCCESS in 5m 03s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/26/check/openstack-tox-pep8/6254d8d/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/26/check/openstack-tox-py27/fdb16cf/ : SUCCESS in 4m 14s\n- openstack-tox-py36 http://logs.openstack.org/69/615469/26/check/openstack-tox-py36/0b5fb52/ : SUCCESS in 4m 23s\n- openstack-tox-py37 http://logs.openstack.org/69/615469/26/check/openstack-tox-py37/8e79437/ : SUCCESS in 3m 30s\n- openstack-tox-docs http://logs.openstack.org/69/615469/26/check/openstack-tox-docs/177d4e8/html/ : SUCCESS in 2m 43s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/26/check/build-openstack-releasenotes/bff6c66/html/ : SUCCESS in 3m 59s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/26/check/kolla-build-ubuntu-binary/a4c59a6/ : SUCCESS in 25m 34s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/26/check/kolla-build-ubuntu-source/33601fa/ : SUCCESS in 37m 21s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/26/check/kolla-ansible-ubuntu-source/680a4d7/ : SUCCESS in 49m 08s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/26/check/kolla-ansible-ubuntu-binary/523775f/ : SUCCESS in 43m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/69/615469/26/check/kolla-ansible-ubuntu-source-upgrade/9f6915b/ : SUCCESS in 1h 00m 36s","accounts_in_message":[],"_revision_number":26},{"id":"6389d52b00fd9fdfccde914d29163993dd9a8407","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-06-27 22:06:09.000000000","message":"Patch Set 26: Code-Review-1","accounts_in_message":[],"_revision_number":26},{"id":"f61cfae1d04f36b868d7b0b576a09f841c6a4cc6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-28 06:36:53.000000000","message":"Patch Set 26: Workflow-1\n\nworkflow -1 is better marker for WIP patch","accounts_in_message":[],"_revision_number":26},{"id":"20a093fd106129b31467255d189e962bebdee03d","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-07-02 16:24:58.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"2966d347064125f6de268fa518e5983fd3063099","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-02 17:11:39.000000000","message":"Patch Set 27:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/69/615469/27/openstack-check/tripleo-build-containers-fedora-28-master/6621b21/ : FAILURE in 6m 31s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/69/615469/27/openstack-check/tripleo-build-containers-fedora-28-stein/22e012e/ : SUCCESS in 45m 47s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"49fe4a4192af2bff19c04c3810b85acca728d055","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 17:56:16.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/69/615469/27/check/test-release-openstack/f998e5d/ : SUCCESS in 2m 09s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/27/check/kolla-build-centos-binary/7d8f3cc/ : SUCCESS in 43m 39s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/27/check/kolla-build-centos-source/c79f073/ : SUCCESS in 31m 08s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/27/check/kolla-ansible-centos-source/b5bc01f/ : SUCCESS in 1h 17m 20s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/27/check/kolla-ansible-centos-binary/f50cb4a/ : SUCCESS in 1h 04m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/69/615469/27/check/kolla-ansible-centos-source-upgrade/fe7bb3b/ : SUCCESS in 1h 25m 54s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/69/615469/27/check/tripleo-build-containers-centos-7/6cb5801/ : SUCCESS in 48m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/27/check/openstack-tox-lower-constraints/d43710c/ : SUCCESS in 4m 11s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/27/check/openstack-tox-pep8/e5c5b54/ : SUCCESS in 3m 57s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/27/check/openstack-tox-py27/9988121/ : SUCCESS in 3m 02s\n- openstack-tox-py36 http://logs.openstack.org/69/615469/27/check/openstack-tox-py36/0ecc6ae/ : SUCCESS in 3m 26s\n- openstack-tox-py37 http://logs.openstack.org/69/615469/27/check/openstack-tox-py37/c5ec5c0/ : SUCCESS in 5m 55s\n- openstack-tox-docs http://logs.openstack.org/69/615469/27/check/openstack-tox-docs/53a08f1/html/ : SUCCESS in 3m 37s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/27/check/build-openstack-releasenotes/1e6fe87/html/ : SUCCESS in 4m 14s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/27/check/kolla-build-ubuntu-binary/afe04f9/ : FAILURE in 30m 33s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/27/check/kolla-build-ubuntu-source/d346a9b/ : SUCCESS in 26m 12s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/27/check/kolla-ansible-ubuntu-source/fb43fe1/ : SUCCESS in 46m 04s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/27/check/kolla-ansible-ubuntu-binary/3afaa00/ : SUCCESS in 47m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/69/615469/27/check/kolla-ansible-ubuntu-source-upgrade/1bbff83/ : SUCCESS in 53m 31s","accounts_in_message":[],"_revision_number":27},{"id":"511cea456238f95ebc831e71063724a8cbec8d7d","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-07-07 05:17:50.000000000","message":"Patch Set 27: Code-Review+1","accounts_in_message":[],"_revision_number":27},{"id":"76f46f04a8d9abc673d529c57e35358520bbf430","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-07-11 22:54:51.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"6c1f1e46fc680d8bf65e9502ff6409b7f0271379","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-07-11 22:56:57.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"d7e1f5b8196275acfe68c22454cb3795ab5851fb","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-07-11 23:05:28.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"3b54d0daaa6a0b30b4d37fb75ad949082fe5f649","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-11 23:50:07.000000000","message":"Patch Set 30:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/69/615469/30/openstack-check/tripleo-build-containers-fedora-28-master/e1d68d5/ : FAILURE in 6m 04s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/69/615469/30/openstack-check/tripleo-build-containers-fedora-28-stein/2e04e18/ : SUCCESS in 42m 05s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"24dc057094577bb2a4da619efbd0b9761452fc67","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-07-12 00:37:54.000000000","message":"Patch Set 30: Code-Review+1","accounts_in_message":[],"_revision_number":30},{"id":"b6c905031854be8bf474450e869dc34e8bad4933","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 00:40:57.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/69/615469/30/check/test-release-openstack/24a0272/ : SUCCESS in 1m 46s\n- kolla-build-centos-binary http://logs.openstack.org/69/615469/30/check/kolla-build-centos-binary/9e42de4/ : SUCCESS in 47m 29s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/615469/30/check/kolla-build-centos-source/03c02b0/ : SUCCESS in 48m 10s\n- kolla-ansible-centos-source http://logs.openstack.org/69/615469/30/check/kolla-ansible-centos-source/0e9cb99/ : SUCCESS in 1h 20m 15s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/615469/30/check/kolla-ansible-centos-binary/fd9ed6e/ : SUCCESS in 1h 16m 11s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/69/615469/30/check/kolla-ansible-centos-source-upgrade/2057cd1/ : SUCCESS in 1h 25m 01s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/69/615469/30/check/tripleo-build-containers-centos-7/350f7d4/ : SUCCESS in 46m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/30/check/openstack-tox-lower-constraints/de8782e/ : SUCCESS in 3m 38s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/30/check/openstack-tox-pep8/637dc8d/ : SUCCESS in 3m 02s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/30/check/openstack-tox-py27/2c423a6/ : SUCCESS in 3m 45s\n- openstack-tox-py36 http://logs.openstack.org/69/615469/30/check/openstack-tox-py36/82ecc6a/ : SUCCESS in 3m 45s\n- openstack-tox-py37 http://logs.openstack.org/69/615469/30/check/openstack-tox-py37/b9e8949/ : SUCCESS in 4m 19s\n- openstack-tox-docs http://logs.openstack.org/69/615469/30/check/openstack-tox-docs/c95293f/html/ : SUCCESS in 2m 40s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/30/check/build-openstack-releasenotes/5baf4a6/html/ : SUCCESS in 4m 14s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/615469/30/check/kolla-build-ubuntu-binary/33da285/ : SUCCESS in 26m 31s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/30/check/kolla-build-ubuntu-source/6141784/ : SUCCESS in 29m 29s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/615469/30/check/kolla-ansible-ubuntu-source/e01e22a/ : SUCCESS in 1h 16m 55s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/615469/30/check/kolla-ansible-ubuntu-binary/df549fd/ : SUCCESS in 1h 12m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/69/615469/30/check/kolla-ansible-ubuntu-source-upgrade/a542e0f/ : SUCCESS in 1h 04m 30s","accounts_in_message":[],"_revision_number":30},{"id":"4fc28e497960c402c1596f8d8233a71625c710b2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-12 11:32:47.000000000","message":"Patch Set 30: Code-Review+2","accounts_in_message":[],"_revision_number":30},{"id":"7c45dcb28910bf8c8881af670d606dade961d8d5","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-07-13 15:55:42.000000000","message":"Patch Set 30: Code-Review+1","accounts_in_message":[],"_revision_number":30},{"id":"bcb6ce920d8ab5e92b24088dc15f3a7c19c6d76b","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-07-13 18:36:22.000000000","message":"Patch Set 30: Code-Review+1","accounts_in_message":[],"_revision_number":30},{"id":"13914b56db4acf27a955d3ad2ffdd6e0a494323c","author":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"date":"2019-07-14 15:27:44.000000000","message":"Patch Set 30: Code-Review+1","accounts_in_message":[],"_revision_number":30},{"id":"6a7d74528e3a46ec38809db1fb2b0f2aa7efc0d0","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-07-14 23:50:29.000000000","message":"Patch Set 30: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":30},{"id":"525105befcf18c0670ce6fd59e278f18e8740bb5","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-07-15 01:58:02.000000000","message":"Patch Set 30: Code-Review+2 Workflow+1\n\nlgtm, thanks","accounts_in_message":[],"_revision_number":30},{"id":"89c473a34397a24bc37cd34de6019186fa78c2c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 01:58:15.000000000","message":"Patch Set 30: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":30},{"id":"96c4a269cc4eb1e404a81198167e5bad2d9e031c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 03:06:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":30},{"id":"9742cc39a5fa8252c720a96d4295c7e3d0e5f446","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 03:06:10.000000000","message":"Patch Set 30: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/69/615469/30/gate/test-release-openstack/1050177/ : SUCCESS in 2m 05s\n- kolla-build-centos-source http://logs.openstack.org/69/615469/30/gate/kolla-build-centos-source/92d08a8/ : SUCCESS in 56m 27s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/69/615469/30/gate/tripleo-build-containers-centos-7/700d115/ : SUCCESS in 46m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/615469/30/gate/openstack-tox-lower-constraints/78eaa54/ : SUCCESS in 3m 49s\n- openstack-tox-pep8 http://logs.openstack.org/69/615469/30/gate/openstack-tox-pep8/b609956/ : SUCCESS in 3m 31s\n- openstack-tox-py27 http://logs.openstack.org/69/615469/30/gate/openstack-tox-py27/6b5508f/ : SUCCESS in 3m 29s\n- openstack-tox-py36 http://logs.openstack.org/69/615469/30/gate/openstack-tox-py36/e9db514/ : SUCCESS in 3m 36s\n- openstack-tox-py37 http://logs.openstack.org/69/615469/30/gate/openstack-tox-py37/819f3f3/ : SUCCESS in 3m 25s\n- openstack-tox-docs http://logs.openstack.org/69/615469/30/gate/openstack-tox-docs/564d448/html/ : SUCCESS in 3m 02s\n- build-openstack-releasenotes http://logs.openstack.org/69/615469/30/gate/build-openstack-releasenotes/a92ba81/html/ : SUCCESS in 4m 24s\n- kolla-build-ubuntu-source http://logs.openstack.org/69/615469/30/gate/kolla-build-ubuntu-source/0b5cc64/ : SUCCESS in 29m 23s","accounts_in_message":[],"_revision_number":30}],"current_revision_number":30,"current_revision":"ce74eb0d7b4c01e1fcdb432e5605835caac6c1d0","revisions":{"32134100a1bdddd64027422194c0fb80650c6051":{"kind":"REWORK","_number":1,"created":"2018-11-05 04:24:06.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/69/615469/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/1"}}},"commit":{"parents":[{"commit":"8aac00972d6b3d5f5635b217a2151f9072e3f75d","subject":"Copy vitrage app.wsgi when deploy install_type binary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8aac00972d6b3d5f5635b217a2151f9072e3f75d"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:25:43.000000000","tz":420},"subject":"Add Masakari containers","message":"Add Masakari containers\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32134100a1bdddd64027422194c0fb80650c6051"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32134100a1bdddd64027422194c0fb80650c6051"}]},"branch":"refs/heads/master"},"3c9512b2aa8fb67173b7babc98e939a8a6daa59f":{"kind":"REWORK","_number":2,"created":"2018-11-05 07:27:21.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/69/615469/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/2"}}},"commit":{"parents":[{"commit":"8aac00972d6b3d5f5635b217a2151f9072e3f75d","subject":"Copy vitrage app.wsgi when deploy install_type binary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8aac00972d6b3d5f5635b217a2151f9072e3f75d"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 07:29:25.000000000","tz":420},"subject":"Add Masakari containers","message":"Add Masakari containers\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c9512b2aa8fb67173b7babc98e939a8a6daa59f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c9512b2aa8fb67173b7babc98e939a8a6daa59f"}]},"branch":"refs/heads/master"},"dd99eb99da85235757a0d503cc2164549faa9cb4":{"kind":"REWORK","_number":3,"created":"2018-11-05 07:29:53.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/69/615469/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/3"}}},"commit":{"parents":[{"commit":"8aac00972d6b3d5f5635b217a2151f9072e3f75d","subject":"Copy vitrage app.wsgi when deploy install_type binary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8aac00972d6b3d5f5635b217a2151f9072e3f75d"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 07:31:56.000000000","tz":420},"subject":"Add Masakari containers","message":"Add Masakari containers\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd99eb99da85235757a0d503cc2164549faa9cb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd99eb99da85235757a0d503cc2164549faa9cb4"}]},"branch":"refs/heads/master"},"3962a54f750028d0229e15b6a073f5e9ab976357":{"kind":"REWORK","_number":4,"created":"2018-11-05 08:11:40.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/69/615469/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/4"}}},"commit":{"parents":[{"commit":"8aac00972d6b3d5f5635b217a2151f9072e3f75d","subject":"Copy vitrage app.wsgi when deploy install_type binary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8aac00972d6b3d5f5635b217a2151f9072e3f75d"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 08:13:39.000000000","tz":420},"subject":"Add Masakari containers","message":"Add Masakari containers\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3962a54f750028d0229e15b6a073f5e9ab976357"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3962a54f750028d0229e15b6a073f5e9ab976357"}]},"branch":"refs/heads/master"},"5d8315785d46a0fb5a9ca8f22e4f1b34ee1d65b0":{"kind":"REWORK","_number":5,"created":"2018-11-05 08:30:48.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/69/615469/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/5"}}},"commit":{"parents":[{"commit":"8aac00972d6b3d5f5635b217a2151f9072e3f75d","subject":"Copy vitrage app.wsgi when deploy install_type binary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8aac00972d6b3d5f5635b217a2151f9072e3f75d"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 08:32:49.000000000","tz":420},"subject":"Add Masakari containers","message":"Add Masakari containers\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d8315785d46a0fb5a9ca8f22e4f1b34ee1d65b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d8315785d46a0fb5a9ca8f22e4f1b34ee1d65b0"}]},"branch":"refs/heads/master"},"93747f9d29b1e96e485fe7a45a22c6e77c391107":{"kind":"REWORK","_number":6,"created":"2018-11-07 01:33:00.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/69/615469/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/6"}}},"commit":{"parents":[{"commit":"cce6157c60d518712aa90586ef4ec0bd0659b5ca","subject":"Merge \"Remove path work around for kibana\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cce6157c60d518712aa90586ef4ec0bd0659b5ca"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-07 01:35:04.000000000","tz":420},"subject":"Add Masakari containers","message":"Add Masakari containers\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93747f9d29b1e96e485fe7a45a22c6e77c391107"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93747f9d29b1e96e485fe7a45a22c6e77c391107"}]},"branch":"refs/heads/master"},"078ad2c5c889e8567557a3b6063ef2ed6c36e0ea":{"kind":"REWORK","_number":7,"created":"2018-11-07 02:10:22.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/69/615469/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/7"}}},"commit":{"parents":[{"commit":"cce6157c60d518712aa90586ef4ec0bd0659b5ca","subject":"Merge \"Remove path work around for kibana\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cce6157c60d518712aa90586ef4ec0bd0659b5ca"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-07 02:12:28.000000000","tz":420},"subject":"Add Masakari containers","message":"Add Masakari containers\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/078ad2c5c889e8567557a3b6063ef2ed6c36e0ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/078ad2c5c889e8567557a3b6063ef2ed6c36e0ea"}]},"branch":"refs/heads/master"},"4ac9b45e3c0b7fb941d03c208374239506d31335":{"kind":"REWORK","_number":8,"created":"2018-11-07 11:11:20.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/69/615469/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/8"}}},"commit":{"parents":[{"commit":"cce6157c60d518712aa90586ef4ec0bd0659b5ca","subject":"Merge \"Remove path work around for kibana\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cce6157c60d518712aa90586ef4ec0bd0659b5ca"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-07 11:13:24.000000000","tz":420},"subject":"Add Masakari containers","message":"Add Masakari containers\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ac9b45e3c0b7fb941d03c208374239506d31335"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ac9b45e3c0b7fb941d03c208374239506d31335"}]},"branch":"refs/heads/master"},"0015605aff608e515bb7abd0541bf5653a38d6f4":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":9,"created":"2018-11-12 02:13:09.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/69/615469/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/9"}}},"commit":{"parents":[{"commit":"f16e410542db39472a7672b4cb9b09883db94930","subject":"Merge \"utils: be Python 2 compatible when squashing is enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f16e410542db39472a7672b4cb9b09883db94930"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-12 02:15:15.000000000","tz":420},"subject":"Add Masakari container images","message":"Add Masakari container images\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0015605aff608e515bb7abd0541bf5653a38d6f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0015605aff608e515bb7abd0541bf5653a38d6f4"}]},"branch":"refs/heads/master"},"163e331f4daddf05869b85b7ed0be05c233588b6":{"kind":"REWORK","_number":10,"created":"2018-11-13 01:49:05.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/69/615469/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/10"}}},"commit":{"parents":[{"commit":"f16e410542db39472a7672b4cb9b09883db94930","subject":"Merge \"utils: be Python 2 compatible when squashing is enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f16e410542db39472a7672b4cb9b09883db94930"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-13 01:51:14.000000000","tz":420},"subject":"Add Masakari container images","message":"Add Masakari container images\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/163e331f4daddf05869b85b7ed0be05c233588b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/163e331f4daddf05869b85b7ed0be05c233588b6"}]},"branch":"refs/heads/master"},"36667a311c574317de2ac583147b47e067899f5f":{"kind":"REWORK","_number":11,"created":"2018-11-13 07:08:14.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/69/615469/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/11"}}},"commit":{"parents":[{"commit":"f16e410542db39472a7672b4cb9b09883db94930","subject":"Merge \"utils: be Python 2 compatible when squashing is enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f16e410542db39472a7672b4cb9b09883db94930"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-13 07:10:18.000000000","tz":420},"subject":"Add Masakari container images","message":"Add Masakari container images\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36667a311c574317de2ac583147b47e067899f5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36667a311c574317de2ac583147b47e067899f5f"}]},"branch":"refs/heads/master"},"ea145e801ca68c6481248ffef9bb5d23fcfa51d8":{"kind":"REWORK","_number":12,"created":"2018-11-17 01:29:23.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/69/615469/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/12"}}},"commit":{"parents":[{"commit":"1affeaa2e223705433f1d1a0e23c6af98f9e2bc3","subject":"Merge \"Replacing the HTTP protocal with HTTPS .\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1affeaa2e223705433f1d1a0e23c6af98f9e2bc3"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-17 01:31:33.000000000","tz":420},"subject":"Add Masakari container images","message":"Add Masakari container images\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea145e801ca68c6481248ffef9bb5d23fcfa51d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea145e801ca68c6481248ffef9bb5d23fcfa51d8"}]},"branch":"refs/heads/master"},"23fc7e7db0e7756112ecdf64aff056184b24a3b0":{"kind":"REWORK","_number":13,"created":"2018-11-18 13:52:03.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/69/615469/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/13"}}},"commit":{"parents":[{"commit":"1affeaa2e223705433f1d1a0e23c6af98f9e2bc3","subject":"Merge \"Replacing the HTTP protocal with HTTPS .\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1affeaa2e223705433f1d1a0e23c6af98f9e2bc3"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-18 13:51:46.000000000","tz":420},"subject":"Add Masakari container images","message":"Add Masakari container images\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23fc7e7db0e7756112ecdf64aff056184b24a3b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23fc7e7db0e7756112ecdf64aff056184b24a3b0"}]},"branch":"refs/heads/master"},"0eefea44d1cd52f76b02f3bd35316736c52d43a9":{"kind":"REWORK","_number":14,"created":"2018-11-20 11:40:01.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/69/615469/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/14"}}},"commit":{"parents":[{"commit":"1affeaa2e223705433f1d1a0e23c6af98f9e2bc3","subject":"Merge \"Replacing the HTTP protocal with HTTPS .\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1affeaa2e223705433f1d1a0e23c6af98f9e2bc3"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-20 11:42:10.000000000","tz":420},"subject":"Add Masakari container images","message":"Add Masakari container images\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eefea44d1cd52f76b02f3bd35316736c52d43a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eefea44d1cd52f76b02f3bd35316736c52d43a9"}]},"branch":"refs/heads/master"},"1e85d38428bf571953d6ebb647ad642c8cae16de":{"kind":"REWORK","_number":15,"created":"2018-11-22 07:49:01.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/69/615469/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/15"}}},"commit":{"parents":[{"commit":"b2f8605f3b78a8f7229fa00d92f971028371a16d","subject":"Fix kafka build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2f8605f3b78a8f7229fa00d92f971028371a16d"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-22 07:51:15.000000000","tz":420},"subject":"Add Masakari container images","message":"Add Masakari container images\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e85d38428bf571953d6ebb647ad642c8cae16de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e85d38428bf571953d6ebb647ad642c8cae16de"}]},"branch":"refs/heads/master"},"83faf78a670c401e4bf693e110a6974ff332c4ca":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2018-12-11 02:17:17.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/69/615469/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/16"}}},"commit":{"parents":[{"commit":"9bc2472705d31704ffcabe09bcc1e38447a5be6b","subject":"Merge \"Update Zaqar image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bc2472705d31704ffcabe09bcc1e38447a5be6b"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-12-11 02:17:17.000000000","tz":0},"subject":"Add Masakari container images","message":"Add Masakari container images\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83faf78a670c401e4bf693e110a6974ff332c4ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83faf78a670c401e4bf693e110a6974ff332c4ca"}]},"branch":"refs/heads/master"},"a998d111569d793f33fcb5d94799ae1f03732b02":{"kind":"REWORK","_number":17,"created":"2018-12-14 01:31:52.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/69/615469/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/17"}}},"commit":{"parents":[{"commit":"38881349080d627a81b9594df36c8f054463c2b1","subject":"Deprecate crane image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38881349080d627a81b9594df36c8f054463c2b1"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-12-14 01:34:31.000000000","tz":420},"subject":"Add Masakari container images","message":"Add Masakari container images\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a998d111569d793f33fcb5d94799ae1f03732b02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a998d111569d793f33fcb5d94799ae1f03732b02"}]},"branch":"refs/heads/master"},"8dddef93e44ee88cc62aa4a26cf68537b0815e53":{"kind":"REWORK","_number":18,"created":"2018-12-19 10:12:11.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/69/615469/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/18"}}},"commit":{"parents":[{"commit":"0f41906089824ed75c445a42fa2a3413d2f26ec1","subject":"Merge \"Install numactl package for Zun images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f41906089824ed75c445a42fa2a3413d2f26ec1"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-12-19 10:12:33.000000000","tz":420},"subject":"Add Masakari container images","message":"Add Masakari container images\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8dddef93e44ee88cc62aa4a26cf68537b0815e53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8dddef93e44ee88cc62aa4a26cf68537b0815e53"}]},"branch":"refs/heads/master"},"c2dc5e536f2c6d366228157496ddc060d731d5ce":{"kind":"REWORK","_number":19,"created":"2018-12-20 01:20:16.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/69/615469/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/19"}}},"commit":{"parents":[{"commit":"3285073aff58de5be787ac46f930dcd56417d217","subject":"Merge \"Split placement-api from nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3285073aff58de5be787ac46f930dcd56417d217"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-12-20 01:22:55.000000000","tz":420},"subject":"Add Masakari container images","message":"Add Masakari container images\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c2dc5e536f2c6d366228157496ddc060d731d5ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c2dc5e536f2c6d366228157496ddc060d731d5ce"}]},"branch":"refs/heads/master"},"a8f9cf120137e9301503bf5f702a2daf207d0c1f":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2019-03-04 07:14:04.000000000","uploader":{"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},"ref":"refs/changes/69/615469/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/20"}}},"commit":{"parents":[{"commit":"8fcd7ea9123840be5fdf382fefe9811086b724b7","subject":"Merge \"Switch to stestr\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8fcd7ea9123840be5fdf382fefe9811086b724b7"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Taeha Kim","email":"kthguru@gmail.com","date":"2019-03-04 07:14:04.000000000","tz":0},"subject":"Add Masakari container images","message":"Add Masakari container images\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8f9cf120137e9301503bf5f702a2daf207d0c1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8f9cf120137e9301503bf5f702a2daf207d0c1f"}]},"branch":"refs/heads/master"},"c0b106d92e2f384cc8e81186ec4fd74d31b36864":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2019-03-10 15:39:55.000000000","uploader":{"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},"ref":"refs/changes/69/615469/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/21"}}},"commit":{"parents":[{"commit":"30e1e8d8a3e73756e0558cefa1c931bb748f8e36","subject":"Merge \"Fix Octavia-API extend_start.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30e1e8d8a3e73756e0558cefa1c931bb748f8e36"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Taeha Kim","email":"kthguru@gmail.com","date":"2019-03-10 15:39:55.000000000","tz":0},"subject":"Add Masakari container images","message":"Add Masakari container images\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0b106d92e2f384cc8e81186ec4fd74d31b36864"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0b106d92e2f384cc8e81186ec4fd74d31b36864"}]},"branch":"refs/heads/master"},"cf6dbad50675855a7ca8afb4c6c608a4fe412891":{"kind":"REWORK","_number":22,"created":"2019-06-25 18:33:03.000000000","uploader":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"ref":"refs/changes/69/615469/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/22"}}},"commit":{"parents":[{"commit":"2777a0c0584e4c707c56ca6aaa8cf2f41a473a83","subject":"Merge \"Make SKIPPED_IMAGES more flexible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2777a0c0584e4c707c56ca6aaa8cf2f41a473a83"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Gaëtan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-06-25 18:33:02.000000000","tz":-240},"subject":"Add Masakari container images","message":"Add Masakari container images\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf6dbad50675855a7ca8afb4c6c608a4fe412891"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf6dbad50675855a7ca8afb4c6c608a4fe412891"}]},"branch":"refs/heads/master"},"287da6b33e9953c642f2aa721afacc1d857b27e2":{"kind":"REWORK","_number":23,"created":"2019-06-25 19:11:19.000000000","uploader":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"ref":"refs/changes/69/615469/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/23"}}},"commit":{"parents":[{"commit":"2777a0c0584e4c707c56ca6aaa8cf2f41a473a83","subject":"Merge \"Make SKIPPED_IMAGES more flexible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2777a0c0584e4c707c56ca6aaa8cf2f41a473a83"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Gaëtan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-06-25 19:11:12.000000000","tz":-240},"subject":"WIP: Add Masakari containers","message":"WIP: Add Masakari containers\n\nMasakari provides Instances High Availability Service for\nOpenStack clouds by automatically recovering failed Instances.\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/287da6b33e9953c642f2aa721afacc1d857b27e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/287da6b33e9953c642f2aa721afacc1d857b27e2"}]},"branch":"refs/heads/master"},"7c60a91e5b7b34d5d6847729f23b59b946553833":{"kind":"REWORK","_number":24,"created":"2019-06-25 19:13:43.000000000","uploader":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"ref":"refs/changes/69/615469/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/24"}}},"commit":{"parents":[{"commit":"2777a0c0584e4c707c56ca6aaa8cf2f41a473a83","subject":"Merge \"Make SKIPPED_IMAGES more flexible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2777a0c0584e4c707c56ca6aaa8cf2f41a473a83"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Gaëtan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-06-25 19:13:38.000000000","tz":-240},"subject":"WIP: Add Masakari containers","message":"WIP: Add Masakari containers\n\nMasakari provides Instances High Availability Service for\nOpenStack clouds by automatically recovering failed Instances.\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c60a91e5b7b34d5d6847729f23b59b946553833"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c60a91e5b7b34d5d6847729f23b59b946553833"}]},"branch":"refs/heads/master"},"aadea8b951b7220664d190683e107cbf5222951b":{"kind":"NO_CODE_CHANGE","_number":25,"created":"2019-06-27 02:59:52.000000000","uploader":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"ref":"refs/changes/69/615469/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/25"}}},"commit":{"parents":[{"commit":"2777a0c0584e4c707c56ca6aaa8cf2f41a473a83","subject":"Merge \"Make SKIPPED_IMAGES more flexible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2777a0c0584e4c707c56ca6aaa8cf2f41a473a83"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Gaëtan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-06-27 02:59:41.000000000","tz":-240},"subject":"Add Masakari containers","message":"Add Masakari containers\n\nMasakari provides Instances High Availability Service for\nOpenStack clouds by automatically recovering failed Instances.\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aadea8b951b7220664d190683e107cbf5222951b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aadea8b951b7220664d190683e107cbf5222951b"}]},"branch":"refs/heads/master"},"3c02095602ca683cd92b6b15c8f1c512b67a2633":{"kind":"REWORK","_number":26,"created":"2019-06-27 15:32:46.000000000","uploader":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"ref":"refs/changes/69/615469/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/26"}}},"commit":{"parents":[{"commit":"2777a0c0584e4c707c56ca6aaa8cf2f41a473a83","subject":"Merge \"Make SKIPPED_IMAGES more flexible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2777a0c0584e4c707c56ca6aaa8cf2f41a473a83"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Gaëtan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-06-27 15:32:38.000000000","tz":-240},"subject":"Add Masakari containers","message":"Add Masakari containers\n\nMasakari provides Instances High Availability Service for\nOpenStack clouds by automatically recovering failed Instances.\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c02095602ca683cd92b6b15c8f1c512b67a2633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c02095602ca683cd92b6b15c8f1c512b67a2633"}]},"branch":"refs/heads/master"},"2860fed161dfd6654a3a0b4aec64006cbf21b938":{"kind":"REWORK","_number":27,"created":"2019-07-02 16:24:58.000000000","uploader":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"ref":"refs/changes/69/615469/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/27"}}},"commit":{"parents":[{"commit":"2777a0c0584e4c707c56ca6aaa8cf2f41a473a83","subject":"Merge \"Make SKIPPED_IMAGES more flexible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2777a0c0584e4c707c56ca6aaa8cf2f41a473a83"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Gaëtan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-07-02 16:24:34.000000000","tz":-240},"subject":"Add Masakari containers","message":"Add Masakari containers\n\nMasakari provides Instances High Availability Service for\nOpenStack clouds by automatically recovering failed Instances.\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2860fed161dfd6654a3a0b4aec64006cbf21b938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2860fed161dfd6654a3a0b4aec64006cbf21b938"}]},"branch":"refs/heads/master"},"993154c0aeea2d64509156e081aa2c369e6bac60":{"kind":"REWORK","_number":28,"created":"2019-07-11 22:54:51.000000000","uploader":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"ref":"refs/changes/69/615469/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/28"}}},"commit":{"parents":[{"commit":"008aa957f27b26caff3be8b18110e71cdaf70772","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/008aa957f27b26caff3be8b18110e71cdaf70772"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Gaëtan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-07-11 22:54:43.000000000","tz":-240},"subject":"Add Masakari containers","message":"Add Masakari containers\n\nMasakari provides Instances High Availability Service for\nOpenStack clouds by automatically recovering failed Instances.\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/993154c0aeea2d64509156e081aa2c369e6bac60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/993154c0aeea2d64509156e081aa2c369e6bac60"}]},"branch":"refs/heads/master"},"94cbb45b9078330023c3db5488904c18aab542e6":{"kind":"REWORK","_number":29,"created":"2019-07-11 22:56:57.000000000","uploader":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"ref":"refs/changes/69/615469/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/29"}}},"commit":{"parents":[{"commit":"008aa957f27b26caff3be8b18110e71cdaf70772","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/008aa957f27b26caff3be8b18110e71cdaf70772"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Gaëtan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-07-11 22:56:41.000000000","tz":-240},"subject":"Add Masakari containers","message":"Add Masakari containers\n\nMasakari provides Instances High Availability Service for\nOpenStack clouds by automatically recovering failed Instances.\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/94cbb45b9078330023c3db5488904c18aab542e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/94cbb45b9078330023c3db5488904c18aab542e6"}]},"branch":"refs/heads/master"},"ce74eb0d7b4c01e1fcdb432e5605835caac6c1d0":{"kind":"REWORK","_number":30,"created":"2019-07-11 23:05:28.000000000","uploader":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"ref":"refs/changes/69/615469/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/615469/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/615469/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/615469/30"}}},"commit":{"parents":[{"commit":"008aa957f27b26caff3be8b18110e71cdaf70772","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/008aa957f27b26caff3be8b18110e71cdaf70772"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-11-05 04:24:55.000000000","tz":420},"committer":{"name":"Gaëtan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-07-11 23:05:13.000000000","tz":-240},"subject":"Add Masakari containers","message":"Add Masakari containers\n\nMasakari provides Instances High Availability Service for\nOpenStack clouds by automatically recovering failed Instances.\n\nChange-Id: I279d6b43f8e4d4e30b39deb607823ceca272a4b8\nImplements: blueprint masakari-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce74eb0d7b4c01e1fcdb432e5605835caac6c1d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce74eb0d7b4c01e1fcdb432e5605835caac6c1d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
