)]}'
{"id":"openstack%2Fkolla~624838","triplet_id":"openstack%2Fkolla~master~Ia6bb2c1de260e099cf44f47b190be032007ac8b1","project":"openstack/kolla","branch":"master","topic":"bp/python3-support","hashtags":[],"change_id":"Ia6bb2c1de260e099cf44f47b190be032007ac8b1","subject":"Fedora support for python3 package testing","status":"ABANDONED","created":"2018-12-13 00:16:54.000000000","updated":"2019-09-20 21:04:03.000000000","total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ef257b821d1c0497b5bc47a19c6e5a3efcb82ad0","_number":624838,"virtual_id_number":624838,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-07-15 16:48:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"date":"2019-07-15 14:56:24.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":-2,"date":"2019-07-15 14:08:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"date":"2019-07-15 14:08:13.000000000","_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-12-13 02:45:56.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2018-12-17 18:47:29.000000000","updated_by":{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},"reviewer":{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},"state":"REVIEWER"},{"updated":"2019-01-09 12:46:03.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2019-01-09 14:40:37.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-01-10 18:06:31.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2019-01-22 15:41:17.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2019-02-26 16:52:35.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2019-02-27 08:46:46.000000000","updated_by":{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},"reviewer":{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},"state":"REVIEWER"},{"updated":"2019-02-28 06:47:50.000000000","updated_by":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"reviewer":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"state":"REVIEWER"},{"updated":"2019-03-06 13:59:47.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2019-03-10 15:38:50.000000000","updated_by":{"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},"reviewer":{"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},"state":"REVIEWER"},{"updated":"2019-06-06 08:13:10.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-06 09:20:32.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"REVIEWER"},{"updated":"2019-06-18 19:21:46.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-06-19 08:16:44.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-15 14:08:13.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-07-15 14:56:24.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-15 16:48:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"626b0afd36df5328c05d0edbe292bfb134b75784","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2018-12-13 00:16:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0af308420a4beadfb13dd379ba7350acb269bf14","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2018-12-13 00:17:12.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"ab067089783dbd182198071e6dee97722fc1e66f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-13 02:38:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/1/check/kolla-build-centos-binary/e4cad5e/ : FAILURE in 1h 02m 43s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/1/check/kolla-build-centos-source/054a0b6/ : FAILURE in 1h 18m 40s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/1/check/kolla-ansible-centos-source/c220f55/ : FAILURE in 46m 39s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/1/check/kolla-ansible-centos-binary/d5c20cf/ : SUCCESS in 1h 26m 41s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/1/check/tripleo-build-containers-centos-7/9b6c6e8/ : FAILURE in 41m 59s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/1/check/kolla-build-oraclelinux-binary/673e72b/ : FAILURE in 1h 03m 43s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/1/check/kolla-build-oraclelinux-source/7fea8bb/ : FAILURE in 1h 00m 27s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/1/check/kolla-ansible-oraclelinux-source/6aa661c/ : FAILURE in 54m 04s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/1/check/kolla-ansible-oraclelinux-binary/97b990e/ : TIMED_OUT in 2h 01m 28s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/38/624838/1/check/openstack-tox-py36/4f25c98/ : SUCCESS in 3m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/1/check/openstack-tox-lower-constraints/b5765c9/ : SUCCESS in 3m 27s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/1/check/openstack-tox-pep8/3db2479/ : FAILURE in 2m 49s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/1/check/openstack-tox-py27/7bb7ba7/ : SUCCESS in 3m 00s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/1/check/openstack-tox-py35/69a07c8/ : SUCCESS in 3m 16s\n- openstack-tox-docs http://logs.openstack.org/38/624838/1/check/openstack-tox-docs/3af5934/html/ : SUCCESS in 2m 50s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/1/check/kolla-build-ubuntu-binary/9276f7e/ : SUCCESS in 1h 32m 20s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/1/check/kolla-build-ubuntu-source/4293796/ : SUCCESS in 39m 54s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/1/check/kolla-ansible-ubuntu-source/c8b632e/ : FAILURE in 45m 14s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/1/check/kolla-ansible-ubuntu-binary/c4f5f31/ : TIMED_OUT in 2h 00m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2df2a8f1da31e398450fb6af2c6d6407c2a8cc5c","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2018-12-13 02:45:56.000000000","message":"Patch Set 1:\n\nI noticed Zuul failed, I think you hit bug(s):\n\n- kolla-ansible-centos-source: https://bugs.launchpad.net/bugs/1708704\n- kolla-ansible-ubuntu-source: unrecognized error\n- kolla-build-centos-source: unrecognized error\n- kolla-build-oraclelinux-source: unrecognized error\n- openstack-tox-pep8: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":1},{"id":"5e3389f0dc082f511e0f5f53b620bafe2a8c9c59","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-08 22:03:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"99240b71e3d544ee569dbbc9cb0d30f11baf9fac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-08 22:52:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/2/check/kolla-build-centos-binary/3fdee4a/ : FAILURE in 13m 59s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/2/check/kolla-build-centos-source/841e3ac/ : FAILURE in 13m 56s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/2/check/kolla-ansible-centos-source/0639b51/ : FAILURE in 10m 10s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/2/check/kolla-ansible-centos-binary/d75c577/ : FAILURE in 9m 56s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/2/check/tripleo-build-containers-centos-7/80c708a/ : FAILURE in 43m 00s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/2/check/kolla-ansible-bifrost-centos-source/aa8f437/ : FAILURE in 10m 25s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/2/check/kolla-build-oraclelinux-binary/d1b297e/ : FAILURE in 13m 16s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/2/check/kolla-build-oraclelinux-source/a376d53/ : FAILURE in 12m 06s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/2/check/kolla-ansible-oraclelinux-source/1acca18/ : FAILURE in 9m 59s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/2/check/kolla-ansible-oraclelinux-binary/f175e42/ : FAILURE in 10m 52s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/2/check/openstack-tox-lower-constraints/023518a/ : SUCCESS in 4m 13s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/2/check/openstack-tox-py36/49ae0b2/ : SUCCESS in 3m 29s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/2/check/openstack-tox-pep8/718c4d2/ : FAILURE in 3m 28s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/2/check/openstack-tox-py27/2731458/ : SUCCESS in 3m 03s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/2/check/openstack-tox-py35/5fee699/ : SUCCESS in 3m 34s\n- openstack-tox-docs http://logs.openstack.org/38/624838/2/check/openstack-tox-docs/5cecaa1/html/ : SUCCESS in 3m 31s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/2/check/kolla-build-ubuntu-binary/9d0ce0d/ : FAILURE in 5m 18s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/2/check/kolla-build-ubuntu-source/f509368/ : FAILURE in 5m 13s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/2/check/kolla-ansible-ubuntu-source/267755c/ : FAILURE in 7m 28s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/2/check/kolla-ansible-ubuntu-binary/8dabc62/ : FAILURE in 8m 04s (non-voting)\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/38/624838/2/check/kolla-ansible-ubuntu-source-zun/b2b93ef/ : FAILURE in 6m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f1144ba5377982468902bdcb0513efe77b1ae76a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-01-09 14:40:37.000000000","message":"Patch Set 2:\n\nrecheck packagecloud.io down","accounts_in_message":[],"_revision_number":2},{"id":"3ebef4cf167ad4e26b26e7ef2442e00165c2845c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-09 15:00:37.000000000","message":"Patch Set 2:\n\n\u003e recheck packagecloud.io down\n\nThe grafana packages were moved from packagecloud to packages.grafana.com. This should now be fixed in master, see https://review.openstack.org/#/c/629380/.","accounts_in_message":[],"_revision_number":2},{"id":"636641a9a0036d98c0bc3a2addeaa96ab1a9b62a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-09 16:30:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/2/check/kolla-build-centos-binary/e30033a/ : FAILURE in 52m 32s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/2/check/kolla-build-centos-source/298d67c/ : FAILURE in 1h 00m 16s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/2/check/kolla-ansible-centos-source/686699e/ : SUCCESS in 1h 30m 50s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/2/check/kolla-ansible-centos-binary/b32812c/ : SUCCESS in 1h 06m 59s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/2/check/tripleo-build-containers-centos-7/0307f89/ : FAILURE in 29m 42s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/2/check/kolla-ansible-bifrost-centos-source/0fb8b7f/ : FAILURE in 33m 43s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/2/check/kolla-build-oraclelinux-binary/1719d23/ : FAILURE in 50m 20s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/2/check/kolla-build-oraclelinux-source/18daf20/ : FAILURE in 1h 13m 42s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/2/check/kolla-ansible-oraclelinux-source/9a2b9d2/ : SUCCESS in 1h 20m 45s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/2/check/kolla-ansible-oraclelinux-binary/7c8aa0f/ : SUCCESS in 1h 10m 25s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/2/check/openstack-tox-lower-constraints/6f4f574/ : SUCCESS in 4m 01s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/2/check/openstack-tox-py36/f94678a/ : SUCCESS in 2m 52s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/2/check/openstack-tox-pep8/848f3bc/ : FAILURE in 3m 00s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/2/check/openstack-tox-py27/4697dac/ : SUCCESS in 3m 18s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/2/check/openstack-tox-py35/4fafd01/ : SUCCESS in 4m 03s\n- openstack-tox-docs http://logs.openstack.org/38/624838/2/check/openstack-tox-docs/5fef96a/html/ : SUCCESS in 3m 02s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/2/check/kolla-build-ubuntu-binary/892a188/ : SUCCESS in 1h 27m 00s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/2/check/kolla-build-ubuntu-source/837ce8f/ : SUCCESS in 34m 06s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/2/check/kolla-ansible-ubuntu-source/cc03930/ : SUCCESS in 1h 01m 27s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/2/check/kolla-ansible-ubuntu-binary/0061f3e/ : SUCCESS in 1h 43m 46s (non-voting)\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/38/624838/2/check/kolla-ansible-ubuntu-source-zun/40fe43a/ : SUCCESS in 1h 08m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9c02169055111cf627f4e1790e950143b0707150","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-09 23:00:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e72ff8bee6e4679af19cd38e7b6f3e2acc09680a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-10 01:03:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/3/check/kolla-build-centos-binary/8a12554/ : SUCCESS in 54m 28s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/3/check/kolla-build-centos-source/82af8e1/ : SUCCESS in 1h 06m 20s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/3/check/kolla-ansible-centos-source/8c1cf14/ : SUCCESS in 1h 17m 39s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/3/check/kolla-ansible-centos-binary/2a7c5dd/ : SUCCESS in 1h 10m 43s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/3/check/tripleo-build-containers-centos-7/a94d98c/ : SUCCESS in 36m 38s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/3/check/kolla-ansible-bifrost-centos-source/89a9841/ : FAILURE in 31m 08s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/3/check/kolla-build-oraclelinux-binary/3a8617d/ : SUCCESS in 1h 14m 25s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/3/check/kolla-build-oraclelinux-source/4f589c5/ : SUCCESS in 1h 00m 39s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/3/check/kolla-ansible-oraclelinux-source/45df015/ : SUCCESS in 1h 21m 30s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/3/check/kolla-ansible-oraclelinux-binary/cc0b3f3/ : SUCCESS in 1h 25m 20s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/3/check/openstack-tox-lower-constraints/35b50ea/ : SUCCESS in 3m 37s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/3/check/openstack-tox-py36/f57271e/ : SUCCESS in 5m 08s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/3/check/openstack-tox-pep8/2914a11/ : SUCCESS in 3m 40s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/3/check/openstack-tox-py27/51a4959/ : SUCCESS in 3m 39s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/3/check/openstack-tox-py35/f41d49b/ : SUCCESS in 3m 46s\n- openstack-tox-docs http://logs.openstack.org/38/624838/3/check/openstack-tox-docs/4b77c2c/html/ : SUCCESS in 2m 59s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/3/check/kolla-build-ubuntu-binary/74983eb/ : FAILURE in 1h 28m 23s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/3/check/kolla-build-ubuntu-source/a4010f9/ : SUCCESS in 35m 32s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/3/check/kolla-ansible-ubuntu-source/c425eb8/ : SUCCESS in 56m 57s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/3/check/kolla-ansible-ubuntu-binary/557e79a/ : TIMED_OUT in 2h 01m 22s (non-voting)\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/38/624838/3/check/kolla-ansible-ubuntu-source-zun/a7f9fe2/ : SUCCESS in 1h 12m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2d4da4969818b7652237cb93cb12b06975e97b29","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-14 15:27:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3cfa7831cffc9c81994418a1c8776bd0e715e976","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-01-14 16:45:38.000000000","message":"Patch Set 4:\n\nCould you alphabetise check for rpm distros? \"if base_distro in [centos, fedora, oraclelinux, rhel]\"\n\nLatest released Fedora is 29 not 28.\n\n\"if distro_version in [28,8]\" as python3 check? Please move it to Kolla\u0027s python code and export it as \u0027rpm_distro_python3\u003d1\u0027 or something? Then update of Fedora release will be done in one place instead of plenty. And add comment that 28 means Fedora while 8 means future release of CentOS/RHEL (and probably oraclelinux too).","accounts_in_message":[],"_revision_number":4},{"id":"21916c688cced5c2120d9c2807a0e93829fe37f6","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-14 16:49:33.000000000","message":"Patch Set 4:\n\nI\u0027m using 28 because currently we are working on getting the python3 openstack items in RDO on a fedora 28 base. This is primarily a WIP as we work to understand what works and what doesn\u0027t. It\u0027s likely that we\u0027ll just drop the fedora bits when it\u0027s ready for merging","accounts_in_message":[],"_revision_number":4},{"id":"ff2b3fb9a8c76ad6bf1e8f60b881c20582d6f8d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-14 17:23:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/4/check/kolla-build-centos-binary/8cc1181/ : SUCCESS in 1h 12m 33s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/4/check/kolla-build-centos-source/9d334c7/ : SUCCESS in 1h 50m 56s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/4/check/kolla-ansible-centos-source/97b798a/ : SUCCESS in 1h 20m 45s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/4/check/kolla-ansible-centos-binary/170ddd0/ : SUCCESS in 1h 04m 58s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/4/check/tripleo-build-containers-centos-7/82fa2f0/ : SUCCESS in 36m 05s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/4/check/kolla-ansible-bifrost-centos-source/b04f407/ : FAILURE in 38m 07s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/4/check/kolla-build-oraclelinux-binary/f0d2ad7/ : SUCCESS in 1h 27m 11s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/4/check/kolla-build-oraclelinux-source/a6b93b2/ : SUCCESS in 1h 50m 41s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/4/check/kolla-ansible-oraclelinux-source/45d83c4/ : SUCCESS in 1h 13m 36s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/4/check/kolla-ansible-oraclelinux-binary/ebc1d0a/ : SUCCESS in 1h 04m 34s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/4/check/openstack-tox-lower-constraints/25ae284/ : SUCCESS in 4m 42s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/4/check/openstack-tox-py36/56280fe/ : SUCCESS in 3m 47s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/4/check/openstack-tox-pep8/56d5238/ : SUCCESS in 4m 00s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/4/check/openstack-tox-py27/bbb2180/ : SUCCESS in 4m 32s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/4/check/openstack-tox-py35/2a088fe/ : SUCCESS in 4m 00s\n- openstack-tox-docs http://logs.openstack.org/38/624838/4/check/openstack-tox-docs/c445507/html/ : SUCCESS in 3m 35s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/4/check/kolla-build-ubuntu-binary/7dae589/ : FAILURE in 1h 18m 42s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/4/check/kolla-build-ubuntu-source/779e10b/ : SUCCESS in 49m 53s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/4/check/kolla-ansible-ubuntu-source/739e9a4/ : SUCCESS in 1h 00m 33s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/4/check/kolla-ansible-ubuntu-binary/71d1d47/ : SUCCESS in 1h 46m 45s (non-voting)\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/38/624838/4/check/kolla-ansible-ubuntu-source-zun/a24aa7d/ : SUCCESS in 1h 04m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8aa414f5b5ca8dd051bdbd325ffc5a0419ff615e","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-01-15 14:41:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4d67c9c0db1009b55c19d11442de4e53736ed006","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-15 23:39:37.000000000","message":"Topic changed from rhel-py3 to bp/python3-support","accounts_in_message":[],"_revision_number":4},{"id":"98a53b2f76b0b61ea0e5d26f6779ed38f584b29b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-16 00:16:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"43cdaa55cf325c6e0f56ec7f3d781a3030a8c69b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-16 00:23:13.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bf3a6ef7e0479c3c9f3e4c9c97111f29049c6096","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-16 00:53:28.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/6/check/kolla-build-centos-binary/6fa765e/ : FAILURE in 18m 31s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/6/check/kolla-build-centos-source/65eaf40/ : FAILURE in 13m 58s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/6/check/kolla-ansible-centos-source/30ef560/ : FAILURE in 13m 37s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/6/check/kolla-ansible-centos-binary/be12a0d/ : FAILURE in 12m 44s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/6/check/tripleo-build-containers-centos-7/8730651/ : FAILURE in 14m 29s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/6/check/kolla-ansible-bifrost-centos-source/4605bd9/ : FAILURE in 10m 53s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/6/check/kolla-build-oraclelinux-binary/581a085/ : FAILURE in 14m 12s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/6/check/kolla-build-oraclelinux-source/f164d31/ : FAILURE in 17m 08s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/6/check/kolla-ansible-oraclelinux-source/7910f40/ : FAILURE in 10m 28s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/6/check/kolla-ansible-oraclelinux-binary/b552663/ : FAILURE in 8m 04s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/6/check/openstack-tox-lower-constraints/237c2c6/ : SUCCESS in 2m 48s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/6/check/openstack-tox-py36/e1f28ec/ : SUCCESS in 2m 27s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/6/check/openstack-tox-pep8/9a8fd0a/ : SUCCESS in 3m 42s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/6/check/openstack-tox-py27/4a4c83b/ : SUCCESS in 3m 06s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/6/check/openstack-tox-py35/ab10203/ : SUCCESS in 2m 41s\n- openstack-tox-docs http://logs.openstack.org/38/624838/6/check/openstack-tox-docs/99b247a/html/ : SUCCESS in 3m 35s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/6/check/kolla-build-ubuntu-binary/7f41547/ : FAILURE in 5m 03s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/6/check/kolla-build-ubuntu-source/cb3aae4/ : FAILURE in 6m 04s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/6/check/kolla-ansible-ubuntu-source/7caa4ec/ : FAILURE in 7m 31s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/6/check/kolla-ansible-ubuntu-binary/d618458/ : FAILURE in 6m 10s (non-voting)\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/38/624838/6/check/kolla-ansible-ubuntu-source-zun/d4864d9/ : FAILURE in 5m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"adcbacc3f4dd6547f79a4a052cf67b1100abbcb3","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-01-16 10:10:01.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cd8faa8a52c2109a119d8eab835525e745ce7cef","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-01-16 10:10:25.000000000","message":"Patch Set 6:\n\nmay be that just me doing kolla wrong, sorry for that then, my -1 is just informational","accounts_in_message":[],"_revision_number":6},{"id":"d5dcd9281ce5ec2029dd166619daaef76d2d2c64","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-01-16 10:21:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b153d4eb7e025573c1e644725767b1e4b217002a","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-01-16 15:37:17.000000000","message":"Patch Set 6:\n\nlocal testing for the topic shows the templating is broken after https://review.openstack.org/#/c/624838/4 :(","accounts_in_message":[],"_revision_number":6},{"id":"4e3a9e7ed8708fcbad5bd6f439207e90f1e0bbb5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-16 15:48:33.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"49a1e531a75109ee37aae972c50dbe8dd665ed77","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-16 15:51:13.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b7cb879be26479d9f0b9a435dc698936203b35cb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-16 15:51:18.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6479bcd1d4825731b4d56d581cf3ec36f4a2637a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-16 16:14:36.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/8/check/kolla-build-centos-binary/cf6405a/ : FAILURE in 14m 16s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/8/check/kolla-build-centos-source/ab5a7c9/ : FAILURE in 10m 33s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/8/check/kolla-ansible-centos-source/3ec936d/ : FAILURE in 12m 04s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/8/check/kolla-ansible-centos-binary/8f9c8c1/ : FAILURE in 12m 40s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/8/check/tripleo-build-containers-centos-7/eeed7f1/ : FAILURE in 13m 59s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/8/check/kolla-ansible-bifrost-centos-source/c1407c1/ : FAILURE in 10m 56s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/8/check/kolla-build-oraclelinux-binary/db46a4f/ : FAILURE in 7m 47s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/8/check/kolla-build-oraclelinux-source/a3b94f0/ : FAILURE in 9m 13s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/8/check/kolla-ansible-oraclelinux-source/d40eaf3/ : FAILURE in 10m 44s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/8/check/kolla-ansible-oraclelinux-binary/e3d0efc/ : FAILURE in 11m 18s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/8/check/openstack-tox-lower-constraints/724c6b1/ : FAILURE in 6m 26s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/8/check/openstack-tox-py36/570e0f8/ : FAILURE in 3m 34s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/8/check/openstack-tox-pep8/e1b19c9/ : SUCCESS in 3m 41s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/8/check/openstack-tox-py27/8004a58/ : FAILURE in 3m 27s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/8/check/openstack-tox-py35/86cc789/ : FAILURE in 3m 41s\n- openstack-tox-docs http://logs.openstack.org/38/624838/8/check/openstack-tox-docs/896a164/html/ : SUCCESS in 3m 28s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/8/check/kolla-build-ubuntu-binary/1277ef6/ : FAILURE in 5m 19s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/8/check/kolla-build-ubuntu-source/fcd71aa/ : FAILURE in 4m 03s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/8/check/kolla-ansible-ubuntu-source/7e491fe/ : FAILURE in 6m 36s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/8/check/kolla-ansible-ubuntu-binary/44b64e2/ : FAILURE in 10m 48s (non-voting)\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/38/624838/8/check/kolla-ansible-ubuntu-source-zun/ef3ef16/ : FAILURE in 9m 09s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"dc99c76630957cba1d36733a073ca259bb60e7d9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-16 17:11:40.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2d365c67da32bf655d63dba9eee59414b04c4ce8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-16 17:29:37.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"0a122e625097ca3a8140e5eccb719df64d984027","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-16 17:58:09.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/10/check/kolla-build-centos-binary/edd06cc/ : FAILURE in 13m 56s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/10/check/kolla-build-centos-source/0d190fe/ : FAILURE in 9m 11s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/10/check/kolla-ansible-centos-source/a65b80e/ : FAILURE in 8m 16s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/10/check/kolla-ansible-centos-binary/ce65b59/ : FAILURE in 8m 17s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/10/check/tripleo-build-containers-centos-7/bf74ee8/ : FAILURE in 10m 11s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/10/check/kolla-ansible-bifrost-centos-source/5df3913/ : FAILURE in 8m 43s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/10/check/kolla-build-oraclelinux-binary/13d6915/ : FAILURE in 13m 34s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/10/check/kolla-build-oraclelinux-source/507f9ad/ : FAILURE in 18m 13s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/10/check/kolla-ansible-oraclelinux-source/aa3b6b8/ : FAILURE in 13m 07s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/10/check/kolla-ansible-oraclelinux-binary/ec0f5a8/ : FAILURE in 10m 50s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/10/check/openstack-tox-lower-constraints/692763a/ : SUCCESS in 5m 28s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/10/check/openstack-tox-py36/440a138/ : SUCCESS in 5m 32s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/10/check/openstack-tox-pep8/a259f8e/ : SUCCESS in 3m 57s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/10/check/openstack-tox-py27/9c4ef1e/ : SUCCESS in 3m 26s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/10/check/openstack-tox-py35/bc8d35e/ : SUCCESS in 3m 42s\n- openstack-tox-docs http://logs.openstack.org/38/624838/10/check/openstack-tox-docs/6f28538/html/ : SUCCESS in 2m 38s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/10/check/kolla-build-ubuntu-binary/a15dc98/ : FAILURE in 4m 18s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/10/check/kolla-build-ubuntu-source/e00d16f/ : FAILURE in 4m 40s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/10/check/kolla-ansible-ubuntu-source/891703f/ : FAILURE in 7m 09s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/10/check/kolla-ansible-ubuntu-binary/6d084d0/ : FAILURE in 11m 49s (non-voting)\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/38/624838/10/check/kolla-ansible-ubuntu-source-zun/31ac6fa/ : FAILURE in 6m 35s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"664965e721b222d727b3eb87f70955c63dd4dd57","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-16 18:01:37.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"6cb2ab10cae80713cbcf2a691b65628cafc50589","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-16 19:00:11.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"0e1efab13a9285ada9cd8be94aca307ae6d599b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-16 20:49:46.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/12/check/kolla-build-centos-binary/3db4736/ : SUCCESS in 1h 02m 32s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/12/check/kolla-build-centos-source/b742bd9/ : SUCCESS in 1h 11m 38s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/12/check/kolla-ansible-centos-source/b392d78/ : SUCCESS in 1h 14m 53s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/12/check/kolla-ansible-centos-binary/ed722be/ : SUCCESS in 1h 05m 40s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/12/check/tripleo-build-containers-centos-7/4fd7f15/ : SUCCESS in 34m 58s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/12/check/kolla-ansible-bifrost-centos-source/5e3ff26/ : FAILURE in 32m 01s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/12/check/kolla-build-oraclelinux-binary/0bf1036/ : SUCCESS in 1h 43m 57s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/12/check/kolla-build-oraclelinux-source/e1ebd22/ : SUCCESS in 1h 15m 33s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/12/check/kolla-ansible-oraclelinux-source/37bd374/ : SUCCESS in 1h 22m 13s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/12/check/kolla-ansible-oraclelinux-binary/3bce14e/ : SUCCESS in 1h 32m 42s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/12/check/openstack-tox-lower-constraints/e9a508f/ : SUCCESS in 3m 56s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/12/check/openstack-tox-py36/0ea5a38/ : SUCCESS in 4m 04s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/12/check/openstack-tox-pep8/c2f73fe/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/12/check/openstack-tox-py27/ed35e35/ : SUCCESS in 4m 01s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/12/check/openstack-tox-py35/61935b9/ : SUCCESS in 3m 51s\n- openstack-tox-docs http://logs.openstack.org/38/624838/12/check/openstack-tox-docs/4c14f8e/html/ : SUCCESS in 5m 17s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/12/check/kolla-build-ubuntu-binary/249d0f9/ : FAILURE in 1h 24m 15s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/12/check/kolla-build-ubuntu-source/ba2cea2/ : SUCCESS in 29m 38s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/12/check/kolla-ansible-ubuntu-source/be534a0/ : SUCCESS in 1h 11m 28s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/12/check/kolla-ansible-ubuntu-binary/42577bb/ : SUCCESS in 1h 44m 56s (non-voting)\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/38/624838/12/check/kolla-ansible-ubuntu-source-zun/097ad91/ : SUCCESS in 57m 24s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"5d7061c2c880d396fbbad7d6708cee364a02b494","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-01-17 12:34:34.000000000","message":"Patch Set 12: Code-Review+1\n\nthis looks much better for my local testing","accounts_in_message":[],"_revision_number":12},{"id":"058951acda5a75f4f196060dca70fdcdf4e66a67","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-01-17 12:55:00.000000000","message":"Patch Set 12:\n\nThough I have the following testing issues remaining https://pastebin.com/MVxtEjeh","accounts_in_message":[],"_revision_number":12},{"id":"e154a3fc13578ecb69b6ee3685faf1168d1487fc","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-01-17 13:01:44.000000000","message":"Patch Set 12: Code-Review-1\n\nI\u0027m not sure my config triggers it right:\n\ncat \u003e /tmp/kolla-build.conf\u003c\u003cEOF\n[DEFAULT]\nbase_package_type\u003drpm\ndistro_python3\u003dtrue\nrpm_setup_config\u003d...\n\nbut openstack-base also templates in python-crypto alongside of python3--crypto https://pastebin.com/NcKz0gfb","accounts_in_message":[],"_revision_number":12},{"id":"530f0ccf6f59bcd1d354323ab75fb2a0d8df71f6","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-01-17 13:26:28.000000000","message":"Patch Set 12:\n\nThat was a tripleo specific issue, sorry! (https://review.openstack.org/#/c/629679)\n\nwrt this change, there is python3-tooz -\u003e python3-zake -\u003e python-kazoo\nand the latter pulls in py2 deps","accounts_in_message":[],"_revision_number":12},{"id":"38b58f34b4b1ac6bce0845d23894daec9594da30","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-01-17 13:41:30.000000000","message":"Patch Set 12:\n\nas a w/a for testing, I removed python3-tooz. But that looks like a py3 readiness bug for the latter","accounts_in_message":[],"_revision_number":12},{"id":"1dae5a1852e2e1a71d5c77bceb49915483f6f0fe","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-17 17:58:59.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"136babd0f4113e4f6fcaf3bc7428405f1175db8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-17 18:27:13.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/13/check/kolla-build-centos-binary/77dbc7e/ : FAILURE in 11m 13s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/13/check/kolla-build-centos-source/f2ef726/ : FAILURE in 11m 45s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/13/check/kolla-ansible-centos-source/b8b5c94/ : FAILURE in 7m 36s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/13/check/kolla-ansible-centos-binary/a33cdd9/ : FAILURE in 7m 50s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/13/check/tripleo-build-containers-centos-7/a9e7733/ : FAILURE in 15m 09s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/13/check/kolla-ansible-bifrost-centos-source/2896e73/ : FAILURE in 8m 38s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/13/check/kolla-build-oraclelinux-binary/ddea530/ : FAILURE in 15m 11s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/13/check/kolla-build-oraclelinux-source/abfe069/ : FAILURE in 13m 17s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/13/check/kolla-ansible-oraclelinux-source/e8901f4/ : FAILURE in 7m 47s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/13/check/kolla-ansible-oraclelinux-binary/0a75ec1/ : FAILURE in 7m 46s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/13/check/openstack-tox-lower-constraints/a113543/ : SUCCESS in 3m 31s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/13/check/openstack-tox-py36/a4de569/ : SUCCESS in 3m 14s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/13/check/openstack-tox-pep8/1df4357/ : SUCCESS in 4m 29s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/13/check/openstack-tox-py27/00de77c/ : SUCCESS in 3m 10s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/13/check/openstack-tox-py35/b9370c8/ : SUCCESS in 2m 59s\n- openstack-tox-docs http://logs.openstack.org/38/624838/13/check/openstack-tox-docs/cf26be5/html/ : SUCCESS in 2m 25s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/13/check/kolla-build-ubuntu-binary/264a891/ : FAILURE in 3m 24s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/13/check/kolla-build-ubuntu-source/a8261ea/ : FAILURE in 3m 47s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/13/check/kolla-ansible-ubuntu-source/cf8da07/ : FAILURE in 6m 09s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/13/check/kolla-ansible-ubuntu-binary/68cd880/ : FAILURE in 5m 58s (non-voting)\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/38/624838/13/check/kolla-ansible-ubuntu-source-zun/c3a1354/ : FAILURE in 6m 11s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"7e86e8269c0e1f27f9120de7782eb7c19539f97d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-17 18:42:24.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"9eb6d9d5d869349bfba7be5905d3e864ba0bf6f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-17 19:11:22.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/14/check/kolla-build-centos-binary/c4ff758/ : FAILURE in 12m 15s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/14/check/kolla-build-centos-source/89d7a58/ : FAILURE in 12m 27s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/14/check/kolla-ansible-centos-source/ed7847f/ : FAILURE in 7m 38s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/14/check/kolla-ansible-centos-binary/e09996a/ : FAILURE in 8m 02s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/14/check/tripleo-build-containers-centos-7/a6e066e/ : FAILURE in 14m 10s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/14/check/kolla-ansible-bifrost-centos-source/c6554f7/ : FAILURE in 7m 33s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/14/check/kolla-build-oraclelinux-binary/8f979a8/ : FAILURE in 10m 25s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/14/check/kolla-build-oraclelinux-source/926a41b/ : FAILURE in 10m 36s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/14/check/kolla-ansible-oraclelinux-source/8dddeb2/ : FAILURE in 7m 43s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/14/check/kolla-ansible-oraclelinux-binary/b1b6187/ : FAILURE in 7m 52s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/14/check/openstack-tox-lower-constraints/f9e3a83/ : SUCCESS in 4m 19s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/14/check/openstack-tox-py36/c5ec8e2/ : SUCCESS in 4m 36s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/14/check/openstack-tox-pep8/f5e222f/ : SUCCESS in 3m 39s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/14/check/openstack-tox-py27/4745a97/ : SUCCESS in 3m 28s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/14/check/openstack-tox-py35/a80809b/ : SUCCESS in 3m 26s\n- openstack-tox-docs http://logs.openstack.org/38/624838/14/check/openstack-tox-docs/fc50598/html/ : SUCCESS in 3m 25s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/14/check/kolla-build-ubuntu-binary/c8af825/ : FAILURE in 4m 28s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/14/check/kolla-build-ubuntu-source/5023daf/ : FAILURE in 3m 48s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/14/check/kolla-ansible-ubuntu-source/7078a85/ : FAILURE in 6m 14s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/14/check/kolla-ansible-ubuntu-binary/fdfd336/ : FAILURE in 6m 06s (non-voting)\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/38/624838/14/check/kolla-ansible-ubuntu-source-zun/01e682a/ : FAILURE in 6m 37s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"40a99e1d3e07e166ffd06783e0e744e3d8b9e8bc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-17 19:20:38.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"30cc8a87f5dc3e499212d6d314918fe9c27fe898","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-17 21:33:20.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/15/check/kolla-build-centos-binary/251d371/ : FAILURE in 1h 13m 05s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/15/check/kolla-build-centos-source/9fdcc9b/ : FAILURE in 10m 27s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/15/check/kolla-ansible-centos-source/5f78448/ : FAILURE in 7m 46s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/15/check/kolla-ansible-centos-binary/8c946b6/ : FAILURE in 23m 29s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/15/check/tripleo-build-containers-centos-7/d7d52a6/ : FAILURE in 47m 05s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/15/check/kolla-ansible-bifrost-centos-source/b83fe66/ : FAILURE in 7m 58s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/15/check/kolla-build-oraclelinux-binary/48ba65c/ : FAILURE in 1h 17m 06s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/15/check/kolla-build-oraclelinux-source/e1fc9fd/ : FAILURE in 12m 30s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/15/check/kolla-ansible-oraclelinux-source/734e457/ : FAILURE in 7m 31s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/15/check/kolla-ansible-oraclelinux-binary/68cfdd3/ : FAILURE in 30m 17s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/15/check/openstack-tox-lower-constraints/5c27101/ : SUCCESS in 3m 37s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/15/check/openstack-tox-py36/6f8b650/ : SUCCESS in 3m 05s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/15/check/openstack-tox-pep8/d6b6f47/ : SUCCESS in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/15/check/openstack-tox-py27/f43ce51/ : SUCCESS in 3m 46s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/15/check/openstack-tox-py35/81d1d7f/ : SUCCESS in 4m 05s\n- openstack-tox-docs http://logs.openstack.org/38/624838/15/check/openstack-tox-docs/d8bd05d/html/ : SUCCESS in 3m 10s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/15/check/kolla-build-ubuntu-binary/5348ae3/ : FAILURE in 54m 09s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/15/check/kolla-build-ubuntu-source/bd98735/ : SUCCESS in 35m 53s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/15/check/kolla-ansible-ubuntu-source/85bbda1/ : SUCCESS in 1h 07m 42s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/15/check/kolla-ansible-ubuntu-binary/da9a7a1/ : SUCCESS in 1h 59m 21s (non-voting)\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/38/624838/15/check/kolla-ansible-ubuntu-source-zun/acf8c5c/ : SUCCESS in 1h 11m 55s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"ac8fff88d2a91ef509ced674daacd0b0c0fd0cd8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-17 21:47:43.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"a986043c5e5c46aec9c9dc619b29bf16ee759383","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-17 23:39:59.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/16/check/kolla-build-centos-binary/413e738/ : FAILURE in 54m 17s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/16/check/kolla-build-centos-source/3dc3c61/ : FAILURE in 1h 08m 33s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/16/check/kolla-ansible-centos-source/f5701a9/ : SUCCESS in 1h 27m 06s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/16/check/kolla-ansible-centos-binary/5b6a240/ : SUCCESS in 1h 05m 59s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/16/check/tripleo-build-containers-centos-7/c3a699e/ : FAILURE in 45m 38s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/16/check/kolla-ansible-bifrost-centos-source/8ba71ea/ : SUCCESS in 42m 32s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/16/check/kolla-build-oraclelinux-binary/2a4f4f4/ : FAILURE in 1h 32m 35s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/16/check/kolla-build-oraclelinux-source/8b7ae99/ : FAILURE in 1h 21m 16s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/16/check/kolla-ansible-oraclelinux-source/b756e83/ : SUCCESS in 1h 23m 02s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/16/check/kolla-ansible-oraclelinux-binary/d8d155a/ : SUCCESS in 1h 11m 03s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/16/check/openstack-tox-lower-constraints/666672c/ : SUCCESS in 4m 07s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/16/check/openstack-tox-py36/4a5593d/ : SUCCESS in 3m 45s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/16/check/openstack-tox-pep8/3988212/ : SUCCESS in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/16/check/openstack-tox-py27/77bde1e/ : SUCCESS in 3m 37s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/16/check/openstack-tox-py35/24e066d/ : SUCCESS in 3m 51s\n- openstack-tox-docs http://logs.openstack.org/38/624838/16/check/openstack-tox-docs/9a160fa/html/ : SUCCESS in 3m 21s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/16/check/kolla-build-ubuntu-binary/061d043/ : FAILURE in 1h 27m 58s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/16/check/kolla-build-ubuntu-source/eca396f/ : SUCCESS in 38m 06s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/16/check/kolla-ansible-ubuntu-source/f93f36f/ : SUCCESS in 1h 27m 02s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/16/check/kolla-ansible-ubuntu-binary/f5c7723/ : SUCCESS in 1h 44m 45s (non-voting)\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/38/624838/16/check/kolla-ansible-ubuntu-source-zun/27d7a6c/ : SUCCESS in 1h 10m 07s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"5c3592c95a8fe558db0191b1cd38d5d367930f48","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-18 00:56:20.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"2f183d650ea712d1c8d965013a62dcdfdacd6965","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-18 03:03:04.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/17/check/kolla-build-centos-binary/cd978e7/ : SUCCESS in 1h 05m 16s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/17/check/kolla-build-centos-source/5340759/ : SUCCESS in 1h 17m 28s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/17/check/kolla-ansible-centos-source/bb0ba48/ : SUCCESS in 1h 17m 59s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/17/check/kolla-ansible-centos-binary/e254a6c/ : SUCCESS in 1h 12m 14s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/17/check/tripleo-build-containers-centos-7/5765178/ : SUCCESS in 43m 51s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/17/check/kolla-ansible-bifrost-centos-source/6482916/ : FAILURE in 33m 09s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/17/check/kolla-build-oraclelinux-binary/385da54/ : SUCCESS in 1h 09m 37s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/17/check/kolla-build-oraclelinux-source/8eab30e/ : SUCCESS in 1h 25m 33s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/17/check/kolla-ansible-oraclelinux-source/92e219f/ : SUCCESS in 1h 20m 40s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/17/check/kolla-ansible-oraclelinux-binary/6a62b65/ : SUCCESS in 1h 43m 26s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/17/check/openstack-tox-lower-constraints/3d4a3fa/ : SUCCESS in 3m 15s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/17/check/openstack-tox-py36/7a40527/ : SUCCESS in 2m 58s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/17/check/openstack-tox-pep8/bcad314/ : SUCCESS in 3m 28s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/17/check/openstack-tox-py27/e530127/ : SUCCESS in 3m 32s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/17/check/openstack-tox-py35/03f2653/ : SUCCESS in 3m 16s\n- openstack-tox-docs http://logs.openstack.org/38/624838/17/check/openstack-tox-docs/974700c/html/ : SUCCESS in 2m 50s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/17/check/kolla-build-ubuntu-binary/8a52a7e/ : FAILURE in 1h 26m 05s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/17/check/kolla-build-ubuntu-source/9e6c8e2/ : SUCCESS in 28m 56s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/17/check/kolla-ansible-ubuntu-source/07d8478/ : SUCCESS in 1h 01m 33s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/17/check/kolla-ansible-ubuntu-binary/5ed3cfa/ : TIMED_OUT in 2h 01m 04s (non-voting)\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/38/624838/17/check/kolla-ansible-ubuntu-source-zun/3008496/ : SUCCESS in 1h 23m 00s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"6985a6511bd4c9c3cfe1de4b82cdf9b1176de487","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-01-21 11:53:37.000000000","message":"Patch Set 17:\n\nAny ideas for fixing  python3-tooz -\u003e python3-zake -\u003e python-kazoo dependency? The latter wants py2 libs in and fails","accounts_in_message":[],"_revision_number":17},{"id":"ed65599553b23831cd1f978d9cf6f6e2339d02cb","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-01-21 11:53:43.000000000","message":"Patch Set 17: Code-Review-1","accounts_in_message":[],"_revision_number":17},{"id":"f35815e6d7e5eadb50553c2f200b59e73ecbfb8a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-21 19:24:25.000000000","message":"Patch Set 17:\n\nthe zake/kazoo issue is fixed in packaging, https://src.fedoraproject.org/rpms/python-zake/c/d5dc21c517359f2a8a903fc5a3ede9fbe26db1d1?branch\u003dmaster","accounts_in_message":[],"_revision_number":17},{"id":"fb27145c997f394a947b32bf3f24d3a7f4bb0bc8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-21 20:02:42.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"fa5b3a0fb8f46447dcb001e0dcf92df1fc7374bd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-21 20:03:49.000000000","message":"Patch Set 18:\n\nI split the packages out from the fedora support bits since we can land the package names but maybe not the fedora support","accounts_in_message":[],"_revision_number":18},{"id":"7351493853fe7eac525920c2c24ace54045e5b78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-21 22:15:05.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/18/check/kolla-build-centos-binary/0601d55/ : FAILURE in 16m 28s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/18/check/kolla-build-centos-source/05ea075/ : SUCCESS in 1h 05m 05s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/18/check/kolla-ansible-centos-source/1396fed/ : SUCCESS in 1h 18m 42s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/18/check/kolla-ansible-centos-binary/3579e5f/ : FAILURE in 1h 43m 37s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/18/check/tripleo-build-containers-centos-7/53cb2d7/ : SUCCESS in 32m 24s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/18/check/kolla-ansible-bifrost-centos-source/99f7092/ : FAILURE in 45m 48s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/18/check/kolla-build-oraclelinux-binary/f006c44/ : SUCCESS in 1h 20m 29s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/18/check/kolla-build-oraclelinux-source/74d140f/ : SUCCESS in 1h 18m 40s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/18/check/kolla-ansible-oraclelinux-source/5f0523a/ : SUCCESS in 1h 26m 41s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/18/check/kolla-ansible-oraclelinux-binary/3b18899/ : FAILURE in 1h 02m 04s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/18/check/openstack-tox-lower-constraints/c786cb9/ : SUCCESS in 4m 00s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/18/check/openstack-tox-py36/755469b/ : SUCCESS in 3m 46s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/18/check/openstack-tox-pep8/d64ba8d/ : SUCCESS in 2m 55s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/18/check/openstack-tox-py27/2158d8c/ : SUCCESS in 4m 09s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/18/check/openstack-tox-py35/689c3a2/ : SUCCESS in 4m 13s\n- openstack-tox-docs http://logs.openstack.org/38/624838/18/check/openstack-tox-docs/bcc3f23/html/ : SUCCESS in 3m 52s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/18/check/kolla-build-ubuntu-binary/1a21ffb/ : FAILURE in 2h 04m 26s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/18/check/kolla-build-ubuntu-source/980b84f/ : SUCCESS in 32m 15s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/18/check/kolla-ansible-ubuntu-source/31a3e03/ : SUCCESS in 1h 11m 05s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/18/check/kolla-ansible-ubuntu-binary/2d720eb/ : SUCCESS in 1h 45m 13s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"509ced2594c183c746eac253389359a340e66ffb","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-01-22 15:41:17.000000000","message":"Patch Set 18: Code-Review+1\n\nthanks!","accounts_in_message":[],"_revision_number":18},{"id":"3597785ea1baa237a9ae68795092454ca36e50fd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-24 15:25:29.000000000","message":"Patch Set 19: Patch Set 18 was rebased","accounts_in_message":[],"_revision_number":19},{"id":"b090d35045dc99fc1b3f5aea8ec19f3714d87ca0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-24 17:04:25.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/19/check/kolla-build-centos-binary/fb04475/ : SUCCESS in 50m 45s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/19/check/kolla-build-centos-source/ac722bf/ : SUCCESS in 58m 17s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/19/check/kolla-ansible-centos-source/e8884f5/ : SUCCESS in 1h 17m 03s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/19/check/kolla-ansible-centos-binary/2180598/ : FAILURE in 1h 10m 31s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/19/check/tripleo-build-containers-centos-7/480b472/ : FAILURE in 42m 16s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/19/check/kolla-ansible-bifrost-centos-source/47e2c5f/ : FAILURE in 39m 30s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/19/check/kolla-build-oraclelinux-binary/8e6ab18/ : SUCCESS in 1h 07m 10s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/19/check/kolla-build-oraclelinux-source/98a97a0/ : SUCCESS in 1h 20m 12s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/19/check/kolla-ansible-oraclelinux-source/80659d4/ : SUCCESS in 1h 16m 59s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/19/check/kolla-ansible-oraclelinux-binary/0fa0169/ : FAILURE in 1h 10m 09s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/19/check/openstack-tox-lower-constraints/9b08b38/ : SUCCESS in 4m 38s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/19/check/openstack-tox-py36/ef4eb71/ : SUCCESS in 3m 54s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/19/check/openstack-tox-pep8/fa266a7/ : SUCCESS in 3m 55s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/19/check/openstack-tox-py27/0f33b52/ : SUCCESS in 4m 17s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/19/check/openstack-tox-py35/a2f687c/ : SUCCESS in 4m 17s\n- openstack-tox-docs http://logs.openstack.org/38/624838/19/check/openstack-tox-docs/d4bfe10/html/ : SUCCESS in 3m 41s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/19/check/kolla-build-ubuntu-binary/d2dc924/ : FAILURE in 1h 26m 05s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/19/check/kolla-build-ubuntu-source/60c7197/ : SUCCESS in 33m 03s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/19/check/kolla-ansible-ubuntu-source/a6d437d/ : SUCCESS in 59m 40s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/19/check/kolla-ansible-ubuntu-binary/095cec1/ : SUCCESS in 1h 29m 39s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"59a968bf69034f359b61d5bd77fb94b4c4e6dec0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-28 17:15:31.000000000","message":"Patch Set 20: Patch Set 19 was rebased","accounts_in_message":[],"_revision_number":20},{"id":"df303eb57d960f3b4e14df9124d3ee7aea226acb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-28 18:59:45.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/20/check/kolla-build-centos-binary/fd5cc75/ : SUCCESS in 1h 15m 36s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/20/check/kolla-build-centos-source/2c2df1d/ : SUCCESS in 1h 24m 45s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/20/check/kolla-ansible-centos-source/3f14be6/ : SUCCESS in 1h 23m 17s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/20/check/kolla-ansible-centos-binary/55c8a84/ : FAILURE in 1h 07m 02s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/20/check/tripleo-build-containers-centos-7/5e319ca/ : FAILURE in 51m 45s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/20/check/kolla-ansible-bifrost-centos-source/3b57d6d/ : FAILURE in 31m 02s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/20/check/kolla-build-oraclelinux-binary/7ec056b/ : SUCCESS in 1h 16m 52s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/20/check/kolla-build-oraclelinux-source/11b8213/ : SUCCESS in 1h 17m 08s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/20/check/kolla-ansible-oraclelinux-source/e9f7d2a/ : SUCCESS in 1h 14m 08s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/20/check/kolla-ansible-oraclelinux-binary/634d37a/ : FAILURE in 1h 02m 32s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/20/check/openstack-tox-lower-constraints/6f26e4f/ : SUCCESS in 4m 31s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/20/check/openstack-tox-py36/1a783a5/ : SUCCESS in 3m 31s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/20/check/openstack-tox-pep8/4790540/ : SUCCESS in 5m 43s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/20/check/openstack-tox-py27/a83f631/ : SUCCESS in 4m 28s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/20/check/openstack-tox-py35/06c98df/ : SUCCESS in 4m 46s\n- openstack-tox-docs http://logs.openstack.org/38/624838/20/check/openstack-tox-docs/ee78f38/html/ : SUCCESS in 4m 21s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/20/check/kolla-build-ubuntu-binary/ff023c3/ : FAILURE in 1h 14m 21s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/20/check/kolla-build-ubuntu-source/ead2b4f/ : SUCCESS in 41m 43s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/20/check/kolla-ansible-ubuntu-source/b6c8bd2/ : SUCCESS in 1h 10m 28s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/20/check/kolla-ansible-ubuntu-binary/8618edc/ : SUCCESS in 1h 38m 36s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"8c0366720b7a0ca48895b2b2c5a3e0181fde2ddd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-29 14:46:07.000000000","message":"Patch Set 21: Patch Set 20 was rebased","accounts_in_message":[],"_revision_number":21},{"id":"e61706d351446990751e339ee00ca536f54894f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-29 17:03:22.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/21/check/kolla-build-centos-binary/fd86df6/ : FAILURE in 1h 16m 55s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/21/check/kolla-build-centos-source/8d5e683/ : FAILURE in 1h 42m 41s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/21/check/kolla-ansible-centos-source/5d94d23/ : SUCCESS in 1h 19m 38s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/21/check/kolla-ansible-centos-binary/3810e75/ : FAILURE in 1h 13m 53s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/21/check/tripleo-build-containers-centos-7/f59dd77/ : SUCCESS in 40m 52s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/21/check/kolla-ansible-bifrost-centos-source/39cae9c/ : SUCCESS in 55m 08s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/21/check/kolla-build-oraclelinux-binary/574e134/ : FAILURE in 1h 35m 22s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/21/check/kolla-build-oraclelinux-source/57e1922/ : FAILURE in 1h 55m 31s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/21/check/kolla-ansible-oraclelinux-source/506700b/ : SUCCESS in 1h 22m 31s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/21/check/kolla-ansible-oraclelinux-binary/7af97b1/ : FAILURE in 1h 14m 56s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/21/check/openstack-tox-lower-constraints/722fb0f/ : SUCCESS in 4m 22s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/21/check/openstack-tox-py36/4e55ace/ : SUCCESS in 5m 31s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/21/check/openstack-tox-pep8/d880415/ : SUCCESS in 4m 33s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/21/check/openstack-tox-py27/53e4acf/ : SUCCESS in 4m 36s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/21/check/openstack-tox-py35/7d99519/ : SUCCESS in 4m 16s\n- openstack-tox-docs http://logs.openstack.org/38/624838/21/check/openstack-tox-docs/9f0f31f/html/ : SUCCESS in 3m 33s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/21/check/kolla-build-ubuntu-binary/13d9684/ : FAILURE in 1h 36m 02s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/21/check/kolla-build-ubuntu-source/8d888f7/ : SUCCESS in 33m 58s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/21/check/kolla-ansible-ubuntu-source/2dc21a1/ : SUCCESS in 1h 03m 01s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/21/check/kolla-ansible-ubuntu-binary/6fe3063/ : SUCCESS in 1h 49m 43s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"01943809209d74af11dcb1f4fd33aa3cd90c3a14","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-31 15:57:57.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"ca17c8284848685df59d322506658292ea81ec06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-31 18:09:13.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/22/check/kolla-build-centos-binary/f3a2e84/ : SUCCESS in 1h 12m 28s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/22/check/kolla-build-centos-source/171926e/ : SUCCESS in 1h 39m 12s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/22/check/kolla-ansible-centos-source/210b952/ : SUCCESS in 1h 21m 52s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/22/check/kolla-ansible-centos-binary/3393506/ : TIMED_OUT in 2h 01m 31s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/22/check/tripleo-build-containers-centos-7/2ecd267/ : FAILURE in 43m 21s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/22/check/kolla-ansible-bifrost-centos-source/e0ace99/ : SUCCESS in 56m 19s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/22/check/kolla-build-oraclelinux-binary/e8b2f2a/ : SUCCESS in 1h 27m 37s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/22/check/kolla-build-oraclelinux-source/46f0f3a/ : SUCCESS in 1h 14m 27s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/22/check/kolla-ansible-oraclelinux-source/bbdf654/ : SUCCESS in 1h 17m 00s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/22/check/kolla-ansible-oraclelinux-binary/15abf94/ : FAILURE in 1h 04m 32s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/22/check/openstack-tox-lower-constraints/4607151/ : SUCCESS in 3m 55s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/22/check/openstack-tox-py36/80537f0/ : SUCCESS in 3m 29s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/22/check/openstack-tox-pep8/7c4a2a0/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/22/check/openstack-tox-py27/32fb2cb/ : SUCCESS in 3m 10s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/22/check/openstack-tox-py35/f8b34d9/ : SUCCESS in 3m 00s\n- openstack-tox-docs http://logs.openstack.org/38/624838/22/check/openstack-tox-docs/b3d77fc/html/ : SUCCESS in 2m 42s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/22/check/kolla-build-ubuntu-binary/47b0b51/ : FAILURE in 59m 57s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/22/check/kolla-build-ubuntu-source/4e35352/ : SUCCESS in 43m 03s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/22/check/kolla-ansible-ubuntu-source/0c60c2e/ : SUCCESS in 1h 07m 25s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/22/check/kolla-ansible-ubuntu-binary/1fed9fc/ : TIMED_OUT in 2h 02m 05s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"667ea2981ca9cd6a940173a6fd5c3ec6de470f29","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-31 19:24:48.000000000","message":"Patch Set 23: Patch Set 22 was rebased","accounts_in_message":[],"_revision_number":23},{"id":"3c7a3a97323d7acc8287eb192c41adc24336ad88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-31 20:56:43.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/23/check/kolla-build-centos-binary/f97fa03/ : FAILURE in 1h 23m 40s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/23/check/kolla-build-centos-source/4478af5/ : FAILURE in 1h 25m 18s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/23/check/kolla-ansible-centos-source/7dceef8/ : FAILURE in 40m 00s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/23/check/kolla-ansible-centos-binary/09837c3/ : FAILURE in 29m 02s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/23/check/tripleo-build-containers-centos-7/79cca11/ : FAILURE in 42m 03s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/23/check/kolla-ansible-bifrost-centos-source/e1555e1/ : FAILURE in 31m 04s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/23/check/kolla-build-oraclelinux-binary/72cba96/ : FAILURE in 1h 00m 26s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/23/check/kolla-build-oraclelinux-source/c4fb3f7/ : FAILURE in 1h 07m 24s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/23/check/kolla-ansible-oraclelinux-source/ec2cee9/ : FAILURE in 42m 15s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/23/check/kolla-ansible-oraclelinux-binary/a0ff55e/ : FAILURE in 30m 10s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/23/check/openstack-tox-lower-constraints/60ccf9c/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/23/check/openstack-tox-py36/2178f04/ : SUCCESS in 3m 23s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/23/check/openstack-tox-pep8/03b67bb/ : SUCCESS in 4m 29s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/23/check/openstack-tox-py27/e6d45cb/ : SUCCESS in 4m 29s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/23/check/openstack-tox-py35/6c31982/ : SUCCESS in 4m 55s\n- openstack-tox-docs http://logs.openstack.org/38/624838/23/check/openstack-tox-docs/92a1c90/html/ : SUCCESS in 2m 53s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/23/check/kolla-build-ubuntu-binary/47a3f2e/ : FAILURE in 1h 17m 31s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/23/check/kolla-build-ubuntu-source/edbe336/ : SUCCESS in 1h 13m 05s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/23/check/kolla-ansible-ubuntu-source/78b2509/ : SUCCESS in 54m 38s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/23/check/kolla-ansible-ubuntu-binary/cc39969/ : SUCCESS in 1h 24m 24s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"edcd07f04ac8d2ea28f5f7271dd3371edd5056f7","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-02-11 15:09:47.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"4b3cca955f766aceb704de2c0a3ec78d031e017c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-11 16:50:31.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/24/check/kolla-build-centos-binary/960483b/ : SUCCESS in 1h 09m 24s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/24/check/kolla-build-centos-source/93e8735/ : SUCCESS in 1h 16m 16s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/24/check/kolla-ansible-centos-source/d0ebe17/ : SUCCESS in 1h 34m 40s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/24/check/kolla-ansible-centos-binary/85f3452/ : FAILURE in 1h 01m 55s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/24/check/tripleo-build-containers-centos-7/86742a5/ : SUCCESS in 52m 00s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/24/check/kolla-ansible-bifrost-centos-source/c887e0b/ : SUCCESS in 1h 02m 05s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/24/check/kolla-build-oraclelinux-binary/1718d28/ : SUCCESS in 1h 06m 39s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/24/check/kolla-build-oraclelinux-source/753d849/ : SUCCESS in 1h 10m 11s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/24/check/kolla-ansible-oraclelinux-source/0176eed/ : SUCCESS in 1h 17m 36s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/24/check/kolla-ansible-oraclelinux-binary/a5a683e/ : FAILURE in 1h 11m 44s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/24/check/openstack-tox-lower-constraints/e8d8782/ : FAILURE in 3m 50s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/24/check/openstack-tox-py36/3f033f4/ : FAILURE in 4m 07s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/24/check/openstack-tox-pep8/25e6f10/ : SUCCESS in 3m 47s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/24/check/openstack-tox-py27/74e63c8/ : SUCCESS in 3m 33s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/24/check/openstack-tox-py35/b958962/ : FAILURE in 3m 12s\n- openstack-tox-docs http://logs.openstack.org/38/624838/24/check/openstack-tox-docs/fc91bd2/html/ : SUCCESS in 4m 09s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/24/check/kolla-build-ubuntu-binary/5a075a1/ : FAILURE in 1h 33m 23s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/24/check/kolla-build-ubuntu-source/339bd66/ : SUCCESS in 33m 55s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/24/check/kolla-ansible-ubuntu-source/f9a8090/ : FAILURE in 29m 16s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/24/check/kolla-ansible-ubuntu-binary/60c2410/ : FAILURE in 1h 15m 42s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"bd25bcdab3a5c7598bce2349a3afb40eacd6fc7d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-02-12 18:08:46.000000000","message":"Patch Set 25: Patch Set 24 was rebased","accounts_in_message":[],"_revision_number":25},{"id":"7d445c5afffdb847bc9f2fec02349569395c131a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-12 19:57:11.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/25/check/kolla-build-centos-binary/8e3514d/ : FAILURE in 15m 06s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/25/check/kolla-build-centos-source/4276362/ : FAILURE in 14m 55s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/25/check/kolla-ansible-centos-source/8d1c8fb/ : SUCCESS in 1h 20m 22s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/25/check/kolla-ansible-centos-binary/616c281/ : FAILURE in 1h 16m 10s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/25/check/tripleo-build-containers-centos-7/790cbc8/ : SUCCESS in 50m 04s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/25/check/kolla-ansible-bifrost-centos-source/4cf80ad/ : FAILURE in 34m 43s (non-voting)\n- tripleo-build-containers-fedora-28 http://logs.openstack.org/38/624838/25/check/tripleo-build-containers-fedora-28/983c5dc/ : SUCCESS in 36m 26s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/25/check/kolla-build-oraclelinux-binary/dfccf92/ : SUCCESS in 1h 21m 38s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/25/check/kolla-build-oraclelinux-source/a00a937/ : SUCCESS in 1h 38m 58s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/25/check/kolla-ansible-oraclelinux-source/b80dba0/ : SUCCESS in 1h 11m 02s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/25/check/kolla-ansible-oraclelinux-binary/b0570af/ : FAILURE in 1h 04m 39s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/25/check/openstack-tox-lower-constraints/3f1e3c6/ : FAILURE in 3m 49s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/25/check/openstack-tox-py36/f1cb0e7/ : FAILURE in 3m 07s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/25/check/openstack-tox-pep8/444d363/ : SUCCESS in 3m 53s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/25/check/openstack-tox-py27/bb2bb26/ : SUCCESS in 3m 37s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/25/check/openstack-tox-py35/4d3cb9d/ : FAILURE in 2m 47s\n- openstack-tox-docs http://logs.openstack.org/38/624838/25/check/openstack-tox-docs/0ac877e/html/ : SUCCESS in 3m 22s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/25/check/kolla-build-ubuntu-binary/666075b/ : FAILURE in 1h 27m 49s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/25/check/kolla-build-ubuntu-source/3426235/ : SUCCESS in 38m 00s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/25/check/kolla-ansible-ubuntu-source/a884492/ : FAILURE in 33m 18s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/25/check/kolla-ansible-ubuntu-binary/8dd8989/ : FAILURE in 1h 01m 18s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"6aed75a7a1d3184326514336ae523304cd966391","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-02-12 23:43:28.000000000","message":"Patch Set 26: Patch Set 25 was rebased","accounts_in_message":[],"_revision_number":26},{"id":"e7cdb5c828df167bd44d04ee7da88ca02c699a63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-13 01:23:01.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/26/check/kolla-build-centos-binary/91a3a14/ : FAILURE in 21m 55s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/26/check/kolla-build-centos-source/0c83c2a/ : FAILURE in 21m 34s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/26/check/kolla-ansible-centos-source/06d9336/ : SUCCESS in 1h 33m 16s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/26/check/kolla-ansible-centos-binary/13470b4/ : FAILURE in 1h 04m 59s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/26/check/tripleo-build-containers-centos-7/bf6e76c/ : SUCCESS in 46m 50s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/26/check/kolla-ansible-bifrost-centos-source/6dda8cb/ : FAILURE in 33m 25s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/26/check/kolla-build-oraclelinux-binary/e18a2f7/ : SUCCESS in 1h 16m 04s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/26/check/kolla-build-oraclelinux-source/a54a5b3/ : SUCCESS in 1h 05m 15s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/26/check/kolla-ansible-oraclelinux-source/84cf2ea/ : SUCCESS in 1h 20m 53s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/26/check/kolla-ansible-oraclelinux-binary/34ee870/ : FAILURE in 1h 12m 30s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/26/check/openstack-tox-lower-constraints/19bc51f/ : FAILURE in 3m 45s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/26/check/openstack-tox-py36/4e64858/ : FAILURE in 2m 29s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/26/check/openstack-tox-pep8/617b261/ : SUCCESS in 3m 42s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/26/check/openstack-tox-py27/41b9b10/ : SUCCESS in 2m 50s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/26/check/openstack-tox-py35/59690ab/ : FAILURE in 2m 47s\n- openstack-tox-docs http://logs.openstack.org/38/624838/26/check/openstack-tox-docs/1df7b66/html/ : SUCCESS in 3m 52s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/26/check/kolla-build-ubuntu-binary/252a2ff/ : FAILURE in 1h 13m 39s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/26/check/kolla-build-ubuntu-source/d7b2b53/ : SUCCESS in 33m 44s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/26/check/kolla-ansible-ubuntu-source/d33e659/ : FAILURE in 31m 24s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/26/check/kolla-ansible-ubuntu-binary/3f1e812/ : FAILURE in 1h 17m 09s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"72d06bc0a495e435dd3f45f0e315afac97c48b57","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-02-14 23:20:06.000000000","message":"Patch Set 27: Patch Set 26 was rebased","accounts_in_message":[],"_revision_number":27},{"id":"0027eed2a6383a8cafeefb3845a93ca5cc387f2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-15 03:00:35.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/27/check/kolla-build-centos-binary/f4ac5b0/ : SUCCESS in 1h 06m 31s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/27/check/kolla-build-centos-source/7a20cf2/ : SUCCESS in 1h 13m 36s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/27/check/kolla-ansible-centos-source/a1df210/ : SUCCESS in 1h 27m 27s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/27/check/kolla-ansible-centos-binary/7cfade2/ : SUCCESS in 1h 13m 16s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/27/check/tripleo-build-containers-centos-7/ba8655f/ : SUCCESS in 51m 10s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/27/check/kolla-ansible-bifrost-centos-source/5aa26ee/ : FAILURE in 18m 05s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/27/check/kolla-build-oraclelinux-binary/bf503c1/ : SUCCESS in 1h 20m 15s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/27/check/kolla-build-oraclelinux-source/516eddd/ : SUCCESS in 1h 07m 35s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/27/check/kolla-ansible-oraclelinux-source/51f81de/ : SUCCESS in 1h 25m 36s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/27/check/kolla-ansible-oraclelinux-binary/7b8e1b8/ : SUCCESS in 1h 15m 41s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/27/check/openstack-tox-lower-constraints/a5881a1/ : SUCCESS in 3m 27s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/27/check/openstack-tox-py36/ef6ddec/ : SUCCESS in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/27/check/openstack-tox-pep8/26a67be/ : SUCCESS in 3m 28s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/27/check/openstack-tox-py27/efba4e2/ : SUCCESS in 3m 09s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/27/check/openstack-tox-py35/a440ebd/ : SUCCESS in 4m 44s\n- openstack-tox-docs http://logs.openstack.org/38/624838/27/check/openstack-tox-docs/c0a3f1d/html/ : SUCCESS in 3m 07s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/27/check/kolla-build-ubuntu-binary/bb310cb/ : FAILURE in 52m 33s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/27/check/kolla-build-ubuntu-source/5f235e1/ : SUCCESS in 29m 52s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/27/check/kolla-ansible-ubuntu-source/074a666/ : SUCCESS in 1h 01m 51s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/27/check/kolla-ansible-ubuntu-binary/0014875/ : SUCCESS in 1h 56m 44s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"27543cd7533d8e0075372480279f67ac74a1c726","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-02-18 17:53:00.000000000","message":"Patch Set 28: Patch Set 27 was rebased","accounts_in_message":[],"_revision_number":28},{"id":"02c17b53bc8e0abdf07109f4e527dedeb12675ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-18 20:06:30.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/28/check/kolla-build-centos-binary/6f05fe3/ : SUCCESS in 1h 13m 38s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/28/check/kolla-build-centos-source/73bac92/ : SUCCESS in 1h 18m 01s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/28/check/kolla-ansible-centos-source/cddb036/ : SUCCESS in 1h 24m 36s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/28/check/kolla-ansible-centos-binary/7eb824a/ : SUCCESS in 1h 18m 35s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/28/check/tripleo-build-containers-centos-7/0e45cdf/ : SUCCESS in 44m 22s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/28/check/kolla-ansible-bifrost-centos-source/43e663f/ : SUCCESS in 1h 07m 49s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/28/check/kolla-build-oraclelinux-binary/ca46dcd/ : SUCCESS in 1h 29m 26s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/28/check/kolla-build-oraclelinux-source/65bce78/ : SUCCESS in 1h 30m 11s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/28/check/kolla-ansible-oraclelinux-source/d731ffc/ : SUCCESS in 1h 18m 47s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/28/check/kolla-ansible-oraclelinux-binary/fb2f17a/ : SUCCESS in 1h 20m 39s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/28/check/openstack-tox-lower-constraints/c26f14c/ : SUCCESS in 3m 54s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/28/check/openstack-tox-py36/2d063bc/ : SUCCESS in 3m 20s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/28/check/openstack-tox-pep8/1b887db/ : SUCCESS in 3m 45s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/28/check/openstack-tox-py27/c769625/ : SUCCESS in 3m 25s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/28/check/openstack-tox-py35/c7dc1af/ : SUCCESS in 3m 34s\n- openstack-tox-docs http://logs.openstack.org/38/624838/28/check/openstack-tox-docs/0604f84/html/ : SUCCESS in 3m 02s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/28/check/kolla-build-ubuntu-binary/32da758/ : FAILURE in 1h 25m 06s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/28/check/kolla-build-ubuntu-source/039db6f/ : SUCCESS in 34m 23s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/28/check/kolla-ansible-ubuntu-source/d9012d1/ : SUCCESS in 1h 07m 24s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/28/check/kolla-ansible-ubuntu-binary/0d9744a/ : SUCCESS in 1h 58m 15s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"2411482ae19e7f78e0bfe627096d91b15c41f6e3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-02-19 15:41:42.000000000","message":"Patch Set 29: Patch Set 28 was rebased","accounts_in_message":[],"_revision_number":29},{"id":"2b1fa75c9abdbd43f4f0f41412a8d866313599e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-19 18:20:33.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/29/check/kolla-build-centos-binary/07f1baa/ : SUCCESS in 1h 12m 06s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/29/check/kolla-build-centos-source/09360ee/ : SUCCESS in 1h 26m 03s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/29/check/kolla-ansible-centos-source/0ef4ddb/ : SUCCESS in 1h 19m 38s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/29/check/kolla-ansible-centos-binary/1d53f63/ : SUCCESS in 1h 22m 58s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/29/check/tripleo-build-containers-centos-7/e11fa1c/ : SUCCESS in 43m 30s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/29/check/kolla-ansible-bifrost-centos-source/06a2237/ : FAILURE in 41m 31s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/29/check/kolla-build-oraclelinux-binary/4ba9f6e/ : SUCCESS in 1h 06m 09s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/29/check/kolla-build-oraclelinux-source/0798f1d/ : SUCCESS in 1h 12m 22s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/29/check/kolla-ansible-oraclelinux-source/0025288/ : SUCCESS in 1h 11m 38s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/29/check/kolla-ansible-oraclelinux-binary/5807afe/ : SUCCESS in 1h 16m 23s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/29/check/openstack-tox-lower-constraints/6d2355a/ : SUCCESS in 5m 31s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/29/check/openstack-tox-py36/24c5522/ : SUCCESS in 5m 45s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/29/check/openstack-tox-pep8/9ee78b9/ : SUCCESS in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/29/check/openstack-tox-py27/a3ea78e/ : SUCCESS in 3m 28s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/29/check/openstack-tox-py35/8b14702/ : SUCCESS in 3m 49s\n- openstack-tox-docs http://logs.openstack.org/38/624838/29/check/openstack-tox-docs/9ac0c41/html/ : SUCCESS in 4m 54s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/29/check/kolla-build-ubuntu-binary/d27b900/ : FAILURE in 1h 21m 50s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/29/check/kolla-build-ubuntu-source/c6c2ac8/ : SUCCESS in 30m 11s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/29/check/kolla-ansible-ubuntu-source/57d94f5/ : SUCCESS in 1h 02m 21s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/29/check/kolla-ansible-ubuntu-binary/d8596f4/ : SUCCESS in 1h 56m 46s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"3bac82f12a0958c2df17a57a53efc597f68ad770","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-02-21 13:44:40.000000000","message":"Patch Set 30: Patch Set 29 was rebased","accounts_in_message":[],"_revision_number":30},{"id":"602496e4256046eacb055c766416e977be0eea09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-21 15:54:51.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/30/check/kolla-build-centos-binary/bfdfabc/ : FAILURE in 18m 59s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/30/check/kolla-build-centos-source/8bc8a65/ : FAILURE in 18m 30s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/30/check/kolla-ansible-centos-source/83b9a86/ : SUCCESS in 1h 17m 10s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/30/check/kolla-ansible-centos-binary/0f8abe7/ : SUCCESS in 1h 19m 03s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/30/check/tripleo-build-containers-centos-7/1131854/ : SUCCESS in 52m 53s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/30/check/kolla-ansible-bifrost-centos-source/fe11d28/ : FAILURE in 50m 08s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/30/check/kolla-build-oraclelinux-binary/5e89875/ : SUCCESS in 1h 19m 27s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/30/check/kolla-build-oraclelinux-source/f84b321/ : SUCCESS in 1h 25m 23s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/30/check/kolla-ansible-oraclelinux-source/208aa89/ : SUCCESS in 1h 25m 42s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/30/check/kolla-ansible-oraclelinux-binary/d060379/ : SUCCESS in 1h 19m 49s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/30/check/openstack-tox-lower-constraints/ac74012/ : SUCCESS in 4m 03s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/30/check/openstack-tox-py36/04cd1cf/ : SUCCESS in 3m 00s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/30/check/openstack-tox-pep8/282d824/ : SUCCESS in 3m 52s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/30/check/openstack-tox-py27/6f0ba42/ : SUCCESS in 2m 59s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/30/check/openstack-tox-py35/31aacef/ : SUCCESS in 3m 06s\n- openstack-tox-docs http://logs.openstack.org/38/624838/30/check/openstack-tox-docs/0a45cdc/html/ : SUCCESS in 2m 55s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/30/check/kolla-build-ubuntu-binary/c3ee712/ : FAILURE in 1h 29m 51s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/30/check/kolla-build-ubuntu-source/b8fe8f7/ : SUCCESS in 1h 07m 29s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/30/check/kolla-ansible-ubuntu-source/3076452/ : SUCCESS in 53m 38s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/30/check/kolla-ansible-ubuntu-binary/ab75cbe/ : TIMED_OUT in 2h 01m 40s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"c43fd94b270a4ee47ac7ccb6ccba0c5ddefd5a01","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-02-25 23:18:51.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"8741bf2467a200ed7e2fbec02e68fdb1ef180fa6","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-02-25 23:19:09.000000000","message":"Topic changed from python3-support to bp/python3-support","accounts_in_message":[],"_revision_number":31},{"id":"4e2d84a7fac954f9d89b03523109d04fbe139e02","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-02-25 23:30:40.000000000","message":"Uploaded patch set 32: Patch Set 31 was rebased.","accounts_in_message":[],"_revision_number":32},{"id":"06245b8478c12d7c402ce46fc1288c22513b0e67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-26 01:32:56.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/32/check/kolla-build-centos-binary/6a461e8/ : SUCCESS in 39m 51s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/32/check/kolla-build-centos-source/47dcf93/ : SUCCESS in 1h 08m 29s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/32/check/kolla-ansible-centos-source/343af33/ : SUCCESS in 1h 26m 28s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/32/check/kolla-ansible-centos-binary/864a106/ : SUCCESS in 1h 42m 31s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/32/check/tripleo-build-containers-centos-7/c21edbd/ : SUCCESS in 44m 10s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/32/check/kolla-ansible-bifrost-centos-source/09b8557/ : FAILURE in 54m 53s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/32/check/kolla-build-oraclelinux-binary/76af6fc/ : SUCCESS in 47m 22s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/32/check/kolla-build-oraclelinux-source/909a1f5/ : SUCCESS in 53m 40s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/32/check/kolla-ansible-oraclelinux-source/b4d84e1/ : SUCCESS in 1h 19m 28s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/32/check/kolla-ansible-oraclelinux-binary/b8a1569/ : SUCCESS in 1h 16m 10s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/32/check/openstack-tox-lower-constraints/54a4905/ : FAILURE in 3m 25s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/32/check/openstack-tox-py36/1067f4e/ : FAILURE in 3m 03s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/32/check/openstack-tox-pep8/2947260/ : SUCCESS in 3m 52s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/32/check/openstack-tox-py27/286dfbb/ : FAILURE in 3m 26s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/32/check/openstack-tox-py35/7abd2a2/ : FAILURE in 2m 46s\n- openstack-tox-docs http://logs.openstack.org/38/624838/32/check/openstack-tox-docs/7bc2d85/html/ : SUCCESS in 3m 18s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/32/check/kolla-build-ubuntu-binary/38b1e0e/ : FAILURE in 1h 21m 05s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/32/check/kolla-build-ubuntu-source/cc1bc8d/ : SUCCESS in 28m 24s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/32/check/kolla-ansible-ubuntu-source/d7dc718/ : SUCCESS in 1h 16m 02s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/32/check/kolla-ansible-ubuntu-binary/d45d1a6/ : SUCCESS in 1h 56m 58s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"0ca40efa623e396666cd7ae5bc5c3de2b2d81ee9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-02-26 18:30:45.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"b77ca85a69fa4e3972db7a9b75fcd8eccd361af1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-26 20:38:10.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/33/check/kolla-build-centos-binary/d157441/ : SUCCESS in 42m 22s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/33/check/kolla-build-centos-source/b473698/ : SUCCESS in 1h 02m 09s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/33/check/kolla-ansible-centos-source/d64cca4/ : FAILURE in 54m 39s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/33/check/kolla-ansible-centos-binary/9b31dde/ : SUCCESS in 1h 17m 12s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/33/check/tripleo-build-containers-centos-7/293d13f/ : SUCCESS in 46m 50s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/33/check/kolla-ansible-bifrost-centos-source/b1f2979/ : FAILURE in 57m 18s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/33/check/kolla-build-oraclelinux-binary/433f93e/ : SUCCESS in 54m 59s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/33/check/kolla-build-oraclelinux-source/91d7342/ : SUCCESS in 47m 23s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/33/check/kolla-ansible-oraclelinux-source/2a3e8ed/ : FAILURE in 56m 41s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/33/check/kolla-ansible-oraclelinux-binary/a261e5d/ : SUCCESS in 1h 10m 05s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/33/check/openstack-tox-lower-constraints/fc110ea/ : SUCCESS in 3m 33s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/33/check/openstack-tox-py36/adf90a0/ : SUCCESS in 3m 22s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/33/check/openstack-tox-pep8/538dc9b/ : SUCCESS in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/33/check/openstack-tox-py27/8fc2453/ : SUCCESS in 3m 43s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/33/check/openstack-tox-py35/6af07fe/ : SUCCESS in 3m 24s\n- openstack-tox-docs http://logs.openstack.org/38/624838/33/check/openstack-tox-docs/21be684/html/ : SUCCESS in 3m 03s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/33/check/kolla-build-ubuntu-binary/612d693/ : FAILURE in 1h 31m 13s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/33/check/kolla-build-ubuntu-source/45037c4/ : SUCCESS in 36m 26s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/33/check/kolla-ansible-ubuntu-source/38db57f/ : FAILURE in 47m 23s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/33/check/kolla-ansible-ubuntu-binary/670c548/ : TIMED_OUT in 2h 01m 46s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"49832eafe9f697ec7db6eb62b5d9c917ba379832","author":{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},"date":"2019-02-27 08:45:53.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"76fca95205733a7af86031c418a2a5b7beaf13a2","author":{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},"date":"2019-02-27 08:46:46.000000000","message":"Patch Set 34:\n\nHave added a patchset with https://review.openstack.org/#/c/639596/1 as commented by lyarwood","accounts_in_message":[],"_revision_number":34},{"id":"e661f409ee36ad95f95705d3b9ac9a9f7760bb69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-27 10:52:46.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/34/check/kolla-build-centos-binary/5014622/ : SUCCESS in 38m 52s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/34/check/kolla-build-centos-source/c107320/ : SUCCESS in 50m 47s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/34/check/kolla-ansible-centos-source/29a4bcf/ : FAILURE in 18m 32s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/34/check/kolla-ansible-centos-binary/082341e/ : SUCCESS in 1h 17m 13s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/34/check/tripleo-build-containers-centos-7/322fd7f/ : SUCCESS in 47m 17s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/34/check/kolla-ansible-bifrost-centos-source/dce1579/ : FAILURE in 50m 10s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/34/check/kolla-build-oraclelinux-binary/8203a35/ : SUCCESS in 54m 32s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/34/check/kolla-build-oraclelinux-source/69a9407/ : SUCCESS in 54m 10s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/34/check/kolla-ansible-oraclelinux-source/faca436/ : FAILURE in 55m 46s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/34/check/kolla-ansible-oraclelinux-binary/ba0dd45/ : SUCCESS in 1h 20m 53s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/34/check/openstack-tox-lower-constraints/9958fe1/ : SUCCESS in 4m 44s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/34/check/openstack-tox-py36/a031580/ : SUCCESS in 4m 02s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/34/check/openstack-tox-pep8/10b066f/ : SUCCESS in 3m 54s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/34/check/openstack-tox-py27/f2f5983/ : SUCCESS in 3m 42s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/34/check/openstack-tox-py35/896b9bf/ : SUCCESS in 3m 15s\n- openstack-tox-docs http://logs.openstack.org/38/624838/34/check/openstack-tox-docs/da1da56/html/ : SUCCESS in 3m 15s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/34/check/kolla-build-ubuntu-binary/d318c2b/ : FAILURE in 1h 35m 16s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/34/check/kolla-build-ubuntu-source/b4509fd/ : SUCCESS in 46m 26s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/34/check/kolla-ansible-ubuntu-source/580e69d/ : FAILURE in 48m 21s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/34/check/kolla-ansible-ubuntu-binary/369d674/ : TIMED_OUT in 2h 01m 23s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"6c94da54e4d696096df2923e3ba597acb2f60f50","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-02-27 16:10:12.000000000","message":"Patch Set 34:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":34},{"id":"cbd0c468e72e8e144b5933513f6a2a27e7201c04","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-02-27 17:32:17.000000000","message":"Uploaded patch set 35: Patch Set 34 was rebased.","accounts_in_message":[],"_revision_number":35},{"id":"39118235a88d0d5df11fdf6277ccf24aaea90d8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-27 19:44:28.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/35/check/kolla-build-centos-binary/5788fba/ : SUCCESS in 37m 03s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/35/check/kolla-build-centos-source/5c59c64/ : SUCCESS in 50m 11s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/35/check/kolla-ansible-centos-source/f57162c/ : FAILURE in 55m 04s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/35/check/kolla-ansible-centos-binary/c006885/ : SUCCESS in 1h 17m 12s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/35/check/tripleo-build-containers-centos-7/a143646/ : SUCCESS in 48m 17s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/35/check/kolla-ansible-bifrost-centos-source/4f8eb22/ : FAILURE in 57m 24s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/35/check/kolla-build-oraclelinux-binary/2813251/ : SUCCESS in 46m 25s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/35/check/kolla-build-oraclelinux-source/962e3f0/ : SUCCESS in 46m 24s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/35/check/kolla-ansible-oraclelinux-source/8860c83/ : FAILURE in 56m 37s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/35/check/kolla-ansible-oraclelinux-binary/3dbee82/ : SUCCESS in 1h 15m 57s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/35/check/openstack-tox-lower-constraints/bf7da87/ : SUCCESS in 4m 01s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/35/check/openstack-tox-py36/9095562/ : SUCCESS in 3m 31s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/35/check/openstack-tox-pep8/7e35849/ : SUCCESS in 3m 53s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/35/check/openstack-tox-py27/fa79b07/ : SUCCESS in 3m 16s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/35/check/openstack-tox-py35/ba7d52b/ : SUCCESS in 4m 33s\n- openstack-tox-docs http://logs.openstack.org/38/624838/35/check/openstack-tox-docs/c1daf34/html/ : SUCCESS in 3m 42s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/35/check/kolla-build-ubuntu-binary/1f8cf58/ : FAILURE in 1h 00m 37s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/35/check/kolla-build-ubuntu-source/401b0d8/ : SUCCESS in 35m 43s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/35/check/kolla-ansible-ubuntu-source/4344200/ : FAILURE in 35m 46s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/35/check/kolla-ansible-ubuntu-binary/3a4cbf8/ : SUCCESS in 2h 00m 40s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"dffb1a27520dc44ba43be28d69de660ad6275618","author":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"date":"2019-02-28 06:47:50.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"2f6c0abf99867fa3a4104ecfc44cabb922643304","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-02-28 08:33:19.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"aade9fd4d9c235da6e4c16dbd6b2c389fefd2d90","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-02-28 15:35:18.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"4acb1adeb02d2f4adaa17eba59dc4a169258a7ed","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-02-28 15:41:44.000000000","message":"Patch Set 35:\n\n(2 comments)","accounts_in_message":[],"_revision_number":35},{"id":"3ca8dc06b648a5adff743c293e17cd7824af2011","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-03-01 21:51:04.000000000","message":"Patch Set 36: Patch Set 35 was rebased","accounts_in_message":[],"_revision_number":36},{"id":"48dc751678da9f1226f9d9c68c86de2ef838761a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-01 22:07:02.000000000","message":"Patch Set 36:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 http://logs.rdoproject.org/38/624838/36/openstack-check/tripleo-build-containers-fedora-28/8b0cc96/ : FAILURE in 7m 10s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"0c9d1315c5beb7c1b35f17d75c832c15dbe95653","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-02 00:00:15.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/36/check/kolla-build-centos-binary/9352d32/ : SUCCESS in 43m 49s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/36/check/kolla-build-centos-source/507bcba/ : FAILURE in 54m 22s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/36/check/kolla-ansible-centos-source/1afd445/ : SUCCESS in 1h 17m 45s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/36/check/kolla-ansible-centos-binary/c02048e/ : SUCCESS in 1h 19m 15s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/36/check/tripleo-build-containers-centos-7/0f759ab/ : SUCCESS in 47m 31s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/36/check/kolla-ansible-bifrost-centos-source/f9f82ec/ : FAILURE in 49m 39s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/36/check/kolla-build-oraclelinux-binary/b1d20fa/ : SUCCESS in 48m 32s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/36/check/kolla-build-oraclelinux-source/e7da507/ : FAILURE in 51m 00s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/36/check/kolla-ansible-oraclelinux-source/534790d/ : SUCCESS in 1h 10m 50s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/36/check/kolla-ansible-oraclelinux-binary/6713716/ : SUCCESS in 1h 12m 50s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/36/check/openstack-tox-lower-constraints/1a8c4fe/ : SUCCESS in 4m 42s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/36/check/openstack-tox-py36/152eddf/ : SUCCESS in 4m 01s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/36/check/openstack-tox-pep8/efcb7b9/ : SUCCESS in 5m 42s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/36/check/openstack-tox-py27/22940fc/ : SUCCESS in 4m 18s\n- openstack-tox-py35 http://logs.openstack.org/38/624838/36/check/openstack-tox-py35/8df4ab4/ : SUCCESS in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/38/624838/36/check/openstack-tox-docs/fb0bbfb/html/ : SUCCESS in 4m 22s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/36/check/kolla-build-ubuntu-binary/c22bd03/ : FAILURE in 1h 23m 33s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/36/check/kolla-build-ubuntu-source/1f2371c/ : FAILURE in 27m 40s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/36/check/kolla-ansible-ubuntu-source/209ceb2/ : SUCCESS in 1h 01m 20s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/36/check/kolla-ansible-ubuntu-binary/736f843/ : TIMED_OUT in 2h 01m 05s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"c3a76fb34923b2fa8a67a80342ed2c6fe9c32cbb","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-03-02 11:49:29.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"96d0ece05d32d291dcb3c95d09c81a5e3043d5ea","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-03-06 13:59:47.000000000","message":"Uploaded patch set 37: Patch Set 36 was rebased.","accounts_in_message":[],"_revision_number":37},{"id":"daccf2643c9d4298ca9cedae4438ff48a9bbedd6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-06 14:37:08.000000000","message":"Patch Set 37:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 http://logs.rdoproject.org/38/624838/37/openstack-check/tripleo-build-containers-fedora-28/cf74ca4/ : SUCCESS in 34m 21s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"b817e5decde39d038ab57b7958ca1d3be14ec03d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-03-06 15:03:46.000000000","message":"Patch Set 38: Patch Set 37 was rebased","accounts_in_message":[],"_revision_number":38},{"id":"43e246c61bb3b27e87256f8b114ee4b647345d95","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-06 15:43:47.000000000","message":"Patch Set 38:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 http://logs.rdoproject.org/38/624838/38/openstack-check/tripleo-build-containers-fedora-28/e0ba40b/ : SUCCESS in 33m 21s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"465ac6954ee8e068ffdd3713a7fb435c62746f44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-06 16:31:43.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/38/check/kolla-build-centos-binary/bdc0c49/ : SUCCESS in 56m 37s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/38/check/kolla-build-centos-source/480a21e/ : SUCCESS in 44m 54s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/38/check/kolla-ansible-centos-source/5f59714/ : SUCCESS in 1h 19m 45s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/38/check/kolla-ansible-centos-binary/cd4197a/ : SUCCESS in 1h 19m 53s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/38/check/tripleo-build-containers-centos-7/d7a1217/ : SUCCESS in 47m 11s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/38/check/kolla-ansible-bifrost-centos-source/b3146ee/ : FAILURE in 39m 39s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/38/check/kolla-build-oraclelinux-binary/bafdb6d/ : SUCCESS in 50m 22s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/38/check/kolla-build-oraclelinux-source/8acbc78/ : SUCCESS in 42m 17s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/38/check/kolla-ansible-oraclelinux-source/9196df3/ : SUCCESS in 1h 19m 29s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/38/check/kolla-ansible-oraclelinux-binary/80bdc94/ : SUCCESS in 1h 12m 16s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/38/check/openstack-tox-lower-constraints/f2e6f2d/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/38/check/openstack-tox-py36/31ec109/ : SUCCESS in 3m 48s\n- openstack-tox-py37 http://logs.openstack.org/38/624838/38/check/openstack-tox-py37/4d2b03d/ : SUCCESS in 4m 35s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/38/check/openstack-tox-pep8/b27f88a/ : SUCCESS in 3m 38s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/38/check/openstack-tox-py27/e1514f7/ : SUCCESS in 3m 42s\n- openstack-tox-docs http://logs.openstack.org/38/624838/38/check/openstack-tox-docs/3a5fa50/html/ : SUCCESS in 3m 24s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/38/check/kolla-build-ubuntu-binary/17eed0c/ : FAILURE in 1h 21m 23s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/38/check/kolla-build-ubuntu-source/99d07ad/ : SUCCESS in 40m 59s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/38/check/kolla-ansible-ubuntu-source/104709b/ : SUCCESS in 56m 11s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/38/check/kolla-ansible-ubuntu-binary/5421ce0/ : SUCCESS in 1h 25m 14s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"49d13ed5582198a7d26cf027273210ab6f31411e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-03-06 17:26:23.000000000","message":"Patch Set 39: Patch Set 38 was rebased","accounts_in_message":[],"_revision_number":39},{"id":"4b9a0f157cddff80261e8c4c9bbf3ab8ba829264","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-06 17:36:28.000000000","message":"Patch Set 39:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 http://logs.rdoproject.org/38/624838/39/openstack-check/tripleo-build-containers-fedora-28/d4514e0/ : FAILURE in 6m 45s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"c798e420fdf2a0f72fc7ac9f78d4d7f43c147cfa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-06 17:53:40.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/39/check/kolla-build-centos-binary/7d15332/ : FAILURE in 5m 32s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/39/check/kolla-build-centos-source/356706d/ : FAILURE in 5m 11s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/39/check/kolla-ansible-centos-source/aa9ee41/ : FAILURE in 8m 07s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/39/check/kolla-ansible-centos-binary/ebd79cf/ : FAILURE in 8m 51s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/39/check/tripleo-build-containers-centos-7/b527b78/ : FAILURE in 12m 28s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/39/check/kolla-ansible-bifrost-centos-source/cd63ac4/ : FAILURE in 8m 05s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/39/check/kolla-build-oraclelinux-binary/faef897/ : FAILURE in 6m 33s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/39/check/kolla-build-oraclelinux-source/ad1da35/ : FAILURE in 8m 57s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/39/check/kolla-ansible-oraclelinux-source/a5daa54/ : FAILURE in 14m 12s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/39/check/kolla-ansible-oraclelinux-binary/927142f/ : FAILURE in 11m 01s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/39/check/openstack-tox-lower-constraints/7c632f7/ : SUCCESS in 3m 39s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/39/check/openstack-tox-py36/af89bbd/ : SUCCESS in 3m 24s\n- openstack-tox-py37 http://logs.openstack.org/38/624838/39/check/openstack-tox-py37/a29b7f3/ : SUCCESS in 4m 13s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/39/check/openstack-tox-pep8/b48babe/ : SUCCESS in 3m 49s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/39/check/openstack-tox-py27/d1324d7/ : SUCCESS in 3m 11s\n- openstack-tox-docs http://logs.openstack.org/38/624838/39/check/openstack-tox-docs/a257961/html/ : SUCCESS in 4m 22s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/39/check/kolla-build-ubuntu-binary/b79aa60/ : FAILURE in 4m 17s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/39/check/kolla-build-ubuntu-source/279aaf3/ : FAILURE in 4m 40s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/39/check/kolla-ansible-ubuntu-source/745acc3/ : FAILURE in 8m 55s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/39/check/kolla-ansible-ubuntu-binary/c5bea75/ : FAILURE in 9m 15s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"d91ec1c1330346a198ea0ed847c77f3bd75ef9bd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-03-06 18:00:41.000000000","message":"Patch Set 40: Patch Set 39 was rebased","accounts_in_message":[],"_revision_number":40},{"id":"f4f5aa9fcc5c67c9fd9fa2675dcea54b88735d24","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-06 18:37:49.000000000","message":"Patch Set 40:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 http://logs.rdoproject.org/38/624838/40/openstack-check/tripleo-build-containers-fedora-28/24fccfd/ : SUCCESS in 31m 17s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"843ab54645038300e90e905c925bddf192c6691d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-06 19:38:24.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/40/check/kolla-build-centos-binary/20db428/ : SUCCESS in 40m 48s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/40/check/kolla-build-centos-source/91b5947/ : SUCCESS in 47m 24s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/40/check/kolla-ansible-centos-source/d98c82b/ : SUCCESS in 1h 22m 30s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/40/check/kolla-ansible-centos-binary/8dc8f59/ : SUCCESS in 1h 08m 37s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/40/check/tripleo-build-containers-centos-7/05496f4/ : SUCCESS in 56m 59s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/40/check/kolla-ansible-bifrost-centos-source/d8bc9d8/ : FAILURE in 34m 44s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/40/check/kolla-build-oraclelinux-binary/426ed19/ : SUCCESS in 49m 20s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/40/check/kolla-build-oraclelinux-source/5a1e3c4/ : SUCCESS in 47m 29s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/40/check/kolla-ansible-oraclelinux-source/e620292/ : SUCCESS in 1h 28m 24s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/40/check/kolla-ansible-oraclelinux-binary/ddbd1e6/ : SUCCESS in 1h 16m 35s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/40/check/openstack-tox-lower-constraints/97c27a6/ : SUCCESS in 2m 59s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/40/check/openstack-tox-py36/388da42/ : SUCCESS in 4m 13s\n- openstack-tox-py37 http://logs.openstack.org/38/624838/40/check/openstack-tox-py37/e1d3ed2/ : SUCCESS in 4m 22s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/40/check/openstack-tox-pep8/d19f786/ : SUCCESS in 3m 33s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/40/check/openstack-tox-py27/d371040/ : SUCCESS in 4m 10s\n- openstack-tox-docs http://logs.openstack.org/38/624838/40/check/openstack-tox-docs/22eb29e/html/ : SUCCESS in 3m 06s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/40/check/kolla-build-ubuntu-binary/25bc9c3/ : SUCCESS in 31m 43s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/40/check/kolla-build-ubuntu-source/bae1a05/ : SUCCESS in 29m 01s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/40/check/kolla-ansible-ubuntu-source/5d4df17/ : SUCCESS in 1h 00m 53s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/40/check/kolla-ansible-ubuntu-binary/52f7fd6/ : FAILURE in 31m 21s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"ea6818118e3828705188692216822432e3804f64","author":{"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},"date":"2019-03-10 15:38:50.000000000","message":"Patch Set 40: Code-Review+1","accounts_in_message":[],"_revision_number":40},{"id":"6e60567b1078c968c29f5e60d8dcf782fc9bf19d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-03-11 15:06:50.000000000","message":"Patch Set 41: Patch Set 40 was rebased","accounts_in_message":[],"_revision_number":41},{"id":"8272f7c488e822a0d118a748d2187c6583cd0932","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-03-11 16:39:36.000000000","message":"Patch Set 42: Patch Set 41 was rebased","accounts_in_message":[],"_revision_number":42},{"id":"bd179afa1b744b84b509be779f0fc2487e7d4f34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-11 16:39:43.000000000","message":"Patch Set 41: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":41},{"id":"096fd98d3bf04514e93bf739e830d7564a892565","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-11 16:39:44.000000000","message":"Patch Set 41:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":41},{"id":"711d667ec8ff0616a6b410a61b52c0e3270a4eb1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-11 18:00:06.000000000","message":"Patch Set 42: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/42/check/kolla-build-centos-binary/7d9825c/ : SUCCESS in 49m 05s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/42/check/kolla-build-centos-source/4fbc804/ : SUCCESS in 54m 01s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/42/check/kolla-ansible-centos-source/421c745/ : SUCCESS in 1h 17m 30s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/42/check/kolla-ansible-centos-binary/7f893d1/ : SUCCESS in 1h 10m 32s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/42/check/tripleo-build-containers-centos-7/dcbdc66/ : SUCCESS in 1h 09m 31s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/42/check/kolla-ansible-bifrost-centos-source/8c64d15/ : FAILURE in 37m 20s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/42/check/kolla-build-oraclelinux-binary/dfdf3b6/ : SUCCESS in 52m 57s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/42/check/kolla-build-oraclelinux-source/896eb3c/ : SUCCESS in 1h 15m 19s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/42/check/kolla-ansible-oraclelinux-source/a9418ca/ : FAILURE in 20m 42s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/42/check/kolla-ansible-oraclelinux-binary/71b0aa7/ : SUCCESS in 1h 18m 50s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/42/check/openstack-tox-lower-constraints/8329a2b/ : SUCCESS in 7m 05s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/42/check/openstack-tox-py36/15ec851/ : SUCCESS in 4m 39s\n- openstack-tox-py37 http://logs.openstack.org/38/624838/42/check/openstack-tox-py37/fec32e7/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/42/check/openstack-tox-pep8/9b436d1/ : SUCCESS in 7m 33s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/42/check/openstack-tox-py27/2189ac7/ : SUCCESS in 5m 12s\n- openstack-tox-docs http://logs.openstack.org/38/624838/42/check/openstack-tox-docs/531ffb0/html/ : SUCCESS in 3m 43s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/42/check/kolla-build-ubuntu-binary/008a3e2/ : SUCCESS in 42m 15s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/42/check/kolla-build-ubuntu-source/977e062/ : SUCCESS in 1h 02m 31s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/42/check/kolla-ansible-ubuntu-source/934a1d3/ : FAILURE in 5m 00s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/42/check/kolla-ansible-ubuntu-binary/ba49d5a/ : FAILURE in 35m 29s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"6706fb063d5b12a290942f920bb05d144bb5e9e3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-03-11 18:45:09.000000000","message":"Patch Set 43: Patch Set 42 was rebased","accounts_in_message":[],"_revision_number":43},{"id":"2387708f55b9e2bc4ead6e325668053c4faf12f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-11 20:20:12.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/43/check/kolla-build-centos-binary/354afd7/ : SUCCESS in 29m 41s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/43/check/kolla-build-centos-source/3e16780/ : SUCCESS in 39m 06s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/43/check/kolla-ansible-centos-source/7139528/ : SUCCESS in 1h 29m 26s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/43/check/kolla-ansible-centos-binary/552c054/ : SUCCESS in 1h 10m 14s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/43/check/tripleo-build-containers-centos-7/7b42e30/ : SUCCESS in 40m 34s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/43/check/kolla-ansible-bifrost-centos-source/5ce82ec/ : FAILURE in 34m 03s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/43/check/kolla-build-oraclelinux-binary/36b0fec/ : SUCCESS in 46m 54s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/43/check/kolla-build-oraclelinux-source/4e8cd45/ : SUCCESS in 40m 50s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/43/check/kolla-ansible-oraclelinux-source/f6b557d/ : SUCCESS in 1h 12m 17s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/43/check/kolla-ansible-oraclelinux-binary/ce06511/ : SUCCESS in 1h 07m 00s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/43/check/openstack-tox-lower-constraints/0c59898/ : SUCCESS in 4m 53s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/43/check/openstack-tox-py36/4148a7c/ : SUCCESS in 3m 07s\n- openstack-tox-py37 http://logs.openstack.org/38/624838/43/check/openstack-tox-py37/cfa6e18/ : SUCCESS in 3m 56s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/43/check/openstack-tox-pep8/a72b4d3/ : SUCCESS in 3m 41s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/43/check/openstack-tox-py27/8a2343c/ : SUCCESS in 4m 19s\n- openstack-tox-docs http://logs.openstack.org/38/624838/43/check/openstack-tox-docs/a8347df/html/ : SUCCESS in 3m 03s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/43/check/kolla-build-ubuntu-binary/373ec0a/ : SUCCESS in 39m 17s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/43/check/kolla-build-ubuntu-source/8474f8e/ : SUCCESS in 31m 28s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/43/check/kolla-ansible-ubuntu-source/a45cfb6/ : SUCCESS in 59m 24s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/43/check/kolla-ansible-ubuntu-binary/2a98777/ : FAILURE in 32m 15s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"c937503a4919baff8077dd3bcdda9340e94d9154","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-11 20:53:24.000000000","message":"Patch Set 43:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 http://logs.rdoproject.org/38/624838/43/openstack-check/tripleo-build-containers-fedora-28/379cc55/ : SUCCESS in 30m 49s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"07eb78be5d42d44105c89b6dc596b56077535530","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-03-12 14:27:48.000000000","message":"Patch Set 44: Patch Set 43 was rebased","accounts_in_message":[],"_revision_number":44},{"id":"c95b5bf16536f14d2a2acf526856e574e0450b9f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-03-12 16:47:47.000000000","message":"Uploaded patch set 45.","accounts_in_message":[],"_revision_number":45},{"id":"e0fb217ddecd34ffaa7215ef6ce3532fc6432706","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-03-12 16:49:28.000000000","message":"Patch Set 46: Patch Set 45 was rebased","accounts_in_message":[],"_revision_number":46},{"id":"f412ca584afea8f784b37dde6c207f2a9e3a6695","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-12 18:19:36.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/46/check/kolla-build-centos-binary/5b934ff/ : SUCCESS in 27m 54s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/46/check/kolla-build-centos-source/72557c3/ : SUCCESS in 43m 31s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/46/check/kolla-ansible-centos-source/1af7ef0/ : SUCCESS in 1h 16m 20s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/46/check/kolla-ansible-centos-binary/188d55d/ : SUCCESS in 1h 13m 25s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/46/check/tripleo-build-containers-centos-7/f445f8c/ : SUCCESS in 41m 23s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/46/check/kolla-ansible-bifrost-centos-source/c41557b/ : FAILURE in 51m 57s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/46/check/kolla-build-oraclelinux-binary/5d51e23/ : SUCCESS in 53m 34s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/46/check/kolla-build-oraclelinux-source/8b80489/ : SUCCESS in 55m 28s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/46/check/kolla-ansible-oraclelinux-source/0786d36/ : SUCCESS in 1h 24m 29s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/46/check/kolla-ansible-oraclelinux-binary/57ecdbb/ : SUCCESS in 1h 13m 51s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/46/check/openstack-tox-lower-constraints/8c754e7/ : SUCCESS in 5m 06s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/46/check/openstack-tox-pep8/1111d7c/ : SUCCESS in 5m 39s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/46/check/openstack-tox-py27/06770fd/ : SUCCESS in 4m 19s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/46/check/openstack-tox-py36/dc6aa0a/ : SUCCESS in 4m 35s\n- openstack-tox-py37 http://logs.openstack.org/38/624838/46/check/openstack-tox-py37/184c2af/ : SUCCESS in 5m 19s\n- openstack-tox-docs http://logs.openstack.org/38/624838/46/check/openstack-tox-docs/3f88e6e/html/ : SUCCESS in 4m 41s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/46/check/kolla-build-ubuntu-binary/b1bbdd8/ : SUCCESS in 31m 23s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/46/check/kolla-build-ubuntu-source/6729aa5/ : SUCCESS in 34m 43s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/46/check/kolla-ansible-ubuntu-source/1267d9d/ : SUCCESS in 1h 08m 41s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/46/check/kolla-ansible-ubuntu-binary/216e31f/ : FAILURE in 28m 24s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"25eaae4d05ea34349d507dab87a956ee96a077be","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-03-12 18:24:34.000000000","message":"Patch Set 47: Patch Set 46 was rebased","accounts_in_message":[],"_revision_number":47},{"id":"cfb72225d728d75846d2f99799054999281285d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-12 20:02:08.000000000","message":"Patch Set 47: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/47/check/kolla-build-centos-binary/34f6a80/ : SUCCESS in 42m 31s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/47/check/kolla-build-centos-source/3241791/ : SUCCESS in 47m 55s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/47/check/kolla-ansible-centos-source/572782c/ : SUCCESS in 1h 18m 33s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/47/check/kolla-ansible-centos-binary/345b99b/ : SUCCESS in 1h 21m 17s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/47/check/tripleo-build-containers-centos-7/20a97c5/ : SUCCESS in 48m 18s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/47/check/kolla-ansible-bifrost-centos-source/6cfa093/ : FAILURE in 59m 12s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/47/check/kolla-build-oraclelinux-binary/4b478d0/ : SUCCESS in 46m 27s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/47/check/kolla-build-oraclelinux-source/883586a/ : SUCCESS in 58m 05s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/47/check/kolla-ansible-oraclelinux-source/1e037b3/ : SUCCESS in 1h 20m 52s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/47/check/kolla-ansible-oraclelinux-binary/de24034/ : SUCCESS in 1h 07m 09s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/47/check/openstack-tox-lower-constraints/f3684d5/ : SUCCESS in 5m 13s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/47/check/openstack-tox-pep8/cf1003b/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/47/check/openstack-tox-py27/790eaeb/ : SUCCESS in 3m 54s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/47/check/openstack-tox-py36/c7af4a0/ : SUCCESS in 3m 28s\n- openstack-tox-py37 http://logs.openstack.org/38/624838/47/check/openstack-tox-py37/9e256cb/ : SUCCESS in 3m 40s\n- openstack-tox-docs http://logs.openstack.org/38/624838/47/check/openstack-tox-docs/a2b1b84/html/ : SUCCESS in 2m 58s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/47/check/kolla-build-ubuntu-binary/7aee6e6/ : FAILURE in 55m 02s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/47/check/kolla-build-ubuntu-source/186aca4/ : SUCCESS in 35m 45s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/47/check/kolla-ansible-ubuntu-source/63382ec/ : SUCCESS in 1h 02m 31s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/47/check/kolla-ansible-ubuntu-binary/580b6d2/ : FAILURE in 27m 58s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"1a8c02a157ba9156c8fe4cdba070078b718ad0ca","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-13 01:50:26.000000000","message":"Patch Set 47:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 http://logs.rdoproject.org/38/624838/47/openstack-check/tripleo-build-containers-fedora-28/bf0f869/ : FAILURE in 35m 08s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"643ed668ff43296ff53d1371f483c384d6e3ba65","author":{"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},"date":"2019-03-14 15:58:19.000000000","message":"Patch Set 48: Patch Set 47 was rebased","accounts_in_message":[],"_revision_number":48},{"id":"923aa568d350b19126e0c84cd43859e78068687b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-14 16:33:19.000000000","message":"Patch Set 48:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 http://logs.rdoproject.org/38/624838/48/openstack-check/tripleo-build-containers-fedora-28/61eac3e/ : FAILURE in 32m 52s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"7b0c5775e2f4d426f0127ed03c3a7366ee900509","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-14 17:34:53.000000000","message":"Patch Set 48: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/48/check/kolla-build-centos-binary/01abf6a/ : SUCCESS in 40m 38s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/48/check/kolla-build-centos-source/536c865/ : SUCCESS in 50m 18s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/48/check/kolla-ansible-centos-source/6bdc293/ : SUCCESS in 1h 18m 39s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/48/check/kolla-ansible-centos-binary/facd649/ : SUCCESS in 1h 23m 37s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/48/check/tripleo-build-containers-centos-7/5659321/ : SUCCESS in 46m 32s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/48/check/kolla-ansible-bifrost-centos-source/73f36db/ : FAILURE in 39m 37s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/48/check/kolla-build-oraclelinux-binary/63ad8f7/ : SUCCESS in 46m 00s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/48/check/kolla-build-oraclelinux-source/c11a32a/ : SUCCESS in 45m 48s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/48/check/kolla-ansible-oraclelinux-source/03a5f46/ : SUCCESS in 1h 27m 31s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/48/check/kolla-ansible-oraclelinux-binary/61a292d/ : SUCCESS in 1h 12m 55s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/48/check/openstack-tox-lower-constraints/fc4223e/ : FAILURE in 4m 22s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/48/check/openstack-tox-pep8/0f71d7e/ : SUCCESS in 3m 20s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/48/check/openstack-tox-py27/71926b5/ : SUCCESS in 3m 25s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/48/check/openstack-tox-py36/8aa609e/ : SUCCESS in 3m 32s\n- openstack-tox-py37 http://logs.openstack.org/38/624838/48/check/openstack-tox-py37/391491f/ : SUCCESS in 3m 43s\n- openstack-tox-docs http://logs.openstack.org/38/624838/48/check/openstack-tox-docs/f9e3e0e/html/ : SUCCESS in 3m 56s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/48/check/kolla-build-ubuntu-binary/49edcfb/ : FAILURE in 1h 13m 02s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/48/check/kolla-build-ubuntu-source/81f3b6f/ : SUCCESS in 33m 24s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/48/check/kolla-ansible-ubuntu-source/e6aceba/ : FAILURE in 41m 33s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/48/check/kolla-ansible-ubuntu-binary/373a7b2/ : FAILURE in 20m 43s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"f76b896d9fecae37474346d6a01a9b2961f96bfb","author":{"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},"date":"2019-03-26 13:12:04.000000000","message":"Patch Set 49: Patch Set 48 was rebased","accounts_in_message":[],"_revision_number":49},{"id":"fe532a826aca20b5834b2fcea0597d26a13685bb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-26 13:52:52.000000000","message":"Patch Set 49:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 http://logs.rdoproject.org/38/624838/49/openstack-check/tripleo-build-containers-fedora-28/ab89e91/ : SUCCESS in 32m 40s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"54b0d3d467abac68eac5c23100a9d3a94f89a333","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-26 14:36:28.000000000","message":"Patch Set 49: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/49/check/kolla-build-centos-binary/9426f3e/ : SUCCESS in 37m 37s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/49/check/kolla-build-centos-source/2561c5a/ : FAILURE in 47m 47s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/49/check/kolla-ansible-centos-source/8ba9dec/ : FAILURE in 37m 01s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/49/check/kolla-ansible-centos-binary/9f77f48/ : SUCCESS in 1h 06m 40s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/49/check/tripleo-build-containers-centos-7/98eadcc/ : SUCCESS in 46m 12s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/49/check/kolla-ansible-bifrost-centos-source/9973044/ : FAILURE in 51m 04s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/49/check/kolla-build-oraclelinux-binary/203ac1f/ : SUCCESS in 1h 07m 30s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/49/check/kolla-build-oraclelinux-source/399cec6/ : FAILURE in 53m 17s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/49/check/kolla-ansible-oraclelinux-source/9494d36/ : FAILURE in 39m 48s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/49/check/kolla-ansible-oraclelinux-binary/9ec8b53/ : SUCCESS in 1h 15m 32s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/49/check/openstack-tox-lower-constraints/2390d95/ : SUCCESS in 3m 56s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/49/check/openstack-tox-pep8/e0e641c/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/49/check/openstack-tox-py27/b775635/ : SUCCESS in 3m 35s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/49/check/openstack-tox-py36/6b15ba0/ : SUCCESS in 3m 17s\n- openstack-tox-py37 http://logs.openstack.org/38/624838/49/check/openstack-tox-py37/03e9dd9/ : SUCCESS in 4m 07s\n- openstack-tox-docs http://logs.openstack.org/38/624838/49/check/openstack-tox-docs/8044b0b/html/ : SUCCESS in 2m 53s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/49/check/kolla-build-ubuntu-binary/aead19b/ : FAILURE in 31m 32s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/49/check/kolla-build-ubuntu-source/4098f73/ : FAILURE in 36m 48s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/49/check/kolla-ansible-ubuntu-source/acbf210/ : FAILURE in 29m 06s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/49/check/kolla-ansible-ubuntu-binary/1507843/ : FAILURE in 29m 14s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"4723c31d0959e29d27b20ed99f1eb9e688dcec9a","author":{"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},"date":"2019-04-15 16:44:20.000000000","message":"Patch Set 50: Patch Set 49 was rebased","accounts_in_message":[],"_revision_number":50},{"id":"0b668939eff852467e206bae26e0eedcc095e5d9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-15 17:22:04.000000000","message":"Patch Set 50:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/38/624838/50/openstack-check/tripleo-build-containers-fedora-28-master/cff5ddd/ : SUCCESS in 33m 29s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/38/624838/50/openstack-check/tripleo-build-containers-fedora-28-stein/e690370/ : SUCCESS in 34m 06s (non-voting)","accounts_in_message":[],"_revision_number":50},{"id":"45b24355107e7f522e66d884d408632b143ffb60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 18:36:53.000000000","message":"Patch Set 50: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/50/check/kolla-build-centos-binary/c65ad4f/ : SUCCESS in 36m 06s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/50/check/kolla-build-centos-source/28f0b6d/ : SUCCESS in 50m 32s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/50/check/kolla-ansible-centos-source/bc70483/ : SUCCESS in 1h 27m 40s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/50/check/kolla-ansible-centos-binary/095e3f3/ : SUCCESS in 1h 09m 42s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/38/624838/50/check/kolla-ansible-centos-source-upgrade/6e3aede/ : RETRY_LIMIT in 1m 02s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/50/check/tripleo-build-containers-centos-7/ed5ddae/ : SUCCESS in 52m 54s\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/50/check/kolla-ansible-bifrost-centos-source/8ae0d6f/ : SUCCESS in 1h 04m 30s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/50/check/kolla-build-oraclelinux-binary/acb629d/ : SUCCESS in 1h 03m 26s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/50/check/kolla-build-oraclelinux-source/4068255/ : SUCCESS in 1h 01m 00s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/50/check/kolla-ansible-oraclelinux-source/2245797/ : SUCCESS in 1h 09m 25s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/50/check/kolla-ansible-oraclelinux-binary/48d45ae/ : SUCCESS in 1h 19m 08s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/50/check/openstack-tox-lower-constraints/b8bbc57/ : SUCCESS in 4m 12s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/50/check/openstack-tox-pep8/fa911a8/ : SUCCESS in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/50/check/openstack-tox-py27/3446f45/ : SUCCESS in 3m 24s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/50/check/openstack-tox-py36/0663338/ : SUCCESS in 3m 47s\n- openstack-tox-py37 http://logs.openstack.org/38/624838/50/check/openstack-tox-py37/ec5ad78/ : SUCCESS in 3m 42s\n- openstack-tox-docs http://logs.openstack.org/38/624838/50/check/openstack-tox-docs/e37f199/html/ : SUCCESS in 3m 33s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/50/check/kolla-build-ubuntu-binary/53d9d94/ : SUCCESS in 47m 12s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/50/check/kolla-build-ubuntu-source/445a465/ : SUCCESS in 54m 52s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/50/check/kolla-ansible-ubuntu-source/161bf82/ : SUCCESS in 1h 04m 35s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/50/check/kolla-ansible-ubuntu-binary/04c33ff/ : FAILURE in 27m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/38/624838/50/check/kolla-ansible-ubuntu-source-upgrade/74eb224/ : SUCCESS in 59m 48s","accounts_in_message":[],"_revision_number":50},{"id":"92a8751d03110451f6c957a31dee2a83c998f02b","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-06-05 19:39:16.000000000","message":"Patch Set 50: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit ac0c434e16b92b807d0eb7c44737939ef8188acc","accounts_in_message":[],"_revision_number":50},{"id":"2ccfdf2f90e9347028b806eec68233157e804048","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-06 08:13:10.000000000","message":"Patch Set 50:\n\nPlease update to Fedora 30.\n\nF28 is EOL.","accounts_in_message":[],"_revision_number":50},{"id":"bcce584b74ceed1f3a206f2d4471a1cd5732eb84","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2019-06-06 09:20:32.000000000","message":"Patch Set 50:\n\nwe are actually going to target centos8 instead, as soon as it is available, bootstrap is in progress: https://wiki.centos.org/About/Building_8","accounts_in_message":[],"_revision_number":50},{"id":"535bc38e3d2f7a00dc5e7fb639ce4d07395e6d93","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-06-18 12:04:41.000000000","message":"Uploaded patch set 51.","accounts_in_message":[],"_revision_number":51},{"id":"a7d0726811ba62bde5c92d25943c9eb93fd61e95","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-06-18 12:07:27.000000000","message":"Patch Set 51: Code-Review+1\n\nLGTM, lets merge it as it is instead of further delays. We do have periodic jobs that needs this for both master and stein in order to\ntest python3 code.\n\nAs long it does not introduce any regressions I do not see why not to  merge it.","accounts_in_message":[],"_revision_number":51},{"id":"b3d5e39c818fc690b8adb3d5e84da25d4f0371ab","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-18 12:42:29.000000000","message":"Patch Set 51:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/38/624838/51/openstack-check/tripleo-build-containers-fedora-28-master/ea25ea7/ : SUCCESS in 36m 13s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/38/624838/51/openstack-check/tripleo-build-containers-fedora-28-stein/4de5084/ : SUCCESS in 33m 51s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"093f981c0b8dbe4052f4b283200078b6e62b7c8a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-18 13:36:56.000000000","message":"Patch Set 51:\n\nAre you hoping to release support for fedora images, or just include it temporarily for testing of python 3?","accounts_in_message":[],"_revision_number":51},{"id":"c13224f7faa376839abe22f4ebfeb8b81f8dbb27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 14:08:19.000000000","message":"Patch Set 51: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/51/check/kolla-build-centos-binary/cbfd397/ : RETRY_LIMIT in 4m 41s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/51/check/kolla-build-centos-source/03efa3c/ : SUCCESS in 47m 06s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/51/check/kolla-ansible-centos-source/74e86cf/ : SUCCESS in 1h 26m 12s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/51/check/kolla-ansible-centos-binary/0bb7ab1/ : SUCCESS in 1h 04m 43s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/38/624838/51/check/kolla-ansible-centos-source-upgrade/4d531e4/ : SUCCESS in 1h 22m 20s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/51/check/tripleo-build-containers-centos-7/2a2f3a6/ : SUCCESS in 38m 01s\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/51/check/kolla-ansible-bifrost-centos-source/63bf4e6/ : SUCCESS in 55m 13s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/624838/51/check/kolla-build-oraclelinux-binary/915af61/ : SUCCESS in 39m 26s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/624838/51/check/kolla-build-oraclelinux-source/38a3a3c/ : RETRY_LIMIT in 4m 45s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/624838/51/check/kolla-ansible-oraclelinux-source/aa640d1/ : SUCCESS in 1h 20m 45s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/624838/51/check/kolla-ansible-oraclelinux-binary/50fe872/ : SUCCESS in 1h 05m 51s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/51/check/openstack-tox-lower-constraints/77fbab5/ : SUCCESS in 3m 31s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/51/check/openstack-tox-pep8/ebc8fb4/ : SUCCESS in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/51/check/openstack-tox-py27/4bbb33b/ : SUCCESS in 5m 07s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/51/check/openstack-tox-py36/09be6bd/ : SUCCESS in 2m 59s\n- openstack-tox-py37 http://logs.openstack.org/38/624838/51/check/openstack-tox-py37/3ee91ea/ : SUCCESS in 3m 39s\n- openstack-tox-docs http://logs.openstack.org/38/624838/51/check/openstack-tox-docs/a23f833/html/ : SUCCESS in 4m 33s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/51/check/kolla-build-ubuntu-binary/ef88b9a/ : SUCCESS in 33m 27s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/51/check/kolla-build-ubuntu-source/8a64675/ : SUCCESS in 30m 42s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/51/check/kolla-ansible-ubuntu-source/2f77449/ : SUCCESS in 41m 32s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/51/check/kolla-ansible-ubuntu-binary/aa0a1af/ : SUCCESS in 54m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/38/624838/51/check/kolla-ansible-ubuntu-source-upgrade/be1c05f/ : SUCCESS in 1h 01m 10s","accounts_in_message":[],"_revision_number":51},{"id":"e937b764c3719f19b33f9377e10ad68b5174ce3e","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-06-18 18:43:33.000000000","message":"Uploaded patch set 52.","accounts_in_message":[],"_revision_number":52},{"id":"4fc32ed59c19f5c59bd6b1e3e2efd67ca0020c50","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-18 18:49:03.000000000","message":"Patch Set 52: Code-Review-1\n\n(2 comments)\n\nAlso repeating Mark\u0027s question:\n\u003e Are you hoping to release support for fedora images, or just include it temporarily for testing of python 3?","accounts_in_message":[],"_revision_number":52},{"id":"c1fd115772664cc0720b512b9a96918e00eff0b4","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-06-18 18:52:49.000000000","message":"Patch Set 52:\n\n(1 comment)","accounts_in_message":[],"_revision_number":52},{"id":"caff848a315bcbf9e490957d7bc08edd0f26f51a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-18 19:13:31.000000000","message":"Patch Set 53: Patch Set 52 was rebased","accounts_in_message":[],"_revision_number":53},{"id":"d3ac6f183ef5485bd468b4884fe8ecddfd98999f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-18 19:15:14.000000000","message":"Patch Set 53: Workflow-1\n\nHonestly no, the fedora28 is supposed to just be our testing for python3 until we get centos8. once we have centos8 we don\u0027t need to carry fedora support.  We are currently using this patch (and the stable/stein) patch for the 3rd party jobs which is why we need them available but not necessarily merged. If kolla folks would like fedora support we can clean this up to land, but since it was previously removed and unlikely to get long term attention, it\u0027s probably not ideal to try and merge these.","accounts_in_message":[],"_revision_number":53},{"id":"5d13207ec830df89625fd43fc3a6119473d325c0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-18 19:21:46.000000000","message":"Patch Set 53: -Code-Review\n\nThanks, Alex. Workflow resets each time so maybe add DNM to commit message so that we no longer come here complaining. :-)","accounts_in_message":[],"_revision_number":53},{"id":"6dc8873e1797effd8016ef458ccc577281abd591","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-18 19:49:24.000000000","message":"Patch Set 53:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/38/624838/53/openstack-check/tripleo-build-containers-fedora-28-master/fcd519c/ : SUCCESS in 34m 44s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/38/624838/53/openstack-check/tripleo-build-containers-fedora-28-stein/45875c5/ : SUCCESS in 32m 08s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"12c62fd1d2038b57c833f393a4fae4ced4e1b36d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 22:20:14.000000000","message":"Patch Set 53: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/53/check/kolla-build-centos-binary/ae81325/ : SUCCESS in 46m 59s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/53/check/kolla-build-centos-source/fb0e930/ : SUCCESS in 37m 41s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/53/check/kolla-ansible-centos-source/08537ac/ : SUCCESS in 1h 05m 20s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/53/check/kolla-ansible-centos-binary/fd58fd5/ : SUCCESS in 1h 04m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/38/624838/53/check/kolla-ansible-centos-source-upgrade/61eaa00/ : SUCCESS in 1h 15m 16s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/53/check/tripleo-build-containers-centos-7/1b19eb5/ : SUCCESS in 40m 42s\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/53/check/kolla-ansible-bifrost-centos-source/144d758/ : SUCCESS in 41m 27s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/53/check/openstack-tox-lower-constraints/183085f/ : SUCCESS in 5m 04s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/53/check/openstack-tox-pep8/891a36d/ : SUCCESS in 3m 18s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/53/check/openstack-tox-py27/9322583/ : SUCCESS in 3m 47s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/53/check/openstack-tox-py36/d70e5d3/ : SUCCESS in 3m 22s\n- openstack-tox-py37 http://logs.openstack.org/38/624838/53/check/openstack-tox-py37/36cf751/ : SUCCESS in 3m 41s\n- openstack-tox-docs http://logs.openstack.org/38/624838/53/check/openstack-tox-docs/ec9a969/html/ : SUCCESS in 2m 38s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/53/check/kolla-build-ubuntu-binary/c6e908a/ : SUCCESS in 23m 33s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/53/check/kolla-build-ubuntu-source/b1f1995/ : SUCCESS in 31m 44s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/53/check/kolla-ansible-ubuntu-source/1d7a3c1/ : SUCCESS in 44m 35s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/53/check/kolla-ansible-ubuntu-binary/ef4132d/ : SUCCESS in 45m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/38/624838/53/check/kolla-ansible-ubuntu-source-upgrade/8d21c5b/ : SUCCESS in 56m 44s","accounts_in_message":[],"_revision_number":53},{"id":"1d10dd3c132d47b31af13c755153cb88fa9f4966","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-19 08:16:44.000000000","message":"Patch Set 53: Code-Review-2\n\nAdding -2 to indicate this will not be merged for the time being.","accounts_in_message":[],"_revision_number":53},{"id":"e1d1257103f39974b553c783fc56f8d414c32f66","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-07-15 14:08:13.000000000","message":"Uploaded patch set 54.","accounts_in_message":[],"_revision_number":54},{"id":"153b7ddc0780c66e8a66b0009d152b586e7434a8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-15 14:56:24.000000000","message":"Patch Set 54:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/38/624838/54/openstack-check/tripleo-build-containers-fedora-28-master/8a10cde/ : FAILURE in 5m 36s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/38/624838/54/openstack-check/tripleo-build-containers-fedora-28-stein/17b4be6/ : SUCCESS in 40m 08s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"6cf982059b8c7fc9fc0b040c3f32e153fae61324","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 16:48:42.000000000","message":"Patch Set 54: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/38/624838/54/check/kolla-build-centos-binary/0323594/ : SUCCESS in 43m 57s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/624838/54/check/kolla-build-centos-source/0b5463d/ : SUCCESS in 55m 27s\n- kolla-ansible-centos-source http://logs.openstack.org/38/624838/54/check/kolla-ansible-centos-source/34234bb/ : SUCCESS in 1h 11m 08s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/624838/54/check/kolla-ansible-centos-binary/0029179/ : SUCCESS in 1h 06m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/38/624838/54/check/kolla-ansible-centos-source-upgrade/cdcc54f/ : SUCCESS in 1h 34m 56s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/624838/54/check/tripleo-build-containers-centos-7/265ee6a/ : SUCCESS in 53m 02s\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/38/624838/54/check/kolla-ansible-bifrost-centos-source/71130e6/ : SUCCESS in 55m 22s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/38/624838/54/check/kolla-ansible-centos-source-ceph/25c2e93/ : SUCCESS in 1h 47m 35s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph http://logs.openstack.org/38/624838/54/check/kolla-ansible-centos-source-upgrade-ceph/0a0635c/ : SUCCESS in 1h 55m 28s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/624838/54/check/openstack-tox-lower-constraints/9283f23/ : SUCCESS in 5m 01s\n- openstack-tox-pep8 http://logs.openstack.org/38/624838/54/check/openstack-tox-pep8/7826060/ : SUCCESS in 7m 32s\n- openstack-tox-py27 http://logs.openstack.org/38/624838/54/check/openstack-tox-py27/ee430f3/ : SUCCESS in 4m 12s\n- openstack-tox-py36 http://logs.openstack.org/38/624838/54/check/openstack-tox-py36/7c2d58c/ : SUCCESS in 4m 36s\n- openstack-tox-py37 http://logs.openstack.org/38/624838/54/check/openstack-tox-py37/00c474e/ : SUCCESS in 3m 29s\n- openstack-tox-docs http://logs.openstack.org/38/624838/54/check/openstack-tox-docs/73dcf06/html/ : SUCCESS in 4m 25s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/624838/54/check/kolla-build-ubuntu-binary/5a322e2/ : SUCCESS in 42m 43s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/624838/54/check/kolla-build-ubuntu-source/58752ab/ : SUCCESS in 27m 18s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/624838/54/check/kolla-ansible-ubuntu-source/8bd5643/ : SUCCESS in 54m 44s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/624838/54/check/kolla-ansible-ubuntu-binary/1a0ce2c/ : SUCCESS in 1h 03m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/38/624838/54/check/kolla-ansible-ubuntu-source-upgrade/56bc489/ : SUCCESS in 1h 17m 49s\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/38/624838/54/check/kolla-ansible-ubuntu-source-ceph/efb73ed/ : SUCCESS in 1h 04m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph http://logs.openstack.org/38/624838/54/check/kolla-ansible-ubuntu-source-upgrade-ceph/82bf652/ : TIMED_OUT in 2h 32m 10s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"ef257b821d1c0497b5bc47a19c6e5a3efcb82ad0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-20 21:04:03.000000000","message":"Abandoned\n\nwe\u0027ve got rhel8 spun up in preparation for centos8. we aren\u0027t testing fedora anymore","accounts_in_message":[],"_revision_number":54}],"current_revision_number":54,"current_revision":"b39180b1bf1085115f7464e12c81b4f9d05c0496","revisions":{"2c28eee8d737c6a7cd6f63a27b934e6d06c8e222":{"kind":"REWORK","_number":1,"created":"2018-12-13 00:16:54.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/1"}}},"commit":{"parents":[{"commit":"32beb0a642b10c2dfe7443e6adda0687d38029c7","subject":"Required changes for mariadb 10.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32beb0a642b10c2dfe7443e6adda0687d38029c7"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-12-07 00:12:57.000000000","tz":0},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-12-13 00:16:01.000000000","tz":-420},"subject":"WIP: Introduce RHEL based python3 support","message":"WIP: Introduce RHEL based python3 support\n\nCurrently RDO is working through the packaging for python3 under RHEL\nbased systems using Fedora.  This patch stubs out some of the dnf and\npackage related work to support this.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c28eee8d737c6a7cd6f63a27b934e6d06c8e222"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c28eee8d737c6a7cd6f63a27b934e6d06c8e222"}]},"branch":"refs/heads/master"},"2925522e87e4fa39fd7a81993908f073a90034f4":{"kind":"REWORK","_number":2,"created":"2019-01-08 22:03:34.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/2"}}},"commit":{"parents":[{"commit":"f46b1c3b084c15757a93d61d9a4b95be9edb8672","subject":"Merge \"Update hacking version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f46b1c3b084c15757a93d61d9a4b95be9edb8672"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-12-07 00:12:57.000000000","tz":0},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-08 22:03:19.000000000","tz":-420},"subject":"WIP: Introduce RHEL based python3 support","message":"WIP: Introduce RHEL based python3 support\n\nCurrently RDO is working through the packaging for python3 under RHEL\nbased systems using Fedora.  This patch stubs out some of the dnf and\npackage related work to support this.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2925522e87e4fa39fd7a81993908f073a90034f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2925522e87e4fa39fd7a81993908f073a90034f4"}]},"branch":"refs/heads/master"},"2a50c2b10778fd6f37f4c3eb7034b6847eead99f":{"kind":"REWORK","_number":3,"created":"2019-01-09 23:00:46.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/3"}}},"commit":{"parents":[{"commit":"96cb18bd0197287b3fd5f41c7e7caae27e03cc4b","subject":"Update the remain hacking version for lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96cb18bd0197287b3fd5f41c7e7caae27e03cc4b"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-12-07 00:12:57.000000000","tz":0},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-09 23:00:29.000000000","tz":-420},"subject":"WIP: Introduce RHEL based python3 support","message":"WIP: Introduce RHEL based python3 support\n\nCurrently RDO is working through the packaging for python3 under RHEL\nbased systems using Fedora.  This patch stubs out some of the dnf and\npackage related work to support this.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a50c2b10778fd6f37f4c3eb7034b6847eead99f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a50c2b10778fd6f37f4c3eb7034b6847eead99f"}]},"branch":"refs/heads/master"},"30419c5260d6331605560af50e584886c0dbaffe":{"kind":"REWORK","_number":4,"created":"2019-01-14 15:27:50.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/4"}}},"commit":{"parents":[{"commit":"c2397e62d86773c243753cab32183b43831e7ac0","subject":"Add python2-sqlalchemy-collectd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c2397e62d86773c243753cab32183b43831e7ac0"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-12-07 00:12:57.000000000","tz":0},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-14 15:27:47.000000000","tz":-420},"subject":"WIP: Introduce RHEL based python3 support","message":"WIP: Introduce RHEL based python3 support\n\nCurrently RDO is working through the packaging for python3 under RHEL\nbased systems using Fedora.  This patch stubs out some of the dnf and\npackage related work to support this.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30419c5260d6331605560af50e584886c0dbaffe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30419c5260d6331605560af50e584886c0dbaffe"}]},"branch":"refs/heads/master"},"c31e92e16f3e3616319b3bbd78a9a6635c840a05":{"kind":"REWORK","_number":5,"created":"2019-01-16 00:16:08.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/5"}}},"commit":{"parents":[{"commit":"4366b2aada7ce45e33709186286ca2841732cce7","subject":"Use base_packaging_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4366b2aada7ce45e33709186286ca2841732cce7"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-12-07 00:12:57.000000000","tz":0},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-16 00:16:02.000000000","tz":-420},"subject":"WIP: Introduce RHEL based python3 support","message":"WIP: Introduce RHEL based python3 support\n\nCurrently RDO is working through the packaging for python3 under RHEL\nbased systems using Fedora.  This patch stubs out some of the dnf and\npackage related work to support this.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c31e92e16f3e3616319b3bbd78a9a6635c840a05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c31e92e16f3e3616319b3bbd78a9a6635c840a05"}]},"branch":"refs/heads/master"},"17b5dd3993848eee74934b79f903463a1665d834":{"kind":"REWORK","_number":6,"created":"2019-01-16 00:23:13.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/6"}}},"commit":{"parents":[{"commit":"4366b2aada7ce45e33709186286ca2841732cce7","subject":"Use base_packaging_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4366b2aada7ce45e33709186286ca2841732cce7"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-12-07 00:12:57.000000000","tz":0},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-16 00:23:09.000000000","tz":-420},"subject":"WIP: Introduce RHEL based python3 support","message":"WIP: Introduce RHEL based python3 support\n\nCurrently RDO is working through the packaging for python3 under RHEL\nbased systems using Fedora.  This patch stubs out some of the dnf and\npackage related work to support this.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17b5dd3993848eee74934b79f903463a1665d834"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17b5dd3993848eee74934b79f903463a1665d834"}]},"branch":"refs/heads/master"},"9c6ddd3ddf992ad9eb92ab783c3b05e02d7b962d":{"kind":"REWORK","_number":7,"created":"2019-01-16 15:48:33.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/7"}}},"commit":{"parents":[{"commit":"a43da2d9c2ccb01316ac51c007dd956750c1e80c","subject":"Use base_package_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a43da2d9c2ccb01316ac51c007dd956750c1e80c"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-12-07 00:12:57.000000000","tz":0},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-16 15:48:27.000000000","tz":-420},"subject":"WIP: Introduce RHEL based python3 support","message":"WIP: Introduce RHEL based python3 support\n\nCurrently RDO is working through the packaging for python3 under RHEL\nbased systems using Fedora.  This patch stubs out some of the dnf and\npackage related work to support this.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c6ddd3ddf992ad9eb92ab783c3b05e02d7b962d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c6ddd3ddf992ad9eb92ab783c3b05e02d7b962d"}]},"branch":"refs/heads/master"},"1b9bca1cc854f8dc6e3680dd5f8fc7babac9f866":{"kind":"REWORK","_number":8,"created":"2019-01-16 15:51:18.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/8"}}},"commit":{"parents":[{"commit":"a43da2d9c2ccb01316ac51c007dd956750c1e80c","subject":"Use base_package_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a43da2d9c2ccb01316ac51c007dd956750c1e80c"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-12-07 00:12:57.000000000","tz":0},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-16 15:51:05.000000000","tz":-420},"subject":"WIP: Introduce RHEL based python3 support","message":"WIP: Introduce RHEL based python3 support\n\nCurrently RDO is working through the packaging for python3 under RHEL\nbased systems using Fedora.  This patch stubs out some of the dnf and\npackage related work to support this.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b9bca1cc854f8dc6e3680dd5f8fc7babac9f866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b9bca1cc854f8dc6e3680dd5f8fc7babac9f866"}]},"branch":"refs/heads/master"},"6b9bedd6ec106cb76a93a21d54d276c0eba48eb2":{"kind":"REWORK","_number":9,"created":"2019-01-16 17:11:40.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/9"}}},"commit":{"parents":[{"commit":"a43da2d9c2ccb01316ac51c007dd956750c1e80c","subject":"Use base_package_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a43da2d9c2ccb01316ac51c007dd956750c1e80c"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-12-07 00:12:57.000000000","tz":0},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-16 16:46:29.000000000","tz":-420},"subject":"WIP: Introduce RHEL based python3 support","message":"WIP: Introduce RHEL based python3 support\n\nCurrently RDO is working through the packaging for python3 under RHEL\nbased systems using Fedora.  This patch stubs out some of the dnf and\npackage related work to support this.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b9bedd6ec106cb76a93a21d54d276c0eba48eb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b9bedd6ec106cb76a93a21d54d276c0eba48eb2"}]},"branch":"refs/heads/master"},"df5d676e63b8044497783fbb1a45c1cd12d59567":{"kind":"NO_CHANGE","_number":10,"created":"2019-01-16 17:29:37.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/10"}}},"commit":{"parents":[{"commit":"0851ca54e645698b5c2e1239db20fb1d25ff9d73","subject":"Use base_package_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0851ca54e645698b5c2e1239db20fb1d25ff9d73"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-12-07 00:12:57.000000000","tz":0},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-16 17:29:37.000000000","tz":0},"subject":"WIP: Introduce RHEL based python3 support","message":"WIP: Introduce RHEL based python3 support\n\nCurrently RDO is working through the packaging for python3 under RHEL\nbased systems using Fedora.  This patch stubs out some of the dnf and\npackage related work to support this.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df5d676e63b8044497783fbb1a45c1cd12d59567"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df5d676e63b8044497783fbb1a45c1cd12d59567"}]},"branch":"refs/heads/master"},"0c8cdf72614185ae2b279de6513f369229698063":{"kind":"REWORK","_number":11,"created":"2019-01-16 18:01:37.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/11"}}},"commit":{"parents":[{"commit":"a43da2d9c2ccb01316ac51c007dd956750c1e80c","subject":"Use base_package_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a43da2d9c2ccb01316ac51c007dd956750c1e80c"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-12-07 00:12:57.000000000","tz":0},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-16 18:01:29.000000000","tz":-420},"subject":"WIP: Introduce RHEL based python3 support","message":"WIP: Introduce RHEL based python3 support\n\nCurrently RDO is working through the packaging for python3 under RHEL\nbased systems using Fedora.  This patch stubs out some of the dnf and\npackage related work to support this.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c8cdf72614185ae2b279de6513f369229698063"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c8cdf72614185ae2b279de6513f369229698063"}]},"branch":"refs/heads/master"},"94049d5905492cf1e0546ddb5b34d28254fd325e":{"kind":"NO_CHANGE","_number":12,"created":"2019-01-16 19:00:11.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/12"}}},"commit":{"parents":[{"commit":"0851ca54e645698b5c2e1239db20fb1d25ff9d73","subject":"Use base_package_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0851ca54e645698b5c2e1239db20fb1d25ff9d73"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-12-07 00:12:57.000000000","tz":0},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-16 19:00:11.000000000","tz":0},"subject":"WIP: Introduce RHEL based python3 support","message":"WIP: Introduce RHEL based python3 support\n\nCurrently RDO is working through the packaging for python3 under RHEL\nbased systems using Fedora.  This patch stubs out some of the dnf and\npackage related work to support this.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/94049d5905492cf1e0546ddb5b34d28254fd325e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/94049d5905492cf1e0546ddb5b34d28254fd325e"}]},"branch":"refs/heads/master"},"504f118989cb31beb0336ff18da60c0a2694d2db":{"kind":"REWORK","_number":13,"created":"2019-01-17 17:58:59.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/13"}}},"commit":{"parents":[{"commit":"ae1322ec10861678abe1137f3013030b34335bc7","subject":"Use base_package_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae1322ec10861678abe1137f3013030b34335bc7"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-12-07 00:12:57.000000000","tz":0},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-17 17:58:38.000000000","tz":-420},"subject":"WIP: Introduce RHEL based python3 support","message":"WIP: Introduce RHEL based python3 support\n\nCurrently RDO is working through the packaging for python3 under RHEL\nbased systems using Fedora.  This patch stubs out some of the dnf and\npackage related work to support this.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/504f118989cb31beb0336ff18da60c0a2694d2db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/504f118989cb31beb0336ff18da60c0a2694d2db"}]},"branch":"refs/heads/master"},"367e5ceaef080c8f491eb6572ff6ff9d204d5628":{"kind":"REWORK","_number":14,"created":"2019-01-17 18:42:24.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/14"}}},"commit":{"parents":[{"commit":"ae1322ec10861678abe1137f3013030b34335bc7","subject":"Use base_package_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae1322ec10861678abe1137f3013030b34335bc7"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-12-07 00:12:57.000000000","tz":0},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-17 18:15:32.000000000","tz":-420},"subject":"WIP: Introduce RHEL based python3 support","message":"WIP: Introduce RHEL based python3 support\n\nCurrently RDO is working through the packaging for python3 under RHEL\nbased systems using Fedora.  This patch stubs out some of the dnf and\npackage related work to support this.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/367e5ceaef080c8f491eb6572ff6ff9d204d5628"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/367e5ceaef080c8f491eb6572ff6ff9d204d5628"}]},"branch":"refs/heads/master"},"08be851fc8d5e357b238bb1eaf8422ecebf576ec":{"kind":"REWORK","_number":15,"created":"2019-01-17 19:20:38.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/15"}}},"commit":{"parents":[{"commit":"ae1322ec10861678abe1137f3013030b34335bc7","subject":"Use base_package_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae1322ec10861678abe1137f3013030b34335bc7"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-12-07 00:12:57.000000000","tz":0},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-17 19:20:32.000000000","tz":-420},"subject":"WIP: Introduce RHEL based python3 support","message":"WIP: Introduce RHEL based python3 support\n\nCurrently RDO is working through the packaging for python3 under RHEL\nbased systems using Fedora.  This patch stubs out some of the dnf and\npackage related work to support this.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08be851fc8d5e357b238bb1eaf8422ecebf576ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08be851fc8d5e357b238bb1eaf8422ecebf576ec"}]},"branch":"refs/heads/master"},"c36e3dad3e4b5e38b27012eeb12dfb223798c4ac":{"kind":"REWORK","_number":16,"created":"2019-01-17 21:47:43.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/16"}}},"commit":{"parents":[{"commit":"ae1322ec10861678abe1137f3013030b34335bc7","subject":"Use base_package_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae1322ec10861678abe1137f3013030b34335bc7"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-12-07 00:12:57.000000000","tz":0},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-17 21:46:32.000000000","tz":-420},"subject":"WIP: Introduce RHEL based python3 support","message":"WIP: Introduce RHEL based python3 support\n\nCurrently RDO is working through the packaging for python3 under RHEL\nbased systems using Fedora.  This patch stubs out some of the dnf and\npackage related work to support this.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c36e3dad3e4b5e38b27012eeb12dfb223798c4ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c36e3dad3e4b5e38b27012eeb12dfb223798c4ac"}]},"branch":"refs/heads/master"},"2f894fca2a10c34a5f9c416ccd4a1df4792905b6":{"kind":"REWORK","_number":17,"created":"2019-01-18 00:56:20.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/17"}}},"commit":{"parents":[{"commit":"ae1322ec10861678abe1137f3013030b34335bc7","subject":"Use base_package_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae1322ec10861678abe1137f3013030b34335bc7"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-12-07 00:12:57.000000000","tz":0},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-17 23:54:30.000000000","tz":-420},"subject":"WIP: Introduce RHEL based python3 support","message":"WIP: Introduce RHEL based python3 support\n\nCurrently RDO is working through the packaging for python3 under RHEL\nbased systems using Fedora.  This patch stubs out some of the dnf and\npackage related work to support this.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2f894fca2a10c34a5f9c416ccd4a1df4792905b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2f894fca2a10c34a5f9c416ccd4a1df4792905b6"}]},"branch":"refs/heads/master"},"586ac7b8b987ac01b3276bd3749ac1d0bd7aa7c7":{"kind":"REWORK","_number":18,"created":"2019-01-21 20:02:42.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/18"}}},"commit":{"parents":[{"commit":"e8b7e532a5880150bbb3203cfaaebed09e97ec98","subject":"Add python3 packages for RHEL systems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e8b7e532a5880150bbb3203cfaaebed09e97ec98"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:02:19.000000000","tz":-420},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/586ac7b8b987ac01b3276bd3749ac1d0bd7aa7c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/586ac7b8b987ac01b3276bd3749ac1d0bd7aa7c7"}]},"branch":"refs/heads/master"},"ef4190305e8dabae58c55c4b2660a437745a6efd":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2019-01-24 15:25:29.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/19"}}},"commit":{"parents":[{"commit":"cc27416e57401624d7ca17a3f15548734a2de8e9","subject":"Add python3 packages for RHEL systems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc27416e57401624d7ca17a3f15548734a2de8e9"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-24 15:25:29.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef4190305e8dabae58c55c4b2660a437745a6efd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef4190305e8dabae58c55c4b2660a437745a6efd"}]},"branch":"refs/heads/master"},"45ca8ab33db1bada2ec8dd46aa0228d5d3a92fed":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2019-01-28 17:15:31.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/20"}}},"commit":{"parents":[{"commit":"2c4e1b016f92aa11827cda412cc09caa0e6c7112","subject":"Add python3 packages for RHEL systems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c4e1b016f92aa11827cda412cc09caa0e6c7112"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-28 17:15:31.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45ca8ab33db1bada2ec8dd46aa0228d5d3a92fed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45ca8ab33db1bada2ec8dd46aa0228d5d3a92fed"}]},"branch":"refs/heads/master"},"6bb282ee4658f11f91b6523a8aecdcf9153fd8cf":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2019-01-29 14:46:07.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/21"}}},"commit":{"parents":[{"commit":"e6e7cc44261105980df09d8da2fdcbc73608840e","subject":"Add python3 packages for RHEL systems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6e7cc44261105980df09d8da2fdcbc73608840e"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-29 14:46:07.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6bb282ee4658f11f91b6523a8aecdcf9153fd8cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6bb282ee4658f11f91b6523a8aecdcf9153fd8cf"}]},"branch":"refs/heads/master"},"7d2c275cda5309b43397b13c3c7ded472a8327d8":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2019-01-31 15:57:57.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/22"}}},"commit":{"parents":[{"commit":"c0878e0931d65fbf3f7438191ce74a47d530e8c6","subject":"Add python3 packages for RHEL systems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0878e0931d65fbf3f7438191ce74a47d530e8c6"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-31 15:57:57.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7d2c275cda5309b43397b13c3c7ded472a8327d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7d2c275cda5309b43397b13c3c7ded472a8327d8"}]},"branch":"refs/heads/master"},"0adb5a1843524393de8c240d6b5497fca34251e1":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2019-01-31 19:24:48.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/23"}}},"commit":{"parents":[{"commit":"0b3c876d46038601b2396f3e595877416338736b","subject":"Add python3 packages for RHEL systems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b3c876d46038601b2396f3e595877416338736b"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-31 19:24:48.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0adb5a1843524393de8c240d6b5497fca34251e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0adb5a1843524393de8c240d6b5497fca34251e1"}]},"branch":"refs/heads/master"},"bf396f9c82c020c8adb5d0b1c75647305a4526a1":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2019-02-11 15:09:47.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/38/624838/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/24"}}},"commit":{"parents":[{"commit":"66410ceaad40505759b49a968b0c65a1803423cd","subject":"Add python3 packages for RHEL systems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/66410ceaad40505759b49a968b0c65a1803423cd"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-02-11 15:07:47.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf396f9c82c020c8adb5d0b1c75647305a4526a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf396f9c82c020c8adb5d0b1c75647305a4526a1"}]},"branch":"refs/heads/master"},"eab05abf10436eabf71deed14a83f491b8a67c18":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2019-02-12 18:08:46.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/25"}}},"commit":{"parents":[{"commit":"679447f8fd8b29db6558055851f9a7b8a3f4cc65","subject":"Add python3 packages for RHEL systems and non-voting fedora job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/679447f8fd8b29db6558055851f9a7b8a3f4cc65"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-02-12 18:08:46.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eab05abf10436eabf71deed14a83f491b8a67c18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eab05abf10436eabf71deed14a83f491b8a67c18"}]},"branch":"refs/heads/master"},"4fdaf73cbe3619c93731347ed1ecd4052f42afe1":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2019-02-12 23:43:28.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/26"}}},"commit":{"parents":[{"commit":"d2ddb0f76ad154207c983e119dfb62f0b9838519","subject":"Add python3 packages for RHEL systems and non-voting fedora job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2ddb0f76ad154207c983e119dfb62f0b9838519"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-02-12 23:43:28.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4fdaf73cbe3619c93731347ed1ecd4052f42afe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4fdaf73cbe3619c93731347ed1ecd4052f42afe1"}]},"branch":"refs/heads/master"},"72ca324fb457c13d0c6c158b7ceab82c17c62e65":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2019-02-14 23:20:06.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/27"}}},"commit":{"parents":[{"commit":"f782ec61568707b16fca9daab61e232ec632c8d5","subject":"Add python3 packages for RHEL systems and non-voting fedora job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f782ec61568707b16fca9daab61e232ec632c8d5"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-02-14 23:20:06.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72ca324fb457c13d0c6c158b7ceab82c17c62e65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72ca324fb457c13d0c6c158b7ceab82c17c62e65"}]},"branch":"refs/heads/master"},"1e058278f98180812043d4caa3975d9f98450135":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2019-02-18 17:53:00.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/28"}}},"commit":{"parents":[{"commit":"b90adf97788d3e2391929da8443790cc287d2ca8","subject":"Add python3 packages for RHEL systems and non-voting fedora job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b90adf97788d3e2391929da8443790cc287d2ca8"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-02-18 17:53:00.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e058278f98180812043d4caa3975d9f98450135"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e058278f98180812043d4caa3975d9f98450135"}]},"branch":"refs/heads/master"},"8dc4755a6544ddf91d04c5fa488a8278b353b6bd":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2019-02-19 15:41:42.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/29"}}},"commit":{"parents":[{"commit":"d8137c0cbc40fb3a1eb9f6ca3333cc7386e2ffb2","subject":"Add python3 packages for RHEL systems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8137c0cbc40fb3a1eb9f6ca3333cc7386e2ffb2"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-02-19 15:41:42.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8dc4755a6544ddf91d04c5fa488a8278b353b6bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8dc4755a6544ddf91d04c5fa488a8278b353b6bd"}]},"branch":"refs/heads/master"},"0789fd41180ced7d9569d70b538153bd945d0a00":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2019-02-21 13:44:40.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/38/624838/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/30"}}},"commit":{"parents":[{"commit":"44aa4252e8080992fb6d898e629c73b1d0b475c8","subject":"Add python3 packages for RHEL systems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44aa4252e8080992fb6d898e629c73b1d0b475c8"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-02-21 13:44:40.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0789fd41180ced7d9569d70b538153bd945d0a00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0789fd41180ced7d9569d70b538153bd945d0a00"}]},"branch":"refs/heads/master"},"cc151c5ad213bdfb6725a1e146c82e3db08aabf0":{"kind":"REWORK","_number":31,"created":"2019-02-25 23:18:51.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/31"}}},"commit":{"parents":[{"commit":"476ab5e530c7c18948b6f456bd6857c6c2fc49d7","subject":"Add distro_package_manager configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/476ab5e530c7c18948b6f456bd6857c6c2fc49d7"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-02-25 23:18:29.000000000","tz":-420},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc151c5ad213bdfb6725a1e146c82e3db08aabf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc151c5ad213bdfb6725a1e146c82e3db08aabf0"}]},"branch":"refs/heads/master"},"a09bc90769c4c976b2eb79e4dc4f82275ee11ef9":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2019-02-25 23:30:40.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/32"}}},"commit":{"parents":[{"commit":"94f6ee09e060b5d4e3cde86c642d60db85cb2bb4","subject":"Add distro_package_manager configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/94f6ee09e060b5d4e3cde86c642d60db85cb2bb4"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-02-25 23:30:31.000000000","tz":-420},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a09bc90769c4c976b2eb79e4dc4f82275ee11ef9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a09bc90769c4c976b2eb79e4dc4f82275ee11ef9"}]},"branch":"refs/heads/master"},"3db9384c82e69437d999bce759c199cfb3bac4a8":{"kind":"REWORK","_number":33,"created":"2019-02-26 18:30:45.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/33"}}},"commit":{"parents":[{"commit":"96cf4f698ab220e4167175a29e0139b29ed2f505","subject":"Add distro_package_manager configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96cf4f698ab220e4167175a29e0139b29ed2f505"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-02-26 18:30:16.000000000","tz":-420},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3db9384c82e69437d999bce759c199cfb3bac4a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3db9384c82e69437d999bce759c199cfb3bac4a8"}]},"branch":"refs/heads/master"},"e7f226b1da03df33a277d28b44fdcc3ece75e699":{"kind":"REWORK","_number":34,"created":"2019-02-27 08:45:53.000000000","uploader":{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},"ref":"refs/changes/38/624838/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/34"}}},"commit":{"parents":[{"commit":"96cf4f698ab220e4167175a29e0139b29ed2f505","subject":"Add distro_package_manager configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96cf4f698ab220e4167175a29e0139b29ed2f505"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Quique Llorente","email":"ellorent@redhat.com","date":"2019-02-27 08:45:24.000000000","tz":60},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7f226b1da03df33a277d28b44fdcc3ece75e699"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7f226b1da03df33a277d28b44fdcc3ece75e699"}]},"branch":"refs/heads/master"},"5962d62a3550faa6ccc6e839efb29d89f567a2f8":{"kind":"TRIVIAL_REBASE","_number":35,"created":"2019-02-27 17:32:17.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/35"}}},"commit":{"parents":[{"commit":"d961c0a1aff121a1e94464695c45873a6612baeb","subject":"Add distro_package_manager configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d961c0a1aff121a1e94464695c45873a6612baeb"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-02-27 17:32:06.000000000","tz":-420},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5962d62a3550faa6ccc6e839efb29d89f567a2f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5962d62a3550faa6ccc6e839efb29d89f567a2f8"}]},"branch":"refs/heads/master"},"be3d98978e4b2523eeed55c38434486b8d889fe1":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2019-03-01 21:51:04.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/36"}}},"commit":{"parents":[{"commit":"c46467a6732363adbd822a308fed73e85e617c35","subject":"Add distro_package_manager configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c46467a6732363adbd822a308fed73e85e617c35"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-03-01 21:51:04.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be3d98978e4b2523eeed55c38434486b8d889fe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be3d98978e4b2523eeed55c38434486b8d889fe1"}]},"branch":"refs/heads/master"},"82d70a27c6da483f9e346553584304757530d3dc":{"kind":"TRIVIAL_REBASE","_number":37,"created":"2019-03-06 13:59:47.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/38/624838/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/37"}}},"commit":{"parents":[{"commit":"1d627128175a5dc14919c00ce62ec9c9dc5c1440","subject":"Add distro_package_manager configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d627128175a5dc14919c00ce62ec9c9dc5c1440"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2019-03-06 13:58:16.000000000","tz":120},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82d70a27c6da483f9e346553584304757530d3dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82d70a27c6da483f9e346553584304757530d3dc"}]},"branch":"refs/heads/master"},"955f10eb3d766d704166678509f43a16c991416c":{"kind":"TRIVIAL_REBASE","_number":38,"created":"2019-03-06 15:03:46.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/38"}}},"commit":{"parents":[{"commit":"42826f1d2178bf1f4cc06be8fe54828d93df12ff","subject":"Add distro_package_manager configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42826f1d2178bf1f4cc06be8fe54828d93df12ff"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-03-06 15:03:46.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/955f10eb3d766d704166678509f43a16c991416c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/955f10eb3d766d704166678509f43a16c991416c"}]},"branch":"refs/heads/master"},"d37b77cc72102ca85a726457585d8e8578f4855c":{"kind":"TRIVIAL_REBASE","_number":39,"created":"2019-03-06 17:26:23.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/39"}}},"commit":{"parents":[{"commit":"6246daac7c745a305c98e830a40b37161b62d2be","subject":"Add distro_package_manager configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6246daac7c745a305c98e830a40b37161b62d2be"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-03-06 17:26:23.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d37b77cc72102ca85a726457585d8e8578f4855c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d37b77cc72102ca85a726457585d8e8578f4855c"}]},"branch":"refs/heads/master"},"4abd7c03f56c36ec63411551c68bedeae25e1f7a":{"kind":"TRIVIAL_REBASE","_number":40,"created":"2019-03-06 18:00:41.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/40"}}},"commit":{"parents":[{"commit":"046bc7be76844d660751d16534a68451d4096865","subject":"Add distro_package_manager configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/046bc7be76844d660751d16534a68451d4096865"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-03-06 18:00:41.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4abd7c03f56c36ec63411551c68bedeae25e1f7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4abd7c03f56c36ec63411551c68bedeae25e1f7a"}]},"branch":"refs/heads/master"},"045e2363127e4a6f5c9b4e7a5a5f79d977f80287":{"kind":"TRIVIAL_REBASE","_number":41,"created":"2019-03-11 15:06:50.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/41"}}},"commit":{"parents":[{"commit":"95c592b9ee0a66ab8a32267219e890519cf3ae28","subject":"Add distro_package_manager configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95c592b9ee0a66ab8a32267219e890519cf3ae28"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-03-11 15:06:50.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/045e2363127e4a6f5c9b4e7a5a5f79d977f80287"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/045e2363127e4a6f5c9b4e7a5a5f79d977f80287"}]},"branch":"refs/heads/master"},"db68c4a5227b50bf8445edc0a91473761bbdd9e6":{"kind":"TRIVIAL_REBASE","_number":42,"created":"2019-03-11 16:39:36.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/42"}}},"commit":{"parents":[{"commit":"493902a4507d94385800498515b01fa416910f1d","subject":"Add distro_package_manager configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/493902a4507d94385800498515b01fa416910f1d"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-03-11 16:39:36.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db68c4a5227b50bf8445edc0a91473761bbdd9e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db68c4a5227b50bf8445edc0a91473761bbdd9e6"}]},"branch":"refs/heads/master"},"dd5bc5e119eee34a1ed548ba5cb2ef4c912f94e8":{"kind":"TRIVIAL_REBASE","_number":43,"created":"2019-03-11 18:45:09.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/43"}}},"commit":{"parents":[{"commit":"9c999ee431ad14977002612e1660adafb62e9f2d","subject":"Add distro_package_manager configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c999ee431ad14977002612e1660adafb62e9f2d"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-03-11 18:45:09.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd5bc5e119eee34a1ed548ba5cb2ef4c912f94e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd5bc5e119eee34a1ed548ba5cb2ef4c912f94e8"}]},"branch":"refs/heads/master"},"90a544271b77bcc3fd49b1d59e6ba9f6afbda8e5":{"kind":"TRIVIAL_REBASE","_number":44,"created":"2019-03-12 14:27:48.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/44"}}},"commit":{"parents":[{"commit":"61b491f3bcc819075aceaad06b957a214d13f64d","subject":"Add distro_package_manager configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61b491f3bcc819075aceaad06b957a214d13f64d"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-03-12 14:27:48.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90a544271b77bcc3fd49b1d59e6ba9f6afbda8e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90a544271b77bcc3fd49b1d59e6ba9f6afbda8e5"}]},"branch":"refs/heads/master"},"24e90df503b3c5bc2e5bc9baff9b9d80bc0fd7f2":{"kind":"REWORK","_number":45,"created":"2019-03-12 16:47:47.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/45"}}},"commit":{"parents":[{"commit":"5598810d288ebd8ace2d740c0a6521a72e776e69","subject":"Add distro_package_manager configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5598810d288ebd8ace2d740c0a6521a72e776e69"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-03-12 16:47:39.000000000","tz":-360},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/24e90df503b3c5bc2e5bc9baff9b9d80bc0fd7f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/24e90df503b3c5bc2e5bc9baff9b9d80bc0fd7f2"}]},"branch":"refs/heads/master"},"87134f3483a4d0346df49dad4e9bfccd20a7b8f6":{"kind":"TRIVIAL_REBASE","_number":46,"created":"2019-03-12 16:49:28.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/46"}}},"commit":{"parents":[{"commit":"bc2e453e05353fd6641b81becbc03ca18944b106","subject":"Add distro_package_manager configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc2e453e05353fd6641b81becbc03ca18944b106"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-03-12 16:49:28.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87134f3483a4d0346df49dad4e9bfccd20a7b8f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87134f3483a4d0346df49dad4e9bfccd20a7b8f6"}]},"branch":"refs/heads/master"},"920bafe0f089e31a1996c3340aaad58f8f45a657":{"kind":"TRIVIAL_REBASE","_number":47,"created":"2019-03-12 18:24:34.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/47"}}},"commit":{"parents":[{"commit":"78c819d6232b3fa5c6fb6aa5ea85a54c596f0b56","subject":"Add distro_package_manager configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78c819d6232b3fa5c6fb6aa5ea85a54c596f0b56"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-03-12 18:24:34.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/920bafe0f089e31a1996c3340aaad58f8f45a657"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/920bafe0f089e31a1996c3340aaad58f8f45a657"}]},"branch":"refs/heads/master"},"466f9e3a75525aad6e5c8deeb6478c1361d7d739":{"kind":"TRIVIAL_REBASE","_number":48,"created":"2019-03-14 15:58:19.000000000","uploader":{"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},"ref":"refs/changes/38/624838/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/48"}}},"commit":{"parents":[{"commit":"c8bbd0b102e176170ae707b310c47f490660cf83","subject":"Merge \"Fix the permission issue of placement-api log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8bbd0b102e176170ae707b310c47f490660cf83"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Taeha Kim","email":"kthguru@gmail.com","date":"2019-03-14 15:58:19.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/466f9e3a75525aad6e5c8deeb6478c1361d7d739"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/466f9e3a75525aad6e5c8deeb6478c1361d7d739"}]},"branch":"refs/heads/master"},"f139ced93e7381b628b366dec8eccf3cdb1b22b4":{"kind":"TRIVIAL_REBASE","_number":49,"created":"2019-03-26 13:12:04.000000000","uploader":{"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},"ref":"refs/changes/38/624838/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/49"}}},"commit":{"parents":[{"commit":"3a2e4081fdef9fde9dd8fed41fef2d86e21ba74e","subject":"Merge \"[docker/prometheus] Bump openstack-exporter version.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a2e4081fdef9fde9dd8fed41fef2d86e21ba74e"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Taeha Kim","email":"kthguru@gmail.com","date":"2019-03-26 13:12:04.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f139ced93e7381b628b366dec8eccf3cdb1b22b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f139ced93e7381b628b366dec8eccf3cdb1b22b4"}]},"branch":"refs/heads/master"},"74db0368fb5a3e61af15fd682c0efebc39cf7cea":{"kind":"TRIVIAL_REBASE","_number":50,"created":"2019-04-15 16:44:20.000000000","uploader":{"_account_id":12882,"name":"Taeha Kim","email":"kthguru@gmail.com","username":"kthguru"},"ref":"refs/changes/38/624838/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/50"}}},"commit":{"parents":[{"commit":"32693b83385c3b76ba0edf7e426bb110c6c335ae","subject":"Merge \"Add config option use_dumb_init to add it to container images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32693b83385c3b76ba0edf7e426bb110c6c335ae"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Taeha Kim","email":"kthguru@gmail.com","date":"2019-04-15 16:44:20.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74db0368fb5a3e61af15fd682c0efebc39cf7cea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74db0368fb5a3e61af15fd682c0efebc39cf7cea"}]},"branch":"refs/heads/master"},"54ccd2147253c819d5fed3b7f44c5a15ffe027db":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":51,"created":"2019-06-18 12:04:41.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/38/624838/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/51"}}},"commit":{"parents":[{"commit":"3b61de2d3193611506ebee37afa3f622ba1b8ecc","subject":"Merge \"Add sha256 in ceph fetch keys\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3b61de2d3193611506ebee37afa3f622ba1b8ecc"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-06-18 12:01:19.000000000","tz":60},"subject":"Fedora support for python3 package testing","message":"Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nFedora28 is used on purpose in order to match RHEL/CentOS 8.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/54ccd2147253c819d5fed3b7f44c5a15ffe027db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/54ccd2147253c819d5fed3b7f44c5a15ffe027db"}]},"branch":"refs/heads/master"},"fa3f9f4bfa4a9566e2c90abcbf4c01fe59518cea":{"kind":"REWORK","_number":52,"created":"2019-06-18 18:43:33.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/38/624838/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/52"}}},"commit":{"parents":[{"commit":"b8a0ed5dfafd7409aab6fa6b494cb05c6a9239f4","subject":"Merge \"Remove crane image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8a0ed5dfafd7409aab6fa6b494cb05c6a9239f4"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2019-06-18 18:43:18.000000000","tz":-360},"subject":"Fedora support for python3 package testing","message":"Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nFedora28 is used on purpose in order to match RHEL/CentOS 8.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa3f9f4bfa4a9566e2c90abcbf4c01fe59518cea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa3f9f4bfa4a9566e2c90abcbf4c01fe59518cea"}]},"branch":"refs/heads/master"},"a7ccfe46ea752e9a3ddffba06e284873cf5173d7":{"kind":"TRIVIAL_REBASE","_number":53,"created":"2019-06-18 19:13:31.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/38/624838/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/53"}}},"commit":{"parents":[{"commit":"86907688802f2633d165271eff08cdcdc0284b33","subject":"Merge \"Disable and remove OracleLinux CI jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/86907688802f2633d165271eff08cdcdc0284b33"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-18 19:13:31.000000000","tz":0},"subject":"Fedora support for python3 package testing","message":"Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nFedora28 is used on purpose in order to match RHEL/CentOS 8.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7ccfe46ea752e9a3ddffba06e284873cf5173d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7ccfe46ea752e9a3ddffba06e284873cf5173d7"}]},"branch":"refs/heads/master"},"b39180b1bf1085115f7464e12c81b4f9d05c0496":{"kind":"REWORK","_number":54,"created":"2019-07-15 14:08:13.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/38/624838/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/624838/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/624838/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/624838/54"}}},"commit":{"parents":[{"commit":"ef6a62de1c91fec16d20485c0bef16cf93346227","subject":"Merge \"Limit OVMF_CODE.secboot.fd to el7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef6a62de1c91fec16d20485c0bef16cf93346227"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-07-15 14:07:52.000000000","tz":330},"subject":"Fedora support for python3 package testing","message":"Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nFedora28 is used on purpose in order to match RHEL/CentOS 8.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b39180b1bf1085115f7464e12c81b4f9d05c0496"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b39180b1bf1085115f7464e12c81b4f9d05c0496"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
