)]}'
{"id":"openstack%2Fkolla~633971","triplet_id":"openstack%2Fkolla~master~I66e54bd18f4256cdca6c0c50626de4f975c246b8","project":"openstack/kolla","branch":"master","topic":"restrict-pkg-binary-distro","hashtags":[],"change_id":"I66e54bd18f4256cdca6c0c50626de4f975c246b8","subject":"Add script to check for forbidden instructions","status":"MERGED","created":"2019-01-30 15:04:24.000000000","updated":"2019-07-15 03:06:11.000000000","submitted":"2019-07-15 03:06:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"633971-1563159971958-0ae53db0","meta_rev_id":"0bbd7e05cd569447894dde9370e81c8f0d1e634c","_number":633971,"virtual_id_number":633971,"owner":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-07-15 03:06:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},{"value":0,"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"all":[{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-07-13 15:55:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":1,"date":"2019-07-14 02:45:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-08 09:12:03.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"value":1,"date":"2019-07-12 16:30:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":2,"date":"2019-07-15 02:06:50.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":1,"date":"2019-07-14 15:27:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":1,"date":"2019-07-14 02:39:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},{"value":0,"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":1,"date":"2019-07-15 02:06:50.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},{"value":0,"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},{"value":0,"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},{"value":0,"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},{"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-07 03:46:11.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-02-07 14:40:00.000000000","updated_by":{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},"reviewer":{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},"state":"REVIEWER"},{"updated":"2019-02-13 16:36:57.000000000","updated_by":{"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},"reviewer":{"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},"state":"REVIEWER"},{"updated":"2019-02-14 18:20:21.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-02-27 06:31:50.000000000","updated_by":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"reviewer":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"state":"REVIEWER"},{"updated":"2019-06-07 22:31:02.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-08 09:12:03.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-12 16:30:27.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-07-13 15:55:27.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-07-14 02:39:42.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-07-14 02:45:05.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-07-14 15:27:39.000000000","updated_by":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"reviewer":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"state":"REVIEWER"},{"updated":"2019-07-15 02:06:50.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-07-15 03:06:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ad339e0ef5409f740f6df27262a1f3fadb069c9","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2019-01-30 15:04:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"261ba7c027fb39d176958c4113f8c806c36c5177","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2019-01-30 15:08:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c22f102c82847f01f97c88712fd5695e9933e103","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2019-01-30 15:22:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"75d7e2e3a4518c9271719684c380ff08b317b5a1","author":{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},"date":"2019-01-30 15:59:16.000000000","message":"Patch Set 3: Code-Review+1\n\nthis looks to be useful! thanks","accounts_in_message":[],"_revision_number":3},{"id":"9678216be241b1a8558a6bc2795de0b2caf52f1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-30 17:26:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/71/633971/3/check/kolla-build-centos-binary/ea8c9de/ : FAILURE in 1h 00m 05s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/71/633971/3/check/kolla-build-centos-source/cb9aba3/ : FAILURE in 1h 14m 54s\n- kolla-ansible-centos-source http://logs.openstack.org/71/633971/3/check/kolla-ansible-centos-source/278ecd8/ : SUCCESS in 1h 23m 10s\n- kolla-ansible-centos-binary http://logs.openstack.org/71/633971/3/check/kolla-ansible-centos-binary/e7ff582/ : FAILURE in 1h 05m 51s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/71/633971/3/check/kolla-build-oraclelinux-binary/7a7dfab/ : FAILURE in 1h 19m 16s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/71/633971/3/check/kolla-build-oraclelinux-source/aba422a/ : FAILURE in 1h 08m 08s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/71/633971/3/check/kolla-ansible-oraclelinux-source/ea01262/ : SUCCESS in 1h 27m 50s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/71/633971/3/check/kolla-ansible-oraclelinux-binary/f23cf8f/ : FAILURE in 1h 12m 42s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/71/633971/3/check/openstack-tox-lower-constraints/11fc96f/ : SUCCESS in 5m 05s\n- openstack-tox-py36 http://logs.openstack.org/71/633971/3/check/openstack-tox-py36/10d43f2/ : SUCCESS in 4m 35s\n- openstack-tox-pep8 http://logs.openstack.org/71/633971/3/check/openstack-tox-pep8/28be580/ : SUCCESS in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/71/633971/3/check/openstack-tox-py27/b1a7aeb/ : SUCCESS in 4m 24s\n- openstack-tox-py35 http://logs.openstack.org/71/633971/3/check/openstack-tox-py35/ef1206c/ : SUCCESS in 4m 27s\n- openstack-tox-docs http://logs.openstack.org/71/633971/3/check/openstack-tox-docs/44c63b7/html/ : SUCCESS in 3m 04s\n- kolla-build-ubuntu-binary http://logs.openstack.org/71/633971/3/check/kolla-build-ubuntu-binary/dde6b03/ : FAILURE in 1h 23m 30s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/71/633971/3/check/kolla-build-ubuntu-source/c415b62/ : FAILURE in 30m 20s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/71/633971/3/check/kolla-ansible-ubuntu-source/3c4afb7/ : SUCCESS in 1h 01m 41s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/71/633971/3/check/kolla-ansible-ubuntu-binary/3e6ce96/ : SUCCESS in 1h 56m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"825c2a4e342d226320b20cac53a0f37e4bf8e13a","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2019-01-31 15:30:23.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"48b562a5d8ac6ae62904b9d15ff66877cf29df39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-31 17:23:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/71/633971/4/check/kolla-build-centos-binary/1194877/ : SUCCESS in 1h 41m 10s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/71/633971/4/check/kolla-build-centos-source/377b474/ : SUCCESS in 1h 15m 04s\n- kolla-ansible-centos-source http://logs.openstack.org/71/633971/4/check/kolla-ansible-centos-source/2aa12a3/ : SUCCESS in 1h 14m 34s\n- kolla-ansible-centos-binary http://logs.openstack.org/71/633971/4/check/kolla-ansible-centos-binary/10aed81/ : FAILURE in 1h 06m 03s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/71/633971/4/check/kolla-build-oraclelinux-binary/df29415/ : SUCCESS in 1h 25m 40s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/71/633971/4/check/kolla-build-oraclelinux-source/175020b/ : SUCCESS in 1h 23m 04s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/71/633971/4/check/kolla-ansible-oraclelinux-source/e727579/ : SUCCESS in 1h 14m 34s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/71/633971/4/check/kolla-ansible-oraclelinux-binary/965efc3/ : FAILURE in 1h 08m 07s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/71/633971/4/check/openstack-tox-lower-constraints/10ac7fe/ : SUCCESS in 3m 54s\n- openstack-tox-py36 http://logs.openstack.org/71/633971/4/check/openstack-tox-py36/cc648cf/ : SUCCESS in 5m 08s\n- openstack-tox-pep8 http://logs.openstack.org/71/633971/4/check/openstack-tox-pep8/fd20d62/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/71/633971/4/check/openstack-tox-py27/3526914/ : SUCCESS in 4m 06s\n- openstack-tox-py35 http://logs.openstack.org/71/633971/4/check/openstack-tox-py35/1316923/ : SUCCESS in 3m 52s\n- openstack-tox-docs http://logs.openstack.org/71/633971/4/check/openstack-tox-docs/8f15072/html/ : SUCCESS in 3m 58s\n- kolla-build-ubuntu-binary http://logs.openstack.org/71/633971/4/check/kolla-build-ubuntu-binary/422ba04/ : FAILURE in 1h 35m 42s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/71/633971/4/check/kolla-build-ubuntu-source/77f43b4/ : SUCCESS in 37m 32s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/71/633971/4/check/kolla-ansible-ubuntu-source/5ba23f4/ : SUCCESS in 1h 09m 46s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/71/633971/4/check/kolla-ansible-ubuntu-binary/05a75e9/ : SUCCESS in 1h 45m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f74111415da2501bd6b0d201d14d518719d2ace2","author":{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},"date":"2019-01-31 20:32:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f211e281ae669ad8f3ac5eed4126d1f873180383","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2019-02-01 09:52:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ac0e3bec7b927438c13e3e61a326f81ff9c91e78","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2019-02-04 15:00:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6a2418c5416d3cc7df139173a5e47555997a4907","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-04 16:47:23.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/71/633971/5/check/kolla-build-centos-binary/965719a/ : SUCCESS in 1h 35m 22s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/71/633971/5/check/kolla-build-centos-source/8d32f16/ : SUCCESS in 1h 32m 33s\n- kolla-ansible-centos-source http://logs.openstack.org/71/633971/5/check/kolla-ansible-centos-source/0820ad4/ : SUCCESS in 1h 27m 30s\n- kolla-ansible-centos-binary http://logs.openstack.org/71/633971/5/check/kolla-ansible-centos-binary/990eee7/ : FAILURE in 1h 23m 18s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/71/633971/5/check/kolla-build-oraclelinux-binary/e58e2ac/ : SUCCESS in 1h 20m 34s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/71/633971/5/check/kolla-build-oraclelinux-source/8f5b72a/ : SUCCESS in 1h 36m 01s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/71/633971/5/check/kolla-ansible-oraclelinux-source/25c9c17/ : SUCCESS in 1h 24m 30s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/71/633971/5/check/kolla-ansible-oraclelinux-binary/05ecb1b/ : FAILURE in 1h 22m 12s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/71/633971/5/check/openstack-tox-lower-constraints/b1003b8/ : SUCCESS in 4m 31s\n- openstack-tox-py36 http://logs.openstack.org/71/633971/5/check/openstack-tox-py36/9a97041/ : SUCCESS in 3m 56s\n- openstack-tox-pep8 http://logs.openstack.org/71/633971/5/check/openstack-tox-pep8/2f78234/ : SUCCESS in 5m 40s\n- openstack-tox-py27 http://logs.openstack.org/71/633971/5/check/openstack-tox-py27/d28c2a2/ : SUCCESS in 4m 31s\n- openstack-tox-py35 http://logs.openstack.org/71/633971/5/check/openstack-tox-py35/74f1f17/ : SUCCESS in 4m 10s\n- openstack-tox-docs http://logs.openstack.org/71/633971/5/check/openstack-tox-docs/666f223/html/ : SUCCESS in 4m 20s\n- kolla-build-ubuntu-binary http://logs.openstack.org/71/633971/5/check/kolla-build-ubuntu-binary/7f89f48/ : FAILURE in 1h 26m 19s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/71/633971/5/check/kolla-build-ubuntu-source/b071d90/ : FAILURE in 39m 54s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/71/633971/5/check/kolla-ansible-ubuntu-source/5897675/ : FAILURE in 29m 55s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/71/633971/5/check/kolla-ansible-ubuntu-binary/308acb2/ : FAILURE in 1h 16m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9d621a181cfc25b1514269c4cabdc760b33b201f","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2019-02-06 11:03:03.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"c8bde65fe62ca755c953678aa3a0b9b6f24531fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-06 13:07:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/71/633971/5/check/kolla-build-centos-binary/70b8fe2/ : SUCCESS in 1h 17m 51s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/71/633971/5/check/kolla-build-centos-source/0c90412/ : SUCCESS in 1h 13m 44s\n- kolla-ansible-centos-source http://logs.openstack.org/71/633971/5/check/kolla-ansible-centos-source/181a15c/ : SUCCESS in 1h 20m 55s\n- kolla-ansible-centos-binary http://logs.openstack.org/71/633971/5/check/kolla-ansible-centos-binary/0c13dd7/ : FAILURE in 1h 16m 56s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/71/633971/5/check/kolla-build-oraclelinux-binary/e1dc8b0/ : SUCCESS in 1h 25m 17s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/71/633971/5/check/kolla-build-oraclelinux-source/d2ff79f/ : SUCCESS in 1h 32m 09s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/71/633971/5/check/kolla-ansible-oraclelinux-source/72a8de0/ : SUCCESS in 1h 28m 43s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/71/633971/5/check/kolla-ansible-oraclelinux-binary/b047484/ : FAILURE in 1h 07m 08s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/71/633971/5/check/openstack-tox-lower-constraints/90cda5c/ : SUCCESS in 4m 03s\n- openstack-tox-py36 http://logs.openstack.org/71/633971/5/check/openstack-tox-py36/a545f7f/ : SUCCESS in 3m 17s\n- openstack-tox-pep8 http://logs.openstack.org/71/633971/5/check/openstack-tox-pep8/25d5074/ : SUCCESS in 4m 39s\n- openstack-tox-py27 http://logs.openstack.org/71/633971/5/check/openstack-tox-py27/0b9d2d3/ : SUCCESS in 3m 37s\n- openstack-tox-py35 http://logs.openstack.org/71/633971/5/check/openstack-tox-py35/38d48a3/ : SUCCESS in 3m 40s\n- openstack-tox-docs http://logs.openstack.org/71/633971/5/check/openstack-tox-docs/0f3be23/html/ : SUCCESS in 3m 33s\n- kolla-build-ubuntu-binary http://logs.openstack.org/71/633971/5/check/kolla-build-ubuntu-binary/c2e120c/ : SUCCESS in 1h 30m 05s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/71/633971/5/check/kolla-build-ubuntu-source/53dfc42/ : SUCCESS in 36m 38s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/71/633971/5/check/kolla-ansible-ubuntu-source/7dc87eb/ : SUCCESS in 1h 02m 26s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/71/633971/5/check/kolla-ansible-ubuntu-binary/57ec050/ : SUCCESS in 1h 58m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"39ae1d34b7bef0d218c15bd053795f51ceaf80e4","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-02-07 03:46:11.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"fccf46fb78718f7ab61d8ad9ea768c8ab3858d92","author":{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},"date":"2019-02-07 14:40:00.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"8743ff09cf1ec67c53ad8b3a0f70e6b887465cb8","author":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"date":"2019-02-12 10:41:52.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"12a25665fb6695efe4d4d671ed463bc5d5247f1a","author":{"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},"date":"2019-02-13 16:36:57.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"90ea0d044587f196cebb9409525c6b19b2d60b92","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-02-14 18:20:21.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"cbce0e734f5d0c4c6e22227065c071ea034ed784","author":{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},"date":"2019-02-22 12:21:48.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nfound another, but I\u0027m wondering if we are missing something with some of this.  \n\nSingle package manager packages (binary install) - trusted sources only\npackage manager binary install + binary install (ruby gem/ sensu plugin, others??) from official tag\u0027s and binary tarballs - no building things at install time, fail if not available (no gcc/make...)\nSource - binary for supporting packages + pip/gem/sensu-install/npm/","accounts_in_message":[],"_revision_number":5},{"id":"add53190cb817cc81e9a2fb56724cf3786d288cf","author":{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},"date":"2019-02-22 15:07:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3cf9883f78c5193b6407c9082c6cd05b2409812b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-22 17:29:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/71/633971/6/check/kolla-build-centos-binary/aace626/ : SUCCESS in 41m 14s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/71/633971/6/check/kolla-build-centos-source/34734dd/ : SUCCESS in 45m 12s\n- kolla-ansible-centos-source http://logs.openstack.org/71/633971/6/check/kolla-ansible-centos-source/2fef8e3/ : SUCCESS in 1h 28m 52s\n- kolla-ansible-centos-binary http://logs.openstack.org/71/633971/6/check/kolla-ansible-centos-binary/707e7ea/ : SUCCESS in 1h 12m 00s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/71/633971/6/check/kolla-build-oraclelinux-binary/d3a7314/ : SUCCESS in 47m 21s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/71/633971/6/check/kolla-build-oraclelinux-source/92eb41c/ : SUCCESS in 47m 30s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/71/633971/6/check/kolla-ansible-oraclelinux-source/78701ed/ : SUCCESS in 1h 11m 57s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/71/633971/6/check/kolla-ansible-oraclelinux-binary/690803a/ : SUCCESS in 1h 15m 08s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/71/633971/6/check/openstack-tox-lower-constraints/1d20a8e/ : SUCCESS in 4m 09s\n- openstack-tox-py36 http://logs.openstack.org/71/633971/6/check/openstack-tox-py36/ac5653a/ : SUCCESS in 3m 33s\n- openstack-tox-pep8 http://logs.openstack.org/71/633971/6/check/openstack-tox-pep8/6cf22f0/ : SUCCESS in 4m 35s\n- openstack-tox-py27 http://logs.openstack.org/71/633971/6/check/openstack-tox-py27/892c7a9/ : SUCCESS in 3m 14s\n- openstack-tox-py35 http://logs.openstack.org/71/633971/6/check/openstack-tox-py35/535df54/ : SUCCESS in 3m 56s\n- openstack-tox-docs http://logs.openstack.org/71/633971/6/check/openstack-tox-docs/b584ad7/html/ : SUCCESS in 3m 45s\n- kolla-build-ubuntu-binary http://logs.openstack.org/71/633971/6/check/kolla-build-ubuntu-binary/4c93ea7/ : SUCCESS in 1h 23m 49s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/71/633971/6/check/kolla-build-ubuntu-source/40aad74/ : SUCCESS in 33m 06s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/71/633971/6/check/kolla-ansible-ubuntu-source/91c7ddf/ : SUCCESS in 1h 05m 50s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/71/633971/6/check/kolla-ansible-ubuntu-binary/b91c68b/ : TIMED_OUT in 2h 02m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5665af5f779e66ef5ebe8efaad936730f20bc438","author":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"date":"2019-02-27 06:31:50.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"046e4e925eaafc4ea83f896e1bc7c7b7bb5f4ae0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-02-27 09:09:54.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)\n\nMy -1 here is due to the muddying of the pep8 output. It\u0027s already quite hard to find the source of pep8 failures from zuul logs, and this makes it much harder.\n\nI\u0027d suggest we don\u0027t run this in the pep8 environment yet.","accounts_in_message":[],"_revision_number":6},{"id":"a4ce7622632e7148d986146f44d000ca96037dfd","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2019-02-27 10:59:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"edfc39b96c2b3c20a8a14b7341624ee06ac4a994","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-02-27 11:02:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5151c8b26363c8d269e0a6ff517b25f7f75515aa","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2019-02-27 11:19:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7341ddf1ac592e067dd0e79d65f99c351b7aeb79","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-07 22:31:02.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"10d5386e283ea895f42e79efa991e97541c0b48c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-07 22:32:42.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d540fe2558a74d48c255e4a167609419a5b37285","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 00:12:25.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/71/633971/7/check/kolla-build-centos-binary/b585a93/ : SUCCESS in 42m 08s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/71/633971/7/check/kolla-build-centos-source/b634a58/ : SUCCESS in 50m 15s\n- kolla-ansible-centos-source http://logs.openstack.org/71/633971/7/check/kolla-ansible-centos-source/08f44af/ : SUCCESS in 1h 09m 09s\n- kolla-ansible-centos-binary http://logs.openstack.org/71/633971/7/check/kolla-ansible-centos-binary/1720409/ : SUCCESS in 1h 09m 13s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/71/633971/7/check/kolla-ansible-centos-source-upgrade/aac4881/ : SUCCESS in 1h 20m 38s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/71/633971/7/check/kolla-build-oraclelinux-binary/20650e8/ : SUCCESS in 33m 56s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/71/633971/7/check/kolla-build-oraclelinux-source/6908972/ : SUCCESS in 49m 17s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/71/633971/7/check/kolla-ansible-oraclelinux-source/135f23d/ : SUCCESS in 1h 10m 29s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/71/633971/7/check/kolla-ansible-oraclelinux-binary/9b65b04/ : SUCCESS in 1h 05m 14s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/71/633971/7/check/openstack-tox-lower-constraints/a4bdba6/ : SUCCESS in 4m 37s\n- openstack-tox-pep8 http://logs.openstack.org/71/633971/7/check/openstack-tox-pep8/a199e23/ : SUCCESS in 4m 16s\n- openstack-tox-py27 http://logs.openstack.org/71/633971/7/check/openstack-tox-py27/21e034f/ : SUCCESS in 3m 04s\n- openstack-tox-py36 http://logs.openstack.org/71/633971/7/check/openstack-tox-py36/c96ad71/ : SUCCESS in 3m 25s\n- openstack-tox-py37 http://logs.openstack.org/71/633971/7/check/openstack-tox-py37/7ab10b2/ : SUCCESS in 3m 37s\n- openstack-tox-docs http://logs.openstack.org/71/633971/7/check/openstack-tox-docs/f089377/html/ : SUCCESS in 4m 05s\n- kolla-build-ubuntu-binary http://logs.openstack.org/71/633971/7/check/kolla-build-ubuntu-binary/01f5744/ : SUCCESS in 33m 54s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/71/633971/7/check/kolla-build-ubuntu-source/2b42a47/ : SUCCESS in 28m 45s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/71/633971/7/check/kolla-ansible-ubuntu-source/f3059e3/ : SUCCESS in 45m 38s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/71/633971/7/check/kolla-ansible-ubuntu-binary/4657a89/ : SUCCESS in 47m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/71/633971/7/check/kolla-ansible-ubuntu-source-upgrade/4a744fa/ : SUCCESS in 55m 07s","accounts_in_message":[],"_revision_number":7},{"id":"c3b46c4f2b008fdf1fea0a980867bdf9997b3176","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-08 09:08:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1ed1522b95a706f9c14c1d9291c23280100af0a0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-08 09:12:03.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"9bd05d555b56a6b588266e5378380865c3599aa9","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-07-12 16:30:27.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"c9450679aa777e84cb8ce1630e7ebe8a30314d51","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-07-13 15:55:27.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"11ffcbbce0d184cb61625310f1286d013d89d98b","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-07-14 02:39:42.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"7dd68df795bf1cc65acacd4c474fb179b0426327","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-07-14 02:45:05.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"b0c91c8004853aa89678a6aeb80099f3f223a86e","author":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"date":"2019-07-14 15:27:39.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"dfa1098825bfc5d8d4f007bf369ad67a402b1124","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-07-15 02:06:50.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nlgtm, thanks","accounts_in_message":[],"_revision_number":7},{"id":"e7d84f1e01cb02b78b9c377e8e1edb9553665c89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 02:07:03.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"42e1edfa65d06a0abae993a1e81fb179d3d481be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 03:06:11.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/71/633971/7/gate/kolla-build-centos-source/cd47051/ : SUCCESS in 35m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/71/633971/7/gate/openstack-tox-lower-constraints/056ac61/ : SUCCESS in 2m 54s\n- openstack-tox-pep8 http://logs.openstack.org/71/633971/7/gate/openstack-tox-pep8/384f09c/ : SUCCESS in 4m 18s\n- openstack-tox-py27 http://logs.openstack.org/71/633971/7/gate/openstack-tox-py27/aa8ade6/ : SUCCESS in 3m 59s\n- openstack-tox-py36 http://logs.openstack.org/71/633971/7/gate/openstack-tox-py36/e4501f9/ : SUCCESS in 4m 09s\n- openstack-tox-py37 http://logs.openstack.org/71/633971/7/gate/openstack-tox-py37/ece2656/ : SUCCESS in 4m 22s\n- openstack-tox-docs http://logs.openstack.org/71/633971/7/gate/openstack-tox-docs/b9717fd/html/ : SUCCESS in 3m 52s\n- kolla-build-ubuntu-source http://logs.openstack.org/71/633971/7/gate/kolla-build-ubuntu-source/ef6ae10/ : SUCCESS in 31m 02s","accounts_in_message":[],"_revision_number":7},{"id":"5d1322097857dd6e80701b7f0e487cb7742be7c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 03:06:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"378a16c3572161687f43ddaf77d12c39bfacd760","revisions":{"0f1ab0601b6b56a1857b04baea86b49b6baacaf9":{"kind":"REWORK","_number":1,"created":"2019-01-30 15:04:24.000000000","uploader":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"ref":"refs/changes/71/633971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/633971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/633971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/633971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/633971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/633971/1"}}},"commit":{"parents":[{"commit":"1d6817e18146ff49486adf5432f4bfe0f1ea1920","subject":"Merge \"Update kubernetes-entrypoint v0.1.1 to 0.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d6817e18146ff49486adf5432f4bfe0f1ea1920"}]}],"author":{"name":"Martin André","email":"m.andre@redhat.com","date":"2019-01-30 13:28:12.000000000","tz":60},"committer":{"name":"Martin André","email":"m.andre@redhat.com","date":"2019-01-30 15:03:40.000000000","tz":60},"subject":"Add script to check for forbidden instructions","message":"Add script to check for forbidden instructions\n\nWhen building images for binary type, it is expected to not install\nexecutables from untrusted sources, and gem or pip packages for instance\nshould come from the distro itself.\n\nThis commit adds a `validate-binary-build.sh` script that checks for\nforbidden commands in the rendered Dockerfile templates for the binary\ntype builds and hooks this script with the pep8 CI job.\n\nThe output is just informative for now until we\u0027ve fixed the culprit\nDockerfiles, then we\u0027ll make the script return an error code and with\nthe ability to fail the pep8 CI check. This should prevent unwanted\ncommands from entering binary type Dockerfiles.\n\nChange-Id: I66e54bd18f4256cdca6c0c50626de4f975c246b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f1ab0601b6b56a1857b04baea86b49b6baacaf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f1ab0601b6b56a1857b04baea86b49b6baacaf9"}]},"branch":"refs/heads/master"},"ea288b6379024b7760ee43530f4407563077d02f":{"kind":"REWORK","_number":2,"created":"2019-01-30 15:08:12.000000000","uploader":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"ref":"refs/changes/71/633971/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/633971/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/633971/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/633971/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/633971/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/633971/2"}}},"commit":{"parents":[{"commit":"1d6817e18146ff49486adf5432f4bfe0f1ea1920","subject":"Merge \"Update kubernetes-entrypoint v0.1.1 to 0.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d6817e18146ff49486adf5432f4bfe0f1ea1920"}]}],"author":{"name":"Martin André","email":"m.andre@redhat.com","date":"2019-01-30 13:28:12.000000000","tz":60},"committer":{"name":"Martin André","email":"m.andre@redhat.com","date":"2019-01-30 15:07:01.000000000","tz":60},"subject":"Add script to check for forbidden instructions","message":"Add script to check for forbidden instructions\n\nWhen building images for binary type, it is expected to not install\nexecutables from untrusted sources, and gem or pip packages for instance\nshould come from the distro itself.\n\nThis commit adds a `validate-binary-build.sh` script that checks for\nforbidden commands in the rendered Dockerfile templates for the binary\ntype builds and hooks this script with the pep8 CI job.\n\nThe output is just informative for now until we\u0027ve fixed the culprit\nDockerfiles, then we\u0027ll make the script return an error code and with\nthe ability to fail the pep8 CI check. This should prevent unwanted\ncommands from entering binary type Dockerfiles.\n\nChange-Id: I66e54bd18f4256cdca6c0c50626de4f975c246b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea288b6379024b7760ee43530f4407563077d02f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea288b6379024b7760ee43530f4407563077d02f"}]},"branch":"refs/heads/master"},"cf787051e51357ff8cea29b2bd8dce31d9df6030":{"kind":"REWORK","_number":3,"created":"2019-01-30 15:22:45.000000000","uploader":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"ref":"refs/changes/71/633971/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/633971/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/633971/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/633971/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/633971/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/633971/3"}}},"commit":{"parents":[{"commit":"1d6817e18146ff49486adf5432f4bfe0f1ea1920","subject":"Merge \"Update kubernetes-entrypoint v0.1.1 to 0.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d6817e18146ff49486adf5432f4bfe0f1ea1920"}]}],"author":{"name":"Martin André","email":"m.andre@redhat.com","date":"2019-01-30 13:28:12.000000000","tz":60},"committer":{"name":"Martin André","email":"m.andre@redhat.com","date":"2019-01-30 15:22:24.000000000","tz":60},"subject":"Add script to check for forbidden instructions","message":"Add script to check for forbidden instructions\n\nWhen building images for binary type, it is expected to not install\nexecutables from untrusted sources, and gem or pip packages for instance\nshould come from the distro itself.\n\nThis commit adds a `validate-binary-build.sh` script that checks for\nforbidden commands in the rendered Dockerfile templates for the binary\ntype builds and hooks this script with the pep8 CI job.\n\nThe output is just informative for now until we\u0027ve fixed the culprit\nDockerfiles, then we\u0027ll make the script return an error code and with\nthe ability to fail the pep8 CI check. This should prevent unwanted\ncommands from entering binary type Dockerfiles.\n\nChange-Id: I66e54bd18f4256cdca6c0c50626de4f975c246b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf787051e51357ff8cea29b2bd8dce31d9df6030"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf787051e51357ff8cea29b2bd8dce31d9df6030"}]},"branch":"refs/heads/master"},"9eafb4ff3a2d04fb82159f5146f935438bd1d358":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-01-31 15:30:23.000000000","uploader":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"ref":"refs/changes/71/633971/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/633971/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/633971/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/633971/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/633971/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/633971/4"}}},"commit":{"parents":[{"commit":"679b9b7acbca9ddca77b25f872e1bcb443362298","subject":"Merge \"Revert \"Update kafka version to 2.1.0\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/679b9b7acbca9ddca77b25f872e1bcb443362298"}]}],"author":{"name":"Martin André","email":"m.andre@redhat.com","date":"2019-01-30 13:28:12.000000000","tz":60},"committer":{"name":"Martin André","email":"m.andre@redhat.com","date":"2019-01-31 14:14:22.000000000","tz":60},"subject":"Add script to check for forbidden instructions","message":"Add script to check for forbidden instructions\n\nWhen building images for binary type, it is expected to not install\nexecutables from untrusted sources, and gem or pip packages for instance\nshould come from the distro itself.\n\nThis commit adds a `validate-binary-build.sh` script that checks for\nforbidden commands in the rendered Dockerfile templates for the binary\ntype builds and hooks this script with the pep8 CI job.\n\nThe output is just informative for now until we\u0027ve fixed the culprit\nDockerfiles, then we\u0027ll make the script return an error code and with\nthe ability to fail the pep8 CI check. This should prevent unwanted\ncommands from entering binary type Dockerfiles.\n\nChange-Id: I66e54bd18f4256cdca6c0c50626de4f975c246b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9eafb4ff3a2d04fb82159f5146f935438bd1d358"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9eafb4ff3a2d04fb82159f5146f935438bd1d358"}]},"branch":"refs/heads/master"},"c8df2300add137b785fc6d10568ef0f66e16df1d":{"kind":"REWORK","_number":5,"created":"2019-02-04 15:00:33.000000000","uploader":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"ref":"refs/changes/71/633971/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/633971/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/633971/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/633971/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/633971/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/633971/5"}}},"commit":{"parents":[{"commit":"679b9b7acbca9ddca77b25f872e1bcb443362298","subject":"Merge \"Revert \"Update kafka version to 2.1.0\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/679b9b7acbca9ddca77b25f872e1bcb443362298"}]}],"author":{"name":"Martin André","email":"m.andre@redhat.com","date":"2019-01-30 13:28:12.000000000","tz":60},"committer":{"name":"Martin André","email":"m.andre@redhat.com","date":"2019-02-04 14:59:59.000000000","tz":60},"subject":"Add script to check for forbidden instructions","message":"Add script to check for forbidden instructions\n\nWhen building images for binary type, it is expected to not install\nexecutables from untrusted sources, and gem or pip packages for instance\nshould come from the distro itself.\n\nThis commit adds a `validate-binary-build.sh` script that checks for\nforbidden commands in the rendered Dockerfile templates for the binary\ntype builds and hooks this script with the pep8 CI job.\n\nThe output is just informative for now until we\u0027ve fixed the culprit\nDockerfiles, then we\u0027ll make the script return an error code and with\nthe ability to fail the pep8 CI check. This should prevent unwanted\ncommands from entering binary type Dockerfiles.\n\nChange-Id: I66e54bd18f4256cdca6c0c50626de4f975c246b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8df2300add137b785fc6d10568ef0f66e16df1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8df2300add137b785fc6d10568ef0f66e16df1d"}]},"branch":"refs/heads/master"},"0f789a92d012ad0744a3d86c05f24d407fc7cd59":{"kind":"REWORK","_number":6,"created":"2019-02-22 15:07:35.000000000","uploader":{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},"ref":"refs/changes/71/633971/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/633971/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/633971/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/633971/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/633971/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/633971/6"}}},"commit":{"parents":[{"commit":"34db656e57283fa800096746156d44f26c7d947e","subject":"Merge \"nova: Only install OVMF packages on RHEL 7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34db656e57283fa800096746156d44f26c7d947e"}]}],"author":{"name":"Martin André","email":"m.andre@redhat.com","date":"2019-01-30 13:28:12.000000000","tz":60},"committer":{"name":"Jon Schlueter","email":"jschluet@redhat.com","date":"2019-02-22 15:07:01.000000000","tz":-300},"subject":"Add script to check for forbidden instructions","message":"Add script to check for forbidden instructions\n\nWhen building images for binary type, it is expected to not install\nexecutables from untrusted sources, and gem, sensu-install, npm or\npip packages for instance should come from the distro itself.\n\nThis commit adds a `validate-binary-build.sh` script that checks for\nforbidden commands in the rendered Dockerfile templates for the binary\ntype builds and hooks this script with the pep8 CI job.\n\nThe output is just informative for now until we\u0027ve fixed the culprit\nDockerfiles, then we\u0027ll make the script return an error code and with\nthe ability to fail the pep8 CI check. This should prevent unwanted\ncommands from entering binary type Dockerfiles.\n\nChange-Id: I66e54bd18f4256cdca6c0c50626de4f975c246b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f789a92d012ad0744a3d86c05f24d407fc7cd59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f789a92d012ad0744a3d86c05f24d407fc7cd59"}]},"branch":"refs/heads/master"},"378a16c3572161687f43ddaf77d12c39bfacd760":{"kind":"REWORK","_number":7,"created":"2019-06-07 22:32:42.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/71/633971/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/633971/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/633971/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/633971/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/633971/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/633971/7"}}},"commit":{"parents":[{"commit":"498c57fcac31da9268918ec6c188fe2aae0c4906","subject":"Merge \"Use packagecloud/ubuntu rabbitmq repo for all archs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/498c57fcac31da9268918ec6c188fe2aae0c4906"}]}],"author":{"name":"Martin André","email":"m.andre@redhat.com","date":"2019-01-30 13:28:12.000000000","tz":60},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-07 22:32:36.000000000","tz":-360},"subject":"Add script to check for forbidden instructions","message":"Add script to check for forbidden instructions\n\nWhen building images for binary type, it is expected to not install\nexecutables from untrusted sources, and gem, sensu-install, npm or\npip packages for instance should come from the distro itself.\n\nThis commit adds a `validate-binary-build.sh` script that checks for\nforbidden commands in the rendered Dockerfile templates for the binary\ntype builds and hooks this script with the pep8 CI job.\n\nThe output is just informative for now until we\u0027ve fixed the culprit\nDockerfiles, then we\u0027ll make the script return an error code and with\nthe ability to fail the pep8 CI check. This should prevent unwanted\ncommands from entering binary type Dockerfiles.\n\nChange-Id: I66e54bd18f4256cdca6c0c50626de4f975c246b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/378a16c3572161687f43ddaf77d12c39bfacd760"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/378a16c3572161687f43ddaf77d12c39bfacd760"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
