)]}'
{"id":"openstack%2Fkolla~634706","triplet_id":"openstack%2Fkolla~master~Ie9f80eded6330a4ca1833018734b785b5be03b2d","project":"openstack/kolla","branch":"master","topic":"restrict-pkg-binary-distro","hashtags":[],"change_id":"Ie9f80eded6330a4ca1833018734b785b5be03b2d","subject":"Do not pip install for binary type","status":"ABANDONED","created":"2019-02-04 15:00:33.000000000","updated":"2021-06-15 08:35:30.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"de8743e0ba5a7876ea3f04ba983e8389a49b70fb","_number":634706,"virtual_id_number":634706,"owner":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":-1,"date":"2019-06-10 19:26:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-06-10 16:59:43.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},{"value":-1,"date":"2019-06-10 16:13:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"date":"2020-02-06 13:25:39.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"date":"2020-02-06 13:25:19.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"date":"2020-02-06 13:33:46.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-05 06:50:11.000000000","updated_by":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"reviewer":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"state":"REVIEWER"},{"updated":"2019-02-05 07:22:07.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-02-25 16:59:15.000000000","updated_by":{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},"reviewer":{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},"state":"REVIEWER"},{"updated":"2019-06-07 22:36:56.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-10 06:38:16.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-06-10 16:59:43.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-10 19:26:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-06 13:25:19.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-02-06 13:25:39.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-02-06 13:33:46.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"47b3e6a396c47beefbb16a77596179827ba5a0e6","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2019-02-04 15:00:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cca444de5013d874ea2e3ab545669907e260f28a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-04 16:53:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/06/634706/1/check/kolla-build-centos-binary/f4bb46f/ : SUCCESS in 1h 07m 41s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/06/634706/1/check/kolla-build-centos-source/6eaf92b/ : SUCCESS in 1h 19m 15s\n- kolla-ansible-centos-source http://logs.openstack.org/06/634706/1/check/kolla-ansible-centos-source/35c499b/ : SUCCESS in 1h 16m 53s\n- kolla-ansible-centos-binary http://logs.openstack.org/06/634706/1/check/kolla-ansible-centos-binary/2627ab1/ : FAILURE in 1h 11m 09s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/06/634706/1/check/tripleo-build-containers-centos-7/2ca637d/ : FAILURE in 35m 01s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/06/634706/1/check/kolla-build-oraclelinux-binary/a01e1cd/ : SUCCESS in 1h 24m 06s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/06/634706/1/check/kolla-build-oraclelinux-source/4d70f4c/ : SUCCESS in 1h 22m 18s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/06/634706/1/check/kolla-ansible-oraclelinux-source/4095055/ : SUCCESS in 1h 42m 52s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/06/634706/1/check/kolla-ansible-oraclelinux-binary/7688b5d/ : FAILURE in 1h 06m 34s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/06/634706/1/check/openstack-tox-lower-constraints/9be9ed4/ : SUCCESS in 4m 20s\n- openstack-tox-py36 http://logs.openstack.org/06/634706/1/check/openstack-tox-py36/6418acf/ : SUCCESS in 4m 29s\n- openstack-tox-pep8 http://logs.openstack.org/06/634706/1/check/openstack-tox-pep8/cd44c5a/ : SUCCESS in 5m 29s\n- openstack-tox-py27 http://logs.openstack.org/06/634706/1/check/openstack-tox-py27/fd33e85/ : SUCCESS in 4m 30s\n- openstack-tox-py35 http://logs.openstack.org/06/634706/1/check/openstack-tox-py35/67d3f3f/ : SUCCESS in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/06/634706/1/check/openstack-tox-docs/9dfdda4/html/ : SUCCESS in 3m 46s\n- build-openstack-releasenotes http://logs.openstack.org/06/634706/1/check/build-openstack-releasenotes/9ae6fdf/html/ : SUCCESS in 5m 02s\n- kolla-build-ubuntu-binary http://logs.openstack.org/06/634706/1/check/kolla-build-ubuntu-binary/45f28ce/ : FAILURE in 1h 23m 40s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/06/634706/1/check/kolla-build-ubuntu-source/cc8005c/ : FAILURE in 31m 00s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/06/634706/1/check/kolla-ansible-ubuntu-source/1858125/ : FAILURE in 32m 02s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/06/634706/1/check/kolla-ansible-ubuntu-binary/86d33e3/ : FAILURE in 1h 22m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4b786b708cd164e6199b94740dba3f91d369fef3","author":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"date":"2019-02-05 06:50:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e2a98143e9732d25887e7d777ebc85cd234e71aa","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-02-05 07:22:07.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e61740444e2027ea35f189626627537083401798","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2019-02-21 10:55:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf60cfe837d3905118c66a18ffdf1acfe1fd8d5d","author":{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},"date":"2019-02-25 16:59:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c1e715ed36e5e2e4b0da2845b112d0c7350c736c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-07 22:36:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b08a8086e5a45ea378f58c95720a75d88294942f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-07 23:14:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/06/634706/2/openstack-check/tripleo-build-containers-fedora-28-master/fe6b877/ : SUCCESS in 35m 38s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/06/634706/2/openstack-check/tripleo-build-containers-fedora-28-stein/0af60ec/ : SUCCESS in 35m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"19a9f352ba466ed65037154372c0f76ca572eec6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 23:19:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/06/634706/2/check/kolla-build-centos-binary/62c3f3e/ : FAILURE in 4m 58s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/06/634706/2/check/kolla-build-centos-source/00e25cf/ : FAILURE in 4m 57s\n- kolla-ansible-centos-source http://logs.openstack.org/06/634706/2/check/kolla-ansible-centos-source/92083c4/ : FAILURE in 7m 07s\n- kolla-ansible-centos-binary http://logs.openstack.org/06/634706/2/check/kolla-ansible-centos-binary/d0dfb80/ : FAILURE in 7m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/06/634706/2/check/kolla-ansible-centos-source-upgrade/bb36152/ : FAILURE in 8m 12s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/06/634706/2/check/tripleo-build-containers-centos-7/2a18804/ : FAILURE in 14m 39s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/06/634706/2/check/kolla-build-oraclelinux-binary/b9ff4ff/ : FAILURE in 5m 17s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/06/634706/2/check/kolla-build-oraclelinux-source/778cf01/ : FAILURE in 4m 54s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/06/634706/2/check/kolla-ansible-oraclelinux-source/2528e9c/ : FAILURE in 8m 15s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/06/634706/2/check/kolla-ansible-oraclelinux-binary/693491c/ : FAILURE in 7m 37s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/06/634706/2/check/openstack-tox-lower-constraints/1a74874/ : SUCCESS in 4m 21s\n- openstack-tox-pep8 http://logs.openstack.org/06/634706/2/check/openstack-tox-pep8/efc28b3/ : SUCCESS in 4m 17s\n- openstack-tox-py27 http://logs.openstack.org/06/634706/2/check/openstack-tox-py27/043e7bb/ : SUCCESS in 3m 25s\n- openstack-tox-py36 http://logs.openstack.org/06/634706/2/check/openstack-tox-py36/ea67f97/ : SUCCESS in 3m 56s\n- openstack-tox-py37 http://logs.openstack.org/06/634706/2/check/openstack-tox-py37/6efe0a4/ : SUCCESS in 5m 00s\n- openstack-tox-docs http://logs.openstack.org/06/634706/2/check/openstack-tox-docs/63d96a6/html/ : SUCCESS in 2m 44s\n- build-openstack-releasenotes http://logs.openstack.org/06/634706/2/check/build-openstack-releasenotes/f8a0876/html/ : SUCCESS in 3m 55s\n- kolla-build-ubuntu-binary http://logs.openstack.org/06/634706/2/check/kolla-build-ubuntu-binary/bfbef97/ : FAILURE in 4m 02s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/06/634706/2/check/kolla-build-ubuntu-source/2946081/ : FAILURE in 4m 27s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/06/634706/2/check/kolla-ansible-ubuntu-source/5bf61ad/ : FAILURE in 8m 14s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/06/634706/2/check/kolla-ansible-ubuntu-binary/cec8710/ : FAILURE in 6m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/06/634706/2/check/kolla-ansible-ubuntu-source-upgrade/8c78550/ : FAILURE in 6m 34s","accounts_in_message":[],"_revision_number":2},{"id":"7170fd2a15ca5835fe8deeeab879f9366bb706bd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-10 06:38:16.000000000","message":"Patch Set 2: Code-Review-1\n\n\u003e docker-py is now installed from distribution packages for dind image.\n\nYeah but this is not changed here (affects commit message and release notes).\n\nI am for the option of considering skipping helm-repository image as a bug-fix rather than depreciation - it should never be binary in the first place it it is not available as such.","accounts_in_message":[],"_revision_number":2},{"id":"24cc0b6ed375d54a281599504b0943affff800d5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-10 06:38:33.000000000","message":"Patch Set 2:\n\nrecheck\n\nodd failure","accounts_in_message":[],"_revision_number":2},{"id":"a5ec335df4dee57472dd25d25d906b899591a2c1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-10 06:46:29.000000000","message":"Patch Set 2:\n\nMy bad, you based it off gem one as well. There is the source of failure.","accounts_in_message":[],"_revision_number":2},{"id":"bafe03f13ed166795c43a55ff0ab553a08603c22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 07:22:27.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/06/634706/2/check/kolla-build-centos-binary/a601787/ : FAILURE in 7m 36s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/06/634706/2/check/kolla-build-centos-source/7c6ee86/ : FAILURE in 5m 33s\n- kolla-ansible-centos-source http://logs.openstack.org/06/634706/2/check/kolla-ansible-centos-source/1181d1c/ : FAILURE in 8m 16s\n- kolla-ansible-centos-binary http://logs.openstack.org/06/634706/2/check/kolla-ansible-centos-binary/7cd4bfe/ : FAILURE in 15m 32s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/06/634706/2/check/kolla-ansible-centos-source-upgrade/c607b4e/ : FAILURE in 8m 28s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/06/634706/2/check/tripleo-build-containers-centos-7/416eb1a/ : FAILURE in 15m 28s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/06/634706/2/check/kolla-build-oraclelinux-binary/19f7543/ : FAILURE in 4m 50s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/06/634706/2/check/kolla-build-oraclelinux-source/8a4d4be/ : FAILURE in 5m 52s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/06/634706/2/check/kolla-ansible-oraclelinux-source/58e1891/ : FAILURE in 7m 41s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/06/634706/2/check/kolla-ansible-oraclelinux-binary/089435a/ : FAILURE in 8m 08s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/06/634706/2/check/openstack-tox-lower-constraints/9b00e28/ : SUCCESS in 3m 48s\n- openstack-tox-pep8 http://logs.openstack.org/06/634706/2/check/openstack-tox-pep8/7b3eb86/ : SUCCESS in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/06/634706/2/check/openstack-tox-py27/362446d/ : SUCCESS in 3m 04s\n- openstack-tox-py36 http://logs.openstack.org/06/634706/2/check/openstack-tox-py36/7e0e46f/ : SUCCESS in 4m 39s\n- openstack-tox-py37 http://logs.openstack.org/06/634706/2/check/openstack-tox-py37/f3d89ef/ : SUCCESS in 4m 15s\n- openstack-tox-docs http://logs.openstack.org/06/634706/2/check/openstack-tox-docs/ea5b2a0/html/ : SUCCESS in 3m 00s\n- build-openstack-releasenotes http://logs.openstack.org/06/634706/2/check/build-openstack-releasenotes/0efcea7/html/ : SUCCESS in 3m 47s\n- kolla-build-ubuntu-binary http://logs.openstack.org/06/634706/2/check/kolla-build-ubuntu-binary/b0625de/ : FAILURE in 3m 48s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/06/634706/2/check/kolla-build-ubuntu-source/118af3f/ : FAILURE in 3m 55s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/06/634706/2/check/kolla-ansible-ubuntu-source/0215a5a/ : FAILURE in 7m 30s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/06/634706/2/check/kolla-ansible-ubuntu-binary/6ca8b3c/ : FAILURE in 6m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/06/634706/2/check/kolla-ansible-ubuntu-source-upgrade/e380a1d/ : FAILURE in 6m 17s","accounts_in_message":[],"_revision_number":2},{"id":"58bd449682deb734ac622eb51f7e566a037f6125","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-10 07:32:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/06/634706/2/openstack-check/tripleo-build-containers-fedora-28-master/43ffcfb/ : SUCCESS in 40m 02s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/06/634706/2/openstack-check/tripleo-build-containers-fedora-28-stein/ae3a32a/ : SUCCESS in 48m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ce8c62813dffd4d86b49e511740aa97f0d152185","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-10 16:13:19.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"6a443f13593745b567665587087f9f90dc17afd5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-10 16:59:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/06/634706/3/openstack-check/tripleo-build-containers-fedora-28-master/fe59925/ : SUCCESS in 43m 18s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/06/634706/3/openstack-check/tripleo-build-containers-fedora-28-stein/1ab2c47/ : SUCCESS in 44m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"334d744197a5c99273bb2884abcc89c8612de8d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 19:26:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/06/634706/3/check/kolla-build-centos-binary/99a4424/ : SUCCESS in 47m 31s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/06/634706/3/check/kolla-build-centos-source/7a957b4/ : SUCCESS in 56m 15s\n- kolla-ansible-centos-source http://logs.openstack.org/06/634706/3/check/kolla-ansible-centos-source/b07af72/ : SUCCESS in 1h 23m 29s\n- kolla-ansible-centos-binary http://logs.openstack.org/06/634706/3/check/kolla-ansible-centos-binary/b7fc6ef/ : SUCCESS in 1h 11m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/06/634706/3/check/kolla-ansible-centos-source-upgrade/92bd331/ : SUCCESS in 1h 15m 31s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/06/634706/3/check/tripleo-build-containers-centos-7/ebd5704/ : SUCCESS in 41m 03s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/06/634706/3/check/kolla-build-oraclelinux-binary/4fe1307/ : SUCCESS in 38m 55s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/06/634706/3/check/kolla-build-oraclelinux-source/540619f/ : SUCCESS in 1h 02m 49s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/06/634706/3/check/kolla-ansible-oraclelinux-source/5ddc16f/ : SUCCESS in 1h 24m 12s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/06/634706/3/check/kolla-ansible-oraclelinux-binary/65b0f8c/ : SUCCESS in 1h 05m 09s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/06/634706/3/check/openstack-tox-lower-constraints/8bd4570/ : SUCCESS in 3m 57s\n- openstack-tox-pep8 http://logs.openstack.org/06/634706/3/check/openstack-tox-pep8/56319f9/ : SUCCESS in 4m 25s\n- openstack-tox-py27 http://logs.openstack.org/06/634706/3/check/openstack-tox-py27/b09c674/ : SUCCESS in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/06/634706/3/check/openstack-tox-py36/0f0edaa/ : SUCCESS in 3m 09s\n- openstack-tox-py37 http://logs.openstack.org/06/634706/3/check/openstack-tox-py37/8ebcdf7/ : SUCCESS in 3m 43s\n- openstack-tox-docs http://logs.openstack.org/06/634706/3/check/openstack-tox-docs/ffc7da9/html/ : SUCCESS in 3m 31s\n- build-openstack-releasenotes http://logs.openstack.org/06/634706/3/check/build-openstack-releasenotes/1e26aaa/html/ : SUCCESS in 4m 47s\n- kolla-build-ubuntu-binary http://logs.openstack.org/06/634706/3/check/kolla-build-ubuntu-binary/2c23bd4/ : FAILURE in 29m 36s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/06/634706/3/check/kolla-build-ubuntu-source/4194081/ : FAILURE in 30m 39s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/06/634706/3/check/kolla-ansible-ubuntu-source/2b9808e/ : SUCCESS in 52m 05s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/06/634706/3/check/kolla-ansible-ubuntu-binary/5b45b46/ : FAILURE in 45m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/06/634706/3/check/kolla-ansible-ubuntu-source-upgrade/7402187/ : SUCCESS in 1h 13m 13s","accounts_in_message":[],"_revision_number":3},{"id":"5ae96fcca9fa05819eabb15d400a752cfad90049","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-06 13:33:46.000000000","message":"Patch Set 3:\n\nNeeds a rebase","accounts_in_message":[],"_revision_number":3},{"id":"de8743e0ba5a7876ea3f04ba983e8389a49b70fb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-15 08:35:30.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7e2a85cb85ed529508ddf833f942c6f5a715ccc4","revisions":{"baa1e13b0e1189f2f7a3eb0c35d41021b17921c1":{"kind":"REWORK","_number":1,"created":"2019-02-04 15:00:33.000000000","uploader":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"ref":"refs/changes/06/634706/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/634706/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/634706/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/634706/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/634706/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/634706/1"}}},"commit":{"parents":[{"commit":"99e7c4c99dfebcf97a35ad4d50bef5abaff4e5b3","subject":"Do not gem install for binary type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/99e7c4c99dfebcf97a35ad4d50bef5abaff4e5b3"}]}],"author":{"name":"Martin André","email":"m.andre@redhat.com","date":"2019-02-01 14:12:46.000000000","tz":60},"committer":{"name":"Martin André","email":"m.andre@redhat.com","date":"2019-02-04 15:00:07.000000000","tz":60},"subject":"Do not pip install for binary type","message":"Do not pip install for binary type\n\nThis removes the few invocations of \"pip install\" that slipped into the\nbinary type templates.\n\ndocker-py is now installed from distribution packages for dind image.\nStop installing cradox via pip in gnocchi binary images as python-rados\nis now new enough.\nHelm-repository image is now skipped for binary images.\n\nChange-Id: Ie9f80eded6330a4ca1833018734b785b5be03b2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/baa1e13b0e1189f2f7a3eb0c35d41021b17921c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/baa1e13b0e1189f2f7a3eb0c35d41021b17921c1"}]},"branch":"refs/heads/master"},"a62a3dd3153384e0b6600d2f919077116d795a8b":{"kind":"REWORK","_number":2,"created":"2019-06-07 22:36:56.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/06/634706/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/634706/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/634706/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/634706/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/634706/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/634706/2"}}},"commit":{"parents":[{"commit":"9afdb60edaa6640b64d5e1346ed00dbfa6306a23","subject":"Do not gem install for binary type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9afdb60edaa6640b64d5e1346ed00dbfa6306a23"}]}],"author":{"name":"Martin André","email":"m.andre@redhat.com","date":"2019-02-01 14:12:46.000000000","tz":60},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-07 22:36:27.000000000","tz":-360},"subject":"Do not pip install for binary type","message":"Do not pip install for binary type\n\nThis removes the few invocations of \"pip install\" that slipped into the\nbinary type templates.\n\ndocker-py is now installed from distribution packages for dind image.\nStop installing cradox via pip in gnocchi binary images as python-rados\nis now new enough.\nHelm-repository image is now skipped for binary images.\n\nChange-Id: Ie9f80eded6330a4ca1833018734b785b5be03b2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a62a3dd3153384e0b6600d2f919077116d795a8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a62a3dd3153384e0b6600d2f919077116d795a8b"}]},"branch":"refs/heads/master"},"7e2a85cb85ed529508ddf833f942c6f5a715ccc4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-10 16:13:19.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/06/634706/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/634706/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/634706/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/634706/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/634706/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/634706/3"}}},"commit":{"parents":[{"commit":"4dff9ee3eb809bc5bbbdeebbf98f854a9f51107a","subject":"Do not gem install for binary type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4dff9ee3eb809bc5bbbdeebbf98f854a9f51107a"}]}],"author":{"name":"Martin André","email":"m.andre@redhat.com","date":"2019-02-01 14:12:46.000000000","tz":60},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-10 16:13:19.000000000","tz":0},"subject":"Do not pip install for binary type","message":"Do not pip install for binary type\n\nThis removes the few invocations of \"pip install\" that slipped into the\nbinary type templates.\n\ndocker-py is now installed from distribution packages for dind image.\nStop installing cradox via pip in gnocchi binary images as python-rados\nis now new enough.\nHelm-repository image is now skipped for binary images.\n\nChange-Id: Ie9f80eded6330a4ca1833018734b785b5be03b2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e2a85cb85ed529508ddf833f942c6f5a715ccc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e2a85cb85ed529508ddf833f942c6f5a715ccc4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
