)]}'
{"id":"openstack%2Fkolla~636403","triplet_id":"openstack%2Fkolla~master~I0a7630e07c5100199236170f745f2c8148df6c42","project":"openstack/kolla","branch":"master","topic":"package-overrides","hashtags":[],"change_id":"I0a7630e07c5100199236170f745f2c8148df6c42","subject":"Add package-replacements files","status":"ABANDONED","created":"2019-02-12 17:56:52.000000000","updated":"2019-11-04 20:02:16.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"03925cdef153042c28283ab7f901dc6c27c7caf7","_number":636403,"virtual_id_number":636403,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"date":"2019-05-21 16:25:11.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-05-20 21:29:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"date":"2019-05-20 20:25:15.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"all":[{"value":1,"date":"2019-05-21 08:36:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":-1,"date":"2019-05-21 17:08:56.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-13 09:44:05.000000000","updated_by":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"reviewer":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"state":"REVIEWER"},{"updated":"2019-02-26 16:52:52.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2019-05-20 20:25:15.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-20 21:29:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-21 08:36:52.000000000","updated_by":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"reviewer":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"state":"REVIEWER"},{"updated":"2019-05-21 16:25:11.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"690da8e149792809b3a0802aa2336c737a018b38","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-02-12 17:56:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80c25f19cc8d31874a49d275656fa293094236b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-12 19:51:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/03/636403/1/check/kolla-build-centos-binary/fbb8d1c/ : FAILURE in 14m 52s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/03/636403/1/check/kolla-build-centos-source/feb53f8/ : FAILURE in 14m 59s\n- kolla-ansible-centos-source http://logs.openstack.org/03/636403/1/check/kolla-ansible-centos-source/bbb5c24/ : SUCCESS in 1h 12m 33s\n- kolla-ansible-centos-binary http://logs.openstack.org/03/636403/1/check/kolla-ansible-centos-binary/0224d4d/ : FAILURE in 1h 03m 28s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/03/636403/1/check/tripleo-build-containers-centos-7/f938a38/ : SUCCESS in 37m 47s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/03/636403/1/check/kolla-build-oraclelinux-binary/cc3683e/ : SUCCESS in 1h 36m 55s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/03/636403/1/check/kolla-build-oraclelinux-source/00143a4/ : SUCCESS in 1h 20m 09s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/03/636403/1/check/kolla-ansible-oraclelinux-source/eb26c89/ : SUCCESS in 1h 14m 12s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/03/636403/1/check/kolla-ansible-oraclelinux-binary/791ed91/ : FAILURE in 1h 03m 42s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/03/636403/1/check/openstack-tox-lower-constraints/f124de8/ : FAILURE in 4m 55s\n- openstack-tox-py36 http://logs.openstack.org/03/636403/1/check/openstack-tox-py36/ca1f111/ : FAILURE in 5m 26s\n- openstack-tox-pep8 http://logs.openstack.org/03/636403/1/check/openstack-tox-pep8/c519946/ : SUCCESS in 4m 24s\n- openstack-tox-py27 http://logs.openstack.org/03/636403/1/check/openstack-tox-py27/a7168c2/ : SUCCESS in 4m 30s\n- openstack-tox-py35 http://logs.openstack.org/03/636403/1/check/openstack-tox-py35/7291437/ : FAILURE in 5m 35s\n- openstack-tox-docs http://logs.openstack.org/03/636403/1/check/openstack-tox-docs/4972c1f/html/ : SUCCESS in 4m 10s\n- kolla-build-ubuntu-binary http://logs.openstack.org/03/636403/1/check/kolla-build-ubuntu-binary/57a90c7/ : SUCCESS in 1h 28m 08s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/03/636403/1/check/kolla-build-ubuntu-source/9cd50b8/ : FAILURE in 55m 28s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/03/636403/1/check/kolla-ansible-ubuntu-source/6a606b6/ : FAILURE in 33m 09s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/03/636403/1/check/kolla-ansible-ubuntu-binary/899aead/ : FAILURE in 1h 15m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d447fdab1e515f667a7d90bb2c9d5289b3708eb7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-02-14 23:19:38.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b2b658dea1d542ab1d2d7ab52506a46aa05fdb9b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-02-14 23:19:41.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"89f4e7797745a76e08c580845565c454945275bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-15 02:25:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/03/636403/3/check/kolla-build-centos-binary/19d2b63/ : SUCCESS in 1h 01m 06s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/03/636403/3/check/kolla-build-centos-source/9a110e2/ : SUCCESS in 1h 30m 12s\n- kolla-ansible-centos-source http://logs.openstack.org/03/636403/3/check/kolla-ansible-centos-source/1ba87e1/ : SUCCESS in 1h 28m 14s\n- kolla-ansible-centos-binary http://logs.openstack.org/03/636403/3/check/kolla-ansible-centos-binary/608d76c/ : SUCCESS in 1h 12m 07s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/03/636403/3/check/tripleo-build-containers-centos-7/10104e1/ : SUCCESS in 48m 21s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/03/636403/3/check/kolla-build-oraclelinux-binary/2fd90bc/ : SUCCESS in 1h 18m 41s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/03/636403/3/check/kolla-build-oraclelinux-source/ce3beb6/ : SUCCESS in 1h 01m 31s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/03/636403/3/check/kolla-ansible-oraclelinux-source/06363e1/ : SUCCESS in 1h 13m 25s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/03/636403/3/check/kolla-ansible-oraclelinux-binary/f02a6d4/ : SUCCESS in 1h 14m 50s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/03/636403/3/check/openstack-tox-lower-constraints/e6560cf/ : SUCCESS in 4m 32s\n- openstack-tox-py36 http://logs.openstack.org/03/636403/3/check/openstack-tox-py36/b6feae6/ : SUCCESS in 3m 46s\n- openstack-tox-pep8 http://logs.openstack.org/03/636403/3/check/openstack-tox-pep8/c52ef0e/ : SUCCESS in 3m 10s\n- openstack-tox-py27 http://logs.openstack.org/03/636403/3/check/openstack-tox-py27/66deb7b/ : SUCCESS in 3m 04s\n- openstack-tox-py35 http://logs.openstack.org/03/636403/3/check/openstack-tox-py35/05cfe55/ : SUCCESS in 3m 41s\n- openstack-tox-docs http://logs.openstack.org/03/636403/3/check/openstack-tox-docs/c1649c2/html/ : SUCCESS in 2m 42s\n- kolla-build-ubuntu-binary http://logs.openstack.org/03/636403/3/check/kolla-build-ubuntu-binary/8158529/ : SUCCESS in 1h 15m 39s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/03/636403/3/check/kolla-build-ubuntu-source/2162a6c/ : SUCCESS in 31m 31s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/03/636403/3/check/kolla-ansible-ubuntu-source/1ec408b/ : SUCCESS in 59m 41s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/03/636403/3/check/kolla-ansible-ubuntu-binary/36a972c/ : TIMED_OUT in 2h 01m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d9c4215e94e59e1e0743106cec0cd8b85e55a68e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-03-29 22:54:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8899ebb137325ac94673a185b7204e27970575e4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-29 23:30:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 http://logs.rdoproject.org/03/636403/4/openstack-check/tripleo-build-containers-fedora-28/62c3501/ : SUCCESS in 32m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cb8474a43be8265304f3cdb0301a79dbcb5a147a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-30 00:16:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/03/636403/4/check/kolla-build-centos-binary/d5a5e2c/ : SUCCESS in 41m 53s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/03/636403/4/check/kolla-build-centos-source/abf29f7/ : SUCCESS in 48m 37s\n- kolla-ansible-centos-source http://logs.openstack.org/03/636403/4/check/kolla-ansible-centos-source/d269f99/ : SUCCESS in 1h 17m 10s\n- kolla-ansible-centos-binary http://logs.openstack.org/03/636403/4/check/kolla-ansible-centos-binary/be931bb/ : SUCCESS in 1h 05m 32s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/03/636403/4/check/tripleo-build-containers-centos-7/26ac982/ : SUCCESS in 48m 36s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/03/636403/4/check/kolla-build-oraclelinux-binary/2ebce9f/ : SUCCESS in 53m 34s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/03/636403/4/check/kolla-build-oraclelinux-source/6f16e31/ : SUCCESS in 45m 51s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/03/636403/4/check/kolla-ansible-oraclelinux-source/bb116cc/ : SUCCESS in 1h 10m 07s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/03/636403/4/check/kolla-ansible-oraclelinux-binary/22c8b2c/ : SUCCESS in 1h 09m 58s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/03/636403/4/check/openstack-tox-lower-constraints/0f76baa/ : SUCCESS in 3m 59s\n- openstack-tox-pep8 http://logs.openstack.org/03/636403/4/check/openstack-tox-pep8/a7ba3f2/ : SUCCESS in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/03/636403/4/check/openstack-tox-py27/eb2da9e/ : SUCCESS in 3m 22s\n- openstack-tox-py36 http://logs.openstack.org/03/636403/4/check/openstack-tox-py36/606008a/ : SUCCESS in 2m 48s\n- openstack-tox-py37 http://logs.openstack.org/03/636403/4/check/openstack-tox-py37/b455cea/ : SUCCESS in 3m 59s\n- openstack-tox-docs http://logs.openstack.org/03/636403/4/check/openstack-tox-docs/e923e54/html/ : SUCCESS in 2m 36s\n- kolla-build-ubuntu-binary http://logs.openstack.org/03/636403/4/check/kolla-build-ubuntu-binary/c6f642f/ : SUCCESS in 38m 49s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/03/636403/4/check/kolla-build-ubuntu-source/e098a49/ : SUCCESS in 31m 11s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/03/636403/4/check/kolla-ansible-ubuntu-source/eaead92/ : SUCCESS in 56m 33s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/03/636403/4/check/kolla-ansible-ubuntu-binary/67be3e8/ : FAILURE in 33m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"62ccc76ca6c9d0dd61a6d8490a58668df42bd5da","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-20 19:48:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b16fa7b35008e8692125be850bc1ff4849458abd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-20 20:25:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/03/636403/5/openstack-check/tripleo-build-containers-fedora-28-master/4e94c8c/ : SUCCESS in 32m 43s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/03/636403/5/openstack-check/tripleo-build-containers-fedora-28-stein/61da206/ : SUCCESS in 35m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"942dc61eee17ab0e2d1c3fe477f2fbd73364aefe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 21:29:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/03/636403/5/check/kolla-build-centos-binary/ab49b06/ : SUCCESS in 49m 04s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/03/636403/5/check/kolla-build-centos-source/eda4f9d/ : SUCCESS in 52m 33s\n- kolla-ansible-centos-source http://logs.openstack.org/03/636403/5/check/kolla-ansible-centos-source/0dedc51/ : SUCCESS in 1h 02m 31s\n- kolla-ansible-centos-binary http://logs.openstack.org/03/636403/5/check/kolla-ansible-centos-binary/803eb22/ : SUCCESS in 1h 02m 49s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/03/636403/5/check/kolla-ansible-centos-source-upgrade/008c957/ : SUCCESS in 1h 22m 24s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/03/636403/5/check/tripleo-build-containers-centos-7/f0d7ee9/ : SUCCESS in 39m 17s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/03/636403/5/check/kolla-build-oraclelinux-binary/193025f/ : SUCCESS in 57m 18s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/03/636403/5/check/kolla-build-oraclelinux-source/b0f8678/ : SUCCESS in 50m 04s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/03/636403/5/check/kolla-ansible-oraclelinux-source/6e1fd2a/ : SUCCESS in 1h 10m 13s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/03/636403/5/check/kolla-ansible-oraclelinux-binary/15415f9/ : SUCCESS in 1h 21m 42s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/03/636403/5/check/openstack-tox-lower-constraints/5310e63/ : SUCCESS in 4m 46s\n- openstack-tox-pep8 http://logs.openstack.org/03/636403/5/check/openstack-tox-pep8/ff32b16/ : SUCCESS in 3m 31s\n- openstack-tox-py27 http://logs.openstack.org/03/636403/5/check/openstack-tox-py27/753cb9d/ : SUCCESS in 3m 06s\n- openstack-tox-py36 http://logs.openstack.org/03/636403/5/check/openstack-tox-py36/c356f52/ : SUCCESS in 3m 00s\n- openstack-tox-py37 http://logs.openstack.org/03/636403/5/check/openstack-tox-py37/b6c7be4/ : SUCCESS in 3m 32s\n- openstack-tox-docs http://logs.openstack.org/03/636403/5/check/openstack-tox-docs/3b4bae9/html/ : SUCCESS in 3m 09s\n- kolla-build-ubuntu-binary http://logs.openstack.org/03/636403/5/check/kolla-build-ubuntu-binary/94eae6f/ : SUCCESS in 28m 17s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/03/636403/5/check/kolla-build-ubuntu-source/7e8481d/ : SUCCESS in 28m 38s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/03/636403/5/check/kolla-ansible-ubuntu-source/3e4bece/ : SUCCESS in 45m 08s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/03/636403/5/check/kolla-ansible-ubuntu-binary/ece4664/ : SUCCESS in 48m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/03/636403/5/check/kolla-ansible-ubuntu-source-upgrade/50d6012/ : SUCCESS in 1h 01m 50s","accounts_in_message":[],"_revision_number":5},{"id":"8b6bac529e93611f0e038cd1e7cfe3ae6431baab","author":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"date":"2019-05-21 08:36:52.000000000","message":"Patch Set 5: Code-Review+1\n\nadding the unit test for this would be good.","accounts_in_message":[],"_revision_number":5},{"id":"40d044b0b981ecf304ce9709368a7d7505b9d29c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-21 08:47:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3929bcfd51967a5153269edf4f0fbc0876b541af","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-21 14:51:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b23933433182d614d0acecce8310bb1f8894047d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-21 16:25:11.000000000","message":"Patch Set 5:\n\n(3 comments)\n\nPart of me is a little concerned this could be use to maintain downstream fixes and could discourage sharing them. Hopefully not.","accounts_in_message":[],"_revision_number":5},{"id":"cb9df34f9e243263dbba7264f4e8451fa4c39867","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-21 17:08:56.000000000","message":"Patch Set 5: Workflow-1\n\nImho it\u0027s not different than override template files for downstream except this is just package names. I think that\u0027s less I trusive than the template override. I\u0027ll try and throw an example somewhere","accounts_in_message":[],"_revision_number":5},{"id":"bb7dc6d5d65361147712baedc30b4e5605ca7210","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-11-04 20:02:16.000000000","message":"Abandoned\n\ndon\u0027t have time to address this","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1d0f125ef3da10411c9d506b8b0351eb97799006","revisions":{"83596252c794cc0cc62916e3af20aaf951fa02d3":{"kind":"REWORK","_number":1,"created":"2019-02-12 17:56:52.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/03/636403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/636403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/636403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/636403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/636403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/636403/1"}}},"commit":{"parents":[{"commit":"5922c22d8af6c34eb5c34250c3fa90dc3f853688","subject":"Merge \"remove pep8 from lower-constraints.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5922c22d8af6c34eb5c34250c3fa90dc3f853688"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-02-12 17:47:58.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-02-12 17:47:58.000000000","tz":-420},"subject":"WIP: Add package-replacement files","message":"WIP: Add package-replacement files\n\nAs an alternative to having if/else for python3 specific packages in the\ndocker files, this adds a --package-replacement file option to allow a\nuser to pass a basic dictionary of { \u0027current-packagename\u0027:\n\u0027new-packagename\u0027, ... } to have the packages replaced when built.  The\nprevious proposal was https://review.openstack.org/#/c/632156/\n\nIMHO the problem with this solution is that things like python versions\nin filenames (for wsgi, etc) are not able to managed in a similar\nfashion.\n\nChange-Id: I0a7630e07c5100199236170f745f2c8148df6c42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83596252c794cc0cc62916e3af20aaf951fa02d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83596252c794cc0cc62916e3af20aaf951fa02d3"}]},"branch":"refs/heads/master"},"78c13b7c10edf3c32d601e893df80d36a4790bc0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-02-14 23:19:38.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/03/636403/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/636403/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/636403/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/636403/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/636403/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/636403/2"}}},"commit":{"parents":[{"commit":"5922c22d8af6c34eb5c34250c3fa90dc3f853688","subject":"Merge \"remove pep8 from lower-constraints.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5922c22d8af6c34eb5c34250c3fa90dc3f853688"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-02-12 17:47:58.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-02-14 23:19:35.000000000","tz":0},"subject":"Add package-replacement files","message":"Add package-replacement files\n\nAs an alternative to having if/else for python3 specific packages in the\ndocker files, this adds a --package-replacement file option to allow a\nuser to pass a basic dictionary of { \u0027current-packagename\u0027:\n\u0027new-packagename\u0027, ... } to have the packages replaced when built.  The\nprevious proposal was https://review.openstack.org/#/c/632156/\n\nIMHO the problem with this solution is that things like python versions\nin filenames (for wsgi, etc) are not able to managed in a similar\nfashion.\n\nChange-Id: I0a7630e07c5100199236170f745f2c8148df6c42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78c13b7c10edf3c32d601e893df80d36a4790bc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78c13b7c10edf3c32d601e893df80d36a4790bc0"}]},"branch":"refs/heads/master"},"010fe652feb3e301d3c5f045930ae0c0d85d2f47":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-02-14 23:19:41.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/03/636403/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/636403/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/636403/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/636403/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/636403/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/636403/3"}}},"commit":{"parents":[{"commit":"a490eef0d5e92a0228656870ebff49bcf26f0472","subject":"Fix py36 gate job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a490eef0d5e92a0228656870ebff49bcf26f0472"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-02-12 17:47:58.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-02-14 23:19:41.000000000","tz":0},"subject":"Add package-replacement files","message":"Add package-replacement files\n\nAs an alternative to having if/else for python3 specific packages in the\ndocker files, this adds a --package-replacement file option to allow a\nuser to pass a basic dictionary of { \u0027current-packagename\u0027:\n\u0027new-packagename\u0027, ... } to have the packages replaced when built.  The\nprevious proposal was https://review.openstack.org/#/c/632156/\n\nIMHO the problem with this solution is that things like python versions\nin filenames (for wsgi, etc) are not able to managed in a similar\nfashion.\n\nChange-Id: I0a7630e07c5100199236170f745f2c8148df6c42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/010fe652feb3e301d3c5f045930ae0c0d85d2f47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/010fe652feb3e301d3c5f045930ae0c0d85d2f47"}]},"branch":"refs/heads/master"},"17a1346910b614621e90b17ac265acd220598e7e":{"kind":"REWORK","_number":4,"created":"2019-03-29 22:54:05.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/03/636403/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/636403/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/636403/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/636403/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/636403/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/636403/4"}}},"commit":{"parents":[{"commit":"636fce475b7b1525f4ad353d2a0740f784d22aad","subject":"Merge \"Fixed file modes in toolbox:/usr/share/ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/636fce475b7b1525f4ad353d2a0740f784d22aad"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-02-12 17:47:58.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-03-29 22:54:00.000000000","tz":-360},"subject":"Add package-replacement files","message":"Add package-replacement files\n\nAs an alternative to having if/else for python3 specific packages in the\ndocker files, this adds a --package-replacement file option to allow a\nuser to pass a basic dictionary of { \u0027current-packagename\u0027:\n\u0027new-packagename\u0027, ... } to have the packages replaced when built.  The\nprevious proposal was https://review.openstack.org/#/c/632156/\n\nIMHO the problem with this solution is that things like python versions\nin filenames (for wsgi, etc) are not able to managed in a similar\nfashion.\n\nChange-Id: I0a7630e07c5100199236170f745f2c8148df6c42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17a1346910b614621e90b17ac265acd220598e7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17a1346910b614621e90b17ac265acd220598e7e"}]},"branch":"refs/heads/master"},"1d0f125ef3da10411c9d506b8b0351eb97799006":{"kind":"REWORK","_number":5,"created":"2019-05-20 19:48:17.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/03/636403/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/636403/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/636403/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/636403/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/636403/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/636403/5"}}},"commit":{"parents":[{"commit":"b55fa2f8b70b143068db57fc1ff844929d29486e","subject":"Merge \"Update Python 3 test runtimes for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b55fa2f8b70b143068db57fc1ff844929d29486e"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-02-12 17:47:58.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-20 19:47:14.000000000","tz":-360},"subject":"Add package-replacements files","message":"Add package-replacements files\n\nAs an alternative to having if/else for python3 specific packages in the\ndocker files, this adds a --package-replacements file option to allow a\nuser to pass a basic dictionary of { \u0027current-packagename\u0027:\n\u0027new-packagename\u0027, ... } to have the packages replaced when built.  The\nprevious proposal was https://review.openstack.org/#/c/632156/\n\nThis would prevent having to do an append/remove from packages list when\nthere\u0027s an alternate package name.\n\nChange-Id: I0a7630e07c5100199236170f745f2c8148df6c42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d0f125ef3da10411c9d506b8b0351eb97799006"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d0f125ef3da10411c9d506b8b0351eb97799006"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
