)]}'
{"id":"openstack%2Fkolla~638692","triplet_id":"openstack%2Fkolla~master~Ic1b25b37d06b429a112f1418a79d385d505bb1e9","project":"openstack/kolla","branch":"master","topic":"pedro/fluentd-ubuntu-master","hashtags":[],"change_id":"Ic1b25b37d06b429a112f1418a79d385d505bb1e9","subject":"Install fluentd gem on debian/ubuntu for non x86_64","status":"MERGED","created":"2019-02-22 15:50:10.000000000","updated":"2019-05-28 19:04:16.000000000","submitted":"2019-05-28 19:04:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"638692-1559070256982-08ba7da0","meta_rev_id":"9ff42c138796e9967238cc404750a6d689038850","_number":638692,"virtual_id_number":638692,"owner":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2019-05-28 16:31:56.000000000","_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},{"value":2,"date":"2019-05-28 19:04:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-28 17:08:15.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":2,"date":"2019-05-07 18:40:47.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-05-28 15:11:54.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-05-28 15:11:54.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-25 12:06:11.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-03-05 03:44:34.000000000","updated_by":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"reviewer":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"state":"REVIEWER"},{"updated":"2019-03-06 08:11:16.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-03-13 14:42:27.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2019-05-07 18:40:47.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-05-28 15:11:54.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-05-28 17:08:15.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-28 19:04:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"36ee1e5ece123277d0898f75895f4fecfec537d0","author":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"date":"2019-02-22 15:50:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83dd494780448582438667cf4f95f3e349d64965","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-22 17:53:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/92/638692/1/check/kolla-build-centos-binary/4722452/ : SUCCESS in 37m 11s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/92/638692/1/check/kolla-build-centos-source/f36bd17/ : SUCCESS in 55m 31s\n- kolla-ansible-centos-source http://logs.openstack.org/92/638692/1/check/kolla-ansible-centos-source/2b90572/ : SUCCESS in 1h 28m 23s\n- kolla-ansible-centos-binary http://logs.openstack.org/92/638692/1/check/kolla-ansible-centos-binary/b25a5dc/ : SUCCESS in 1h 12m 39s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/92/638692/1/check/tripleo-build-containers-centos-7/c66a168/ : SUCCESS in 47m 36s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/92/638692/1/check/kolla-build-oraclelinux-binary/92b094d/ : SUCCESS in 41m 06s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/92/638692/1/check/kolla-build-oraclelinux-source/d30ecf9/ : SUCCESS in 52m 26s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/92/638692/1/check/kolla-ansible-oraclelinux-source/e5d3c01/ : SUCCESS in 1h 15m 29s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/92/638692/1/check/kolla-ansible-oraclelinux-binary/c38d71a/ : SUCCESS in 1h 13m 33s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/92/638692/1/check/openstack-tox-lower-constraints/3342952/ : SUCCESS in 3m 52s\n- openstack-tox-py36 http://logs.openstack.org/92/638692/1/check/openstack-tox-py36/33b4863/ : SUCCESS in 3m 42s\n- openstack-tox-pep8 http://logs.openstack.org/92/638692/1/check/openstack-tox-pep8/be63298/ : SUCCESS in 3m 54s\n- openstack-tox-py27 http://logs.openstack.org/92/638692/1/check/openstack-tox-py27/d8cb70c/ : SUCCESS in 4m 02s\n- openstack-tox-py35 http://logs.openstack.org/92/638692/1/check/openstack-tox-py35/9e4e745/ : SUCCESS in 3m 43s\n- openstack-tox-docs http://logs.openstack.org/92/638692/1/check/openstack-tox-docs/bb693ef/html/ : SUCCESS in 2m 52s\n- kolla-build-ubuntu-binary http://logs.openstack.org/92/638692/1/check/kolla-build-ubuntu-binary/cbb92cb/ : SUCCESS in 1h 33m 30s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/92/638692/1/check/kolla-build-ubuntu-source/435c9ac/ : SUCCESS in 30m 11s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/92/638692/1/check/kolla-ansible-ubuntu-source/5b2e062/ : SUCCESS in 1h 00m 28s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/92/638692/1/check/kolla-ansible-ubuntu-binary/553c933/ : TIMED_OUT in 2h 01m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b30611f258e445226a436aa3bbe6e5224563c9a0","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-02-25 12:06:11.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5c422a0c9270032d6a79278efa95b0b127c69531","author":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"date":"2019-02-25 12:20:18.000000000","message":"Patch Set 1:\n\n(3 comments)\n\nThanks for the comments :)\n\nI think the first of your suggestions is not possible though, without moving more things around.","accounts_in_message":[],"_revision_number":1},{"id":"ad73c58666a54b87e9111d941402b7fbb8ca143b","author":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"date":"2019-02-25 12:46:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d5d44f650c025f5265f99f4d02d64f7ace6e87c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-25 14:25:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/92/638692/2/check/kolla-build-centos-binary/7f823db/ : SUCCESS in 44m 43s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/92/638692/2/check/kolla-build-centos-source/84671de/ : SUCCESS in 40m 36s\n- kolla-ansible-centos-source http://logs.openstack.org/92/638692/2/check/kolla-ansible-centos-source/b915eae/ : SUCCESS in 1h 12m 01s\n- kolla-ansible-centos-binary http://logs.openstack.org/92/638692/2/check/kolla-ansible-centos-binary/dd35956/ : SUCCESS in 1h 09m 35s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/92/638692/2/check/tripleo-build-containers-centos-7/21753d6/ : SUCCESS in 55m 04s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/92/638692/2/check/kolla-build-oraclelinux-binary/3ea9101/ : SUCCESS in 39m 26s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/92/638692/2/check/kolla-build-oraclelinux-source/ca34368/ : SUCCESS in 39m 23s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/92/638692/2/check/kolla-ansible-oraclelinux-source/48be5ba/ : SUCCESS in 1h 15m 30s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/92/638692/2/check/kolla-ansible-oraclelinux-binary/81731a8/ : SUCCESS in 1h 13m 38s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/92/638692/2/check/openstack-tox-lower-constraints/3f0d214/ : SUCCESS in 4m 15s\n- openstack-tox-py36 http://logs.openstack.org/92/638692/2/check/openstack-tox-py36/c432d57/ : SUCCESS in 2m 47s\n- openstack-tox-pep8 http://logs.openstack.org/92/638692/2/check/openstack-tox-pep8/8e2b2e5/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/92/638692/2/check/openstack-tox-py27/7e26547/ : SUCCESS in 3m 16s\n- openstack-tox-py35 http://logs.openstack.org/92/638692/2/check/openstack-tox-py35/57e2128/ : SUCCESS in 4m 06s\n- openstack-tox-docs http://logs.openstack.org/92/638692/2/check/openstack-tox-docs/3f47a78/html/ : SUCCESS in 2m 50s\n- kolla-build-ubuntu-binary http://logs.openstack.org/92/638692/2/check/kolla-build-ubuntu-binary/033e7dd/ : SUCCESS in 1h 21m 11s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/92/638692/2/check/kolla-build-ubuntu-source/bcc4108/ : SUCCESS in 26m 09s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/92/638692/2/check/kolla-ansible-ubuntu-source/aa620a9/ : SUCCESS in 1h 01m 07s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/92/638692/2/check/kolla-ansible-ubuntu-binary/d64ca70/ : SUCCESS in 1h 28m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ecddb47f570ce3f2425d22a0ab354c6868254c17","author":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"date":"2019-02-26 09:20:22.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"05ccb2b876bd37937fb2e338095120732597dc47","author":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"date":"2019-03-05 03:44:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1e6b419a5e4b9e04d459cc711e67db938c60e1ea","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-03-06 08:11:16.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b74aaa755f0181e9c3cc6b25db87e16ca4edcebd","author":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"date":"2019-03-12 10:24:44.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"316ee2f094bd9d05818ba3ff4f470ca2e7ec541f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-12 12:06:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/92/638692/3/check/kolla-build-centos-binary/c617252/ : SUCCESS in 42m 59s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/92/638692/3/check/kolla-build-centos-source/cf012fa/ : SUCCESS in 48m 46s\n- kolla-ansible-centos-source http://logs.openstack.org/92/638692/3/check/kolla-ansible-centos-source/22c50cd/ : SUCCESS in 1h 35m 09s\n- kolla-ansible-centos-binary http://logs.openstack.org/92/638692/3/check/kolla-ansible-centos-binary/19ea33d/ : SUCCESS in 1h 10m 03s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/92/638692/3/check/tripleo-build-containers-centos-7/fb59d00/ : SUCCESS in 42m 42s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/92/638692/3/check/kolla-build-oraclelinux-binary/0e2e412/ : SUCCESS in 42m 26s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/92/638692/3/check/kolla-build-oraclelinux-source/557ca97/ : SUCCESS in 53m 47s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/92/638692/3/check/kolla-ansible-oraclelinux-source/3fc4c7f/ : SUCCESS in 1h 14m 35s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/92/638692/3/check/kolla-ansible-oraclelinux-binary/b11a527/ : SUCCESS in 1h 10m 45s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/92/638692/3/check/openstack-tox-lower-constraints/1a9de3a/ : SUCCESS in 5m 08s\n- openstack-tox-pep8 http://logs.openstack.org/92/638692/3/check/openstack-tox-pep8/f1aff46/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/92/638692/3/check/openstack-tox-py27/a68ce54/ : SUCCESS in 4m 41s\n- openstack-tox-py36 http://logs.openstack.org/92/638692/3/check/openstack-tox-py36/111e810/ : SUCCESS in 3m 40s\n- openstack-tox-py37 http://logs.openstack.org/92/638692/3/check/openstack-tox-py37/6691907/ : SUCCESS in 5m 22s\n- openstack-tox-docs http://logs.openstack.org/92/638692/3/check/openstack-tox-docs/7b55fb5/html/ : SUCCESS in 3m 57s\n- kolla-build-ubuntu-binary http://logs.openstack.org/92/638692/3/check/kolla-build-ubuntu-binary/ffe5d1c/ : SUCCESS in 31m 09s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/92/638692/3/check/kolla-build-ubuntu-source/1e88b11/ : FAILURE in 17m 06s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/92/638692/3/check/kolla-ansible-ubuntu-source/2083e40/ : SUCCESS in 1h 10m 19s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/92/638692/3/check/kolla-ansible-ubuntu-binary/f9bdbbe/ : FAILURE in 30m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3b57617b0bea5a5881b4be5725a5285f7f20692b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-12 12:25:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 tripleo-build-containers-fedora-28 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"adfc346cc94540a91c08698489ef5c48755bf4d2","author":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"date":"2019-03-12 12:26:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c2dd924b652ebec50a8fb3f6f65e3d2628dc4ba5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-12 13:51:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/92/638692/3/check/kolla-build-centos-binary/7f63292/ : SUCCESS in 38m 58s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/92/638692/3/check/kolla-build-centos-source/8fb33f2/ : SUCCESS in 44m 44s\n- kolla-ansible-centos-source http://logs.openstack.org/92/638692/3/check/kolla-ansible-centos-source/386741d/ : SUCCESS in 1h 15m 08s\n- kolla-ansible-centos-binary http://logs.openstack.org/92/638692/3/check/kolla-ansible-centos-binary/d05d826/ : SUCCESS in 1h 13m 11s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/92/638692/3/check/tripleo-build-containers-centos-7/5eaa1a4/ : SUCCESS in 43m 55s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/92/638692/3/check/kolla-build-oraclelinux-binary/48cc935/ : SUCCESS in 48m 52s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/92/638692/3/check/kolla-build-oraclelinux-source/c2e1f49/ : SUCCESS in 43m 44s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/92/638692/3/check/kolla-ansible-oraclelinux-source/df9bc19/ : SUCCESS in 1h 20m 05s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/92/638692/3/check/kolla-ansible-oraclelinux-binary/8491645/ : SUCCESS in 1h 13m 45s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/92/638692/3/check/openstack-tox-lower-constraints/8732845/ : SUCCESS in 4m 06s\n- openstack-tox-pep8 http://logs.openstack.org/92/638692/3/check/openstack-tox-pep8/8516801/ : SUCCESS in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/92/638692/3/check/openstack-tox-py27/ee076c4/ : SUCCESS in 5m 21s\n- openstack-tox-py36 http://logs.openstack.org/92/638692/3/check/openstack-tox-py36/a992b9c/ : SUCCESS in 3m 11s\n- openstack-tox-py37 http://logs.openstack.org/92/638692/3/check/openstack-tox-py37/f72d7a7/ : SUCCESS in 3m 10s\n- openstack-tox-docs http://logs.openstack.org/92/638692/3/check/openstack-tox-docs/996f0ab/html/ : SUCCESS in 2m 28s\n- kolla-build-ubuntu-binary http://logs.openstack.org/92/638692/3/check/kolla-build-ubuntu-binary/da11cd1/ : SUCCESS in 36m 49s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/92/638692/3/check/kolla-build-ubuntu-source/210dbfc/ : SUCCESS in 39m 46s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/92/638692/3/check/kolla-ansible-ubuntu-source/99211d3/ : SUCCESS in 1h 05m 08s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/92/638692/3/check/kolla-ansible-ubuntu-binary/8732782/ : FAILURE in 32m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3db906020aed3a988a2dd6f3eef92de0f0a03ec5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-12 14:04:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 tripleo-build-containers-fedora-28 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"97facd0014aac249ed424dddc78c0523e8891f33","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2019-03-13 14:42:27.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"4962e409d4f3cc8c58862b9b1b31c29e303d41e3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-03-20 16:12:31.000000000","message":"Patch Set 3: Code-Review+1\n\nHow many of gems installed with fluentd can be installed from packages?\n\nFor binary builds I would prefer to install them from packages.","accounts_in_message":[],"_revision_number":3},{"id":"1be781f18353b30354755677fcd53881575d7837","author":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"date":"2019-03-20 16:58:16.000000000","message":"Patch Set 3: Workflow-1\n\nI just noticed that I need to also change some bits in `docker/fluentd/extend_start.sh`","accounts_in_message":[],"_revision_number":3},{"id":"3b0f4352afb3a7cc3fcf3b98f969da53adef73c4","author":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"date":"2019-03-20 17:00:37.000000000","message":"Patch Set 3:\n\n\u003e How many of gems installed with fluentd can be installed from\n \u003e packages?\n \u003e \n \u003e For binary builds I would prefer to install them from packages.\n\nI haven\u0027t looked into that, given that kolla was already using gems for those, and I wanted to do the minimal change possible to support arm64 builds.\n\nThis patch only installs an extra gem that we weren\u0027t doing before, only for debian/ubuntu on architectures other than x86_64: fluentd","accounts_in_message":[],"_revision_number":3},{"id":"ab6716b44da37142c918a97ebe8e99d52dadde38","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-03-20 17:07:36.000000000","message":"Patch Set 3:\n\nok then","accounts_in_message":[],"_revision_number":3},{"id":"f3b9b82d4c445e2c012188003a2b34c342a2a33c","author":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"date":"2019-03-28 12:14:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"226871e9b93a8f04c23586984a452a34253e7ddc","author":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"date":"2019-03-28 12:16:46.000000000","message":"Topic changed from fluend-on-arm to pedro/fluentd-ubuntu-master","accounts_in_message":[],"_revision_number":4},{"id":"7b5f281c62e652c4fa11d32b2af20d222bd84e48","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-28 12:54:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 http://logs.rdoproject.org/92/638692/4/openstack-check/tripleo-build-containers-fedora-28/4179ad2/ : SUCCESS in 35m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dc5bc438cdc4992bf77a01f8c6dd153ded29243c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-28 13:37:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/92/638692/4/check/kolla-build-centos-binary/4cca456/ : FAILURE in 50m 45s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/92/638692/4/check/kolla-build-centos-source/5ac2c00/ : FAILURE in 46m 40s\n- kolla-ansible-centos-source http://logs.openstack.org/92/638692/4/check/kolla-ansible-centos-source/a862e65/ : SUCCESS in 1h 17m 03s\n- kolla-ansible-centos-binary http://logs.openstack.org/92/638692/4/check/kolla-ansible-centos-binary/d7fb004/ : SUCCESS in 1h 20m 46s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/92/638692/4/check/tripleo-build-containers-centos-7/df0e07b/ : SUCCESS in 51m 49s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/92/638692/4/check/kolla-build-oraclelinux-binary/a4fe22e/ : FAILURE in 33m 56s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/92/638692/4/check/kolla-build-oraclelinux-source/272f527/ : FAILURE in 1h 00m 35s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/92/638692/4/check/kolla-ansible-oraclelinux-source/3dca696/ : SUCCESS in 1h 15m 32s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/92/638692/4/check/kolla-ansible-oraclelinux-binary/168763e/ : SUCCESS in 1h 13m 39s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/92/638692/4/check/openstack-tox-lower-constraints/d93b8bd/ : SUCCESS in 3m 50s\n- openstack-tox-pep8 http://logs.openstack.org/92/638692/4/check/openstack-tox-pep8/d20bc9a/ : SUCCESS in 2m 59s\n- openstack-tox-py27 http://logs.openstack.org/92/638692/4/check/openstack-tox-py27/2855d5e/ : SUCCESS in 3m 10s\n- openstack-tox-py36 http://logs.openstack.org/92/638692/4/check/openstack-tox-py36/5d9b989/ : SUCCESS in 3m 24s\n- openstack-tox-py37 http://logs.openstack.org/92/638692/4/check/openstack-tox-py37/d8de88d/ : SUCCESS in 3m 40s\n- openstack-tox-docs http://logs.openstack.org/92/638692/4/check/openstack-tox-docs/dfe9232/html/ : SUCCESS in 3m 23s\n- kolla-build-ubuntu-binary http://logs.openstack.org/92/638692/4/check/kolla-build-ubuntu-binary/8d275c7/ : FAILURE in 32m 16s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/92/638692/4/check/kolla-build-ubuntu-source/c87c2c4/ : FAILURE in 29m 08s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/92/638692/4/check/kolla-ansible-ubuntu-source/71d38c7/ : SUCCESS in 59m 47s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/92/638692/4/check/kolla-ansible-ubuntu-binary/f721909/ : FAILURE in 26m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2fafce3181f4d1ab9e79335cc0eb172558f6e656","author":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"date":"2019-03-28 14:31:40.000000000","message":"Patch Set 4:\n\nrecheck:\n\n     INFO:kolla.common.utils.kafka:curl (http://apache.osuosl.org/kafka/1.0.2/kafka_2.12-1.0.2.tgz): response: 404, time: 0.058, size: 297","accounts_in_message":[],"_revision_number":4},{"id":"8441f4dba9fbb640ae676b5dbf01a00e3e1825c7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-28 15:16:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 http://logs.rdoproject.org/92/638692/4/openstack-check/tripleo-build-containers-fedora-28/6b5c901/ : SUCCESS in 30m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"576a0bcc0f15ed2706f4b9d982d8214e6be29e72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-28 15:52:25.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/92/638692/4/check/kolla-build-centos-binary/3471ae3/ : FAILURE in 53m 42s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/92/638692/4/check/kolla-build-centos-source/56259f8/ : FAILURE in 52m 26s\n- kolla-ansible-centos-source http://logs.openstack.org/92/638692/4/check/kolla-ansible-centos-source/361b54d/ : SUCCESS in 1h 04m 23s\n- kolla-ansible-centos-binary http://logs.openstack.org/92/638692/4/check/kolla-ansible-centos-binary/a6a5f67/ : SUCCESS in 1h 07m 38s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/92/638692/4/check/tripleo-build-containers-centos-7/ffe7c1f/ : SUCCESS in 46m 25s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/92/638692/4/check/kolla-build-oraclelinux-binary/0131756/ : FAILURE in 49m 16s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/92/638692/4/check/kolla-build-oraclelinux-source/60618fc/ : FAILURE in 56m 19s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/92/638692/4/check/kolla-ansible-oraclelinux-source/464e269/ : SUCCESS in 1h 02m 04s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/92/638692/4/check/kolla-ansible-oraclelinux-binary/5c73e66/ : SUCCESS in 1h 05m 17s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/92/638692/4/check/openstack-tox-lower-constraints/063ea06/ : SUCCESS in 4m 16s\n- openstack-tox-pep8 http://logs.openstack.org/92/638692/4/check/openstack-tox-pep8/ec5589d/ : SUCCESS in 3m 45s\n- openstack-tox-py27 http://logs.openstack.org/92/638692/4/check/openstack-tox-py27/0db5990/ : SUCCESS in 4m 08s\n- openstack-tox-py36 http://logs.openstack.org/92/638692/4/check/openstack-tox-py36/f0036da/ : SUCCESS in 4m 18s\n- openstack-tox-py37 http://logs.openstack.org/92/638692/4/check/openstack-tox-py37/d0d770a/ : SUCCESS in 3m 57s\n- openstack-tox-docs http://logs.openstack.org/92/638692/4/check/openstack-tox-docs/9be43a5/html/ : SUCCESS in 3m 41s\n- kolla-build-ubuntu-binary http://logs.openstack.org/92/638692/4/check/kolla-build-ubuntu-binary/0829bd4/ : FAILURE in 33m 48s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/92/638692/4/check/kolla-build-ubuntu-source/370a4e2/ : FAILURE in 31m 59s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/92/638692/4/check/kolla-ansible-ubuntu-source/f39ca55/ : SUCCESS in 1h 00m 35s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/92/638692/4/check/kolla-ansible-ubuntu-binary/fb5758c/ : FAILURE in 27m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b4cd315a3b0bcdbbdab67d4744d5d13b58e360e8","author":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"date":"2019-03-28 16:26:22.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"367c2fe2c8143bc70f3e9cc190c8f386dfcca070","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-28 17:04:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 http://logs.rdoproject.org/92/638692/5/openstack-check/tripleo-build-containers-fedora-28/5ce7d80/ : SUCCESS in 35m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7925caf00704952f594c591bb9a459150fbd62ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-28 17:41:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/92/638692/5/check/kolla-build-centos-binary/5797c08/ : SUCCESS in 41m 22s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/92/638692/5/check/kolla-build-centos-source/0d9f81e/ : SUCCESS in 51m 59s\n- kolla-ansible-centos-source http://logs.openstack.org/92/638692/5/check/kolla-ansible-centos-source/b63bbbb/ : SUCCESS in 1h 03m 19s\n- kolla-ansible-centos-binary http://logs.openstack.org/92/638692/5/check/kolla-ansible-centos-binary/b82760e/ : SUCCESS in 58m 52s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/92/638692/5/check/tripleo-build-containers-centos-7/2c1f3ff/ : SUCCESS in 42m 29s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/92/638692/5/check/kolla-build-oraclelinux-binary/04be71b/ : SUCCESS in 1h 01m 10s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/92/638692/5/check/kolla-build-oraclelinux-source/b177e11/ : SUCCESS in 50m 52s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/92/638692/5/check/kolla-ansible-oraclelinux-source/d47ebea/ : SUCCESS in 1h 06m 16s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/92/638692/5/check/kolla-ansible-oraclelinux-binary/e596c02/ : SUCCESS in 1h 08m 25s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/92/638692/5/check/openstack-tox-lower-constraints/6d1ac85/ : SUCCESS in 4m 11s\n- openstack-tox-pep8 http://logs.openstack.org/92/638692/5/check/openstack-tox-pep8/11f7c1c/ : SUCCESS in 3m 25s\n- openstack-tox-py27 http://logs.openstack.org/92/638692/5/check/openstack-tox-py27/31703c6/ : SUCCESS in 3m 44s\n- openstack-tox-py36 http://logs.openstack.org/92/638692/5/check/openstack-tox-py36/78e409d/ : SUCCESS in 5m 16s\n- openstack-tox-py37 http://logs.openstack.org/92/638692/5/check/openstack-tox-py37/2e0d2cb/ : SUCCESS in 3m 53s\n- openstack-tox-docs http://logs.openstack.org/92/638692/5/check/openstack-tox-docs/c156756/html/ : SUCCESS in 3m 50s\n- kolla-build-ubuntu-binary http://logs.openstack.org/92/638692/5/check/kolla-build-ubuntu-binary/9269faa/ : SUCCESS in 32m 08s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/92/638692/5/check/kolla-build-ubuntu-source/624eff0/ : SUCCESS in 37m 27s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/92/638692/5/check/kolla-ansible-ubuntu-source/9d50f4c/ : SUCCESS in 1h 01m 00s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/92/638692/5/check/kolla-ansible-ubuntu-binary/a06ebef/ : FAILURE in 33m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"542c410c9182521f9a2b8812d27445ae11b0d2d9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-07 18:40:47.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"d7c068b4658e466a49373cb774d8a451a3fb07bc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-28 15:11:54.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"27c991b33d2193a7792771726170ab909b3a8a39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 15:12:08.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"aaa7569ed5960f048d44a65659d7ac828363b9b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 16:29:27.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source http://logs.openstack.org/92/638692/5/gate/kolla-build-centos-source/4dfe1e0/ : SUCCESS in 55m 03s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/92/638692/5/gate/tripleo-build-containers-centos-7/4c77a9f/ : SUCCESS in 43m 28s\n- kolla-build-oraclelinux-source http://logs.openstack.org/92/638692/5/gate/kolla-build-oraclelinux-source/df85ce0/ : FAILURE in 3m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/638692/5/gate/openstack-tox-lower-constraints/633ba4c/ : SUCCESS in 4m 10s\n- openstack-tox-pep8 http://logs.openstack.org/92/638692/5/gate/openstack-tox-pep8/d181f3b/ : SUCCESS in 3m 49s\n- openstack-tox-py27 http://logs.openstack.org/92/638692/5/gate/openstack-tox-py27/f151670/ : SUCCESS in 2m 57s\n- openstack-tox-py36 http://logs.openstack.org/92/638692/5/gate/openstack-tox-py36/7236b65/ : SUCCESS in 3m 44s\n- openstack-tox-py37 http://logs.openstack.org/92/638692/5/gate/openstack-tox-py37/518f44b/ : SUCCESS in 4m 33s\n- openstack-tox-docs http://logs.openstack.org/92/638692/5/gate/openstack-tox-docs/47e9cfe/html/ : SUCCESS in 2m 24s\n- kolla-build-ubuntu-source http://logs.openstack.org/92/638692/5/gate/kolla-build-ubuntu-source/be0beaf/ : SUCCESS in 38m 46s","accounts_in_message":[],"_revision_number":5},{"id":"c786ba8c8d3c7b7fd4c3f14851a1f00e4f781c91","author":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"date":"2019-05-28 16:31:56.000000000","message":"Patch Set 5:\n\nrecheck as it looks like a temporary issue:\n\n    Could not fetch/save url https://download.docker.com/linux/centos/docker-ce.repo to file /etc/yum.repos.d/docker-ce.repo: [Errno 14] curl#7 - \"Failed connect to download.docker.com:443; Operation now in progress\"","accounts_in_message":[],"_revision_number":5},{"id":"6fcd2ac419851765e4d8eb44feda21493a1968b2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-28 17:08:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/92/638692/5/openstack-check/tripleo-build-containers-fedora-28-master/1061c7a/ : SUCCESS in 33m 52s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/92/638692/5/openstack-check/tripleo-build-containers-fedora-28-stein/a8c4b92/ : SUCCESS in 32m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d6a6badd7ffdc7e0c4b6a17c6bfe8edddb83fa2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 18:05:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/92/638692/5/check/kolla-build-centos-binary/d6ea040/ : FAILURE in 3m 36s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/92/638692/5/check/kolla-build-centos-source/271f26f/ : SUCCESS in 1h 03m 41s\n- kolla-ansible-centos-source http://logs.openstack.org/92/638692/5/check/kolla-ansible-centos-source/884db3a/ : SUCCESS in 1h 15m 32s\n- kolla-ansible-centos-binary http://logs.openstack.org/92/638692/5/check/kolla-ansible-centos-binary/99025d4/ : SUCCESS in 1h 02m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/92/638692/5/check/kolla-ansible-centos-source-upgrade/35bebac/ : SUCCESS in 1h 20m 14s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/92/638692/5/check/tripleo-build-containers-centos-7/ddc271f/ : SUCCESS in 42m 41s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/92/638692/5/check/kolla-build-oraclelinux-binary/f3828d1/ : SUCCESS in 58m 41s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/92/638692/5/check/kolla-build-oraclelinux-source/bd30529/ : SUCCESS in 38m 53s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/92/638692/5/check/kolla-ansible-oraclelinux-source/49a76c8/ : SUCCESS in 1h 11m 58s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/92/638692/5/check/kolla-ansible-oraclelinux-binary/cdec775/ : SUCCESS in 1h 05m 28s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/92/638692/5/check/openstack-tox-lower-constraints/b788607/ : SUCCESS in 5m 04s\n- openstack-tox-pep8 http://logs.openstack.org/92/638692/5/check/openstack-tox-pep8/ca85bda/ : SUCCESS in 3m 18s\n- openstack-tox-py27 http://logs.openstack.org/92/638692/5/check/openstack-tox-py27/0f8dfcd/ : SUCCESS in 2m 37s\n- openstack-tox-py36 http://logs.openstack.org/92/638692/5/check/openstack-tox-py36/da4ddf9/ : SUCCESS in 2m 47s\n- openstack-tox-py37 http://logs.openstack.org/92/638692/5/check/openstack-tox-py37/ce0f7ab/ : SUCCESS in 3m 19s\n- openstack-tox-docs http://logs.openstack.org/92/638692/5/check/openstack-tox-docs/9d4269c/html/ : SUCCESS in 2m 45s\n- kolla-build-ubuntu-binary http://logs.openstack.org/92/638692/5/check/kolla-build-ubuntu-binary/57a1399/ : SUCCESS in 27m 47s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/92/638692/5/check/kolla-build-ubuntu-source/58f3d16/ : SUCCESS in 34m 06s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/92/638692/5/check/kolla-ansible-ubuntu-source/fc6f90f/ : SUCCESS in 52m 00s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/92/638692/5/check/kolla-ansible-ubuntu-binary/3c172b6/ : SUCCESS in 45m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/92/638692/5/check/kolla-ansible-ubuntu-source-upgrade/682ad9f/ : SUCCESS in 1h 01m 39s","accounts_in_message":[],"_revision_number":5},{"id":"674fc80eb07950f8f79d1d96542948dd2bb1ebb9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 18:06:11.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"02d2eafb3149ed3ad34e768615ba4222b4883f0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 19:04:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"9ff42c138796e9967238cc404750a6d689038850","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 19:04:16.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/92/638692/5/gate/kolla-build-centos-source/b7fcf9e/ : SUCCESS in 49m 18s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/92/638692/5/gate/tripleo-build-containers-centos-7/760e460/ : SUCCESS in 50m 44s\n- kolla-build-oraclelinux-source http://logs.openstack.org/92/638692/5/gate/kolla-build-oraclelinux-source/0a9bab6/ : SUCCESS in 49m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/638692/5/gate/openstack-tox-lower-constraints/4caf6da/ : SUCCESS in 3m 49s\n- openstack-tox-pep8 http://logs.openstack.org/92/638692/5/gate/openstack-tox-pep8/05c7f3b/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/92/638692/5/gate/openstack-tox-py27/26ec7e1/ : SUCCESS in 2m 48s\n- openstack-tox-py36 http://logs.openstack.org/92/638692/5/gate/openstack-tox-py36/297168f/ : SUCCESS in 3m 26s\n- openstack-tox-py37 http://logs.openstack.org/92/638692/5/gate/openstack-tox-py37/ee1516f/ : SUCCESS in 3m 27s\n- openstack-tox-docs http://logs.openstack.org/92/638692/5/gate/openstack-tox-docs/8b98105/html/ : SUCCESS in 2m 56s\n- kolla-build-ubuntu-source http://logs.openstack.org/92/638692/5/gate/kolla-build-ubuntu-source/930eb69/ : SUCCESS in 26m 53s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"35394e25a33c6bbfdc2e1853adfa8f2a1a5614c1","revisions":{"e4426749ca84fd883a87fa9661abf65276e110c7":{"kind":"REWORK","_number":1,"created":"2019-02-22 15:50:10.000000000","uploader":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"ref":"refs/changes/92/638692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/638692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/638692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/638692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/638692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/638692/1"}}},"commit":{"parents":[{"commit":"725bb6156a18e7639365b66387f067e492cc692e","subject":"Merge \"debian/ubuntu: drop mysql compat stuff, unify on openssl 1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/725bb6156a18e7639365b66387f067e492cc692e"}]}],"author":{"name":"Pedro Alvarez","email":"pedro.alvarez@codethink.co.uk","date":"2019-02-21 16:46:21.000000000","tz":0},"committer":{"name":"Pedro Alvarez","email":"pedro.alvarez@codethink.co.uk","date":"2019-02-22 15:50:05.000000000","tz":0},"subject":"Install fluentd gem on debian/ubuntu for non x86_64","message":"Install fluentd gem on debian/ubuntu for non x86_64\n\nThe package used to install fluentd on debian/ubuntu is not\navailable for architectures other than x86_64. For those cases\ninstall fluentd as a gem until the package \u0027td-agent\u0027 is available\nfor more architectures.\n\nChange-Id: Ic1b25b37d06b429a112f1418a79d385d505bb1e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4426749ca84fd883a87fa9661abf65276e110c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4426749ca84fd883a87fa9661abf65276e110c7"}]},"branch":"refs/heads/master"},"f5541dfc13a5938dabe330e5a5da2ad61c2f78c3":{"kind":"REWORK","_number":2,"created":"2019-02-25 12:46:56.000000000","uploader":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"ref":"refs/changes/92/638692/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/638692/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/638692/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/638692/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/638692/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/638692/2"}}},"commit":{"parents":[{"commit":"725bb6156a18e7639365b66387f067e492cc692e","subject":"Merge \"debian/ubuntu: drop mysql compat stuff, unify on openssl 1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/725bb6156a18e7639365b66387f067e492cc692e"}]}],"author":{"name":"Pedro Alvarez","email":"pedro.alvarez@codethink.co.uk","date":"2019-02-21 16:46:21.000000000","tz":0},"committer":{"name":"Pedro Alvarez","email":"pedro.alvarez@codethink.co.uk","date":"2019-02-25 12:25:42.000000000","tz":0},"subject":"Install fluentd gem on debian/ubuntu for non x86_64","message":"Install fluentd gem on debian/ubuntu for non x86_64\n\nThe package used to install fluentd on debian/ubuntu is not\navailable for architectures other than x86_64. For those cases\ninstall fluentd as a gem until the package \u0027td-agent\u0027 is available\nfor more architectures.\n\nChange-Id: Ic1b25b37d06b429a112f1418a79d385d505bb1e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f5541dfc13a5938dabe330e5a5da2ad61c2f78c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f5541dfc13a5938dabe330e5a5da2ad61c2f78c3"}]},"branch":"refs/heads/master"},"d5fd09577850e5a2cf2fceeff230fef40291e207":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-03-12 10:24:44.000000000","uploader":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"ref":"refs/changes/92/638692/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/638692/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/638692/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/638692/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/638692/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/638692/3"}}},"commit":{"parents":[{"commit":"c2804ca14f47702bc4a26e624ee91028b9ae8662","subject":"Merge \"Introduce \u0027get_pip\u0027 macro\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c2804ca14f47702bc4a26e624ee91028b9ae8662"}]}],"author":{"name":"Pedro Alvarez","email":"pedro.alvarez@codethink.co.uk","date":"2019-02-21 16:46:21.000000000","tz":0},"committer":{"name":"Pedro Alvarez Piedehierro","email":"pedro.alvarez@codethink.co.uk","date":"2019-03-12 10:24:44.000000000","tz":0},"subject":"Install fluentd gem on debian/ubuntu for non x86_64","message":"Install fluentd gem on debian/ubuntu for non x86_64\n\nThe package used to install fluentd on debian/ubuntu is not\navailable for architectures other than x86_64. For those cases\ninstall fluentd as a gem until the package \u0027td-agent\u0027 is available\nfor more architectures.\n\nChange-Id: Ic1b25b37d06b429a112f1418a79d385d505bb1e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5fd09577850e5a2cf2fceeff230fef40291e207"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5fd09577850e5a2cf2fceeff230fef40291e207"}]},"branch":"refs/heads/master"},"a40930a8e6622b8c4531d021903f2fd3fa255153":{"kind":"REWORK","_number":4,"created":"2019-03-28 12:14:44.000000000","uploader":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"ref":"refs/changes/92/638692/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/638692/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/638692/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/638692/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/638692/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/638692/4"}}},"commit":{"parents":[{"commit":"09f37541d8e57d85aa568a0ad6a23c44ef11a36e","subject":"Merge \"Add boto3 python dependencies into gnocchi-base image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09f37541d8e57d85aa568a0ad6a23c44ef11a36e"}]}],"author":{"name":"Pedro Alvarez","email":"pedro.alvarez@codethink.co.uk","date":"2019-02-21 16:46:21.000000000","tz":0},"committer":{"name":"Pedro Alvarez","email":"pedro.alvarez@codethink.co.uk","date":"2019-03-28 12:13:12.000000000","tz":60},"subject":"Install fluentd gem on debian/ubuntu for non x86_64","message":"Install fluentd gem on debian/ubuntu for non x86_64\n\nThe package used to install fluentd on debian/ubuntu is not\navailable for architectures other than x86_64. For those cases\ninstall fluentd as a gem until the package \u0027td-agent\u0027 is available\nfor more architectures.\n\nChange-Id: Ic1b25b37d06b429a112f1418a79d385d505bb1e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a40930a8e6622b8c4531d021903f2fd3fa255153"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a40930a8e6622b8c4531d021903f2fd3fa255153"}]},"branch":"refs/heads/master"},"35394e25a33c6bbfdc2e1853adfa8f2a1a5614c1":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-03-28 16:26:22.000000000","uploader":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"ref":"refs/changes/92/638692/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/638692/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/638692/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/638692/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/638692/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/638692/5"}}},"commit":{"parents":[{"commit":"83e8be4f0597c4572f58a1052010d3660064cdb9","subject":"Set requirements branch in version-check.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83e8be4f0597c4572f58a1052010d3660064cdb9"}]}],"author":{"name":"Pedro Alvarez","email":"pedro.alvarez@codethink.co.uk","date":"2019-02-21 16:46:21.000000000","tz":0},"committer":{"name":"Pedro Alvarez Piedehierro","email":"pedro.alvarez@codethink.co.uk","date":"2019-03-28 16:26:22.000000000","tz":0},"subject":"Install fluentd gem on debian/ubuntu for non x86_64","message":"Install fluentd gem on debian/ubuntu for non x86_64\n\nThe package used to install fluentd on debian/ubuntu is not\navailable for architectures other than x86_64. For those cases\ninstall fluentd as a gem until the package \u0027td-agent\u0027 is available\nfor more architectures.\n\nChange-Id: Ic1b25b37d06b429a112f1418a79d385d505bb1e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35394e25a33c6bbfdc2e1853adfa8f2a1a5614c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35394e25a33c6bbfdc2e1853adfa8f2a1a5614c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
