)]}'
{"id":"openstack%2Fkolla~640099","triplet_id":"openstack%2Fkolla~master~Id65d7832298e537c82df214db8fd0d1ae4f81d20","project":"openstack/kolla","branch":"master","topic":"640099","hashtags":[],"change_id":"Id65d7832298e537c82df214db8fd0d1ae4f81d20","subject":"Fix image squashing","status":"ABANDONED","created":"2019-02-28 17:01:52.000000000","updated":"2019-06-24 14:53:37.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"763d51f717808ff86d8980a140f0f5aa5fa82955","_number":640099,"virtual_id_number":640099,"owner":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":-1,"date":"2019-03-12 12:09:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-03-12 12:25:50.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22582,"name":"Duong Ha-Quang","email":"quangduongha@gmail.com","username":"duonghq"},{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"disliked":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":1,"date":"2019-03-12 10:19:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"date":"2019-03-12 10:18:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},{"value":-1,"date":"2019-03-12 17:40:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22582,"name":"Duong Ha-Quang","email":"quangduongha@gmail.com","username":"duonghq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"all":[{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":-1,"date":"2019-03-13 09:56:24.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22582,"name":"Duong Ha-Quang","email":"quangduongha@gmail.com","username":"duonghq"},{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22582,"name":"Duong Ha-Quang","email":"quangduongha@gmail.com","username":"duonghq"},{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22582,"name":"Duong Ha-Quang","email":"quangduongha@gmail.com","username":"duonghq"},{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22582,"name":"Duong Ha-Quang","email":"quangduongha@gmail.com","username":"duonghq"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-03 09:06:47.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2019-03-03 09:06:47.000000000","updated_by":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"reviewer":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"state":"REVIEWER"},{"updated":"2019-03-03 09:06:47.000000000","updated_by":{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},"reviewer":{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},"state":"REVIEWER"},{"updated":"2019-03-03 09:06:47.000000000","updated_by":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"reviewer":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"state":"REVIEWER"},{"updated":"2019-03-03 09:06:47.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-03-03 09:06:47.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-03-03 09:06:47.000000000","updated_by":{"_account_id":22582,"name":"Duong Ha-Quang","email":"quangduongha@gmail.com","username":"duonghq"},"reviewer":{"_account_id":22582,"name":"Duong Ha-Quang","email":"quangduongha@gmail.com","username":"duonghq"},"state":"REVIEWER"},{"updated":"2019-03-03 09:06:47.000000000","updated_by":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"reviewer":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"state":"REVIEWER"},{"updated":"2019-03-10 20:40:03.000000000","updated_by":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"reviewer":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"state":"REVIEWER"},{"updated":"2019-03-12 10:18:24.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-03-12 10:19:30.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-03-12 12:09:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-12 12:25:50.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-12 17:40:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"2528a45215f43067b1a5f9256ca772a14c2aa575","author":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"date":"2019-02-28 17:01:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c58530d41545594cb0564819cf88862465f9cd8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-02-28 17:53:38.000000000","message":"Patch Set 1: Code-Review-1\n\nI think the squash option is intended for cases where you rarely have shared history between layers, e.g. when images are built individually in a CI/CD pipeline. If they have shared history there is less benefit to squashing.","accounts_in_message":[],"_revision_number":1},{"id":"be0c7318dab143d0a264bd99a2d91d473e77d054","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-02-28 17:55:31.000000000","message":"Patch Set 1:\n\n\u003e I think the squash option is intended for cases where you rarely\n \u003e have shared history between layers, e.g. when images are built\n \u003e individually in a CI/CD pipeline. If they have shared history there\n \u003e is less benefit to squashing.\n\nActually, looks like this might also produce squashed \u0027base\u0027 images (base, openstack-base, nova-base) etc, with a single layer per Dockerfile. Have I got that right? If so it should be useful. We still might want to allow both types of squashing - a full squash might be more efficient if there is no shared history between images.","accounts_in_message":[],"_revision_number":1},{"id":"1e579c0ab3540e099cfa8280e378224456a02977","author":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"date":"2019-02-28 18:10:15.000000000","message":"Patch Set 1:\n\n\u003e this might also produce squashed \u0027base\u0027 images (base, openstack-base, nova-base) etc, with a single layer per Dockerfile.\n\nExactly, that was one of the goals. Unfortunately, top-level layer misses link to parent after squashing, so `docker history` gives you the following\n\n$ docker history kolla/ubuntu-binary-keystone-ssh:5.0.5\nIMAGE               CREATED             CREATED BY                                      SIZE                COMMENT\naed50e78b9a4        45 minutes ago                                                      7.71MB              \n\u003cmissing\u003e           About an hour ago                                                   78.2MB              \n\u003cmissing\u003e           About an hour ago                                                   137MB               \n\u003cmissing\u003e           About an hour ago                                                   130MB               \n\u003cmissing\u003e           5 weeks ago         /bin/sh -c #(nop)  CMD [\"/bin/bash\"]            0B                  \n\u003cmissing\u003e           5 weeks ago         /bin/sh -c mkdir -p /run/systemd \u0026\u0026 echo \u0027do…   7B                  \n\u003cmissing\u003e           5 weeks ago         /bin/sh -c rm -rf /var/lib/apt/lists/*          0B                  \n\u003cmissing\u003e           5 weeks ago         /bin/sh -c set -xe   \u0026\u0026 echo \u0027#!/bin/sh\u0027 \u003e /…   745B                \n\u003cmissing\u003e           5 weeks ago         /bin/sh -c #(nop) ADD file:916a45030dee881bb…   117MB \n\n5 layers at the bottom are from ubuntu, then goes base, openstack-base, keystone-base, keystone-ssh","accounts_in_message":[],"_revision_number":1},{"id":"58cc79677859d313c21dc38d3e43821b516502c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-28 18:56:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/99/640099/1/check/kolla-build-centos-binary/82a5dff/ : SUCCESS in 54m 07s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/99/640099/1/check/kolla-build-centos-source/8dcad97/ : FAILURE in 47m 25s\n- kolla-ansible-centos-source http://logs.openstack.org/99/640099/1/check/kolla-ansible-centos-source/90a5b6d/ : FAILURE in 52m 28s\n- kolla-ansible-centos-binary http://logs.openstack.org/99/640099/1/check/kolla-ansible-centos-binary/58e5598/ : SUCCESS in 1h 08m 13s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/99/640099/1/check/tripleo-build-containers-centos-7/39545c5/ : SUCCESS in 35m 56s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/99/640099/1/check/kolla-build-oraclelinux-binary/240b63d/ : SUCCESS in 41m 26s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/99/640099/1/check/kolla-build-oraclelinux-source/a72c3ba/ : FAILURE in 49m 37s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/99/640099/1/check/kolla-ansible-oraclelinux-source/b55cf6c/ : FAILURE in 1h 05m 21s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/99/640099/1/check/kolla-ansible-oraclelinux-binary/56318b9/ : SUCCESS in 1h 12m 53s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/99/640099/1/check/openstack-tox-lower-constraints/73b95b5/ : SUCCESS in 4m 33s\n- openstack-tox-py36 http://logs.openstack.org/99/640099/1/check/openstack-tox-py36/5e9329a/ : SUCCESS in 3m 46s\n- openstack-tox-pep8 http://logs.openstack.org/99/640099/1/check/openstack-tox-pep8/6a4fa6e/ : SUCCESS in 3m 36s\n- openstack-tox-py27 http://logs.openstack.org/99/640099/1/check/openstack-tox-py27/7b287c9/ : SUCCESS in 4m 30s\n- openstack-tox-py35 http://logs.openstack.org/99/640099/1/check/openstack-tox-py35/9e704c2/ : SUCCESS in 4m 54s\n- openstack-tox-docs http://logs.openstack.org/99/640099/1/check/openstack-tox-docs/28ea836/html/ : SUCCESS in 4m 11s\n- kolla-build-ubuntu-binary http://logs.openstack.org/99/640099/1/check/kolla-build-ubuntu-binary/2cf19d4/ : SUCCESS in 1h 25m 15s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/99/640099/1/check/kolla-build-ubuntu-source/965a7ab/ : FAILURE in 41m 21s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/99/640099/1/check/kolla-ansible-ubuntu-source/b33c41d/ : FAILURE in 43m 24s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/99/640099/1/check/kolla-ansible-ubuntu-binary/fd7aa03/ : SUCCESS in 1h 50m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb05b3d87629067189372797e587f4768b2813eb","author":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"date":"2019-03-01 11:12:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9afccf49e6f053465b83721c78307f5d29e9466e","author":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"date":"2019-03-01 11:15:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f4e3ddf0843fd1603de9fe39d70ff6f2cd74c9c3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-01 11:27:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 http://logs.rdoproject.org/99/640099/3/openstack-check/tripleo-build-containers-fedora-28/b7bd814/ : FAILURE in 7m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b622a018e1aec8c87556d49918aefff1a3cf9f32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-01 13:22:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/99/640099/3/check/kolla-build-centos-binary/c48f0f2/ : SUCCESS in 31m 33s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/99/640099/3/check/kolla-build-centos-source/b30961f/ : FAILURE in 46m 24s\n- kolla-ansible-centos-source http://logs.openstack.org/99/640099/3/check/kolla-ansible-centos-source/25e1a38/ : SUCCESS in 1h 12m 52s\n- kolla-ansible-centos-binary http://logs.openstack.org/99/640099/3/check/kolla-ansible-centos-binary/cca29e8/ : SUCCESS in 1h 14m 28s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/99/640099/3/check/tripleo-build-containers-centos-7/64bc6c1/ : SUCCESS in 46m 22s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/99/640099/3/check/kolla-build-oraclelinux-binary/56b88f4/ : SUCCESS in 45m 21s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/99/640099/3/check/kolla-build-oraclelinux-source/8342d60/ : FAILURE in 52m 02s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/99/640099/3/check/kolla-ansible-oraclelinux-source/bcfaeff/ : SUCCESS in 1h 31m 04s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/99/640099/3/check/kolla-ansible-oraclelinux-binary/bcecba9/ : SUCCESS in 1h 21m 06s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/99/640099/3/check/openstack-tox-lower-constraints/75d26ec/ : SUCCESS in 4m 11s\n- openstack-tox-py36 http://logs.openstack.org/99/640099/3/check/openstack-tox-py36/090034e/ : SUCCESS in 5m 13s\n- openstack-tox-pep8 http://logs.openstack.org/99/640099/3/check/openstack-tox-pep8/d81536c/ : SUCCESS in 3m 52s\n- openstack-tox-py27 http://logs.openstack.org/99/640099/3/check/openstack-tox-py27/91a6fdc/ : SUCCESS in 4m 11s\n- openstack-tox-py35 http://logs.openstack.org/99/640099/3/check/openstack-tox-py35/ad2d71d/ : SUCCESS in 6m 37s\n- openstack-tox-docs http://logs.openstack.org/99/640099/3/check/openstack-tox-docs/89e0be5/html/ : SUCCESS in 4m 13s\n- kolla-build-ubuntu-binary http://logs.openstack.org/99/640099/3/check/kolla-build-ubuntu-binary/3a5290f/ : FAILURE in 1h 00m 26s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/99/640099/3/check/kolla-build-ubuntu-source/81f4cc4/ : FAILURE in 36m 12s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/99/640099/3/check/kolla-ansible-ubuntu-source/6dbeae6/ : SUCCESS in 59m 04s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/99/640099/3/check/kolla-ansible-ubuntu-binary/27c32b5/ : SUCCESS in 1h 58m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8b57bac4c1da6ee0d55488e6220b3b3321ed19f0","author":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"date":"2019-03-02 18:43:41.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"eca20d6b4499d9ddfcad9b324dc113399bcf70d6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-02 19:16:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 http://logs.rdoproject.org/99/640099/4/openstack-check/tripleo-build-containers-fedora-28/18f0509/ : SUCCESS in 27m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"16666c0ad4695b6774dd02368ecb045662fd4fc3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-02 20:37:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/99/640099/4/check/kolla-build-centos-binary/3fd4026/ : SUCCESS in 39m 03s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/99/640099/4/check/kolla-build-centos-source/a656866/ : SUCCESS in 31m 16s\n- kolla-ansible-centos-source http://logs.openstack.org/99/640099/4/check/kolla-ansible-centos-source/19b2f97/ : SUCCESS in 1h 11m 15s\n- kolla-ansible-centos-binary http://logs.openstack.org/99/640099/4/check/kolla-ansible-centos-binary/8e33244/ : SUCCESS in 1h 36m 44s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/99/640099/4/check/tripleo-build-containers-centos-7/a9e3e7b/ : SUCCESS in 44m 28s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/99/640099/4/check/kolla-build-oraclelinux-binary/538bb03/ : SUCCESS in 53m 43s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/99/640099/4/check/kolla-build-oraclelinux-source/5b9dd7f/ : SUCCESS in 41m 16s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/99/640099/4/check/kolla-ansible-oraclelinux-source/ef4edda/ : SUCCESS in 1h 05m 28s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/99/640099/4/check/kolla-ansible-oraclelinux-binary/8b1b949/ : SUCCESS in 1h 23m 38s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/99/640099/4/check/openstack-tox-lower-constraints/3da4f37/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/99/640099/4/check/openstack-tox-py36/4f9c161/ : SUCCESS in 2m 38s\n- openstack-tox-pep8 http://logs.openstack.org/99/640099/4/check/openstack-tox-pep8/1958458/ : SUCCESS in 2m 44s\n- openstack-tox-py27 http://logs.openstack.org/99/640099/4/check/openstack-tox-py27/18e91a1/ : SUCCESS in 3m 19s\n- openstack-tox-py35 http://logs.openstack.org/99/640099/4/check/openstack-tox-py35/d7f5de4/ : SUCCESS in 3m 03s\n- openstack-tox-docs http://logs.openstack.org/99/640099/4/check/openstack-tox-docs/7c74e9f/html/ : SUCCESS in 2m 26s\n- kolla-build-ubuntu-binary http://logs.openstack.org/99/640099/4/check/kolla-build-ubuntu-binary/66489e8/ : SUCCESS in 1h 17m 23s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/99/640099/4/check/kolla-build-ubuntu-source/9a176d4/ : SUCCESS in 36m 51s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/99/640099/4/check/kolla-ansible-ubuntu-source/15c7ade/ : SUCCESS in 1h 06m 44s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/99/640099/4/check/kolla-ansible-ubuntu-binary/1241e4b/ : SUCCESS in 1h 49m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bf0fa5f3ea6cc4ab55cba2a031d09a0971d36842","author":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"date":"2019-03-05 08:56:09.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"bc265104a8f6697697a070d5e2ca828775d2533e","author":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"date":"2019-03-10 10:22:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e1977e91d56ababcc77fe2cbce232b392b9fc6bf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-10 10:58:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 http://logs.rdoproject.org/99/640099/5/openstack-check/tripleo-build-containers-fedora-28/cd4bc23/ : SUCCESS in 30m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cd8042c04d9c384a2a0ee6049075450ba41e8c53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-10 11:57:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/99/640099/5/check/kolla-build-centos-binary/6c78c85/ : FAILURE in 16m 09s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/99/640099/5/check/kolla-build-centos-source/8ed2f3f/ : SUCCESS in 57m 03s\n- kolla-ansible-centos-source http://logs.openstack.org/99/640099/5/check/kolla-ansible-centos-source/569f283/ : SUCCESS in 1h 12m 13s\n- kolla-ansible-centos-binary http://logs.openstack.org/99/640099/5/check/kolla-ansible-centos-binary/c061230/ : SUCCESS in 1h 11m 57s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/99/640099/5/check/tripleo-build-containers-centos-7/5d53531/ : SUCCESS in 40m 18s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/99/640099/5/check/kolla-build-oraclelinux-binary/720e386/ : SUCCESS in 52m 59s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/99/640099/5/check/kolla-build-oraclelinux-source/c8d5af1/ : FAILURE in 15m 05s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/99/640099/5/check/kolla-ansible-oraclelinux-source/f88054a/ : SUCCESS in 1h 29m 05s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/99/640099/5/check/kolla-ansible-oraclelinux-binary/b8b63e9/ : SUCCESS in 1h 17m 52s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/99/640099/5/check/openstack-tox-lower-constraints/daa87dd/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/99/640099/5/check/openstack-tox-py36/629637b/ : SUCCESS in 3m 34s\n- openstack-tox-py37 http://logs.openstack.org/99/640099/5/check/openstack-tox-py37/8db8046/ : SUCCESS in 2m 59s\n- openstack-tox-pep8 http://logs.openstack.org/99/640099/5/check/openstack-tox-pep8/c576dc2/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/99/640099/5/check/openstack-tox-py27/2d500f6/ : SUCCESS in 2m 38s\n- openstack-tox-docs http://logs.openstack.org/99/640099/5/check/openstack-tox-docs/6106c7c/html/ : SUCCESS in 2m 46s\n- kolla-build-ubuntu-binary http://logs.openstack.org/99/640099/5/check/kolla-build-ubuntu-binary/0f2d318/ : SUCCESS in 29m 28s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/99/640099/5/check/kolla-build-ubuntu-source/26c114f/ : SUCCESS in 33m 36s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/99/640099/5/check/kolla-ansible-ubuntu-source/650e874/ : SUCCESS in 1h 00m 16s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/99/640099/5/check/kolla-ansible-ubuntu-binary/04b5ad3/ : FAILURE in 30m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6ca70931c06c625e9d4ba2829196e6d6b0d91635","author":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"date":"2019-03-10 15:23:38.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"39653331c2fc048500781fb2eaf1faa611f06b02","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-10 16:30:00.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 http://logs.rdoproject.org/99/640099/6/openstack-check/tripleo-build-containers-fedora-28/099ce91/ : SUCCESS in 30m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"be95d33b1ee827e3bdef4fd7a5c306820e295546","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-10 16:48:36.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/99/640099/6/check/kolla-build-centos-binary/af0ba7b/ : SUCCESS in 44m 34s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/99/640099/6/check/kolla-build-centos-source/7d2a81b/ : SUCCESS in 44m 55s\n- kolla-ansible-centos-source http://logs.openstack.org/99/640099/6/check/kolla-ansible-centos-source/ed4de57/ : SUCCESS in 1h 13m 13s\n- kolla-ansible-centos-binary http://logs.openstack.org/99/640099/6/check/kolla-ansible-centos-binary/0e953fa/ : SUCCESS in 1h 18m 40s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/99/640099/6/check/tripleo-build-containers-centos-7/5c228f8/ : SUCCESS in 52m 00s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/99/640099/6/check/kolla-build-oraclelinux-binary/da82877/ : SUCCESS in 45m 00s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/99/640099/6/check/kolla-build-oraclelinux-source/241dbc8/ : SUCCESS in 46m 09s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/99/640099/6/check/kolla-ansible-oraclelinux-source/19c79b4/ : SUCCESS in 1h 19m 26s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/99/640099/6/check/kolla-ansible-oraclelinux-binary/f3f6704/ : SUCCESS in 1h 14m 25s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/99/640099/6/check/openstack-tox-lower-constraints/67b09d2/ : SUCCESS in 2m 57s\n- openstack-tox-py36 http://logs.openstack.org/99/640099/6/check/openstack-tox-py36/273bdc5/ : SUCCESS in 2m 54s\n- openstack-tox-py37 http://logs.openstack.org/99/640099/6/check/openstack-tox-py37/32d08da/ : SUCCESS in 3m 14s\n- openstack-tox-pep8 http://logs.openstack.org/99/640099/6/check/openstack-tox-pep8/17b39de/ : SUCCESS in 3m 13s\n- openstack-tox-py27 http://logs.openstack.org/99/640099/6/check/openstack-tox-py27/37cefc4/ : SUCCESS in 2m 46s\n- openstack-tox-docs http://logs.openstack.org/99/640099/6/check/openstack-tox-docs/03d7d27/html/ : SUCCESS in 2m 34s\n- kolla-build-ubuntu-binary http://logs.openstack.org/99/640099/6/check/kolla-build-ubuntu-binary/4c0bd66/ : SUCCESS in 28m 01s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/99/640099/6/check/kolla-build-ubuntu-source/fd6e620/ : SUCCESS in 34m 30s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/99/640099/6/check/kolla-ansible-ubuntu-source/b5f973b/ : SUCCESS in 1h 04m 27s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/99/640099/6/check/kolla-ansible-ubuntu-binary/aaf382b/ : FAILURE in 29m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f8a22a0b27d3a4027b832dc1548beca5fd0344e7","author":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"date":"2019-03-10 20:40:03.000000000","message":"Patch Set 6: Code-Review-1\n\nI thought existing squashing is producing layers per each \u0027parent\u0027. So for example there would be:\n- os layer\n- base layer\n- openstack-base layer\n- component-base layer\n- final component layer\n\nWhich is exactly as optimised as possible. I need to check though.","accounts_in_message":[],"_revision_number":6},{"id":"21f1f8c58e53458a98e6d6da0e688fd0a5060885","author":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"date":"2019-03-11 20:40:51.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1f3cecdcbc829de33a9216feede15c950091640d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-11 23:01:00.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/99/640099/7/check/kolla-build-centos-binary/67c0acc/ : SUCCESS in 43m 39s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/99/640099/7/check/kolla-build-centos-source/eef6ca7/ : SUCCESS in 1h 00m 24s\n- kolla-ansible-centos-source http://logs.openstack.org/99/640099/7/check/kolla-ansible-centos-source/1cb5d8a/ : SUCCESS in 1h 31m 33s\n- kolla-ansible-centos-binary http://logs.openstack.org/99/640099/7/check/kolla-ansible-centos-binary/66e70cb/ : SUCCESS in 1h 12m 49s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/99/640099/7/check/tripleo-build-containers-centos-7/c353f90/ : SUCCESS in 41m 53s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/99/640099/7/check/kolla-build-oraclelinux-binary/ea9de64/ : SUCCESS in 53m 47s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/99/640099/7/check/kolla-build-oraclelinux-source/dfc2fc5/ : SUCCESS in 54m 08s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/99/640099/7/check/kolla-ansible-oraclelinux-source/35d087a/ : SUCCESS in 1h 13m 57s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/99/640099/7/check/kolla-ansible-oraclelinux-binary/aae10a5/ : SUCCESS in 1h 12m 26s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/99/640099/7/check/openstack-tox-lower-constraints/0b57976/ : SUCCESS in 3m 25s\n- openstack-tox-py36 http://logs.openstack.org/99/640099/7/check/openstack-tox-py36/f8850f7/ : SUCCESS in 3m 22s\n- openstack-tox-py37 http://logs.openstack.org/99/640099/7/check/openstack-tox-py37/a14af8e/ : SUCCESS in 3m 09s\n- openstack-tox-pep8 http://logs.openstack.org/99/640099/7/check/openstack-tox-pep8/bcc802b/ : FAILURE in 3m 26s\n- openstack-tox-py27 http://logs.openstack.org/99/640099/7/check/openstack-tox-py27/cfdbe80/ : SUCCESS in 4m 09s\n- openstack-tox-docs http://logs.openstack.org/99/640099/7/check/openstack-tox-docs/754deb5/html/ : SUCCESS in 3m 27s\n- kolla-build-ubuntu-binary http://logs.openstack.org/99/640099/7/check/kolla-build-ubuntu-binary/1d8dd95/ : SUCCESS in 29m 07s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/99/640099/7/check/kolla-build-ubuntu-source/14b4c8a/ : SUCCESS in 32m 29s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/99/640099/7/check/kolla-ansible-ubuntu-source/943084b/ : SUCCESS in 1h 18m 29s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/99/640099/7/check/kolla-ansible-ubuntu-binary/fa7c4d0/ : FAILURE in 32m 04s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"33a7f7a32a6e87c80d50ee92e2148840f021efce","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-12 01:07:06.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 tripleo-build-containers-fedora-28 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"41f88bf9a61396a9e04cfa156a5fc6cfca6b93ac","author":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"date":"2019-03-12 04:59:32.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"0191067da6d01c0067c78609f4fd3589d45eccb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-12 06:36:36.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/99/640099/8/check/kolla-build-centos-binary/71d9dbd/ : SUCCESS in 46m 57s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/99/640099/8/check/kolla-build-centos-source/b360866/ : SUCCESS in 53m 27s\n- kolla-ansible-centos-source http://logs.openstack.org/99/640099/8/check/kolla-ansible-centos-source/f0c5748/ : SUCCESS in 1h 17m 29s\n- kolla-ansible-centos-binary http://logs.openstack.org/99/640099/8/check/kolla-ansible-centos-binary/c4f6a91/ : SUCCESS in 1h 23m 32s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/99/640099/8/check/tripleo-build-containers-centos-7/5defe87/ : SUCCESS in 51m 56s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/99/640099/8/check/kolla-build-oraclelinux-binary/52f44ac/ : SUCCESS in 44m 36s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/99/640099/8/check/kolla-build-oraclelinux-source/b9472c7/ : SUCCESS in 56m 02s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/99/640099/8/check/kolla-ansible-oraclelinux-source/a8dd566/ : SUCCESS in 1h 31m 57s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/99/640099/8/check/kolla-ansible-oraclelinux-binary/179ffd0/ : SUCCESS in 1h 14m 14s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/99/640099/8/check/openstack-tox-lower-constraints/97d47e9/ : SUCCESS in 3m 05s\n- openstack-tox-py36 http://logs.openstack.org/99/640099/8/check/openstack-tox-py36/8f75e21/ : SUCCESS in 3m 37s\n- openstack-tox-py37 http://logs.openstack.org/99/640099/8/check/openstack-tox-py37/cdf78cd/ : SUCCESS in 2m 46s\n- openstack-tox-pep8 http://logs.openstack.org/99/640099/8/check/openstack-tox-pep8/be00a5e/ : SUCCESS in 2m 53s\n- openstack-tox-py27 http://logs.openstack.org/99/640099/8/check/openstack-tox-py27/0ddc674/ : SUCCESS in 3m 01s\n- openstack-tox-docs http://logs.openstack.org/99/640099/8/check/openstack-tox-docs/2283a0e/html/ : SUCCESS in 2m 28s\n- kolla-build-ubuntu-binary http://logs.openstack.org/99/640099/8/check/kolla-build-ubuntu-binary/b210138/ : SUCCESS in 29m 04s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/99/640099/8/check/kolla-build-ubuntu-source/02667ef/ : SUCCESS in 38m 13s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/99/640099/8/check/kolla-ansible-ubuntu-source/e3606f3/ : SUCCESS in 1h 01m 13s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/99/640099/8/check/kolla-ansible-ubuntu-binary/8ba6982/ : FAILURE in 27m 55s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"66e99bab39251d96e6f818ab7b077e4febc77efb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-03-12 09:22:08.000000000","message":"Patch Set 8:\n\nINFO:kolla.common.utils.base:Removing intermediate container 612b863a99f3\nINFO:kolla.common.utils.base: ---\u003e 9efa91073649\nINFO:kolla.common.utils.base:Step 24/27 : RUN curl -sSL http://deb.debian.org/debian/pool/main/d/dumb-init/dumb-init_1.2.2-1.1_arm64.deb -o dumb-init_1.2.2-1.1_arm64.deb     \u0026\u0026 ar -x dumb-init_1.2.2-1.1_arm64.deb data.tar.xz     \u0026\u0026 tar xf data.tar.xz ./usr/bin/dumb-init     \u0026\u0026 install -d -m 0755 /usr/local/bin     \u0026\u0026 mv usr/bin/dumb-init /usr/local/bin/dumb-init     \u0026\u0026 rm data.tar.xz dumb-init_1.2.2-1.1_arm64.deb\nINFO:kolla.common.utils.base: ---\u003e Running in 103b3bd5fff8\nINFO:kolla.common.utils.base:curl (http://cdn-fastly.deb.debian.org/debian/pool/main/d/dumb-init/dumb-init_1.2.2-1.1_arm64.deb): response: 200, time: 0.223058, size: 13472\nINFO:kolla.common.utils.base:Removing intermediate container 103b3bd5fff8\nINFO:kolla.common.utils.base: ---\u003e 2f60a563fdb3\nINFO:kolla.common.utils.base:Step 25/27 : ENTRYPOINT [\"dumb-init\", \"--single-child\", \"--\"]\nINFO:kolla.common.utils.base: ---\u003e Running in 6d63044946a9\nINFO:kolla.common.utils.base:Removing intermediate container 6d63044946a9\nINFO:kolla.common.utils.base: ---\u003e e4e4667af6e7\nINFO:kolla.common.utils.base:Step 26/27 : RUN touch /usr/local/bin/kolla_extend_start     \u0026\u0026 chmod 755 /usr/local/bin/kolla_start /usr/local/bin/kolla_extend_start /usr/local/bin/kolla_set_configs     \u0026\u0026 chmod 440 /etc/sudoers     \u0026\u0026 mkdir -p /var/log/kolla     \u0026\u0026 chown :kolla /var/log/kolla     \u0026\u0026 chmod 2775 /var/log/kolla     \u0026\u0026 rm -f /tmp/kolla_bashrc\nINFO:kolla.common.utils.base: ---\u003e Running in b527b1cc89cc\nINFO:kolla.common.utils.base:Removing intermediate container b527b1cc89cc\nINFO:kolla.common.utils.base: ---\u003e 7cb6b031e901\nINFO:kolla.common.utils.base:Step 27/27 : CMD [\"kolla_start\"]\nINFO:kolla.common.utils.base: ---\u003e Running in a9ec52d10ca7\nINFO:kolla.common.utils.base:Removing intermediate container a9ec52d10ca7\nINFO:kolla.common.utils.base: ---\u003e e8c39b18c737\nINFO:kolla.common.utils.base:Successfully built e8c39b18c737\nINFO:kolla.common.utils.base:Successfully tagged squash/debian-source-base:8.0.0\nERROR:kolla.common.utils.base:Unknown error when building\nTraceback (most recent call last):\n  File \"/home/hrw/devel/kolla/kolla/kolla/image/build.py\", line 598, in builder\n    self.squash()\n  File \"/home/hrw/devel/kolla/kolla/kolla/image/build.py\", line 620, in squash\n    squash_from \u003d tagged[1]\nTypeError: \u0027filter\u0027 object is not subscriptable\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nINFO:kolla.common.utils:Images that failed to build\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nERROR:kolla.common.utils:base Failed with status: error","accounts_in_message":[],"_revision_number":8},{"id":"d70e11c12f95f696a9b287fe07bfaffa0b5599aa","author":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"date":"2019-03-12 09:46:13.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d6cfd70cd88ec1060a74eb6b773b2ab5684943cb","author":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"date":"2019-03-12 10:10:12.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"fae5fb02a3a6cb2713b88fd64ad00f5c6a5fd0aa","author":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"date":"2019-03-12 10:13:51.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"0b9be0277a575c7cf1dbefb3d2d5c662110c5c2b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-03-12 10:18:24.000000000","message":"Patch Set 11: Code-Review+2\n\nWith all those changes I can give +2.\n\ndebian-source-nova-compute went from 42 layers to just 6.","accounts_in_message":[],"_revision_number":11},{"id":"96804234993b8b99b7a6b7304d28bfe7e61f7722","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-03-12 10:19:30.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"44752a3bdddf4c82ae56cc30b2c98246ce0bc90a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-12 12:09:51.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/99/640099/11/check/kolla-build-centos-binary/3bfeda3/ : SUCCESS in 56m 23s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/99/640099/11/check/kolla-build-centos-source/23109c8/ : SUCCESS in 1h 00m 30s\n- kolla-ansible-centos-source http://logs.openstack.org/99/640099/11/check/kolla-ansible-centos-source/451fef8/ : SUCCESS in 1h 30m 57s\n- kolla-ansible-centos-binary http://logs.openstack.org/99/640099/11/check/kolla-ansible-centos-binary/895004a/ : SUCCESS in 1h 11m 42s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/99/640099/11/check/tripleo-build-containers-centos-7/d1fd99a/ : SUCCESS in 39m 18s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/99/640099/11/check/kolla-build-oraclelinux-binary/b4c745b/ : SUCCESS in 53m 29s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/99/640099/11/check/kolla-build-oraclelinux-source/be5448e/ : SUCCESS in 36m 22s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/99/640099/11/check/kolla-ansible-oraclelinux-source/b229c70/ : SUCCESS in 1h 18m 57s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/99/640099/11/check/kolla-ansible-oraclelinux-binary/e498cd7/ : SUCCESS in 1h 06m 11s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/99/640099/11/check/openstack-tox-lower-constraints/3382016/ : SUCCESS in 5m 34s\n- openstack-tox-pep8 http://logs.openstack.org/99/640099/11/check/openstack-tox-pep8/23d6a18/ : SUCCESS in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/99/640099/11/check/openstack-tox-py27/51b1710/ : SUCCESS in 5m 39s\n- openstack-tox-py36 http://logs.openstack.org/99/640099/11/check/openstack-tox-py36/ff1ff7a/ : SUCCESS in 5m 24s\n- openstack-tox-py37 http://logs.openstack.org/99/640099/11/check/openstack-tox-py37/c409ce6/ : SUCCESS in 6m 26s\n- requirements-check http://logs.openstack.org/99/640099/11/check/requirements-check/a0801c4/ : FAILURE in 3m 59s\n- openstack-tox-docs http://logs.openstack.org/99/640099/11/check/openstack-tox-docs/78c1d70/html/ : SUCCESS in 3m 19s\n- kolla-build-ubuntu-binary http://logs.openstack.org/99/640099/11/check/kolla-build-ubuntu-binary/5e37653/ : SUCCESS in 33m 40s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/99/640099/11/check/kolla-build-ubuntu-source/5260361/ : SUCCESS in 37m 27s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/99/640099/11/check/kolla-ansible-ubuntu-source/d91f9d0/ : SUCCESS in 57m 12s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/99/640099/11/check/kolla-ansible-ubuntu-binary/662b152/ : FAILURE in 29m 48s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d2c2b4fbacacd4a7cf8bebe53e6a05d3f3bc2bdb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-12 12:25:50.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 tripleo-build-containers-fedora-28 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"5b7701fe7be3e4f35457bc0160c8fde738bfa80d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-03-12 17:40:16.000000000","message":"Patch Set 11: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"a352668987f3172cf6b545d38f301d2ed51b55ef","author":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"date":"2019-03-13 09:56:24.000000000","message":"Patch Set 11: Workflow-1\n\nNeed to reproduce the issue I met when start digging into this","accounts_in_message":[],"_revision_number":11},{"id":"e4e1adcbaf1d0aea3ab268430b56d8880b2b93c1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-18 08:19:44.000000000","message":"Patch Set 11:\n\nany plans for updating it?","accounts_in_message":[],"_revision_number":11},{"id":"6dde817aaea7b3669a39fe5482da1febde87dc84","author":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"date":"2019-06-19 10:59:18.000000000","message":"Patch Set 11:\n\nSeems that I was wrong and the only thing in this CR that remains is exception handling fix and I\u0027m not sure it worths.","accounts_in_message":[],"_revision_number":11},{"id":"ba9c65c7002b833ebea74e5a220c7b585695b179","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-24 09:57:50.000000000","message":"Removed Code-Review+2 by Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n","accounts_in_message":[],"_revision_number":11},{"id":"9fc6f1a343ea03e375c9f22e464c8088f9825de9","author":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"date":"2019-06-24 14:53:37.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"2926d6ea1fb53b56b4b6ca1a0f746208e24cfd29","revisions":{"d29c49b8e9961fb7a2d0050422c0e2a9d404f200":{"kind":"REWORK","_number":1,"created":"2019-02-28 17:01:52.000000000","uploader":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"ref":"refs/changes/99/640099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/640099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/640099/1"}}},"commit":{"parents":[{"commit":"8fcd7ea9123840be5fdf382fefe9811086b724b7","subject":"Merge \"Switch to stestr\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8fcd7ea9123840be5fdf382fefe9811086b724b7"}]}],"author":{"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","date":"2019-02-28 16:53:06.000000000","tz":180},"committer":{"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","date":"2019-02-28 17:01:50.000000000","tz":180},"subject":"[WIP] Change number of squashed levels","message":"[WIP] Change number of squashed levels\n\nKolla squashes images down to the latest (actually, first)\nlayer in image\u0027s history, and this have the following drawbacks:\n\n* all meaning of having a chain of containers is lost as after\n  squashing you have one big image that contains only one layer.\n\n* sometimes this leads to problems when building multiple images\n  at once as some layers just disappears because of squashing.\n\nThis commit changes this a bit, squashing only layers from top\n(newest layer) to the first that have tags. This allows to keep\nolder layers intact, and effectively reduce size of container\nbeing built.\n\nChange-Id: Id65d7832298e537c82df214db8fd0d1ae4f81d20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d29c49b8e9961fb7a2d0050422c0e2a9d404f200"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d29c49b8e9961fb7a2d0050422c0e2a9d404f200"}]},"branch":"refs/heads/master"},"7267a6130feb109d70acca956341e0f8c67b48cb":{"kind":"REWORK","_number":2,"created":"2019-03-01 11:12:58.000000000","uploader":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"ref":"refs/changes/99/640099/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/640099/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/640099/2"}}},"commit":{"parents":[{"commit":"8fcd7ea9123840be5fdf382fefe9811086b724b7","subject":"Merge \"Switch to stestr\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8fcd7ea9123840be5fdf382fefe9811086b724b7"}]}],"author":{"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","date":"2019-02-28 16:53:06.000000000","tz":180},"committer":{"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","date":"2019-03-01 10:59:22.000000000","tz":180},"subject":"[WIP] Improve image squashing","message":"[WIP] Improve image squashing\n\nKolla squashes images staring from the first layer (oldest image in history),\nand this have the following drawbacks:\n\n* all meaning of having a chain of containers is lost as after\n  squashing you have one big image that contains only one layer.\n\n* this leads to failures when building multiple images at once because\n  docker-squash tryies to remove layers after squashing, but some\n  layers were used when building \u0027child\u0027 images, so docker-squash reports\n  an error.\n\nThis commit changes this a bit, allowing to specify an image where to\nstart squashing. For this a new option \u0027--squash-from\u0027 was added. It has\na format of \u0027\u003ckey\u003e:\u003cvalue\u003e\u0027 string. \u003ckey\u003e could be either \u0027tag\u0027 or\n\u0027tag-index\u0027:\n\nIf \u0027tag\u0027:\n\n  Use \u0027--squash-from tag:\u003cimage tag\u003e\u0027 to tell docker-squash\n  that image should be squashed starting at layer tagged with \u003cimage tag\u003e\n\nIf \u0027tag-index\u0027:\n\n  Use \u0027--squash-from tag-index:\u003cN\u003e\u0027 to tell docker-squash\n  that image should be squashed from Nth taggerd layer in image\u0027s history.\n  Intermediate layers that are not tagged are ignored.\n  Index \u00270\u0027 refers to current image (newest), index \u0027-1\u0027 is oldest one\n  (usually this is OS image). Most useful case is to have index \u00271\u0027,\n  that means \u0027squash starting from previous tagged image\u0027.\n\nChange-Id: Id65d7832298e537c82df214db8fd0d1ae4f81d20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7267a6130feb109d70acca956341e0f8c67b48cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7267a6130feb109d70acca956341e0f8c67b48cb"}]},"branch":"refs/heads/master"},"35881978e73cc85df70287240213d51b962ac6e6":{"kind":"REWORK","_number":3,"created":"2019-03-01 11:15:09.000000000","uploader":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"ref":"refs/changes/99/640099/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/640099/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/640099/3"}}},"commit":{"parents":[{"commit":"8fcd7ea9123840be5fdf382fefe9811086b724b7","subject":"Merge \"Switch to stestr\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8fcd7ea9123840be5fdf382fefe9811086b724b7"}]}],"author":{"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","date":"2019-02-28 16:53:06.000000000","tz":180},"committer":{"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","date":"2019-03-01 11:14:52.000000000","tz":180},"subject":"[WIP] Improve image squashing","message":"[WIP] Improve image squashing\n\nKolla squashes images staring from the first layer (oldest image in history),\nand this have the following drawbacks:\n\n* all meaning of having a chain of containers is lost as after\n  squashing you have one big image that contains only one layer.\n\n* this leads to failures when building multiple images at once because\n  docker-squash tryies to remove layers after squashing, but some\n  layers were used when building \u0027child\u0027 images, so docker-squash reports\n  an error.\n\nThis commit changes this a bit, allowing to specify an image where to\nstart squashing. For this a new option \u0027--squash-from\u0027 was added. It has\na format of \u0027\u003ckey\u003e:\u003cvalue\u003e\u0027 string. \u003ckey\u003e could be either \u0027tag\u0027 or\n\u0027tag-index\u0027:\n\nIf \u0027tag\u0027:\n\n  Use \u0027--squash-from tag:\u003cimage tag\u003e\u0027 to tell docker-squash\n  that image should be squashed starting at layer tagged with \u003cimage tag\u003e\n\nIf \u0027tag-index\u0027:\n\n  Use \u0027--squash-from tag-index:\u003cN\u003e\u0027 to tell docker-squash\n  that image should be squashed from Nth taggerd layer in image\u0027s history.\n  Intermediate layers that are not tagged are ignored.\n  Index \u00270\u0027 refers to current image (newest), index \u0027-1\u0027 is oldest one\n  (usually this is OS image). Most useful case is to have index \u00271\u0027,\n  that means \u0027squash starting from previous tagged image\u0027.\n\nChange-Id: Id65d7832298e537c82df214db8fd0d1ae4f81d20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35881978e73cc85df70287240213d51b962ac6e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35881978e73cc85df70287240213d51b962ac6e6"}]},"branch":"refs/heads/master"},"85ffc8d0e80b911d3f29a4c5ef24a973812d5947":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-03-02 18:43:41.000000000","uploader":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"ref":"refs/changes/99/640099/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/640099/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/640099/4"}}},"commit":{"parents":[{"commit":"8fcd7ea9123840be5fdf382fefe9811086b724b7","subject":"Merge \"Switch to stestr\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8fcd7ea9123840be5fdf382fefe9811086b724b7"}]}],"author":{"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","date":"2019-02-28 16:53:06.000000000","tz":180},"committer":{"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","date":"2019-03-02 18:43:20.000000000","tz":180},"subject":"Improve image squashing","message":"Improve image squashing\n\nKolla squashes images staring from the first layer (oldest image in history),\nand this have the following drawbacks:\n\n* all meaning of having a chain of containers is lost as after\n  squashing you have one big image that contains only one layer.\n\n* this leads to failures when building multiple images at once because\n  docker-squash tryies to remove layers after squashing, but some\n  layers were used when building \u0027child\u0027 images, so docker-squash reports\n  an error.\n\nThis commit changes this a bit, allowing to specify an image where to\nstart squashing. For this a new option \u0027--squash-from\u0027 was added. It has\na format of \u0027\u003ckey\u003e:\u003cvalue\u003e\u0027 string. \u003ckey\u003e could be either \u0027tag\u0027 or\n\u0027tag-index\u0027:\n\nIf \u0027tag\u0027:\n\n  Use \u0027--squash-from tag:\u003cimage tag\u003e\u0027 to tell docker-squash\n  that image should be squashed starting at layer tagged with \u003cimage tag\u003e\n\nIf \u0027tag-index\u0027:\n\n  Use \u0027--squash-from tag-index:\u003cN\u003e\u0027 to tell docker-squash\n  that image should be squashed from Nth taggerd layer in image\u0027s history.\n  Intermediate layers that are not tagged are ignored.\n  Index \u00270\u0027 refers to current image (newest), index \u0027-1\u0027 is oldest one\n  (usually this is OS image). Most useful case is to have index \u00271\u0027,\n  that means \u0027squash starting from previous tagged image\u0027.\n\nChange-Id: Id65d7832298e537c82df214db8fd0d1ae4f81d20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85ffc8d0e80b911d3f29a4c5ef24a973812d5947"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85ffc8d0e80b911d3f29a4c5ef24a973812d5947"}]},"branch":"refs/heads/master"},"7cafee4e29725af9fe27202c3be43f7c16275e28":{"kind":"REWORK","_number":5,"created":"2019-03-10 10:22:19.000000000","uploader":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"ref":"refs/changes/99/640099/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/640099/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/640099/5"}}},"commit":{"parents":[{"commit":"8fcd7ea9123840be5fdf382fefe9811086b724b7","subject":"Merge \"Switch to stestr\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8fcd7ea9123840be5fdf382fefe9811086b724b7"}]}],"author":{"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","date":"2019-02-28 16:53:06.000000000","tz":180},"committer":{"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","date":"2019-03-10 10:12:03.000000000","tz":180},"subject":"Improve image squashing","message":"Improve image squashing\n\nKolla squashes images staring from the first layer (oldest image in history),\nand this have the following drawbacks:\n\n* after squashing you have one big image that contains only one layer,\n  so it\u0027s not possible to have a chain of layers to use several relatively\n  big images as common ancestors to multiple small (final) layers.\n\n* this leads to failures when building multiple images at once because\n  docker-squash tryies to remove layers after squashing, but some\n  layers were used when building \u0027child\u0027 images, so docker-squash reports\n  an error.\n\nThis commit changes this a bit, allowing to specify an image where to\nstart squashing. For this a new option \u0027--squash-from\u0027 was added. It has\na format of \u0027\u003ckey\u003e:\u003cvalue\u003e\u0027 string. \u003ckey\u003e could be either \u0027tag\u0027 or\n\u0027tag-index\u0027:\n\nIf \u0027tag\u0027:\n\n  Use \u0027--squash-from tag:\u003cimage tag\u003e\u0027 to tell docker-squash\n  that image should be squashed starting at layer tagged with \u003cimage tag\u003e\n\nIf \u0027tag-index\u0027:\n\n  Use \u0027--squash-from tag-index:\u003cN\u003e\u0027 to tell docker-squash\n  that image should be squashed from Nth taggerd layer in image\u0027s history.\n  Intermediate layers that are not tagged are ignored. Index could be\n  used the same way as for python lists:\n    0 - index of an oldest layer (at the bottom of \u0027docker history\u0027 output)\n    1 - index of the first tagged image (counting from bottom), usualli this\n        is an operating system image\n    -1 - index of the last tagged image\n  The most typical usage should be \u0027--squash-from tag-index:-1\u0027 to squash\n  all layers produced during the build of the current container.\n\nChange-Id: Id65d7832298e537c82df214db8fd0d1ae4f81d20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cafee4e29725af9fe27202c3be43f7c16275e28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cafee4e29725af9fe27202c3be43f7c16275e28"}]},"branch":"refs/heads/master"},"c919464cc6dd724ffea8f50a2f7157c952f5f710":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-03-10 15:23:38.000000000","uploader":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"ref":"refs/changes/99/640099/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/640099/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/640099/6"}}},"commit":{"parents":[{"commit":"30e1e8d8a3e73756e0558cefa1c931bb748f8e36","subject":"Merge \"Fix Octavia-API extend_start.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30e1e8d8a3e73756e0558cefa1c931bb748f8e36"}]}],"author":{"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","date":"2019-02-28 16:53:06.000000000","tz":180},"committer":{"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","date":"2019-03-10 15:23:38.000000000","tz":0},"subject":"Improve image squashing","message":"Improve image squashing\n\nKolla squashes images staring from the first layer (oldest image in history),\nand this have the following drawbacks:\n\n* after squashing you have one big image that contains only one layer,\n  so it\u0027s not possible to have a chain of layers to use several relatively\n  big images as common ancestors to multiple small (final) layers.\n\n* this leads to failures when building multiple images at once because\n  docker-squash tryies to remove layers after squashing, but some\n  layers were used when building \u0027child\u0027 images, so docker-squash reports\n  an error.\n\nThis commit changes this a bit, allowing to specify an image where to\nstart squashing. For this a new option \u0027--squash-from\u0027 was added. It has\na format of \u0027\u003ckey\u003e:\u003cvalue\u003e\u0027 string. \u003ckey\u003e could be either \u0027tag\u0027 or\n\u0027tag-index\u0027:\n\nIf \u0027tag\u0027:\n\n  Use \u0027--squash-from tag:\u003cimage tag\u003e\u0027 to tell docker-squash\n  that image should be squashed starting at layer tagged with \u003cimage tag\u003e\n\nIf \u0027tag-index\u0027:\n\n  Use \u0027--squash-from tag-index:\u003cN\u003e\u0027 to tell docker-squash\n  that image should be squashed from Nth taggerd layer in image\u0027s history.\n  Intermediate layers that are not tagged are ignored. Index could be\n  used the same way as for python lists:\n    0 - index of an oldest layer (at the bottom of \u0027docker history\u0027 output)\n    1 - index of the first tagged image (counting from bottom), usualli this\n        is an operating system image\n    -1 - index of the last tagged image\n  The most typical usage should be \u0027--squash-from tag-index:-1\u0027 to squash\n  all layers produced during the build of the current container.\n\nChange-Id: Id65d7832298e537c82df214db8fd0d1ae4f81d20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c919464cc6dd724ffea8f50a2f7157c952f5f710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c919464cc6dd724ffea8f50a2f7157c952f5f710"}]},"branch":"refs/heads/master"},"935b747fd544d964a8688cc3dcc51b8e5c948a5d":{"kind":"REWORK","_number":7,"created":"2019-03-11 20:40:51.000000000","uploader":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"ref":"refs/changes/99/640099/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/640099/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/640099/7"}}},"commit":{"parents":[{"commit":"30e1e8d8a3e73756e0558cefa1c931bb748f8e36","subject":"Merge \"Fix Octavia-API extend_start.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30e1e8d8a3e73756e0558cefa1c931bb748f8e36"}]}],"author":{"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","date":"2019-02-28 16:53:06.000000000","tz":180},"committer":{"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","date":"2019-03-11 20:40:40.000000000","tz":180},"subject":"Fix image squashing","message":"Fix image squashing\n\nKolla squashes images staring from the first (oldest in the history) layer,\nand this have the following drawbacks:\n\n* after squashing you have one big image that contains only one layer,\n  so it\u0027s not possible to have a chain of layers to use several relatively\n  big images as common ancestors to multiple small (final) layers.\n\n* this leads to failures when building multiple images at once because\n  docker-squash tryies to remove layers after squashing, but some\n  layers were used when building \u0027child\u0027 images, so docker-squash reports\n  an error.\n\nThis commit fixes this by selecting previous tagged image as \u0027starting point\u0027.\nIf there is no such images for some reason then squashing skipped with\nwarning message.\n\nChange-Id: Id65d7832298e537c82df214db8fd0d1ae4f81d20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/935b747fd544d964a8688cc3dcc51b8e5c948a5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/935b747fd544d964a8688cc3dcc51b8e5c948a5d"}]},"branch":"refs/heads/master"},"4d8c15c4fd2c58034e91468d6c1ae9c3a8fe37df":{"kind":"REWORK","_number":8,"created":"2019-03-12 04:59:32.000000000","uploader":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"ref":"refs/changes/99/640099/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/640099/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/640099/8"}}},"commit":{"parents":[{"commit":"30e1e8d8a3e73756e0558cefa1c931bb748f8e36","subject":"Merge \"Fix Octavia-API extend_start.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30e1e8d8a3e73756e0558cefa1c931bb748f8e36"}]}],"author":{"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","date":"2019-02-28 16:53:06.000000000","tz":180},"committer":{"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","date":"2019-03-12 04:59:11.000000000","tz":180},"subject":"Fix image squashing","message":"Fix image squashing\n\nKolla squashes images staring from the first (oldest in the history) layer,\nand this have the following drawbacks:\n\n* after squashing you have one big image that contains only one layer,\n  so it\u0027s not possible to have a chain of layers to use several relatively\n  big images as common ancestors to multiple small (final) layers.\n\n* this leads to failures when building multiple images at once because\n  docker-squash tryies to remove layers after squashing, but some\n  layers were used when building \u0027child\u0027 images, so docker-squash reports\n  an error.\n\nThis commit fixes this by selecting previous tagged image as \u0027starting point\u0027.\nIf there is no such images for some reason then squashing skipped with\nwarning message.\n\nChange-Id: Id65d7832298e537c82df214db8fd0d1ae4f81d20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d8c15c4fd2c58034e91468d6c1ae9c3a8fe37df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d8c15c4fd2c58034e91468d6c1ae9c3a8fe37df"}]},"branch":"refs/heads/master"},"bdefb94f032ad07c5d4599ccb363ebd4816f286c":{"kind":"REWORK","_number":9,"created":"2019-03-12 09:46:13.000000000","uploader":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"ref":"refs/changes/99/640099/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/640099/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/640099/9"}}},"commit":{"parents":[{"commit":"30e1e8d8a3e73756e0558cefa1c931bb748f8e36","subject":"Merge \"Fix Octavia-API extend_start.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30e1e8d8a3e73756e0558cefa1c931bb748f8e36"}]}],"author":{"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","date":"2019-02-28 16:53:06.000000000","tz":180},"committer":{"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","date":"2019-03-12 09:43:15.000000000","tz":180},"subject":"Fix image squashing","message":"Fix image squashing\n\nKolla squashes images staring from the first (oldest in the history) layer,\nand this have the following drawbacks:\n\n* after squashing you have one big image that contains only one layer,\n  so it\u0027s not possible to have a chain of layers to use several relatively\n  big images as common ancestors to multiple small (final) layers.\n\n* this leads to failures when building multiple images at once because\n  docker-squash tryies to remove layers after squashing, but some\n  layers were used when building \u0027child\u0027 images, so docker-squash reports\n  an error.\n\nThis commit fixes this by selecting previous tagged image as \u0027starting point\u0027.\nIf there is no such images for some reason then squashing skipped with\nwarning message.\n\nIt also disables squashing if \u0027docker-squash\u0027 is not installed instead\nof failure.\n\nChange-Id: Id65d7832298e537c82df214db8fd0d1ae4f81d20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bdefb94f032ad07c5d4599ccb363ebd4816f286c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bdefb94f032ad07c5d4599ccb363ebd4816f286c"}]},"branch":"refs/heads/master"},"da967521b95c37fdc81ce34f677a6b25a787abc7":{"kind":"REWORK","_number":10,"created":"2019-03-12 10:10:12.000000000","uploader":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"ref":"refs/changes/99/640099/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/640099/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/640099/10"}}},"commit":{"parents":[{"commit":"30e1e8d8a3e73756e0558cefa1c931bb748f8e36","subject":"Merge \"Fix Octavia-API extend_start.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30e1e8d8a3e73756e0558cefa1c931bb748f8e36"}]}],"author":{"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","date":"2019-02-28 16:53:06.000000000","tz":180},"committer":{"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","date":"2019-03-12 10:09:57.000000000","tz":180},"subject":"Fix image squashing","message":"Fix image squashing\n\nKolla squashes images staring from the first (oldest in the history) layer,\nand this have the following drawbacks:\n\n* after squashing you have one big image that contains only one layer,\n  so it\u0027s not possible to have a chain of layers to use several relatively\n  big images as common ancestors to multiple small (final) layers.\n\n* this leads to failures when building multiple images at once because\n  docker-squash tryies to remove layers after squashing, but some\n  layers were used when building \u0027child\u0027 images, so docker-squash reports\n  an error.\n\nThis commit fixes this by selecting previous tagged image as \u0027starting point\u0027.\nIf there is no such images for some reason then squashing skipped with\nwarning message.\n\nIt also disables squashing if \u0027docker-squash\u0027 is not installed instead\nof failure.\n\nChange-Id: Id65d7832298e537c82df214db8fd0d1ae4f81d20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/da967521b95c37fdc81ce34f677a6b25a787abc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/da967521b95c37fdc81ce34f677a6b25a787abc7"}]},"branch":"refs/heads/master"},"2926d6ea1fb53b56b4b6ca1a0f746208e24cfd29":{"kind":"REWORK","_number":11,"created":"2019-03-12 10:13:51.000000000","uploader":{"_account_id":7562,"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","username":"dteselkin"},"ref":"refs/changes/99/640099/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/640099/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/640099/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/640099/11"}}},"commit":{"parents":[{"commit":"30e1e8d8a3e73756e0558cefa1c931bb748f8e36","subject":"Merge \"Fix Octavia-API extend_start.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30e1e8d8a3e73756e0558cefa1c931bb748f8e36"}]}],"author":{"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","date":"2019-02-28 16:53:06.000000000","tz":180},"committer":{"name":"Dmitry Teselkin","email":"dteselkin@mirantis.com","date":"2019-03-12 10:13:02.000000000","tz":180},"subject":"Fix image squashing","message":"Fix image squashing\n\nKolla squashes images staring from the first (oldest in the history) layer,\nand this have the following drawbacks:\n\n* after squashing you have one big image that contains only one layer,\n  so it\u0027s not possible to have a chain of layers to use several relatively\n  big images as common ancestors to multiple small (final) layers.\n\n* this leads to failures when building multiple images at once because\n  docker-squash tryies to remove layers after squashing, but some\n  layers were used when building \u0027child\u0027 images, so docker-squash reports\n  an error.\n\nThis commit fixes this by selecting previous tagged image as \u0027starting point\u0027.\nIf there is no such images for some reason then squashing skipped with\nwarning message.\n\nIt also adds \u0027docker-squash\u0027 to requirements.txt.\n\nChange-Id: Id65d7832298e537c82df214db8fd0d1ae4f81d20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2926d6ea1fb53b56b4b6ca1a0f746208e24cfd29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2926d6ea1fb53b56b4b6ca1a0f746208e24cfd29"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
