)]}'
{"id":"openstack%2Fkolla~642416","triplet_id":"openstack%2Fkolla~master~I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa","project":"openstack/kolla","branch":"master","topic":"feature/bump_up_toolbox","hashtags":[],"change_id":"I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa","subject":"Bump up pkg versions in kolla-toolbox","status":"MERGED","created":"2019-03-11 10:49:55.000000000","updated":"2019-08-09 08:59:30.000000000","submitted":"2019-08-08 18:25:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"642416-1565288706819-25fc65c6","meta_rev_id":"6a5c0fbdfbdf3bfbc48f94418b13759db3eb56e1","_number":642416,"virtual_id_number":642416,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-08-08 18:25:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"date":"2019-08-08 16:29:45.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-08 13:58:47.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-08-08 14:02:32.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-08-09 08:59:30.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 09:09:48.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-08-05 14:26:20.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-08-06 06:00:38.000000000","updated_by":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"reviewer":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"state":"REVIEWER"},{"updated":"2019-08-08 12:58:04.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-08-08 14:02:32.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-08-08 16:29:45.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-08 18:25:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"def526689ec20e0518d74b4e7677dc0c49f49296","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-03-11 10:49:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"016a88ec4590733cda5c9cebcc611da14d01ebf8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-11 11:48:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/16/642416/1/check/kolla-build-centos-binary/2772b99/ : SUCCESS in 44m 25s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/16/642416/1/check/kolla-build-centos-source/2e322a6/ : SUCCESS in 47m 41s\n- kolla-ansible-centos-source http://logs.openstack.org/16/642416/1/check/kolla-ansible-centos-source/eed89d3/ : FAILURE in 43m 49s\n- kolla-ansible-centos-binary http://logs.openstack.org/16/642416/1/check/kolla-ansible-centos-binary/8b3022e/ : FAILURE in 36m 27s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/16/642416/1/check/tripleo-build-containers-centos-7/8c61148/ : SUCCESS in 53m 52s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/16/642416/1/check/kolla-build-oraclelinux-binary/68efce4/ : SUCCESS in 54m 25s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/16/642416/1/check/kolla-build-oraclelinux-source/b7ab72f/ : SUCCESS in 49m 52s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/16/642416/1/check/kolla-ansible-oraclelinux-source/e38b56c/ : FAILURE in 55m 32s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/16/642416/1/check/kolla-ansible-oraclelinux-binary/434e9be/ : FAILURE in 53m 16s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/16/642416/1/check/openstack-tox-lower-constraints/1fd5434/ : SUCCESS in 4m 06s\n- openstack-tox-py36 http://logs.openstack.org/16/642416/1/check/openstack-tox-py36/432c110/ : SUCCESS in 3m 40s\n- openstack-tox-py37 http://logs.openstack.org/16/642416/1/check/openstack-tox-py37/249059d/ : SUCCESS in 3m 54s\n- openstack-tox-pep8 http://logs.openstack.org/16/642416/1/check/openstack-tox-pep8/66d3ae4/ : SUCCESS in 3m 54s\n- openstack-tox-py27 http://logs.openstack.org/16/642416/1/check/openstack-tox-py27/c7395b8/ : SUCCESS in 3m 04s\n- openstack-tox-docs http://logs.openstack.org/16/642416/1/check/openstack-tox-docs/8910a75/html/ : SUCCESS in 3m 10s\n- kolla-build-ubuntu-binary http://logs.openstack.org/16/642416/1/check/kolla-build-ubuntu-binary/9f7bc09/ : SUCCESS in 27m 52s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/16/642416/1/check/kolla-build-ubuntu-source/09406d8/ : SUCCESS in 30m 10s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/16/642416/1/check/kolla-ansible-ubuntu-source/c9e6005/ : FAILURE in 32m 15s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/16/642416/1/check/kolla-ansible-ubuntu-binary/66f7355/ : FAILURE in 29m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3fcbf943ee8c97f9be392451f6e83875b0f4f468","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-03-11 12:08:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d1d58ad2a397d8a2aab78b0a28c3640f379e4691","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-11 12:40:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 http://logs.rdoproject.org/16/642416/2/openstack-check/tripleo-build-containers-fedora-28/11be2f3/ : SUCCESS in 28m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bceba7597b2ae66321de61de16b026013fd88a44","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-03-11 12:55:12.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c333a88b8a42bc14a281fac80728837fff67f113","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-11 12:55:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/16/642416/2/check/kolla-build-centos-binary/35e41f0/ : SUCCESS in 27m 55s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/16/642416/2/check/kolla-build-centos-source/875c6dd/ : SUCCESS in 45m 06s\n- kolla-ansible-centos-source http://logs.openstack.org/16/642416/2/check/kolla-ansible-centos-source/255d33e/ : FAILURE in 41m 25s\n- kolla-ansible-centos-binary http://logs.openstack.org/16/642416/2/check/kolla-ansible-centos-binary/768114b/ : FAILURE in 42m 53s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/16/642416/2/check/tripleo-build-containers-centos-7/b0540a7/ : SUCCESS in 37m 00s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/16/642416/2/check/kolla-build-oraclelinux-binary/f6f04e7/ : SUCCESS in 33m 27s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/16/642416/2/check/kolla-build-oraclelinux-source/3668c0e/ : SUCCESS in 35m 02s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/16/642416/2/check/kolla-ansible-oraclelinux-source/aa54d52/ : FAILURE in 40m 55s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/16/642416/2/check/kolla-ansible-oraclelinux-binary/a9a8568/ : FAILURE in 44m 13s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/16/642416/2/check/openstack-tox-lower-constraints/c10208b/ : SUCCESS in 4m 09s\n- openstack-tox-py36 http://logs.openstack.org/16/642416/2/check/openstack-tox-py36/877d3ca/ : SUCCESS in 3m 19s\n- openstack-tox-py37 http://logs.openstack.org/16/642416/2/check/openstack-tox-py37/80d253c/ : SUCCESS in 4m 59s\n- openstack-tox-pep8 http://logs.openstack.org/16/642416/2/check/openstack-tox-pep8/3d522f4/ : SUCCESS in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/16/642416/2/check/openstack-tox-py27/62f941c/ : SUCCESS in 3m 37s\n- openstack-tox-docs http://logs.openstack.org/16/642416/2/check/openstack-tox-docs/fbde1e1/html/ : SUCCESS in 3m 00s\n- kolla-build-ubuntu-binary http://logs.openstack.org/16/642416/2/check/kolla-build-ubuntu-binary/e3c623e/ : SUCCESS in 31m 41s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/16/642416/2/check/kolla-build-ubuntu-source/afca346/ : SUCCESS in 33m 22s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/16/642416/2/check/kolla-ansible-ubuntu-source/46ec461/ : FAILURE in 29m 11s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/16/642416/2/check/kolla-ansible-ubuntu-binary/bf1f69d/ : FAILURE in 28m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"10b23ae3b059644415f34fa32e744cc4c31d5fb3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-03-11 13:01:43.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"a7e9e725038c705044eb46e9b00686f7f850901e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-03-11 13:26:57.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"02ac59c49f3412061bfaca892733acff1a447375","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-03-11 13:41:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ad25181f89bc8ace654bf9e15806f2041e3b25c9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-03-11 13:56:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a76f1749a9c725658eadc832abf779b72920d6b1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-03-11 13:56:55.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"f7b95e24fc68bfd607e838d78aca9094bde7d765","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-03-11 14:01:17.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"883472d81a1f830cf14f97c8a15056c7c87600db","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-03-11 14:21:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6502ff8577bd54f82a9d853d47840ba810b8910b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-11 14:55:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/16/642416/3/check/kolla-build-centos-binary/a962227/ : SUCCESS in 34m 28s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/16/642416/3/check/kolla-build-centos-source/4318324/ : SUCCESS in 46m 24s\n- kolla-ansible-centos-source http://logs.openstack.org/16/642416/3/check/kolla-ansible-centos-source/819fe44/ : FAILURE in 51m 37s\n- kolla-ansible-centos-binary http://logs.openstack.org/16/642416/3/check/kolla-ansible-centos-binary/5cb8630/ : FAILURE in 36m 55s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/16/642416/3/check/tripleo-build-containers-centos-7/9f836ee/ : SUCCESS in 41m 53s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/16/642416/3/check/kolla-build-oraclelinux-binary/b61b2a6/ : SUCCESS in 40m 37s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/16/642416/3/check/kolla-build-oraclelinux-source/c900231/ : SUCCESS in 42m 34s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/16/642416/3/check/kolla-ansible-oraclelinux-source/e3aed65/ : FAILURE in 45m 16s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/16/642416/3/check/kolla-ansible-oraclelinux-binary/58995d0/ : FAILURE in 37m 22s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/16/642416/3/check/openstack-tox-lower-constraints/e803ed7/ : SUCCESS in 3m 29s\n- openstack-tox-py36 http://logs.openstack.org/16/642416/3/check/openstack-tox-py36/25a7c24/ : SUCCESS in 3m 18s\n- openstack-tox-py37 http://logs.openstack.org/16/642416/3/check/openstack-tox-py37/61202e8/ : SUCCESS in 4m 07s\n- openstack-tox-pep8 http://logs.openstack.org/16/642416/3/check/openstack-tox-pep8/88372c5/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/16/642416/3/check/openstack-tox-py27/803f473/ : SUCCESS in 3m 42s\n- openstack-tox-docs http://logs.openstack.org/16/642416/3/check/openstack-tox-docs/af9c4c2/html/ : SUCCESS in 3m 28s\n- kolla-build-ubuntu-binary http://logs.openstack.org/16/642416/3/check/kolla-build-ubuntu-binary/bad58dd/ : SUCCESS in 32m 27s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/16/642416/3/check/kolla-build-ubuntu-source/d17f381/ : SUCCESS in 38m 03s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/16/642416/3/check/kolla-ansible-ubuntu-source/f142dd2/ : FAILURE in 29m 05s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/16/642416/3/check/kolla-ansible-ubuntu-binary/fa2e532/ : FAILURE in 27m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cbd8b1d389e246f6c18e09807dbcab84758a60ea","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-03-11 15:18:19.000000000","message":"Patch Set 3:\n\nSeems Ansible modules API changed between 2.2.0.0 and 2.7.x, will investigate.","accounts_in_message":[],"_revision_number":3},{"id":"9a27f852a70954791cf2ac8ddf2da0d0dac310ad","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-11 18:29:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 http://logs.rdoproject.org/16/642416/3/openstack-check/tripleo-build-containers-fedora-28/efce60b/ : SUCCESS in 28m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7c367b4269e0c6c5a83fb0d98cfaad06fdde3d87","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-08 15:01:08.000000000","message":"Patch Set 3:\n\nChristian Berendt raised a bug about this: https://bugs.launchpad.net/kolla/+bug/1823736. Please add Closes-Bug.","accounts_in_message":[],"_revision_number":3},{"id":"54afd3b03737d94819ab89e1b0582bff2fcfeceb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-21 12:34:48.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"621908ae36ed0b499bc47e2d9c53808ea4038701","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-21 12:35:01.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"e05c357730b3862a04a053c983936b7af023b7c9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-21 13:16:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/16/642416/4/openstack-check/tripleo-build-containers-fedora-28-master/553f4e5/ : SUCCESS in 36m 26s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/16/642416/4/openstack-check/tripleo-build-containers-fedora-28-stein/36f9331/ : SUCCESS in 33m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"76eaed9af2d2e5f2d1dcb1f8bb10acb406a011a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 13:58:36.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/16/642416/4/check/kolla-build-centos-binary/bd20a38/ : SUCCESS in 1h 01m 29s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/16/642416/4/check/kolla-build-centos-source/6b82bca/ : SUCCESS in 34m 47s\n- kolla-ansible-centos-source http://logs.openstack.org/16/642416/4/check/kolla-ansible-centos-source/13832b2/ : FAILURE in 47m 15s\n- kolla-ansible-centos-binary http://logs.openstack.org/16/642416/4/check/kolla-ansible-centos-binary/792204b/ : FAILURE in 42m 18s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/16/642416/4/check/kolla-ansible-centos-source-upgrade/a1c0724/ : FAILURE in 1h 15m 20s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/16/642416/4/check/tripleo-build-containers-centos-7/5ac5201/ : SUCCESS in 45m 38s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/16/642416/4/check/kolla-build-oraclelinux-binary/c6b31b5/ : SUCCESS in 51m 45s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/16/642416/4/check/kolla-build-oraclelinux-source/1f18576/ : SUCCESS in 48m 40s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/16/642416/4/check/kolla-ansible-oraclelinux-source/41bb095/ : FAILURE in 54m 25s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/16/642416/4/check/kolla-ansible-oraclelinux-binary/cf94f19/ : FAILURE in 46m 54s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/16/642416/4/check/openstack-tox-lower-constraints/705ac03/ : SUCCESS in 3m 40s\n- openstack-tox-pep8 http://logs.openstack.org/16/642416/4/check/openstack-tox-pep8/c55f7cf/ : SUCCESS in 3m 57s\n- openstack-tox-py27 http://logs.openstack.org/16/642416/4/check/openstack-tox-py27/f43fae4/ : SUCCESS in 5m 02s\n- openstack-tox-py36 http://logs.openstack.org/16/642416/4/check/openstack-tox-py36/8155938/ : SUCCESS in 3m 33s\n- openstack-tox-py37 http://logs.openstack.org/16/642416/4/check/openstack-tox-py37/1f78a75/ : SUCCESS in 3m 26s\n- openstack-tox-docs http://logs.openstack.org/16/642416/4/check/openstack-tox-docs/4d7cb19/html/ : SUCCESS in 2m 59s\n- kolla-build-ubuntu-binary http://logs.openstack.org/16/642416/4/check/kolla-build-ubuntu-binary/a6252e0/ : SUCCESS in 28m 30s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/16/642416/4/check/kolla-build-ubuntu-source/a3b8f8b/ : SUCCESS in 31m 13s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/16/642416/4/check/kolla-ansible-ubuntu-source/ce1574c/ : FAILURE in 29m 25s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/16/642416/4/check/kolla-ansible-ubuntu-binary/770f536/ : FAILURE in 29m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/16/642416/4/check/kolla-ansible-ubuntu-source-upgrade/20c02bc/ : FAILURE in 54m 10s","accounts_in_message":[],"_revision_number":4},{"id":"1e70857de905818ba9eb42f0957355457b2ac9ad","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-21 16:05:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5887375199a2c912ee926ac0ec7ba8d10855d18e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-06 06:53:08.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"c75872cd6803d55590c069caf81a4e7c35205602","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-06 07:10:01.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f6da66b25474d634d3fb2943d5dc6990b6948189","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-06 07:12:07.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"4985454f1efa0b056fbe06a51e7d7af65ce4d8de","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 07:49:21.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/16/642416/6/openstack-check/tripleo-build-containers-fedora-28-master/32d2417/ : SUCCESS in 36m 27s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/16/642416/6/openstack-check/tripleo-build-containers-fedora-28-stein/bcc7346/ : SUCCESS in 33m 58s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a52e5891ffb82f0dd8648c085690e8d91af69c95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 08:44:24.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/16/642416/6/check/kolla-build-centos-binary/3e1a287/ : SUCCESS in 40m 16s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/16/642416/6/check/kolla-build-centos-source/8b52312/ : SUCCESS in 59m 43s\n- kolla-ansible-centos-source http://logs.openstack.org/16/642416/6/check/kolla-ansible-centos-source/43d7e08/ : FAILURE in 52m 17s\n- kolla-ansible-centos-binary http://logs.openstack.org/16/642416/6/check/kolla-ansible-centos-binary/1ad51b2/ : FAILURE in 38m 15s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/16/642416/6/check/kolla-ansible-centos-source-upgrade/b97cfc0/ : FAILURE in 1h 23m 35s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/16/642416/6/check/tripleo-build-containers-centos-7/9e60780/ : SUCCESS in 44m 11s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/16/642416/6/check/kolla-build-oraclelinux-binary/1baea26/ : SUCCESS in 56m 53s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/16/642416/6/check/kolla-build-oraclelinux-source/66f2b2e/ : SUCCESS in 51m 00s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/16/642416/6/check/kolla-ansible-oraclelinux-source/3c32a31/ : FAILURE in 44m 22s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/16/642416/6/check/kolla-ansible-oraclelinux-binary/8690e1f/ : FAILURE in 39m 15s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/16/642416/6/check/openstack-tox-lower-constraints/b844c64/ : SUCCESS in 4m 43s\n- openstack-tox-pep8 http://logs.openstack.org/16/642416/6/check/openstack-tox-pep8/a7e60ea/ : SUCCESS in 3m 09s\n- openstack-tox-py27 http://logs.openstack.org/16/642416/6/check/openstack-tox-py27/0b67185/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/16/642416/6/check/openstack-tox-py36/2851014/ : SUCCESS in 3m 03s\n- openstack-tox-py37 http://logs.openstack.org/16/642416/6/check/openstack-tox-py37/61cfdaa/ : SUCCESS in 3m 29s\n- openstack-tox-docs http://logs.openstack.org/16/642416/6/check/openstack-tox-docs/86f55b0/html/ : SUCCESS in 2m 41s\n- kolla-build-ubuntu-binary http://logs.openstack.org/16/642416/6/check/kolla-build-ubuntu-binary/ea0ebb1/ : SUCCESS in 58m 28s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/16/642416/6/check/kolla-build-ubuntu-source/403656d/ : SUCCESS in 35m 04s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/16/642416/6/check/kolla-ansible-ubuntu-source/d37d007/ : FAILURE in 31m 26s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/16/642416/6/check/kolla-ansible-ubuntu-binary/f5e4d83/ : FAILURE in 26m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/16/642416/6/check/kolla-ansible-ubuntu-source-upgrade/fb2a1ee/ : FAILURE in 52m 08s","accounts_in_message":[],"_revision_number":6},{"id":"18e9ba6ef66cb5679f6f9e58a023f8f20c51d8a4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-06 14:36:23.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"7444728b338966e56cf6bf841629439dccd9120c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-06 14:37:40.000000000","message":"Patch Set 6:\n\nLooks like kolla_keystone_user needs to set no_log on its password argument:\n\n[WARNING]: Module did not set no_log for password","accounts_in_message":[],"_revision_number":6},{"id":"78ceaca74b51931a69e29c10e18b21149f08b7d2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-07 07:06:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d96de81d487bc0b981b6e8af1878bfa4c958a793","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-07 07:51:52.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/16/642416/7/openstack-check/tripleo-build-containers-fedora-28-master/eb2b204/ : SUCCESS in 37m 02s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/16/642416/7/openstack-check/tripleo-build-containers-fedora-28-stein/81fc87b/ : SUCCESS in 39m 36s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9e773364fe8071b23a17b05f1e7acb440014d2c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 08:37:11.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/16/642416/7/check/kolla-build-centos-binary/2cb0880/ : SUCCESS in 39m 31s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/16/642416/7/check/kolla-build-centos-source/c184380/ : SUCCESS in 1h 00m 40s\n- kolla-ansible-centos-source http://logs.openstack.org/16/642416/7/check/kolla-ansible-centos-source/e223e24/ : FAILURE in 55m 21s\n- kolla-ansible-centos-binary http://logs.openstack.org/16/642416/7/check/kolla-ansible-centos-binary/d4670e1/ : FAILURE in 43m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/16/642416/7/check/kolla-ansible-centos-source-upgrade/b518740/ : FAILURE in 59m 58s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/16/642416/7/check/tripleo-build-containers-centos-7/7be76ba/ : SUCCESS in 40m 16s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/16/642416/7/check/kolla-build-oraclelinux-binary/6b7f731/ : SUCCESS in 37m 38s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/16/642416/7/check/kolla-build-oraclelinux-source/49a37c8/ : SUCCESS in 55m 24s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/16/642416/7/check/kolla-ansible-oraclelinux-source/679f678/ : FAILURE in 48m 29s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/16/642416/7/check/kolla-ansible-oraclelinux-binary/40e3bdb/ : FAILURE in 43m 50s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/16/642416/7/check/openstack-tox-lower-constraints/39fc0ba/ : SUCCESS in 4m 09s\n- openstack-tox-pep8 http://logs.openstack.org/16/642416/7/check/openstack-tox-pep8/0c1c90d/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/16/642416/7/check/openstack-tox-py27/312bed2/ : SUCCESS in 2m 42s\n- openstack-tox-py36 http://logs.openstack.org/16/642416/7/check/openstack-tox-py36/4e1d9cd/ : SUCCESS in 3m 07s\n- openstack-tox-py37 http://logs.openstack.org/16/642416/7/check/openstack-tox-py37/b34b628/ : SUCCESS in 3m 11s\n- openstack-tox-docs http://logs.openstack.org/16/642416/7/check/openstack-tox-docs/540cb2c/html/ : SUCCESS in 2m 59s\n- kolla-build-ubuntu-binary http://logs.openstack.org/16/642416/7/check/kolla-build-ubuntu-binary/acf607c/ : SUCCESS in 30m 01s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/16/642416/7/check/kolla-build-ubuntu-source/87a3a00/ : SUCCESS in 30m 57s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/16/642416/7/check/kolla-ansible-ubuntu-source/257e41d/ : FAILURE in 25m 16s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/16/642416/7/check/kolla-ansible-ubuntu-binary/4b122f2/ : FAILURE in 28m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/16/642416/7/check/kolla-ansible-ubuntu-source-upgrade/574ee68/ : FAILURE in 57m 11s","accounts_in_message":[],"_revision_number":7},{"id":"6e36125870550dd091b0f65247eeb44454a39d68","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-07 08:39:06.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"bd063346966813fe3b83e6970e9205079fa02157","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-07 08:46:51.000000000","message":"Patch Set 7:\n\nno_log is one thing, but other is: \nok: [primary] \u003d\u003e (item\u003d{u\u0027interface\u0027: u\u0027admin\u0027, u\u0027url\u0027: u\u0027http://169.254.169.10:35357\u0027}) \u003d\u003e {\n    \"ansible_loop_var\": \"item\", \n    \"changed\": false, \n    \"invocation\": {\n        \"module_args\": {\n            \"api_version\": \"auto\", \n            \"module_args\": {\n                \"auth\": {\n                    \"auth_url\": \"http://169.254.169.10:35357\", \n                    \"domain_name\": \"default\", \n                    \"password\": \"bMOLdXMNDYat0w6taRdhuW7tAg3ggX5zr6jfkjMN\", \n                    \"project_name\": \"admin\", \n                    \"user_domain_name\": \"default\", \n                    \"username\": \"admin\"\n                }, \n                \"description\": \"Openstack Identity Service\", \n                \"endpoint_region\": \"RegionOne\", \n                \"endpoint_type\": \"admin\", \n                \"interface\": \"admin\", \n                \"region_name\": \"RegionOne\", \n                \"service_name\": \"keystone\", \n                \"service_type\": \"identity\", \n                \"url\": \"http://169.254.169.10:35357\"\n            }, \n            \"module_extra_vars\": null, \n            \"module_name\": \"kolla_keystone_service\", \n            \"timeout\": 180\n        }\n    }, \n    \"item\": {\n        \"interface\": \"admin\", \n        \"url\": \"http://169.254.169.10:35357\"\n    }, \n    \"msg\": \"Unsupported parameters for (kolla_keystone_service) module: endpoint_type Supported parameters include: api_timeout, auth, auth_type, availability_zone, ca_cert, client_cert, client_key, cloud, description, endpoint_region, interface, region_name, service_name, service_type, timeout, url, validate_certs, wait\"\n}","accounts_in_message":[],"_revision_number":7},{"id":"f2a86602bd255922d3e82d2d80c5241b02eb5690","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-07 08:51:06.000000000","message":"Patch Set 7:\n\nSo - Ansible no longer uses shade - https://github.com/ansible/ansible/commit/89ce826a9fb53c304238923a73667ab820711338#diff-1acc83f61082dea5e335b0c691eae23e","accounts_in_message":[],"_revision_number":7},{"id":"366a2fe530dc08c2432da0773a93626ab9fb3dec","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-07 09:09:21.000000000","message":"Patch Set 7:\n\n\u003e So - Ansible no longer uses shade - https://github.com/ansible/ansible/commit/89ce826a9fb53c304238923a73667ab820711338#diff-1acc83f61082dea5e335b0c691eae23e\n\nYeah - openstacksdk now. There is a nice utility method that generates module arguments and gives you a cloud connection object. See openstack_cloud_from_module.","accounts_in_message":[],"_revision_number":7},{"id":"8330de53766b6cf98e4f01c7ad906140218c02de","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-07 09:09:48.000000000","message":"Patch Set 7:\n\n\u003e \u003e So - Ansible no longer uses shade - https://github.com/ansible/ansible/commit/89ce826a9fb53c304238923a73667ab820711338#diff-1acc83f61082dea5e335b0c691eae23e\n \u003e \n \u003e Yeah - openstacksdk now. There is a nice utility method that\n \u003e generates module arguments and gives you a cloud connection object.\n \u003e See openstack_cloud_from_module.\n\nAlternatively we switch to upstream modules like you said. Would be more work upfront though.","accounts_in_message":[],"_revision_number":7},{"id":"92bc0762de512953b740cbd4cc3c87591857813e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-07 17:18:48.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"c47110be39fb2014bdebf75a9c7e84960e42274b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-07 18:05:53.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/16/642416/8/openstack-check/tripleo-build-containers-fedora-28-master/8d7b018/ : SUCCESS in 36m 12s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/16/642416/8/openstack-check/tripleo-build-containers-fedora-28-stein/320dce8/ : SUCCESS in 43m 35s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fb80c253a10f2ac52c2fda98505865f143925771","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 18:34:52.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/16/642416/8/check/kolla-build-centos-binary/5091c0a/ : SUCCESS in 42m 58s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/16/642416/8/check/kolla-build-centos-source/4e141b8/ : SUCCESS in 48m 51s\n- kolla-ansible-centos-source http://logs.openstack.org/16/642416/8/check/kolla-ansible-centos-source/6dcbf34/ : FAILURE in 52m 00s\n- kolla-ansible-centos-binary http://logs.openstack.org/16/642416/8/check/kolla-ansible-centos-binary/d3d2fa1/ : FAILURE in 38m 51s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/16/642416/8/check/kolla-ansible-centos-source-upgrade/b2624ce/ : FAILURE in 1h 07m 30s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/16/642416/8/check/tripleo-build-containers-centos-7/b152553/ : SUCCESS in 45m 57s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/16/642416/8/check/kolla-build-oraclelinux-binary/07fed93/ : SUCCESS in 49m 20s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/16/642416/8/check/kolla-build-oraclelinux-source/63dede0/ : SUCCESS in 52m 54s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/16/642416/8/check/kolla-ansible-oraclelinux-source/3024813/ : FAILURE in 47m 13s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/16/642416/8/check/kolla-ansible-oraclelinux-binary/e6ab87e/ : FAILURE in 44m 04s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/16/642416/8/check/openstack-tox-lower-constraints/3230329/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/16/642416/8/check/openstack-tox-pep8/430b71a/ : SUCCESS in 3m 18s\n- openstack-tox-py27 http://logs.openstack.org/16/642416/8/check/openstack-tox-py27/eeb5c32/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/16/642416/8/check/openstack-tox-py36/ee21685/ : SUCCESS in 2m 58s\n- openstack-tox-py37 http://logs.openstack.org/16/642416/8/check/openstack-tox-py37/5e082b9/ : SUCCESS in 3m 20s\n- openstack-tox-docs http://logs.openstack.org/16/642416/8/check/openstack-tox-docs/c2a9a82/html/ : SUCCESS in 2m 30s\n- kolla-build-ubuntu-binary http://logs.openstack.org/16/642416/8/check/kolla-build-ubuntu-binary/1391993/ : SUCCESS in 24m 44s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/16/642416/8/check/kolla-build-ubuntu-source/67ffe98/ : SUCCESS in 34m 02s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/16/642416/8/check/kolla-ansible-ubuntu-source/5f0a838/ : FAILURE in 28m 13s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/16/642416/8/check/kolla-ansible-ubuntu-binary/eecf2bd/ : FAILURE in 24m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/16/642416/8/check/kolla-ansible-ubuntu-source-upgrade/5139aa3/ : FAILURE in 48m 24s","accounts_in_message":[],"_revision_number":8},{"id":"b30e97ab74b10aeb6f6a7954faf8f6788f396e6d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-07 20:53:19.000000000","message":"Patch Set 8:\n\nadd description to commit message please\n\nthis change has a reason, there was discussion...","accounts_in_message":[],"_revision_number":8},{"id":"f2fa8e3badb4977bc009ee677826e92d354afc11","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-10 09:15:34.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"65d5014070d2967f544ff9ef6545611facf362be","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-10 10:04:08.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/16/642416/9/openstack-check/tripleo-build-containers-fedora-28-master/f5de231/ : SUCCESS in 45m 17s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/16/642416/9/openstack-check/tripleo-build-containers-fedora-28-stein/64c7193/ : SUCCESS in 44m 06s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f9b63e64369fea20a0331c837ac998fd40ad9668","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 10:38:41.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/16/642416/9/check/kolla-build-centos-binary/98722d7/ : SUCCESS in 31m 31s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/16/642416/9/check/kolla-build-centos-source/f6d943b/ : SUCCESS in 57m 05s\n- kolla-ansible-centos-source http://logs.openstack.org/16/642416/9/check/kolla-ansible-centos-source/d686dd2/ : FAILURE in 49m 32s\n- kolla-ansible-centos-binary http://logs.openstack.org/16/642416/9/check/kolla-ansible-centos-binary/c9b62ad/ : FAILURE in 45m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/16/642416/9/check/kolla-ansible-centos-source-upgrade/8a3971f/ : FAILURE in 1h 10m 10s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/16/642416/9/check/tripleo-build-containers-centos-7/38ee0a5/ : SUCCESS in 46m 54s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/16/642416/9/check/kolla-build-oraclelinux-binary/2b747e2/ : SUCCESS in 51m 05s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/16/642416/9/check/kolla-build-oraclelinux-source/0c23663/ : SUCCESS in 46m 47s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/16/642416/9/check/kolla-ansible-oraclelinux-source/8446ce4/ : FAILURE in 46m 04s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/16/642416/9/check/kolla-ansible-oraclelinux-binary/51d2eac/ : FAILURE in 42m 18s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/16/642416/9/check/openstack-tox-lower-constraints/b0ebf52/ : SUCCESS in 4m 03s\n- openstack-tox-pep8 http://logs.openstack.org/16/642416/9/check/openstack-tox-pep8/cb30c04/ : SUCCESS in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/16/642416/9/check/openstack-tox-py27/2440824/ : SUCCESS in 3m 22s\n- openstack-tox-py36 http://logs.openstack.org/16/642416/9/check/openstack-tox-py36/4f1013b/ : SUCCESS in 2m 49s\n- openstack-tox-py37 http://logs.openstack.org/16/642416/9/check/openstack-tox-py37/6118b45/ : SUCCESS in 5m 19s\n- openstack-tox-docs http://logs.openstack.org/16/642416/9/check/openstack-tox-docs/0d46fc5/html/ : SUCCESS in 3m 43s\n- kolla-build-ubuntu-binary http://logs.openstack.org/16/642416/9/check/kolla-build-ubuntu-binary/8ee2b09/ : SUCCESS in 31m 42s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/16/642416/9/check/kolla-build-ubuntu-source/7716bc8/ : SUCCESS in 32m 43s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/16/642416/9/check/kolla-ansible-ubuntu-source/f2ee316/ : FAILURE in 28m 14s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/16/642416/9/check/kolla-ansible-ubuntu-binary/7c8289c/ : FAILURE in 40m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/16/642416/9/check/kolla-ansible-ubuntu-source-upgrade/1d69f0d/ : FAILURE in 1h 12m 57s","accounts_in_message":[],"_revision_number":9},{"id":"dde016be1712a204732c03e8a68b0766671aca92","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-11 14:24:57.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"2af53f338e64117c20fc77b303a555aa8b725a33","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-11 14:28:52.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"706a580b1923639af52e23c7c7909799455ff9ea","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-11 15:40:03.000000000","message":"Patch Set 11:\n\nSeems some problem with placement now: http://logs.openstack.org/16/642416/11/check/kolla-ansible-centos-source/91d62f6/primary/logs/kolla/nova/nova-compute.txt.gz#_2019-06-11_15_23_47_408","accounts_in_message":[],"_revision_number":11},{"id":"e53211de8d79c48d38318d6ed30f315ae6ea0309","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 15:57:25.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/16/642416/11/check/kolla-build-centos-binary/ff03417/ : SUCCESS in 42m 26s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/16/642416/11/check/kolla-build-centos-source/a910f3b/ : SUCCESS in 40m 40s\n- kolla-ansible-centos-source http://logs.openstack.org/16/642416/11/check/kolla-ansible-centos-source/91d62f6/ : FAILURE in 45m 07s\n- kolla-ansible-centos-binary http://logs.openstack.org/16/642416/11/check/kolla-ansible-centos-binary/96cce28/ : FAILURE in 44m 49s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/16/642416/11/check/kolla-ansible-centos-source-upgrade/1f5de5c/ : SUCCESS in 1h 11m 46s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/16/642416/11/check/tripleo-build-containers-centos-7/06f57d4/ : SUCCESS in 46m 11s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/16/642416/11/check/kolla-build-oraclelinux-binary/8dcc041/ : SUCCESS in 54m 12s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/16/642416/11/check/kolla-build-oraclelinux-source/438e20f/ : SUCCESS in 49m 11s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/16/642416/11/check/kolla-ansible-oraclelinux-source/843672a/ : FAILURE in 58m 40s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/16/642416/11/check/kolla-ansible-oraclelinux-binary/86d765a/ : FAILURE in 52m 59s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/16/642416/11/check/openstack-tox-lower-constraints/c713b37/ : FAILURE in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/16/642416/11/check/openstack-tox-pep8/5f69d64/ : SUCCESS in 3m 17s\n- openstack-tox-py27 http://logs.openstack.org/16/642416/11/check/openstack-tox-py27/a99eb93/ : SUCCESS in 3m 00s\n- openstack-tox-py36 http://logs.openstack.org/16/642416/11/check/openstack-tox-py36/cfddd63/ : SUCCESS in 3m 14s\n- openstack-tox-py37 http://logs.openstack.org/16/642416/11/check/openstack-tox-py37/1f75b67/ : SUCCESS in 5m 10s\n- openstack-tox-docs http://logs.openstack.org/16/642416/11/check/openstack-tox-docs/b46fab2/html/ : SUCCESS in 3m 06s\n- kolla-build-ubuntu-binary http://logs.openstack.org/16/642416/11/check/kolla-build-ubuntu-binary/21a81b0/ : SUCCESS in 24m 54s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/16/642416/11/check/kolla-build-ubuntu-source/647bd09/ : SUCCESS in 29m 46s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/16/642416/11/check/kolla-ansible-ubuntu-source/1d5f88a/ : FAILURE in 32m 28s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/16/642416/11/check/kolla-ansible-ubuntu-binary/06219c6/ : FAILURE in 33m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/16/642416/11/check/kolla-ansible-ubuntu-source-upgrade/4421347/ : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":11},{"id":"76f7c72b5f5eaf35abc0b74d2f57065add01f943","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-11 16:02:52.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/16/642416/11/openstack-check/tripleo-build-containers-fedora-28-master/4b589be/ : SUCCESS in 44m 41s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/16/642416/11/openstack-check/tripleo-build-containers-fedora-28-stein/56ac642/ : SUCCESS in 54m 25s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"31e8fcd141735df41657e74a4c41eb737656f33f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-13 07:04:48.000000000","message":"Patch Set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"bb9cb74cebf98f3519914984b05695f71cec6c9b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-13 07:18:50.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"548f7e587611074e459f4814f82faaaec61144ec","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-13 07:45:05.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/16/642416/12/openstack-check/tripleo-build-containers-fedora-28-master/ab43637/ : SUCCESS in 34m 44s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/16/642416/12/openstack-check/tripleo-build-containers-fedora-28-stein/5cb535a/ : SUCCESS in 33m 06s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"a97f52764cd02cb91c3b8808e0fa5ff2fa7c9767","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 08:23:58.000000000","message":"Patch Set 12: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"5dbea0445110ef66620daf1064375c2bc9856dc2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-13 17:21:21.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"735e1fcd062dfe212ab4f4923e980b6f23f6d6c5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-13 18:02:10.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/16/642416/12/openstack-check/tripleo-build-containers-fedora-28-master/a846868/ : SUCCESS in 30m 05s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/16/642416/12/openstack-check/tripleo-build-containers-fedora-28-stein/b873cc2/ : SUCCESS in 34m 40s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"6c778f73c0f18d4cec5060a47066163783a0aedd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 19:42:51.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/16/642416/12/check/kolla-build-centos-binary/a648a1b/ : SUCCESS in 43m 02s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/16/642416/12/check/kolla-build-centos-source/8a98799/ : SUCCESS in 40m 26s\n- kolla-ansible-centos-source http://logs.openstack.org/16/642416/12/check/kolla-ansible-centos-source/1ff462a/ : FAILURE in 51m 26s\n- kolla-ansible-centos-binary http://logs.openstack.org/16/642416/12/check/kolla-ansible-centos-binary/140b902/ : FAILURE in 47m 15s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/16/642416/12/check/kolla-ansible-centos-source-upgrade/04f7066/ : SUCCESS in 1h 20m 20s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/16/642416/12/check/tripleo-build-containers-centos-7/39d735c/ : SUCCESS in 48m 14s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/16/642416/12/check/kolla-build-oraclelinux-binary/732dbc4/ : SUCCESS in 52m 16s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/16/642416/12/check/kolla-build-oraclelinux-source/780acdd/ : SUCCESS in 48m 27s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/16/642416/12/check/kolla-ansible-oraclelinux-source/c4028d8/ : FAILURE in 1h 05m 09s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/16/642416/12/check/kolla-ansible-oraclelinux-binary/8026ee7/ : FAILURE in 57m 24s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/16/642416/12/check/openstack-tox-lower-constraints/2f691e2/ : FAILURE in 3m 13s\n- openstack-tox-pep8 http://logs.openstack.org/16/642416/12/check/openstack-tox-pep8/0cda7f8/ : SUCCESS in 3m 21s\n- openstack-tox-py27 http://logs.openstack.org/16/642416/12/check/openstack-tox-py27/8faeea8/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/16/642416/12/check/openstack-tox-py36/3febf66/ : SUCCESS in 3m 19s\n- openstack-tox-py37 http://logs.openstack.org/16/642416/12/check/openstack-tox-py37/fb71de6/ : SUCCESS in 3m 31s\n- openstack-tox-docs http://logs.openstack.org/16/642416/12/check/openstack-tox-docs/0997259/html/ : SUCCESS in 4m 05s\n- kolla-build-ubuntu-binary http://logs.openstack.org/16/642416/12/check/kolla-build-ubuntu-binary/f4de43d/ : FAILURE in 46m 14s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/16/642416/12/check/kolla-build-ubuntu-source/61ddb44/ : SUCCESS in 33m 52s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/16/642416/12/check/kolla-ansible-ubuntu-source/b60a53b/ : FAILURE in 33m 58s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/16/642416/12/check/kolla-ansible-ubuntu-binary/3c6dd3b/ : FAILURE in 1h 10m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/16/642416/12/check/kolla-ansible-ubuntu-source-upgrade/0c71f07/ : SUCCESS in 1h 02m 17s","accounts_in_message":[],"_revision_number":12},{"id":"64fe3a93ea85506dd2f7c233edd48da35ea33b5f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-17 14:25:20.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"4b7a860185fe07685d8e2043237cd2293550d583","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 15:19:41.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/16/642416/12/openstack-check/tripleo-build-containers-fedora-28-master/2145562/ : FAILURE in 6m 50s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/16/642416/12/openstack-check/tripleo-build-containers-fedora-28-stein/8d85e2f/ : SUCCESS in 52m 03s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"088dcd4d52611bb27c33161e5573ffcdc54583ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 16:00:16.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/16/642416/12/check/kolla-build-centos-binary/0dca502/ : SUCCESS in 41m 49s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/16/642416/12/check/kolla-build-centos-source/c9df36a/ : SUCCESS in 50m 15s\n- kolla-ansible-centos-source http://logs.openstack.org/16/642416/12/check/kolla-ansible-centos-source/070cdb5/ : FAILURE in 1h 00m 34s\n- kolla-ansible-centos-binary http://logs.openstack.org/16/642416/12/check/kolla-ansible-centos-binary/cc87faa/ : FAILURE in 46m 22s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/16/642416/12/check/kolla-ansible-centos-source-upgrade/bf294ac/ : SUCCESS in 1h 29m 34s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/16/642416/12/check/tripleo-build-containers-centos-7/72467bc/ : SUCCESS in 48m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/642416/12/check/openstack-tox-lower-constraints/8a9a8ba/ : SUCCESS in 4m 09s\n- openstack-tox-pep8 http://logs.openstack.org/16/642416/12/check/openstack-tox-pep8/bf1c676/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/16/642416/12/check/openstack-tox-py27/2ef40e9/ : SUCCESS in 3m 35s\n- openstack-tox-py36 http://logs.openstack.org/16/642416/12/check/openstack-tox-py36/67c625e/ : SUCCESS in 5m 08s\n- openstack-tox-py37 http://logs.openstack.org/16/642416/12/check/openstack-tox-py37/8c1e763/ : SUCCESS in 4m 14s\n- openstack-tox-docs http://logs.openstack.org/16/642416/12/check/openstack-tox-docs/87fd931/html/ : SUCCESS in 4m 22s\n- kolla-build-ubuntu-binary http://logs.openstack.org/16/642416/12/check/kolla-build-ubuntu-binary/131086a/ : SUCCESS in 32m 54s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/16/642416/12/check/kolla-build-ubuntu-source/67a679d/ : SUCCESS in 29m 02s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/16/642416/12/check/kolla-ansible-ubuntu-source/db70653/ : FAILURE in 38m 51s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/16/642416/12/check/kolla-ansible-ubuntu-binary/63a8bf5/ : FAILURE in 39m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/16/642416/12/check/kolla-ansible-ubuntu-source-upgrade/00cf592/ : SUCCESS in 1h 03m 08s","accounts_in_message":[],"_revision_number":12},{"id":"f4241c2008c0fa3d49602c2585e638f66bd96d7d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-07-23 09:48:31.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"8aa578193601103a65e53d02c74517482cd0585d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-23 09:54:55.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"976a77184fd2dd365c836153ae923e51f7b0562d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-07-23 10:01:48.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"4742831ac1ae4166b3111adc2194a41adcb4eed1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-23 10:35:54.000000000","message":"Patch Set 13:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/16/642416/13/openstack-check/tripleo-build-containers-fedora-28-master/f35869f/ : FAILURE in 6m 03s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/16/642416/13/openstack-check/tripleo-build-containers-fedora-28-stein/fdc0ba1/ : SUCCESS in 44m 42s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"bdb794a5556c81f606c555cdf1811ca38de1c151","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-23 11:10:19.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"3b004ef03c869a46fa353fef9dadd4711484cf41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 11:21:03.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/16/642416/13/check/kolla-build-centos-binary/0e5cb23/ : SUCCESS in 36m 01s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/16/642416/13/check/kolla-build-centos-source/7912e06/ : SUCCESS in 1h 00m 47s\n- kolla-ansible-centos-source http://logs.openstack.org/16/642416/13/check/kolla-ansible-centos-source/812d7ba/ : FAILURE in 57m 06s\n- kolla-ansible-centos-binary http://logs.openstack.org/16/642416/13/check/kolla-ansible-centos-binary/4589a0a/ : FAILURE in 10m 24s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/16/642416/13/check/kolla-ansible-centos-source-upgrade/443c1c8/ : SUCCESS in 1h 21m 05s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/16/642416/13/check/tripleo-build-containers-centos-7/d09b534/ : SUCCESS in 58m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/642416/13/check/openstack-tox-lower-constraints/c7b4e1c/ : SUCCESS in 3m 32s\n- openstack-tox-pep8 http://logs.openstack.org/16/642416/13/check/openstack-tox-pep8/eb5791e/ : SUCCESS in 4m 04s\n- openstack-tox-py27 http://logs.openstack.org/16/642416/13/check/openstack-tox-py27/02b244f/ : SUCCESS in 3m 16s\n- openstack-tox-py36 http://logs.openstack.org/16/642416/13/check/openstack-tox-py36/f5796ff/ : SUCCESS in 3m 35s\n- openstack-tox-py37 http://logs.openstack.org/16/642416/13/check/openstack-tox-py37/848b970/ : SUCCESS in 6m 14s\n- openstack-tox-docs http://logs.openstack.org/16/642416/13/check/openstack-tox-docs/f89f3e8/html/ : SUCCESS in 5m 43s\n- kolla-build-ubuntu-binary http://logs.openstack.org/16/642416/13/check/kolla-build-ubuntu-binary/bb47465/ : SUCCESS in 25m 17s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/16/642416/13/check/kolla-build-ubuntu-source/e5f198a/ : SUCCESS in 26m 53s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/16/642416/13/check/kolla-ansible-ubuntu-source/00e3dc8/ : FAILURE in 39m 22s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/16/642416/13/check/kolla-ansible-ubuntu-binary/3ce01b8/ : FAILURE in 34m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/16/642416/13/check/kolla-ansible-ubuntu-source-upgrade/f19a41f/ : SUCCESS in 1h 12m 43s","accounts_in_message":[],"_revision_number":13},{"id":"f6b57c0b3e79729f72307fc43ef3d9d832655c5f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-01 07:43:03.000000000","message":"Patch Set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"d1dcc0379435f9f08e6eb2b9eb05284e92ea9d8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 08:01:49.000000000","message":"Patch Set 14: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"57faf06af3b6809330d89b87c14268c9d614f5fb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 08:01:49.000000000","message":"Patch Set 14:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"882383496105ff2647390d36410480ed14ee00df","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-01 15:36:07.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"9b134e988c29942d3e6dd35c68d96bb1bc030433","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 16:15:19.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/16/642416/14/openstack-check/tripleo-build-containers-rhel-8-master/8b0a30a/ : SUCCESS in 37m 44s","accounts_in_message":[],"_revision_number":14},{"id":"8190dc2772da2f4ca2ff47c4f96bddf19f455526","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 17:09:14.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://logs.opendev.org/16/642416/14/check/kolla-build-centos-binary/5c2522f/ : SUCCESS in 45m 54s (non-voting)\n- kolla-build-centos-source https://logs.opendev.org/16/642416/14/check/kolla-build-centos-source/f84e142/ : SUCCESS in 47m 17s\n- kolla-ansible-centos-source https://logs.opendev.org/16/642416/14/check/kolla-ansible-centos-source/c0aaa00/ : FAILURE in 58m 40s\n- kolla-ansible-centos-binary https://logs.opendev.org/16/642416/14/check/kolla-ansible-centos-binary/68924ed/ : FAILURE in 48m 15s (non-voting)\n- kolla-ansible-centos-source-upgrade https://logs.opendev.org/16/642416/14/check/kolla-ansible-centos-source-upgrade/5843cc7/ : SUCCESS in 1h 26m 22s\n- tripleo-build-containers-centos-7 https://logs.opendev.org/16/642416/14/check/tripleo-build-containers-centos-7/e2be5c2/ : SUCCESS in 54m 54s\n- openstack-tox-lower-constraints https://logs.opendev.org/16/642416/14/check/openstack-tox-lower-constraints/0f39700/ : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://logs.opendev.org/16/642416/14/check/openstack-tox-pep8/a343fdd/ : SUCCESS in 4m 16s\n- openstack-tox-py27 https://logs.opendev.org/16/642416/14/check/openstack-tox-py27/da59c93/ : SUCCESS in 3m 20s\n- openstack-tox-py36 https://logs.opendev.org/16/642416/14/check/openstack-tox-py36/128c50e/ : SUCCESS in 3m 44s\n- openstack-tox-py37 https://logs.opendev.org/16/642416/14/check/openstack-tox-py37/39255db/ : SUCCESS in 6m 03s\n- openstack-tox-docs https://logs.opendev.org/16/642416/14/check/openstack-tox-docs/da6b62d/html/ : SUCCESS in 6m 21s\n- kolla-build-ubuntu-binary https://logs.opendev.org/16/642416/14/check/kolla-build-ubuntu-binary/59999c3/ : SUCCESS in 34m 40s (non-voting)\n- kolla-build-ubuntu-source https://logs.opendev.org/16/642416/14/check/kolla-build-ubuntu-source/4162651/ : SUCCESS in 43m 19s\n- kolla-ansible-ubuntu-source https://logs.opendev.org/16/642416/14/check/kolla-ansible-ubuntu-source/812d749/ : FAILURE in 47m 25s\n- kolla-ansible-ubuntu-binary https://logs.opendev.org/16/642416/14/check/kolla-ansible-ubuntu-binary/6a1644f/ : FAILURE in 46m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://logs.opendev.org/16/642416/14/check/kolla-ansible-ubuntu-source-upgrade/c3413f0/ : SUCCESS in 1h 13m 26s","accounts_in_message":[],"_revision_number":14},{"id":"57e77547b6afdb6b941358b917f08e76bd1f2477","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-02 06:19:03.000000000","message":"Patch Set 15: Commit message was updated.","accounts_in_message":[],"_revision_number":15},{"id":"85d2067b244753103ebc13a95e6bb757d33ff59e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-02 06:20:56.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"b6860a2e7dd62dafc13d331d8ca1e86bd619393b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-02 06:35:48.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"e35ee71bd78daeb776fce4dc88c6fd4a3c945df8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-02 06:38:11.000000000","message":"Patch Set 18: Commit message was updated.","accounts_in_message":[],"_revision_number":18},{"id":"3737aacdf2c1491d9d15939548b6c8f3bfd0263e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 07:17:35.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/16/642416/18/openstack-check/tripleo-build-containers-rhel-8-master/6427c4b/ : SUCCESS in 38m 38s","accounts_in_message":[],"_revision_number":18},{"id":"f5d03756dd27e1527e6122319832ece7ca033d01","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-02 07:53:02.000000000","message":"Patch Set 19: Patch Set 18 was rebased","accounts_in_message":[],"_revision_number":19},{"id":"10b402c14a7f22f9525982e03bae1a27fc55935b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-02 07:53:16.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"787b31222d0b92ea697f31d2b5c7a413b540c58a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 08:30:02.000000000","message":"Patch Set 20:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/16/642416/20/openstack-check/tripleo-build-containers-rhel-8-master/03ef225/ : SUCCESS in 35m 56s","accounts_in_message":[],"_revision_number":20},{"id":"af000af17c0ff5cbfd0e4cdf618f314fecb88449","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-02 09:11:14.000000000","message":"Patch Set 20: Review-Priority+1","accounts_in_message":[],"_revision_number":20},{"id":"f51f802c0289b7438c398be82812267b3a5425f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 09:26:46.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://logs.opendev.org/16/642416/20/check/kolla-build-centos-binary/a58e481/ : SUCCESS in 45m 26s (non-voting)\n- kolla-build-centos-source https://logs.opendev.org/16/642416/20/check/kolla-build-centos-source/ffd9fd3/ : SUCCESS in 59m 30s\n- kolla-ansible-centos-source https://logs.opendev.org/16/642416/20/check/kolla-ansible-centos-source/4b87705/ : SUCCESS in 1h 22m 11s\n- kolla-ansible-centos-binary https://logs.opendev.org/16/642416/20/check/kolla-ansible-centos-binary/55a5122/ : SUCCESS in 1h 21m 01s (non-voting)\n- kolla-ansible-centos-source-upgrade https://logs.opendev.org/16/642416/20/check/kolla-ansible-centos-source-upgrade/afbb737/ : SUCCESS in 1h 26m 58s\n- tripleo-build-containers-centos-7 https://logs.opendev.org/16/642416/20/check/tripleo-build-containers-centos-7/22e5f37/ : SUCCESS in 48m 49s\n- openstack-tox-lower-constraints https://logs.opendev.org/16/642416/20/check/openstack-tox-lower-constraints/613c871/ : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://logs.opendev.org/16/642416/20/check/openstack-tox-pep8/d5a0582/ : SUCCESS in 5m 16s\n- openstack-tox-py27 https://logs.opendev.org/16/642416/20/check/openstack-tox-py27/87be7c7/ : SUCCESS in 4m 06s\n- openstack-tox-py36 https://logs.opendev.org/16/642416/20/check/openstack-tox-py36/bd46060/ : SUCCESS in 3m 22s\n- openstack-tox-py37 https://logs.opendev.org/16/642416/20/check/openstack-tox-py37/60b4458/ : SUCCESS in 4m 26s\n- openstack-tox-docs https://logs.opendev.org/16/642416/20/check/openstack-tox-docs/199338d/html/ : SUCCESS in 4m 08s\n- kolla-build-ubuntu-binary https://logs.opendev.org/16/642416/20/check/kolla-build-ubuntu-binary/20591b5/ : SUCCESS in 29m 01s (non-voting)\n- kolla-build-ubuntu-source https://logs.opendev.org/16/642416/20/check/kolla-build-ubuntu-source/d14302e/ : SUCCESS in 28m 34s\n- kolla-ansible-ubuntu-source https://logs.opendev.org/16/642416/20/check/kolla-ansible-ubuntu-source/8e43c71/ : SUCCESS in 57m 14s\n- kolla-ansible-ubuntu-binary https://logs.opendev.org/16/642416/20/check/kolla-ansible-ubuntu-binary/02865a2/ : SUCCESS in 49m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://logs.opendev.org/16/642416/20/check/kolla-ansible-ubuntu-source-upgrade/b2e9db7/ : SUCCESS in 1h 04m 11s","accounts_in_message":[],"_revision_number":20},{"id":"d48aaf01056d74c90c7fbeb8587940eac703ab5a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-02 09:26:57.000000000","message":"Patch Set 20: Code-Review+2","accounts_in_message":[],"_revision_number":20},{"id":"2fe12a87e999bc95ef07a284908a080de3a9f7c2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-02 09:28:25.000000000","message":"Patch Set 21: Published edit on patch set 20.","accounts_in_message":[],"_revision_number":21},{"id":"d7fad98f4bb5f49e8944b7ae1acf0fa916eef54e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-02 09:30:03.000000000","message":"Patch Set 22: Published edit on patch set 21.","accounts_in_message":[],"_revision_number":22},{"id":"6f1079d6292236ad7e88323d0972afcd77a800cc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-02 09:38:18.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"651c6ea8873abc16dfda4329d006c8c0d4b2c59d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-02 09:39:04.000000000","message":"Patch Set 23: Published edit on patch set 22.","accounts_in_message":[],"_revision_number":23},{"id":"a05d0c8955fd8e49a5664ec5970ae7bab53cfe0c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-02 09:39:55.000000000","message":"Patch Set 23: Code-Review+2","accounts_in_message":[],"_revision_number":23},{"id":"360c8ba3d02667220f6e8cf939040e5bfc145899","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-02 09:41:33.000000000","message":"Patch Set 23:\n\n(2 comments)","accounts_in_message":[],"_revision_number":23},{"id":"14a429f21856ad6701e0be69b16dfe0e3ecbd027","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-02 09:45:50.000000000","message":"Patch Set 24: Commit message was updated.","accounts_in_message":[],"_revision_number":24},{"id":"9c08d4926e633b731760c39292b89c297cefb14d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-02 09:49:23.000000000","message":"Patch Set 24: Code-Review+2","accounts_in_message":[],"_revision_number":24},{"id":"d6385eeacc7cad7e9373992527237d62ad3d7ab1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-02 09:50:59.000000000","message":"Patch Set 24: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"5503970562b126617e2dec235ccfbb254c0012fd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-02 09:55:01.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"17de0994ef929cb4b56bf5dd0f8e4be0c12db5bb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-02 09:58:19.000000000","message":"Patch Set 24:\n\n\u003e (1 comment)\n\nIt\u0027s required, when your other module arg contains \"*pass*\" :-)\nI\u0027ll add that in the commit message after this Zuul run.","accounts_in_message":[],"_revision_number":24},{"id":"76def9f2af3e8f87e70242660ab5f2b075ff7828","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 10:30:24.000000000","message":"Patch Set 24:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/16/642416/24/openstack-check/tripleo-build-containers-rhel-8-master/b95b73d/ : SUCCESS in 43m 44s","accounts_in_message":[],"_revision_number":24},{"id":"77af6948f635fa7f12bb57e92ba9da4bc854ea2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 10:47:39.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://logs.opendev.org/16/642416/24/check/kolla-build-centos-binary/7bbc57f/ : FAILURE in 53m 10s (non-voting)\n- kolla-build-centos-source https://logs.opendev.org/16/642416/24/check/kolla-build-centos-source/0b541e3/ : FAILURE in 52m 55s\n- kolla-ansible-centos-source https://logs.opendev.org/16/642416/24/check/kolla-ansible-centos-source/c78f1fb/ : FAILURE in 35m 09s\n- kolla-ansible-centos-binary https://logs.opendev.org/16/642416/24/check/kolla-ansible-centos-binary/bcd9834/ : FAILURE in 29m 47s (non-voting)\n- kolla-ansible-centos-source-upgrade https://logs.opendev.org/16/642416/24/check/kolla-ansible-centos-source-upgrade/464c1a9/ : FAILURE in 32m 11s\n- tripleo-build-containers-centos-7 https://logs.opendev.org/16/642416/24/check/tripleo-build-containers-centos-7/e3f54c3/ : SUCCESS in 54m 47s\n- openstack-tox-lower-constraints https://logs.opendev.org/16/642416/24/check/openstack-tox-lower-constraints/e0ab510/ : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://logs.opendev.org/16/642416/24/check/openstack-tox-pep8/0a076ce/ : SUCCESS in 5m 11s\n- openstack-tox-py27 https://logs.opendev.org/16/642416/24/check/openstack-tox-py27/7b910dd/ : SUCCESS in 3m 35s\n- openstack-tox-py36 https://logs.opendev.org/16/642416/24/check/openstack-tox-py36/08cbf1c/ : SUCCESS in 4m 06s\n- openstack-tox-py37 https://logs.opendev.org/16/642416/24/check/openstack-tox-py37/3ae3a1b/ : SUCCESS in 3m 44s\n- openstack-tox-docs https://logs.opendev.org/16/642416/24/check/openstack-tox-docs/119127e/html/ : SUCCESS in 4m 35s\n- kolla-build-ubuntu-binary https://logs.opendev.org/16/642416/24/check/kolla-build-ubuntu-binary/1411632/ : FAILURE in 32m 48s (non-voting)\n- kolla-build-ubuntu-source https://logs.opendev.org/16/642416/24/check/kolla-build-ubuntu-source/4b7903d/ : FAILURE in 30m 06s\n- kolla-ansible-ubuntu-source https://logs.opendev.org/16/642416/24/check/kolla-ansible-ubuntu-source/0ad7408/ : FAILURE in 19m 40s\n- kolla-ansible-ubuntu-binary https://logs.opendev.org/16/642416/24/check/kolla-ansible-ubuntu-binary/031a4db/ : FAILURE in 17m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://logs.opendev.org/16/642416/24/check/kolla-ansible-ubuntu-source-upgrade/3c55bb3/ : FAILURE in 18m 39s","accounts_in_message":[],"_revision_number":24},{"id":"8dd68d809cf6eff340f102b1cb0433b95402b162","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-02 11:46:29.000000000","message":"Patch Set 25: Published edit on patch set 24.","accounts_in_message":[],"_revision_number":25},{"id":"2e5be42278fdb3b1889fa76efa221a9b30dc71f8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 12:40:13.000000000","message":"Patch Set 25:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/16/642416/25/openstack-check/tripleo-build-containers-rhel-8-master/473c563/ : SUCCESS in 52m 46s","accounts_in_message":[],"_revision_number":25},{"id":"fe0ee35e8fbf8b125ba67d074744ddcc8272eb04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 12:51:13.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://logs.opendev.org/16/642416/25/check/kolla-build-centos-binary/87ff981/ : FAILURE in 32m 12s (non-voting)\n- kolla-build-centos-source https://logs.opendev.org/16/642416/25/check/kolla-build-centos-source/1a1af19/ : FAILURE in 56m 13s\n- kolla-ansible-centos-source https://logs.opendev.org/16/642416/25/check/kolla-ansible-centos-source/5853b5f/ : FAILURE in 39m 40s\n- kolla-ansible-centos-binary https://logs.opendev.org/16/642416/25/check/kolla-ansible-centos-binary/3a63eb5/ : FAILURE in 36m 22s (non-voting)\n- kolla-ansible-centos-source-upgrade https://logs.opendev.org/16/642416/25/check/kolla-ansible-centos-source-upgrade/0502810/ : FAILURE in 39m 49s\n- tripleo-build-containers-centos-7 https://logs.opendev.org/16/642416/25/check/tripleo-build-containers-centos-7/ae81f59/ : SUCCESS in 48m 01s\n- openstack-tox-lower-constraints https://logs.opendev.org/16/642416/25/check/openstack-tox-lower-constraints/bac7f07/ : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://logs.opendev.org/16/642416/25/check/openstack-tox-pep8/c40bfe3/ : SUCCESS in 5m 22s\n- openstack-tox-py27 https://logs.opendev.org/16/642416/25/check/openstack-tox-py27/b92803a/ : SUCCESS in 3m 42s\n- openstack-tox-py36 https://logs.opendev.org/16/642416/25/check/openstack-tox-py36/6159604/ : SUCCESS in 3m 32s\n- openstack-tox-py37 https://logs.opendev.org/16/642416/25/check/openstack-tox-py37/7d98c9f/ : SUCCESS in 6m 01s\n- openstack-tox-docs https://logs.opendev.org/16/642416/25/check/openstack-tox-docs/1b88fee/html/ : SUCCESS in 4m 18s\n- kolla-build-ubuntu-binary https://logs.opendev.org/16/642416/25/check/kolla-build-ubuntu-binary/428aa5c/ : FAILURE in 25m 33s (non-voting)\n- kolla-build-ubuntu-source https://logs.opendev.org/16/642416/25/check/kolla-build-ubuntu-source/eb2ae7b/ : FAILURE in 28m 43s\n- kolla-ansible-ubuntu-source https://logs.opendev.org/16/642416/25/check/kolla-ansible-ubuntu-source/bf7e6f1/ : FAILURE in 18m 13s\n- kolla-ansible-ubuntu-binary https://logs.opendev.org/16/642416/25/check/kolla-ansible-ubuntu-binary/6518563/ : FAILURE in 17m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://logs.opendev.org/16/642416/25/check/kolla-ansible-ubuntu-source-upgrade/c14f6c2/ : FAILURE in 20m 38s","accounts_in_message":[],"_revision_number":25},{"id":"14e4dec2bf1d17e4b0f5ef93a5223422c6c8bf61","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-02 13:01:05.000000000","message":"Patch Set 26: Published edit on patch set 25.","accounts_in_message":[],"_revision_number":26},{"id":"1720814ceaf44a3c0d46ab867171a5fc08017da6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 13:42:32.000000000","message":"Patch Set 26:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/16/642416/26/openstack-check/tripleo-build-containers-rhel-8-master/d26b212/ : SUCCESS in 40m 36s","accounts_in_message":[],"_revision_number":26},{"id":"5324a431abfeaf2c2d3f43d26e4948e3d9ff2c1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 14:36:43.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://logs.opendev.org/16/642416/26/check/kolla-build-centos-binary/94712ff/ : FAILURE in 29m 52s (non-voting)\n- kolla-build-centos-source https://logs.opendev.org/16/642416/26/check/kolla-build-centos-source/43a9b73/ : SUCCESS in 51m 28s\n- kolla-ansible-centos-source https://logs.opendev.org/16/642416/26/check/kolla-ansible-centos-source/b20e689/ : SUCCESS in 1h 18m 35s\n- kolla-ansible-centos-binary https://logs.opendev.org/16/642416/26/check/kolla-ansible-centos-binary/4a184ce/ : FAILURE in 34m 13s (non-voting)\n- kolla-ansible-centos-source-upgrade https://logs.opendev.org/16/642416/26/check/kolla-ansible-centos-source-upgrade/5cc0607/ : SUCCESS in 1h 27m 15s\n- tripleo-build-containers-centos-7 https://logs.opendev.org/16/642416/26/check/tripleo-build-containers-centos-7/be2d897/ : SUCCESS in 47m 33s\n- openstack-tox-lower-constraints https://logs.opendev.org/16/642416/26/check/openstack-tox-lower-constraints/c075f81/ : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://logs.opendev.org/16/642416/26/check/openstack-tox-pep8/a5f351e/ : SUCCESS in 4m 38s\n- openstack-tox-py27 https://logs.opendev.org/16/642416/26/check/openstack-tox-py27/be386c0/ : SUCCESS in 3m 05s\n- openstack-tox-py36 https://logs.opendev.org/16/642416/26/check/openstack-tox-py36/4013ef4/ : SUCCESS in 3m 47s\n- openstack-tox-py37 https://logs.opendev.org/16/642416/26/check/openstack-tox-py37/fd3a41e/ : SUCCESS in 4m 48s\n- openstack-tox-docs https://logs.opendev.org/16/642416/26/check/openstack-tox-docs/e12d975/html/ : SUCCESS in 3m 13s\n- kolla-build-ubuntu-binary https://logs.opendev.org/16/642416/26/check/kolla-build-ubuntu-binary/9d4a4a6/ : FAILURE in 22m 15s (non-voting)\n- kolla-build-ubuntu-source https://logs.opendev.org/16/642416/26/check/kolla-build-ubuntu-source/2270d38/ : SUCCESS in 25m 06s\n- kolla-ansible-ubuntu-source https://logs.opendev.org/16/642416/26/check/kolla-ansible-ubuntu-source/3f3fcb2/ : SUCCESS in 42m 13s\n- kolla-ansible-ubuntu-binary https://logs.opendev.org/16/642416/26/check/kolla-ansible-ubuntu-binary/3f40cbc/ : FAILURE in 16m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://logs.opendev.org/16/642416/26/check/kolla-ansible-ubuntu-source-upgrade/d62b946/ : SUCCESS in 49m 42s","accounts_in_message":[],"_revision_number":26},{"id":"747e29d3c4bb2a36a476a40d62247f9828b46268","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-04 09:27:01.000000000","message":"Patch Set 26:\n\nseems it needs an exception for the binary case still","accounts_in_message":[],"_revision_number":26},{"id":"040312d4951aac48ca4867238ede413686d923c4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-05 09:14:33.000000000","message":"Patch Set 26: Code-Review+1\n\nonce pass zuul will get +2","accounts_in_message":[],"_revision_number":26},{"id":"dc7723df34967bfe6cfa7e5d19dedf076f76f8af","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-08-05 14:26:20.000000000","message":"Patch Set 26: Code-Review+1","accounts_in_message":[],"_revision_number":26},{"id":"698239125b064f3f357d21389013d4e62ad5b5eb","author":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"date":"2019-08-06 06:00:38.000000000","message":"Patch Set 26: Code-Review+1","accounts_in_message":[],"_revision_number":26},{"id":"a930c1bc6b4b973eab0b33c7a8cf696137b2e790","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-08 07:10:13.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"0a6ef6a3b0ba4346dd35b9fdd458b85f18720ed0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-08 07:11:11.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.","accounts_in_message":[],"_revision_number":28},{"id":"245d771736d936a8e72238f4d69b7c8b517468a9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 07:57:27.000000000","message":"Patch Set 28:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/16/642416/28/openstack-check/tripleo-build-containers-rhel-8-master/24d28c5/ : FAILURE in 45m 21s","accounts_in_message":[],"_revision_number":28},{"id":"d69a8208e7858955269e4b88155614da4233b46a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 09:05:16.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://logs.opendev.org/16/642416/28/check/kolla-build-centos-binary/87702fe/ : FAILURE in 44m 39s (non-voting)\n- kolla-build-centos-source https://logs.opendev.org/16/642416/28/check/kolla-build-centos-source/929af66/ : FAILURE in 54m 39s\n- kolla-ansible-centos-source https://logs.opendev.org/16/642416/28/check/kolla-ansible-centos-source/a8ebd3e/ : FAILURE in 30m 32s\n- kolla-ansible-centos-binary https://logs.opendev.org/16/642416/28/check/kolla-ansible-centos-binary/86a7be2/ : FAILURE in 1h 03m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade https://logs.opendev.org/16/642416/28/check/kolla-ansible-centos-source-upgrade/ea6e943/ : FAILURE in 32m 32s\n- tripleo-build-containers-centos-7 https://logs.opendev.org/16/642416/28/check/tripleo-build-containers-centos-7/d9dd475/ : SUCCESS in 1h 52m 08s\n- openstack-tox-lower-constraints https://logs.opendev.org/16/642416/28/check/openstack-tox-lower-constraints/8890541/ : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://logs.opendev.org/16/642416/28/check/openstack-tox-pep8/f827eb9/ : SUCCESS in 6m 06s\n- openstack-tox-py27 https://logs.opendev.org/16/642416/28/check/openstack-tox-py27/1e79ac5/ : SUCCESS in 3m 55s\n- openstack-tox-py36 https://logs.opendev.org/16/642416/28/check/openstack-tox-py36/46ecee5/ : SUCCESS in 5m 54s\n- openstack-tox-py37 https://logs.opendev.org/16/642416/28/check/openstack-tox-py37/92a6da0/ : SUCCESS in 4m 46s\n- openstack-tox-docs https://logs.opendev.org/16/642416/28/check/openstack-tox-docs/026fbea/html/ : SUCCESS in 3m 30s\n- kolla-build-ubuntu-binary https://logs.opendev.org/16/642416/28/check/kolla-build-ubuntu-binary/2764057/ : FAILURE in 27m 00s (non-voting)\n- kolla-build-ubuntu-source https://logs.opendev.org/16/642416/28/check/kolla-build-ubuntu-source/1096ada/ : FAILURE in 31m 35s\n- kolla-ansible-ubuntu-source https://logs.opendev.org/16/642416/28/check/kolla-ansible-ubuntu-source/568b17c/ : FAILURE in 18m 26s\n- kolla-ansible-ubuntu-binary https://logs.opendev.org/16/642416/28/check/kolla-ansible-ubuntu-binary/8369c80/ : FAILURE in 21m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://logs.opendev.org/16/642416/28/check/kolla-ansible-ubuntu-source-upgrade/bb74d24/ : FAILURE in 23m 06s","accounts_in_message":[],"_revision_number":28},{"id":"cd30f32f7d60b2cc66042b1f20f191940356462b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-08 09:34:58.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"dda0d17e9d046ad2a6a2d68748125d919b1c4d98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 10:34:26.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://logs.opendev.org/16/642416/29/check/kolla-build-centos-binary/a7184df/ : FAILURE in 44m 53s (non-voting)\n- kolla-build-centos-source https://logs.opendev.org/16/642416/29/check/kolla-build-centos-source/08ae45f/ : FAILURE in 52m 46s\n- kolla-ansible-centos-source https://logs.opendev.org/16/642416/29/check/kolla-ansible-centos-source/bb7c1ed/ : FAILURE in 34m 30s\n- kolla-ansible-centos-binary https://logs.opendev.org/16/642416/29/check/kolla-ansible-centos-binary/630747d/ : FAILURE in 29m 42s (non-voting)\n- kolla-ansible-centos-source-upgrade https://logs.opendev.org/16/642416/29/check/kolla-ansible-centos-source-upgrade/1093b31/ : FAILURE in 33m 32s\n- tripleo-build-containers-centos-7 https://logs.opendev.org/16/642416/29/check/tripleo-build-containers-centos-7/287f4b3/ : SUCCESS in 53m 05s\n- openstack-tox-lower-constraints https://logs.opendev.org/16/642416/29/check/openstack-tox-lower-constraints/9501a8f/ : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://logs.opendev.org/16/642416/29/check/openstack-tox-pep8/4643349/ : SUCCESS in 5m 00s\n- openstack-tox-py27 https://logs.opendev.org/16/642416/29/check/openstack-tox-py27/9cfc9a1/ : SUCCESS in 5m 54s\n- openstack-tox-py36 https://logs.opendev.org/16/642416/29/check/openstack-tox-py36/7f23a3d/ : SUCCESS in 4m 58s\n- openstack-tox-py37 https://logs.opendev.org/16/642416/29/check/openstack-tox-py37/de96356/ : SUCCESS in 4m 11s\n- openstack-tox-docs https://logs.opendev.org/16/642416/29/check/openstack-tox-docs/46ff94a/html/ : SUCCESS in 5m 27s\n- kolla-build-ubuntu-binary https://logs.opendev.org/16/642416/29/check/kolla-build-ubuntu-binary/52f73ff/ : FAILURE in 29m 07s (non-voting)\n- kolla-build-ubuntu-source https://logs.opendev.org/16/642416/29/check/kolla-build-ubuntu-source/6b24b98/ : FAILURE in 35m 17s\n- kolla-ansible-ubuntu-source https://logs.opendev.org/16/642416/29/check/kolla-ansible-ubuntu-source/d75be82/ : FAILURE in 30m 36s\n- kolla-ansible-ubuntu-binary https://logs.opendev.org/16/642416/29/check/kolla-ansible-ubuntu-binary/f42920d/ : FAILURE in 24m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://logs.opendev.org/16/642416/29/check/kolla-ansible-ubuntu-source-upgrade/cbbfca3/ : FAILURE in 20m 03s","accounts_in_message":[],"_revision_number":29},{"id":"f8704e8b9be6b67c3f1df3b3e1c01382cb3def67","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-08 11:22:11.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"3d5c01182ec4a76ef65a82e81750b60b72546add","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 12:02:25.000000000","message":"Patch Set 30:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/16/642416/30/openstack-check/tripleo-build-containers-rhel-8-master/18b7866/ : FAILURE in 37m 19s","accounts_in_message":[],"_revision_number":30},{"id":"ae2467af6aebe23acddaa00762b9981709ec211c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 12:44:37.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://logs.opendev.org/16/642416/30/check/kolla-build-centos-binary/9fa3f0d/ : SUCCESS in 45m 52s (non-voting)\n- kolla-build-centos-source https://logs.opendev.org/16/642416/30/check/kolla-build-centos-source/249ed92/ : SUCCESS in 50m 28s\n- kolla-ansible-centos-source https://logs.opendev.org/16/642416/30/check/kolla-ansible-centos-source/9da0f87/ : SUCCESS in 1h 17m 42s\n- kolla-ansible-centos-binary https://logs.opendev.org/16/642416/30/check/kolla-ansible-centos-binary/7d767f8/ : SUCCESS in 1h 16m 59s (non-voting)\n- kolla-ansible-centos-source-upgrade https://logs.opendev.org/16/642416/30/check/kolla-ansible-centos-source-upgrade/5f0eeb5/ : SUCCESS in 1h 10m 24s\n- tripleo-build-containers-centos-7 https://logs.opendev.org/16/642416/30/check/tripleo-build-containers-centos-7/4e9b112/ : SUCCESS in 56m 39s\n- openstack-tox-lower-constraints https://logs.opendev.org/16/642416/30/check/openstack-tox-lower-constraints/6ddcdd4/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://logs.opendev.org/16/642416/30/check/openstack-tox-pep8/d21c485/ : SUCCESS in 5m 41s\n- openstack-tox-py27 https://logs.opendev.org/16/642416/30/check/openstack-tox-py27/30f8c08/ : SUCCESS in 4m 28s\n- openstack-tox-py36 https://logs.opendev.org/16/642416/30/check/openstack-tox-py36/3b7213d/ : SUCCESS in 4m 55s\n- openstack-tox-py37 https://logs.opendev.org/16/642416/30/check/openstack-tox-py37/40a9bff/ : SUCCESS in 4m 17s\n- openstack-tox-docs https://logs.opendev.org/16/642416/30/check/openstack-tox-docs/640ffcf/html/ : SUCCESS in 3m 31s\n- kolla-build-ubuntu-binary https://logs.opendev.org/16/642416/30/check/kolla-build-ubuntu-binary/262f554/ : SUCCESS in 25m 12s (non-voting)\n- kolla-build-ubuntu-source https://logs.opendev.org/16/642416/30/check/kolla-build-ubuntu-source/93ae3f7/ : SUCCESS in 32m 41s\n- kolla-ansible-ubuntu-source https://logs.opendev.org/16/642416/30/check/kolla-ansible-ubuntu-source/848023b/ : SUCCESS in 56m 54s\n- kolla-ansible-ubuntu-binary https://logs.opendev.org/16/642416/30/check/kolla-ansible-ubuntu-binary/622a081/ : SUCCESS in 51m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://logs.opendev.org/16/642416/30/check/kolla-ansible-ubuntu-source-upgrade/c061f88/ : SUCCESS in 59m 14s","accounts_in_message":[],"_revision_number":30},{"id":"9767582f8d45f5c87784ff4d72033dac6f8bba38","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-08 12:55:43.000000000","message":"Patch Set 30: Review-Priority+1","accounts_in_message":[],"_revision_number":30},{"id":"5ebc6b6d56ce12dade37601d06d668211b8f4bd6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-08 12:58:04.000000000","message":"Patch Set 30: Code-Review+2","accounts_in_message":[],"_revision_number":30},{"id":"37cf5406674dfe5d25c5833962279c2ce2b1a122","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-08 13:26:07.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"68b671ef88d345cfe6a577732483eefadc78f1c5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-08 13:56:15.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"53a174080bdccc8379788b4e528ed37fa38f0f20","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-08 13:56:48.000000000","message":"Patch Set 31: Published edit on patch set 30.","accounts_in_message":[],"_revision_number":31},{"id":"4093cfac1df5ea9d559c3b5d4c5fe07a0b27d033","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-08 13:58:47.000000000","message":"Patch Set 31: Code-Review+2","accounts_in_message":[],"_revision_number":31},{"id":"7baf840f91a282b75699fb4177b7245568cd4e80","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-08 14:02:32.000000000","message":"Patch Set 31: Workflow+1\n\nit had another +2 already","accounts_in_message":[],"_revision_number":31},{"id":"08044ae64eb73e480d07858a55ed3ffef87f267d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 14:37:27.000000000","message":"Patch Set 31:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/16/642416/31/openstack-check/tripleo-build-containers-rhel-8-master/c24733a/ : FAILURE in 39m 31s","accounts_in_message":[],"_revision_number":31},{"id":"7c07584c6344da3531f15c5da7b80ade3bfd6f48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 15:29:12.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://logs.opendev.org/16/642416/31/check/kolla-build-centos-binary/6eaa8fd/ : SUCCESS in 46m 30s (non-voting)\n- kolla-build-centos-source https://logs.opendev.org/16/642416/31/check/kolla-build-centos-source/c300c6e/ : SUCCESS in 43m 09s\n- kolla-ansible-centos-source finger://ze11.openstack.org/c6b623eaa5214676b31f69497ec8c3c4 : RETRY_LIMIT in 1m 00s\n- kolla-ansible-centos-binary https://logs.opendev.org/16/642416/31/check/kolla-ansible-centos-binary/b49e330/ : SUCCESS in 1h 13m 24s (non-voting)\n- kolla-ansible-centos-source-upgrade https://logs.opendev.org/16/642416/31/check/kolla-ansible-centos-source-upgrade/be8bcb4/ : SUCCESS in 1h 23m 53s\n- tripleo-build-containers-centos-7 https://logs.opendev.org/16/642416/31/check/tripleo-build-containers-centos-7/e4e0d9c/ : SUCCESS in 56m 33s\n- openstack-tox-lower-constraints https://logs.opendev.org/16/642416/31/check/openstack-tox-lower-constraints/b28744a/ : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://logs.opendev.org/16/642416/31/check/openstack-tox-pep8/5114817/ : SUCCESS in 5m 50s\n- openstack-tox-py27 https://logs.opendev.org/16/642416/31/check/openstack-tox-py27/0ca63e3/ : SUCCESS in 4m 21s\n- openstack-tox-py36 https://logs.opendev.org/16/642416/31/check/openstack-tox-py36/7189f8e/ : SUCCESS in 4m 49s\n- openstack-tox-py37 https://logs.opendev.org/16/642416/31/check/openstack-tox-py37/5d34815/ : SUCCESS in 4m 40s\n- openstack-tox-docs https://logs.opendev.org/16/642416/31/check/openstack-tox-docs/4519bac/html/ : SUCCESS in 3m 47s\n- kolla-build-ubuntu-binary https://logs.opendev.org/16/642416/31/check/kolla-build-ubuntu-binary/49cbcaf/ : SUCCESS in 30m 07s (non-voting)\n- kolla-build-ubuntu-source https://logs.opendev.org/16/642416/31/check/kolla-build-ubuntu-source/db50f73/ : SUCCESS in 34m 02s\n- kolla-ansible-ubuntu-source https://logs.opendev.org/16/642416/31/check/kolla-ansible-ubuntu-source/8caf072/ : SUCCESS in 51m 39s\n- kolla-ansible-ubuntu-binary https://logs.opendev.org/16/642416/31/check/kolla-ansible-ubuntu-binary/a334410/ : SUCCESS in 56m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://logs.opendev.org/16/642416/31/check/kolla-ansible-ubuntu-source-upgrade/4a141ed/ : SUCCESS in 1h 03m 10s","accounts_in_message":[],"_revision_number":31},{"id":"2220f8c748b17ea23262ab602d78a5573b659a77","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-08 15:47:49.000000000","message":"Patch Set 31:\n\nrecheck\n\nZuul finger","accounts_in_message":[],"_revision_number":31},{"id":"9820d35a4e001a84c5553359630429ca8caee6fb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 16:29:45.000000000","message":"Patch Set 31:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/16/642416/31/openstack-check/tripleo-build-containers-rhel-8-master/4db3243/ : FAILURE in 40m 56s","accounts_in_message":[],"_revision_number":31},{"id":"8ec7181c7ef4eb22e977089080b676990e4be23c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 17:15:49.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://logs.opendev.org/16/642416/31/check/kolla-build-centos-binary/17de29f/ : SUCCESS in 46m 43s (non-voting)\n- kolla-build-centos-source https://logs.opendev.org/16/642416/31/check/kolla-build-centos-source/d022bde/ : SUCCESS in 39m 36s\n- kolla-ansible-centos-source https://logs.opendev.org/16/642416/31/check/kolla-ansible-centos-source/112ed02/ : SUCCESS in 1h 12m 38s\n- kolla-ansible-centos-binary https://logs.opendev.org/16/642416/31/check/kolla-ansible-centos-binary/7894a2f/ : SUCCESS in 1h 18m 24s (non-voting)\n- kolla-ansible-centos-source-upgrade https://logs.opendev.org/16/642416/31/check/kolla-ansible-centos-source-upgrade/74dd9a2/ : SUCCESS in 1h 22m 00s\n- tripleo-build-containers-centos-7 https://logs.opendev.org/16/642416/31/check/tripleo-build-containers-centos-7/45f46bd/ : SUCCESS in 58m 38s\n- openstack-tox-lower-constraints https://logs.opendev.org/16/642416/31/check/openstack-tox-lower-constraints/6cad139/ : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://logs.opendev.org/16/642416/31/check/openstack-tox-pep8/6e9b152/ : SUCCESS in 3m 56s\n- openstack-tox-py27 https://logs.opendev.org/16/642416/31/check/openstack-tox-py27/1600239/ : SUCCESS in 2m 50s\n- openstack-tox-py36 https://logs.opendev.org/16/642416/31/check/openstack-tox-py36/b6c8964/ : SUCCESS in 3m 37s\n- openstack-tox-py37 https://logs.opendev.org/16/642416/31/check/openstack-tox-py37/b3cde0e/ : SUCCESS in 5m 00s\n- openstack-tox-docs https://logs.opendev.org/16/642416/31/check/openstack-tox-docs/cdea6db/html/ : SUCCESS in 3m 39s\n- kolla-build-ubuntu-binary https://logs.opendev.org/16/642416/31/check/kolla-build-ubuntu-binary/a4ceedc/ : SUCCESS in 24m 10s (non-voting)\n- kolla-build-ubuntu-source https://logs.opendev.org/16/642416/31/check/kolla-build-ubuntu-source/701bd2a/ : SUCCESS in 28m 25s\n- kolla-ansible-ubuntu-source https://logs.opendev.org/16/642416/31/check/kolla-ansible-ubuntu-source/39949be/ : SUCCESS in 53m 14s\n- kolla-ansible-ubuntu-binary https://logs.opendev.org/16/642416/31/check/kolla-ansible-ubuntu-binary/5f8a9cc/ : SUCCESS in 57m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://logs.opendev.org/16/642416/31/check/kolla-ansible-ubuntu-source-upgrade/4dbec92/ : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":31},{"id":"6bbe7c7f1dc21bf60f130e60fc48de7c825b27ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 17:16:04.000000000","message":"Patch Set 31: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":31},{"id":"95db65296cc1cc69e3683b8ab870558bf2865ddb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 18:25:06.000000000","message":"Patch Set 31: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://logs.opendev.org/16/642416/31/gate/kolla-build-centos-source/3907d8d/ : SUCCESS in 34m 57s\n- tripleo-build-containers-centos-7 https://logs.opendev.org/16/642416/31/gate/tripleo-build-containers-centos-7/e81862f/ : SUCCESS in 1h 03m 42s\n- openstack-tox-lower-constraints https://logs.opendev.org/16/642416/31/gate/openstack-tox-lower-constraints/bd0044e/ : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://logs.opendev.org/16/642416/31/gate/openstack-tox-pep8/1e448b0/ : SUCCESS in 4m 57s\n- openstack-tox-py27 https://logs.opendev.org/16/642416/31/gate/openstack-tox-py27/2de35a1/ : SUCCESS in 4m 37s\n- openstack-tox-py36 https://logs.opendev.org/16/642416/31/gate/openstack-tox-py36/1760f28/ : SUCCESS in 3m 53s\n- openstack-tox-py37 https://logs.opendev.org/16/642416/31/gate/openstack-tox-py37/071f2f5/ : SUCCESS in 4m 20s\n- openstack-tox-docs https://logs.opendev.org/16/642416/31/gate/openstack-tox-docs/2d4a32a/html/ : SUCCESS in 3m 39s\n- kolla-build-ubuntu-source https://logs.opendev.org/16/642416/31/gate/kolla-build-ubuntu-source/7732db6/ : SUCCESS in 34m 25s","accounts_in_message":[],"_revision_number":31},{"id":"799c76447d270072aae3f0151ed883263d3a32fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 18:25:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":31},{"id":"0f66937e52a69cf31f906e0bd008137dc368965f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-09 08:59:30.000000000","message":"Patch Set 31:\n\n(2 comments)","accounts_in_message":[],"_revision_number":31}],"current_revision_number":31,"current_revision":"e3108d93c69dbad0c575da719f842d36c195a673","revisions":{"058816dcc42a4d8b7f3806e9783fca78a7d7d3b9":{"kind":"REWORK","_number":1,"created":"2019-03-11 10:49:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/1"}}},"commit":{"parents":[{"commit":"fc8b5fa89df77f2f9f1f3a66849d3fbadd7f2299","subject":"Merge \"openstack-base: drop pika-pool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc8b5fa89df77f2f9f1f3a66849d3fbadd7f2299"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/058816dcc42a4d8b7f3806e9783fca78a7d7d3b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/058816dcc42a4d8b7f3806e9783fca78a7d7d3b9"}]},"branch":"refs/heads/master"},"33b542561ee80d73dd384529e5e6bff034117d91":{"kind":"REWORK","_number":2,"created":"2019-03-11 12:08:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/2"}}},"commit":{"parents":[{"commit":"fc8b5fa89df77f2f9f1f3a66849d3fbadd7f2299","subject":"Merge \"openstack-base: drop pika-pool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc8b5fa89df77f2f9f1f3a66849d3fbadd7f2299"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 12:08:21.000000000","tz":60},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33b542561ee80d73dd384529e5e6bff034117d91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33b542561ee80d73dd384529e5e6bff034117d91"}]},"branch":"refs/heads/master"},"99c937462163a06cddc20e4222016eaf44806896":{"kind":"REWORK","_number":3,"created":"2019-03-11 13:56:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/3"}}},"commit":{"parents":[{"commit":"fc8b5fa89df77f2f9f1f3a66849d3fbadd7f2299","subject":"Merge \"openstack-base: drop pika-pool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc8b5fa89df77f2f9f1f3a66849d3fbadd7f2299"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 13:56:09.000000000","tz":60},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/99c937462163a06cddc20e4222016eaf44806896"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/99c937462163a06cddc20e4222016eaf44806896"}]},"branch":"refs/heads/master"},"75592d987344c901c3c750206288bea37a0d470a":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-05-21 12:34:48.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/16/642416/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/4"}}},"commit":{"parents":[{"commit":"fc8b5fa89df77f2f9f1f3a66849d3fbadd7f2299","subject":"Merge \"openstack-base: drop pika-pool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc8b5fa89df77f2f9f1f3a66849d3fbadd7f2299"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-21 12:34:36.000000000","tz":0},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCloses-bug: 1823736\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75592d987344c901c3c750206288bea37a0d470a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75592d987344c901c3c750206288bea37a0d470a"}]},"branch":"refs/heads/master"},"daced73186acd78e66ba43e9c639b1755dc9a0da":{"kind":"REWORK","_number":5,"created":"2019-06-06 06:53:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/5"}}},"commit":{"parents":[{"commit":"fc8b5fa89df77f2f9f1f3a66849d3fbadd7f2299","subject":"Merge \"openstack-base: drop pika-pool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc8b5fa89df77f2f9f1f3a66849d3fbadd7f2299"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-06-06 06:53:05.000000000","tz":0},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCloses-bug: 1823736\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/daced73186acd78e66ba43e9c639b1755dc9a0da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/daced73186acd78e66ba43e9c639b1755dc9a0da"}]},"branch":"refs/heads/master"},"573bf07c4b8a22bcfcbb7117e4e79c5d40510e91":{"kind":"REWORK","_number":6,"created":"2019-06-06 07:12:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/6"}}},"commit":{"parents":[{"commit":"fc8b5fa89df77f2f9f1f3a66849d3fbadd7f2299","subject":"Merge \"openstack-base: drop pika-pool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc8b5fa89df77f2f9f1f3a66849d3fbadd7f2299"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-06-06 07:12:04.000000000","tz":0},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCloses-bug: 1823736\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/573bf07c4b8a22bcfcbb7117e4e79c5d40510e91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/573bf07c4b8a22bcfcbb7117e4e79c5d40510e91"}]},"branch":"refs/heads/master"},"beea28794b8e042b5ada96729d1f4f365cae45e5":{"kind":"REWORK","_number":7,"created":"2019-06-07 07:06:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/7"}}},"commit":{"parents":[{"commit":"03ca8c63256549ab2f204e89d225abb8b9ed0e6b","subject":"Merge \"Blacklist bandit 1.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03ca8c63256549ab2f204e89d225abb8b9ed0e6b"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-06-07 07:06:18.000000000","tz":120},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCloses-bug: 1823736\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/beea28794b8e042b5ada96729d1f4f365cae45e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/beea28794b8e042b5ada96729d1f4f365cae45e5"}]},"branch":"refs/heads/master"},"32aa4c3630e366f8d67537337aed85da28cfef29":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2019-06-07 17:18:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/8"}}},"commit":{"parents":[{"commit":"03ca8c63256549ab2f204e89d225abb8b9ed0e6b","subject":"Merge \"Blacklist bandit 1.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03ca8c63256549ab2f204e89d225abb8b9ed0e6b"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-06-07 17:18:42.000000000","tz":0},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCloses-bug: 1823736\nDepends-On: https://review.opendev.org/664009\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32aa4c3630e366f8d67537337aed85da28cfef29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32aa4c3630e366f8d67537337aed85da28cfef29"}]},"branch":"refs/heads/master"},"4fb26aeeec72cb2566d3c404ccc36b7ef694a2f4":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2019-06-10 09:15:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/9"}}},"commit":{"parents":[{"commit":"03ca8c63256549ab2f204e89d225abb8b9ed0e6b","subject":"Merge \"Blacklist bandit 1.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03ca8c63256549ab2f204e89d225abb8b9ed0e6b"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-06-10 09:15:31.000000000","tz":0},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCurrently we use Ansible 2.2 in kolla_toolbox, while lowest supported\nversion is 2.6 - bumping up to 2.8 and resolving issues in dependent\nchanges.\n\nThis is part of the effort to move to newer Ansible, including Python3.\n\nCloses-bug: 1823736\nDepends-On: https://review.opendev.org/664009\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4fb26aeeec72cb2566d3c404ccc36b7ef694a2f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4fb26aeeec72cb2566d3c404ccc36b7ef694a2f4"}]},"branch":"refs/heads/master"},"ab98e58198ca2bb527ea09f05519b7020fbc29a8":{"kind":"REWORK","_number":10,"created":"2019-06-11 14:24:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/10"}}},"commit":{"parents":[{"commit":"b5c8b595a7f9c4aebb35b0ceacc236ec5f4eb350","subject":"Merge \"Fix issues with Percona in Zuul and clean up a little bit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5c8b595a7f9c4aebb35b0ceacc236ec5f4eb350"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-06-11 14:24:54.000000000","tz":120},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCurrently we use Ansible 2.2 in kolla_toolbox, while lowest supported\nversion is 2.6 - bumping up to 2.8 and resolving issues in dependent\nchanges.\n\nSome changes are needed in kolla_* modules to support no_log\u003dTrue when\npassing passwords.\n\nThis is part of the effort to move to newer Ansible, including Python3.\n\nCloses-bug: 1823736\nDepends-On: https://review.opendev.org/664009\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab98e58198ca2bb527ea09f05519b7020fbc29a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab98e58198ca2bb527ea09f05519b7020fbc29a8"}]},"branch":"refs/heads/master"},"119b61de0f5e68b7993333f7f3866789c1a0e66b":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2019-06-11 14:28:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/11"}}},"commit":{"parents":[{"commit":"b5c8b595a7f9c4aebb35b0ceacc236ec5f4eb350","subject":"Merge \"Fix issues with Percona in Zuul and clean up a little bit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5c8b595a7f9c4aebb35b0ceacc236ec5f4eb350"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-06-11 14:28:50.000000000","tz":0},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCurrently we use Ansible 2.2 in kolla_toolbox, while lowest supported\nversion is 2.6 - bumping up to 2.8 and resolving issues in dependent\nchanges.\n\nSome changes are needed in kolla_* modules to support no_log\u003dTrue when\npassing passwords.\n\nThis is part of the effort to move to newer Ansible, including Python3.\n\nCloses-bug: 1823736\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/119b61de0f5e68b7993333f7f3866789c1a0e66b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/119b61de0f5e68b7993333f7f3866789c1a0e66b"}]},"branch":"refs/heads/master"},"a79ca2798702e2270c465c41b1832634ad43a2b7":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2019-06-13 07:04:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/12"}}},"commit":{"parents":[{"commit":"b5c8b595a7f9c4aebb35b0ceacc236ec5f4eb350","subject":"Merge \"Fix issues with Percona in Zuul and clean up a little bit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5c8b595a7f9c4aebb35b0ceacc236ec5f4eb350"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-06-13 07:04:41.000000000","tz":0},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCurrently we use Ansible 2.2 in kolla_toolbox, while lowest supported\nversion is 2.6 - bumping up to 2.8 and resolving issues in dependent\nchanges.\n\nSome changes are needed in kolla_* modules to support no_log\u003dTrue when\npassing passwords.\n\nThis is part of the effort to move to newer Ansible, including Python3.\n\nDepends-On: https://review.opendev.org/664631\n\nCloses-bug: 1823736\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a79ca2798702e2270c465c41b1832634ad43a2b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a79ca2798702e2270c465c41b1832634ad43a2b7"}]},"branch":"refs/heads/master"},"fab5de1f81e8c8c45087c60d8913225a169a5fcd":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-07-23 09:48:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/13"}}},"commit":{"parents":[{"commit":"ae12359b83c514473793a9a42281a786d553b64f","subject":"Merge \"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae12359b83c514473793a9a42281a786d553b64f"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-07-23 09:48:31.000000000","tz":0},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCurrently we use Ansible 2.2 in kolla_toolbox, while lowest supported\nversion is 2.6 - bumping up to 2.8 and resolving issues in dependent\nchanges.\n\nSome changes are needed in kolla_* modules to support no_log\u003dTrue when\npassing passwords.\n\nThis is part of the effort to move to newer Ansible, including Python3.\n\nDepends-On: https://review.opendev.org/664631\n\nCloses-bug: 1823736\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fab5de1f81e8c8c45087c60d8913225a169a5fcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fab5de1f81e8c8c45087c60d8913225a169a5fcd"}]},"branch":"refs/heads/master"},"1237b9713e435f24a49f1b7fa3c14a7bd8ffa9fc":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2019-08-01 07:43:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/14"}}},"commit":{"parents":[{"commit":"ae12359b83c514473793a9a42281a786d553b64f","subject":"Merge \"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae12359b83c514473793a9a42281a786d553b64f"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-01 07:42:43.000000000","tz":0},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCurrently we use Ansible 2.2 in kolla_toolbox, while lowest supported\nversion is 2.6 - bumping up to 2.8 and resolving issues in dependent\nchanges.\n\nSome changes are needed in kolla_* modules to support no_log\u003dTrue when\npassing passwords.\n\nThis is part of the effort to move to newer Ansible, including Python3.\n\nDepends-On: https://review.opendev.org/673949\n\nCloses-bug: 1823736\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1237b9713e435f24a49f1b7fa3c14a7bd8ffa9fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1237b9713e435f24a49f1b7fa3c14a7bd8ffa9fc"}]},"branch":"refs/heads/master"},"2711ee8c85ad0924ffdcc17e197a9ed7778550a3":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2019-08-02 06:19:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/15"}}},"commit":{"parents":[{"commit":"ae12359b83c514473793a9a42281a786d553b64f","subject":"Merge \"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae12359b83c514473793a9a42281a786d553b64f"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-02 06:18:58.000000000","tz":0},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCurrently we use Ansible 2.2 in kolla_toolbox, while lowest supported\nversion is 2.6 - bumping up to 2.8 and resolving issues in dependent\nchanges.\n\nSome changes are needed in kolla_* modules to support no_log\u003dTrue when\npassing passwords.\n\nThis is part of the effort to move to newer Ansible, including Python3.\n\nDepends-On: https://review.opendev.org/673949\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2711ee8c85ad0924ffdcc17e197a9ed7778550a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2711ee8c85ad0924ffdcc17e197a9ed7778550a3"}]},"branch":"refs/heads/master"},"11d3d264d12087d152da25e398312a9f502871e9":{"kind":"REWORK","_number":16,"created":"2019-08-02 06:20:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/16"}}},"commit":{"parents":[{"commit":"ae12359b83c514473793a9a42281a786d553b64f","subject":"Merge \"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae12359b83c514473793a9a42281a786d553b64f"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-02 06:20:53.000000000","tz":0},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCurrently we use Ansible 2.2 in kolla_toolbox, while lowest supported\nversion is 2.6 - bumping up to 2.8 and resolving issues in dependent\nchanges.\n\nSome changes are needed in kolla_* modules to support no_log\u003dTrue when\npassing passwords.\n\nThis is part of the effort to move to newer Ansible, including Python3.\n\nDepends-On: https://review.opendev.org/673949\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11d3d264d12087d152da25e398312a9f502871e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11d3d264d12087d152da25e398312a9f502871e9"}]},"branch":"refs/heads/master"},"b4feb076aeecc5d86d8a8685a52e5d6945d7fdc4":{"kind":"REWORK","_number":17,"created":"2019-08-02 06:35:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/17"}}},"commit":{"parents":[{"commit":"ae12359b83c514473793a9a42281a786d553b64f","subject":"Merge \"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae12359b83c514473793a9a42281a786d553b64f"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-02 06:35:44.000000000","tz":0},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCurrently we use Ansible 2.2 in kolla_toolbox, while lowest supported\nversion is 2.6 - bumping up to 2.8 and resolving issues in dependent\nchanges.\n\nSome changes are needed in kolla_* modules to support no_log\u003dTrue when\npassing passwords.\n\nThis is part of the effort to move to newer Ansible, including Python3.\n\nDepends-On: https://review.opendev.org/673949\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4feb076aeecc5d86d8a8685a52e5d6945d7fdc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4feb076aeecc5d86d8a8685a52e5d6945d7fdc4"}]},"branch":"refs/heads/master"},"40d77dc9bfcb5a83682a7580427ce7717ee990cf":{"kind":"NO_CODE_CHANGE","_number":18,"created":"2019-08-02 06:38:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/18"}}},"commit":{"parents":[{"commit":"ae12359b83c514473793a9a42281a786d553b64f","subject":"Merge \"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae12359b83c514473793a9a42281a786d553b64f"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-02 06:38:06.000000000","tz":0},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCurrently we use Ansible 2.2 in kolla_toolbox, while lowest supported\nversion is 2.6 - bumping up to 2.8 and resolving issues in dependent\nchanges.\n\nSome changes are needed in kolla_* modules to support no_log\u003dTrue when\npassing passwords.\n\nThis is part of the effort to move to newer Ansible, including Python3.\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/40d77dc9bfcb5a83682a7580427ce7717ee990cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/40d77dc9bfcb5a83682a7580427ce7717ee990cf"}]},"branch":"refs/heads/master"},"21c06a308be96e322a097448c9729732fd73a7d5":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2019-08-02 07:53:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/19"}}},"commit":{"parents":[{"commit":"6295f17985b5e691e136269a2b2fdefed8d722ed","subject":"Merge \"Update build profiles for Placement split\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6295f17985b5e691e136269a2b2fdefed8d722ed"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-02 07:53:02.000000000","tz":0},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCurrently we use Ansible 2.2 in kolla_toolbox, while lowest supported\nversion is 2.6 - bumping up to 2.8 and resolving issues in dependent\nchanges.\n\nSome changes are needed in kolla_* modules to support no_log\u003dTrue when\npassing passwords.\n\nThis is part of the effort to move to newer Ansible, including Python3.\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21c06a308be96e322a097448c9729732fd73a7d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21c06a308be96e322a097448c9729732fd73a7d5"}]},"branch":"refs/heads/master"},"b4ff696808dfd9cdf6ccca700f0e886be40bb49a":{"kind":"REWORK","_number":20,"created":"2019-08-02 07:53:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/20"}}},"commit":{"parents":[{"commit":"6295f17985b5e691e136269a2b2fdefed8d722ed","subject":"Merge \"Update build profiles for Placement split\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6295f17985b5e691e136269a2b2fdefed8d722ed"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-02 07:53:13.000000000","tz":0},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCurrently we use Ansible 2.2 in kolla_toolbox, while lowest supported\nversion is 2.6 - bumping up to 2.8 and resolving issues in dependent\nchanges.\n\nSome changes are needed in kolla_* modules to support no_log\u003dTrue when\npassing passwords.\n\nThis is part of the effort to move to newer Ansible, including Python3.\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4ff696808dfd9cdf6ccca700f0e886be40bb49a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4ff696808dfd9cdf6ccca700f0e886be40bb49a"}]},"branch":"refs/heads/master"},"3fe9b439bbc1e22fcddc116fab7be2e23b6e4575":{"kind":"REWORK","_number":21,"created":"2019-08-02 09:28:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/21"}}},"commit":{"parents":[{"commit":"6295f17985b5e691e136269a2b2fdefed8d722ed","subject":"Merge \"Update build profiles for Placement split\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6295f17985b5e691e136269a2b2fdefed8d722ed"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-02 09:28:12.000000000","tz":0},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCurrently we use Ansible 2.2 in kolla_toolbox, while lowest supported\nversion is 2.6 - bumping up to 2.8 and resolving issues in dependent\nchanges.\n\nSome changes are needed in kolla_* modules to support no_log\u003dTrue when\npassing passwords.\n\nThis is part of the effort to move to newer Ansible, including Python3.\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3fe9b439bbc1e22fcddc116fab7be2e23b6e4575"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3fe9b439bbc1e22fcddc116fab7be2e23b6e4575"}]},"branch":"refs/heads/master"},"fa9ba6af6432e3c817ae4d8a8dbafd5a4bcf155c":{"kind":"REWORK","_number":22,"created":"2019-08-02 09:30:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/22"}}},"commit":{"parents":[{"commit":"6295f17985b5e691e136269a2b2fdefed8d722ed","subject":"Merge \"Update build profiles for Placement split\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6295f17985b5e691e136269a2b2fdefed8d722ed"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-02 09:29:59.000000000","tz":0},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCurrently we use Ansible 2.2 in kolla_toolbox, while lowest supported\nversion is 2.6 - bumping up to 2.8 and resolving issues in dependent\nchanges.\n\nSome changes are needed in kolla_* modules to support no_log\u003dTrue when\npassing passwords.\n\nThis is part of the effort to move to newer Ansible, including Python3.\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa9ba6af6432e3c817ae4d8a8dbafd5a4bcf155c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa9ba6af6432e3c817ae4d8a8dbafd5a4bcf155c"}]},"branch":"refs/heads/master"},"d592b872fcbe46500cb65a3aacad04734508c815":{"kind":"REWORK","_number":23,"created":"2019-08-02 09:39:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/23"}}},"commit":{"parents":[{"commit":"6295f17985b5e691e136269a2b2fdefed8d722ed","subject":"Merge \"Update build profiles for Placement split\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6295f17985b5e691e136269a2b2fdefed8d722ed"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-02 09:39:01.000000000","tz":0},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCurrently we use Ansible 2.2 in kolla_toolbox, while lowest supported\nversion is 2.6 - bumping up to 2.8 and resolving issues in dependent\nchanges.\n\nSome changes are needed in kolla_* modules to support no_log\u003dTrue when\npassing passwords.\n\nThis is part of the effort to move to newer Ansible, including Python3.\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d592b872fcbe46500cb65a3aacad04734508c815"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d592b872fcbe46500cb65a3aacad04734508c815"}]},"branch":"refs/heads/master"},"eabc0b6d68d891c42f4033f174c1be6fc9e7ea2f":{"kind":"NO_CODE_CHANGE","_number":24,"created":"2019-08-02 09:45:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/24"}}},"commit":{"parents":[{"commit":"6295f17985b5e691e136269a2b2fdefed8d722ed","subject":"Merge \"Update build profiles for Placement split\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6295f17985b5e691e136269a2b2fdefed8d722ed"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-02 09:45:45.000000000","tz":0},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCurrently we use Ansible 2.2 in kolla_toolbox, while lowest supported\nversion is 2.6 - bumping up to 2.8.\n\n* Use upper constrains in Dockerfile.j2, instead of pinning them in\nhere - Ansible is pinned to 2.8.*\n* Add no_log\u003dTrue in kolla_keystone_user\n* Add endpoint_type as a param in kolla_keystone_service - now\nAnsibleModule requires all options to be defined, otherwise it throws\nan UnsupportedParameter error\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eabc0b6d68d891c42f4033f174c1be6fc9e7ea2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eabc0b6d68d891c42f4033f174c1be6fc9e7ea2f"}]},"branch":"refs/heads/master"},"437935950f31394500e5930fdd220e9d54983abe":{"kind":"REWORK","_number":25,"created":"2019-08-02 11:46:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/25"}}},"commit":{"parents":[{"commit":"6295f17985b5e691e136269a2b2fdefed8d722ed","subject":"Merge \"Update build profiles for Placement split\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6295f17985b5e691e136269a2b2fdefed8d722ed"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-02 11:46:17.000000000","tz":0},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCurrently we use Ansible 2.2 in kolla_toolbox, while lowest supported\nversion is 2.6 - bumping up to 2.8.\n\n* Use upper constrains in Dockerfile.j2, instead of pinning them in\nhere - Ansible is pinned to 2.8.*\n* Add no_log\u003dTrue in kolla_keystone_user\n* Add endpoint_type as a param in kolla_keystone_service - now\nAnsibleModule requires all options to be defined, otherwise it throws\nan UnsupportedParameter error\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/437935950f31394500e5930fdd220e9d54983abe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/437935950f31394500e5930fdd220e9d54983abe"}]},"branch":"refs/heads/master"},"8992b14c4389eb83ad42db7b1caf38c84cb2b69c":{"kind":"REWORK","_number":26,"created":"2019-08-02 13:01:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/26"}}},"commit":{"parents":[{"commit":"6295f17985b5e691e136269a2b2fdefed8d722ed","subject":"Merge \"Update build profiles for Placement split\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6295f17985b5e691e136269a2b2fdefed8d722ed"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-02 13:01:01.000000000","tz":0},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCurrently we use Ansible 2.2 in kolla_toolbox, while lowest supported\nversion is 2.6 - bumping up to 2.8.\n\n* Use upper constrains in Dockerfile.j2, instead of pinning them in\nhere - Ansible is pinned to 2.8.*\n* Add no_log\u003dTrue in kolla_keystone_user\n* Add endpoint_type as a param in kolla_keystone_service - now\nAnsibleModule requires all options to be defined, otherwise it throws\nan UnsupportedParameter error\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8992b14c4389eb83ad42db7b1caf38c84cb2b69c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8992b14c4389eb83ad42db7b1caf38c84cb2b69c"}]},"branch":"refs/heads/master"},"111034d3076ff581a60b6a8d0b7b805dd911c1ea":{"kind":"REWORK","_number":27,"created":"2019-08-08 07:10:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/27"}}},"commit":{"parents":[{"commit":"bc717cba729815a7c8165c541006f13ac194dd4e","subject":"Merge \"Change monasca-grafana source branch to use grafana v4.2.0 with keystone changes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc717cba729815a7c8165c541006f13ac194dd4e"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-08 07:10:11.000000000","tz":120},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCurrently we use Ansible 2.2 in kolla_toolbox, while lowest supported\nversion is 2.6 - bumping up to 2.8.\n\n* Use upper constrains in Dockerfile.j2, instead of pinning them in\nhere - Ansible is pinned to 2.8.*\n* Add no_log\u003dTrue in kolla_keystone_user\n* Add endpoint_type as a param in kolla_keystone_service - now\nAnsibleModule requires all options to be defined, otherwise it throws\nan UnsupportedParameter error\n* Add openstack_release variable in kolla/common/config.py which\n  is used for fetching upper-constraints.txt for desired release\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/111034d3076ff581a60b6a8d0b7b805dd911c1ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/111034d3076ff581a60b6a8d0b7b805dd911c1ea"}]},"branch":"refs/heads/master"},"8c31c48e4ca0d79eb304b62c7c6f9ba0b5c8baf8":{"kind":"NO_CHANGE","_number":28,"created":"2019-08-08 07:11:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/28"}}},"commit":{"parents":[{"commit":"bc717cba729815a7c8165c541006f13ac194dd4e","subject":"Merge \"Change monasca-grafana source branch to use grafana v4.2.0 with keystone changes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc717cba729815a7c8165c541006f13ac194dd4e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-08 07:10:56.000000000","tz":120},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCurrently we use Ansible 2.2 in kolla_toolbox, while lowest supported\nversion is 2.6 - bumping up to 2.8.\n\n* Use upper constrains in Dockerfile.j2, instead of pinning them in\nhere - Ansible is pinned to 2.8.*\n* Add no_log\u003dTrue in kolla_keystone_user\n* Add endpoint_type as a param in kolla_keystone_service - now\nAnsibleModule requires all options to be defined, otherwise it throws\nan UnsupportedParameter error\n* Add openstack_release variable in kolla/common/config.py which\n  is used for fetching upper-constraints.txt for desired release\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c31c48e4ca0d79eb304b62c7c6f9ba0b5c8baf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c31c48e4ca0d79eb304b62c7c6f9ba0b5c8baf8"}]},"branch":"refs/heads/master"},"b46f230efd109d2570f1618db849574f5500a761":{"kind":"REWORK","_number":29,"created":"2019-08-08 09:34:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/29"}}},"commit":{"parents":[{"commit":"bc717cba729815a7c8165c541006f13ac194dd4e","subject":"Merge \"Change monasca-grafana source branch to use grafana v4.2.0 with keystone changes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc717cba729815a7c8165c541006f13ac194dd4e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-08 09:34:48.000000000","tz":120},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCurrently we use Ansible 2.2 in kolla_toolbox, while lowest supported\nversion is 2.6 - bumping up to 2.8.\n\n* Use upper constrains in Dockerfile.j2, instead of pinning them in\nhere - Ansible is pinned to 2.8.*\n* Add no_log\u003dTrue in kolla_keystone_user\n* Add endpoint_type as a param in kolla_keystone_service - now\nAnsibleModule requires all options to be defined, otherwise it throws\nan UnsupportedParameter error\n* Add openstack_release variable in kolla/common/config.py which\n  is used for fetching upper-constraints.txt for desired release\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b46f230efd109d2570f1618db849574f5500a761"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b46f230efd109d2570f1618db849574f5500a761"}]},"branch":"refs/heads/master"},"c4bf229ecdee8ef1c7990e44771177b7d843c0bd":{"kind":"REWORK","_number":30,"created":"2019-08-08 11:22:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/30"}}},"commit":{"parents":[{"commit":"bc717cba729815a7c8165c541006f13ac194dd4e","subject":"Merge \"Change monasca-grafana source branch to use grafana v4.2.0 with keystone changes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc717cba729815a7c8165c541006f13ac194dd4e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-08 11:22:01.000000000","tz":120},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCurrently we use Ansible 2.2 in kolla_toolbox, while lowest supported\nversion is 2.6 - bumping up to 2.8.\n\n* Use upper constrains in Dockerfile.j2, instead of pinning them in\nhere - Ansible is pinned to 2.8.*\n* Add no_log\u003dTrue in kolla_keystone_user\n* Add endpoint_type as a param in kolla_keystone_service - now\nAnsibleModule requires all options to be defined, otherwise it throws\nan UnsupportedParameter error\n* Add openstack_release variable in kolla/common/config.py which\n  is used for fetching upper-constraints.txt for desired release\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4bf229ecdee8ef1c7990e44771177b7d843c0bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4bf229ecdee8ef1c7990e44771177b7d843c0bd"}]},"branch":"refs/heads/master"},"e3108d93c69dbad0c575da719f842d36c195a673":{"kind":"REWORK","_number":31,"created":"2019-08-08 13:56:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/642416/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/642416/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/642416/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/642416/31"}}},"commit":{"parents":[{"commit":"bc717cba729815a7c8165c541006f13ac194dd4e","subject":"Merge \"Change monasca-grafana source branch to use grafana v4.2.0 with keystone changes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc717cba729815a7c8165c541006f13ac194dd4e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-03-11 10:49:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-08 13:56:42.000000000","tz":0},"subject":"Bump up pkg versions in kolla-toolbox","message":"Bump up pkg versions in kolla-toolbox\n\nCurrently we use Ansible 2.2 in kolla_toolbox, while lowest supported\nversion is 2.6 - bumping up to 2.8.\n\n* Use upper constrains in Dockerfile.j2, instead of pinning them in\nhere - Ansible is pinned to 2.8.*\n* Add no_log\u003dTrue in kolla_keystone_user\n* Add endpoint_type as a param in kolla_keystone_service - now\nAnsibleModule requires all options to be defined, otherwise it throws\nan UnsupportedParameter error\n* Add openstack_release variable in kolla/common/config.py which\n  is used for fetching upper-constraints.txt for desired release\n\nChange-Id: I2cb0ea2fde1b408d48b39bf7a54b898e850d31aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3108d93c69dbad0c575da719f842d36c195a673"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3108d93c69dbad0c575da719f842d36c195a673"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
