)]}'
{"id":"openstack%2Fkolla~649292","triplet_id":"openstack%2Fkolla~master~I78b74903f733451d5284caec48f10d4ff39dcf3c","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I78b74903f733451d5284caec48f10d4ff39dcf3c","subject":"DNM: Test bifrost fix","status":"ABANDONED","created":"2019-04-02 12:13:28.000000000","updated":"2019-04-02 16:20:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a6cdddd3c1f4df1978363b3e47f1cf716d65fcff","_number":649292,"virtual_id_number":649292,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-04-02 15:12:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-04-02 13:19:59.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-02 13:19:59.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-02 15:12:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2b638570ded6cbea360cdec52804187823acbe1d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-02 12:13:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42687592fbce3ebbdc05f29546aed5f4e2d2fc28","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-02 13:19:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 http://logs.rdoproject.org/92/649292/1/openstack-check/tripleo-build-containers-fedora-28/e1cbc6d/ : SUCCESS in 35m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ba8733c44652cbfdca052e32a40aa2c59c7fcfb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 15:12:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/92/649292/1/check/kolla-build-centos-binary/a4cf06e/ : FAILURE in 52m 31s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/92/649292/1/check/kolla-build-centos-source/9104d47/ : FAILURE in 52m 49s\n- kolla-ansible-centos-source http://logs.openstack.org/92/649292/1/check/kolla-ansible-centos-source/9d6faa8/ : SUCCESS in 1h 08m 30s\n- kolla-ansible-centos-binary http://logs.openstack.org/92/649292/1/check/kolla-ansible-centos-binary/7ffd4c5/ : SUCCESS in 1h 09m 36s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/92/649292/1/check/tripleo-build-containers-centos-7/7afb261/ : SUCCESS in 42m 36s (non-voting)\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/92/649292/1/check/kolla-ansible-bifrost-centos-source/2b557a7/ : FAILURE in 1h 00m 16s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/92/649292/1/check/kolla-build-oraclelinux-binary/493c486/ : FAILURE in 54m 39s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/92/649292/1/check/kolla-build-oraclelinux-source/763da92/ : FAILURE in 1h 04m 58s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/92/649292/1/check/kolla-ansible-oraclelinux-source/de48ee9/ : SUCCESS in 1h 10m 51s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/92/649292/1/check/kolla-ansible-oraclelinux-binary/ca788a0/ : SUCCESS in 1h 02m 34s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/92/649292/1/check/openstack-tox-lower-constraints/5478fb8/ : SUCCESS in 5m 04s\n- openstack-tox-pep8 http://logs.openstack.org/92/649292/1/check/openstack-tox-pep8/378ebb7/ : SUCCESS in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/92/649292/1/check/openstack-tox-py27/21d3c6f/ : SUCCESS in 4m 33s\n- openstack-tox-py36 http://logs.openstack.org/92/649292/1/check/openstack-tox-py36/cc4c76f/ : SUCCESS in 4m 26s\n- openstack-tox-py37 http://logs.openstack.org/92/649292/1/check/openstack-tox-py37/aea6d7f/ : SUCCESS in 4m 09s\n- openstack-tox-docs http://logs.openstack.org/92/649292/1/check/openstack-tox-docs/fae1a5e/html/ : SUCCESS in 3m 55s\n- kolla-build-ubuntu-binary http://logs.openstack.org/92/649292/1/check/kolla-build-ubuntu-binary/b746d34/ : TIMED_OUT in 2h 51m 12s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/92/649292/1/check/kolla-build-ubuntu-source/bb03c12/ : FAILURE in 18m 44s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/92/649292/1/check/kolla-ansible-ubuntu-source/c0c1594/ : FAILURE in 29m 09s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/92/649292/1/check/kolla-ansible-ubuntu-binary/5b4d5ca/ : TIMED_OUT in 2h 01m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a6cdddd3c1f4df1978363b3e47f1cf716d65fcff","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-02 16:20:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b50a508550d1c722a302b65eb6b3dc0e19ceb7b","revisions":{"9b50a508550d1c722a302b65eb6b3dc0e19ceb7b":{"kind":"REWORK","_number":1,"created":"2019-04-02 12:13:28.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/92/649292/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/649292/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/649292/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/649292/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/649292/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/649292/1"}}},"commit":{"parents":[{"commit":"c8d91a1dafa58a116255994111838c94a628df86","subject":"Support building source images with in-review changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8d91a1dafa58a116255994111838c94a628df86"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-04-02 12:12:27.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-04-02 12:12:27.000000000","tz":60},"subject":"DNM: Test bifrost fix","message":"DNM: Test bifrost fix\n\nChange-Id: I78b74903f733451d5284caec48f10d4ff39dcf3c\nDepends-On: https://review.openstack.org/649291\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b50a508550d1c722a302b65eb6b3dc0e19ceb7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b50a508550d1c722a302b65eb6b3dc0e19ceb7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
