)]}'
{"id":"openstack%2Fkolla~651601","triplet_id":"openstack%2Fkolla~stable%2Fstein~I1df5dd28adc65ff461a893b60c57d7b5da96d653","project":"openstack/kolla","branch":"stable/stein","topic":"centos-rdo-stein","hashtags":[],"change_id":"I1df5dd28adc65ff461a893b60c57d7b5da96d653","subject":"Switch to RDO Stein release on CentOS \u0026 Oracle","status":"MERGED","created":"2019-04-10 16:29:42.000000000","updated":"2019-08-19 14:14:43.000000000","submitted":"2019-06-07 10:55:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"651601-1559904924117-e9fba48b","meta_rev_id":"ae63f12ce4e9722bf10ea662b557d9d3beef4b77","_number":651601,"virtual_id_number":651601,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-06-07 10:55:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-07 08:50:07.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":2,"date":"2019-06-07 07:04:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-06-06 13:33:08.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-06-07 07:10:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":1,"date":"2019-06-07 07:04:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-06-06 15:08:46.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-29 06:42:37.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-06-06 13:33:08.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-07 07:04:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-06-07 07:10:02.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-06-07 08:50:07.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-07 10:55:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c10328cd339f9ad8434c44b6fa68400b75cda4b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-10 16:29:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f039ba83a6b84d1f6a63b9edeb0655d578034fa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-10 16:30:14.000000000","message":"Patch Set 1: Workflow-1\n\nRepo not available yet.","accounts_in_message":[],"_revision_number":1},{"id":"bbc4820ed33a47e0dbd588722a61deb88ba15fb9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-10 17:07:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28 http://logs.rdoproject.org/01/651601/1/openstack-check/tripleo-build-containers-fedora-28/b348016/ : FAILURE in 34m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"19f2d9325305189ed7dca56d5fee0a5fe9ea52b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-10 17:40:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/01/651601/1/check/kolla-build-centos-binary/12eca56/ : FAILURE in 7m 45s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/01/651601/1/check/kolla-build-centos-source/b8f404c/ : FAILURE in 8m 59s\n- kolla-ansible-centos-source http://logs.openstack.org/01/651601/1/check/kolla-ansible-centos-source/7789edb/ : FAILURE in 10m 29s\n- kolla-ansible-centos-binary http://logs.openstack.org/01/651601/1/check/kolla-ansible-centos-binary/d055711/ : FAILURE in 11m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/01/651601/1/check/kolla-ansible-centos-source-upgrade/354e21f/ : FAILURE in 10m 07s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/01/651601/1/check/tripleo-build-containers-centos-7/9ba708a/ : FAILURE in 10m 59s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/01/651601/1/check/kolla-build-oraclelinux-binary/b507c27/ : FAILURE in 8m 52s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/01/651601/1/check/kolla-build-oraclelinux-source/52ddf7a/ : FAILURE in 8m 39s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/01/651601/1/check/kolla-ansible-oraclelinux-source/a1d597d/ : FAILURE in 11m 30s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/01/651601/1/check/kolla-ansible-oraclelinux-binary/1f33b42/ : FAILURE in 11m 20s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/01/651601/1/check/openstack-tox-lower-constraints/99e162f/ : SUCCESS in 7m 14s\n- openstack-tox-pep8 http://logs.openstack.org/01/651601/1/check/openstack-tox-pep8/6cdaff9/ : SUCCESS in 4m 01s\n- openstack-tox-py27 http://logs.openstack.org/01/651601/1/check/openstack-tox-py27/e15af7c/ : SUCCESS in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/01/651601/1/check/openstack-tox-py36/f546823/ : SUCCESS in 4m 33s\n- openstack-tox-py37 http://logs.openstack.org/01/651601/1/check/openstack-tox-py37/1560446/ : SUCCESS in 7m 45s\n- openstack-tox-docs http://logs.openstack.org/01/651601/1/check/openstack-tox-docs/845fa3c/html/ : SUCCESS in 5m 40s\n- kolla-build-ubuntu-binary http://logs.openstack.org/01/651601/1/check/kolla-build-ubuntu-binary/02a614e/ : SUCCESS in 43m 27s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/01/651601/1/check/kolla-build-ubuntu-source/1fcb700/ : SUCCESS in 31m 18s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/01/651601/1/check/kolla-ansible-ubuntu-source/325df64/ : SUCCESS in 56m 37s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/01/651601/1/check/kolla-ansible-ubuntu-binary/7912b50/ : FAILURE in 28m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/01/651601/1/check/kolla-ansible-ubuntu-source-upgrade/7f99d61/ : SUCCESS in 1h 03m 27s","accounts_in_message":[],"_revision_number":1},{"id":"740313cf46d5040cc62c6102265ab43f4cc44f64","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-04-22 06:53:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f55e9705524b58ccbd318cbd149d0df037767d86","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-22 07:01:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/01/651601/1/openstack-check/tripleo-build-containers-fedora-28-master/031da63/ : FAILURE in 5m 28s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/01/651601/1/openstack-check/tripleo-build-containers-fedora-28-stein/7f35543/ : FAILURE in 6m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"95920de71f2f6bdead6c5f915274fe25d301ac26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 08:33:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/01/651601/1/check/kolla-build-centos-binary/9271a8e/ : FAILURE in 7m 26s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/01/651601/1/check/kolla-build-centos-source/83b032f/ : FAILURE in 7m 14s\n- kolla-ansible-centos-source http://logs.openstack.org/01/651601/1/check/kolla-ansible-centos-source/2850029/ : FAILURE in 10m 39s\n- kolla-ansible-centos-binary http://logs.openstack.org/01/651601/1/check/kolla-ansible-centos-binary/592c22d/ : FAILURE in 11m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/01/651601/1/check/kolla-ansible-centos-source-upgrade/6b005f2/ : FAILURE in 9m 44s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/01/651601/1/check/tripleo-build-containers-centos-7/6683be0/ : FAILURE in 17m 37s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/01/651601/1/check/kolla-build-oraclelinux-binary/2dd4d81/ : FAILURE in 12m 54s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/01/651601/1/check/kolla-build-oraclelinux-source/13e8063/ : FAILURE in 10m 31s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/01/651601/1/check/kolla-ansible-oraclelinux-source/180db61/ : FAILURE in 11m 16s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/01/651601/1/check/kolla-ansible-oraclelinux-binary/9fa3b05/ : FAILURE in 10m 52s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/01/651601/1/check/openstack-tox-lower-constraints/ddd21e6/ : SUCCESS in 3m 51s\n- openstack-tox-pep8 http://logs.openstack.org/01/651601/1/check/openstack-tox-pep8/0547e30/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/01/651601/1/check/openstack-tox-py27/542b54c/ : SUCCESS in 3m 21s\n- openstack-tox-py36 http://logs.openstack.org/01/651601/1/check/openstack-tox-py36/c3bad77/ : SUCCESS in 3m 30s\n- openstack-tox-py37 http://logs.openstack.org/01/651601/1/check/openstack-tox-py37/f5c5379/ : SUCCESS in 3m 34s\n- openstack-tox-docs http://logs.openstack.org/01/651601/1/check/openstack-tox-docs/80db9f6/html/ : SUCCESS in 3m 28s\n- kolla-build-ubuntu-binary http://logs.openstack.org/01/651601/1/check/kolla-build-ubuntu-binary/f4dd52f/ : SUCCESS in 32m 42s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/01/651601/1/check/kolla-build-ubuntu-source/1f94e57/ : SUCCESS in 35m 23s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/01/651601/1/check/kolla-ansible-ubuntu-source/c93ad18/ : SUCCESS in 59m 37s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/01/651601/1/check/kolla-ansible-ubuntu-binary/67bfab8/ : SUCCESS in 48m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/01/651601/1/check/kolla-ansible-ubuntu-source-upgrade/1811367/ : SUCCESS in 1h 37m 39s","accounts_in_message":[],"_revision_number":1},{"id":"c2796d70765e1c89ac957fd8611b9b8830e05806","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-04-24 07:20:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"68ef83e3f572011cceee81687c510d13f2620489","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-24 08:01:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/01/651601/1/openstack-check/tripleo-build-containers-fedora-28-master/18c8e37/ : SUCCESS in 36m 40s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/01/651601/1/openstack-check/tripleo-build-containers-fedora-28-stein/e6d852d/ : SUCCESS in 37m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"84795dc8d74b257213d1b43b3d8b14ab3d7919ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 08:32:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/01/651601/1/check/kolla-build-centos-binary/c1e332b/ : FAILURE in 6m 53s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/01/651601/1/check/kolla-build-centos-source/dfb3a80/ : FAILURE in 7m 40s\n- kolla-ansible-centos-source http://logs.openstack.org/01/651601/1/check/kolla-ansible-centos-source/238ebb6/ : FAILURE in 9m 18s\n- kolla-ansible-centos-binary http://logs.openstack.org/01/651601/1/check/kolla-ansible-centos-binary/8520c2e/ : FAILURE in 9m 39s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/01/651601/1/check/kolla-ansible-centos-source-upgrade/628a170/ : FAILURE in 9m 12s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/01/651601/1/check/tripleo-build-containers-centos-7/536ea77/ : FAILURE in 14m 14s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/01/651601/1/check/kolla-build-oraclelinux-binary/acd6ad7/ : FAILURE in 8m 48s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/01/651601/1/check/kolla-build-oraclelinux-source/7e58765/ : FAILURE in 10m 05s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/01/651601/1/check/kolla-ansible-oraclelinux-source/286a67a/ : FAILURE in 12m 00s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/01/651601/1/check/kolla-ansible-oraclelinux-binary/cf87541/ : FAILURE in 11m 12s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/01/651601/1/check/openstack-tox-lower-constraints/378ec61/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 http://logs.openstack.org/01/651601/1/check/openstack-tox-pep8/d7fc6bf/ : SUCCESS in 2m 47s\n- openstack-tox-py27 http://logs.openstack.org/01/651601/1/check/openstack-tox-py27/d12f70d/ : SUCCESS in 2m 32s\n- openstack-tox-py36 http://logs.openstack.org/01/651601/1/check/openstack-tox-py36/b25fe0e/ : SUCCESS in 2m 50s\n- openstack-tox-py37 http://logs.openstack.org/01/651601/1/check/openstack-tox-py37/8758e12/ : SUCCESS in 3m 30s\n- openstack-tox-docs http://logs.openstack.org/01/651601/1/check/openstack-tox-docs/6158c0c/html/ : SUCCESS in 2m 40s\n- kolla-build-ubuntu-binary http://logs.openstack.org/01/651601/1/check/kolla-build-ubuntu-binary/a06e32d/ : SUCCESS in 33m 23s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/01/651601/1/check/kolla-build-ubuntu-source/50c61bb/ : SUCCESS in 34m 46s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/01/651601/1/check/kolla-ansible-ubuntu-source/ee08fb8/ : SUCCESS in 1h 01m 30s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/01/651601/1/check/kolla-ansible-ubuntu-binary/2c92b7f/ : SUCCESS in 53m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/01/651601/1/check/kolla-ansible-ubuntu-source-upgrade/1159fac/ : SUCCESS in 1h 05m 08s","accounts_in_message":[],"_revision_number":1},{"id":"6a1cd1bab06ebd13450f54a4681748fa7722ca74","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-04-29 06:42:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3a915cff6f6880818f75593d3958324677a5da13","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-29 07:08:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/01/651601/1/openstack-check/tripleo-build-containers-fedora-28-master/407c7e3/ : RETRY_LIMIT in 2m 24s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/01/651601/1/openstack-check/tripleo-build-containers-fedora-28-stein/f6e69af/ : RETRY_LIMIT in 3m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e0e983d4fcdf3e80d785f0ac201f15ad32572ef3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 07:47:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/01/651601/1/check/kolla-build-centos-binary/64389aa/ : FAILURE in 7m 06s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/01/651601/1/check/kolla-build-centos-source/2f6e610/ : FAILURE in 7m 15s\n- kolla-ansible-centos-source http://logs.openstack.org/01/651601/1/check/kolla-ansible-centos-source/f6eb181/ : FAILURE in 10m 57s\n- kolla-ansible-centos-binary http://logs.openstack.org/01/651601/1/check/kolla-ansible-centos-binary/1b10ec3/ : FAILURE in 10m 52s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/01/651601/1/check/kolla-ansible-centos-source-upgrade/3819ff8/ : FAILURE in 11m 16s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/01/651601/1/check/tripleo-build-containers-centos-7/2645cb6/ : FAILURE in 14m 12s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/01/651601/1/check/kolla-build-oraclelinux-binary/11928c6/ : FAILURE in 8m 28s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/01/651601/1/check/kolla-build-oraclelinux-source/df7c595/ : FAILURE in 11m 39s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/01/651601/1/check/kolla-ansible-oraclelinux-source/69c0c9c/ : FAILURE in 12m 26s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/01/651601/1/check/kolla-ansible-oraclelinux-binary/00cc443/ : FAILURE in 11m 20s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/01/651601/1/check/openstack-tox-lower-constraints/95b11bf/ : SUCCESS in 4m 26s\n- openstack-tox-pep8 http://logs.openstack.org/01/651601/1/check/openstack-tox-pep8/64a0bb2/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/01/651601/1/check/openstack-tox-py27/904012c/ : SUCCESS in 2m 49s\n- openstack-tox-py36 http://logs.openstack.org/01/651601/1/check/openstack-tox-py36/83c69c3/ : SUCCESS in 3m 38s\n- openstack-tox-py37 http://logs.openstack.org/01/651601/1/check/openstack-tox-py37/d712e80/ : SUCCESS in 3m 30s\n- openstack-tox-docs http://logs.openstack.org/01/651601/1/check/openstack-tox-docs/4d0ef36/html/ : SUCCESS in 2m 17s\n- kolla-build-ubuntu-binary http://logs.openstack.org/01/651601/1/check/kolla-build-ubuntu-binary/b6fe4c7/ : SUCCESS in 27m 13s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/01/651601/1/check/kolla-build-ubuntu-source/2ec837d/ : SUCCESS in 41m 44s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/01/651601/1/check/kolla-ansible-ubuntu-source/ec1e1db/ : SUCCESS in 43m 55s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/01/651601/1/check/kolla-ansible-ubuntu-binary/0dc7ebc/ : SUCCESS in 1h 04m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/01/651601/1/check/kolla-ansible-ubuntu-source-upgrade/0dd914f/ : SUCCESS in 57m 54s","accounts_in_message":[],"_revision_number":1},{"id":"9f5be70fa67a19a5a0bd71c2209912d8c8a4c96f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-08 15:15:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e732a82e254dc95b5c2a701d7fceb4d832abee42","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-08 15:50:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/01/651601/1/openstack-check/tripleo-build-containers-fedora-28-master/91e2b08/ : SUCCESS in 30m 49s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/01/651601/1/openstack-check/tripleo-build-containers-fedora-28-stein/7138ea8/ : SUCCESS in 33m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"299423ac40f324f7778289b20f5cf6e43342a185","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 16:37:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/01/651601/1/check/kolla-build-centos-binary/2a65632/ : FAILURE in 7m 09s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/01/651601/1/check/kolla-build-centos-source/9bfe4e2/ : FAILURE in 7m 55s\n- kolla-ansible-centos-source http://logs.openstack.org/01/651601/1/check/kolla-ansible-centos-source/8e3d7b5/ : FAILURE in 10m 36s\n- kolla-ansible-centos-binary http://logs.openstack.org/01/651601/1/check/kolla-ansible-centos-binary/6d92cb4/ : FAILURE in 11m 36s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/01/651601/1/check/kolla-ansible-centos-source-upgrade/e537916/ : FAILURE in 13m 23s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/01/651601/1/check/tripleo-build-containers-centos-7/84fa0f4/ : SUCCESS in 42m 31s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/01/651601/1/check/kolla-build-oraclelinux-binary/28c759c/ : FAILURE in 54m 48s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/01/651601/1/check/kolla-build-oraclelinux-source/91681f3/ : SUCCESS in 50m 28s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/01/651601/1/check/kolla-ansible-oraclelinux-source/0b3e7b2/ : SUCCESS in 1h 06m 40s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/01/651601/1/check/kolla-ansible-oraclelinux-binary/ce88131/ : SUCCESS in 1h 05m 30s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/01/651601/1/check/openstack-tox-lower-constraints/18b4dfa/ : SUCCESS in 3m 47s\n- openstack-tox-pep8 http://logs.openstack.org/01/651601/1/check/openstack-tox-pep8/44341a9/ : SUCCESS in 2m 51s\n- openstack-tox-py27 http://logs.openstack.org/01/651601/1/check/openstack-tox-py27/fab0696/ : SUCCESS in 3m 25s\n- openstack-tox-py36 http://logs.openstack.org/01/651601/1/check/openstack-tox-py36/ff8131b/ : SUCCESS in 3m 26s\n- openstack-tox-py37 http://logs.openstack.org/01/651601/1/check/openstack-tox-py37/f06a30a/ : SUCCESS in 3m 34s\n- openstack-tox-docs http://logs.openstack.org/01/651601/1/check/openstack-tox-docs/d8e74e6/html/ : SUCCESS in 3m 57s\n- kolla-build-ubuntu-binary http://logs.openstack.org/01/651601/1/check/kolla-build-ubuntu-binary/5d5667d/ : SUCCESS in 27m 21s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/01/651601/1/check/kolla-build-ubuntu-source/967216e/ : SUCCESS in 29m 06s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/01/651601/1/check/kolla-ansible-ubuntu-source/7233fb4/ : SUCCESS in 49m 40s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/01/651601/1/check/kolla-ansible-ubuntu-binary/c76b165/ : SUCCESS in 48m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/01/651601/1/check/kolla-ansible-ubuntu-source-upgrade/092c31e/ : SUCCESS in 1h 04m 54s","accounts_in_message":[],"_revision_number":1},{"id":"389892056b27b17047ac5d98e050b61363cc813b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-08 20:04:41.000000000","message":"Patch Set 1:\n\nINFO:kolla.common.utils.base:\u001b[91mError: centos-release-ceph-nautilus conflicts with centos-release-ceph-luminous-1.1-2.el7.centos.noarch\n\n:(","accounts_in_message":[],"_revision_number":1},{"id":"b6fd97ff064f03d5dde65e336a97a1e08b2f16b1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-05 11:01:05.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f2a5f575da80bd067601b8e2c40da70a0eef2966","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-05 11:01:20.000000000","message":"Patch Set 2:\n\nShould pass now that ceph upgrade has merged.","accounts_in_message":[],"_revision_number":2},{"id":"c4f82204bd16849d1821e7dd03e82192dd49702d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-05 11:39:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/01/651601/2/openstack-check/tripleo-build-containers-fedora-28-master/4335cdc/ : SUCCESS in 32m 13s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/01/651601/2/openstack-check/tripleo-build-containers-fedora-28-stein/949f427/ : SUCCESS in 36m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d659dd2c5edb1cb0d2da44e41315ad030da28bbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 12:27:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/01/651601/2/check/kolla-build-centos-binary/186978a/ : FAILURE in 7m 39s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/01/651601/2/check/kolla-build-centos-source/73912eb/ : FAILURE in 7m 19s\n- kolla-ansible-centos-source http://logs.openstack.org/01/651601/2/check/kolla-ansible-centos-source/f5b9e1e/ : FAILURE in 10m 21s\n- kolla-ansible-centos-binary http://logs.openstack.org/01/651601/2/check/kolla-ansible-centos-binary/19f40d0/ : FAILURE in 14m 21s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/01/651601/2/check/kolla-ansible-centos-source-upgrade/a6104c5/ : FAILURE in 9m 45s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/01/651601/2/check/tripleo-build-containers-centos-7/34f42c1/ : SUCCESS in 51m 20s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/01/651601/2/check/kolla-build-oraclelinux-binary/699d757/ : FAILURE in 40m 20s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/01/651601/2/check/kolla-build-oraclelinux-source/5ce4961/ : SUCCESS in 41m 05s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/01/651601/2/check/kolla-ansible-oraclelinux-source/881f33e/ : SUCCESS in 1h 10m 29s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/01/651601/2/check/kolla-ansible-oraclelinux-binary/126cd2f/ : SUCCESS in 1h 11m 29s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/01/651601/2/check/openstack-tox-lower-constraints/84b509e/ : SUCCESS in 3m 51s\n- openstack-tox-pep8 http://logs.openstack.org/01/651601/2/check/openstack-tox-pep8/e0a3999/ : SUCCESS in 2m 33s\n- openstack-tox-py27 http://logs.openstack.org/01/651601/2/check/openstack-tox-py27/1c5f430/ : SUCCESS in 3m 07s\n- openstack-tox-py36 http://logs.openstack.org/01/651601/2/check/openstack-tox-py36/7458123/ : SUCCESS in 2m 58s\n- openstack-tox-py37 http://logs.openstack.org/01/651601/2/check/openstack-tox-py37/d8359b0/ : SUCCESS in 3m 07s\n- openstack-tox-docs http://logs.openstack.org/01/651601/2/check/openstack-tox-docs/d1dc52e/html/ : SUCCESS in 2m 49s\n- kolla-build-ubuntu-binary http://logs.openstack.org/01/651601/2/check/kolla-build-ubuntu-binary/5b798db/ : SUCCESS in 38m 53s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/01/651601/2/check/kolla-build-ubuntu-source/9f05973/ : SUCCESS in 38m 33s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/01/651601/2/check/kolla-ansible-ubuntu-source/6e44004/ : SUCCESS in 51m 21s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/01/651601/2/check/kolla-ansible-ubuntu-binary/e1cd0d9/ : SUCCESS in 48m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/01/651601/2/check/kolla-ansible-ubuntu-source-upgrade/489413c/ : SUCCESS in 1h 00m 59s","accounts_in_message":[],"_revision_number":2},{"id":"c2b352b93712d67aecc9fe265479c435cabe23d4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-05 12:49:22.000000000","message":"Patch Set 2: Workflow-1\n\nAh, need ceph nautilus stein backport.","accounts_in_message":[],"_revision_number":2},{"id":"d5dd2139decebf241968f446e7c30a707f7af4bb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-05 13:20:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"141773c26a306cfe839f14bd0294845a0270e556","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-05 14:04:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/01/651601/3/openstack-check/tripleo-build-containers-fedora-28-master/012c55c/ : SUCCESS in 36m 02s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/01/651601/3/openstack-check/tripleo-build-containers-fedora-28-stein/e52b079/ : SUCCESS in 41m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4c3720dc30f238fd5d7ff636b85dd2aa56ce52ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 15:19:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/01/651601/3/check/kolla-build-centos-binary/7dea4be/ : FAILURE in 30m 54s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/01/651601/3/check/kolla-build-centos-source/78aabfc/ : FAILURE in 54m 40s\n- kolla-ansible-centos-source http://logs.openstack.org/01/651601/3/check/kolla-ansible-centos-source/e4c71e1/ : SUCCESS in 1h 22m 55s\n- kolla-ansible-centos-binary http://logs.openstack.org/01/651601/3/check/kolla-ansible-centos-binary/452aa77/ : SUCCESS in 1h 07m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/01/651601/3/check/kolla-ansible-centos-source-upgrade/a576f70/ : SUCCESS in 1h 21m 49s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/01/651601/3/check/tripleo-build-containers-centos-7/6b87cd4/ : SUCCESS in 48m 43s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/01/651601/3/check/kolla-build-oraclelinux-binary/e0311c2/ : FAILURE in 41m 29s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/01/651601/3/check/kolla-build-oraclelinux-source/427e210/ : SUCCESS in 49m 35s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/01/651601/3/check/kolla-ansible-oraclelinux-source/6412e14/ : FAILURE in 32m 15s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/01/651601/3/check/kolla-ansible-oraclelinux-binary/11da72f/ : SUCCESS in 1h 05m 55s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/01/651601/3/check/openstack-tox-lower-constraints/f48f9b3/ : SUCCESS in 4m 53s\n- openstack-tox-pep8 http://logs.openstack.org/01/651601/3/check/openstack-tox-pep8/9e7c1ff/ : SUCCESS in 3m 16s\n- openstack-tox-py27 http://logs.openstack.org/01/651601/3/check/openstack-tox-py27/301090a/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/01/651601/3/check/openstack-tox-py36/f0211bc/ : SUCCESS in 2m 54s\n- openstack-tox-py37 http://logs.openstack.org/01/651601/3/check/openstack-tox-py37/34edd40/ : SUCCESS in 3m 45s\n- openstack-tox-docs http://logs.openstack.org/01/651601/3/check/openstack-tox-docs/55786d4/html/ : SUCCESS in 2m 42s\n- kolla-build-ubuntu-binary http://logs.openstack.org/01/651601/3/check/kolla-build-ubuntu-binary/0ffb6f9/ : SUCCESS in 47m 44s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/01/651601/3/check/kolla-build-ubuntu-source/d336e77/ : SUCCESS in 49m 07s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/01/651601/3/check/kolla-ansible-ubuntu-source/eaf8f69/ : SUCCESS in 45m 47s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/01/651601/3/check/kolla-ansible-ubuntu-binary/904bc4e/ : SUCCESS in 1h 06m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/01/651601/3/check/kolla-ansible-ubuntu-source-upgrade/1ee6a55/ : SUCCESS in 1h 20m 47s","accounts_in_message":[],"_revision_number":3},{"id":"f100081203311cba9925618f9ce9a6f3a609c6e7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-06 08:16:17.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"fe6d1d5b07dd3b7bf687ebea9f70fda065ed9a39","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 08:59:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/01/651601/3/openstack-check/tripleo-build-containers-fedora-28-master/af833c8/ : SUCCESS in 34m 39s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/01/651601/3/openstack-check/tripleo-build-containers-fedora-28-stein/d2d7c1c/ : SUCCESS in 37m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6d4d012764891f3e45d404fb5da55c790f1ceb94","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-06 09:21:03.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"093c4d7e236882dbff56e5f462b40e0777c8fe68","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-06 09:22:19.000000000","message":"Patch Set 4:\n\nRebased due to merge conflicts.","accounts_in_message":[],"_revision_number":4},{"id":"4385dcd0353395aca3285a7ea9a23cf7a16b7a25","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 09:58:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/01/651601/4/openstack-check/tripleo-build-containers-fedora-28-master/187f239/ : SUCCESS in 36m 13s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/01/651601/4/openstack-check/tripleo-build-containers-fedora-28-stein/d2bd3d1/ : SUCCESS in 35m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dc7ba0eb80eb2c3dde8874032fc63fa0d19f545a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 10:47:28.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/01/651601/4/check/kolla-build-centos-binary/09ebf26/ : FAILURE in 41m 56s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/01/651601/4/check/kolla-build-centos-source/6765b76/ : FAILURE in 48m 09s\n- kolla-ansible-centos-source http://logs.openstack.org/01/651601/4/check/kolla-ansible-centos-source/0eac7a2/ : SUCCESS in 1h 12m 12s\n- kolla-ansible-centos-binary http://logs.openstack.org/01/651601/4/check/kolla-ansible-centos-binary/291d3f3/ : SUCCESS in 1h 03m 48s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/01/651601/4/check/kolla-ansible-centos-source-upgrade/b582751/ : SUCCESS in 1h 16m 53s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/01/651601/4/check/tripleo-build-containers-centos-7/209c405/ : SUCCESS in 45m 45s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/01/651601/4/check/kolla-build-oraclelinux-binary/a18b008/ : FAILURE in 53m 36s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/01/651601/4/check/kolla-build-oraclelinux-source/24268d5/ : SUCCESS in 54m 22s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/01/651601/4/check/kolla-ansible-oraclelinux-source/adf4af1/ : SUCCESS in 1h 15m 46s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/01/651601/4/check/kolla-ansible-oraclelinux-binary/35b3442/ : SUCCESS in 1h 04m 02s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/01/651601/4/check/openstack-tox-lower-constraints/a44fc4d/ : SUCCESS in 4m 11s\n- openstack-tox-pep8 http://logs.openstack.org/01/651601/4/check/openstack-tox-pep8/a0fd275/ : SUCCESS in 2m 46s\n- openstack-tox-py27 http://logs.openstack.org/01/651601/4/check/openstack-tox-py27/569aab6/ : SUCCESS in 2m 41s\n- openstack-tox-py36 http://logs.openstack.org/01/651601/4/check/openstack-tox-py36/d9ed8a0/ : SUCCESS in 2m 42s\n- openstack-tox-py37 http://logs.openstack.org/01/651601/4/check/openstack-tox-py37/a401adb/ : SUCCESS in 3m 13s\n- openstack-tox-docs http://logs.openstack.org/01/651601/4/check/openstack-tox-docs/4d81b73/html/ : SUCCESS in 2m 18s\n- kolla-build-ubuntu-binary http://logs.openstack.org/01/651601/4/check/kolla-build-ubuntu-binary/264be62/ : SUCCESS in 47m 11s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/01/651601/4/check/kolla-build-ubuntu-source/b09729b/ : SUCCESS in 34m 48s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/01/651601/4/check/kolla-ansible-ubuntu-source/c28e286/ : SUCCESS in 52m 24s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/01/651601/4/check/kolla-ansible-ubuntu-binary/c0e05f6/ : SUCCESS in 46m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/01/651601/4/check/kolla-ansible-ubuntu-source-upgrade/4bc2258/ : SUCCESS in 58m 11s","accounts_in_message":[],"_revision_number":4},{"id":"e77d86aa7a52abb32ab0ee5f214a607979b3f5f2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-06 13:19:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bf5b6c1cd3f99a45221d9a17334a6e9986555e76","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-06 13:27:30.000000000","message":"Patch Set 5: Code-Review-1\n\nforgot to disable repo for oraclelinux","accounts_in_message":[],"_revision_number":5},{"id":"948a4f18e7d771a42083a6d926c2a72e53e56538","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-06 13:33:08.000000000","message":"Patch Set 5: Code-Review+2\n\nafter discussion on irc","accounts_in_message":[],"_revision_number":5},{"id":"8fd02b4beff78608b9a1bc9e60b59b8fff9bbfb8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-06 13:33:28.000000000","message":"Patch Set 5:\n\n\u003e forgot to disable repo for oraclelinux\n\nOracle does this:\n\nrpm -Uvh --nodeps \u003cstein RDO RPM URL\u003e\n\nSo does not pull in ceph.","accounts_in_message":[],"_revision_number":5},{"id":"6c65ec2f8508f114eb6c6b772a13e5148eb3f9c5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 13:57:01.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/01/651601/5/openstack-check/tripleo-build-containers-fedora-28-master/9bced13/ : SUCCESS in 34m 42s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/01/651601/5/openstack-check/tripleo-build-containers-fedora-28-stein/31df80f/ : SUCCESS in 34m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4a94f96e8b4358f742b366908bc4203f1f388bbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 14:55:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/01/651601/5/check/kolla-build-centos-binary/b24bae1/ : SUCCESS in 43m 04s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/01/651601/5/check/kolla-build-centos-source/a43ee0c/ : SUCCESS in 56m 17s\n- kolla-ansible-centos-source http://logs.openstack.org/01/651601/5/check/kolla-ansible-centos-source/34b7f62/ : SUCCESS in 1h 08m 46s\n- kolla-ansible-centos-binary http://logs.openstack.org/01/651601/5/check/kolla-ansible-centos-binary/39cf591/ : SUCCESS in 1h 06m 15s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/01/651601/5/check/kolla-ansible-centos-source-upgrade/9cdb146/ : SUCCESS in 1h 17m 20s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/01/651601/5/check/tripleo-build-containers-centos-7/65bca65/ : SUCCESS in 50m 16s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/01/651601/5/check/kolla-build-oraclelinux-binary/f4038d2/ : SUCCESS in 58m 55s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/01/651601/5/check/kolla-build-oraclelinux-source/dae936d/ : SUCCESS in 46m 58s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/01/651601/5/check/kolla-ansible-oraclelinux-source/a71c636/ : SUCCESS in 1h 27m 33s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/01/651601/5/check/kolla-ansible-oraclelinux-binary/d528841/ : SUCCESS in 1h 11m 35s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/01/651601/5/check/openstack-tox-lower-constraints/48b2367/ : SUCCESS in 3m 56s\n- openstack-tox-pep8 http://logs.openstack.org/01/651601/5/check/openstack-tox-pep8/e78944f/ : SUCCESS in 4m 11s\n- openstack-tox-py27 http://logs.openstack.org/01/651601/5/check/openstack-tox-py27/d3c365b/ : SUCCESS in 3m 20s\n- openstack-tox-py36 http://logs.openstack.org/01/651601/5/check/openstack-tox-py36/2113a39/ : SUCCESS in 2m 51s\n- openstack-tox-py37 http://logs.openstack.org/01/651601/5/check/openstack-tox-py37/3951f62/ : SUCCESS in 3m 44s\n- openstack-tox-docs http://logs.openstack.org/01/651601/5/check/openstack-tox-docs/e123fae/html/ : SUCCESS in 3m 05s\n- kolla-build-ubuntu-binary http://logs.openstack.org/01/651601/5/check/kolla-build-ubuntu-binary/df2e805/ : SUCCESS in 1h 08m 21s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/01/651601/5/check/kolla-build-ubuntu-source/3bcbee8/ : SUCCESS in 55m 36s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/01/651601/5/check/kolla-ansible-ubuntu-source/b48f93d/ : SUCCESS in 1h 06m 49s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/01/651601/5/check/kolla-ansible-ubuntu-binary/c940d0b/ : SUCCESS in 1h 17m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/01/651601/5/check/kolla-ansible-ubuntu-source-upgrade/9466f50/ : SUCCESS in 1h 06m 47s","accounts_in_message":[],"_revision_number":5},{"id":"b00ceaa3539055ca3fedf8b23c7708fcb711f74f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-06 15:08:46.000000000","message":"Patch Set 5: Review-Priority+2","accounts_in_message":[],"_revision_number":5},{"id":"92e334120d76bffd3b309c9b43f65a0ddd61d8ba","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-07 07:04:14.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"96572c2ae1b6478bdc64694bd1c4913d700a9b48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 07:04:26.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"116f4618ae53d6e5f94825401aef8917afe80f3a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-07 07:10:02.000000000","message":"Patch Set 5: Code-Review+1\n\nGlad disabling the repo worked in this case.","accounts_in_message":[],"_revision_number":5},{"id":"79526e15cd22406ea5da287c45a9084d79d30f9c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 08:08:43.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source http://logs.openstack.org/01/651601/5/gate/kolla-build-centos-source/0c744b6/ : FAILURE in 15m 59s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/01/651601/5/gate/tripleo-build-containers-centos-7/84fb4c6/ : SUCCESS in 49m 29s\n- kolla-build-oraclelinux-source http://logs.openstack.org/01/651601/5/gate/kolla-build-oraclelinux-source/b515044/ : SUCCESS in 45m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/651601/5/gate/openstack-tox-lower-constraints/5114a24/ : SUCCESS in 5m 15s\n- openstack-tox-pep8 http://logs.openstack.org/01/651601/5/gate/openstack-tox-pep8/eebf34f/ : SUCCESS in 4m 25s\n- openstack-tox-py27 http://logs.openstack.org/01/651601/5/gate/openstack-tox-py27/53fdd3d/ : SUCCESS in 3m 01s\n- openstack-tox-py36 http://logs.openstack.org/01/651601/5/gate/openstack-tox-py36/020b5d7/ : SUCCESS in 4m 14s\n- openstack-tox-py37 http://logs.openstack.org/01/651601/5/gate/openstack-tox-py37/7f875b9/ : SUCCESS in 4m 23s\n- openstack-tox-docs http://logs.openstack.org/01/651601/5/gate/openstack-tox-docs/aecbccb/html/ : SUCCESS in 2m 55s\n- kolla-build-ubuntu-source http://logs.openstack.org/01/651601/5/gate/kolla-build-ubuntu-source/a5bbac4/ : SUCCESS in 29m 00s","accounts_in_message":[],"_revision_number":5},{"id":"9f7ef6e417af37fd14b8e75763f3640a8cfa7f7f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-07 08:12:03.000000000","message":"Patch Set 5:\n\nrecheck\n\nFailed connect to www.percona.com:443; Connection timed out","accounts_in_message":[],"_revision_number":5},{"id":"e1fe0d3538a68dfbe8b831253350188aeb97154d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-07 08:50:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/01/651601/5/openstack-check/tripleo-build-containers-fedora-28-master/446606a/ : SUCCESS in 37m 23s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/01/651601/5/openstack-check/tripleo-build-containers-fedora-28-stein/7a194ae/ : SUCCESS in 34m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"83d4dc935da5a014b853e84929039bd8d9ef2655","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 09:43:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/01/651601/5/check/kolla-build-centos-binary/fc0ecf1/ : SUCCESS in 38m 31s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/01/651601/5/check/kolla-build-centos-source/a1d86e5/ : SUCCESS in 53m 17s\n- kolla-ansible-centos-source http://logs.openstack.org/01/651601/5/check/kolla-ansible-centos-source/b4c6f8e/ : SUCCESS in 1h 05m 12s\n- kolla-ansible-centos-binary http://logs.openstack.org/01/651601/5/check/kolla-ansible-centos-binary/e8a9d95/ : SUCCESS in 1h 04m 41s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/01/651601/5/check/kolla-ansible-centos-source-upgrade/60fc884/ : SUCCESS in 1h 25m 55s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/01/651601/5/check/tripleo-build-containers-centos-7/46c39f9/ : SUCCESS in 52m 07s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/01/651601/5/check/kolla-build-oraclelinux-binary/7abc1dc/ : SUCCESS in 49m 17s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/01/651601/5/check/kolla-build-oraclelinux-source/57ddc70/ : SUCCESS in 37m 46s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/01/651601/5/check/kolla-ansible-oraclelinux-source/1fe2c50/ : SUCCESS in 1h 13m 35s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/01/651601/5/check/kolla-ansible-oraclelinux-binary/7e01fcc/ : SUCCESS in 1h 01m 49s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/01/651601/5/check/openstack-tox-lower-constraints/0c01146/ : SUCCESS in 4m 12s\n- openstack-tox-pep8 http://logs.openstack.org/01/651601/5/check/openstack-tox-pep8/57b6954/ : SUCCESS in 3m 26s\n- openstack-tox-py27 http://logs.openstack.org/01/651601/5/check/openstack-tox-py27/4eff4c7/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/01/651601/5/check/openstack-tox-py36/9cde080/ : SUCCESS in 3m 07s\n- openstack-tox-py37 http://logs.openstack.org/01/651601/5/check/openstack-tox-py37/c343774/ : SUCCESS in 3m 21s\n- openstack-tox-docs http://logs.openstack.org/01/651601/5/check/openstack-tox-docs/2f4ed56/html/ : SUCCESS in 2m 53s\n- kolla-build-ubuntu-binary http://logs.openstack.org/01/651601/5/check/kolla-build-ubuntu-binary/3932a06/ : SUCCESS in 31m 46s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/01/651601/5/check/kolla-build-ubuntu-source/4bd57e5/ : SUCCESS in 58m 18s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/01/651601/5/check/kolla-ansible-ubuntu-source/4c6583c/ : SUCCESS in 49m 37s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/01/651601/5/check/kolla-ansible-ubuntu-binary/26ad7ac/ : SUCCESS in 49m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/01/651601/5/check/kolla-ansible-ubuntu-source-upgrade/28bb1a0/ : SUCCESS in 58m 40s","accounts_in_message":[],"_revision_number":5},{"id":"7fb1916616b10cd7c1baa24c3313c1fabe9d8650","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 09:43:22.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"05bf43fbe910833733a2a29dc38c982a50b7ae28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 10:55:23.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/01/651601/5/gate/kolla-build-centos-source/f223b21/ : SUCCESS in 1h 02m 11s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/01/651601/5/gate/tripleo-build-containers-centos-7/2794a75/ : SUCCESS in 34m 09s\n- kolla-build-oraclelinux-source http://logs.openstack.org/01/651601/5/gate/kolla-build-oraclelinux-source/c9be691/ : SUCCESS in 43m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/651601/5/gate/openstack-tox-lower-constraints/857970b/ : SUCCESS in 3m 43s\n- openstack-tox-pep8 http://logs.openstack.org/01/651601/5/gate/openstack-tox-pep8/81b3380/ : SUCCESS in 3m 09s\n- openstack-tox-py27 http://logs.openstack.org/01/651601/5/gate/openstack-tox-py27/2ee93a3/ : SUCCESS in 3m 05s\n- openstack-tox-py36 http://logs.openstack.org/01/651601/5/gate/openstack-tox-py36/6f62d15/ : SUCCESS in 3m 26s\n- openstack-tox-py37 http://logs.openstack.org/01/651601/5/gate/openstack-tox-py37/239200c/ : SUCCESS in 3m 34s\n- openstack-tox-docs http://logs.openstack.org/01/651601/5/gate/openstack-tox-docs/5322ab8/html/ : SUCCESS in 2m 46s\n- kolla-build-ubuntu-source http://logs.openstack.org/01/651601/5/gate/kolla-build-ubuntu-source/c04e7dd/ : SUCCESS in 34m 24s","accounts_in_message":[],"_revision_number":5},{"id":"119adef65ebc89e164870920a3e247c4d786c74e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 10:55:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"02ec5084634bf14530a189222d1d68c05ec21fcb","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2019-06-18 13:02:50.000000000","message":"Patch Set 5:\n\nwe build a new centos repo for nfs-ganesha 2.8 and will make ceph/nautilus depend on the new nfs-ganesha repo\n\ncan you guys consider switching back to the centos repos for ceph/nautilus? and if so, add me to the submission so that I can follow things? :D","accounts_in_message":[],"_revision_number":5},{"id":"88460d455370081bf7e24385f6ee4dc622509321","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-08-19 14:14:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c5d2e9683d8c26e382356a1c1b6d431e78e0f398","revisions":{"7d42910db34d035dbd1ba6ce442644ca8a51037a":{"kind":"REWORK","_number":1,"created":"2019-04-10 16:29:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/651601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/651601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/651601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/651601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/651601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/651601/1"}}},"commit":{"parents":[{"commit":"77e1745046a507f2b1e503cc38e08f92fc8c0475","subject":"Update UPPER_CONSTRAINTS_FILE for stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77e1745046a507f2b1e503cc38e08f92fc8c0475"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-04-10 16:27:15.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-04-10 16:27:15.000000000","tz":60},"subject":"Switch to RDO Stein release on CentOS \u0026 Oracle","message":"Switch to RDO Stein release on CentOS \u0026 Oracle\n\nThis change switches from the Delorean yum repositories to the RDO Stein\nrelease repository.\n\nChange-Id: I1df5dd28adc65ff461a893b60c57d7b5da96d653\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7d42910db34d035dbd1ba6ce442644ca8a51037a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7d42910db34d035dbd1ba6ce442644ca8a51037a"}]},"branch":"refs/heads/stable/stein"},"c7efe626ffd720373f1ea258d6acbe2037ad2719":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-05 11:01:05.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/651601/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/651601/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/651601/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/651601/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/651601/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/651601/2"}}},"commit":{"parents":[{"commit":"49f40264f6c4b3cebbd138d97732d0c6c578408d","subject":"Merge \"Fix multiple mon deployment failures\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49f40264f6c4b3cebbd138d97732d0c6c578408d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-04-10 16:27:15.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-05 11:01:05.000000000","tz":0},"subject":"Switch to RDO Stein release on CentOS \u0026 Oracle","message":"Switch to RDO Stein release on CentOS \u0026 Oracle\n\nThis change switches from the Delorean yum repositories to the RDO Stein\nrelease repository.\n\nChange-Id: I1df5dd28adc65ff461a893b60c57d7b5da96d653\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c7efe626ffd720373f1ea258d6acbe2037ad2719"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c7efe626ffd720373f1ea258d6acbe2037ad2719"}]},"branch":"refs/heads/stable/stein"},"e7f6d47d7c69a00b56ce2a73adda1a455da455cf":{"kind":"REWORK","_number":3,"created":"2019-06-05 13:20:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/651601/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/651601/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/651601/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/651601/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/651601/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/651601/3"}}},"commit":{"parents":[{"commit":"e7e537d638b214975ac431fe444383987544fb13","subject":"Ceph: Update to Nautilus on CentOS and OracleLinux","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7e537d638b214975ac431fe444383987544fb13"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-04-10 16:27:15.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-05 13:15:35.000000000","tz":60},"subject":"Switch to RDO Stein release on CentOS \u0026 Oracle","message":"Switch to RDO Stein release on CentOS \u0026 Oracle\n\nThis change switches from the Delorean yum repositories to the RDO Stein\nrelease repository.\n\nChange-Id: I1df5dd28adc65ff461a893b60c57d7b5da96d653\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7f6d47d7c69a00b56ce2a73adda1a455da455cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7f6d47d7c69a00b56ce2a73adda1a455da455cf"}]},"branch":"refs/heads/stable/stein"},"550d330d0f0903d964cffd500b4d99ab369ac401":{"kind":"NO_CHANGE","_number":4,"created":"2019-06-06 09:21:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/651601/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/651601/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/651601/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/651601/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/651601/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/651601/4"}}},"commit":{"parents":[{"commit":"78f862b0d7e390b16a57770a6af8519c27f05631","subject":"Ceph: Update to Nautilus on CentOS and OracleLinux","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78f862b0d7e390b16a57770a6af8519c27f05631"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-04-10 16:27:15.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-06 09:20:39.000000000","tz":60},"subject":"Switch to RDO Stein release on CentOS \u0026 Oracle","message":"Switch to RDO Stein release on CentOS \u0026 Oracle\n\nThis change switches from the Delorean yum repositories to the RDO Stein\nrelease repository.\n\nChange-Id: I1df5dd28adc65ff461a893b60c57d7b5da96d653\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/550d330d0f0903d964cffd500b4d99ab369ac401"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/550d330d0f0903d964cffd500b4d99ab369ac401"}]},"branch":"refs/heads/stable/stein"},"c5d2e9683d8c26e382356a1c1b6d431e78e0f398":{"kind":"REWORK","_number":5,"created":"2019-06-06 13:19:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/651601/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/651601/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/651601/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/651601/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/651601/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/651601/5"}}},"commit":{"parents":[{"commit":"78f862b0d7e390b16a57770a6af8519c27f05631","subject":"Ceph: Update to Nautilus on CentOS and OracleLinux","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78f862b0d7e390b16a57770a6af8519c27f05631"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-04-10 16:27:15.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-06 13:19:24.000000000","tz":60},"subject":"Switch to RDO Stein release on CentOS \u0026 Oracle","message":"Switch to RDO Stein release on CentOS \u0026 Oracle\n\nThis change switches from the Delorean yum repositories to the RDO Stein\nrelease repository.\n\ncentos-release-openstack-stein pulls in centos-release-ceph-nautilus,\nwhich conflicts with upstream ceph packages. Disable the CentOS Ceph\nrepo to work around this.\n\nChange-Id: I1df5dd28adc65ff461a893b60c57d7b5da96d653\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5d2e9683d8c26e382356a1c1b6d431e78e0f398"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5d2e9683d8c26e382356a1c1b6d431e78e0f398"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
