)]}'
{"id":"openstack%2Fkolla~652606","triplet_id":"openstack%2Fkolla~master~Idbfc6741d371adcb06ed97d578d75d3bce862743","project":"openstack/kolla","branch":"master","topic":"bug/1824783","hashtags":[],"change_id":"Idbfc6741d371adcb06ed97d578d75d3bce862743","subject":"Fix multiple mon deployment failures","status":"MERGED","created":"2019-04-15 09:38:07.000000000","updated":"2019-05-23 01:25:31.000000000","submitted":"2019-05-22 11:22:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652606-1558524150089-16b933f0","meta_rev_id":"0b6ec3c1d7db613b4c1fcbc4f102638d36f81a7d","_number":652606,"virtual_id_number":652606,"owner":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"date":"2019-04-25 01:27:32.000000000","_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-05-22 11:22:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-18 11:49:40.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"all":[{"value":1,"date":"2019-04-29 02:53:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":2,"date":"2019-04-24 18:47:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-05-03 13:27:56.000000000","_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":1,"date":"2019-05-03 03:20:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":2,"date":"2019-05-22 10:20:32.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":1,"date":"2019-05-22 10:20:32.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-18 11:49:40.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-24 18:47:36.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-04-29 02:53:49.000000000","updated_by":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"reviewer":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"state":"REVIEWER"},{"updated":"2019-05-03 03:20:35.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-05-03 13:27:56.000000000","updated_by":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"reviewer":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"state":"REVIEWER"},{"updated":"2019-05-22 10:20:32.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-05-22 11:22:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c73f9df7c6f416fa9e6191f620886527b2613c6c","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-04-15 09:38:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d16a82b31096115f0864c99f25c0618700353bd1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-15 10:26:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/06/652606/1/openstack-check/tripleo-build-containers-fedora-28-master/337b2cf/ : SUCCESS in 36m 44s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/06/652606/1/openstack-check/tripleo-build-containers-fedora-28-stein/f7196fa/ : SUCCESS in 31m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6cc15326fcdf63ea4d1bf47b5b4de694a99eb68b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 11:21:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/06/652606/1/check/kolla-build-centos-binary/346e027/ : SUCCESS in 30m 30s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/06/652606/1/check/kolla-build-centos-source/145fbc2/ : SUCCESS in 51m 00s\n- kolla-ansible-centos-source http://logs.openstack.org/06/652606/1/check/kolla-ansible-centos-source/8fb6569/ : SUCCESS in 1h 07m 22s\n- kolla-ansible-centos-binary http://logs.openstack.org/06/652606/1/check/kolla-ansible-centos-binary/80d911c/ : SUCCESS in 59m 02s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/06/652606/1/check/kolla-ansible-centos-source-upgrade/2bd825b/ : SUCCESS in 1h 25m 30s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/06/652606/1/check/tripleo-build-containers-centos-7/74ef2a7/ : SUCCESS in 46m 01s\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/06/652606/1/check/kolla-ansible-centos-source-ceph/9fcabdd/ : SUCCESS in 1h 36m 24s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/06/652606/1/check/kolla-build-oraclelinux-binary/20c3bd0/ : SUCCESS in 52m 56s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/06/652606/1/check/kolla-build-oraclelinux-source/ad714ef/ : SUCCESS in 50m 41s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/06/652606/1/check/kolla-ansible-oraclelinux-source/c68308a/ : SUCCESS in 1h 06m 40s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/06/652606/1/check/kolla-ansible-oraclelinux-binary/5aed29b/ : SUCCESS in 1h 04m 08s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/06/652606/1/check/kolla-ansible-oraclelinux-source-ceph/500c852/ : SUCCESS in 1h 26m 31s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/06/652606/1/check/openstack-tox-lower-constraints/be69714/ : SUCCESS in 3m 33s\n- openstack-tox-pep8 http://logs.openstack.org/06/652606/1/check/openstack-tox-pep8/b0a29b5/ : SUCCESS in 3m 17s\n- openstack-tox-py27 http://logs.openstack.org/06/652606/1/check/openstack-tox-py27/3436c25/ : SUCCESS in 3m 33s\n- openstack-tox-py36 http://logs.openstack.org/06/652606/1/check/openstack-tox-py36/56f67e7/ : SUCCESS in 3m 29s\n- openstack-tox-py37 http://logs.openstack.org/06/652606/1/check/openstack-tox-py37/56b662e/ : SUCCESS in 3m 31s\n- openstack-tox-docs http://logs.openstack.org/06/652606/1/check/openstack-tox-docs/f54976d/html/ : SUCCESS in 3m 00s\n- kolla-build-ubuntu-binary http://logs.openstack.org/06/652606/1/check/kolla-build-ubuntu-binary/76562eb/ : SUCCESS in 25m 02s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/06/652606/1/check/kolla-build-ubuntu-source/0224c7d/ : SUCCESS in 36m 18s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/06/652606/1/check/kolla-ansible-ubuntu-source/3f691fa/ : SUCCESS in 52m 10s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/06/652606/1/check/kolla-ansible-ubuntu-binary/0bc72dd/ : FAILURE in 26m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/06/652606/1/check/kolla-ansible-ubuntu-source-upgrade/92736f4/ : SUCCESS in 1h 09m 07s\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/06/652606/1/check/kolla-ansible-ubuntu-source-ceph/ce2ad1b/ : SUCCESS in 1h 17m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b6249be591531ba5c47c7cf012c7cbde0d10e70e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-15 12:48:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fd75a4aeffd14b4aacb0d1c734758e33319ac2be","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-04-16 06:55:25.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\n\nYes, it seems that when adding mon, it is best to import monmap, I will look for a better approach.","accounts_in_message":[],"_revision_number":1},{"id":"56b1881a9db2f44e57f870269054c1d02249d851","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-04-18 05:22:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8be84168609272531d1e424e16cbb36e965aee58","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-04-18 05:24:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"11b1c5678a7e8005d7ecbe2ad34be41f903cbcf1","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-04-18 05:27:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0902abd1a61a9be00574e19dab5de774da80dbce","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-18 06:16:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/06/652606/4/openstack-check/tripleo-build-containers-fedora-28-master/7dda42b/ : SUCCESS in 45m 20s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/06/652606/4/openstack-check/tripleo-build-containers-fedora-28-stein/ee10b4c/ : SUCCESS in 34m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cd418a564f078f9488ceb150fd463013bf13a2cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 07:37:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/06/652606/4/check/kolla-build-centos-binary/527558a/ : SUCCESS in 36m 08s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/06/652606/4/check/kolla-build-centos-source/4b90e5a/ : SUCCESS in 43m 52s\n- kolla-ansible-centos-source http://logs.openstack.org/06/652606/4/check/kolla-ansible-centos-source/4d7456f/ : SUCCESS in 1h 05m 42s\n- kolla-ansible-centos-binary http://logs.openstack.org/06/652606/4/check/kolla-ansible-centos-binary/c618963/ : SUCCESS in 1h 05m 05s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/06/652606/4/check/kolla-ansible-centos-source-upgrade/f4902f4/ : SUCCESS in 1h 27m 43s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/06/652606/4/check/tripleo-build-containers-centos-7/ed4ac70/ : SUCCESS in 53m 20s\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/06/652606/4/check/kolla-ansible-centos-source-ceph/683d8c1/ : SUCCESS in 1h 29m 57s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/06/652606/4/check/kolla-build-oraclelinux-binary/d7d494e/ : SUCCESS in 56m 14s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/06/652606/4/check/kolla-build-oraclelinux-source/16db212/ : SUCCESS in 57m 48s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/06/652606/4/check/kolla-ansible-oraclelinux-source/1abd270/ : SUCCESS in 1h 20m 18s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/06/652606/4/check/kolla-ansible-oraclelinux-binary/693785d/ : SUCCESS in 1h 04m 07s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/06/652606/4/check/kolla-ansible-oraclelinux-source-ceph/bf2e50d/ : SUCCESS in 1h 26m 38s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/06/652606/4/check/openstack-tox-lower-constraints/567ab71/ : SUCCESS in 5m 18s\n- openstack-tox-pep8 http://logs.openstack.org/06/652606/4/check/openstack-tox-pep8/b6736bc/ : SUCCESS in 3m 16s\n- openstack-tox-py27 http://logs.openstack.org/06/652606/4/check/openstack-tox-py27/68e55ca/ : SUCCESS in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/06/652606/4/check/openstack-tox-py36/1712066/ : SUCCESS in 3m 52s\n- openstack-tox-py37 http://logs.openstack.org/06/652606/4/check/openstack-tox-py37/d6c674e/ : SUCCESS in 3m 53s\n- openstack-tox-docs http://logs.openstack.org/06/652606/4/check/openstack-tox-docs/ee9c922/html/ : SUCCESS in 3m 53s\n- kolla-build-ubuntu-binary http://logs.openstack.org/06/652606/4/check/kolla-build-ubuntu-binary/fbb5af4/ : SUCCESS in 33m 40s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/06/652606/4/check/kolla-build-ubuntu-source/96091d1/ : SUCCESS in 53m 59s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/06/652606/4/check/kolla-ansible-ubuntu-source/aca082a/ : SUCCESS in 1h 57m 06s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/06/652606/4/check/kolla-ansible-ubuntu-binary/2957b15/ : FAILURE in 30m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/06/652606/4/check/kolla-ansible-ubuntu-source-upgrade/b44f2f2/ : SUCCESS in 1h 04m 53s\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/06/652606/4/check/kolla-ansible-ubuntu-source-ceph/0f0d220/ : SUCCESS in 1h 19m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0ff5146f9c0700957ca3e5820180adc939ca51a0","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-04-18 10:20:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"28db6923b4ab514c2f735a41ce3a86815d40b4fd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-18 11:49:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/06/652606/5/openstack-check/tripleo-build-containers-fedora-28-master/18800e7/ : SUCCESS in 34m 35s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/06/652606/5/openstack-check/tripleo-build-containers-fedora-28-stein/4623166/ : SUCCESS in 33m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dc5f22eabf3b68c5ef7e28fb44679902370476cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 12:29:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/06/652606/5/check/kolla-build-centos-binary/cf2337e/ : SUCCESS in 44m 40s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/06/652606/5/check/kolla-build-centos-source/e1d2a0a/ : SUCCESS in 52m 22s\n- kolla-ansible-centos-source http://logs.openstack.org/06/652606/5/check/kolla-ansible-centos-source/8f8f283/ : SUCCESS in 1h 18m 10s\n- kolla-ansible-centos-binary http://logs.openstack.org/06/652606/5/check/kolla-ansible-centos-binary/d6c3094/ : SUCCESS in 1h 08m 48s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/06/652606/5/check/kolla-ansible-centos-source-upgrade/9b5ebab/ : SUCCESS in 1h 16m 45s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/06/652606/5/check/tripleo-build-containers-centos-7/ad03380/ : SUCCESS in 39m 12s\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/06/652606/5/check/kolla-ansible-centos-source-ceph/4e8f6a1/ : SUCCESS in 1h 34m 22s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/06/652606/5/check/kolla-build-oraclelinux-binary/30f7f47/ : SUCCESS in 50m 48s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/06/652606/5/check/kolla-build-oraclelinux-source/bd85534/ : SUCCESS in 57m 52s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/06/652606/5/check/kolla-ansible-oraclelinux-source/ff9f9e0/ : SUCCESS in 1h 14m 05s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/06/652606/5/check/kolla-ansible-oraclelinux-binary/e5c10f7/ : SUCCESS in 1h 21m 55s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/06/652606/5/check/kolla-ansible-oraclelinux-source-ceph/ba3c5bd/ : SUCCESS in 1h 31m 28s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/06/652606/5/check/openstack-tox-lower-constraints/3322870/ : SUCCESS in 3m 57s\n- openstack-tox-pep8 http://logs.openstack.org/06/652606/5/check/openstack-tox-pep8/f60561e/ : SUCCESS in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/06/652606/5/check/openstack-tox-py27/76328be/ : SUCCESS in 2m 59s\n- openstack-tox-py36 http://logs.openstack.org/06/652606/5/check/openstack-tox-py36/18dd52e/ : SUCCESS in 2m 58s\n- openstack-tox-py37 http://logs.openstack.org/06/652606/5/check/openstack-tox-py37/203031c/ : SUCCESS in 3m 58s\n- openstack-tox-docs http://logs.openstack.org/06/652606/5/check/openstack-tox-docs/d33bf2c/html/ : SUCCESS in 2m 57s\n- kolla-build-ubuntu-binary http://logs.openstack.org/06/652606/5/check/kolla-build-ubuntu-binary/a2d0ec8/ : SUCCESS in 31m 12s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/06/652606/5/check/kolla-build-ubuntu-source/c1628fa/ : SUCCESS in 32m 06s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/06/652606/5/check/kolla-ansible-ubuntu-source/4e3a930/ : SUCCESS in 1h 03m 04s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/06/652606/5/check/kolla-ansible-ubuntu-binary/2bb1214/ : SUCCESS in 47m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/06/652606/5/check/kolla-ansible-ubuntu-source-upgrade/ee9e97d/ : SUCCESS in 1h 02m 35s\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/06/652606/5/check/kolla-ansible-ubuntu-source-ceph/193c9c1/ : SUCCESS in 1h 13m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1bd80576898355ae77c48867d960adcd68eb9e48","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-04-22 02:38:39.000000000","message":"Patch Set 5:\n\n\u003e (1 comment)\n\nhi, after comparing with ceph-deploy, I found that when the cluster was\ninitialized, it did not use monmap. When adding a new mon node, it\nneeds to use monmap. According to the above situation, I fixed the\nproblem.","accounts_in_message":[],"_revision_number":5},{"id":"2bcbdb6355ae91e4c2f9348796d8029980764c4d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-04-24 18:47:36.000000000","message":"Patch Set 5: Code-Review+2\n\nlooks better now","accounts_in_message":[],"_revision_number":5},{"id":"3177fdd3ece921f8ff709d7841736e8d885a588e","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-04-25 01:27:32.000000000","message":"Patch Set 5:\n\n\u003e looks better now\n\nThanks for your review.^^","accounts_in_message":[],"_revision_number":5},{"id":"548a6ba5ed457b095c64ae07c1e30af98bdef379","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-04-29 02:53:49.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"0597605d7134d15e1a40b5dd979ed493154fd226","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-05-03 03:20:35.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"708f9cefe342243d115449ff3a8ae97314d1f886","author":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"date":"2019-05-03 13:27:56.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"02944c979f5c145ffd07717b499f0a5d8a4c68bc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-22 10:20:32.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7a3606f0cefcce72e357a3066e7a20bfe31d17d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 10:20:45.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"507a5e986deac53535db714ac11f1ec61515aea4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 11:22:29.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/06/652606/5/gate/kolla-build-centos-source/c855636/ : SUCCESS in 41m 08s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/06/652606/5/gate/tripleo-build-containers-centos-7/4520058/ : SUCCESS in 54m 42s\n- kolla-build-oraclelinux-source http://logs.openstack.org/06/652606/5/gate/kolla-build-oraclelinux-source/6dcb508/ : SUCCESS in 55m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/06/652606/5/gate/openstack-tox-lower-constraints/fd586b8/ : SUCCESS in 5m 12s\n- openstack-tox-pep8 http://logs.openstack.org/06/652606/5/gate/openstack-tox-pep8/bec9eb1/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/06/652606/5/gate/openstack-tox-py27/11057b9/ : SUCCESS in 4m 32s\n- openstack-tox-py36 http://logs.openstack.org/06/652606/5/gate/openstack-tox-py36/4118dee/ : SUCCESS in 3m 03s\n- openstack-tox-py37 http://logs.openstack.org/06/652606/5/gate/openstack-tox-py37/7db64bd/ : SUCCESS in 4m 48s\n- openstack-tox-docs http://logs.openstack.org/06/652606/5/gate/openstack-tox-docs/654179f/html/ : SUCCESS in 3m 37s\n- kolla-build-ubuntu-source http://logs.openstack.org/06/652606/5/gate/kolla-build-ubuntu-source/ad04778/ : SUCCESS in 35m 44s","accounts_in_message":[],"_revision_number":5},{"id":"fa4bef64dd45909099679b163d236f67f8ea4c4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 11:22:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"894ce672ca84f5d908fd3d07943266d2d49b521b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-22 13:04:01.000000000","message":"Patch Set 5:\n\nDoes this need to be backported?","accounts_in_message":[],"_revision_number":5},{"id":"fb3bbae12f83d995a3f6f8287bf35f4db7f6cbc9","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-05-23 01:18:03.000000000","message":"Patch Set 5:\n\n\u003e Does this need to be backported?\n\nyes, I will do it.","accounts_in_message":[],"_revision_number":5},{"id":"1a3069092459c433ca5e063c0f44fb845e580c46","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-05-23 01:19:02.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 877df152ddc6dcf83e5a92964a275de2dfc1444b","accounts_in_message":[],"_revision_number":5},{"id":"8b01e9f0217fa84b2d1d05dc180c34cad8e93a4a","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-05-23 01:25:31.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 63d63ab97c87d4470be1cd72575869c32e1cb9c9","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"285814353af29395932e15d78c6a154796978a21","revisions":{"ee64718187d8301ac8f9630d4694676b2bdce2a8":{"kind":"REWORK","_number":1,"created":"2019-04-15 09:38:07.000000000","uploader":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"ref":"refs/changes/06/652606/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/652606/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/652606/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/652606/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/652606/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/652606/1"}}},"commit":{"parents":[{"commit":"32693b83385c3b76ba0edf7e426bb110c6c335ae","subject":"Merge \"Add config option use_dumb_init to add it to container images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32693b83385c3b76ba0edf7e426bb110c6c335ae"}]}],"author":{"name":"wangwei","email":"wangwei@navercorp.com","date":"2019-04-15 09:36:40.000000000","tz":540},"committer":{"name":"wangwei","email":"wangwei@navercorp.com","date":"2019-04-15 09:36:40.000000000","tz":540},"subject":"Fix multiple mon deployment failures","message":"Fix multiple mon deployment failures\n\nWhen 4 or 5 mons are deployed, all mons will always be in the election\nstate, and the ceph cluster cannot be accessed.\n\nCompared with ceph-deploy, I found that the initialization command has\none more parameter \u0027--monmap\u0027 than ceph-deploy. When this parameter is\nremoved, the deployment of mons will return to normal.\n\nChange-Id: Idbfc6741d371adcb06ed97d578d75d3bce862743\nCloses-bug: #1824783\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee64718187d8301ac8f9630d4694676b2bdce2a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee64718187d8301ac8f9630d4694676b2bdce2a8"}]},"branch":"refs/heads/master"},"2881b4dae5ebc274798a311b9bbdeea32fa8895a":{"kind":"REWORK","_number":2,"created":"2019-04-18 05:22:41.000000000","uploader":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"ref":"refs/changes/06/652606/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/652606/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/652606/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/652606/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/652606/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/652606/2"}}},"commit":{"parents":[{"commit":"32693b83385c3b76ba0edf7e426bb110c6c335ae","subject":"Merge \"Add config option use_dumb_init to add it to container images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32693b83385c3b76ba0edf7e426bb110c6c335ae"}]}],"author":{"name":"wangwei","email":"wangwei@navercorp.com","date":"2019-04-15 09:36:40.000000000","tz":540},"committer":{"name":"wangwei","email":"wangwei@navercorp.com","date":"2019-04-18 05:15:43.000000000","tz":540},"subject":"Fix multiple mon deployment failures","message":"Fix multiple mon deployment failures\n\nWhen 4 or 5 mons are deployed, all mons will always be in the election\nstate, and the ceph cluster cannot be accessed.\n\nAfter comparing with ceph-deploy, I found that when the cluster was\ninitialized, it did not use monmap. When adding a new mon node, it\nneeds to use monmap. According to the above situation, I fixed the\nproblem.\n\nChange-Id: Idbfc6741d371adcb06ed97d578d75d3bce862743\nCloses-bug: #1824783\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2881b4dae5ebc274798a311b9bbdeea32fa8895a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2881b4dae5ebc274798a311b9bbdeea32fa8895a"}]},"branch":"refs/heads/master"},"a223a1349206e60ac53fe234e1dbeb0205b5e93c":{"kind":"REWORK","_number":3,"created":"2019-04-18 05:24:26.000000000","uploader":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"ref":"refs/changes/06/652606/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/652606/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/652606/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/652606/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/652606/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/652606/3"}}},"commit":{"parents":[{"commit":"32693b83385c3b76ba0edf7e426bb110c6c335ae","subject":"Merge \"Add config option use_dumb_init to add it to container images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32693b83385c3b76ba0edf7e426bb110c6c335ae"}]}],"author":{"name":"wangwei","email":"wangwei@navercorp.com","date":"2019-04-15 09:36:40.000000000","tz":540},"committer":{"name":"wangwei","email":"wangwei@navercorp.com","date":"2019-04-18 05:24:12.000000000","tz":540},"subject":"Fix multiple mon deployment failures","message":"Fix multiple mon deployment failures\n\nWhen 4 or 5 mons are deployed, all mons will always be in the election\nstate, and the ceph cluster cannot be accessed.\n\nAfter comparing with ceph-deploy, I found that when the cluster was\ninitialized, it did not use monmap. When adding a new mon node, it\nneeds to use monmap. According to the above situation, I fixed the\nproblem.\n\nChange-Id: Idbfc6741d371adcb06ed97d578d75d3bce862743\nCloses-bug: #1824783\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a223a1349206e60ac53fe234e1dbeb0205b5e93c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a223a1349206e60ac53fe234e1dbeb0205b5e93c"}]},"branch":"refs/heads/master"},"79d13f2abacd40fb183858603d0699706a9b5df5":{"kind":"REWORK","_number":4,"created":"2019-04-18 05:27:14.000000000","uploader":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"ref":"refs/changes/06/652606/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/652606/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/652606/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/652606/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/652606/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/652606/4"}}},"commit":{"parents":[{"commit":"32693b83385c3b76ba0edf7e426bb110c6c335ae","subject":"Merge \"Add config option use_dumb_init to add it to container images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32693b83385c3b76ba0edf7e426bb110c6c335ae"}]}],"author":{"name":"wangwei","email":"wangwei@navercorp.com","date":"2019-04-15 09:36:40.000000000","tz":540},"committer":{"name":"wangwei","email":"wangwei@navercorp.com","date":"2019-04-18 05:26:58.000000000","tz":540},"subject":"Fix multiple mon deployment failures","message":"Fix multiple mon deployment failures\n\nWhen 4 or 5 mons are deployed, all mons will always be in the election\nstate, and the ceph cluster cannot be accessed.\n\nAfter comparing with ceph-deploy, I found that when the cluster was\ninitialized, it did not use monmap. When adding a new mon node, it\nneeds to use monmap. According to the above situation, I fixed the\nproblem.\n\nChange-Id: Idbfc6741d371adcb06ed97d578d75d3bce862743\nCloses-bug: #1824783\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79d13f2abacd40fb183858603d0699706a9b5df5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79d13f2abacd40fb183858603d0699706a9b5df5"}]},"branch":"refs/heads/master"},"285814353af29395932e15d78c6a154796978a21":{"kind":"REWORK","_number":5,"created":"2019-04-18 10:20:11.000000000","uploader":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"ref":"refs/changes/06/652606/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/652606/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/652606/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/652606/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/652606/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/652606/5"}}},"commit":{"parents":[{"commit":"32693b83385c3b76ba0edf7e426bb110c6c335ae","subject":"Merge \"Add config option use_dumb_init to add it to container images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32693b83385c3b76ba0edf7e426bb110c6c335ae"}]}],"author":{"name":"wangwei","email":"wangwei@navercorp.com","date":"2019-04-15 09:36:40.000000000","tz":540},"committer":{"name":"wangwei","email":"wangwei@navercorp.com","date":"2019-04-18 10:19:59.000000000","tz":540},"subject":"Fix multiple mon deployment failures","message":"Fix multiple mon deployment failures\n\nWhen 4 or 5 mons are deployed, all mons will always be in the election\nstate, and the ceph cluster cannot be accessed.\n\nAfter comparing with ceph-deploy, I found that when the cluster was\ninitialized, it did not use monmap. When adding a new mon node, it\nneeds to use monmap. According to the above situation, I fixed the\nproblem.\n\nChange-Id: Idbfc6741d371adcb06ed97d578d75d3bce862743\nCloses-bug: #1824783\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/285814353af29395932e15d78c6a154796978a21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/285814353af29395932e15d78c6a154796978a21"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
