)]}'
{"id":"openstack%2Fkolla~652937","triplet_id":"openstack%2Fkolla~master~I402f58ce53fdce426582cf2fc31349da6d2f5ae0","project":"openstack/kolla","branch":"master","topic":"tempest","hashtags":[],"change_id":"I402f58ce53fdce426582cf2fc31349da6d2f5ae0","subject":"Reuse /var/lib/tempest symlink for rpm also","status":"MERGED","created":"2019-04-16 08:01:46.000000000","updated":"2019-05-06 05:26:28.000000000","submitted":"2019-04-29 14:30:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652937-1556548249954-b3e99d47","meta_rev_id":"9d722f6dc5a4cf6068e2362ce6e00c7c97dbacf6","_number":652937,"virtual_id_number":652937,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2019-04-25 12:02:49.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-04-29 14:30:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-17 07:30:43.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"value":0,"_account_id":8944,"name":"Chandan Kumar","username":"Chandankumar","inactive":true},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":22582,"name":"Duong Ha-Quang","email":"quangduongha@gmail.com","username":"duonghq"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":1,"date":"2019-04-29 03:04:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":1,"date":"2019-04-25 11:44:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":2,"date":"2019-04-29 13:17:37.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2019-04-25 12:20:47.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"value":0,"_account_id":8944,"name":"Chandan Kumar","username":"Chandankumar","inactive":true},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":22582,"name":"Duong Ha-Quang","email":"quangduongha@gmail.com","username":"duonghq"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":1,"date":"2019-04-29 13:17:37.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"value":0,"_account_id":8944,"name":"Chandan Kumar","username":"Chandankumar","inactive":true},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":22582,"name":"Duong Ha-Quang","email":"quangduongha@gmail.com","username":"duonghq"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"value":0,"_account_id":8944,"name":"Chandan Kumar","username":"Chandankumar","inactive":true},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":22582,"name":"Duong Ha-Quang","email":"quangduongha@gmail.com","username":"duonghq"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-04-25 11:45:10.000000000","_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"date":"2019-04-25 11:44:57.000000000","_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"date":"2019-04-25 11:45:10.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2019-04-25 11:45:10.000000000","_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-17 06:31:20.000000000","_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"value":0,"_account_id":8944,"name":"Chandan Kumar","username":"Chandankumar","inactive":true},{"value":0,"date":"2019-04-25 11:45:10.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"date":"2019-04-25 11:45:10.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"date":"2019-04-25 11:45:10.000000000","_account_id":22582,"name":"Duong Ha-Quang","email":"quangduongha@gmail.com","username":"duonghq"},{"value":0,"date":"2019-04-17 06:31:14.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"date":"2019-04-25 11:45:10.000000000","_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"_account_id":8944,"name":"Chandan Kumar","username":"Chandankumar","inactive":true},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22582,"name":"Duong Ha-Quang","email":"quangduongha@gmail.com","username":"duonghq"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-16 08:01:46.000000000","updated_by":{"_account_id":8944,"name":"Chandan Kumar","username":"Chandankumar","inactive":true},"reviewer":{"_account_id":8944,"name":"Chandan Kumar","username":"Chandankumar","inactive":true},"state":"REVIEWER"},{"updated":"2019-04-17 06:31:14.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-04-17 06:31:20.000000000","updated_by":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"reviewer":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"state":"REVIEWER"},{"updated":"2019-04-17 07:30:43.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-25 11:44:29.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-04-25 11:44:57.000000000","updated_by":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"reviewer":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"state":"REVIEWER"},{"updated":"2019-04-25 11:45:10.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2019-04-25 11:45:10.000000000","updated_by":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"reviewer":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"state":"REVIEWER"},{"updated":"2019-04-25 11:45:10.000000000","updated_by":{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},"reviewer":{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},"state":"REVIEWER"},{"updated":"2019-04-25 11:45:10.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-04-25 11:45:10.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-04-25 11:45:10.000000000","updated_by":{"_account_id":22582,"name":"Duong Ha-Quang","email":"quangduongha@gmail.com","username":"duonghq"},"reviewer":{"_account_id":22582,"name":"Duong Ha-Quang","email":"quangduongha@gmail.com","username":"duonghq"},"state":"REVIEWER"},{"updated":"2019-04-25 11:45:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-04-25 12:20:47.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-04-29 03:04:29.000000000","updated_by":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"reviewer":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"state":"REVIEWER"},{"updated":"2019-04-29 13:17:37.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-04-29 14:30:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"489edd5ba0433eea1574c71b2f7630d4e2c85257","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-16 08:01:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"030b0878d95a6a28cd9c82688e87681c6f0257f6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-16 08:37:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/37/652937/1/openstack-check/tripleo-build-containers-fedora-28-master/cfa423e/ : SUCCESS in 30m 59s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/37/652937/1/openstack-check/tripleo-build-containers-fedora-28-stein/3bc1be2/ : SUCCESS in 32m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0abb88c91f82e58121dbfc8be984eb1801cf9ae5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 09:48:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/37/652937/1/check/kolla-build-centos-binary/0e7d642/ : SUCCESS in 41m 57s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/37/652937/1/check/kolla-build-centos-source/172c59a/ : SUCCESS in 53m 17s\n- kolla-ansible-centos-source http://logs.openstack.org/37/652937/1/check/kolla-ansible-centos-source/4f69ced/ : SUCCESS in 1h 06m 42s\n- kolla-ansible-centos-binary http://logs.openstack.org/37/652937/1/check/kolla-ansible-centos-binary/22c697b/ : SUCCESS in 1h 01m 39s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/37/652937/1/check/kolla-ansible-centos-source-upgrade/b48c010/ : SUCCESS in 1h 17m 15s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/37/652937/1/check/tripleo-build-containers-centos-7/280de1c/ : SUCCESS in 50m 31s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/37/652937/1/check/kolla-build-oraclelinux-binary/d11e1c4/ : SUCCESS in 56m 51s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/37/652937/1/check/kolla-build-oraclelinux-source/8742a98/ : SUCCESS in 48m 39s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/37/652937/1/check/kolla-ansible-oraclelinux-source/1488009/ : SUCCESS in 1h 11m 07s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/37/652937/1/check/kolla-ansible-oraclelinux-binary/20289f9/ : SUCCESS in 1h 05m 58s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/37/652937/1/check/openstack-tox-lower-constraints/fe01abb/ : SUCCESS in 4m 37s\n- openstack-tox-pep8 http://logs.openstack.org/37/652937/1/check/openstack-tox-pep8/faf0afb/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/37/652937/1/check/openstack-tox-py27/0fc468a/ : SUCCESS in 3m 21s\n- openstack-tox-py36 http://logs.openstack.org/37/652937/1/check/openstack-tox-py36/8283513/ : SUCCESS in 3m 24s\n- openstack-tox-py37 http://logs.openstack.org/37/652937/1/check/openstack-tox-py37/f8f68d2/ : SUCCESS in 3m 36s\n- openstack-tox-docs http://logs.openstack.org/37/652937/1/check/openstack-tox-docs/d53016e/html/ : SUCCESS in 2m 50s\n- kolla-build-ubuntu-binary http://logs.openstack.org/37/652937/1/check/kolla-build-ubuntu-binary/7df7ddb/ : SUCCESS in 31m 49s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/37/652937/1/check/kolla-build-ubuntu-source/fc6fae1/ : SUCCESS in 39m 24s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/37/652937/1/check/kolla-ansible-ubuntu-source/e384957/ : SUCCESS in 56m 15s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/37/652937/1/check/kolla-ansible-ubuntu-binary/cd1c81a/ : FAILURE in 25m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/37/652937/1/check/kolla-ansible-ubuntu-source-upgrade/9ca30f9/ : SUCCESS in 1h 05m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e463baead5ddb49700e21374f8932202e3756d87","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-17 06:30:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f5b572cc771eaabc2a2fdcd27627d2d63df722e6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-17 07:30:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/37/652937/2/openstack-check/tripleo-build-containers-fedora-28-master/375d354/ : SUCCESS in 35m 06s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/37/652937/2/openstack-check/tripleo-build-containers-fedora-28-stein/224c479/ : SUCCESS in 31m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4bd95c4727e50be7d4cf290a536b366bc009fb68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 08:06:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/37/652937/2/check/kolla-build-centos-binary/2b42136/ : SUCCESS in 45m 05s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/37/652937/2/check/kolla-build-centos-source/61dfa1d/ : SUCCESS in 49m 42s\n- kolla-ansible-centos-source http://logs.openstack.org/37/652937/2/check/kolla-ansible-centos-source/c031fcf/ : SUCCESS in 1h 08m 25s\n- kolla-ansible-centos-binary http://logs.openstack.org/37/652937/2/check/kolla-ansible-centos-binary/52bd9a6/ : SUCCESS in 1h 03m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/37/652937/2/check/kolla-ansible-centos-source-upgrade/8ae7ab3/ : SUCCESS in 1h 26m 58s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/37/652937/2/check/tripleo-build-containers-centos-7/aa9f7e0/ : SUCCESS in 50m 32s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/37/652937/2/check/kolla-build-oraclelinux-binary/9b25513/ : SUCCESS in 57m 11s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/37/652937/2/check/kolla-build-oraclelinux-source/31dccb4/ : SUCCESS in 38m 32s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/37/652937/2/check/kolla-ansible-oraclelinux-source/8afb314/ : SUCCESS in 1h 23m 15s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/37/652937/2/check/kolla-ansible-oraclelinux-binary/a45e0cf/ : SUCCESS in 1h 11m 57s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/37/652937/2/check/openstack-tox-lower-constraints/38aff29/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 http://logs.openstack.org/37/652937/2/check/openstack-tox-pep8/7e852cc/ : SUCCESS in 4m 11s\n- openstack-tox-py27 http://logs.openstack.org/37/652937/2/check/openstack-tox-py27/78791bd/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/37/652937/2/check/openstack-tox-py36/eba338b/ : SUCCESS in 2m 58s\n- openstack-tox-py37 http://logs.openstack.org/37/652937/2/check/openstack-tox-py37/4ef816f/ : SUCCESS in 3m 54s\n- openstack-tox-docs http://logs.openstack.org/37/652937/2/check/openstack-tox-docs/ddf4d29/html/ : SUCCESS in 2m 46s\n- kolla-build-ubuntu-binary http://logs.openstack.org/37/652937/2/check/kolla-build-ubuntu-binary/7fd95eb/ : SUCCESS in 1h 10m 27s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/37/652937/2/check/kolla-build-ubuntu-source/beaa632/ : SUCCESS in 1h 15m 41s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/37/652937/2/check/kolla-ansible-ubuntu-source/1efd423/ : SUCCESS in 1h 08m 48s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/37/652937/2/check/kolla-ansible-ubuntu-binary/21a1647/ : FAILURE in 1h 06m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/37/652937/2/check/kolla-ansible-ubuntu-source-upgrade/729ace9/ : SUCCESS in 1h 30m 25s","accounts_in_message":[],"_revision_number":2},{"id":"82b3a3d8bb03e35231f29d3317405b8d5cd4842d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-04-18 11:58:05.000000000","message":"Patch Set 2: Code-Review+1\n\ndoes it affect kolla-ansible?","accounts_in_message":[],"_revision_number":2},{"id":"9d6dd423cf5622c38decd6a4ba3f910546e8e3ef","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-04-25 11:44:29.000000000","message":"Patch Set 2: Code-Review+1\n\nyes /usr/share/openstack-tempest is not provided by tempest package, so good to get rid off.","accounts_in_message":[],"_revision_number":2},{"id":"c04fba46cd790873b4bfe458cda9e3b26f402947","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-25 12:02:49.000000000","message":"Patch Set 2:\n\n\u003e does it affect kolla-ansible?\n\nNope, It is just related to RDO packaging.","accounts_in_message":[],"_revision_number":2},{"id":"119b43e197be6f9c81f5cdf16e2bbbdb5d4369dc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-25 12:20:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"94315fa92a0681b844b3f736e9afa4ec167efdd6","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-04-29 03:04:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"53cd786d595a640159e118fa5fe1bed119929dd4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-04-29 13:17:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b634d86d7ada8aa2e0203b4ecf457fa88d6ec542","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 13:17:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6cc3e53b379d8d1cebe6daa2be321267a23e28b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 14:30:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b22a7cf1a357386d55d32c661c6d52f0337aa497","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 14:30:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/37/652937/2/gate/kolla-build-centos-source/48aa4d1/ : SUCCESS in 57m 03s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/37/652937/2/gate/tripleo-build-containers-centos-7/b2d5e58/ : SUCCESS in 53m 18s\n- kolla-build-oraclelinux-source http://logs.openstack.org/37/652937/2/gate/kolla-build-oraclelinux-source/0670b93/ : SUCCESS in 59m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/37/652937/2/gate/openstack-tox-lower-constraints/064d23b/ : SUCCESS in 4m 33s\n- openstack-tox-pep8 http://logs.openstack.org/37/652937/2/gate/openstack-tox-pep8/c590977/ : SUCCESS in 2m 43s\n- openstack-tox-py27 http://logs.openstack.org/37/652937/2/gate/openstack-tox-py27/52f0f5a/ : SUCCESS in 2m 47s\n- openstack-tox-py36 http://logs.openstack.org/37/652937/2/gate/openstack-tox-py36/0f63cfe/ : SUCCESS in 2m 56s\n- openstack-tox-py37 http://logs.openstack.org/37/652937/2/gate/openstack-tox-py37/dbfb46e/ : SUCCESS in 3m 37s\n- openstack-tox-docs http://logs.openstack.org/37/652937/2/gate/openstack-tox-docs/f7c571d/html/ : SUCCESS in 2m 51s\n- kolla-build-ubuntu-source http://logs.openstack.org/37/652937/2/gate/kolla-build-ubuntu-source/1823a62/ : SUCCESS in 35m 22s","accounts_in_message":[],"_revision_number":2},{"id":"3cabc557ba2f2c7b23e22708e309dfbe2f2dd8f8","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-05-06 05:26:28.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 2648c12c8d3567ecafc6c4b6c3c1ce9c6090c972","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8b8958459809940b812364542a6994a89624f981","revisions":{"ac577ab505ac085fce95f44d55ab56bee44c9cf6":{"kind":"REWORK","_number":1,"created":"2019-04-16 08:01:46.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/37/652937/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/652937/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/652937/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/652937/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/652937/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/652937/1"}}},"commit":{"parents":[{"commit":"89b1ee6ebac5f5b40cbf09e5e7b0b69ae0d24110","subject":"Merge \"Disable publishing images on Ocata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89b1ee6ebac5f5b40cbf09e5e7b0b69ae0d24110"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-16 07:59:37.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-16 08:01:27.000000000","tz":330},"subject":"Use /var/lib/tempest symlink for tempest container","message":"Use /var/lib/tempest symlink for tempest container\n\nCurrently with in tempest Dockerfile, we have two symlink\n/usr/share/openstack and /var/lib/tempest, we cannot create\ntwo symlink for a single directory. It is causing failure to\nbuild container using buildah.\n\nChange-Id: I402f58ce53fdce426582cf2fc31349da6d2f5ae0\nSigned-off-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac577ab505ac085fce95f44d55ab56bee44c9cf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac577ab505ac085fce95f44d55ab56bee44c9cf6"}]},"branch":"refs/heads/master"},"8b8958459809940b812364542a6994a89624f981":{"kind":"REWORK","_number":2,"created":"2019-04-17 06:30:53.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/37/652937/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/652937/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/652937/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/652937/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/652937/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/652937/2"}}},"commit":{"parents":[{"commit":"89b1ee6ebac5f5b40cbf09e5e7b0b69ae0d24110","subject":"Merge \"Disable publishing images on Ocata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89b1ee6ebac5f5b40cbf09e5e7b0b69ae0d24110"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-17 06:26:03.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-17 06:26:03.000000000","tz":330},"subject":"Reuse /var/lib/tempest symlink for rpm also","message":"Reuse /var/lib/tempest symlink for rpm also\n\nhttps://review.openstack.org/#/c/605980/ with this review in\ntripleo, we use /var/lib/tempest path but in kolla tempest\ndockerfile /usr/share/openstack-tempest is used which is not\nmounted in tripleo deployment.\nLet\u0027s reuse the same volume with tempest container instead\nof /usr/share/openstack-tempest.\n\nChange-Id: I402f58ce53fdce426582cf2fc31349da6d2f5ae0\nSigned-off-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b8958459809940b812364542a6994a89624f981"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b8958459809940b812364542a6994a89624f981"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
