)]}'
{"id":"openstack%2Fkolla~653915","triplet_id":"openstack%2Fkolla~stable%2Frocky~Ifdc3a1fd5fe37da529b2aee9811b12f744cff3bf","project":"openstack/kolla","branch":"stable/rocky","topic":"bug/1821681-stable/rocky","hashtags":[],"change_id":"Ifdc3a1fd5fe37da529b2aee9811b12f744cff3bf","subject":"Fix the problem that the crush map is changed","status":"MERGED","created":"2019-04-19 07:21:44.000000000","updated":"2019-06-06 16:31:13.000000000","submitted":"2019-06-06 16:31:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653915-1559838673521-2f66b382","meta_rev_id":"9d91e55fcae7e7cbee1997cb804a56e73eaffc84","_number":653915,"virtual_id_number":653915,"owner":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2019-04-22 05:08:21.000000000","_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-06-06 16:31:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-22 05:31:31.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-04-24 17:14:32.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":2,"date":"2019-06-06 15:30:40.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":1,"date":"2019-06-06 15:30:40.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 05:31:31.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-24 17:14:32.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-06 15:30:40.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-06 16:31:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"19711be81772f125fb9f95f73b7cf2473d98931c","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-04-19 07:21:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"82532e7e71070ad17f6646c5d2d0ea51ec6893fa","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-19 07:52:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/15/653915/1/openstack-check/tripleo-build-containers-fedora-28-master/7fd2737/ : RETRY_LIMIT in 3m 13s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/15/653915/1/openstack-check/tripleo-build-containers-fedora-28-stein/707bd7c/ : SUCCESS in 28m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e316e26799b3039f193ee80b7255e389fa1959e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 09:36:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/15/653915/1/check/kolla-build-centos-binary/68ff85e/ : SUCCESS in 36m 00s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/15/653915/1/check/kolla-build-centos-source/830c9a7/ : SUCCESS in 49m 28s\n- kolla-ansible-centos-source finger://ze11.openstack.org/79998fcf2656400cb5207129f9752a16 : RETRY_LIMIT in 59m 04s\n- kolla-ansible-centos-binary finger://ze03.openstack.org/20eca3ef2d6542f186a45f290e4c271e : RETRY_LIMIT in 1h 06m 01s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/15/653915/1/check/kolla-build-oraclelinux-binary/97ad626/ : RETRY_LIMIT in 14m 24s (non-voting)\n- kolla-build-oraclelinux-source finger://ze10.openstack.org/3cee016d18ba4b6b8359a2a0d558d720 : RETRY_LIMIT in 40m 49s\n- kolla-ansible-oraclelinux-source finger://ze08.openstack.org/2aa4c73ad6c347ee9896dca0d72232dd : RETRY_LIMIT in 45m 39s (non-voting)\n- kolla-ansible-oraclelinux-binary finger://ze05.openstack.org/b17b7b5c14c54f88bfceda7a6dad6d0f : RETRY_LIMIT in 1h 09m 20s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/15/653915/1/check/openstack-tox-lower-constraints/716abe7/ : SUCCESS in 4m 58s\n- openstack-tox-pep8 http://logs.openstack.org/15/653915/1/check/openstack-tox-pep8/dd2ee97/ : SUCCESS in 4m 51s\n- openstack-tox-py27 http://logs.openstack.org/15/653915/1/check/openstack-tox-py27/9e0b41a/ : SUCCESS in 6m 29s\n- openstack-tox-py35 http://logs.openstack.org/15/653915/1/check/openstack-tox-py35/9c8b7f1/ : SUCCESS in 4m 42s\n- build-openstack-sphinx-docs http://logs.openstack.org/15/653915/1/check/build-openstack-sphinx-docs/7ef218d/html/ : SUCCESS in 4m 28s\n- kolla-build-ubuntu-binary http://logs.openstack.org/15/653915/1/check/kolla-build-ubuntu-binary/cb58a64/ : SUCCESS in 1h 02m 13s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/15/653915/1/check/kolla-build-ubuntu-source/bc84490/ : SUCCESS in 36m 59s\n- kolla-ansible-ubuntu-source finger://ze08.openstack.org/2d4beaaa6a034fcc8a4cffd239774338 : RETRY_LIMIT in 23m 24s\n- kolla-ansible-ubuntu-binary finger://ze10.openstack.org/9ad75f7390b84e6ba8e75f2d0254510b : RETRY_LIMIT in 1h 04m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c5ca9e5be74d8fc4ed3094e9c76de0220e391265","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-04-22 05:08:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"60e2028d6ade65d00d18ea398df1fbe48c2be92b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-22 05:31:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/15/653915/1/openstack-check/tripleo-build-containers-fedora-28-master/91efbf4/ : RETRY_LIMIT in 2m 27s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/15/653915/1/openstack-check/tripleo-build-containers-fedora-28-stein/0d12f5c/ : FAILURE in 5m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"32359bd9ba47cb2c94d96804e4003408bf8adbb3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 06:56:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/15/653915/1/check/kolla-build-centos-binary/84ea2c4/ : SUCCESS in 28m 17s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/15/653915/1/check/kolla-build-centos-source/b6c0e26/ : SUCCESS in 52m 32s\n- kolla-ansible-centos-source http://logs.openstack.org/15/653915/1/check/kolla-ansible-centos-source/46aeb91/ : SUCCESS in 1h 11m 35s\n- kolla-ansible-centos-binary http://logs.openstack.org/15/653915/1/check/kolla-ansible-centos-binary/baa113f/ : SUCCESS in 1h 00m 55s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/15/653915/1/check/kolla-build-oraclelinux-binary/e1cc00d/ : SUCCESS in 43m 56s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/15/653915/1/check/kolla-build-oraclelinux-source/ac94837/ : SUCCESS in 49m 04s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/15/653915/1/check/kolla-ansible-oraclelinux-source/bbff241/ : SUCCESS in 1h 26m 29s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/15/653915/1/check/kolla-ansible-oraclelinux-binary/2cc3846/ : SUCCESS in 1h 16m 57s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/15/653915/1/check/openstack-tox-lower-constraints/080241f/ : SUCCESS in 4m 28s\n- openstack-tox-pep8 http://logs.openstack.org/15/653915/1/check/openstack-tox-pep8/a4571a9/ : SUCCESS in 3m 12s\n- openstack-tox-py27 http://logs.openstack.org/15/653915/1/check/openstack-tox-py27/1af06f2/ : SUCCESS in 2m 35s\n- openstack-tox-py35 http://logs.openstack.org/15/653915/1/check/openstack-tox-py35/d6c2efb/ : SUCCESS in 3m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/15/653915/1/check/build-openstack-sphinx-docs/0e36194/html/ : SUCCESS in 3m 20s\n- kolla-build-ubuntu-binary http://logs.openstack.org/15/653915/1/check/kolla-build-ubuntu-binary/b034cd3/ : SUCCESS in 59m 11s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/15/653915/1/check/kolla-build-ubuntu-source/71e3ef6/ : SUCCESS in 36m 11s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/15/653915/1/check/kolla-ansible-ubuntu-source/1e28b1b/ : SUCCESS in 1h 00m 16s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/15/653915/1/check/kolla-ansible-ubuntu-binary/1bd005e/ : SUCCESS in 1h 47m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5f2313ac84e2ebe51f6e8bf0fdbbea182ecfc4c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-04-24 17:14:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"32726e7c1b9ffb0ee3db225ae27883b2e02a765d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-06 15:30:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"db731c346c81f0757d6d75bd150342e18aabe49d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 15:30:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"492b57cb6461f4c7ef4443fe16778a5d3df3c19c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 16:31:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/15/653915/1/gate/kolla-build-centos-source/3b4a6c5/ : SUCCESS in 54m 11s\n- kolla-build-oraclelinux-source http://logs.openstack.org/15/653915/1/gate/kolla-build-oraclelinux-source/9dc0593/ : SUCCESS in 52m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/15/653915/1/gate/openstack-tox-lower-constraints/84265f3/ : SUCCESS in 3m 36s\n- openstack-tox-pep8 http://logs.openstack.org/15/653915/1/gate/openstack-tox-pep8/00801cb/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/15/653915/1/gate/openstack-tox-py27/7decc0b/ : SUCCESS in 3m 19s\n- openstack-tox-py35 http://logs.openstack.org/15/653915/1/gate/openstack-tox-py35/4a62a6c/ : SUCCESS in 3m 13s\n- build-openstack-sphinx-docs http://logs.openstack.org/15/653915/1/gate/build-openstack-sphinx-docs/473a1c9/html/ : SUCCESS in 3m 26s\n- kolla-build-ubuntu-source http://logs.openstack.org/15/653915/1/gate/kolla-build-ubuntu-source/ff84960/ : SUCCESS in 58m 37s","accounts_in_message":[],"_revision_number":1},{"id":"89394c228f943a95eaaac882d0e71be89c6e67f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 16:31:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e8ede2b49a2c34d7bea9906564bf57662f53d1d","revisions":{"8e8ede2b49a2c34d7bea9906564bf57662f53d1d":{"kind":"REWORK","_number":1,"created":"2019-04-19 07:21:44.000000000","uploader":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"ref":"refs/changes/15/653915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/653915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/653915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/653915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/653915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/653915/1"}}},"commit":{"parents":[{"commit":"9f85941324904fc61f88bc14175e4e364cee7f6f","subject":"Fix the problem of osd initialization failed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f85941324904fc61f88bc14175e4e364cee7f6f"}]}],"author":{"name":"wangwei","email":"wangwei@navercorp.com","date":"2019-03-26 06:49:18.000000000","tz":540},"committer":{"name":"wangwei","email":"wang.wei@navercorp.com","date":"2019-04-19 07:21:44.000000000","tz":0},"subject":"Fix the problem that the crush map is changed","message":"Fix the problem that the crush map is changed\n\nWhen deploying ceph osd with kolla, each osd deployment will move the\nhost bucket to default in the crush map.\n\nIf we adjust the crush map, we have to re-adjust the crush map after\nadding osd or fixing osd.\nThis commit fixes this problem.\n\nChange-Id: Ifdc3a1fd5fe37da529b2aee9811b12f744cff3bf\nCloses-bug: #1821681\n(cherry picked from commit 3d4a7360a7c51a1f1c806a70026a7c8b20e20789)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e8ede2b49a2c34d7bea9906564bf57662f53d1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e8ede2b49a2c34d7bea9906564bf57662f53d1d"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
