)]}'
{"id":"openstack%2Fkolla~653916","triplet_id":"openstack%2Fkolla~stable%2Fstein~Ifdc3a1fd5fe37da529b2aee9811b12f744cff3bf","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1821681-stable/stein","hashtags":[],"change_id":"Ifdc3a1fd5fe37da529b2aee9811b12f744cff3bf","subject":"Fix the problem that the crush map is changed","status":"MERGED","created":"2019-04-19 07:22:22.000000000","updated":"2019-04-24 19:54:55.000000000","submitted":"2019-04-24 19:54:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653916-1556135695587-d1a6a7d1","meta_rev_id":"5f3f7fc6b9aef39e5e87f48a68b0cf94e1ed7b74","_number":653916,"virtual_id_number":653916,"owner":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2019-04-22 07:28:43.000000000","_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":2,"date":"2019-04-24 19:54:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-22 07:40:36.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},"all":[{"value":1,"date":"2019-04-24 03:44:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},{"value":2,"date":"2019-04-24 18:41:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-04-24 17:14:28.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},{"value":1,"date":"2019-04-24 18:41:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 07:40:36.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-24 03:44:03.000000000","updated_by":{"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},"reviewer":{"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},"state":"REVIEWER"},{"updated":"2019-04-24 17:14:28.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-04-24 18:41:46.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-04-24 19:54:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a82f0747661909f725250b754f9724a8dde55c8a","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-04-19 07:22:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"99c5f1854d2228747eaffe79fe5edf3082f6c6cc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-19 08:07:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/16/653916/1/openstack-check/tripleo-build-containers-fedora-28-master/6a335af/ : SUCCESS in 29m 59s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/16/653916/1/openstack-check/tripleo-build-containers-fedora-28-stein/53acd03/ : SUCCESS in 32m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bcd0db6b7949ddddc8ab3c070a6e757d889a4754","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 10:54:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/16/653916/1/check/kolla-build-centos-binary/c28e7e2/ : SUCCESS in 38m 18s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/16/653916/1/check/kolla-build-centos-source/3e49ead/ : SUCCESS in 42m 40s\n- kolla-ansible-centos-source finger://ze02.openstack.org/b4fdf13ce5044fbb910257e16553a7a2 : RETRY_LIMIT in 51m 19s\n- kolla-ansible-centos-binary finger://ze11.openstack.org/6491276a36cf4c56a051d681706e4c27 : RETRY_LIMIT in 47m 15s (non-voting)\n- kolla-ansible-centos-source-upgrade finger://ze12.openstack.org/53b8dfeba6c2440f8759c44be21f8753 : RETRY_LIMIT in 59m 40s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/16/653916/1/check/tripleo-build-containers-centos-7/fda4a4e/ : SUCCESS in 40m 59s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/16/653916/1/check/kolla-build-oraclelinux-binary/b4c67e9/ : SUCCESS in 49m 41s (non-voting)\n- kolla-build-oraclelinux-source finger://ze01.openstack.org/a9f27198f1414385b1aab6350ec67733 : RETRY_LIMIT in 56m 31s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/16/653916/1/check/kolla-ansible-oraclelinux-source/c8a670f/ : SUCCESS in 1h 03m 00s (non-voting)\n- kolla-ansible-oraclelinux-binary finger://ze09.openstack.org/4c3786546dd44b459a8aa4e1340f65eb : RETRY_LIMIT in 1h 01m 38s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/16/653916/1/check/openstack-tox-lower-constraints/6170fd1/ : SUCCESS in 6m 03s\n- openstack-tox-pep8 http://logs.openstack.org/16/653916/1/check/openstack-tox-pep8/9628411/ : SUCCESS in 4m 48s\n- openstack-tox-py27 http://logs.openstack.org/16/653916/1/check/openstack-tox-py27/c999ddb/ : SUCCESS in 5m 37s\n- openstack-tox-py36 http://logs.openstack.org/16/653916/1/check/openstack-tox-py36/8a2842d/ : SUCCESS in 5m 36s\n- openstack-tox-py37 http://logs.openstack.org/16/653916/1/check/openstack-tox-py37/a2d6b66/ : SUCCESS in 4m 58s\n- openstack-tox-docs http://logs.openstack.org/16/653916/1/check/openstack-tox-docs/169e94b/html/ : SUCCESS in 4m 01s\n- kolla-build-ubuntu-binary http://logs.openstack.org/16/653916/1/check/kolla-build-ubuntu-binary/d88bceb/ : SUCCESS in 30m 13s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/16/653916/1/check/kolla-build-ubuntu-source/2ce735b/ : SUCCESS in 31m 05s\n- kolla-ansible-ubuntu-source finger://ze11.openstack.org/ea405d5431344d68b7682cc18fad5164 : RETRY_LIMIT in 1h 02m 07s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/16/653916/1/check/kolla-ansible-ubuntu-binary/d56d5bf/ : FAILURE in 32m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/16/653916/1/check/kolla-ansible-ubuntu-source-upgrade/4b4dfb8/ : SUCCESS in 58m 36s","accounts_in_message":[],"_revision_number":1},{"id":"cd17181974d9e79bf28a0a8186678ad2deb2d5b4","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-04-22 07:28:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"54fe4061008e5e58a9cce17c40c4b0d8ea2ed9f8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-22 07:40:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/16/653916/1/openstack-check/tripleo-build-containers-fedora-28-master/448c0e6/ : FAILURE in 5m 49s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/16/653916/1/openstack-check/tripleo-build-containers-fedora-28-stein/2637d0e/ : FAILURE in 6m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4698337219245f9c26d6247d5795806f481692ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 09:14:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/16/653916/1/check/kolla-build-centos-binary/4c68fca/ : SUCCESS in 41m 37s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/16/653916/1/check/kolla-build-centos-source/48daf9e/ : SUCCESS in 42m 04s\n- kolla-ansible-centos-source http://logs.openstack.org/16/653916/1/check/kolla-ansible-centos-source/d2b5aea/ : SUCCESS in 1h 12m 28s\n- kolla-ansible-centos-binary http://logs.openstack.org/16/653916/1/check/kolla-ansible-centos-binary/42374c7/ : SUCCESS in 52m 56s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/16/653916/1/check/kolla-ansible-centos-source-upgrade/535e42c/ : SUCCESS in 1h 11m 38s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/16/653916/1/check/tripleo-build-containers-centos-7/a3fd2cf/ : SUCCESS in 43m 09s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/16/653916/1/check/kolla-build-oraclelinux-binary/97dab2d/ : SUCCESS in 1h 00m 18s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/16/653916/1/check/kolla-build-oraclelinux-source/c18f8ee/ : SUCCESS in 1h 00m 53s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/16/653916/1/check/kolla-ansible-oraclelinux-source/b8d4662/ : SUCCESS in 1h 24m 12s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/16/653916/1/check/kolla-ansible-oraclelinux-binary/39627a9/ : SUCCESS in 1h 03m 29s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/16/653916/1/check/openstack-tox-lower-constraints/c7249c0/ : SUCCESS in 3m 56s\n- openstack-tox-pep8 http://logs.openstack.org/16/653916/1/check/openstack-tox-pep8/961fdc0/ : SUCCESS in 3m 17s\n- openstack-tox-py27 http://logs.openstack.org/16/653916/1/check/openstack-tox-py27/5e52739/ : SUCCESS in 3m 12s\n- openstack-tox-py36 http://logs.openstack.org/16/653916/1/check/openstack-tox-py36/20abc04/ : SUCCESS in 3m 34s\n- openstack-tox-py37 http://logs.openstack.org/16/653916/1/check/openstack-tox-py37/3fa8c36/ : SUCCESS in 3m 21s\n- openstack-tox-docs http://logs.openstack.org/16/653916/1/check/openstack-tox-docs/c469f5b/html/ : SUCCESS in 3m 04s\n- kolla-build-ubuntu-binary http://logs.openstack.org/16/653916/1/check/kolla-build-ubuntu-binary/fd68b8d/ : SUCCESS in 42m 59s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/16/653916/1/check/kolla-build-ubuntu-source/3934459/ : SUCCESS in 1h 02m 39s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/16/653916/1/check/kolla-ansible-ubuntu-source/d8b94c3/ : SUCCESS in 57m 25s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/16/653916/1/check/kolla-ansible-ubuntu-binary/3aea9fb/ : SUCCESS in 48m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/16/653916/1/check/kolla-ansible-ubuntu-source-upgrade/006c504/ : SUCCESS in 54m 46s","accounts_in_message":[],"_revision_number":1},{"id":"cfbd329fb5272dc33974130d2feb2fd52957bd89","author":{"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},"date":"2019-04-24 03:44:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a8d81bcd17cf0028e209ff820cedffa20135c72e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-04-24 17:14:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"73b182e26fd9040d67bd6bfd0472b9bb4ba46b2c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-04-24 18:41:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"08f3e638f607b904da872807ea50c65be976224e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 18:42:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e0eca229820dfe7a9ab08cdd48edb453d90ced18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:54:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/16/653916/1/gate/kolla-build-centos-source/6154d36/ : SUCCESS in 47m 53s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/16/653916/1/gate/tripleo-build-containers-centos-7/de2925d/ : SUCCESS in 43m 41s\n- kolla-build-oraclelinux-source http://logs.openstack.org/16/653916/1/gate/kolla-build-oraclelinux-source/cccd427/ : SUCCESS in 1h 01m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/653916/1/gate/openstack-tox-lower-constraints/8339b7a/ : SUCCESS in 4m 57s\n- openstack-tox-pep8 http://logs.openstack.org/16/653916/1/gate/openstack-tox-pep8/91503f8/ : SUCCESS in 4m 54s\n- openstack-tox-py27 http://logs.openstack.org/16/653916/1/gate/openstack-tox-py27/ffcc88b/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/16/653916/1/gate/openstack-tox-py36/b6c5039/ : SUCCESS in 4m 01s\n- openstack-tox-py37 http://logs.openstack.org/16/653916/1/gate/openstack-tox-py37/12c01d6/ : SUCCESS in 3m 57s\n- openstack-tox-docs http://logs.openstack.org/16/653916/1/gate/openstack-tox-docs/55cbf14/html/ : SUCCESS in 5m 14s\n- kolla-build-ubuntu-source http://logs.openstack.org/16/653916/1/gate/kolla-build-ubuntu-source/333711f/ : SUCCESS in 34m 41s","accounts_in_message":[],"_revision_number":1},{"id":"3c919fa34fb48f44b458af729838479c1aa1a3c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:54:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8371ecb2acc5be214dcf9503c83a9ccb4081f151","revisions":{"8371ecb2acc5be214dcf9503c83a9ccb4081f151":{"kind":"REWORK","_number":1,"created":"2019-04-19 07:22:22.000000000","uploader":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"ref":"refs/changes/16/653916/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/653916/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/653916/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/653916/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/653916/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/653916/1"}}},"commit":{"parents":[{"commit":"f958a9d36871c724a091f8ad4c0591a8769127eb","subject":"Bump Bifrost version to 6.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f958a9d36871c724a091f8ad4c0591a8769127eb"}]}],"author":{"name":"wangwei","email":"wangwei@navercorp.com","date":"2019-03-26 06:49:18.000000000","tz":540},"committer":{"name":"wangwei","email":"wang.wei@navercorp.com","date":"2019-04-19 07:22:22.000000000","tz":0},"subject":"Fix the problem that the crush map is changed","message":"Fix the problem that the crush map is changed\n\nWhen deploying ceph osd with kolla, each osd deployment will move the\nhost bucket to default in the crush map.\n\nIf we adjust the crush map, we have to re-adjust the crush map after\nadding osd or fixing osd.\nThis commit fixes this problem.\n\nChange-Id: Ifdc3a1fd5fe37da529b2aee9811b12f744cff3bf\nCloses-bug: #1821681\n(cherry picked from commit 3d4a7360a7c51a1f1c806a70026a7c8b20e20789)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8371ecb2acc5be214dcf9503c83a9ccb4081f151"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8371ecb2acc5be214dcf9503c83a9ccb4081f151"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
