)]}'
{"id":"openstack%2Fkolla~654396","triplet_id":"openstack%2Fkolla~master~I377ea63827b6f3995ee275fdc466c1b2100eafc7","project":"openstack/kolla","branch":"master","topic":"opendev","hashtags":[],"change_id":"I377ea63827b6f3995ee275fdc466c1b2100eafc7","subject":"Move to opendev","status":"MERGED","created":"2019-04-22 07:28:42.000000000","updated":"2019-04-29 14:30:48.000000000","submitted":"2019-04-29 14:30:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654396-1556548248297-47422c5d","meta_rev_id":"c8612c164431cfb37b5a43a620783495c8c7e133","_number":654396,"virtual_id_number":654396,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"date":"2019-04-24 07:24:51.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-04-29 14:30:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-24 08:02:26.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"recommended":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"all":[{"value":1,"date":"2019-04-28 07:31:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":1,"date":"2019-04-29 04:35:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-04-29 13:17:10.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-04-24 11:22:02.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-04-28 01:41:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2019-04-29 13:17:10.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 05:41:29.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-04-23 06:54:03.000000000","updated_by":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"reviewer":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"state":"REVIEWER"},{"updated":"2019-04-24 02:49:43.000000000","updated_by":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"reviewer":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"state":"REVIEWER"},{"updated":"2019-04-24 07:20:10.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-04-24 08:02:26.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-24 11:22:02.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-04-28 01:41:04.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-04-28 07:31:23.000000000","updated_by":{"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},"reviewer":{"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},"state":"REVIEWER"},{"updated":"2019-04-29 04:35:42.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-04-29 13:17:10.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-04-29 14:30:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f69abe93f8d9ee1f4879e77435630d94ecdd4bfb","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-22 07:28:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9a187bfdc4baf169fef11ae1c6c575c2e181047","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-22 07:40:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/96/654396/1/openstack-check/tripleo-build-containers-fedora-28-master/0d9e50b/ : FAILURE in 5m 31s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/96/654396/1/openstack-check/tripleo-build-containers-fedora-28-stein/9dc80cf/ : FAILURE in 5m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"72234e8dc0d235a5a303c9f2141865bc994bd1cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 09:06:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/96/654396/1/check/test-release-openstack/cf7eea2/ : SUCCESS in 1m 55s\n- kolla-build-centos-binary http://logs.openstack.org/96/654396/1/check/kolla-build-centos-binary/c7fced6/ : SUCCESS in 42m 01s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/96/654396/1/check/kolla-build-centos-source/ff9755e/ : SUCCESS in 50m 05s\n- kolla-ansible-centos-source http://logs.openstack.org/96/654396/1/check/kolla-ansible-centos-source/ad45f87/ : SUCCESS in 1h 15m 49s\n- kolla-ansible-centos-binary http://logs.openstack.org/96/654396/1/check/kolla-ansible-centos-binary/b46b52a/ : SUCCESS in 1h 07m 07s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/96/654396/1/check/kolla-ansible-centos-source-upgrade/97a28ff/ : SUCCESS in 1h 19m 36s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/96/654396/1/check/tripleo-build-containers-centos-7/beb2f43/ : SUCCESS in 49m 01s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/96/654396/1/check/kolla-build-oraclelinux-binary/1c284fe/ : SUCCESS in 51m 24s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/96/654396/1/check/kolla-build-oraclelinux-source/0135478/ : SUCCESS in 53m 21s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/96/654396/1/check/kolla-ansible-oraclelinux-source/029782f/ : SUCCESS in 1h 10m 20s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/96/654396/1/check/kolla-ansible-oraclelinux-binary/bad26de/ : SUCCESS in 1h 07m 14s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/96/654396/1/check/openstack-tox-lower-constraints/7654e07/ : SUCCESS in 4m 04s\n- openstack-tox-pep8 http://logs.openstack.org/96/654396/1/check/openstack-tox-pep8/38d456b/ : SUCCESS in 2m 55s\n- openstack-tox-py27 http://logs.openstack.org/96/654396/1/check/openstack-tox-py27/89261a5/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/96/654396/1/check/openstack-tox-py36/f3d5f07/ : SUCCESS in 3m 12s\n- openstack-tox-py37 http://logs.openstack.org/96/654396/1/check/openstack-tox-py37/89ae890/ : SUCCESS in 3m 32s\n- openstack-tox-docs http://logs.openstack.org/96/654396/1/check/openstack-tox-docs/8093cc4/html/ : SUCCESS in 2m 52s\n- build-openstack-releasenotes http://logs.openstack.org/96/654396/1/check/build-openstack-releasenotes/daf5a8b/html/ : SUCCESS in 4m 02s\n- kolla-build-ubuntu-binary http://logs.openstack.org/96/654396/1/check/kolla-build-ubuntu-binary/ebcdf0b/ : SUCCESS in 34m 24s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/96/654396/1/check/kolla-build-ubuntu-source/6796232/ : SUCCESS in 51m 08s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/96/654396/1/check/kolla-ansible-ubuntu-source/48cf36e/ : SUCCESS in 51m 03s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/96/654396/1/check/kolla-ansible-ubuntu-binary/cfe24a4/ : SUCCESS in 48m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/96/654396/1/check/kolla-ansible-ubuntu-source-upgrade/9a1664c/ : SUCCESS in 1h 07m 58s","accounts_in_message":[],"_revision_number":1},{"id":"6d55c4fd166b626b18966c8b04808424f8123d8d","author":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"date":"2019-04-23 05:06:12.000000000","message":"Patch Set 1: Code-Review-1\n\nwould be great if you add discussion link, it is always helpful for the review to check the reference and readout the discussion http://lists.openstack.org/pipermail/openstack-discuss/2019-March/003603.html\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"ff1537dd6bab00f5fad71a623353e7f5924deddd","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-23 05:30:59.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a531652aa521ba94e71b24d185cd08055509b1ac","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-04-23 05:41:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a90693f7b1a4d4a1949547e35c1623ceae336154","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-23 06:13:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/96/654396/2/openstack-check/tripleo-build-containers-fedora-28-master/7aed527/ : SUCCESS in 36m 41s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/96/654396/2/openstack-check/tripleo-build-containers-fedora-28-stein/818ecd4/ : SUCCESS in 40m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aeb4554916ccf186bb6c706a0c7751f8e628ded4","author":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"date":"2019-04-23 06:54:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"481da618adffe6044fe50824e25a32970fa41239","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 07:10:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/96/654396/2/check/test-release-openstack/e055ada/ : SUCCESS in 2m 34s\n- kolla-build-centos-binary http://logs.openstack.org/96/654396/2/check/kolla-build-centos-binary/7bc5dd9/ : SUCCESS in 41m 04s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/96/654396/2/check/kolla-build-centos-source/3ceea48/ : SUCCESS in 53m 03s\n- kolla-ansible-centos-source http://logs.openstack.org/96/654396/2/check/kolla-ansible-centos-source/33b7392/ : SUCCESS in 1h 09m 06s\n- kolla-ansible-centos-binary http://logs.openstack.org/96/654396/2/check/kolla-ansible-centos-binary/30ceb60/ : SUCCESS in 1h 05m 33s (non-voting)\n- kolla-ansible-centos-source-upgrade finger://ze11.openstack.org/8155b2ddee8244f3aa146b16d1d1bdbe : RETRY_LIMIT in 1h 34m 00s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/96/654396/2/check/tripleo-build-containers-centos-7/5e48e19/ : SUCCESS in 44m 45s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/96/654396/2/check/kolla-build-oraclelinux-binary/4f3c034/ : SUCCESS in 1h 02m 11s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/96/654396/2/check/kolla-build-oraclelinux-source/af12869/ : SUCCESS in 52m 48s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/96/654396/2/check/kolla-ansible-oraclelinux-source/b6e1b2e/ : SUCCESS in 1h 08m 48s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/96/654396/2/check/kolla-ansible-oraclelinux-binary/2bbba40/ : SUCCESS in 1h 05m 51s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/96/654396/2/check/openstack-tox-lower-constraints/102db16/ : SUCCESS in 4m 36s\n- openstack-tox-pep8 http://logs.openstack.org/96/654396/2/check/openstack-tox-pep8/2d07b01/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/96/654396/2/check/openstack-tox-py27/05c510a/ : SUCCESS in 4m 01s\n- openstack-tox-py36 http://logs.openstack.org/96/654396/2/check/openstack-tox-py36/eb31846/ : SUCCESS in 3m 42s\n- openstack-tox-py37 http://logs.openstack.org/96/654396/2/check/openstack-tox-py37/af952a6/ : SUCCESS in 3m 58s\n- openstack-tox-docs http://logs.openstack.org/96/654396/2/check/openstack-tox-docs/2248ec5/html/ : SUCCESS in 3m 34s\n- build-openstack-releasenotes http://logs.openstack.org/96/654396/2/check/build-openstack-releasenotes/477c97f/html/ : SUCCESS in 4m 17s\n- kolla-build-ubuntu-binary http://logs.openstack.org/96/654396/2/check/kolla-build-ubuntu-binary/b3b92e5/ : SUCCESS in 29m 41s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/96/654396/2/check/kolla-build-ubuntu-source/3f4d0a3/ : SUCCESS in 37m 22s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/96/654396/2/check/kolla-ansible-ubuntu-source/5df5280/ : SUCCESS in 58m 31s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/96/654396/2/check/kolla-ansible-ubuntu-binary/8232e11/ : SUCCESS in 50m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/96/654396/2/check/kolla-ansible-ubuntu-source-upgrade/fe81524/ : SUCCESS in 1h 12m 01s","accounts_in_message":[],"_revision_number":2},{"id":"389352dcce8b20d3edca14b5cd6992ccf40e672c","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-23 07:26:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"38b14dae1bc56dff69e734b90db178be1a184b9f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-23 08:02:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/96/654396/2/openstack-check/tripleo-build-containers-fedora-28-master/b49873e/ : SUCCESS in 33m 31s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/96/654396/2/openstack-check/tripleo-build-containers-fedora-28-stein/f5263aa/ : SUCCESS in 31m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8d9b675b6e6e3e60efb4ce0ffbc9f3794f120aa1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 09:02:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/96/654396/2/check/test-release-openstack/4118fd9/ : SUCCESS in 1m 49s\n- kolla-build-centos-binary http://logs.openstack.org/96/654396/2/check/kolla-build-centos-binary/434cf79/ : SUCCESS in 40m 35s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/96/654396/2/check/kolla-build-centos-source/1360cf4/ : SUCCESS in 1h 00m 05s\n- kolla-ansible-centos-source http://logs.openstack.org/96/654396/2/check/kolla-ansible-centos-source/53e9dbe/ : SUCCESS in 1h 11m 37s\n- kolla-ansible-centos-binary http://logs.openstack.org/96/654396/2/check/kolla-ansible-centos-binary/fec4bc9/ : SUCCESS in 1h 08m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/96/654396/2/check/kolla-ansible-centos-source-upgrade/f04e37b/ : SUCCESS in 1h 13m 31s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/96/654396/2/check/tripleo-build-containers-centos-7/6ef4faf/ : SUCCESS in 42m 43s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/96/654396/2/check/kolla-build-oraclelinux-binary/e70e856/ : SUCCESS in 1h 05m 51s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/96/654396/2/check/kolla-build-oraclelinux-source/7ce6933/ : SUCCESS in 49m 05s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/96/654396/2/check/kolla-ansible-oraclelinux-source/0a26d1c/ : SUCCESS in 1h 19m 00s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/96/654396/2/check/kolla-ansible-oraclelinux-binary/c49b327/ : SUCCESS in 1h 08m 26s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/96/654396/2/check/openstack-tox-lower-constraints/1d66cfe/ : SUCCESS in 3m 58s\n- openstack-tox-pep8 http://logs.openstack.org/96/654396/2/check/openstack-tox-pep8/7180678/ : SUCCESS in 3m 17s\n- openstack-tox-py27 http://logs.openstack.org/96/654396/2/check/openstack-tox-py27/206c976/ : SUCCESS in 3m 06s\n- openstack-tox-py36 http://logs.openstack.org/96/654396/2/check/openstack-tox-py36/aef975e/ : SUCCESS in 3m 23s\n- openstack-tox-py37 http://logs.openstack.org/96/654396/2/check/openstack-tox-py37/8785d6a/ : SUCCESS in 3m 27s\n- openstack-tox-docs http://logs.openstack.org/96/654396/2/check/openstack-tox-docs/064b486/html/ : SUCCESS in 2m 52s\n- build-openstack-releasenotes http://logs.openstack.org/96/654396/2/check/build-openstack-releasenotes/524ca74/html/ : SUCCESS in 3m 37s\n- kolla-build-ubuntu-binary http://logs.openstack.org/96/654396/2/check/kolla-build-ubuntu-binary/202b386/ : SUCCESS in 37m 43s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/96/654396/2/check/kolla-build-ubuntu-source/5090886/ : SUCCESS in 29m 22s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/96/654396/2/check/kolla-ansible-ubuntu-source/f983f53/ : SUCCESS in 54m 58s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/96/654396/2/check/kolla-ansible-ubuntu-binary/bace098/ : SUCCESS in 49m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/96/654396/2/check/kolla-ansible-ubuntu-source-upgrade/8926e2d/ : SUCCESS in 1h 05m 54s","accounts_in_message":[],"_revision_number":2},{"id":"8f83d3f61ea67b9a82148ed51ff69a9051d71670","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-04-24 02:49:43.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2f8b59d3abefc4da01da53d6c08eeb6a2089fd08","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-04-24 07:20:10.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"44d795e80547dba1fa0ced38d95d467d8a703f8a","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-24 07:23:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f378f1b08ba68858499f25df31dd84d8ba7e6382","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-24 07:24:41.000000000","message":"Patch Set 3:\n\n\u003e (1 comment)\n\ndone","accounts_in_message":[],"_revision_number":3},{"id":"e2129b745b05e10c42f9e3368cd2c2b2f4ef2c21","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-24 07:24:51.000000000","message":"Patch Set 3:\n\nI checked all the urls and they all available. Every place in this project has changed.","accounts_in_message":[],"_revision_number":3},{"id":"1710e36364aac7536c2b82676a824b4155dd5a51","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-24 08:02:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/96/654396/3/openstack-check/tripleo-build-containers-fedora-28-master/2bee7a7/ : SUCCESS in 36m 09s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/96/654396/3/openstack-check/tripleo-build-containers-fedora-28-stein/96f9c04/ : SUCCESS in 37m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b28203bd71400790c6be76bd76349e51b3c2428a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 09:13:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/96/654396/3/check/test-release-openstack/7c5dbff/ : SUCCESS in 2m 14s\n- kolla-build-centos-binary http://logs.openstack.org/96/654396/3/check/kolla-build-centos-binary/40ab770/ : SUCCESS in 46m 08s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/96/654396/3/check/kolla-build-centos-source/e3e1e83/ : SUCCESS in 52m 43s\n- kolla-ansible-centos-source http://logs.openstack.org/96/654396/3/check/kolla-ansible-centos-source/7b60743/ : SUCCESS in 1h 10m 00s\n- kolla-ansible-centos-binary http://logs.openstack.org/96/654396/3/check/kolla-ansible-centos-binary/2d2f52f/ : SUCCESS in 1h 00m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/96/654396/3/check/kolla-ansible-centos-source-upgrade/289e558/ : SUCCESS in 1h 29m 01s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/96/654396/3/check/tripleo-build-containers-centos-7/058e155/ : SUCCESS in 41m 47s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/96/654396/3/check/kolla-build-oraclelinux-binary/b209cd9/ : SUCCESS in 57m 42s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/96/654396/3/check/kolla-build-oraclelinux-source/84db400/ : SUCCESS in 51m 48s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/96/654396/3/check/kolla-ansible-oraclelinux-source/89a1c71/ : SUCCESS in 1h 19m 31s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/96/654396/3/check/kolla-ansible-oraclelinux-binary/cf6ae60/ : SUCCESS in 1h 02m 55s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/96/654396/3/check/openstack-tox-lower-constraints/d420515/ : SUCCESS in 4m 23s\n- openstack-tox-pep8 http://logs.openstack.org/96/654396/3/check/openstack-tox-pep8/fb51239/ : SUCCESS in 2m 50s\n- openstack-tox-py27 http://logs.openstack.org/96/654396/3/check/openstack-tox-py27/fee8ed1/ : SUCCESS in 3m 27s\n- openstack-tox-py36 http://logs.openstack.org/96/654396/3/check/openstack-tox-py36/18d9ecc/ : SUCCESS in 3m 23s\n- openstack-tox-py37 http://logs.openstack.org/96/654396/3/check/openstack-tox-py37/62d15bd/ : SUCCESS in 4m 01s\n- openstack-tox-docs http://logs.openstack.org/96/654396/3/check/openstack-tox-docs/125a3d2/html/ : SUCCESS in 2m 55s\n- build-openstack-releasenotes http://logs.openstack.org/96/654396/3/check/build-openstack-releasenotes/3733765/html/ : SUCCESS in 3m 51s\n- kolla-build-ubuntu-binary http://logs.openstack.org/96/654396/3/check/kolla-build-ubuntu-binary/14c7335/ : SUCCESS in 41m 10s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/96/654396/3/check/kolla-build-ubuntu-source/e3395f2/ : SUCCESS in 38m 47s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/96/654396/3/check/kolla-ansible-ubuntu-source/f13133a/ : SUCCESS in 53m 31s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/96/654396/3/check/kolla-ansible-ubuntu-binary/f4ee6a3/ : SUCCESS in 48m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/96/654396/3/check/kolla-ansible-ubuntu-source-upgrade/de488a7/ : SUCCESS in 1h 05m 31s","accounts_in_message":[],"_revision_number":3},{"id":"a7c84b13ec95013f70d539e8ed61b7c53e005fac","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-24 11:22:02.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"80df6d613be765a8f088b1b583202eb76840ca84","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-04-28 01:41:04.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"801ca275d56055b4e39f4751e114dabe9fc0845a","author":{"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},"date":"2019-04-28 07:31:23.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f0158bbe7b1694017f71bae07611eae8d7aec70d","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-29 04:35:42.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3e1ac78132cd339c2c9c7b7a7b015acbcc273d51","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-04-29 13:17:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f4bac922f384bf6bbef212e5401a60e07c0b9fd1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 13:17:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"956d34f836398867e8a5573f4aab9b0c9a793160","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 14:30:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/96/654396/3/gate/test-release-openstack/b029bcc/ : SUCCESS in 1m 48s\n- kolla-build-centos-source http://logs.openstack.org/96/654396/3/gate/kolla-build-centos-source/49617b4/ : SUCCESS in 58m 15s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/96/654396/3/gate/tripleo-build-containers-centos-7/5638f20/ : SUCCESS in 43m 46s\n- kolla-build-oraclelinux-source http://logs.openstack.org/96/654396/3/gate/kolla-build-oraclelinux-source/01ee2d8/ : SUCCESS in 59m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/654396/3/gate/openstack-tox-lower-constraints/1233404/ : SUCCESS in 3m 16s\n- openstack-tox-pep8 http://logs.openstack.org/96/654396/3/gate/openstack-tox-pep8/02ed517/ : SUCCESS in 3m 20s\n- openstack-tox-py27 http://logs.openstack.org/96/654396/3/gate/openstack-tox-py27/7799b90/ : SUCCESS in 2m 41s\n- openstack-tox-py36 http://logs.openstack.org/96/654396/3/gate/openstack-tox-py36/92f82cf/ : SUCCESS in 3m 43s\n- openstack-tox-py37 http://logs.openstack.org/96/654396/3/gate/openstack-tox-py37/1aac273/ : SUCCESS in 3m 29s\n- openstack-tox-docs http://logs.openstack.org/96/654396/3/gate/openstack-tox-docs/3fcfbca/html/ : SUCCESS in 2m 42s\n- build-openstack-releasenotes http://logs.openstack.org/96/654396/3/gate/build-openstack-releasenotes/a1a371f/html/ : SUCCESS in 3m 50s\n- kolla-build-ubuntu-source http://logs.openstack.org/96/654396/3/gate/kolla-build-ubuntu-source/d799aaa/ : SUCCESS in 38m 36s","accounts_in_message":[],"_revision_number":3},{"id":"23ee057c354d477dbbc6f30d4aaae76d709602c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 14:30:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4fc5f1cc6ec7a3722f518ca12cca50677ac801cb","revisions":{"0c2c8f799152e3024c8bafff824057a6350a98f9":{"kind":"REWORK","_number":1,"created":"2019-04-22 07:28:42.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/96/654396/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/654396/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/654396/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/654396/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/654396/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/654396/1"}}},"commit":{"parents":[{"commit":"4e17e475eddc613f00bf84c4adc298bfd5a35bf3","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e17e475eddc613f00bf84c4adc298bfd5a35bf3"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-22 07:24:23.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-22 07:28:25.000000000","tz":480},"subject":"Move to opendev","message":"Move to opendev\n\n1.Use opendev.org instead of git.openstack.org.\n2.Use review.opendev.org instead of review.openstack.org.\n\nChange-Id: I377ea63827b6f3995ee275fdc466c1b2100eafc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c2c8f799152e3024c8bafff824057a6350a98f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c2c8f799152e3024c8bafff824057a6350a98f9"}]},"branch":"refs/heads/master"},"bed03952d50450c77d254c113b7415dd47b0bd7d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-04-23 05:30:59.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/96/654396/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/654396/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/654396/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/654396/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/654396/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/654396/2"}}},"commit":{"parents":[{"commit":"4e17e475eddc613f00bf84c4adc298bfd5a35bf3","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e17e475eddc613f00bf84c4adc298bfd5a35bf3"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-22 07:24:23.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-23 05:30:38.000000000","tz":480},"subject":"Move to opendev","message":"Move to opendev\n\n1.Use opendev.org instead of git.openstack.org.\n2.Use review.opendev.org instead of review.openstack.org.\n\nYou can see the discussion below:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-March/003603.html\n\nChange-Id: I377ea63827b6f3995ee275fdc466c1b2100eafc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bed03952d50450c77d254c113b7415dd47b0bd7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bed03952d50450c77d254c113b7415dd47b0bd7d"}]},"branch":"refs/heads/master"},"4fc5f1cc6ec7a3722f518ca12cca50677ac801cb":{"kind":"REWORK","_number":3,"created":"2019-04-24 07:23:31.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/96/654396/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/654396/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/654396/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/654396/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/654396/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/654396/3"}}},"commit":{"parents":[{"commit":"4e17e475eddc613f00bf84c4adc298bfd5a35bf3","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e17e475eddc613f00bf84c4adc298bfd5a35bf3"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-22 07:24:23.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-24 07:22:57.000000000","tz":480},"subject":"Move to opendev","message":"Move to opendev\n\n1.Use opendev.org instead of git.openstack.org.\n2.Use review.opendev.org instead of review.openstack.org.\n\nYou can see the discussion below:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-March/003603.html\n\nChange-Id: I377ea63827b6f3995ee275fdc466c1b2100eafc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4fc5f1cc6ec7a3722f518ca12cca50677ac801cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4fc5f1cc6ec7a3722f518ca12cca50677ac801cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
