)]}'
{"id":"openstack%2Fkolla~655369","triplet_id":"openstack%2Fkolla~master~I6c5746606ab59fd88e35525ecc17a62bfa9211e6","project":"openstack/kolla","branch":"master","topic":"bug/1825784","hashtags":[],"change_id":"I6c5746606ab59fd88e35525ecc17a62bfa9211e6","subject":"Install udevadm in ironic-conductor image","status":"MERGED","created":"2019-04-24 09:58:38.000000000","updated":"2019-04-25 13:24:25.000000000","submitted":"2019-04-25 12:39:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655369-1556195979542-7a5aba1f","meta_rev_id":"d4f9dd050d9f67b859575d13cbef80d5c38c6558","_number":655369,"virtual_id_number":655369,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-04-25 12:39:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-25 10:20:49.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":2,"date":"2019-04-24 14:50:33.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":2,"date":"2019-04-24 18:36:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":1,"date":"2019-04-24 18:36:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-04-24 11:37:25.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 14:50:33.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-04-24 18:36:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-04-25 10:20:49.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-25 12:39:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db19a6c7a84da4590760077e3d121a377d5f4b43","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-24 09:58:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02d638a0ea084884a2b8fa962c3b5cdb766acf05","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-24 10:40:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/69/655369/1/openstack-check/tripleo-build-containers-fedora-28-master/ffbe3c8/ : SUCCESS in 34m 42s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/69/655369/1/openstack-check/tripleo-build-containers-fedora-28-stein/b64c5b7/ : SUCCESS in 39m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a3ed79730d8ceaf957d90a5639c4951a72c82d2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 11:33:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/69/655369/1/check/kolla-build-centos-binary/76a7cd6/ : SUCCESS in 41m 33s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/655369/1/check/kolla-build-centos-source/3ba120e/ : SUCCESS in 55m 13s\n- kolla-ansible-centos-source http://logs.openstack.org/69/655369/1/check/kolla-ansible-centos-source/5991c6d/ : SUCCESS in 1h 08m 02s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/655369/1/check/kolla-ansible-centos-binary/ea69866/ : SUCCESS in 1h 12m 18s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/69/655369/1/check/kolla-ansible-centos-source-upgrade/f457ab3/ : SUCCESS in 1h 18m 05s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/69/655369/1/check/tripleo-build-containers-centos-7/74d6608/ : SUCCESS in 39m 38s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/655369/1/check/kolla-build-oraclelinux-binary/ae17f15/ : SUCCESS in 48m 00s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/655369/1/check/kolla-build-oraclelinux-source/b7723fc/ : SUCCESS in 49m 26s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/655369/1/check/kolla-ansible-oraclelinux-source/fa1a837/ : SUCCESS in 1h 14m 26s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/655369/1/check/kolla-ansible-oraclelinux-binary/4021d57/ : SUCCESS in 1h 14m 10s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/69/655369/1/check/openstack-tox-lower-constraints/d9843b4/ : SUCCESS in 3m 47s\n- openstack-tox-pep8 http://logs.openstack.org/69/655369/1/check/openstack-tox-pep8/6f1f23f/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/69/655369/1/check/openstack-tox-py27/0dce56b/ : SUCCESS in 3m 07s\n- openstack-tox-py36 http://logs.openstack.org/69/655369/1/check/openstack-tox-py36/021c7e6/ : SUCCESS in 3m 03s\n- openstack-tox-py37 http://logs.openstack.org/69/655369/1/check/openstack-tox-py37/b369212/ : SUCCESS in 3m 27s\n- openstack-tox-docs http://logs.openstack.org/69/655369/1/check/openstack-tox-docs/a9bbcc1/html/ : SUCCESS in 2m 20s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/655369/1/check/kolla-build-ubuntu-binary/79fb396/ : SUCCESS in 35m 40s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/655369/1/check/kolla-build-ubuntu-source/58bef4e/ : SUCCESS in 38m 41s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/655369/1/check/kolla-ansible-ubuntu-source/7d24518/ : SUCCESS in 50m 02s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/655369/1/check/kolla-ansible-ubuntu-binary/b6209e6/ : SUCCESS in 56m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/69/655369/1/check/kolla-ansible-ubuntu-source-upgrade/be3f470/ : SUCCESS in 1h 04m 42s","accounts_in_message":[],"_revision_number":1},{"id":"83c3cd11d3beb94f7aad23d0e8b79a14644edd05","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-24 11:37:25.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"536ab399018d336cb812dc129bdb61aff115f00d","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-04-24 14:50:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"58527b4226d75fed57b8dcc8b0ea22f084259c4a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-04-24 18:36:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c42b7df120e74bd2baba92f7ec9585588490eb80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 18:36:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"61ffe80eb6ec571195c34caa8da541511d26e520","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:54:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source http://logs.openstack.org/69/655369/1/gate/kolla-build-centos-source/97f2307/ : SUCCESS in 1h 02m 47s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/69/655369/1/gate/tripleo-build-containers-centos-7/97f1b39/ : SUCCESS in 44m 51s\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/655369/1/gate/kolla-build-oraclelinux-source/c9f13f2/ : SUCCESS in 54m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/655369/1/gate/openstack-tox-lower-constraints/3ffaa39/ : SUCCESS in 5m 14s\n- openstack-tox-pep8 http://logs.openstack.org/69/655369/1/gate/openstack-tox-pep8/3514b47/ : SUCCESS in 4m 55s\n- openstack-tox-py27 http://logs.openstack.org/69/655369/1/gate/openstack-tox-py27/0576bfd/ : SUCCESS in 2m 25s\n- openstack-tox-py36 http://logs.openstack.org/69/655369/1/gate/openstack-tox-py36/abfd879/ : SUCCESS in 2m 51s\n- openstack-tox-py37 http://logs.openstack.org/69/655369/1/gate/openstack-tox-py37/1baa195/ : SUCCESS in 7m 05s\n- openstack-tox-docs http://logs.openstack.org/69/655369/1/gate/openstack-tox-docs/7d433f6/html/ : SUCCESS in 4m 25s\n- kolla-build-ubuntu-source http://logs.openstack.org/69/655369/1/gate/kolla-build-ubuntu-source/509c989/ : RETRY_LIMIT in 1m 39s","accounts_in_message":[],"_revision_number":1},{"id":"c344ef8506446d3a49e873c764fe5c69af31b644","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-25 10:00:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"da201ea89be268d3f45c601040d8add85fa135b5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-25 10:20:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/69/655369/1/openstack-check/tripleo-build-containers-fedora-28-master/19b8396/ : RETRY_LIMIT in 5m 12s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/69/655369/1/openstack-check/tripleo-build-containers-fedora-28-stein/fd44bd7/ : RETRY_LIMIT in 2m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a375aa3c3625824f726a35c28d205d27cdf32d6f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 11:42:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/69/655369/1/check/kolla-build-centos-binary/ad8d90a/ : SUCCESS in 39m 18s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/655369/1/check/kolla-build-centos-source/7256de8/ : SUCCESS in 52m 33s\n- kolla-ansible-centos-source http://logs.openstack.org/69/655369/1/check/kolla-ansible-centos-source/404a927/ : SUCCESS in 1h 10m 16s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/655369/1/check/kolla-ansible-centos-binary/14453fd/ : SUCCESS in 1h 08m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/69/655369/1/check/kolla-ansible-centos-source-upgrade/66a6007/ : SUCCESS in 1h 28m 44s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/69/655369/1/check/tripleo-build-containers-centos-7/256da7a/ : SUCCESS in 46m 37s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/655369/1/check/kolla-build-oraclelinux-binary/b6a3e0b/ : SUCCESS in 53m 52s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/655369/1/check/kolla-build-oraclelinux-source/e62b17c/ : SUCCESS in 55m 49s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/655369/1/check/kolla-ansible-oraclelinux-source/98eef92/ : SUCCESS in 1h 18m 23s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/655369/1/check/kolla-ansible-oraclelinux-binary/7820c37/ : SUCCESS in 1h 10m 23s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/69/655369/1/check/openstack-tox-lower-constraints/e23ac85/ : SUCCESS in 3m 39s\n- openstack-tox-pep8 http://logs.openstack.org/69/655369/1/check/openstack-tox-pep8/a5a7233/ : SUCCESS in 3m 07s\n- openstack-tox-py27 http://logs.openstack.org/69/655369/1/check/openstack-tox-py27/87c2f34/ : SUCCESS in 4m 47s\n- openstack-tox-py36 http://logs.openstack.org/69/655369/1/check/openstack-tox-py36/e38f55d/ : SUCCESS in 2m 45s\n- openstack-tox-py37 http://logs.openstack.org/69/655369/1/check/openstack-tox-py37/2200a14/ : SUCCESS in 4m 02s\n- openstack-tox-docs http://logs.openstack.org/69/655369/1/check/openstack-tox-docs/6b9a65c/html/ : SUCCESS in 2m 29s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/655369/1/check/kolla-build-ubuntu-binary/9face87/ : SUCCESS in 30m 54s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/655369/1/check/kolla-build-ubuntu-source/18dbde3/ : SUCCESS in 39m 27s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/655369/1/check/kolla-ansible-ubuntu-source/9b49ed2/ : SUCCESS in 50m 04s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/655369/1/check/kolla-ansible-ubuntu-binary/24b72a3/ : SUCCESS in 44m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/69/655369/1/check/kolla-ansible-ubuntu-source-upgrade/4f73606/ : SUCCESS in 57m 40s","accounts_in_message":[],"_revision_number":1},{"id":"1cc5404219c1731e55843bbdf0d012cc17a1beeb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 11:42:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"883399e4daca6366cd64127277aecfb378ed3221","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 12:39:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/69/655369/1/gate/kolla-build-centos-source/78ff5bb/ : SUCCESS in 50m 11s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/69/655369/1/gate/tripleo-build-containers-centos-7/87d110a/ : SUCCESS in 54m 03s\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/655369/1/gate/kolla-build-oraclelinux-source/5ff0544/ : SUCCESS in 51m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/655369/1/gate/openstack-tox-lower-constraints/975820a/ : SUCCESS in 3m 31s\n- openstack-tox-pep8 http://logs.openstack.org/69/655369/1/gate/openstack-tox-pep8/5218847/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/69/655369/1/gate/openstack-tox-py27/3383e86/ : SUCCESS in 3m 03s\n- openstack-tox-py36 http://logs.openstack.org/69/655369/1/gate/openstack-tox-py36/5405d60/ : SUCCESS in 2m 54s\n- openstack-tox-py37 http://logs.openstack.org/69/655369/1/gate/openstack-tox-py37/764b896/ : SUCCESS in 3m 45s\n- openstack-tox-docs http://logs.openstack.org/69/655369/1/gate/openstack-tox-docs/512228b/html/ : SUCCESS in 3m 04s\n- kolla-build-ubuntu-source http://logs.openstack.org/69/655369/1/gate/kolla-build-ubuntu-source/9c9fc94/ : SUCCESS in 36m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8982f9db6351cf6866a6bceff9f56dd5b1277dd3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 12:39:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"41ffe31c49b33946271263c90d6dea4ebd21bf33","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-25 13:19:03.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 3e9b4406a499f395975992e016d03d60a6c61037","accounts_in_message":[],"_revision_number":1},{"id":"4bdcd69f01ad0d38d70c8a66b4ee4430ccd6098a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-25 13:24:25.000000000","message":"Patch Set 1:\n\nRocky: https://review.opendev.org/655690\nQueens: https://review.opendev.org/655691","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"856b2c8824967e582c3d2e98851b58efab6757f8","revisions":{"856b2c8824967e582c3d2e98851b58efab6757f8":{"kind":"REWORK","_number":1,"created":"2019-04-24 09:58:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/69/655369/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/655369/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/655369/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/655369/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/655369/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/655369/1"}}},"commit":{"parents":[{"commit":"4e17e475eddc613f00bf84c4adc298bfd5a35bf3","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e17e475eddc613f00bf84c4adc298bfd5a35bf3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-04-24 09:54:50.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-04-24 09:54:50.000000000","tz":60},"subject":"Install udevadm in ironic-conductor image","message":"Install udevadm in ironic-conductor image\n\nbackport: rocky, queens\n\nA dependency on udevadm was recently added to ironic-lib:\n\nhttps://review.opendev.org/#/q/I9db48a3462422749290bbb887c14816734ab0478\n\nChange-Id: I6c5746606ab59fd88e35525ecc17a62bfa9211e6\nCloses-Bug: #1825784\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/856b2c8824967e582c3d2e98851b58efab6757f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/856b2c8824967e582c3d2e98851b58efab6757f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
