)]}'
{"id":"openstack%2Fkolla~655416","triplet_id":"openstack%2Fkolla~master~I3cd9a23af1e925b7e0d739ba2715412ef5911e74","project":"openstack/kolla","branch":"master","topic":"oschecks-deprecate","hashtags":[],"change_id":"I3cd9a23af1e925b7e0d739ba2715412ef5911e74","subject":"Remove oschecks installation","status":"MERGED","created":"2019-04-24 11:50:32.000000000","updated":"2019-04-25 06:08:49.000000000","submitted":"2019-04-24 15:56:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655416-1556121410318-dfe5b5ea","meta_rev_id":"a53fa7d93738d73ea0a7fc5f02ac36bdfb1923cf","_number":655416,"virtual_id_number":655416,"owner":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-04-24 15:56:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-24 13:02:55.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":2,"date":"2019-04-24 14:50:04.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":2,"date":"2019-04-24 12:57:23.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":1,"date":"2019-04-24 14:50:04.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 12:57:23.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-04-24 13:02:55.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-24 14:50:04.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-04-24 15:56:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0db1cd50cf467ddc2804694ac6404dafac5dcdba","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2019-04-24 11:50:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a3730323f45f48dd8e1b67fbaadaac137f0b122","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-24 12:57:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3895a85119f0263a338caedaa3f68810a653a3cc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-24 13:02:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/16/655416/1/openstack-check/tripleo-build-containers-fedora-28-master/36a80de/ : SUCCESS in 34m 29s (non-voting)\n- tripleo-build-containers-fedora-28-stein tripleo-build-containers-fedora-28-stein : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4fd030519d31c69eb64c148e97cd4dc59f48052","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 13:35:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/16/655416/1/check/kolla-build-centos-binary/f05c024/ : SUCCESS in 40m 39s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/16/655416/1/check/kolla-build-centos-source/d20aacb/ : SUCCESS in 55m 19s\n- kolla-ansible-centos-source http://logs.openstack.org/16/655416/1/check/kolla-ansible-centos-source/6f7ce5c/ : SUCCESS in 1h 14m 57s\n- kolla-ansible-centos-binary http://logs.openstack.org/16/655416/1/check/kolla-ansible-centos-binary/c649d8b/ : SUCCESS in 1h 01m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/16/655416/1/check/kolla-ansible-centos-source-upgrade/76f198c/ : SUCCESS in 1h 28m 36s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/16/655416/1/check/tripleo-build-containers-centos-7/39d7796/ : SUCCESS in 40m 42s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/16/655416/1/check/kolla-build-oraclelinux-binary/0d0a605/ : SUCCESS in 53m 58s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/16/655416/1/check/kolla-build-oraclelinux-source/0b5576f/ : SUCCESS in 58m 14s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/16/655416/1/check/kolla-ansible-oraclelinux-source/4222f3a/ : SUCCESS in 1h 10m 42s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/16/655416/1/check/kolla-ansible-oraclelinux-binary/3bc59e8/ : SUCCESS in 1h 14m 51s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/16/655416/1/check/openstack-tox-lower-constraints/741e332/ : SUCCESS in 3m 38s\n- openstack-tox-pep8 http://logs.openstack.org/16/655416/1/check/openstack-tox-pep8/41b7d57/ : SUCCESS in 3m 09s\n- openstack-tox-py27 http://logs.openstack.org/16/655416/1/check/openstack-tox-py27/3d44343/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/16/655416/1/check/openstack-tox-py36/e994945/ : SUCCESS in 3m 13s\n- openstack-tox-py37 http://logs.openstack.org/16/655416/1/check/openstack-tox-py37/031d184/ : SUCCESS in 3m 24s\n- openstack-tox-docs http://logs.openstack.org/16/655416/1/check/openstack-tox-docs/6c1547b/html/ : SUCCESS in 2m 40s\n- kolla-build-ubuntu-binary http://logs.openstack.org/16/655416/1/check/kolla-build-ubuntu-binary/6203bc2/ : SUCCESS in 29m 56s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/16/655416/1/check/kolla-build-ubuntu-source/6485c91/ : SUCCESS in 35m 10s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/16/655416/1/check/kolla-ansible-ubuntu-source/5080e76/ : SUCCESS in 54m 38s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/16/655416/1/check/kolla-ansible-ubuntu-binary/8d7532e/ : SUCCESS in 46m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/16/655416/1/check/kolla-ansible-ubuntu-source-upgrade/dd9c931/ : SUCCESS in 1h 13m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d6135c64d2f7e5900d4b1b6b88773a613c1ed8ac","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-04-24 14:50:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ea1d5b890003098119b60832e5e96ab32fa221d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 14:50:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac67b2e1b1729ed2dbc93e3f530fc8e561fa0462","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 15:56:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/16/655416/1/gate/kolla-build-centos-source/0d2eb69/ : SUCCESS in 47m 49s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/16/655416/1/gate/tripleo-build-containers-centos-7/a0f7e84/ : SUCCESS in 49m 21s\n- kolla-build-oraclelinux-source http://logs.openstack.org/16/655416/1/gate/kolla-build-oraclelinux-source/0468df0/ : SUCCESS in 49m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/655416/1/gate/openstack-tox-lower-constraints/0fa77e2/ : SUCCESS in 3m 50s\n- openstack-tox-pep8 http://logs.openstack.org/16/655416/1/gate/openstack-tox-pep8/1865d1b/ : SUCCESS in 3m 00s\n- openstack-tox-py27 http://logs.openstack.org/16/655416/1/gate/openstack-tox-py27/fef230f/ : SUCCESS in 3m 03s\n- openstack-tox-py36 http://logs.openstack.org/16/655416/1/gate/openstack-tox-py36/357fd77/ : SUCCESS in 3m 12s\n- openstack-tox-py37 http://logs.openstack.org/16/655416/1/gate/openstack-tox-py37/ea3bf4e/ : SUCCESS in 3m 41s\n- openstack-tox-docs http://logs.openstack.org/16/655416/1/gate/openstack-tox-docs/9f9b44a/html/ : SUCCESS in 3m 09s\n- kolla-build-ubuntu-source http://logs.openstack.org/16/655416/1/gate/kolla-build-ubuntu-source/e8cca5e/ : SUCCESS in 36m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f48e8c7131116c5ff1a5a8eb12700aa4a42070dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 15:56:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3c72eb0faf72180dc739ccc013a484b5b7c95179","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2019-04-25 06:08:49.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 10f0e55fd7df0fce2c737e7d47fb80d52abb883e","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"45c96112008a9fe22b3aa7ce75ce314ab9acc711","revisions":{"45c96112008a9fe22b3aa7ce75ce314ab9acc711":{"kind":"REWORK","_number":1,"created":"2019-04-24 11:50:32.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/16/655416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/655416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/655416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/655416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/655416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/655416/1"}}},"commit":{"parents":[{"commit":"4e17e475eddc613f00bf84c4adc298bfd5a35bf3","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e17e475eddc613f00bf84c4adc298bfd5a35bf3"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-04-24 11:48:32.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-04-24 11:48:32.000000000","tz":120},"subject":"Remove oschecks installation","message":"Remove oschecks installation\n\nThe projects is almost dead and has been deprecated in TripleO project.\nThis patch removes installation of the package on RPM based distros.\n\nChange-Id: I3cd9a23af1e925b7e0d739ba2715412ef5911e74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45c96112008a9fe22b3aa7ce75ce314ab9acc711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45c96112008a9fe22b3aa7ce75ce314ab9acc711"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
