)]}'
{"id":"openstack%2Fkolla~655537","triplet_id":"openstack%2Fkolla~master~Ife905cb6b5dff201da357aecf100a34e599bbe80","project":"openstack/kolla","branch":"master","topic":"feature/remove_bandit_blacklist","hashtags":[],"change_id":"Ife905cb6b5dff201da357aecf100a34e599bbe80","subject":"Remove skipping B303 for bandit checks","status":"MERGED","created":"2019-04-24 19:11:49.000000000","updated":"2020-02-20 23:00:00.000000000","submitted":"2020-02-20 21:06:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655537-1582232795349-3390009c","meta_rev_id":"dd6e703d0e6b19d0bbac4a0fd8276b78fe64a50a","_number":655537,"virtual_id_number":655537,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-20 21:06:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"recommended":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"all":[{"value":1,"date":"2019-04-29 02:57:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-04-29 13:17:53.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-04-25 09:50:53.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-02-20 19:19:57.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"date":"2020-02-20 19:11:07.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2019-04-29 13:17:53.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-02-20 19:19:57.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-25 09:50:53.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-04-29 02:57:41.000000000","updated_by":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"reviewer":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"state":"REVIEWER"},{"updated":"2019-04-29 13:17:53.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-02-20 19:19:57.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-02-20 21:06:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b2c8a7bb2ad9f37199560620998adad0350019f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-04-24 19:11:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6c450002b00d0dd575943d3ac8e60d940e02dfd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 20:44:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/37/655537/1/check/kolla-build-centos-binary/0fffa34/ : SUCCESS in 40m 16s (non-voting)\n- kolla-build-centos-source finger://ze09.openstack.org/6b31eb913bf8424983c94a89698bc5cf : RETRY_LIMIT in 46m 09s\n- kolla-ansible-centos-source http://logs.openstack.org/37/655537/1/check/kolla-ansible-centos-source/6e68c8c/ : SUCCESS in 1h 08m 45s\n- kolla-ansible-centos-binary finger://ze05.openstack.org/746946c6d3824da08a45a6cc4be910ac : RETRY_LIMIT in 1h 00m 26s (non-voting)\n- kolla-ansible-centos-source-upgrade finger://ze07.openstack.org/b9127ab575544821a2fb4949e770052f : RETRY_LIMIT in 1h 04m 06s\n- kolla-build-oraclelinux-binary finger://ze10.openstack.org/e03d6afb47684424a7ea2f7745e4684f : RETRY_LIMIT in 38m 31s (non-voting)\n- kolla-build-oraclelinux-source finger://ze08.openstack.org/65ce87525bff4c58bca3c00c8ed1c7dd : RETRY_LIMIT in 56m 09s\n- kolla-ansible-oraclelinux-source finger://ze01.openstack.org/e13aaad7d9404ec7af03d8ebb01c3aa1 : RETRY_LIMIT in 58m 17s (non-voting)\n- kolla-ansible-oraclelinux-binary finger://ze07.openstack.org/5611c2113ed6433986e53e8667752485 : RETRY_LIMIT in 59m 23s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/37/655537/1/check/openstack-tox-lower-constraints/dce7576/ : SUCCESS in 4m 02s\n- openstack-tox-pep8 http://logs.openstack.org/37/655537/1/check/openstack-tox-pep8/eb172af/ : FAILURE in 4m 06s\n- openstack-tox-py27 http://logs.openstack.org/37/655537/1/check/openstack-tox-py27/2e855d4/ : SUCCESS in 6m 18s\n- openstack-tox-py36 http://logs.openstack.org/37/655537/1/check/openstack-tox-py36/d683a5d/ : SUCCESS in 3m 32s\n- openstack-tox-py37 http://logs.openstack.org/37/655537/1/check/openstack-tox-py37/bf8ed1c/ : SUCCESS in 3m 43s\n- openstack-tox-docs http://logs.openstack.org/37/655537/1/check/openstack-tox-docs/6f4e813/html/ : SUCCESS in 2m 37s\n- build-openstack-releasenotes http://logs.openstack.org/37/655537/1/check/build-openstack-releasenotes/664c4aa/html/ : SUCCESS in 3m 28s\n- kolla-build-ubuntu-binary http://logs.openstack.org/37/655537/1/check/kolla-build-ubuntu-binary/e87ea29/ : FAILURE in 4m 05s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/37/655537/1/check/kolla-build-ubuntu-source/51b0e67/ : SUCCESS in 32m 26s\n- kolla-ansible-ubuntu-source finger://ze04.openstack.org/2238760043c24ddfac754a7311b439c3 : RETRY_LIMIT in 1h 00m 23s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/37/655537/1/check/kolla-ansible-ubuntu-binary/3299b71/ : SUCCESS in 47m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade finger://ze04.openstack.org/446078bbed4b4486a5fb865ba829a380 : RETRY_LIMIT in 1h 16m 26s","accounts_in_message":[],"_revision_number":1},{"id":"3686644b930af4d90fa7f107544dd4839ce0e26f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-04-25 07:08:59.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ec37adae179ec328d4c234d50ad891bb8bcf51bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 09:28:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/37/655537/2/check/kolla-build-centos-binary/161eaca/ : SUCCESS in 46m 37s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/37/655537/2/check/kolla-build-centos-source/c217b31/ : SUCCESS in 56m 28s\n- kolla-ansible-centos-source http://logs.openstack.org/37/655537/2/check/kolla-ansible-centos-source/51da7b7/ : SUCCESS in 1h 13m 37s\n- kolla-ansible-centos-binary http://logs.openstack.org/37/655537/2/check/kolla-ansible-centos-binary/707218e/ : SUCCESS in 1h 05m 16s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/37/655537/2/check/kolla-ansible-centos-source-upgrade/f8f537c/ : SUCCESS in 1h 21m 38s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/37/655537/2/check/kolla-build-oraclelinux-binary/f5c4f47/ : SUCCESS in 59m 02s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/37/655537/2/check/kolla-build-oraclelinux-source/6374f95/ : SUCCESS in 56m 56s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/37/655537/2/check/kolla-ansible-oraclelinux-source/3306b4d/ : SUCCESS in 1h 21m 04s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/37/655537/2/check/kolla-ansible-oraclelinux-binary/1218569/ : FAILURE in 20m 05s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/37/655537/2/check/openstack-tox-lower-constraints/5b60f08/ : SUCCESS in 3m 39s\n- openstack-tox-pep8 http://logs.openstack.org/37/655537/2/check/openstack-tox-pep8/c3d5670/ : SUCCESS in 2m 58s\n- openstack-tox-py27 http://logs.openstack.org/37/655537/2/check/openstack-tox-py27/b1a24d6/ : SUCCESS in 3m 26s\n- openstack-tox-py36 http://logs.openstack.org/37/655537/2/check/openstack-tox-py36/81237a6/ : SUCCESS in 3m 17s\n- openstack-tox-py37 http://logs.openstack.org/37/655537/2/check/openstack-tox-py37/3beed56/ : SUCCESS in 3m 22s\n- openstack-tox-docs http://logs.openstack.org/37/655537/2/check/openstack-tox-docs/9893573/html/ : SUCCESS in 2m 34s\n- build-openstack-releasenotes http://logs.openstack.org/37/655537/2/check/build-openstack-releasenotes/bf70b60/html/ : SUCCESS in 3m 53s\n- kolla-build-ubuntu-binary http://logs.openstack.org/37/655537/2/check/kolla-build-ubuntu-binary/ac82ca4/ : SUCCESS in 27m 48s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/37/655537/2/check/kolla-build-ubuntu-source/8d44133/ : SUCCESS in 28m 57s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/37/655537/2/check/kolla-ansible-ubuntu-source/a059e06/ : SUCCESS in 50m 21s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/37/655537/2/check/kolla-ansible-ubuntu-binary/a5fe97a/ : SUCCESS in 49m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/37/655537/2/check/kolla-ansible-ubuntu-source-upgrade/a93bf78/ : SUCCESS in 1h 00m 46s","accounts_in_message":[],"_revision_number":2},{"id":"b7f5b26dac2aa009c7e0a83e070a93bdad6a2402","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-25 09:50:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cf38c87788a8bf021f3150787d39bc0dd5b62b64","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-04-29 02:57:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"762e6c591469f9f8f0405e1ac091bfaa280023a8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-04-29 13:17:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7bbef049e81c9ee593ce75aec00deb90a0dc9b0a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-17 14:34:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"54a8940f69bb168957eb5232f6aa0eba3434506c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 14:34:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"343e6389a6692b6f8f6d46da92231c1386531097","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 15:36:07.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source http://logs.openstack.org/37/655537/2/gate/kolla-build-centos-source/05235cd/ : SUCCESS in 56m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/37/655537/2/gate/openstack-tox-lower-constraints/24630d6/ : SUCCESS in 5m 22s\n- openstack-tox-pep8 http://logs.openstack.org/37/655537/2/gate/openstack-tox-pep8/65765fd/ : FAILURE in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/37/655537/2/gate/openstack-tox-py27/b9c2198/ : SUCCESS in 4m 27s\n- openstack-tox-py36 http://logs.openstack.org/37/655537/2/gate/openstack-tox-py36/ad03039/ : SUCCESS in 4m 35s\n- openstack-tox-py37 http://logs.openstack.org/37/655537/2/gate/openstack-tox-py37/14ef5bb/ : SUCCESS in 4m 20s\n- openstack-tox-docs http://logs.openstack.org/37/655537/2/gate/openstack-tox-docs/5449bc7/html/ : SUCCESS in 3m 20s\n- build-openstack-releasenotes http://logs.openstack.org/37/655537/2/gate/build-openstack-releasenotes/abf7b82/html/ : SUCCESS in 4m 23s\n- kolla-build-ubuntu-source http://logs.openstack.org/37/655537/2/gate/kolla-build-ubuntu-source/acf0a02/ : SUCCESS in 35m 15s","accounts_in_message":[],"_revision_number":2},{"id":"e1cf8748787cf6ed8857aa3c51a50d6bc3b20d87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 16:05:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/37/655537/2/check/kolla-build-centos-binary/bbc7b8f/ : SUCCESS in 42m 46s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/37/655537/2/check/kolla-build-centos-source/0a3f64b/ : SUCCESS in 59m 23s\n- kolla-ansible-centos-source http://logs.openstack.org/37/655537/2/check/kolla-ansible-centos-source/168ea79/ : FAILURE in 49m 15s\n- kolla-ansible-centos-binary http://logs.openstack.org/37/655537/2/check/kolla-ansible-centos-binary/665c436/ : SUCCESS in 1h 13m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/37/655537/2/check/kolla-ansible-centos-source-upgrade/42fb8f7/ : SUCCESS in 1h 24m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/37/655537/2/check/openstack-tox-lower-constraints/740bd08/ : SUCCESS in 5m 16s\n- openstack-tox-pep8 http://logs.openstack.org/37/655537/2/check/openstack-tox-pep8/e5c6d47/ : FAILURE in 5m 15s\n- openstack-tox-py27 http://logs.openstack.org/37/655537/2/check/openstack-tox-py27/4d06f5e/ : SUCCESS in 4m 56s\n- openstack-tox-py36 http://logs.openstack.org/37/655537/2/check/openstack-tox-py36/1e179bc/ : SUCCESS in 3m 46s\n- openstack-tox-py37 http://logs.openstack.org/37/655537/2/check/openstack-tox-py37/2bcc536/ : SUCCESS in 4m 24s\n- openstack-tox-docs http://logs.openstack.org/37/655537/2/check/openstack-tox-docs/4956603/html/ : SUCCESS in 2m 47s\n- build-openstack-releasenotes http://logs.openstack.org/37/655537/2/check/build-openstack-releasenotes/ff77be2/html/ : SUCCESS in 4m 07s\n- kolla-build-ubuntu-binary http://logs.openstack.org/37/655537/2/check/kolla-build-ubuntu-binary/4bc50fe/ : SUCCESS in 23m 28s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/37/655537/2/check/kolla-build-ubuntu-source/85f7159/ : SUCCESS in 32m 58s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/37/655537/2/check/kolla-ansible-ubuntu-source/5c5170f/ : SUCCESS in 45m 55s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/37/655537/2/check/kolla-ansible-ubuntu-binary/ddde072/ : SUCCESS in 47m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/37/655537/2/check/kolla-ansible-ubuntu-source-upgrade/f26791b/ : SUCCESS in 1h 03m 50s","accounts_in_message":[],"_revision_number":2},{"id":"f20af8f009dc1f5dec99db954c6f875d10e33abf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-17 19:11:54.000000000","message":"Patch Set 2:\n\nsha1 still used","accounts_in_message":[],"_revision_number":2},{"id":"ad7710fec9318f3af09af40382a91a79d733ca62","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-05 09:11:10.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c4a6b0a760a65267946d9db971731cca4ba19a6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 10:58:22.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://logs.opendev.org/37/655537/2/check/kolla-build-centos-binary/b973618/ : SUCCESS in 32m 33s (non-voting)\n- kolla-build-centos-source https://logs.opendev.org/37/655537/2/check/kolla-build-centos-source/c0bf861/ : SUCCESS in 53m 06s\n- kolla-ansible-centos-source https://logs.opendev.org/37/655537/2/check/kolla-ansible-centos-source/d55a353/ : SUCCESS in 1h 21m 45s\n- kolla-ansible-centos-binary https://logs.opendev.org/37/655537/2/check/kolla-ansible-centos-binary/931c640/ : SUCCESS in 1h 13m 55s (non-voting)\n- kolla-ansible-centos-source-upgrade https://logs.opendev.org/37/655537/2/check/kolla-ansible-centos-source-upgrade/8d1fa75/ : SUCCESS in 1h 26m 07s\n- openstack-tox-lower-constraints https://logs.opendev.org/37/655537/2/check/openstack-tox-lower-constraints/10dbbc2/ : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://logs.opendev.org/37/655537/2/check/openstack-tox-pep8/fe55bee/ : FAILURE in 6m 05s\n- openstack-tox-py27 https://logs.opendev.org/37/655537/2/check/openstack-tox-py27/1a11a76/ : SUCCESS in 4m 39s\n- openstack-tox-py36 https://logs.opendev.org/37/655537/2/check/openstack-tox-py36/ea1ec4a/ : SUCCESS in 3m 45s\n- openstack-tox-py37 https://logs.opendev.org/37/655537/2/check/openstack-tox-py37/464f56b/ : SUCCESS in 3m 56s\n- openstack-tox-docs https://logs.opendev.org/37/655537/2/check/openstack-tox-docs/c733198/html/ : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://logs.opendev.org/37/655537/2/check/build-openstack-releasenotes/2d4b895/html/ : SUCCESS in 6m 10s\n- kolla-build-ubuntu-binary https://logs.opendev.org/37/655537/2/check/kolla-build-ubuntu-binary/69323b5/ : SUCCESS in 39m 01s (non-voting)\n- kolla-build-ubuntu-source https://logs.opendev.org/37/655537/2/check/kolla-build-ubuntu-source/c742b40/ : SUCCESS in 35m 02s\n- kolla-ansible-ubuntu-source https://logs.opendev.org/37/655537/2/check/kolla-ansible-ubuntu-source/c3d4a86/ : SUCCESS in 50m 01s\n- kolla-ansible-ubuntu-binary https://logs.opendev.org/37/655537/2/check/kolla-ansible-ubuntu-binary/a0cf27b/ : SUCCESS in 48m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://logs.opendev.org/37/655537/2/check/kolla-ansible-ubuntu-source-upgrade/f99da0c/ : SUCCESS in 1h 01m 18s","accounts_in_message":[],"_revision_number":2},{"id":"8932cf4c58300cd7b26b50d52cf770e56fc0d199","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-07 11:51:58.000000000","message":"Patch Set 2: Workflow-1\n\nIt needs to wait until we can remove sha1.","accounts_in_message":[],"_revision_number":2},{"id":"9256df1e78ef1a6f213dc33396a9b0d4fc3537f3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-09-06 13:28:25.000000000","message":"Removed Workflow+1 by Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"a373891bff8d9d06bdf7025007ffdd2afd6524d6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-09-26 09:06:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cdc6355f9513fea82e2a19a6432bb9faf7f5addf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 10:34:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/180e7059966f404b9282a0d7648ebc13 : SUCCESS in 1h 01m 35s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7f727d92a9cb49db9a39c3a9cb457483 : SUCCESS in 56m 17s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bef4506e96484564b1ae194c3163375d : SUCCESS in 1h 18m 22s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/743764584cb4489a8773f316b43b6a74 : SUCCESS in 1h 23m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b2b369220021450bb2ff61775bd83d3a : SUCCESS in 1h 17m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/73acdd839e2446c9ae45da4d5262ef92 : SUCCESS in 33m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37514e7f6fff4f27823c23d6b0f20a98 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17870b03c31649eb97371c1028cfabf8 : FAILURE in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/389f761572a7461ea5f81e507032713a : SUCCESS in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa8c74407cf04ac9bfedb248592ab87c : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f85b0cdb149b4c3e8b2c89b49a735418 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fe8acc3bd414aa59c996113ccfc2e85 : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36d43554f8974afa9e1a4d02ee6bb2ce : SUCCESS in 5m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/770078c32c9b47f78996fd267368ae33 : SUCCESS in 33m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/05d2c1a86cf240b18bd36576bac94fa1 : SUCCESS in 28m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e6d662340d93445791eb26f5b5a7256f : SUCCESS in 54m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/56c162a81a9e4ae2a2d467da96979335 : SUCCESS in 1h 02m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/36b1cd8f71a0471fa2ba40428b6827fe : SUCCESS in 1h 10m 43s","accounts_in_message":[],"_revision_number":2},{"id":"a6de2c9081342d24f6f8b157cdf9f70eee888434","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-03 09:33:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"03f0b2522bc67b401b23b141550e60bbd7693aa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 10:32:01.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9f2bf9333c0d488fafa662018d9bd712 : FAILURE in 8m 31s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0149e9ce54da4e46a57cdee95982bec6 : FAILURE in 8m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dfc35607865d426caa4a4cbacc6d1f98 : FAILURE in 5m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0490019601a4477c9d34264298e38652 : SUCCESS in 23m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c912d8221d3446ea7a8da670e535672 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba7a2a834f3b45c6a6c0c542ccad2cf2 : FAILURE in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a80ab0a11a5641aeb9b1e392edf89ba6 : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ddb36970dad4fea86ea89cd2f31295a : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/834604034f3042f5a16623add8251dbe : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0eb69959429d45c383bb996aec112df3 : SUCCESS in 4m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1631b78797fc4ec68d77efebbc60983d : SUCCESS in 28m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/525f9f79179b47609aa39b5853e6916c : SUCCESS in 47m 10s","accounts_in_message":[],"_revision_number":2},{"id":"b49c1fc4b8c49d5df50057dd268633f62e0d7aba","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-20 19:11:07.000000000","message":"Patch Set 2: -Workflow\n\nrecheck (ceph is gone!)","accounts_in_message":[],"_revision_number":2},{"id":"fae2c85ccb16173fc001fd65deb74609bc314be2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-20 19:19:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(I see pep8 passed already)","accounts_in_message":[],"_revision_number":2},{"id":"c3890d4a540687cefe7ea9ea410620e90103b1b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 20:25:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/fbbadc248fdd4aa68f61610b180f0ddb : SUCCESS in 1h 05m 51s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/34ae28bef4354c1bbb97f4186259e33c : SUCCESS in 1h 11m 42s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7bf73e858b074536a8a966470b013ff0 : SUCCESS in 32m 59s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3201b5b1f06542c6ac7ef34b2a594ebe : SUCCESS in 45m 01s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f486085e62a9461285e9cb773d08d427 : SUCCESS in 26m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ab48703a731842569eda38c7ea733235 : SUCCESS in 41m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d3b47deba344c29bcd17bc69524ccec : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0e63170270547ea85c448ac72779aff : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b1e28fdbb9d448bab6488fc108c839f : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3908ca6a2f384131ae881a4e08421ca6 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de80497a68e144a2b3e4931076db7a32 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4a1566fdfa246de978441948662bd94 : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fbb4fc717cf41cf8c06eef13f85019c : SUCCESS in 6m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/66ddb036c90e4fedbb9d16f452de8225 : SUCCESS in 47m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2606479993684511b75e3dd325ed5c15 : SUCCESS in 50m 06s","accounts_in_message":[],"_revision_number":2},{"id":"996ce43845bf33d2703bbc79ab9b715976e6596a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 20:25:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b47bf0de133eed4dc89eb139520156164a1eed1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 21:06:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"478f9f39628acfa4464f1199aab9367b3ba33998","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 21:06:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/042dce88ec1347ecbcc29132a3780b5c : SUCCESS in 38m 51s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e82f4a4b63534da8bd36e9b901e207d0 : SUCCESS in 27m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de589e04e2b64164a609824f32751cd1 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bf0271ad5a8491da47fac151ae41f5d : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/306541e2146848db86d902d03b6af1a6 : SUCCESS in 5m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bfc6e46969884922a3058b6a396a407d : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69e52483c4354b2ea3f6c39d5798c04e : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46f9d04268f84b4a979f7b0bf3e44a0e : SUCCESS in 6m 12s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/01400ad127ff4a8aa24666cc39569584 : SUCCESS in 34m 08s","accounts_in_message":[],"_revision_number":2},{"id":"a9c082b95bd982c23ac256927d1cd6e1c827d3dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 21:08:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b6cd8024c7e459c9a04f45e26b2661e : SUCCESS in 1m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05656d9a95a04bec927fd4bcadd6b641 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":2},{"id":"f5a5305c542c86d01453602a472f5d424f59f76d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 23:00:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/aab9db31f9b34874bdf1b201de825417 : TIMED_OUT in 3h 01m 15s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"075f1b71aa111b93f72f865acd3059de11946fdc","revisions":{"b07fbcb7236de4302e64c6e77b9661d6add93389":{"kind":"REWORK","_number":1,"created":"2019-04-24 19:11:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/37/655537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/655537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/655537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/655537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/655537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/655537/1"}}},"commit":{"parents":[{"commit":"cbdce1eff79da7b62c4cbc16bea2b98226d01233","subject":"Merge \"Update the min version of tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbdce1eff79da7b62c4cbc16bea2b98226d01233"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-04-24 19:11:35.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-04-24 19:11:35.000000000","tz":120},"subject":"Remove skipping B303 for bandit checks","message":"Remove skipping B303 for bandit checks\n\nChange-Id: Ife905cb6b5dff201da357aecf100a34e599bbe80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b07fbcb7236de4302e64c6e77b9661d6add93389"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b07fbcb7236de4302e64c6e77b9661d6add93389"}]},"branch":"refs/heads/master"},"075f1b71aa111b93f72f865acd3059de11946fdc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-04-25 07:08:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/37/655537/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/655537/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/655537/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/655537/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/655537/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/655537/2"}}},"commit":{"parents":[{"commit":"cbdce1eff79da7b62c4cbc16bea2b98226d01233","subject":"Merge \"Update the min version of tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbdce1eff79da7b62c4cbc16bea2b98226d01233"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-04-24 19:11:35.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-04-25 07:08:50.000000000","tz":0},"subject":"Remove skipping B303 for bandit checks","message":"Remove skipping B303 for bandit checks\n\nDepends-On: https://review.opendev.org/655623/\nChange-Id: Ife905cb6b5dff201da357aecf100a34e599bbe80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/075f1b71aa111b93f72f865acd3059de11946fdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/075f1b71aa111b93f72f865acd3059de11946fdc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
