)]}'
{"id":"openstack%2Fkolla~655854","triplet_id":"openstack%2Fkolla~stable%2Frocky~I0f82cc10ebfc2ea60d8e4378e54c63c770e936e7","project":"openstack/kolla","branch":"stable/rocky","topic":"bp/monasca-containers","hashtags":[],"change_id":"I0f82cc10ebfc2ea60d8e4378e54c63c770e936e7","subject":"Add missing dependency for Monasca Grafana image","status":"MERGED","created":"2019-04-26 10:01:07.000000000","updated":"2019-05-20 18:33:12.000000000","submitted":"2019-05-20 18:33:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655854-1558377192464-a5bd43bf","meta_rev_id":"15437f9f3dbc56914be18ba277f9b7defcd3d8ac","_number":655854,"virtual_id_number":655854,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-05-20 18:33:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-26 10:18:00.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-05-07 18:19:17.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-05-20 17:25:13.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-05-20 17:25:13.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-26 10:18:00.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-07 18:19:17.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-05-20 17:25:13.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-05-20 18:33:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"637abf4d67445a36ce81786cfdf8cebacbcf8aa9","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-04-26 10:01:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4770382d04ab8c26f594b408df90231fefb8d188","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-26 10:18:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/54/655854/1/openstack-check/tripleo-build-containers-fedora-28-master/3a5eb96/ : RETRY_LIMIT in 2m 24s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/54/655854/1/openstack-check/tripleo-build-containers-fedora-28-stein/a074600/ : RETRY_LIMIT in 2m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45ca9f772b522a76105a963b63213c71f1e5f6a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 12:11:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/54/655854/1/check/kolla-build-centos-binary/160bd7a/ : SUCCESS in 45m 14s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/54/655854/1/check/kolla-build-centos-source/8ed3112/ : SUCCESS in 53m 48s\n- kolla-ansible-centos-source http://logs.openstack.org/54/655854/1/check/kolla-ansible-centos-source/44d619f/ : SUCCESS in 1h 16m 22s\n- kolla-ansible-centos-binary http://logs.openstack.org/54/655854/1/check/kolla-ansible-centos-binary/fdd0e69/ : SUCCESS in 1h 09m 07s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/54/655854/1/check/kolla-build-oraclelinux-binary/39f1aa2/ : SUCCESS in 46m 53s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/54/655854/1/check/kolla-build-oraclelinux-source/188a46d/ : SUCCESS in 1h 00m 38s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/54/655854/1/check/kolla-ansible-oraclelinux-source/1c6b31f/ : SUCCESS in 1h 26m 13s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/54/655854/1/check/kolla-ansible-oraclelinux-binary/c84e9b0/ : SUCCESS in 1h 27m 37s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/54/655854/1/check/openstack-tox-lower-constraints/4dc696b/ : SUCCESS in 3m 28s\n- openstack-tox-pep8 http://logs.openstack.org/54/655854/1/check/openstack-tox-pep8/c57eadc/ : SUCCESS in 3m 42s\n- openstack-tox-py27 http://logs.openstack.org/54/655854/1/check/openstack-tox-py27/97cf9c8/ : SUCCESS in 2m 57s\n- openstack-tox-py35 http://logs.openstack.org/54/655854/1/check/openstack-tox-py35/77aafa8/ : SUCCESS in 4m 18s\n- build-openstack-sphinx-docs http://logs.openstack.org/54/655854/1/check/build-openstack-sphinx-docs/6735acc/html/ : SUCCESS in 4m 18s\n- kolla-build-ubuntu-binary http://logs.openstack.org/54/655854/1/check/kolla-build-ubuntu-binary/5e0e006/ : SUCCESS in 1h 24m 22s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/54/655854/1/check/kolla-build-ubuntu-source/3a6a243/ : SUCCESS in 35m 14s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/54/655854/1/check/kolla-ansible-ubuntu-source/bc91634/ : SUCCESS in 1h 00m 21s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/54/655854/1/check/kolla-ansible-ubuntu-binary/393574a/ : TIMED_OUT in 2h 00m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ca06c36141efe41a6b65d5590ab9660221ab8753","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-07 18:19:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7123111a6527f83723f5efb0e20e340340d50eff","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-20 17:25:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"414f010f47015e8993dd2a68bb79d1b6b6ea6e0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 17:29:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ffd12bbe674f73f275c8a19298c4a47de5a8096c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 18:33:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/54/655854/1/gate/kolla-build-centos-source/d8a653b/ : SUCCESS in 45m 46s\n- kolla-build-oraclelinux-source http://logs.openstack.org/54/655854/1/gate/kolla-build-oraclelinux-source/bbf56ed/ : SUCCESS in 50m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/54/655854/1/gate/openstack-tox-lower-constraints/eba0b6d/ : SUCCESS in 3m 11s\n- openstack-tox-pep8 http://logs.openstack.org/54/655854/1/gate/openstack-tox-pep8/dc39d9e/ : SUCCESS in 4m 36s\n- openstack-tox-py27 http://logs.openstack.org/54/655854/1/gate/openstack-tox-py27/ed4a0b6/ : SUCCESS in 4m 24s\n- openstack-tox-py35 http://logs.openstack.org/54/655854/1/gate/openstack-tox-py35/47bc6d3/ : SUCCESS in 3m 40s\n- build-openstack-sphinx-docs http://logs.openstack.org/54/655854/1/gate/build-openstack-sphinx-docs/5c787f3/html/ : SUCCESS in 3m 10s\n- kolla-build-ubuntu-source http://logs.openstack.org/54/655854/1/gate/kolla-build-ubuntu-source/9caac8b/ : SUCCESS in 37m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f53a1effa91310343b01ebba08b89b505f72a8ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 18:33:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c83d8ab11113a373c94ece2f7204bd77becb05e","revisions":{"9c83d8ab11113a373c94ece2f7204bd77becb05e":{"kind":"REWORK","_number":1,"created":"2019-04-26 10:01:07.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/54/655854/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/655854/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/655854/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/655854/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/655854/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/655854/1"}}},"commit":{"parents":[{"commit":"6af1eb8f44fc75556cf5fcaa7f5b48cc2369602e","subject":"Add Monasca Grafana image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6af1eb8f44fc75556cf5fcaa7f5b48cc2369602e"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-04-18 10:50:33.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-04-18 10:50:37.000000000","tz":0},"subject":"Add missing dependency for Monasca Grafana image","message":"Add missing dependency for Monasca Grafana image\n\nThis is split out from Ica9867448dc20864f2fd4614a295a23a4a625af4\nwhich includes an unrelated fix to Percona which is not in Rocky.\n\nCo-Authored-By: Martin André \u003cm.andre@redhat.com\u003e\nChange-Id: I0f82cc10ebfc2ea60d8e4378e54c63c770e936e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c83d8ab11113a373c94ece2f7204bd77becb05e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c83d8ab11113a373c94ece2f7204bd77becb05e"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
