)]}'
{"id":"openstack%2Fkolla~655857","triplet_id":"openstack%2Fkolla~stable%2Frocky~Id961134e424117ebefb088021faaf37fa02757d7","project":"openstack/kolla","branch":"stable/rocky","topic":"bp/monasca-containers","hashtags":[],"change_id":"Id961134e424117ebefb088021faaf37fa02757d7","subject":"Fix potential Monasca-thresh start failure","status":"MERGED","created":"2019-04-26 10:01:07.000000000","updated":"2019-05-21 11:22:04.000000000","submitted":"2019-05-21 11:22:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655857-1558437724821-8079ff73","meta_rev_id":"eef22491a57439c7749929277f0623c2c68c6c9f","_number":655857,"virtual_id_number":655857,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-05-21 11:22:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-21 09:07:23.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-04-29 13:13:42.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-05-20 17:26:24.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-05-20 17:26:24.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-29 13:13:42.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-05-20 17:26:24.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-05-21 09:07:23.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-21 11:22:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21e24d2e2f01754c59a1574a6248ec305b3b9d3d","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-04-26 10:01:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"447f1c577b3a4422a8b329ec96cb5f024b073af5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-26 10:17:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/57/655857/1/openstack-check/tripleo-build-containers-fedora-28-master/2d98710/ : RETRY_LIMIT in 2m 29s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/57/655857/1/openstack-check/tripleo-build-containers-fedora-28-stein/2c5436f/ : RETRY_LIMIT in 4m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3eb8c6458a2a5322809cade0a964e29762cb0369","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 12:14:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/57/655857/1/check/kolla-build-centos-binary/2f99565/ : SUCCESS in 39m 46s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/57/655857/1/check/kolla-build-centos-source/5309984/ : SUCCESS in 45m 47s\n- kolla-ansible-centos-source http://logs.openstack.org/57/655857/1/check/kolla-ansible-centos-source/96e732a/ : SUCCESS in 1h 06m 49s\n- kolla-ansible-centos-binary http://logs.openstack.org/57/655857/1/check/kolla-ansible-centos-binary/4d69017/ : RETRY_LIMIT in 4m 03s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/57/655857/1/check/kolla-build-oraclelinux-binary/6440d82/ : SUCCESS in 46m 10s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/57/655857/1/check/kolla-build-oraclelinux-source/746c9c8/ : SUCCESS in 44m 23s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/57/655857/1/check/kolla-ansible-oraclelinux-source/c053a5a/ : SUCCESS in 1h 10m 03s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/57/655857/1/check/kolla-ansible-oraclelinux-binary/83f953d/ : SUCCESS in 1h 04m 52s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/57/655857/1/check/openstack-tox-lower-constraints/bd737be/ : SUCCESS in 3m 39s\n- openstack-tox-pep8 http://logs.openstack.org/57/655857/1/check/openstack-tox-pep8/1105fd3/ : SUCCESS in 3m 43s\n- openstack-tox-py27 http://logs.openstack.org/57/655857/1/check/openstack-tox-py27/4eb5063/ : SUCCESS in 3m 27s\n- openstack-tox-py35 http://logs.openstack.org/57/655857/1/check/openstack-tox-py35/eb10bc8/ : SUCCESS in 3m 05s\n- build-openstack-sphinx-docs http://logs.openstack.org/57/655857/1/check/build-openstack-sphinx-docs/2c4eb66/html/ : SUCCESS in 2m 53s\n- kolla-build-ubuntu-binary http://logs.openstack.org/57/655857/1/check/kolla-build-ubuntu-binary/338c2f0/ : SUCCESS in 1h 23m 21s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/57/655857/1/check/kolla-build-ubuntu-source/f8d1aea/ : SUCCESS in 29m 58s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/57/655857/1/check/kolla-ansible-ubuntu-source/5aee3c9/ : SUCCESS in 1h 03m 04s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/57/655857/1/check/kolla-ansible-ubuntu-binary/028d22b/ : SUCCESS in 1h 55m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2be955ddb5b9a164012a280689d8781ee9d04c35","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-04-29 13:13:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6ce5aefe3942bfa896550dd2ac51d583d7e72ff5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-20 17:26:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9ce81e4c653f6c3de9f27a6e8de341b06f2c9ecd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 17:29:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"515dc4ad1499a44192c313cc4cd445cbd46deb5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 19:54:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source http://logs.openstack.org/57/655857/1/gate/kolla-build-centos-source/194191b/ : SUCCESS in 48m 53s\n- kolla-build-oraclelinux-source http://logs.openstack.org/57/655857/1/gate/kolla-build-oraclelinux-source/a19a375/ : SUCCESS in 47m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/655857/1/gate/openstack-tox-lower-constraints/18cee86/ : SUCCESS in 3m 33s\n- openstack-tox-pep8 http://logs.openstack.org/57/655857/1/gate/openstack-tox-pep8/86766c7/ : SUCCESS in 3m 18s\n- openstack-tox-py27 http://logs.openstack.org/57/655857/1/gate/openstack-tox-py27/a0aa3a2/ : SUCCESS in 3m 08s\n- openstack-tox-py35 http://logs.openstack.org/57/655857/1/gate/openstack-tox-py35/6ac4186/ : SUCCESS in 3m 32s\n- build-openstack-sphinx-docs http://logs.openstack.org/57/655857/1/gate/build-openstack-sphinx-docs/5934900/html/ : SUCCESS in 4m 13s\n- kolla-build-ubuntu-source http://logs.openstack.org/57/655857/1/gate/kolla-build-ubuntu-source/fbb5524/ : FAILURE in 2h 05m 11s","accounts_in_message":[],"_revision_number":1},{"id":"e975d0c07f8c8ccae9c35eda7d251b2a62e3361f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-21 08:20:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"22f9ffdbdb0bc071469884ffcdb27792cf872b70","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-21 09:07:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/57/655857/1/openstack-check/tripleo-build-containers-fedora-28-master/8a80ec3/ : RETRY_LIMIT in 2m 52s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/57/655857/1/openstack-check/tripleo-build-containers-fedora-28-stein/68e2333/ : SUCCESS in 32m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e07d7cabc674e0c1837629b5780d67f3dba56156","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 10:23:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/57/655857/1/check/kolla-build-centos-binary/d9d1dd8/ : SUCCESS in 39m 27s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/57/655857/1/check/kolla-build-centos-source/e61ed9e/ : SUCCESS in 52m 42s\n- kolla-ansible-centos-source http://logs.openstack.org/57/655857/1/check/kolla-ansible-centos-source/c027ce0/ : SUCCESS in 1h 28m 27s\n- kolla-ansible-centos-binary http://logs.openstack.org/57/655857/1/check/kolla-ansible-centos-binary/e2744ed/ : TIMED_OUT in 2h 01m 11s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/57/655857/1/check/kolla-build-oraclelinux-binary/26b84a4/ : SUCCESS in 40m 24s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/57/655857/1/check/kolla-build-oraclelinux-source/ddc1563/ : SUCCESS in 37m 07s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/57/655857/1/check/kolla-ansible-oraclelinux-source/454c72a/ : SUCCESS in 1h 07m 50s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/57/655857/1/check/kolla-ansible-oraclelinux-binary/0383309/ : SUCCESS in 1h 08m 20s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/57/655857/1/check/openstack-tox-lower-constraints/dc3bccc/ : SUCCESS in 3m 41s\n- openstack-tox-pep8 http://logs.openstack.org/57/655857/1/check/openstack-tox-pep8/1515ea9/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/57/655857/1/check/openstack-tox-py27/253b49a/ : SUCCESS in 3m 12s\n- openstack-tox-py35 http://logs.openstack.org/57/655857/1/check/openstack-tox-py35/ef7a4f0/ : SUCCESS in 3m 02s\n- build-openstack-sphinx-docs http://logs.openstack.org/57/655857/1/check/build-openstack-sphinx-docs/7b1c611/html/ : SUCCESS in 3m 47s\n- kolla-build-ubuntu-binary http://logs.openstack.org/57/655857/1/check/kolla-build-ubuntu-binary/ef1d226/ : SUCCESS in 1h 11m 55s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/57/655857/1/check/kolla-build-ubuntu-source/3721911/ : SUCCESS in 37m 57s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/57/655857/1/check/kolla-ansible-ubuntu-source/d4ad4ce/ : SUCCESS in 1h 00m 18s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/57/655857/1/check/kolla-ansible-ubuntu-binary/09cb2d6/ : SUCCESS in 1h 44m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3367965aed4ec532554b3bffe5e3a6fad0c70e09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 10:23:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ab6c7dcac591c52483772eb91fe14656df9ff3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 11:22:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f984b688f6cda7cf865b3c51b251e6707ee142ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 11:22:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/57/655857/1/gate/kolla-build-centos-source/388bb9e/ : SUCCESS in 52m 11s\n- kolla-build-oraclelinux-source http://logs.openstack.org/57/655857/1/gate/kolla-build-oraclelinux-source/1a2be9c/ : SUCCESS in 37m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/655857/1/gate/openstack-tox-lower-constraints/929a6ef/ : SUCCESS in 3m 22s\n- openstack-tox-pep8 http://logs.openstack.org/57/655857/1/gate/openstack-tox-pep8/49879e5/ : SUCCESS in 4m 39s\n- openstack-tox-py27 http://logs.openstack.org/57/655857/1/gate/openstack-tox-py27/8869735/ : SUCCESS in 3m 24s\n- openstack-tox-py35 http://logs.openstack.org/57/655857/1/gate/openstack-tox-py35/667dbf8/ : SUCCESS in 3m 16s\n- build-openstack-sphinx-docs http://logs.openstack.org/57/655857/1/gate/build-openstack-sphinx-docs/691b4f4/html/ : SUCCESS in 3m 23s\n- kolla-build-ubuntu-source http://logs.openstack.org/57/655857/1/gate/kolla-build-ubuntu-source/b42f85e/ : SUCCESS in 34m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a300a83966a7459c0205f0a130e7d9ca2f9fac3e","revisions":{"a300a83966a7459c0205f0a130e7d9ca2f9fac3e":{"kind":"REWORK","_number":1,"created":"2019-04-26 10:01:07.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/57/655857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/655857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/655857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/655857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/655857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/655857/1"}}},"commit":{"parents":[{"commit":"85bf3dcbec87eeffd3344a5d5a71dd2db1539ce9","subject":"Fix naming of override template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85bf3dcbec87eeffd3344a5d5a71dd2db1539ce9"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2018-10-01 16:09:41.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-04-26 09:59:12.000000000","tz":0},"subject":"Fix potential Monasca-thresh start failure","message":"Fix potential Monasca-thresh start failure\n\nMonasca-thresh inherits from the Storm container, but we want\nit to write logs to the Monasca logs directory. In this commit\nwe overwrite the script which configures the Storm log directory\nwith the Monasca script.\n\nChange-Id: Id961134e424117ebefb088021faaf37fa02757d7\nPartially-Implements: blueprint monasca-containers\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a300a83966a7459c0205f0a130e7d9ca2f9fac3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a300a83966a7459c0205f0a130e7d9ca2f9fac3e"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
