)]}'
{"id":"openstack%2Fkolla~655859","triplet_id":"openstack%2Fkolla~stable%2Frocky~I89084ac0076eff099e567b32de249df9f0115e82","project":"openstack/kolla","branch":"stable/rocky","topic":"bp/monasca-containers","hashtags":[],"change_id":"I89084ac0076eff099e567b32de249df9f0115e82","subject":"Use alembic migrations to deploy Monasca DB","status":"MERGED","created":"2019-04-26 10:01:07.000000000","updated":"2019-05-21 18:45:21.000000000","submitted":"2019-05-21 18:45:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655859-1558464321430-a79c9f46","meta_rev_id":"a2b984baa918abc337f27b6e029e23e9eb188031","_number":655859,"virtual_id_number":655859,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-05-21 18:45:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-21 16:33:27.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-04-29 13:14:49.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-05-20 17:28:28.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-05-20 17:28:28.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-29 13:14:49.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-05-20 17:28:28.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-05-21 16:33:27.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-21 18:45:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"74eee83855b4a9140afb65c72cfb552afac4b6e3","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-04-26 10:01:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df80fbc79f5cc33a9c908a1f481e1a813b06b387","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-26 11:03:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/59/655859/1/openstack-check/tripleo-build-containers-fedora-28-master/aad4cc4/ : RETRY_LIMIT in 2m 25s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/59/655859/1/openstack-check/tripleo-build-containers-fedora-28-stein/4b8033e/ : RETRY_LIMIT in 2m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a9412446bbeca31c81c7e05f26b7ecfa2537ee3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 12:10:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/59/655859/1/check/kolla-build-centos-binary/6b19210/ : SUCCESS in 39m 14s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/59/655859/1/check/kolla-build-centos-source/df2be58/ : SUCCESS in 42m 50s\n- kolla-ansible-centos-source http://logs.openstack.org/59/655859/1/check/kolla-ansible-centos-source/e289c0d/ : SUCCESS in 1h 08m 48s\n- kolla-ansible-centos-binary http://logs.openstack.org/59/655859/1/check/kolla-ansible-centos-binary/a03d34b/ : SUCCESS in 1h 13m 31s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/59/655859/1/check/kolla-build-oraclelinux-binary/4231b0a/ : SUCCESS in 42m 07s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/59/655859/1/check/kolla-build-oraclelinux-source/2211c53/ : SUCCESS in 48m 24s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/59/655859/1/check/kolla-ansible-oraclelinux-source/a54d21f/ : SUCCESS in 1h 13m 52s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/59/655859/1/check/kolla-ansible-oraclelinux-binary/797e8cf/ : SUCCESS in 1h 09m 25s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/655859/1/check/openstack-tox-lower-constraints/d28c186/ : SUCCESS in 2m 58s\n- openstack-tox-pep8 http://logs.openstack.org/59/655859/1/check/openstack-tox-pep8/af51122/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/59/655859/1/check/openstack-tox-py27/a12e3db/ : SUCCESS in 3m 09s\n- openstack-tox-py35 http://logs.openstack.org/59/655859/1/check/openstack-tox-py35/e0d1c7f/ : SUCCESS in 2m 57s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/655859/1/check/build-openstack-sphinx-docs/398914d/html/ : SUCCESS in 2m 53s\n- kolla-build-ubuntu-binary http://logs.openstack.org/59/655859/1/check/kolla-build-ubuntu-binary/7763667/ : SUCCESS in 1h 23m 19s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/59/655859/1/check/kolla-build-ubuntu-source/b6e8c67/ : SUCCESS in 29m 17s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/59/655859/1/check/kolla-ansible-ubuntu-source/8f84b2f/ : SUCCESS in 1h 01m 44s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/59/655859/1/check/kolla-ansible-ubuntu-binary/c4bd647/ : SUCCESS in 1h 51m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"034d1a864ad15547a1bda000ee0408cbdf304f25","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-04-29 13:14:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"80a64511841f6576e96fb18a0089073834d8f377","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-20 17:28:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"15f69da7e07d04b51ede98936c0cb2d4df4a29e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 17:29:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c0ff3eab05282ac0f065ec39cb1bc9e70429f7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 19:54:26.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"e9819e4c6e8c899dd3333e2db0494677fe140bae","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-21 08:21:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9fd3e59268f7baaee48cafef093685db4485bf7e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-21 08:58:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/59/655859/1/openstack-check/tripleo-build-containers-fedora-28-master/17438c6/ : RETRY_LIMIT in 3m 07s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/59/655859/1/openstack-check/tripleo-build-containers-fedora-28-stein/38a0a88/ : SUCCESS in 30m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58c36ee4f520617e861f1ca612a60d139c671070","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 10:25:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/59/655859/1/check/kolla-build-centos-binary/95f8fff/ : SUCCESS in 34m 01s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/59/655859/1/check/kolla-build-centos-source/380cb49/ : SUCCESS in 58m 10s\n- kolla-ansible-centos-source http://logs.openstack.org/59/655859/1/check/kolla-ansible-centos-source/430a791/ : SUCCESS in 1h 00m 29s\n- kolla-ansible-centos-binary http://logs.openstack.org/59/655859/1/check/kolla-ansible-centos-binary/e8b9663/ : SUCCESS in 1h 03m 18s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/59/655859/1/check/kolla-build-oraclelinux-binary/758f106/ : SUCCESS in 51m 40s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/59/655859/1/check/kolla-build-oraclelinux-source/d1ba858/ : SUCCESS in 39m 28s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/59/655859/1/check/kolla-ansible-oraclelinux-source/8a23f1e/ : SUCCESS in 1h 13m 16s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/59/655859/1/check/kolla-ansible-oraclelinux-binary/d352f27/ : SUCCESS in 1h 02m 44s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/655859/1/check/openstack-tox-lower-constraints/466665e/ : SUCCESS in 3m 44s\n- openstack-tox-pep8 http://logs.openstack.org/59/655859/1/check/openstack-tox-pep8/3e29910/ : SUCCESS in 3m 54s\n- openstack-tox-py27 http://logs.openstack.org/59/655859/1/check/openstack-tox-py27/ea634b2/ : SUCCESS in 3m 07s\n- openstack-tox-py35 http://logs.openstack.org/59/655859/1/check/openstack-tox-py35/d680359/ : SUCCESS in 3m 15s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/655859/1/check/build-openstack-sphinx-docs/efea0de/html/ : SUCCESS in 3m 26s\n- kolla-build-ubuntu-binary http://logs.openstack.org/59/655859/1/check/kolla-build-ubuntu-binary/d3d9a1c/ : SUCCESS in 1h 14m 32s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/59/655859/1/check/kolla-build-ubuntu-source/c0ff25a/ : SUCCESS in 32m 38s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/59/655859/1/check/kolla-ansible-ubuntu-source/c6ea20e/ : SUCCESS in 1h 04m 05s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/59/655859/1/check/kolla-ansible-ubuntu-binary/1da51a2/ : SUCCESS in 1h 58m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2533d61029e482a50a18c550535befbfae33d59d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 10:26:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"79b98cfb9d41df58db065b27c6670ea1b110ed92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 11:22:06.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"f076a3fe92155190a98eed0c2b7199c0581ccbad","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-21 13:00:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a1d92a984ebbe537e869bea19b43f12aee22406c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-21 13:42:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/59/655859/1/openstack-check/tripleo-build-containers-fedora-28-master/b2dbe47/ : RETRY_LIMIT in 5m 15s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/59/655859/1/openstack-check/tripleo-build-containers-fedora-28-stein/c1b7ba1/ : SUCCESS in 38m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fb26b6dd3c6af2d2a75b9ece94d46f7e3e0625e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 15:13:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/59/655859/1/check/kolla-build-centos-binary/76478f1/ : SUCCESS in 28m 12s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/59/655859/1/check/kolla-build-centos-source/4ceffc0/ : SUCCESS in 36m 09s\n- kolla-ansible-centos-source http://logs.openstack.org/59/655859/1/check/kolla-ansible-centos-source/7fade93/ : TIMED_OUT in 2h 01m 20s\n- kolla-ansible-centos-binary http://logs.openstack.org/59/655859/1/check/kolla-ansible-centos-binary/7f9b411/ : SUCCESS in 1h 11m 38s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/59/655859/1/check/kolla-build-oraclelinux-binary/ae76f99/ : SUCCESS in 48m 10s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/59/655859/1/check/kolla-build-oraclelinux-source/6da0067/ : SUCCESS in 47m 07s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/59/655859/1/check/kolla-ansible-oraclelinux-source/0395490/ : SUCCESS in 1h 14m 49s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/59/655859/1/check/kolla-ansible-oraclelinux-binary/06baf47/ : SUCCESS in 1h 13m 02s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/655859/1/check/openstack-tox-lower-constraints/958cb5f/ : SUCCESS in 4m 59s\n- openstack-tox-pep8 http://logs.openstack.org/59/655859/1/check/openstack-tox-pep8/c4f9b43/ : SUCCESS in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/59/655859/1/check/openstack-tox-py27/35d160a/ : SUCCESS in 3m 01s\n- openstack-tox-py35 http://logs.openstack.org/59/655859/1/check/openstack-tox-py35/d8b5f35/ : SUCCESS in 4m 36s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/655859/1/check/build-openstack-sphinx-docs/676508a/html/ : SUCCESS in 3m 44s\n- kolla-build-ubuntu-binary http://logs.openstack.org/59/655859/1/check/kolla-build-ubuntu-binary/56be7d8/ : SUCCESS in 1h 21m 21s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/59/655859/1/check/kolla-build-ubuntu-source/ac8570a/ : SUCCESS in 29m 45s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/59/655859/1/check/kolla-ansible-ubuntu-source/5ebe987/ : SUCCESS in 1h 05m 07s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/59/655859/1/check/kolla-ansible-ubuntu-binary/689e33b/ : SUCCESS in 1h 32m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3379af2b176d8745a5ce84458cfea967cec7bbcd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-21 15:46:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1c3407c0f1183d59d0536287cb00e873b94701d5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-21 16:33:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/59/655859/1/openstack-check/tripleo-build-containers-fedora-28-master/c93fdc6/ : RETRY_LIMIT in 3m 56s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/59/655859/1/openstack-check/tripleo-build-containers-fedora-28-stein/0e20e43/ : SUCCESS in 32m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2f1d2c06002418355e8da94736cd8876b776ec8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 17:42:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/59/655859/1/check/kolla-build-centos-binary/d0a2d44/ : SUCCESS in 36m 12s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/59/655859/1/check/kolla-build-centos-source/632c1c1/ : SUCCESS in 58m 06s\n- kolla-ansible-centos-source http://logs.openstack.org/59/655859/1/check/kolla-ansible-centos-source/f03d7b9/ : SUCCESS in 1h 12m 10s\n- kolla-ansible-centos-binary http://logs.openstack.org/59/655859/1/check/kolla-ansible-centos-binary/ebb8692/ : SUCCESS in 1h 34m 13s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/59/655859/1/check/kolla-build-oraclelinux-binary/6b72fa8/ : SUCCESS in 47m 04s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/59/655859/1/check/kolla-build-oraclelinux-source/5e04514/ : SUCCESS in 41m 48s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/59/655859/1/check/kolla-ansible-oraclelinux-source/fbdc9a1/ : SUCCESS in 1h 05m 45s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/59/655859/1/check/kolla-ansible-oraclelinux-binary/46f26d0/ : SUCCESS in 1h 09m 45s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/655859/1/check/openstack-tox-lower-constraints/b245f47/ : SUCCESS in 4m 53s\n- openstack-tox-pep8 http://logs.openstack.org/59/655859/1/check/openstack-tox-pep8/a7974a0/ : SUCCESS in 5m 06s\n- openstack-tox-py27 http://logs.openstack.org/59/655859/1/check/openstack-tox-py27/6f68c0a/ : SUCCESS in 3m 34s\n- openstack-tox-py35 http://logs.openstack.org/59/655859/1/check/openstack-tox-py35/072db27/ : SUCCESS in 4m 20s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/655859/1/check/build-openstack-sphinx-docs/2ccb6f4/html/ : SUCCESS in 3m 32s\n- kolla-build-ubuntu-binary http://logs.openstack.org/59/655859/1/check/kolla-build-ubuntu-binary/d15c4bd/ : SUCCESS in 1h 00m 30s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/59/655859/1/check/kolla-build-ubuntu-source/560ce3f/ : SUCCESS in 35m 06s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/59/655859/1/check/kolla-ansible-ubuntu-source/8624372/ : SUCCESS in 1h 02m 14s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/59/655859/1/check/kolla-ansible-ubuntu-binary/57235cd/ : SUCCESS in 1h 28m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b4672f983205567e337de349c41862628788908d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 17:42:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a117da7088ebf118ecb36207169605aee1b9c6f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 18:45:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/59/655859/1/gate/kolla-build-centos-source/64bc4a8/ : SUCCESS in 41m 38s\n- kolla-build-oraclelinux-source http://logs.openstack.org/59/655859/1/gate/kolla-build-oraclelinux-source/bffa015/ : SUCCESS in 46m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/655859/1/gate/openstack-tox-lower-constraints/6cbf724/ : SUCCESS in 3m 36s\n- openstack-tox-pep8 http://logs.openstack.org/59/655859/1/gate/openstack-tox-pep8/b318ebd/ : SUCCESS in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/59/655859/1/gate/openstack-tox-py27/ae72940/ : SUCCESS in 3m 42s\n- openstack-tox-py35 http://logs.openstack.org/59/655859/1/gate/openstack-tox-py35/a374547/ : SUCCESS in 3m 36s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/655859/1/gate/build-openstack-sphinx-docs/ab1a74e/html/ : SUCCESS in 3m 28s\n- kolla-build-ubuntu-source http://logs.openstack.org/59/655859/1/gate/kolla-build-ubuntu-source/17b4070/ : SUCCESS in 56m 40s","accounts_in_message":[],"_revision_number":1},{"id":"beac73b39d20d35329020359a841a3367e29eda4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 18:45:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"521b802ba4f882413d05f7641ad8f146eb47c523","revisions":{"521b802ba4f882413d05f7641ad8f146eb47c523":{"kind":"REWORK","_number":1,"created":"2019-04-26 10:01:07.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/59/655859/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/655859/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/655859/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/655859/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/655859/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/655859/1"}}},"commit":{"parents":[{"commit":"aa445283e7dad667726ef572fec2cfa994059603","subject":"Add monasca user to Monasca thresh container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa445283e7dad667726ef572fec2cfa994059603"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2018-09-19 16:34:04.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-04-26 09:59:43.000000000","tz":0},"subject":"Use alembic migrations to deploy Monasca DB","message":"Use alembic migrations to deploy Monasca DB\n\nSupport has recently been added for managing the Monasca DB\nwith alembic. This change takes advantage of that.\n\nPartially-Implements: blueprint monasca-containers\nChange-Id: I89084ac0076eff099e567b32de249df9f0115e82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/521b802ba4f882413d05f7641ad8f146eb47c523"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/521b802ba4f882413d05f7641ad8f146eb47c523"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
