)]}'
{"id":"openstack%2Fkolla~656502","triplet_id":"openstack%2Fkolla~master~I209c3c60d5c2edb6856e81eb374fbcad648e3904","project":"openstack/kolla","branch":"master","topic":"bug/1827025","hashtags":[],"change_id":"I209c3c60d5c2edb6856e81eb374fbcad648e3904","subject":"Add template override for grafana plugins","status":"MERGED","created":"2019-04-30 11:58:19.000000000","updated":"2019-05-07 19:28:48.000000000","submitted":"2019-05-07 19:28:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656502-1557257328040-7de1955d","meta_rev_id":"bdf0d97c794a8ee330543ae837560ce8f4ba0f8a","_number":656502,"virtual_id_number":656502,"owner":{"_account_id":30354,"name":"Arnaud PARENT","email":"arnaud@parent.mx","username":"pandarouxbsd"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"date":"2019-04-30 11:58:31.000000000","_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-05-07 19:28:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-30 12:18:25.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"recommended":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"all":[{"value":1,"date":"2019-04-30 12:11:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"value":1,"date":"2019-05-06 10:43:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":1,"date":"2019-04-30 12:06:49.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2019-05-03 03:19:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2019-05-07 18:23:32.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-05-07 16:23:31.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":1,"date":"2019-05-07 18:23:32.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-30 11:58:31.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2019-04-30 12:06:49.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-04-30 12:11:45.000000000","updated_by":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"reviewer":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"state":"REVIEWER"},{"updated":"2019-04-30 12:18:25.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-03 03:19:06.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-05-06 10:43:28.000000000","updated_by":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"reviewer":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"state":"REVIEWER"},{"updated":"2019-05-07 16:23:31.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-05-07 18:23:32.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-05-07 19:28:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"55c20d0027608e6bacb848ef8adb398207d2f949","author":{"_account_id":30354,"name":"Arnaud PARENT","email":"arnaud@parent.mx","username":"pandarouxbsd"},"date":"2019-04-30 11:58:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2372598d4595939996e99b2f33837b029592cdd5","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2019-04-30 11:58:31.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"43dcd8f44e1426bcc03209cb55cf4ea16e7b9150","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-04-30 12:06:49.000000000","message":"Patch Set 1: Code-Review+1\n\nSeems reasonable to me.","accounts_in_message":[],"_revision_number":1},{"id":"be22c4def6b734a95502da509f4a011ec2b2bbdd","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2019-04-30 12:11:45.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks !","accounts_in_message":[],"_revision_number":1},{"id":"04e9f1dc0cac50a6340a775d62769205e5a12b6d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-30 12:18:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/02/656502/1/openstack-check/tripleo-build-containers-fedora-28-master/b48f2f2/ : RETRY_LIMIT in 2m 22s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/02/656502/1/openstack-check/tripleo-build-containers-fedora-28-stein/4a910ed/ : RETRY_LIMIT in 2m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ebfa87b9e5de57daf15c33b4221a938a13729d7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 14:16:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/02/656502/1/check/kolla-build-centos-binary/f6ea94b/ : SUCCESS in 31m 05s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/02/656502/1/check/kolla-build-centos-source/5060a8b/ : SUCCESS in 47m 35s\n- kolla-ansible-centos-source http://logs.openstack.org/02/656502/1/check/kolla-ansible-centos-source/f8ab3eb/ : SUCCESS in 1h 08m 20s\n- kolla-ansible-centos-binary http://logs.openstack.org/02/656502/1/check/kolla-ansible-centos-binary/55d0bd3/ : SUCCESS in 1h 04m 16s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/02/656502/1/check/kolla-ansible-centos-source-upgrade/f67561e/ : SUCCESS in 1h 15m 02s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/02/656502/1/check/tripleo-build-containers-centos-7/06ff9d8/ : SUCCESS in 42m 11s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/02/656502/1/check/kolla-build-oraclelinux-binary/f87708c/ : SUCCESS in 37m 38s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/02/656502/1/check/kolla-build-oraclelinux-source/67ed95e/ : SUCCESS in 53m 26s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/02/656502/1/check/kolla-ansible-oraclelinux-source/8fb144b/ : TIMED_OUT in 2h 01m 20s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/02/656502/1/check/kolla-ansible-oraclelinux-binary/54c9375/ : SUCCESS in 1h 17m 41s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/02/656502/1/check/openstack-tox-lower-constraints/8b3f98e/ : SUCCESS in 3m 56s\n- openstack-tox-pep8 http://logs.openstack.org/02/656502/1/check/openstack-tox-pep8/6cdc958/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/02/656502/1/check/openstack-tox-py27/3cbd9dd/ : SUCCESS in 3m 24s\n- openstack-tox-py36 http://logs.openstack.org/02/656502/1/check/openstack-tox-py36/685460b/ : SUCCESS in 3m 20s\n- openstack-tox-py37 http://logs.openstack.org/02/656502/1/check/openstack-tox-py37/2ac7096/ : SUCCESS in 3m 33s\n- openstack-tox-docs http://logs.openstack.org/02/656502/1/check/openstack-tox-docs/dcad32c/html/ : SUCCESS in 4m 35s\n- kolla-build-ubuntu-binary http://logs.openstack.org/02/656502/1/check/kolla-build-ubuntu-binary/bc66146/ : SUCCESS in 41m 02s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/02/656502/1/check/kolla-build-ubuntu-source/ab95883/ : SUCCESS in 31m 47s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/02/656502/1/check/kolla-ansible-ubuntu-source/f1b41d2/ : SUCCESS in 51m 54s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/02/656502/1/check/kolla-ansible-ubuntu-binary/2787b3c/ : SUCCESS in 43m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/02/656502/1/check/kolla-ansible-ubuntu-source-upgrade/b12a1e5/ : SUCCESS in 59m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c31f967ba8a10d72a670ce92e6ed44f1b2eec849","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-05-03 03:19:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a8baac78480084aadb73e9fcfd21007a7970b0eb","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-05-06 10:43:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cdf78063fc9f9581793565e77de6eefcb6a77f5d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-07 16:23:31.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"fd57b4496b9a30ee8bac5aea6c2fcc5b60af3da2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-07 18:23:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d8e13c981c10710c9597645ba7bd30cd4e01421b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 18:23:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3b1b1608f1c9497d4735e8e00c0032b7e1351e56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 19:28:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/02/656502/1/gate/kolla-build-centos-source/a38f515/ : SUCCESS in 48m 51s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/02/656502/1/gate/tripleo-build-containers-centos-7/6097668/ : SUCCESS in 45m 26s\n- kolla-build-oraclelinux-source http://logs.openstack.org/02/656502/1/gate/kolla-build-oraclelinux-source/2cd90ea/ : SUCCESS in 56m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/656502/1/gate/openstack-tox-lower-constraints/56411a9/ : SUCCESS in 3m 55s\n- openstack-tox-pep8 http://logs.openstack.org/02/656502/1/gate/openstack-tox-pep8/e336213/ : SUCCESS in 3m 56s\n- openstack-tox-py27 http://logs.openstack.org/02/656502/1/gate/openstack-tox-py27/c1920e1/ : SUCCESS in 3m 00s\n- openstack-tox-py36 http://logs.openstack.org/02/656502/1/gate/openstack-tox-py36/43a4a58/ : SUCCESS in 3m 08s\n- openstack-tox-py37 http://logs.openstack.org/02/656502/1/gate/openstack-tox-py37/ce94cb7/ : SUCCESS in 3m 35s\n- openstack-tox-docs http://logs.openstack.org/02/656502/1/gate/openstack-tox-docs/0fca66b/html/ : SUCCESS in 2m 58s\n- kolla-build-ubuntu-source http://logs.openstack.org/02/656502/1/gate/kolla-build-ubuntu-source/b51eb15/ : SUCCESS in 1h 00m 58s","accounts_in_message":[],"_revision_number":1},{"id":"1347942abcf96697f3ab7dcbe2bde3f894fcf4f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 19:28:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a873677392f1736a73726d3f25a01c2f7cb1ade9","revisions":{"a873677392f1736a73726d3f25a01c2f7cb1ade9":{"kind":"REWORK","_number":1,"created":"2019-04-30 11:58:19.000000000","uploader":{"_account_id":30354,"name":"Arnaud PARENT","email":"arnaud@parent.mx","username":"pandarouxbsd"},"ref":"refs/changes/02/656502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/656502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/656502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/656502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/656502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/656502/1"}}},"commit":{"parents":[{"commit":"9e9e31ed2c671d7547354708bb8decc05c737ecd","subject":"Merge \"Reuse /var/lib/tempest symlink for rpm also\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e9e31ed2c671d7547354708bb8decc05c737ecd"}]}],"author":{"name":"Arnaud PARENT","email":"arnaud@parent.mx","date":"2019-04-30 11:31:50.000000000","tz":0},"committer":{"name":"Arnaud PARENT","email":"arnaud@parent.mx","date":"2019-04-30 11:57:53.000000000","tz":0},"subject":"Add template override for grafana plugins","message":"Add template override for grafana plugins\n\nChange-Id: I209c3c60d5c2edb6856e81eb374fbcad648e3904\nCloses-Bug: #1827025\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a873677392f1736a73726d3f25a01c2f7cb1ade9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a873677392f1736a73726d3f25a01c2f7cb1ade9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
