)]}'
{"id":"openstack%2Fkolla~657314","triplet_id":"openstack%2Fkolla~stable%2Fstein~I402f58ce53fdce426582cf2fc31349da6d2f5ae0","project":"openstack/kolla","branch":"stable/stein","topic":"tempest-stable/stein","hashtags":[],"change_id":"I402f58ce53fdce426582cf2fc31349da6d2f5ae0","subject":"Reuse /var/lib/tempest symlink for rpm also","status":"MERGED","created":"2019-05-06 05:26:28.000000000","updated":"2019-05-06 10:39:57.000000000","submitted":"2019-05-06 10:39:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657314-1557139197139-5260f087","meta_rev_id":"462252245eb6b035a79244345da08a57b5dc5484","_number":657314,"virtual_id_number":657314,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-05-06 10:39:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-06 07:23:44.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2019-05-06 05:28:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2019-05-06 09:40:24.000000000","_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":2,"date":"2019-05-06 09:51:32.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":1,"date":"2019-05-06 09:51:32.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2019-05-06 05:27:20.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-06 05:27:20.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-05-06 05:28:07.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-05-06 07:23:44.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-06 09:40:24.000000000","updated_by":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"reviewer":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"state":"REVIEWER"},{"updated":"2019-05-06 09:51:32.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-05-06 10:39:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7761e20ec60d5ef79d2eb9a6cdfef4139d8e47ec","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-05-06 05:26:28.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"653f715efaa216a2e33b15774a6202c7dd04c54d","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-05-06 05:28:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8acbbe15780e47fb289f0c89e5ce01058b221faf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-06 05:40:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/14/657314/1/openstack-check/tripleo-build-containers-fedora-28-master/ae38eb0/ : RETRY_LIMIT in 2m 27s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/14/657314/1/openstack-check/tripleo-build-containers-fedora-28-stein/1aabefb/ : RETRY_LIMIT in 2m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d16e22b3ac959221a69ffddd06b2fe7bc688e4a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 06:50:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/14/657314/1/check/kolla-build-centos-binary/bb9f201/ : SUCCESS in 39m 36s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/14/657314/1/check/kolla-build-centos-source/c5074c2/ : SUCCESS in 46m 31s\n- kolla-ansible-centos-source http://logs.openstack.org/14/657314/1/check/kolla-ansible-centos-source/c3a6b16/ : FAILURE in 17m 47s\n- kolla-ansible-centos-binary http://logs.openstack.org/14/657314/1/check/kolla-ansible-centos-binary/f6a5cdd/ : SUCCESS in 1h 06m 38s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/14/657314/1/check/kolla-ansible-centos-source-upgrade/c29ee2f/ : SUCCESS in 1h 11m 25s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/14/657314/1/check/tripleo-build-containers-centos-7/4245df9/ : SUCCESS in 42m 43s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/14/657314/1/check/kolla-build-oraclelinux-binary/af4e179/ : SUCCESS in 58m 47s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/14/657314/1/check/kolla-build-oraclelinux-source/091e074/ : SUCCESS in 39m 31s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/14/657314/1/check/kolla-ansible-oraclelinux-source/76c59db/ : SUCCESS in 1h 11m 14s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/14/657314/1/check/kolla-ansible-oraclelinux-binary/e488fe5/ : SUCCESS in 1h 10m 09s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/14/657314/1/check/openstack-tox-lower-constraints/6a69cdf/ : SUCCESS in 3m 13s\n- openstack-tox-pep8 http://logs.openstack.org/14/657314/1/check/openstack-tox-pep8/39f384d/ : SUCCESS in 3m 53s\n- openstack-tox-py27 http://logs.openstack.org/14/657314/1/check/openstack-tox-py27/a74afaa/ : SUCCESS in 3m 42s\n- openstack-tox-py36 http://logs.openstack.org/14/657314/1/check/openstack-tox-py36/387de72/ : SUCCESS in 2m 50s\n- openstack-tox-py37 http://logs.openstack.org/14/657314/1/check/openstack-tox-py37/78a3c8c/ : SUCCESS in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/14/657314/1/check/openstack-tox-docs/59c01a0/html/ : SUCCESS in 3m 38s\n- kolla-build-ubuntu-binary http://logs.openstack.org/14/657314/1/check/kolla-build-ubuntu-binary/2793203/ : SUCCESS in 36m 13s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/14/657314/1/check/kolla-build-ubuntu-source/10b0d47/ : SUCCESS in 29m 40s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/14/657314/1/check/kolla-ansible-ubuntu-source/9a8bf1d/ : SUCCESS in 50m 37s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/14/657314/1/check/kolla-ansible-ubuntu-binary/3bf8a63/ : SUCCESS in 52m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/14/657314/1/check/kolla-ansible-ubuntu-source-upgrade/9df2193/ : SUCCESS in 1h 03m 06s","accounts_in_message":[],"_revision_number":1},{"id":"083c75c7d28f21c0f5da20e679f977b8e4c8db25","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-05-06 07:07:01.000000000","message":"Patch Set 1:\n\n\u003e Build failed (check pipeline).  For information on how to proceed,\n \u003e see\n \u003e http://docs.openstack.org/infra/manual/developers.html#automated-testing\n \u003e \n \u003e \n \u003e - kolla-build-centos-binary http://logs.openstack.org/14/657314/1/check/kolla-build-centos-binary/bb9f201/\n \u003e : SUCCESS in 39m 36s (non-voting)\n \u003e - kolla-build-centos-source http://logs.openstack.org/14/657314/1/check/kolla-build-centos-source/c5074c2/\n \u003e : SUCCESS in 46m 31s\n \u003e - kolla-ansible-centos-source http://logs.openstack.org/14/657314/1/check/kolla-ansible-centos-source/c3a6b16/\n \u003e : FAILURE in 17m 47s\nseems some network issue:-\nINFO:kolla.common.utils.base:\u001b[91mcurl: (7) Failed connect to www.percona.com:443; Connection timed out\nINFO:kolla.common.utils.base:\u001b[0m\nINFO:kolla.common.utils.base:\u001b[91merror: https://www.percona.com/downloads/RPM-GPG-KEY-percona: import read failed(2).\nINFO:kolla.common.utils.base:\u001b[0m\nINFO:kolla.common.utils.base:Removing intermediate container 35e87b6a363a\nERROR:kolla.common.utils.base:Error\u0027d with the following message\nERROR:kolla.common.utils.base:The command \u0027/bin/sh -c rpm --import https://artifacts.elastic.co/GPG-KEY-elasticsearch     \u0026\u0026 rpm --import https://packages.grafana.com/gpg.key     \u0026\u0026 rpm --import https://repos.influxdata.com/influxdb.key     \u0026\u0026 rpm --import https://packagecloud.io/gpg.key     \u0026\u0026 rpm --import https://www.percona.com/downloads/RPM-GPG-KEY-percona     \u0026\u0026 rpm --import /etc/pki/rpm-gpg/PERCONA-PACKAGING-KEY     \u0026\u0026 rpm --import https://packages.treasuredata.com/GPG-KEY-td-agent\u0027 returned a non-zero code: 1\n \u003e - kolla-ansible-centos-binary http://logs.openstack.org/14/657314/1/check/kolla-ansible-centos-binary/f6a5cdd/\n \u003e : SUCCESS in 1h 06m 38s (non-voting)\n \u003e - kolla-ansible-centos-source-upgrade http://logs.openstack.org/14/657314/1/check/kolla-ansible-centos-source-upgrade/c29ee2f/\n \u003e : SUCCESS in 1h 11m 25s\n \u003e - tripleo-build-containers-centos-7 http://logs.openstack.org/14/657314/1/check/tripleo-build-containers-centos-7/4245df9/\n \u003e : SUCCESS in 42m 43s\n \u003e - kolla-build-oraclelinux-binary http://logs.openstack.org/14/657314/1/check/kolla-build-oraclelinux-binary/af4e179/\n \u003e : SUCCESS in 58m 47s (non-voting)\n \u003e - kolla-build-oraclelinux-source http://logs.openstack.org/14/657314/1/check/kolla-build-oraclelinux-source/091e074/\n \u003e : SUCCESS in 39m 31s\n \u003e - kolla-ansible-oraclelinux-source http://logs.openstack.org/14/657314/1/check/kolla-ansible-oraclelinux-source/76c59db/\n \u003e : SUCCESS in 1h 11m 14s (non-voting)\n \u003e - kolla-ansible-oraclelinux-binary http://logs.openstack.org/14/657314/1/check/kolla-ansible-oraclelinux-binary/e488fe5/\n \u003e : SUCCESS in 1h 10m 09s (non-voting)\n \u003e - openstack-tox-lower-constraints http://logs.openstack.org/14/657314/1/check/openstack-tox-lower-constraints/6a69cdf/\n \u003e : SUCCESS in 3m 13s\n \u003e - openstack-tox-pep8 http://logs.openstack.org/14/657314/1/check/openstack-tox-pep8/39f384d/\n \u003e : SUCCESS in 3m 53s\n \u003e - openstack-tox-py27 http://logs.openstack.org/14/657314/1/check/openstack-tox-py27/a74afaa/\n \u003e : SUCCESS in 3m 42s\n \u003e - openstack-tox-py36 http://logs.openstack.org/14/657314/1/check/openstack-tox-py36/387de72/\n \u003e : SUCCESS in 2m 50s\n \u003e - openstack-tox-py37 http://logs.openstack.org/14/657314/1/check/openstack-tox-py37/78a3c8c/\n \u003e : SUCCESS in 4m 30s\n \u003e - openstack-tox-docs http://logs.openstack.org/14/657314/1/check/openstack-tox-docs/59c01a0/html/\n \u003e : SUCCESS in 3m 38s\n \u003e - kolla-build-ubuntu-binary http://logs.openstack.org/14/657314/1/check/kolla-build-ubuntu-binary/2793203/\n \u003e : SUCCESS in 36m 13s (non-voting)\n \u003e - kolla-build-ubuntu-source http://logs.openstack.org/14/657314/1/check/kolla-build-ubuntu-source/10b0d47/\n \u003e : SUCCESS in 29m 40s\n \u003e - kolla-ansible-ubuntu-source http://logs.openstack.org/14/657314/1/check/kolla-ansible-ubuntu-source/9a8bf1d/\n \u003e : SUCCESS in 50m 37s\n \u003e - kolla-ansible-ubuntu-binary http://logs.openstack.org/14/657314/1/check/kolla-ansible-ubuntu-binary/3bf8a63/\n \u003e : SUCCESS in 52m 38s (non-voting)\n \u003e - kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/14/657314/1/check/kolla-ansible-ubuntu-source-upgrade/9df2193/\n \u003e : SUCCESS in 1h 03m 06s","accounts_in_message":[],"_revision_number":1},{"id":"46e76c12d61de599acf0cd58377186276d6bb9ce","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-05-06 07:07:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c0ce521f7fddd904c10b387703c17c06ef9c0e0f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-06 07:23:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/14/657314/1/openstack-check/tripleo-build-containers-fedora-28-master/2abb72b/ : RETRY_LIMIT in 2m 32s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/14/657314/1/openstack-check/tripleo-build-containers-fedora-28-stein/e5b7848/ : RETRY_LIMIT in 2m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7625f5698033b131a06eddb4b9b4d160be44c6fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 08:54:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/14/657314/1/check/kolla-build-centos-binary/1fe2146/ : SUCCESS in 32m 29s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/14/657314/1/check/kolla-build-centos-source/9f36d21/ : SUCCESS in 49m 32s\n- kolla-ansible-centos-source http://logs.openstack.org/14/657314/1/check/kolla-ansible-centos-source/717692e/ : SUCCESS in 1h 22m 07s\n- kolla-ansible-centos-binary http://logs.openstack.org/14/657314/1/check/kolla-ansible-centos-binary/c51bc8c/ : SUCCESS in 1h 07m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/14/657314/1/check/kolla-ansible-centos-source-upgrade/0623321/ : SUCCESS in 1h 19m 00s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/14/657314/1/check/tripleo-build-containers-centos-7/347631e/ : SUCCESS in 42m 17s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/14/657314/1/check/kolla-build-oraclelinux-binary/20a0782/ : SUCCESS in 54m 34s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/14/657314/1/check/kolla-build-oraclelinux-source/1ce51cd/ : SUCCESS in 59m 15s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/14/657314/1/check/kolla-ansible-oraclelinux-source/b50dbb4/ : SUCCESS in 1h 05m 22s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/14/657314/1/check/kolla-ansible-oraclelinux-binary/6d162c0/ : SUCCESS in 1h 11m 29s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/14/657314/1/check/openstack-tox-lower-constraints/a70cdec/ : SUCCESS in 4m 31s\n- openstack-tox-pep8 http://logs.openstack.org/14/657314/1/check/openstack-tox-pep8/27c1e66/ : SUCCESS in 3m 55s\n- openstack-tox-py27 http://logs.openstack.org/14/657314/1/check/openstack-tox-py27/930d286/ : SUCCESS in 2m 56s\n- openstack-tox-py36 http://logs.openstack.org/14/657314/1/check/openstack-tox-py36/8b608ef/ : SUCCESS in 2m 58s\n- openstack-tox-py37 http://logs.openstack.org/14/657314/1/check/openstack-tox-py37/c0d4e86/ : SUCCESS in 4m 35s\n- openstack-tox-docs http://logs.openstack.org/14/657314/1/check/openstack-tox-docs/168404b/html/ : SUCCESS in 2m 37s\n- kolla-build-ubuntu-binary http://logs.openstack.org/14/657314/1/check/kolla-build-ubuntu-binary/b1df66d/ : SUCCESS in 27m 51s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/14/657314/1/check/kolla-build-ubuntu-source/85b9812/ : SUCCESS in 42m 20s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/14/657314/1/check/kolla-ansible-ubuntu-source/6f4adc4/ : SUCCESS in 52m 29s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/14/657314/1/check/kolla-ansible-ubuntu-binary/637280e/ : SUCCESS in 48m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/14/657314/1/check/kolla-ansible-ubuntu-source-upgrade/cd6faf6/ : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f9039ab9253ff7116d16768e92a7522ab16e5309","author":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"date":"2019-05-06 09:40:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ebd2c14ef33b5acc101ba7485dc23e1f7dcc5c20","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-06 09:51:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f2aabef09f20d3addd46bc14f6ed5281ce47d214","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 09:51:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"107be6a27ad80e3285bf1f47190fd7ff847dd13d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 10:39:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/14/657314/1/gate/kolla-build-centos-source/284709a/ : SUCCESS in 35m 26s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/14/657314/1/gate/tripleo-build-containers-centos-7/7dcc45a/ : SUCCESS in 40m 09s\n- kolla-build-oraclelinux-source http://logs.openstack.org/14/657314/1/gate/kolla-build-oraclelinux-source/1fc2d57/ : SUCCESS in 37m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/14/657314/1/gate/openstack-tox-lower-constraints/1ca3d06/ : SUCCESS in 3m 29s\n- openstack-tox-pep8 http://logs.openstack.org/14/657314/1/gate/openstack-tox-pep8/f7910f7/ : SUCCESS in 2m 30s\n- openstack-tox-py27 http://logs.openstack.org/14/657314/1/gate/openstack-tox-py27/eccf040/ : SUCCESS in 3m 36s\n- openstack-tox-py36 http://logs.openstack.org/14/657314/1/gate/openstack-tox-py36/5eb039c/ : SUCCESS in 3m 19s\n- openstack-tox-py37 http://logs.openstack.org/14/657314/1/gate/openstack-tox-py37/24cea71/ : SUCCESS in 3m 57s\n- openstack-tox-docs http://logs.openstack.org/14/657314/1/gate/openstack-tox-docs/e56f4f5/html/ : SUCCESS in 3m 17s\n- kolla-build-ubuntu-source http://logs.openstack.org/14/657314/1/gate/kolla-build-ubuntu-source/b13eb85/ : SUCCESS in 29m 24s","accounts_in_message":[],"_revision_number":1},{"id":"d156a712e8324d501b7034140eb4f7592798c02d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 10:39:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2648c12c8d3567ecafc6c4b6c3c1ce9c6090c972","revisions":{"2648c12c8d3567ecafc6c4b6c3c1ce9c6090c972":{"kind":"REWORK","_number":1,"created":"2019-05-06 05:26:28.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/14/657314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/657314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/657314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/657314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/657314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/657314/1"}}},"commit":{"parents":[{"commit":"eccd7bc1a7817be50c92bcd7881dd45718ab362d","subject":"Merge \"Install udevadm in ironic-conductor image\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eccd7bc1a7817be50c92bcd7881dd45718ab362d"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-17 06:26:03.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-05-06 05:26:28.000000000","tz":0},"subject":"Reuse /var/lib/tempest symlink for rpm also","message":"Reuse /var/lib/tempest symlink for rpm also\n\nhttps://review.openstack.org/#/c/605980/ with this review in\ntripleo, we use /var/lib/tempest path but in kolla tempest\ndockerfile /usr/share/openstack-tempest is used which is not\nmounted in tripleo deployment.\nLet\u0027s reuse the same volume with tempest container instead\nof /usr/share/openstack-tempest.\n\nChange-Id: I402f58ce53fdce426582cf2fc31349da6d2f5ae0\nSigned-off-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n(cherry picked from commit 8b8958459809940b812364542a6994a89624f981)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2648c12c8d3567ecafc6c4b6c3c1ce9c6090c972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2648c12c8d3567ecafc6c4b6c3c1ce9c6090c972"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
