)]}'
{"id":"openstack%2Fkolla~657814","triplet_id":"openstack%2Fkolla~stable%2Fstein~I3e0e86026f5a4a78473bed824cd1682d3a020cd5","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1819461-stable/stein","hashtags":[],"change_id":"I3e0e86026f5a4a78473bed824cd1682d3a020cd5","subject":"Drop systemd support from PAM","status":"MERGED","created":"2019-05-08 14:10:44.000000000","updated":"2019-05-16 11:59:03.000000000","submitted":"2019-05-16 11:59:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657814-1558007943302-4a183efd","meta_rev_id":"3e64a1fc26c39331ab04127292b54f51fb42a83e","_number":657814,"virtual_id_number":657814,"owner":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},{"value":2,"date":"2019-05-16 11:59:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"date":"2019-05-16 10:13:49.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"},{"value":0,"date":"2019-05-16 09:32:12.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"recommended":{"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":1,"date":"2019-05-13 09:26:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":2,"date":"2019-05-09 07:18:58.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-05-09 10:14:50.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2019-05-13 11:34:15.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-09 08:55:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-05-09 10:14:50.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-05-08 14:12:58.000000000","_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},{"value":0,"date":"2019-05-08 14:12:42.000000000","_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-08 14:14:15.000000000","_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"},{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-08 14:12:42.000000000","updated_by":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"reviewer":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"state":"REVIEWER"},{"updated":"2019-05-08 14:12:58.000000000","updated_by":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"reviewer":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"state":"REVIEWER"},{"updated":"2019-05-08 14:14:15.000000000","updated_by":{"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"},"reviewer":{"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"},"state":"REVIEWER"},{"updated":"2019-05-09 07:18:58.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-05-09 08:55:36.000000000","updated_by":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"reviewer":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"state":"REVIEWER"},{"updated":"2019-05-09 10:14:50.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-05-13 09:26:50.000000000","updated_by":{"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},"reviewer":{"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},"state":"REVIEWER"},{"updated":"2019-05-13 11:34:15.000000000","updated_by":{"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},"reviewer":{"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},"state":"REVIEWER"},{"updated":"2019-05-16 10:13:49.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-16 11:59:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b3d9adc26d4d4cd7d23baeef1665356ba9a46c7d","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-05-08 14:10:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4af97c17bfa11de752b82945c26059476a072aa6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-08 14:50:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/14/657814/1/openstack-check/tripleo-build-containers-fedora-28-master/4e8f566/ : SUCCESS in 28m 29s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/14/657814/1/openstack-check/tripleo-build-containers-fedora-28-stein/32da122/ : SUCCESS in 34m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"808856f972e4b54fc843e72f1b2cc25256a398cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 15:55:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/14/657814/1/check/kolla-build-centos-binary/8054e43/ : FAILURE in 41m 14s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/14/657814/1/check/kolla-build-centos-source/b9cebbb/ : SUCCESS in 47m 08s\n- kolla-ansible-centos-source http://logs.openstack.org/14/657814/1/check/kolla-ansible-centos-source/77c6912/ : SUCCESS in 1h 03m 10s\n- kolla-ansible-centos-binary http://logs.openstack.org/14/657814/1/check/kolla-ansible-centos-binary/84d5a05/ : FAILURE in 29m 33s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/14/657814/1/check/kolla-ansible-centos-source-upgrade/72832ac/ : SUCCESS in 1h 18m 12s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/14/657814/1/check/tripleo-build-containers-centos-7/615d2b2/ : SUCCESS in 51m 38s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/14/657814/1/check/kolla-build-oraclelinux-binary/33cc122/ : FAILURE in 49m 20s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/14/657814/1/check/kolla-build-oraclelinux-source/ee4f3af/ : SUCCESS in 52m 09s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/14/657814/1/check/kolla-ansible-oraclelinux-source/90c8d5a/ : SUCCESS in 1h 28m 19s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/14/657814/1/check/kolla-ansible-oraclelinux-binary/f47b6c2/ : FAILURE in 35m 43s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/14/657814/1/check/openstack-tox-lower-constraints/76d1a7e/ : SUCCESS in 4m 16s\n- openstack-tox-pep8 http://logs.openstack.org/14/657814/1/check/openstack-tox-pep8/b6134e4/ : SUCCESS in 2m 57s\n- openstack-tox-py27 http://logs.openstack.org/14/657814/1/check/openstack-tox-py27/c26ab5a/ : SUCCESS in 3m 03s\n- openstack-tox-py36 http://logs.openstack.org/14/657814/1/check/openstack-tox-py36/a487e61/ : SUCCESS in 3m 11s\n- openstack-tox-py37 http://logs.openstack.org/14/657814/1/check/openstack-tox-py37/7dc1dde/ : SUCCESS in 2m 59s\n- openstack-tox-docs http://logs.openstack.org/14/657814/1/check/openstack-tox-docs/1f20ac6/html/ : SUCCESS in 2m 36s\n- kolla-build-ubuntu-binary http://logs.openstack.org/14/657814/1/check/kolla-build-ubuntu-binary/34e1e0e/ : SUCCESS in 35m 10s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/14/657814/1/check/kolla-build-ubuntu-source/c9d1665/ : SUCCESS in 31m 00s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/14/657814/1/check/kolla-ansible-ubuntu-source/63f7376/ : SUCCESS in 58m 22s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/14/657814/1/check/kolla-ansible-ubuntu-binary/8e0d729/ : SUCCESS in 44m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/14/657814/1/check/kolla-ansible-ubuntu-source-upgrade/bc14b70/ : SUCCESS in 55m 09s","accounts_in_message":[],"_revision_number":1},{"id":"6c83ce7e2494959fb068a96154c67e0e07ef30e0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-09 07:18:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d05cf382dde6735bd4ea4f9bc22e3a120eb7b494","author":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"date":"2019-05-09 08:55:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c0b879e80f4332e22a005376716a9b3576e4a0ac","author":{"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},"date":"2019-05-09 08:58:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"60f2107155822e79bf3214e021b1638dea7253e5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-09 10:14:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bb9d3188bf350858577ccebddeecfb4c0a408b39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 10:15:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40eb31777dfdbbbb9c1958d89a0a18ffd7a699d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 11:13:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source http://logs.openstack.org/14/657814/1/gate/kolla-build-centos-source/d8e3d26/ : SUCCESS in 44m 06s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/14/657814/1/gate/tripleo-build-containers-centos-7/85fe123/ : SUCCESS in 43m 41s\n- kolla-build-oraclelinux-source http://logs.openstack.org/14/657814/1/gate/kolla-build-oraclelinux-source/2895e78/ : SUCCESS in 43m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/14/657814/1/gate/openstack-tox-lower-constraints/48d0820/ : SUCCESS in 3m 40s\n- openstack-tox-pep8 http://logs.openstack.org/14/657814/1/gate/openstack-tox-pep8/5b7bd1d/ : FAILURE in 2m 46s\n- openstack-tox-py27 http://logs.openstack.org/14/657814/1/gate/openstack-tox-py27/10884c0/ : SUCCESS in 3m 18s\n- openstack-tox-py36 http://logs.openstack.org/14/657814/1/gate/openstack-tox-py36/32c52ab/ : SUCCESS in 3m 15s\n- openstack-tox-py37 http://logs.openstack.org/14/657814/1/gate/openstack-tox-py37/60d21be/ : SUCCESS in 3m 31s\n- openstack-tox-docs http://logs.openstack.org/14/657814/1/gate/openstack-tox-docs/ce44314/html/ : SUCCESS in 2m 37s\n- kolla-build-ubuntu-source http://logs.openstack.org/14/657814/1/gate/kolla-build-ubuntu-source/8c8788c/ : SUCCESS in 36m 53s","accounts_in_message":[],"_revision_number":1},{"id":"75dae75cef42a122ccb78403f9f0425a8fd65648","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-09 11:15:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"519155500d617172208da667954533139ebb46e9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-09 11:54:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/14/657814/1/openstack-check/tripleo-build-containers-fedora-28-master/05d502e/ : SUCCESS in 32m 39s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/14/657814/1/openstack-check/tripleo-build-containers-fedora-28-stein/bb62e1f/ : SUCCESS in 34m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a352d30c2640d3e9cba90390a0fe9aca16d26393","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 13:05:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/14/657814/1/check/kolla-build-centos-binary/e829039/ : FAILURE in 25m 48s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/14/657814/1/check/kolla-build-centos-source/9d0e85d/ : SUCCESS in 37m 51s\n- kolla-ansible-centos-source http://logs.openstack.org/14/657814/1/check/kolla-ansible-centos-source/e2d90f6/ : SUCCESS in 1h 18m 09s\n- kolla-ansible-centos-binary http://logs.openstack.org/14/657814/1/check/kolla-ansible-centos-binary/c3f782b/ : FAILURE in 29m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/14/657814/1/check/kolla-ansible-centos-source-upgrade/6c8e784/ : SUCCESS in 1h 19m 09s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/14/657814/1/check/tripleo-build-containers-centos-7/424ae4c/ : SUCCESS in 46m 38s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/14/657814/1/check/kolla-build-oraclelinux-binary/eb1453f/ : FAILURE in 50m 42s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/14/657814/1/check/kolla-build-oraclelinux-source/817f45e/ : SUCCESS in 57m 47s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/14/657814/1/check/kolla-ansible-oraclelinux-source/8553e69/ : SUCCESS in 1h 08m 42s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/14/657814/1/check/kolla-ansible-oraclelinux-binary/754578c/ : FAILURE in 35m 52s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/14/657814/1/check/openstack-tox-lower-constraints/4a74340/ : SUCCESS in 3m 41s\n- openstack-tox-pep8 http://logs.openstack.org/14/657814/1/check/openstack-tox-pep8/9427c36/ : FAILURE in 2m 38s\n- openstack-tox-py27 http://logs.openstack.org/14/657814/1/check/openstack-tox-py27/7796b34/ : SUCCESS in 3m 32s\n- openstack-tox-py36 http://logs.openstack.org/14/657814/1/check/openstack-tox-py36/ba47361/ : SUCCESS in 3m 15s\n- openstack-tox-py37 http://logs.openstack.org/14/657814/1/check/openstack-tox-py37/4b69086/ : SUCCESS in 4m 05s\n- openstack-tox-docs http://logs.openstack.org/14/657814/1/check/openstack-tox-docs/afd3792/html/ : SUCCESS in 3m 32s\n- kolla-build-ubuntu-binary http://logs.openstack.org/14/657814/1/check/kolla-build-ubuntu-binary/69775ad/ : SUCCESS in 1h 47m 11s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/14/657814/1/check/kolla-build-ubuntu-source/b056a22/ : SUCCESS in 41m 26s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/14/657814/1/check/kolla-ansible-ubuntu-source/cc2df18/ : SUCCESS in 46m 05s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/14/657814/1/check/kolla-ansible-ubuntu-binary/fe1919e/ : SUCCESS in 47m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/14/657814/1/check/kolla-ansible-ubuntu-source-upgrade/7bc2b72/ : SUCCESS in 58m 40s","accounts_in_message":[],"_revision_number":1},{"id":"d9b11825c2dbd466962436b8f006a86a3d302538","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-05-10 06:40:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"28b0f28524f930ed130ca9a9b489613304f58e87","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-10 07:15:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/14/657814/1/openstack-check/tripleo-build-containers-fedora-28-master/da238c9/ : SUCCESS in 33m 49s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/14/657814/1/openstack-check/tripleo-build-containers-fedora-28-stein/5e87a07/ : SUCCESS in 33m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"575447f522bee054e7e803bd14cbb10baa6b9985","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 08:38:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/14/657814/1/check/kolla-build-centos-binary/6646d92/ : FAILURE in 42m 47s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/14/657814/1/check/kolla-build-centos-source/143ad4c/ : SUCCESS in 50m 01s\n- kolla-ansible-centos-source http://logs.openstack.org/14/657814/1/check/kolla-ansible-centos-source/d09d852/ : SUCCESS in 1h 07m 38s\n- kolla-ansible-centos-binary http://logs.openstack.org/14/657814/1/check/kolla-ansible-centos-binary/5ba8c83/ : FAILURE in 28m 05s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/14/657814/1/check/kolla-ansible-centos-source-upgrade/b93952e/ : SUCCESS in 1h 15m 45s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/14/657814/1/check/tripleo-build-containers-centos-7/2f0259a/ : SUCCESS in 47m 05s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/14/657814/1/check/kolla-build-oraclelinux-binary/86f2069/ : FAILURE in 1h 06m 36s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/14/657814/1/check/kolla-build-oraclelinux-source/3cd4528/ : SUCCESS in 46m 28s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/14/657814/1/check/kolla-ansible-oraclelinux-source/b09fa6d/ : SUCCESS in 1h 49m 42s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/14/657814/1/check/kolla-ansible-oraclelinux-binary/cec05f0/ : FAILURE in 34m 43s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/14/657814/1/check/openstack-tox-lower-constraints/f85fb85/ : SUCCESS in 3m 48s\n- openstack-tox-pep8 http://logs.openstack.org/14/657814/1/check/openstack-tox-pep8/e99b9f4/ : FAILURE in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/14/657814/1/check/openstack-tox-py27/fdc97ad/ : SUCCESS in 2m 46s\n- openstack-tox-py36 http://logs.openstack.org/14/657814/1/check/openstack-tox-py36/82871c3/ : SUCCESS in 2m 59s\n- openstack-tox-py37 http://logs.openstack.org/14/657814/1/check/openstack-tox-py37/73cf5fc/ : SUCCESS in 3m 33s\n- openstack-tox-docs http://logs.openstack.org/14/657814/1/check/openstack-tox-docs/c274cec/html/ : SUCCESS in 2m 33s\n- kolla-build-ubuntu-binary http://logs.openstack.org/14/657814/1/check/kolla-build-ubuntu-binary/3748248/ : FAILURE in 29m 14s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/14/657814/1/check/kolla-build-ubuntu-source/4dbbb07/ : FAILURE in 31m 35s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/14/657814/1/check/kolla-ansible-ubuntu-source/f60cef6/ : SUCCESS in 54m 52s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/14/657814/1/check/kolla-ansible-ubuntu-binary/87d2f4d/ : SUCCESS in 45m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/14/657814/1/check/kolla-ansible-ubuntu-source-upgrade/ea59adb/ : SUCCESS in 57m 23s","accounts_in_message":[],"_revision_number":1},{"id":"08fbac98686a9221dffd42416b184dd52fab2f13","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-05-13 06:20:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a06bb42ca641a473dde7f26f302a4cce68ad47c8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-13 07:04:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/14/657814/1/openstack-check/tripleo-build-containers-fedora-28-master/3c8bbf7/ : SUCCESS in 31m 57s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/14/657814/1/openstack-check/tripleo-build-containers-fedora-28-stein/1c1637c/ : SUCCESS in 30m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"930367b8579a932f7765439259061b3c81b51c4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 07:47:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/14/657814/1/check/kolla-build-centos-binary/f001e90/ : FAILURE in 45m 23s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/14/657814/1/check/kolla-build-centos-source/0dc2267/ : SUCCESS in 54m 13s\n- kolla-ansible-centos-source http://logs.openstack.org/14/657814/1/check/kolla-ansible-centos-source/64ded6b/ : SUCCESS in 1h 13m 53s\n- kolla-ansible-centos-binary http://logs.openstack.org/14/657814/1/check/kolla-ansible-centos-binary/eb9872d/ : FAILURE in 32m 32s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/14/657814/1/check/kolla-ansible-centos-source-upgrade/51a8e1e/ : SUCCESS in 1h 19m 04s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/14/657814/1/check/tripleo-build-containers-centos-7/7a95c80/ : SUCCESS in 52m 00s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/14/657814/1/check/kolla-build-oraclelinux-binary/8dbc7e8/ : FAILURE in 1h 07m 51s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/14/657814/1/check/kolla-build-oraclelinux-source/6ff0d04/ : SUCCESS in 59m 49s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/14/657814/1/check/kolla-ansible-oraclelinux-source/2cda5af/ : SUCCESS in 1h 12m 09s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/14/657814/1/check/kolla-ansible-oraclelinux-binary/3bce59c/ : FAILURE in 33m 00s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/14/657814/1/check/openstack-tox-lower-constraints/494764f/ : SUCCESS in 4m 04s\n- openstack-tox-pep8 http://logs.openstack.org/14/657814/1/check/openstack-tox-pep8/1ac272d/ : FAILURE in 3m 14s\n- openstack-tox-py27 http://logs.openstack.org/14/657814/1/check/openstack-tox-py27/8d2cc3f/ : SUCCESS in 3m 18s\n- openstack-tox-py36 http://logs.openstack.org/14/657814/1/check/openstack-tox-py36/6378597/ : SUCCESS in 3m 04s\n- openstack-tox-py37 http://logs.openstack.org/14/657814/1/check/openstack-tox-py37/314c69a/ : SUCCESS in 3m 40s\n- openstack-tox-docs http://logs.openstack.org/14/657814/1/check/openstack-tox-docs/3b6d2f5/html/ : SUCCESS in 2m 50s\n- kolla-build-ubuntu-binary http://logs.openstack.org/14/657814/1/check/kolla-build-ubuntu-binary/0ade8af/ : SUCCESS in 27m 37s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/14/657814/1/check/kolla-build-ubuntu-source/173992d/ : SUCCESS in 28m 37s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/14/657814/1/check/kolla-ansible-ubuntu-source/b465106/ : SUCCESS in 48m 28s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/14/657814/1/check/kolla-ansible-ubuntu-binary/b8d4297/ : SUCCESS in 44m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/14/657814/1/check/kolla-ansible-ubuntu-source-upgrade/9b5d957/ : SUCCESS in 54m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ce8be38b588fff94ffc14094e8c523a212248ee5","author":{"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},"date":"2019-05-13 09:26:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"07f38021a031d53b16f84bb5b12267ef8d80f154","author":{"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},"date":"2019-05-13 11:34:15.000000000","message":"Patch Set 1: -Code-Review\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e0031683b3f4fc3dec74c192d7ae982cfa105b7e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-13 12:06:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/14/657814/1/openstack-check/tripleo-build-containers-fedora-28-master/00af481/ : SUCCESS in 30m 17s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/14/657814/1/openstack-check/tripleo-build-containers-fedora-28-stein/c9c5f05/ : SUCCESS in 30m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1dcda7676200603725978a10a42eb7620067ddbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 13:16:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/14/657814/1/check/kolla-build-centos-binary/a7c2d96/ : FAILURE in 37m 34s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/14/657814/1/check/kolla-build-centos-source/cd26532/ : SUCCESS in 38m 51s\n- kolla-ansible-centos-source http://logs.openstack.org/14/657814/1/check/kolla-ansible-centos-source/0515bfc/ : SUCCESS in 1h 08m 41s\n- kolla-ansible-centos-binary http://logs.openstack.org/14/657814/1/check/kolla-ansible-centos-binary/787b587/ : FAILURE in 28m 00s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/14/657814/1/check/kolla-ansible-centos-source-upgrade/300089b/ : SUCCESS in 1h 32m 45s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/14/657814/1/check/tripleo-build-containers-centos-7/8d672e7/ : SUCCESS in 43m 15s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/14/657814/1/check/kolla-build-oraclelinux-binary/ee8895c/ : FAILURE in 48m 43s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/14/657814/1/check/kolla-build-oraclelinux-source/bf84350/ : SUCCESS in 51m 43s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/14/657814/1/check/kolla-ansible-oraclelinux-source/926ed91/ : SUCCESS in 1h 09m 19s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/14/657814/1/check/kolla-ansible-oraclelinux-binary/1118f14/ : FAILURE in 34m 14s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/14/657814/1/check/openstack-tox-lower-constraints/7853e8c/ : SUCCESS in 3m 31s\n- openstack-tox-pep8 http://logs.openstack.org/14/657814/1/check/openstack-tox-pep8/31cf13f/ : FAILURE in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/14/657814/1/check/openstack-tox-py27/fcdd147/ : SUCCESS in 4m 04s\n- openstack-tox-py36 http://logs.openstack.org/14/657814/1/check/openstack-tox-py36/6fdea3b/ : SUCCESS in 3m 15s\n- openstack-tox-py37 http://logs.openstack.org/14/657814/1/check/openstack-tox-py37/f95a24c/ : SUCCESS in 4m 06s\n- openstack-tox-docs http://logs.openstack.org/14/657814/1/check/openstack-tox-docs/adb98d5/html/ : SUCCESS in 3m 53s\n- kolla-build-ubuntu-binary http://logs.openstack.org/14/657814/1/check/kolla-build-ubuntu-binary/6fbd4b3/ : SUCCESS in 31m 51s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/14/657814/1/check/kolla-build-ubuntu-source/6087938/ : SUCCESS in 35m 02s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/14/657814/1/check/kolla-ansible-ubuntu-source/ed3a637/ : SUCCESS in 59m 51s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/14/657814/1/check/kolla-ansible-ubuntu-binary/c91ba6c/ : SUCCESS in 47m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/14/657814/1/check/kolla-ansible-ubuntu-source-upgrade/9646381/ : SUCCESS in 59m 09s","accounts_in_message":[],"_revision_number":1},{"id":"abcfbcf029d3841357c66dc76c04562f1f236f84","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-05-16 09:32:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b6bcc1731830abdf7ef4d6aa6796a00db7a991cf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-16 10:13:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/14/657814/1/openstack-check/tripleo-build-containers-fedora-28-master/a44f449/ : SUCCESS in 32m 44s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/14/657814/1/openstack-check/tripleo-build-containers-fedora-28-stein/b7fa64f/ : SUCCESS in 34m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6ade5a258fa71e397564c07c2c8477db24302288","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 10:59:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/14/657814/1/check/kolla-build-centos-binary/e216533/ : FAILURE in 52m 49s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/14/657814/1/check/kolla-build-centos-source/e72e342/ : SUCCESS in 52m 34s\n- kolla-ansible-centos-source http://logs.openstack.org/14/657814/1/check/kolla-ansible-centos-source/da37f03/ : SUCCESS in 1h 07m 49s\n- kolla-ansible-centos-binary http://logs.openstack.org/14/657814/1/check/kolla-ansible-centos-binary/1d6b12c/ : FAILURE in 33m 01s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/14/657814/1/check/kolla-ansible-centos-source-upgrade/50594bd/ : SUCCESS in 1h 23m 06s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/14/657814/1/check/tripleo-build-containers-centos-7/5c377c2/ : SUCCESS in 42m 31s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/14/657814/1/check/kolla-build-oraclelinux-binary/29177ba/ : FAILURE in 53m 23s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/14/657814/1/check/kolla-build-oraclelinux-source/3cef37e/ : SUCCESS in 42m 45s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/14/657814/1/check/kolla-ansible-oraclelinux-source/f5106cb/ : SUCCESS in 1h 11m 26s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/14/657814/1/check/kolla-ansible-oraclelinux-binary/bce4067/ : FAILURE in 31m 21s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/14/657814/1/check/openstack-tox-lower-constraints/a204f30/ : SUCCESS in 5m 05s\n- openstack-tox-pep8 http://logs.openstack.org/14/657814/1/check/openstack-tox-pep8/76a7420/ : SUCCESS in 4m 42s\n- openstack-tox-py27 http://logs.openstack.org/14/657814/1/check/openstack-tox-py27/a3b969f/ : SUCCESS in 4m 19s\n- openstack-tox-py36 http://logs.openstack.org/14/657814/1/check/openstack-tox-py36/d41480a/ : SUCCESS in 4m 46s\n- openstack-tox-py37 http://logs.openstack.org/14/657814/1/check/openstack-tox-py37/5d3db70/ : SUCCESS in 5m 17s\n- openstack-tox-docs http://logs.openstack.org/14/657814/1/check/openstack-tox-docs/0849cab/html/ : SUCCESS in 2m 44s\n- kolla-build-ubuntu-binary http://logs.openstack.org/14/657814/1/check/kolla-build-ubuntu-binary/81d46b2/ : SUCCESS in 28m 07s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/14/657814/1/check/kolla-build-ubuntu-source/3631a0f/ : SUCCESS in 29m 48s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/14/657814/1/check/kolla-ansible-ubuntu-source/f707c2e/ : SUCCESS in 51m 40s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/14/657814/1/check/kolla-ansible-ubuntu-binary/67ae619/ : SUCCESS in 43m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/14/657814/1/check/kolla-ansible-ubuntu-source-upgrade/70d4e0a/ : SUCCESS in 56m 37s","accounts_in_message":[],"_revision_number":1},{"id":"fa71a127ece58398cdd5708a660e2450afa99392","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 10:59:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"737c8b643efea8217e53e018da910782065ec997","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 11:59:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7c06240a1fdfb77fb42c30dfae87d651ac53d938","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 11:59:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/14/657814/1/gate/kolla-build-centos-source/b76f8b0/ : SUCCESS in 54m 11s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/14/657814/1/gate/tripleo-build-containers-centos-7/0b3e404/ : SUCCESS in 45m 34s\n- kolla-build-oraclelinux-source http://logs.openstack.org/14/657814/1/gate/kolla-build-oraclelinux-source/e03813c/ : SUCCESS in 49m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/14/657814/1/gate/openstack-tox-lower-constraints/43aa0ad/ : SUCCESS in 4m 04s\n- openstack-tox-pep8 http://logs.openstack.org/14/657814/1/gate/openstack-tox-pep8/616c1ad/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/14/657814/1/gate/openstack-tox-py27/57a331a/ : SUCCESS in 3m 24s\n- openstack-tox-py36 http://logs.openstack.org/14/657814/1/gate/openstack-tox-py36/6d654c5/ : SUCCESS in 3m 23s\n- openstack-tox-py37 http://logs.openstack.org/14/657814/1/gate/openstack-tox-py37/078e60f/ : SUCCESS in 3m 35s\n- openstack-tox-docs http://logs.openstack.org/14/657814/1/gate/openstack-tox-docs/eb5730f/html/ : SUCCESS in 2m 52s\n- kolla-build-ubuntu-source http://logs.openstack.org/14/657814/1/gate/kolla-build-ubuntu-source/135bbd8/ : SUCCESS in 30m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c56ee10d21496f0d1e8f2757226526686f524c21","revisions":{"c56ee10d21496f0d1e8f2757226526686f524c21":{"kind":"REWORK","_number":1,"created":"2019-05-08 14:10:44.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/14/657814/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/657814/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/657814/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/657814/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/657814/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/657814/1"}}},"commit":{"parents":[{"commit":"2648c12c8d3567ecafc6c4b6c3c1ce9c6090c972","subject":"Reuse /var/lib/tempest symlink for rpm also","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2648c12c8d3567ecafc6c4b6c3c1ce9c6090c972"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-04-30 14:11:46.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-05-08 14:10:44.000000000","tz":0},"subject":"Drop systemd support from PAM","message":"Drop systemd support from PAM\n\nThe line being removed is the cause of audit.log spam with a\nSELinux enforcing (and permissive) system. They are due to the\nfact we\u0027re calling \"sudo\" from within some containers, and PAM\nwants to connect to the host \"dbus\" socket (mounted in /run).\nThis action is obviously denied by SELinux.\n\nThe \"-\" prefix makes this module optional anyway, but apparently\nPAM does see it and tries to use it.\n\nDropping this module has no consequences other than preventing\nlog spamming.\n\nChange-Id: I3e0e86026f5a4a78473bed824cd1682d3a020cd5\nCo-Authored-By: Michele Baldessari \u003cmbaldess@redhat.com\u003e\nResolves-Bug: #1819461\n(cherry picked from commit c4b62490714ea849cedfe07d4b9a6aebf1f815c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c56ee10d21496f0d1e8f2757226526686f524c21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c56ee10d21496f0d1e8f2757226526686f524c21"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
