)]}'
{"id":"openstack%2Fkolla~658169","triplet_id":"openstack%2Fkolla~stable%2Frocky~I9828c23e4bcec7553a69aa5f20e990e0ef0538ea","project":"openstack/kolla","branch":"stable/rocky","topic":"bug/1828416-stable/rocky","hashtags":[],"change_id":"I9828c23e4bcec7553a69aa5f20e990e0ef0538ea","subject":"Fix bandit failure B105 in keystone fernet","status":"MERGED","created":"2019-05-09 18:14:00.000000000","updated":"2019-05-10 10:58:39.000000000","submitted":"2019-05-10 10:58:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658169-1557485919779-4a183efd","meta_rev_id":"2e64ba5cc744695274102c22ddfbcfe690a87d82","_number":658169,"virtual_id_number":658169,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-05-10 10:58:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-09 18:51:00.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"recommended":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"all":[{"value":1,"date":"2019-05-10 06:10:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"value":2,"date":"2019-05-10 08:31:15.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-05-09 18:14:06.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"value":1,"date":"2019-05-10 08:31:15.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 18:51:00.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-10 06:10:33.000000000","updated_by":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"reviewer":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"state":"REVIEWER"},{"updated":"2019-05-10 08:31:15.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-05-10 10:58:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"29580323df540a7a5c66760bedad5b6d945d7ecc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-09 18:14:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d030b1d96b4bf4c8ec5581759a5baba57c266a05","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-09 18:14:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4343011e264b4034787f0b9c8d7188cf26dd8b98","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-09 18:51:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/69/658169/1/openstack-check/tripleo-build-containers-fedora-28-master/d8e63b4/ : RETRY_LIMIT in 2m 36s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/69/658169/1/openstack-check/tripleo-build-containers-fedora-28-stein/038e0fc/ : SUCCESS in 33m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae9f5663c64608f155c38185d61adeb4799b949b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 20:28:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/69/658169/1/check/kolla-build-centos-binary/7ce92d1/ : SUCCESS in 38m 04s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/658169/1/check/kolla-build-centos-source/b57b6fb/ : SUCCESS in 37m 48s\n- kolla-ansible-centos-source http://logs.openstack.org/69/658169/1/check/kolla-ansible-centos-source/2338210/ : SUCCESS in 1h 16m 55s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/658169/1/check/kolla-ansible-centos-binary/fd688a1/ : SUCCESS in 1h 13m 34s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/658169/1/check/kolla-build-oraclelinux-binary/3bcbe45/ : SUCCESS in 47m 05s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/658169/1/check/kolla-build-oraclelinux-source/32ce5fa/ : SUCCESS in 45m 43s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/658169/1/check/kolla-ansible-oraclelinux-source/c5b6a0d/ : FAILURE in 53m 08s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/658169/1/check/kolla-ansible-oraclelinux-binary/4f75bd4/ : SUCCESS in 1h 07m 59s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/69/658169/1/check/openstack-tox-lower-constraints/c4f4fff/ : SUCCESS in 2m 57s\n- openstack-tox-pep8 http://logs.openstack.org/69/658169/1/check/openstack-tox-pep8/69a6c76/ : SUCCESS in 3m 46s\n- openstack-tox-py27 http://logs.openstack.org/69/658169/1/check/openstack-tox-py27/e2ca371/ : SUCCESS in 3m 09s\n- openstack-tox-py35 http://logs.openstack.org/69/658169/1/check/openstack-tox-py35/5a8ac45/ : SUCCESS in 4m 17s\n- build-openstack-sphinx-docs http://logs.openstack.org/69/658169/1/check/build-openstack-sphinx-docs/1784732/html/ : SUCCESS in 3m 28s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/658169/1/check/kolla-build-ubuntu-binary/e3ee40d/ : SUCCESS in 1h 08m 27s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/658169/1/check/kolla-build-ubuntu-source/6559926/ : SUCCESS in 33m 32s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/658169/1/check/kolla-ansible-ubuntu-source/c087faa/ : SUCCESS in 1h 05m 25s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/658169/1/check/kolla-ansible-ubuntu-binary/9bf211d/ : TIMED_OUT in 2h 00m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45545568ac071508b58e87900947456c3d23889a","author":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"date":"2019-05-10 06:10:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1dd94e72ddedc63c33a980efd17ddaa8074245e9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-10 08:31:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4343c777df5c5e9ee916b9a6ae9f063effc7e1ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 08:31:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"74cde8988b85f63ec78587712f0a9a03594560d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 10:58:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"624826ce78799b4818c676bb694daf3848054c88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 10:58:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/69/658169/1/gate/kolla-build-centos-source/45a086e/ : SUCCESS in 39m 16s\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/658169/1/gate/kolla-build-oraclelinux-source/f7b0a48/ : SUCCESS in 32m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/658169/1/gate/openstack-tox-lower-constraints/1a42752/ : SUCCESS in 3m 25s\n- openstack-tox-pep8 http://logs.openstack.org/69/658169/1/gate/openstack-tox-pep8/80c5b19/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/69/658169/1/gate/openstack-tox-py27/f710c0a/ : SUCCESS in 3m 21s\n- openstack-tox-py35 http://logs.openstack.org/69/658169/1/gate/openstack-tox-py35/7cf560d/ : SUCCESS in 3m 05s\n- build-openstack-sphinx-docs http://logs.openstack.org/69/658169/1/gate/build-openstack-sphinx-docs/73c5519/html/ : SUCCESS in 3m 18s\n- kolla-build-ubuntu-source http://logs.openstack.org/69/658169/1/gate/kolla-build-ubuntu-source/02ca97f/ : SUCCESS in 29m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0eb7f031c29a5d4c6e05c1787af2e6a7d09749f5","revisions":{"0eb7f031c29a5d4c6e05c1787af2e6a7d09749f5":{"kind":"REWORK","_number":1,"created":"2019-05-09 18:14:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/69/658169/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/658169/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/658169/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/658169/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/658169/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/658169/1"}}},"commit":{"parents":[{"commit":"eae9b95690b6d3d0962f6801740b8d9a7cb52075","subject":"Merge \"Bump OpenStack service versions (Rocky)\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eae9b95690b6d3d0962f6801740b8d9a7cb52075"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-09 13:28:21.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-09 18:14:00.000000000","tz":0},"subject":"Fix bandit failure B105 in keystone fernet","message":"Fix bandit failure B105 in keystone fernet\n\n\u003e\u003e Issue: [B105:hardcoded_password_string] Possible hardcoded password: \u0027/etc/keystone/fernet-keys\u0027\n   Severity: Low Confidence: Medium\n      Location: docker/keystone/keystone-fernet/fetch_fernet_tokens.py:27\n         More Info: https://bandit.readthedocs.io/en/latest/plugins/b105_hardcoded_password_string.html\n\nIgnore bandit failures on this line.\n\nChange-Id: I9828c23e4bcec7553a69aa5f20e990e0ef0538ea\nCloses-Bug: #1828416\n(cherry picked from commit e6387634c6a7170e96606b9a8ff680f08d81e801)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eb7f031c29a5d4c6e05c1787af2e6a7d09749f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eb7f031c29a5d4c6e05c1787af2e6a7d09749f5"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
