)]}'
{"id":"openstack%2Fkolla~658456","triplet_id":"openstack%2Fkolla~master~I6b205db2f3c01c6802e637723167ffc6d3c66d17","project":"openstack/kolla","branch":"master","topic":"bug/1828542","hashtags":[],"change_id":"I6b205db2f3c01c6802e637723167ffc6d3c66d17","subject":"Add RPM GPG keys import retry","status":"ABANDONED","created":"2019-05-10 19:17:41.000000000","updated":"2022-04-04 15:19:16.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"32909ac994d7f92901b9a5c1387a96db315908d2","_number":658456,"virtual_id_number":658456,"owner":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"date":"2019-05-13 17:48:25.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":-1,"date":"2019-05-10 20:54:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-05-13 14:23:04.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"disliked":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"all":[{"value":1,"date":"2019-05-13 11:30:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":-1,"date":"2019-05-13 14:50:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-10 20:54:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-13 11:30:01.000000000","updated_by":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"reviewer":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"state":"REVIEWER"},{"updated":"2019-05-13 14:23:04.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-13 17:48:25.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"5dd8d89e309426f8538cb3f5b3600d5dec101031","author":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"date":"2019-05-10 19:17:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"577b344405b667ec13671f6779866995b9f1d019","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-10 19:53:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/56/658456/1/openstack-check/tripleo-build-containers-fedora-28-master/933cb14/ : SUCCESS in 32m 17s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/56/658456/1/openstack-check/tripleo-build-containers-fedora-28-stein/5424b44/ : SUCCESS in 31m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"640a665c2ec9c6d56e833676f4c5ecfca9291a94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 20:54:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/56/658456/1/check/kolla-build-centos-binary/633713f/ : FAILURE in 47m 35s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/56/658456/1/check/kolla-build-centos-source/843dd5e/ : FAILURE in 41m 12s\n- kolla-ansible-centos-source http://logs.openstack.org/56/658456/1/check/kolla-ansible-centos-source/1db1dc5/ : SUCCESS in 1h 10m 24s\n- kolla-ansible-centos-binary http://logs.openstack.org/56/658456/1/check/kolla-ansible-centos-binary/5124146/ : RETRY_LIMIT in 3m 01s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/56/658456/1/check/kolla-ansible-centos-source-upgrade/76844e8/ : SUCCESS in 1h 20m 30s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/56/658456/1/check/tripleo-build-containers-centos-7/efbd9d5/ : SUCCESS in 50m 34s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/56/658456/1/check/kolla-build-oraclelinux-binary/bd5c3be/ : FAILURE in 44m 21s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/56/658456/1/check/kolla-build-oraclelinux-source/faf77a5/ : FAILURE in 50m 22s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/56/658456/1/check/kolla-ansible-oraclelinux-source/b4b72e0/ : SUCCESS in 1h 13m 26s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/56/658456/1/check/kolla-ansible-oraclelinux-binary/a8e8669/ : SUCCESS in 1h 08m 33s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/658456/1/check/openstack-tox-lower-constraints/23bc5e4/ : SUCCESS in 5m 36s\n- openstack-tox-pep8 http://logs.openstack.org/56/658456/1/check/openstack-tox-pep8/d634b64/ : SUCCESS in 3m 07s\n- openstack-tox-py27 http://logs.openstack.org/56/658456/1/check/openstack-tox-py27/f0f4d75/ : SUCCESS in 2m 55s\n- openstack-tox-py36 http://logs.openstack.org/56/658456/1/check/openstack-tox-py36/d2f618a/ : SUCCESS in 3m 11s\n- openstack-tox-py37 http://logs.openstack.org/56/658456/1/check/openstack-tox-py37/ec83531/ : SUCCESS in 3m 23s\n- openstack-tox-docs http://logs.openstack.org/56/658456/1/check/openstack-tox-docs/1f3b0e6/html/ : SUCCESS in 2m 54s\n- kolla-build-ubuntu-binary http://logs.openstack.org/56/658456/1/check/kolla-build-ubuntu-binary/6a6523a/ : SUCCESS in 29m 25s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/56/658456/1/check/kolla-build-ubuntu-source/288dada/ : SUCCESS in 33m 00s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/56/658456/1/check/kolla-ansible-ubuntu-source/1a9de27/ : SUCCESS in 53m 39s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/56/658456/1/check/kolla-ansible-ubuntu-binary/6790122/ : SUCCESS in 57m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/56/658456/1/check/kolla-ansible-ubuntu-source-upgrade/8de5fa4/ : SUCCESS in 59m 51s","accounts_in_message":[],"_revision_number":1},{"id":"2b685e661dce110e4707c3ce17976a0974bae7a0","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-05-13 11:30:01.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6856e15155eb4a0d4015055c0b5736112c8a7f2b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-13 12:01:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/56/658456/1/openstack-check/tripleo-build-containers-fedora-28-master/71bb34c/ : SUCCESS in 27m 23s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/56/658456/1/openstack-check/tripleo-build-containers-fedora-28-stein/673cd7c/ : SUCCESS in 29m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a73ae56f00f0b60528b783b933a14b2b20c1993","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 13:03:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/56/658456/1/check/kolla-build-centos-binary/f759e0a/ : FAILURE in 40m 25s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/56/658456/1/check/kolla-build-centos-source/3f16feb/ : FAILURE in 48m 46s\n- kolla-ansible-centos-source http://logs.openstack.org/56/658456/1/check/kolla-ansible-centos-source/b2af653/ : SUCCESS in 1h 09m 47s\n- kolla-ansible-centos-binary http://logs.openstack.org/56/658456/1/check/kolla-ansible-centos-binary/cb84f45/ : SUCCESS in 1h 12m 31s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/56/658456/1/check/kolla-ansible-centos-source-upgrade/578868e/ : SUCCESS in 1h 19m 40s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/56/658456/1/check/tripleo-build-containers-centos-7/19ca2ef/ : SUCCESS in 34m 22s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/56/658456/1/check/kolla-build-oraclelinux-binary/e864660/ : FAILURE in 1h 02m 27s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/56/658456/1/check/kolla-build-oraclelinux-source/4ebfd8a/ : FAILURE in 45m 47s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/56/658456/1/check/kolla-ansible-oraclelinux-source/0e70305/ : SUCCESS in 1h 05m 47s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/56/658456/1/check/kolla-ansible-oraclelinux-binary/ae62a20/ : SUCCESS in 1h 04m 05s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/658456/1/check/openstack-tox-lower-constraints/37b5267/ : SUCCESS in 3m 45s\n- openstack-tox-pep8 http://logs.openstack.org/56/658456/1/check/openstack-tox-pep8/48a8932/ : SUCCESS in 2m 57s\n- openstack-tox-py27 http://logs.openstack.org/56/658456/1/check/openstack-tox-py27/0ea374b/ : SUCCESS in 3m 03s\n- openstack-tox-py36 http://logs.openstack.org/56/658456/1/check/openstack-tox-py36/f7fbc14/ : SUCCESS in 2m 35s\n- openstack-tox-py37 http://logs.openstack.org/56/658456/1/check/openstack-tox-py37/65f4163/ : SUCCESS in 3m 28s\n- openstack-tox-docs http://logs.openstack.org/56/658456/1/check/openstack-tox-docs/33524ff/html/ : SUCCESS in 2m 38s\n- kolla-build-ubuntu-binary http://logs.openstack.org/56/658456/1/check/kolla-build-ubuntu-binary/cab80f2/ : SUCCESS in 28m 56s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/56/658456/1/check/kolla-build-ubuntu-source/950d58a/ : SUCCESS in 42m 18s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/56/658456/1/check/kolla-ansible-ubuntu-source/14bbfa6/ : SUCCESS in 57m 12s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/56/658456/1/check/kolla-ansible-ubuntu-binary/09ddba1/ : SUCCESS in 41m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/56/658456/1/check/kolla-ansible-ubuntu-source-upgrade/ed71fdf/ : SUCCESS in 1h 02m 24s","accounts_in_message":[],"_revision_number":1},{"id":"dffc8d1a5cd091cf4439750c356adf7ea8db606c","author":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"date":"2019-05-13 13:49:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3598252bf367515c15847c09cc50e50a289c9a7e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-13 14:23:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/56/658456/1/openstack-check/tripleo-build-containers-fedora-28-master/10573d8/ : SUCCESS in 32m 15s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/56/658456/1/openstack-check/tripleo-build-containers-fedora-28-stein/0188b77/ : SUCCESS in 31m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ea737863c7037c770ef6dd59cf1e3ae44cb2258d","author":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"date":"2019-05-13 14:50:49.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"9d63a766f8fbb707730f9d57449ebf89bf898ffa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 15:18:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/56/658456/1/check/kolla-build-centos-binary/0acac2d/ : FAILURE in 38m 15s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/56/658456/1/check/kolla-build-centos-source/7e2bac2/ : FAILURE in 56m 32s\n- kolla-ansible-centos-source http://logs.openstack.org/56/658456/1/check/kolla-ansible-centos-source/32fe44a/ : SUCCESS in 1h 09m 49s\n- kolla-ansible-centos-binary http://logs.openstack.org/56/658456/1/check/kolla-ansible-centos-binary/f639f21/ : SUCCESS in 1h 06m 40s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/56/658456/1/check/kolla-ansible-centos-source-upgrade/5611876/ : SUCCESS in 1h 19m 05s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/56/658456/1/check/tripleo-build-containers-centos-7/c8c67f3/ : SUCCESS in 41m 22s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/56/658456/1/check/kolla-build-oraclelinux-binary/7409e26/ : FAILURE in 51m 11s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/56/658456/1/check/kolla-build-oraclelinux-source/2bdf71e/ : FAILURE in 49m 44s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/56/658456/1/check/kolla-ansible-oraclelinux-source/ccd7c66/ : SUCCESS in 1h 28m 43s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/56/658456/1/check/kolla-ansible-oraclelinux-binary/ad7f081/ : SUCCESS in 1h 11m 06s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/658456/1/check/openstack-tox-lower-constraints/46ed045/ : SUCCESS in 3m 55s\n- openstack-tox-pep8 http://logs.openstack.org/56/658456/1/check/openstack-tox-pep8/ba8c533/ : SUCCESS in 3m 17s\n- openstack-tox-py27 http://logs.openstack.org/56/658456/1/check/openstack-tox-py27/5b59b3d/ : SUCCESS in 3m 04s\n- openstack-tox-py36 http://logs.openstack.org/56/658456/1/check/openstack-tox-py36/faea6ee/ : SUCCESS in 3m 16s\n- openstack-tox-py37 http://logs.openstack.org/56/658456/1/check/openstack-tox-py37/1fdd23e/ : SUCCESS in 3m 29s\n- openstack-tox-docs http://logs.openstack.org/56/658456/1/check/openstack-tox-docs/83f12a3/html/ : SUCCESS in 2m 21s\n- kolla-build-ubuntu-binary http://logs.openstack.org/56/658456/1/check/kolla-build-ubuntu-binary/1cb633e/ : SUCCESS in 32m 36s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/56/658456/1/check/kolla-build-ubuntu-source/c6b7f83/ : SUCCESS in 37m 15s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/56/658456/1/check/kolla-ansible-ubuntu-source/354edb8/ : SUCCESS in 52m 34s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/56/658456/1/check/kolla-ansible-ubuntu-binary/5f82733/ : SUCCESS in 47m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/56/658456/1/check/kolla-ansible-ubuntu-source-upgrade/26589da/ : SUCCESS in 1h 00m 43s","accounts_in_message":[],"_revision_number":1},{"id":"71b31f4a899211e27c645424500b621654be6882","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-13 17:48:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"32909ac994d7f92901b9a5c1387a96db315908d2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-04 15:19:16.000000000","message":"Abandoned\n\ncleaning old patches","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3ab8bef202255385ae0b1b80f8966a942300ea5","revisions":{"e3ab8bef202255385ae0b1b80f8966a942300ea5":{"kind":"REWORK","_number":1,"created":"2019-05-10 19:17:41.000000000","uploader":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"ref":"refs/changes/56/658456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/658456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/658456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/658456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/658456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/658456/1"}}},"commit":{"parents":[{"commit":"e6387634c6a7170e96606b9a8ff680f08d81e801","subject":"Fix bandit failure B105 in keystone fernet","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6387634c6a7170e96606b9a8ff680f08d81e801"}]}],"author":{"name":"Radoslaw Smigielski","email":"radoslaw.smigielski@nokia.com","date":"2019-05-10 13:39:50.000000000","tz":120},"committer":{"name":"Radoslaw Smigielski","email":"radoslaw.smigielski@nokia.com","date":"2019-05-10 13:39:50.000000000","tz":120},"subject":"Add RPM GPG keys import retry","message":"Add RPM GPG keys import retry\n\nIt\u0027s has become more and more often that some zuul job fails on downloading\nRPM GPG key from external source.\nThis change slightly modify the way how the GPG keys are imported.\nIt\u0027s a trade of between ability of \u0027rpm --import\u0027 to use file or url\nand curl which allows to retry download.\n\nChange-Id: I6b205db2f3c01c6802e637723167ffc6d3c66d17\nCloses-Bug: 1828542\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3ab8bef202255385ae0b1b80f8966a942300ea5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3ab8bef202255385ae0b1b80f8966a942300ea5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
