)]}'
{"id":"openstack%2Fkolla~659422","triplet_id":"openstack%2Fkolla~master~Idebf5c8756a0d0225adcce29b45a399120bb5eb6","project":"openstack/kolla","branch":"master","topic":"bp/qinling-support","hashtags":[],"change_id":"Idebf5c8756a0d0225adcce29b45a399120bb5eb6","subject":"Add Qinling containers","status":"MERGED","created":"2019-05-16 02:44:49.000000000","updated":"2019-05-21 07:46:44.000000000","submitted":"2019-05-21 07:46:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659422-1558424804127-38c90c1d","meta_rev_id":"11ed7a8a09ecb54c7acf1e92738ec916bb18a617","_number":659422,"virtual_id_number":659422,"owner":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"value":0,"_account_id":18665,"name":"Pierre Blanc","email":"pierreblanc75@gmail.com","username":"pblanc"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-05-21 07:46:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"date":"2019-05-20 14:33:10.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":0,"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},{"value":2,"date":"2019-05-21 06:45:23.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"value":0,"_account_id":18665,"name":"Pierre Blanc","email":"pierreblanc75@gmail.com","username":"pblanc"},{"value":2,"date":"2019-05-20 14:04:54.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":0,"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},{"value":1,"date":"2019-05-21 06:45:23.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"value":0,"_account_id":18665,"name":"Pierre Blanc","email":"pierreblanc75@gmail.com","username":"pblanc"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"value":0,"_account_id":18665,"name":"Pierre Blanc","email":"pierreblanc75@gmail.com","username":"pblanc"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"value":0,"_account_id":18665,"name":"Pierre Blanc","email":"pierreblanc75@gmail.com","username":"pblanc"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"_account_id":18665,"name":"Pierre Blanc","email":"pierreblanc75@gmail.com","username":"pblanc"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 15:43:35.000000000","updated_by":{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},"reviewer":{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},"state":"REVIEWER"},{"updated":"2019-05-17 15:56:13.000000000","updated_by":{"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},"reviewer":{"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},"state":"REVIEWER"},{"updated":"2019-05-17 16:23:01.000000000","updated_by":{"_account_id":18665,"name":"Pierre Blanc","email":"pierreblanc75@gmail.com","username":"pblanc"},"reviewer":{"_account_id":18665,"name":"Pierre Blanc","email":"pierreblanc75@gmail.com","username":"pblanc"},"state":"REVIEWER"},{"updated":"2019-05-18 04:24:45.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-05-20 14:04:54.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-05-20 14:33:10.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-21 06:45:23.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-05-21 07:46:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4db4c75ae8c5bb31e4c33adf448afcdcaee88da7","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-05-16 02:44:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d853876bd0aa581da9fcda7aba19a2f5f5c93c3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-16 03:40:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/22/659422/1/openstack-check/tripleo-build-containers-fedora-28-master/ec9b8ab/ : SUCCESS in 33m 05s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/22/659422/1/openstack-check/tripleo-build-containers-fedora-28-stein/c3d87bc/ : SUCCESS in 40m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f75d8c9611b7e8a024539ec90306629bf548f36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 04:06:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/22/659422/1/check/test-release-openstack/779b0e6/ : SUCCESS in 2m 41s\n- kolla-build-centos-binary http://logs.openstack.org/22/659422/1/check/kolla-build-centos-binary/0c88168/ : FAILURE in 4m 03s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/22/659422/1/check/kolla-build-centos-source/7b465c7/ : FAILURE in 51m 57s\n- kolla-ansible-centos-source http://logs.openstack.org/22/659422/1/check/kolla-ansible-centos-source/db99781/ : SUCCESS in 1h 15m 19s\n- kolla-ansible-centos-binary http://logs.openstack.org/22/659422/1/check/kolla-ansible-centos-binary/62eeca3/ : FAILURE in 18m 56s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/22/659422/1/check/kolla-ansible-centos-source-upgrade/2e0220e/ : SUCCESS in 1h 15m 11s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/22/659422/1/check/tripleo-build-containers-centos-7/bfb3abe/ : FAILURE in 29m 32s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/22/659422/1/check/kolla-build-oraclelinux-binary/63c188e/ : FAILURE in 35m 51s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/22/659422/1/check/kolla-build-oraclelinux-source/f745fe9/ : FAILURE in 57m 23s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/22/659422/1/check/kolla-ansible-oraclelinux-source/86cb173/ : SUCCESS in 1h 09m 52s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/22/659422/1/check/kolla-ansible-oraclelinux-binary/2cced50/ : FAILURE in 21m 28s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/22/659422/1/check/openstack-tox-lower-constraints/ebd0919/ : SUCCESS in 4m 01s\n- openstack-tox-pep8 http://logs.openstack.org/22/659422/1/check/openstack-tox-pep8/942ae98/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/22/659422/1/check/openstack-tox-py27/eef7a99/ : SUCCESS in 4m 14s\n- openstack-tox-py36 http://logs.openstack.org/22/659422/1/check/openstack-tox-py36/b3eb7a2/ : SUCCESS in 3m 29s\n- openstack-tox-py37 http://logs.openstack.org/22/659422/1/check/openstack-tox-py37/fc85b9c/ : SUCCESS in 3m 39s\n- openstack-tox-docs http://logs.openstack.org/22/659422/1/check/openstack-tox-docs/d88c3d3/html/ : SUCCESS in 2m 43s\n- build-openstack-releasenotes http://logs.openstack.org/22/659422/1/check/build-openstack-releasenotes/1f1ac5a/html/ : SUCCESS in 3m 45s\n- kolla-build-ubuntu-binary http://logs.openstack.org/22/659422/1/check/kolla-build-ubuntu-binary/36923fe/ : SUCCESS in 28m 29s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/22/659422/1/check/kolla-build-ubuntu-source/bd31e30/ : FAILURE in 32m 35s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/22/659422/1/check/kolla-ansible-ubuntu-source/a610809/ : SUCCESS in 49m 45s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/22/659422/1/check/kolla-ansible-ubuntu-binary/e78d662/ : SUCCESS in 53m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/22/659422/1/check/kolla-ansible-ubuntu-source-upgrade/731bfad/ : SUCCESS in 1h 04m 44s","accounts_in_message":[],"_revision_number":1},{"id":"59d35a62cd3dd6e5892c163201012c5e0d30121d","author":{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},"date":"2019-05-16 12:42:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8dd54a7355d8c35e2818ab6978fd13061ba8445f","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-05-16 13:13:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5978121292b2f8c6a0383d180f9e606e25bdc21b","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-05-16 13:13:49.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n\nGood catch, thanks!","accounts_in_message":[],"_revision_number":2},{"id":"dc3a553e0f61b849f8b2947593d9efaec2b61dc2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 14:31:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/22/659422/2/check/test-release-openstack/0b8326b/ : SUCCESS in 2m 20s\n- kolla-build-centos-binary http://logs.openstack.org/22/659422/2/check/kolla-build-centos-binary/f5bb121/ : FAILURE in 31m 01s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/22/659422/2/check/kolla-build-centos-source/f617dd6/ : SUCCESS in 41m 22s\n- kolla-ansible-centos-source http://logs.openstack.org/22/659422/2/check/kolla-ansible-centos-source/ccf1f5d/ : SUCCESS in 1h 06m 30s\n- kolla-ansible-centos-binary http://logs.openstack.org/22/659422/2/check/kolla-ansible-centos-binary/67e0de7/ : FAILURE in 19m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/22/659422/2/check/kolla-ansible-centos-source-upgrade/4fbee0f/ : SUCCESS in 1h 15m 40s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/22/659422/2/check/tripleo-build-containers-centos-7/a4f29e2/ : FAILURE in 29m 22s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/22/659422/2/check/kolla-build-oraclelinux-binary/7dcc998/ : FAILURE in 35m 06s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/22/659422/2/check/kolla-build-oraclelinux-source/96217d7/ : SUCCESS in 41m 56s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/22/659422/2/check/kolla-ansible-oraclelinux-source/115e825/ : SUCCESS in 1h 15m 01s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/22/659422/2/check/kolla-ansible-oraclelinux-binary/13089b2/ : FAILURE in 23m 51s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/22/659422/2/check/openstack-tox-lower-constraints/ca285b5/ : SUCCESS in 5m 16s\n- openstack-tox-pep8 http://logs.openstack.org/22/659422/2/check/openstack-tox-pep8/d15364c/ : SUCCESS in 3m 20s\n- openstack-tox-py27 http://logs.openstack.org/22/659422/2/check/openstack-tox-py27/de7d51f/ : SUCCESS in 4m 01s\n- openstack-tox-py36 http://logs.openstack.org/22/659422/2/check/openstack-tox-py36/c21ead1/ : SUCCESS in 3m 23s\n- openstack-tox-py37 http://logs.openstack.org/22/659422/2/check/openstack-tox-py37/251d4d1/ : SUCCESS in 4m 20s\n- openstack-tox-docs http://logs.openstack.org/22/659422/2/check/openstack-tox-docs/dd7143d/html/ : SUCCESS in 3m 19s\n- build-openstack-releasenotes http://logs.openstack.org/22/659422/2/check/build-openstack-releasenotes/dff2e99/html/ : SUCCESS in 3m 55s\n- kolla-build-ubuntu-binary http://logs.openstack.org/22/659422/2/check/kolla-build-ubuntu-binary/83e517c/ : SUCCESS in 36m 57s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/22/659422/2/check/kolla-build-ubuntu-source/6683ea5/ : SUCCESS in 31m 19s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/22/659422/2/check/kolla-ansible-ubuntu-source/bafc4d3/ : SUCCESS in 42m 15s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/22/659422/2/check/kolla-ansible-ubuntu-binary/904da83/ : SUCCESS in 46m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/22/659422/2/check/kolla-ansible-ubuntu-source-upgrade/4948eef/ : SUCCESS in 1h 04m 37s","accounts_in_message":[],"_revision_number":2},{"id":"8532e4340a64bd6a77483decffeecd6fa1f27370","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-05-16 14:54:48.000000000","message":"Patch Set 2:\n\n*-binary complain about:\n\n  CRITICAL:tests.test_build:\u003e\u003e\u003e Expected image \u0027openstack-base\u0027 to succeed!","accounts_in_message":[],"_revision_number":2},{"id":"65eb931e242c70632cf8007f596b1d5cb1025af2","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-05-16 17:21:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"48466e4b8b50a6dc57e7918843ba98ff556ed377","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-16 17:56:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/22/659422/3/openstack-check/tripleo-build-containers-fedora-28-master/7f96d03/ : SUCCESS in 28m 11s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/22/659422/3/openstack-check/tripleo-build-containers-fedora-28-stein/1a0a88a/ : SUCCESS in 29m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d6242875c68652ef8dacc382b292acd885e1ad26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 19:27:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/22/659422/3/check/test-release-openstack/b5a6b9c/ : SUCCESS in 2m 01s\n- kolla-build-centos-binary http://logs.openstack.org/22/659422/3/check/kolla-build-centos-binary/d22b5c4/ : FAILURE in 47m 10s (non-voting)\n- kolla-build-centos-source kolla-build-centos-source : MERGER_FAILURE\n- kolla-ansible-centos-source http://logs.openstack.org/22/659422/3/check/kolla-ansible-centos-source/0723c71/ : SUCCESS in 1h 09m 03s\n- kolla-ansible-centos-binary http://logs.openstack.org/22/659422/3/check/kolla-ansible-centos-binary/31c4269/ : FAILURE in 33m 33s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/22/659422/3/check/kolla-ansible-centos-source-upgrade/d0cbb28/ : SUCCESS in 1h 18m 38s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/22/659422/3/check/tripleo-build-containers-centos-7/716281f/ : FAILURE in 43m 09s\n- kolla-build-oraclelinux-binary kolla-build-oraclelinux-binary : MERGER_FAILURE (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/22/659422/3/check/kolla-build-oraclelinux-source/471d69b/ : SUCCESS in 50m 19s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/22/659422/3/check/kolla-ansible-oraclelinux-source/a125562/ : SUCCESS in 1h 16m 56s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/22/659422/3/check/kolla-ansible-oraclelinux-binary/3caaa2b/ : FAILURE in 36m 42s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/22/659422/3/check/openstack-tox-lower-constraints/8bf405a/ : SUCCESS in 3m 30s\n- openstack-tox-pep8 http://logs.openstack.org/22/659422/3/check/openstack-tox-pep8/353d210/ : SUCCESS in 2m 56s\n- openstack-tox-py27 http://logs.openstack.org/22/659422/3/check/openstack-tox-py27/a8805b0/ : SUCCESS in 2m 54s\n- openstack-tox-py36 http://logs.openstack.org/22/659422/3/check/openstack-tox-py36/7df9b9c/ : SUCCESS in 3m 03s\n- openstack-tox-py37 http://logs.openstack.org/22/659422/3/check/openstack-tox-py37/ad69df2/ : SUCCESS in 3m 04s\n- openstack-tox-docs http://logs.openstack.org/22/659422/3/check/openstack-tox-docs/4abd104/html/ : SUCCESS in 2m 57s\n- build-openstack-releasenotes http://logs.openstack.org/22/659422/3/check/build-openstack-releasenotes/3ec8486/html/ : SUCCESS in 4m 15s\n- kolla-build-ubuntu-binary http://logs.openstack.org/22/659422/3/check/kolla-build-ubuntu-binary/4349ee8/ : SUCCESS in 1h 56m 41s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/22/659422/3/check/kolla-build-ubuntu-source/6c8bcc8/ : SUCCESS in 38m 58s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/22/659422/3/check/kolla-ansible-ubuntu-source/b59e188/ : SUCCESS in 54m 55s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/22/659422/3/check/kolla-ansible-ubuntu-binary/49d697a/ : SUCCESS in 51m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/22/659422/3/check/kolla-ansible-ubuntu-source-upgrade/49ecfb9/ : SUCCESS in 1h 11m 40s","accounts_in_message":[],"_revision_number":3},{"id":"b4bf3ee4587702ba657da3d3ae5a72b3210204d7","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-05-16 20:15:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8cca7cf7c9843d0563016a876e453b53ae5f0fe2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-16 21:14:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/22/659422/4/openstack-check/tripleo-build-containers-fedora-28-master/d6a8bb3/ : SUCCESS in 33m 10s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/22/659422/4/openstack-check/tripleo-build-containers-fedora-28-stein/5cd3f35/ : SUCCESS in 31m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5cb851781534eb4f9677c4927c091dc04fc00bdc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 21:42:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/22/659422/4/check/test-release-openstack/7a7d97e/ : SUCCESS in 2m 12s\n- kolla-build-centos-binary http://logs.openstack.org/22/659422/4/check/kolla-build-centos-binary/09bd97b/ : SUCCESS in 43m 42s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/22/659422/4/check/kolla-build-centos-source/bc3ec1f/ : SUCCESS in 34m 50s\n- kolla-ansible-centos-source http://logs.openstack.org/22/659422/4/check/kolla-ansible-centos-source/eac85af/ : SUCCESS in 1h 07m 58s\n- kolla-ansible-centos-binary http://logs.openstack.org/22/659422/4/check/kolla-ansible-centos-binary/235348d/ : SUCCESS in 1h 05m 38s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/22/659422/4/check/kolla-ansible-centos-source-upgrade/14f595b/ : SUCCESS in 1h 20m 00s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/22/659422/4/check/tripleo-build-containers-centos-7/b85afbc/ : SUCCESS in 42m 05s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/22/659422/4/check/kolla-build-oraclelinux-binary/0704b69/ : SUCCESS in 43m 50s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/22/659422/4/check/kolla-build-oraclelinux-source/f7bf16a/ : SUCCESS in 54m 26s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/22/659422/4/check/kolla-ansible-oraclelinux-source/cfd343d/ : SUCCESS in 1h 17m 00s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/22/659422/4/check/kolla-ansible-oraclelinux-binary/1f505ca/ : SUCCESS in 1h 08m 21s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/22/659422/4/check/openstack-tox-lower-constraints/1520bdc/ : SUCCESS in 5m 39s\n- openstack-tox-pep8 http://logs.openstack.org/22/659422/4/check/openstack-tox-pep8/c992bb5/ : SUCCESS in 3m 02s\n- openstack-tox-py27 http://logs.openstack.org/22/659422/4/check/openstack-tox-py27/b50a2cd/ : SUCCESS in 3m 14s\n- openstack-tox-py36 http://logs.openstack.org/22/659422/4/check/openstack-tox-py36/a2b9903/ : SUCCESS in 2m 56s\n- openstack-tox-py37 http://logs.openstack.org/22/659422/4/check/openstack-tox-py37/5670775/ : SUCCESS in 6m 03s\n- openstack-tox-docs http://logs.openstack.org/22/659422/4/check/openstack-tox-docs/6818635/html/ : SUCCESS in 2m 38s\n- build-openstack-releasenotes http://logs.openstack.org/22/659422/4/check/build-openstack-releasenotes/dd8a19b/html/ : SUCCESS in 3m 50s\n- kolla-build-ubuntu-binary http://logs.openstack.org/22/659422/4/check/kolla-build-ubuntu-binary/1f3c96d/ : SUCCESS in 29m 09s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/22/659422/4/check/kolla-build-ubuntu-source/d48e9a8/ : SUCCESS in 35m 55s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/22/659422/4/check/kolla-ansible-ubuntu-source/cf7a394/ : SUCCESS in 57m 01s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/22/659422/4/check/kolla-ansible-ubuntu-binary/07d792a/ : SUCCESS in 55m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/22/659422/4/check/kolla-ansible-ubuntu-source-upgrade/da9f649/ : SUCCESS in 1h 01m 08s","accounts_in_message":[],"_revision_number":4},{"id":"b9982237c7addb826b6fd9b074b98d547846f934","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-05-17 13:45:27.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"828012727bc3b579d0b80336d06d3847605738c5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-17 14:15:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/22/659422/5/openstack-check/tripleo-build-containers-fedora-28-master/e3916e5/ : SUCCESS in 26m 44s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/22/659422/5/openstack-check/tripleo-build-containers-fedora-28-stein/af50493/ : SUCCESS in 27m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c76ac9aa16a4056c1f1da0b5f91480233083e06a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 15:37:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/22/659422/5/check/test-release-openstack/4c095fb/ : SUCCESS in 1m 38s\n- kolla-build-centos-binary http://logs.openstack.org/22/659422/5/check/kolla-build-centos-binary/368271a/ : SUCCESS in 41m 42s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/22/659422/5/check/kolla-build-centos-source/c0f38ec/ : SUCCESS in 40m 22s\n- kolla-ansible-centos-source http://logs.openstack.org/22/659422/5/check/kolla-ansible-centos-source/c688b90/ : SUCCESS in 1h 05m 13s\n- kolla-ansible-centos-binary http://logs.openstack.org/22/659422/5/check/kolla-ansible-centos-binary/6d251e5/ : SUCCESS in 1h 12m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/22/659422/5/check/kolla-ansible-centos-source-upgrade/91f4efd/ : SUCCESS in 1h 32m 00s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/22/659422/5/check/tripleo-build-containers-centos-7/ef802e0/ : SUCCESS in 38m 40s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/22/659422/5/check/kolla-build-oraclelinux-binary/650e739/ : FAILURE in 15m 08s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/22/659422/5/check/kolla-build-oraclelinux-source/9ba83e4/ : SUCCESS in 58m 23s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/22/659422/5/check/kolla-ansible-oraclelinux-source/2a67f86/ : SUCCESS in 1h 28m 01s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/22/659422/5/check/kolla-ansible-oraclelinux-binary/0ba4bdc/ : SUCCESS in 1h 13m 40s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/22/659422/5/check/openstack-tox-lower-constraints/fa02791/ : SUCCESS in 3m 43s\n- openstack-tox-pep8 http://logs.openstack.org/22/659422/5/check/openstack-tox-pep8/596faaa/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/22/659422/5/check/openstack-tox-py27/70d35ee/ : SUCCESS in 3m 03s\n- openstack-tox-py36 http://logs.openstack.org/22/659422/5/check/openstack-tox-py36/0305218/ : SUCCESS in 3m 12s\n- openstack-tox-py37 http://logs.openstack.org/22/659422/5/check/openstack-tox-py37/6367ce0/ : SUCCESS in 3m 27s\n- openstack-tox-docs http://logs.openstack.org/22/659422/5/check/openstack-tox-docs/aaa9a8d/html/ : SUCCESS in 2m 45s\n- build-openstack-releasenotes http://logs.openstack.org/22/659422/5/check/build-openstack-releasenotes/2fda44d/html/ : SUCCESS in 3m 45s\n- kolla-build-ubuntu-binary http://logs.openstack.org/22/659422/5/check/kolla-build-ubuntu-binary/ed8be25/ : SUCCESS in 29m 37s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/22/659422/5/check/kolla-build-ubuntu-source/f6c09f9/ : SUCCESS in 38m 09s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/22/659422/5/check/kolla-ansible-ubuntu-source/253d8e8/ : SUCCESS in 52m 34s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/22/659422/5/check/kolla-ansible-ubuntu-binary/86d6d55/ : SUCCESS in 1h 03m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/22/659422/5/check/kolla-ansible-ubuntu-source-upgrade/7924b67/ : SUCCESS in 1h 14m 28s","accounts_in_message":[],"_revision_number":5},{"id":"7ff2cbd6a559d89e2d5b10bbc180a841514fd0c5","author":{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},"date":"2019-05-17 15:43:35.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"e424e3ee89467cdb3847cc4df512766924cc859c","author":{"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},"date":"2019-05-17 15:56:13.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"c642d634a2fd6040a519bd37df9b6dd9e148aae9","author":{"_account_id":18665,"name":"Pierre Blanc","email":"pierreblanc75@gmail.com","username":"pblanc"},"date":"2019-05-17 16:23:01.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"645a36d4b66c50c6e8dd031227cab4f5cdbd9463","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-20 11:46:35.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"297a80bd49c795ce7ed9308f10ae4c4c4146e9f0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-20 11:46:53.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nMade a small edit to the release note.","accounts_in_message":[],"_revision_number":6},{"id":"fceda48049d5e2b91b99516b89786e3a969b5b4f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-20 12:21:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/22/659422/6/openstack-check/tripleo-build-containers-fedora-28-master/a7505aa/ : SUCCESS in 31m 16s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/22/659422/6/openstack-check/tripleo-build-containers-fedora-28-stein/8b3e7bc/ : SUCCESS in 30m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"962f35e15f8b8fd792230f52bda09072141f7202","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 13:28:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/22/659422/6/check/test-release-openstack/df091e4/ : SUCCESS in 3m 00s\n- kolla-build-centos-binary http://logs.openstack.org/22/659422/6/check/kolla-build-centos-binary/b6b1e99/ : SUCCESS in 40m 22s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/22/659422/6/check/kolla-build-centos-source/db081be/ : SUCCESS in 54m 17s\n- kolla-ansible-centos-source http://logs.openstack.org/22/659422/6/check/kolla-ansible-centos-source/29a7fb4/ : SUCCESS in 1h 11m 03s\n- kolla-ansible-centos-binary http://logs.openstack.org/22/659422/6/check/kolla-ansible-centos-binary/820fb37/ : SUCCESS in 1h 13m 05s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/22/659422/6/check/kolla-ansible-centos-source-upgrade/1a691e1/ : SUCCESS in 1h 29m 15s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/22/659422/6/check/tripleo-build-containers-centos-7/231219b/ : SUCCESS in 46m 23s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/22/659422/6/check/kolla-build-oraclelinux-binary/5be4b56/ : SUCCESS in 59m 51s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/22/659422/6/check/kolla-build-oraclelinux-source/0819e47/ : SUCCESS in 56m 10s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/22/659422/6/check/kolla-ansible-oraclelinux-source/2d3014a/ : SUCCESS in 1h 11m 56s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/22/659422/6/check/kolla-ansible-oraclelinux-binary/c452d13/ : SUCCESS in 1h 11m 32s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/22/659422/6/check/openstack-tox-lower-constraints/54f80eb/ : SUCCESS in 4m 19s\n- openstack-tox-pep8 http://logs.openstack.org/22/659422/6/check/openstack-tox-pep8/478f253/ : SUCCESS in 3m 18s\n- openstack-tox-py27 http://logs.openstack.org/22/659422/6/check/openstack-tox-py27/4d3fa83/ : SUCCESS in 3m 35s\n- openstack-tox-py36 http://logs.openstack.org/22/659422/6/check/openstack-tox-py36/cffd7dd/ : SUCCESS in 4m 36s\n- openstack-tox-py37 http://logs.openstack.org/22/659422/6/check/openstack-tox-py37/792f9eb/ : SUCCESS in 3m 46s\n- openstack-tox-docs http://logs.openstack.org/22/659422/6/check/openstack-tox-docs/1db8aea/html/ : SUCCESS in 2m 38s\n- build-openstack-releasenotes http://logs.openstack.org/22/659422/6/check/build-openstack-releasenotes/f621658/html/ : SUCCESS in 3m 27s\n- kolla-build-ubuntu-binary http://logs.openstack.org/22/659422/6/check/kolla-build-ubuntu-binary/f2e6ac4/ : SUCCESS in 26m 25s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/22/659422/6/check/kolla-build-ubuntu-source/136c2d8/ : SUCCESS in 31m 13s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/22/659422/6/check/kolla-ansible-ubuntu-source/90f418e/ : SUCCESS in 53m 56s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/22/659422/6/check/kolla-ansible-ubuntu-binary/ccb6c75/ : SUCCESS in 49m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/22/659422/6/check/kolla-ansible-ubuntu-source-upgrade/297fe77/ : SUCCESS in 1h 01m 53s","accounts_in_message":[],"_revision_number":6},{"id":"88638811fd062bc026b3321a27fc8697fc78a880","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-05-20 13:53:38.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"992c96f0df8969e8d25d48ba5ddc01de5ce13265","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-05-20 13:54:53.000000000","message":"Patch Set 7:\n\n@Mark, you are right.\nI removed the python-qinlingclient from openstack-base Docker image.","accounts_in_message":[],"_revision_number":7},{"id":"40d02a1ee194fe330e8582376a8b7933001785c6","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-05-20 13:56:10.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"cf0582db2749d8d3488e113b22a275fa589d8bb9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-20 14:04:54.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"ac6cf7ec996132cad8ec0b253db0eb40ee72f039","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-20 14:33:10.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/22/659422/8/openstack-check/tripleo-build-containers-fedora-28-master/a365430/ : SUCCESS in 31m 00s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/22/659422/8/openstack-check/tripleo-build-containers-fedora-28-stein/7a0c45c/ : SUCCESS in 35m 20s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3dfd53b73b68a33903302852e0fe1c4cb8ae5a45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 15:54:10.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/22/659422/8/check/test-release-openstack/6fb3304/ : SUCCESS in 2m 19s\n- kolla-build-centos-binary http://logs.openstack.org/22/659422/8/check/kolla-build-centos-binary/5ee49d4/ : SUCCESS in 45m 02s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/22/659422/8/check/kolla-build-centos-source/aa8250d/ : SUCCESS in 48m 59s\n- kolla-ansible-centos-source http://logs.openstack.org/22/659422/8/check/kolla-ansible-centos-source/a3eb295/ : SUCCESS in 1h 09m 59s\n- kolla-ansible-centos-binary http://logs.openstack.org/22/659422/8/check/kolla-ansible-centos-binary/1e7e64f/ : SUCCESS in 1h 06m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/22/659422/8/check/kolla-ansible-centos-source-upgrade/f597d02/ : SUCCESS in 1h 17m 43s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/22/659422/8/check/tripleo-build-containers-centos-7/e5b389b/ : SUCCESS in 40m 14s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/22/659422/8/check/kolla-build-oraclelinux-binary/1d0b5e0/ : SUCCESS in 51m 58s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/22/659422/8/check/kolla-build-oraclelinux-source/b836ce3/ : SUCCESS in 55m 14s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/22/659422/8/check/kolla-ansible-oraclelinux-source/df9ac25/ : SUCCESS in 1h 33m 40s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/22/659422/8/check/kolla-ansible-oraclelinux-binary/0fd3057/ : SUCCESS in 1h 15m 41s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/22/659422/8/check/openstack-tox-lower-constraints/66e62ab/ : SUCCESS in 3m 53s\n- openstack-tox-pep8 http://logs.openstack.org/22/659422/8/check/openstack-tox-pep8/76a5b83/ : SUCCESS in 3m 21s\n- openstack-tox-py27 http://logs.openstack.org/22/659422/8/check/openstack-tox-py27/381166b/ : SUCCESS in 3m 10s\n- openstack-tox-py36 http://logs.openstack.org/22/659422/8/check/openstack-tox-py36/660ee68/ : SUCCESS in 3m 08s\n- openstack-tox-py37 http://logs.openstack.org/22/659422/8/check/openstack-tox-py37/ac3b548/ : SUCCESS in 2m 56s\n- openstack-tox-docs http://logs.openstack.org/22/659422/8/check/openstack-tox-docs/f19eb7f/html/ : SUCCESS in 2m 48s\n- build-openstack-releasenotes http://logs.openstack.org/22/659422/8/check/build-openstack-releasenotes/03043f8/html/ : SUCCESS in 3m 44s\n- kolla-build-ubuntu-binary http://logs.openstack.org/22/659422/8/check/kolla-build-ubuntu-binary/94d5c20/ : SUCCESS in 30m 34s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/22/659422/8/check/kolla-build-ubuntu-source/2b94c75/ : SUCCESS in 35m 38s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/22/659422/8/check/kolla-ansible-ubuntu-source/802da25/ : SUCCESS in 58m 59s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/22/659422/8/check/kolla-ansible-ubuntu-binary/f4523a5/ : SUCCESS in 46m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/22/659422/8/check/kolla-ansible-ubuntu-source-upgrade/3bdcc29/ : SUCCESS in 1h 06m 11s","accounts_in_message":[],"_revision_number":8},{"id":"28d4d8386c0c43ba31359ee6ecd2a374cac5c322","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-05-21 06:45:23.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"0b883b22bd7c9c12c2d6f4e7c629c6a5cfa1d8b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 06:45:35.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"6b3281333a441b4e701ffcee9b025a63c6d3b2f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 07:46:43.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/22/659422/8/gate/test-release-openstack/7df18a2/ : SUCCESS in 2m 22s\n- kolla-build-centos-source http://logs.openstack.org/22/659422/8/gate/kolla-build-centos-source/d0ed1f6/ : SUCCESS in 50m 58s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/22/659422/8/gate/tripleo-build-containers-centos-7/9614a1d/ : SUCCESS in 50m 44s\n- kolla-build-oraclelinux-source http://logs.openstack.org/22/659422/8/gate/kolla-build-oraclelinux-source/60575bb/ : SUCCESS in 55m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/22/659422/8/gate/openstack-tox-lower-constraints/29974e5/ : SUCCESS in 4m 05s\n- openstack-tox-pep8 http://logs.openstack.org/22/659422/8/gate/openstack-tox-pep8/ede8d08/ : SUCCESS in 2m 57s\n- openstack-tox-py27 http://logs.openstack.org/22/659422/8/gate/openstack-tox-py27/b8aec1b/ : SUCCESS in 2m 34s\n- openstack-tox-py36 http://logs.openstack.org/22/659422/8/gate/openstack-tox-py36/f0f0448/ : SUCCESS in 2m 36s\n- openstack-tox-py37 http://logs.openstack.org/22/659422/8/gate/openstack-tox-py37/8e792ed/ : SUCCESS in 4m 27s\n- openstack-tox-docs http://logs.openstack.org/22/659422/8/gate/openstack-tox-docs/de82a9e/html/ : SUCCESS in 3m 23s\n- build-openstack-releasenotes http://logs.openstack.org/22/659422/8/gate/build-openstack-releasenotes/0f0c871/html/ : SUCCESS in 3m 46s\n- kolla-build-ubuntu-source http://logs.openstack.org/22/659422/8/gate/kolla-build-ubuntu-source/9e6ef8b/ : SUCCESS in 31m 09s","accounts_in_message":[],"_revision_number":8},{"id":"019a0de1289d44babf037d0d405d880ba291c0c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 07:46:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"dbdf927d391904252464ab401cb5724e10a45a72","revisions":{"ac8bcfe725fa056b26c84158da3a7abe2d5d584a":{"kind":"REWORK","_number":1,"created":"2019-05-16 02:44:49.000000000","uploader":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"ref":"refs/changes/22/659422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/659422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/659422/1"}}},"commit":{"parents":[{"commit":"b55fa2f8b70b143068db57fc1ff844929d29486e","subject":"Merge \"Update Python 3 test runtimes for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b55fa2f8b70b143068db57fc1ff844929d29486e"}]}],"author":{"name":"Gaetan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-05-16 02:43:11.000000000","tz":-240},"committer":{"name":"Gaetan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-05-16 02:44:48.000000000","tz":-240},"subject":"Add Qinling containers","message":"Add Qinling containers\n\nAdd Qinling Dockerfile , Qinling is an OpenStack project\nto provide \"Function as a service\". This project aims to\nprovide a platform to support serverless functions.\n\nChange-Id: Idebf5c8756a0d0225adcce29b45a399120bb5eb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac8bcfe725fa056b26c84158da3a7abe2d5d584a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac8bcfe725fa056b26c84158da3a7abe2d5d584a"}]},"branch":"refs/heads/master"},"dfc5f3bad904c55b0d2bb31f032ee3f354efe380":{"kind":"REWORK","_number":2,"created":"2019-05-16 13:13:24.000000000","uploader":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"ref":"refs/changes/22/659422/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/659422/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/659422/2"}}},"commit":{"parents":[{"commit":"b55fa2f8b70b143068db57fc1ff844929d29486e","subject":"Merge \"Update Python 3 test runtimes for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b55fa2f8b70b143068db57fc1ff844929d29486e"}]}],"author":{"name":"Gaetan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-05-16 02:43:11.000000000","tz":-240},"committer":{"name":"Gaetan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-05-16 13:13:11.000000000","tz":-240},"subject":"Add Qinling containers","message":"Add Qinling containers\n\nAdd Qinling Dockerfile , Qinling is an OpenStack project\nto provide \"Function as a service\". This project aims to\nprovide a platform to support serverless functions.\n\nChange-Id: Idebf5c8756a0d0225adcce29b45a399120bb5eb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dfc5f3bad904c55b0d2bb31f032ee3f354efe380"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dfc5f3bad904c55b0d2bb31f032ee3f354efe380"}]},"branch":"refs/heads/master"},"5a9bc3c6a48bb1a275d6b1ab247bd14c58fb062f":{"kind":"REWORK","_number":3,"created":"2019-05-16 17:21:42.000000000","uploader":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"ref":"refs/changes/22/659422/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/659422/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/659422/3"}}},"commit":{"parents":[{"commit":"b55fa2f8b70b143068db57fc1ff844929d29486e","subject":"Merge \"Update Python 3 test runtimes for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b55fa2f8b70b143068db57fc1ff844929d29486e"}]}],"author":{"name":"Gaetan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-05-16 02:43:11.000000000","tz":-240},"committer":{"name":"Gaetan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-05-16 17:21:02.000000000","tz":-240},"subject":"Add Qinling containers","message":"Add Qinling containers\n\nAdd Qinling Dockerfile , Qinling is an OpenStack project\nto provide \"Function as a service\". This project aims to\nprovide a platform to support serverless functions.\n\nChange-Id: Idebf5c8756a0d0225adcce29b45a399120bb5eb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a9bc3c6a48bb1a275d6b1ab247bd14c58fb062f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a9bc3c6a48bb1a275d6b1ab247bd14c58fb062f"}]},"branch":"refs/heads/master"},"38ca78082153deeb6b9e07960b93318b44db2c3c":{"kind":"REWORK","_number":4,"created":"2019-05-16 20:15:03.000000000","uploader":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"ref":"refs/changes/22/659422/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/659422/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/659422/4"}}},"commit":{"parents":[{"commit":"b55fa2f8b70b143068db57fc1ff844929d29486e","subject":"Merge \"Update Python 3 test runtimes for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b55fa2f8b70b143068db57fc1ff844929d29486e"}]}],"author":{"name":"Gaetan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-05-16 02:43:11.000000000","tz":-240},"committer":{"name":"Gaetan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-05-16 20:14:57.000000000","tz":-240},"subject":"Add Qinling containers","message":"Add Qinling containers\n\nAdd Qinling Dockerfile , Qinling is an OpenStack project\nto provide \"Function as a service\". This project aims to\nprovide a platform to support serverless functions.\n\nChange-Id: Idebf5c8756a0d0225adcce29b45a399120bb5eb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38ca78082153deeb6b9e07960b93318b44db2c3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38ca78082153deeb6b9e07960b93318b44db2c3c"}]},"branch":"refs/heads/master"},"aab5e4cd5ab8b9ad8fb4499bfec2c59958e7f969":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-05-17 13:45:27.000000000","uploader":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"ref":"refs/changes/22/659422/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/659422/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/659422/5"}}},"commit":{"parents":[{"commit":"b55fa2f8b70b143068db57fc1ff844929d29486e","subject":"Merge \"Update Python 3 test runtimes for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b55fa2f8b70b143068db57fc1ff844929d29486e"}]}],"author":{"name":"Gaetan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-05-16 02:43:11.000000000","tz":-240},"committer":{"name":"Gaetan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-05-17 13:44:16.000000000","tz":-240},"subject":"Add Qinling containers","message":"Add Qinling containers\n\nAdd Qinling Dockerfile , Qinling is an OpenStack project\nto provide \"Function as a service\". This project aims to\nprovide a platform to support serverless functions.\n\nChange-Id: Idebf5c8756a0d0225adcce29b45a399120bb5eb6\nImplements: blueprint qinling-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aab5e4cd5ab8b9ad8fb4499bfec2c59958e7f969"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aab5e4cd5ab8b9ad8fb4499bfec2c59958e7f969"}]},"branch":"refs/heads/master"},"a825d156a57822fd4e5b0bd255fee1fb082dd457":{"kind":"REWORK","_number":6,"created":"2019-05-20 11:46:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/22/659422/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/659422/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/659422/6"}}},"commit":{"parents":[{"commit":"b55fa2f8b70b143068db57fc1ff844929d29486e","subject":"Merge \"Update Python 3 test runtimes for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b55fa2f8b70b143068db57fc1ff844929d29486e"}]}],"author":{"name":"Gaetan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-05-16 02:43:11.000000000","tz":-240},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-20 11:46:31.000000000","tz":0},"subject":"Add Qinling containers","message":"Add Qinling containers\n\nAdd Qinling Dockerfile , Qinling is an OpenStack project\nto provide \"Function as a service\". This project aims to\nprovide a platform to support serverless functions.\n\nChange-Id: Idebf5c8756a0d0225adcce29b45a399120bb5eb6\nImplements: blueprint qinling-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a825d156a57822fd4e5b0bd255fee1fb082dd457"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a825d156a57822fd4e5b0bd255fee1fb082dd457"}]},"branch":"refs/heads/master"},"f68ce7dd9480b6db23ec4ef7e84805bb29a0c558":{"kind":"REWORK","_number":7,"created":"2019-05-20 13:53:38.000000000","uploader":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"ref":"refs/changes/22/659422/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/659422/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/659422/7"}}},"commit":{"parents":[{"commit":"b55fa2f8b70b143068db57fc1ff844929d29486e","subject":"Merge \"Update Python 3 test runtimes for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b55fa2f8b70b143068db57fc1ff844929d29486e"}]}],"author":{"name":"Gaetan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-05-16 02:43:11.000000000","tz":-240},"committer":{"name":"Gaetan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-05-20 13:53:27.000000000","tz":-240},"subject":"Add Qinling containers","message":"Add Qinling containers\n\nAdd Qinling Dockerfile , Qinling is an OpenStack project\nto provide \"Function as a service\". This project aims to\nprovide a platform to support serverless functions.\n\nChange-Id: Idebf5c8756a0d0225adcce29b45a399120bb5eb6\nImplements: blueprint qinling-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f68ce7dd9480b6db23ec4ef7e84805bb29a0c558"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f68ce7dd9480b6db23ec4ef7e84805bb29a0c558"}]},"branch":"refs/heads/master"},"dbdf927d391904252464ab401cb5724e10a45a72":{"kind":"REWORK","_number":8,"created":"2019-05-20 13:56:10.000000000","uploader":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"ref":"refs/changes/22/659422/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/659422/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/659422/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/659422/8"}}},"commit":{"parents":[{"commit":"b55fa2f8b70b143068db57fc1ff844929d29486e","subject":"Merge \"Update Python 3 test runtimes for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b55fa2f8b70b143068db57fc1ff844929d29486e"}]}],"author":{"name":"Gaetan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-05-16 02:43:11.000000000","tz":-240},"committer":{"name":"Gaetan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-05-20 13:56:03.000000000","tz":-240},"subject":"Add Qinling containers","message":"Add Qinling containers\n\nAdd Qinling Dockerfile , Qinling is an OpenStack project\nto provide \"Function as a service\". This project aims to\nprovide a platform to support serverless functions.\n\nChange-Id: Idebf5c8756a0d0225adcce29b45a399120bb5eb6\nImplements: blueprint qinling-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbdf927d391904252464ab401cb5724e10a45a72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbdf927d391904252464ab401cb5724e10a45a72"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
