)]}'
{"id":"openstack%2Fkolla~660088","triplet_id":"openstack%2Fkolla~stable%2Fqueens~If691820ea3dea3e8758819be42db60732680aad9","project":"openstack/kolla","branch":"stable/queens","topic":"bug/1828774","hashtags":[],"change_id":"If691820ea3dea3e8758819be42db60732680aad9","subject":"Install ceph-mds in ceph-base on ubuntu/debian","status":"MERGED","created":"2019-05-20 10:47:09.000000000","updated":"2019-05-22 10:14:11.000000000","submitted":"2019-05-22 10:11:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660088-1558519900330-3248f38c","meta_rev_id":"27f683a16b914d0aa90a283eac4e7a363983f78d","_number":660088,"virtual_id_number":660088,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2019-05-20 13:01:47.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-05-22 10:11:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-20 13:56:30.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-05-22 09:23:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-05-20 10:51:58.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-05-22 09:23:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 10:51:58.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-05-20 13:56:30.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-22 09:23:22.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-05-22 10:11:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ded510fb2505907f92b5b18a28ffb7c3ba4f846f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-20 10:47:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13df2a46dd14b0829c5d00c4d3d7b5b062c69a46","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-20 10:51:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e545baa1ce18a6c1e162a2392516af882231db26","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-20 11:18:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/88/660088/1/openstack-check/tripleo-build-containers-fedora-28-master/defb97a/ : RETRY_LIMIT in 2m 32s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/88/660088/1/openstack-check/tripleo-build-containers-fedora-28-stein/c7439b8/ : SUCCESS in 29m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9a2e2bd2dbc62f86da3831058b603d4c63a0f89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 12:15:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/88/660088/1/check/kolla-build-centos-binary/760f158/ : SUCCESS in 36m 13s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/88/660088/1/check/kolla-build-centos-source/c412801/ : SUCCESS in 46m 58s\n- kolla-ansible-centos-source http://logs.openstack.org/88/660088/1/check/kolla-ansible-centos-source/668e3c0/ : SUCCESS in 1h 02m 48s (non-voting)\n- kolla-ansible-centos-binary http://logs.openstack.org/88/660088/1/check/kolla-ansible-centos-binary/448cb99/ : SUCCESS in 56m 55s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/88/660088/1/check/kolla-build-oraclelinux-binary/a827848/ : SUCCESS in 46m 24s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/88/660088/1/check/kolla-build-oraclelinux-source/6052002/ : SUCCESS in 42m 37s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/88/660088/1/check/kolla-ansible-oraclelinux-source/59aecdc/ : SUCCESS in 1h 16m 51s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/88/660088/1/check/kolla-ansible-oraclelinux-binary/0ad1b6b/ : SUCCESS in 1h 00m 44s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/88/660088/1/check/openstack-tox-pep8/8a589c2/ : FAILURE in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/88/660088/1/check/openstack-tox-py27/02d3f97/ : SUCCESS in 3m 24s\n- openstack-tox-py35 http://logs.openstack.org/88/660088/1/check/openstack-tox-py35/d0ccd33/ : SUCCESS in 3m 39s\n- build-openstack-sphinx-docs http://logs.openstack.org/88/660088/1/check/build-openstack-sphinx-docs/6b11946/html/ : SUCCESS in 4m 00s\n- kolla-build-ubuntu-binary http://logs.openstack.org/88/660088/1/check/kolla-build-ubuntu-binary/91baa1f/ : SUCCESS in 22m 24s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/88/660088/1/check/kolla-build-ubuntu-source/123011a/ : SUCCESS in 26m 39s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/88/660088/1/check/kolla-ansible-ubuntu-source/2e1a62e/ : SUCCESS in 56m 08s (non-voting)\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/88/660088/1/check/kolla-ansible-ubuntu-binary/001d5fd/ : SUCCESS in 48m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"658b1b3d9fa72fe0bf65286f53cb5c93602f0c7a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-20 13:01:47.000000000","message":"Patch Set 1:\n\nrecheck\n\nBandit patch just merged","accounts_in_message":[],"_revision_number":1},{"id":"0a480475e6c979f9b9a5c9b66b28469a65e4eae5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-20 13:56:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/88/660088/1/openstack-check/tripleo-build-containers-fedora-28-master/0b73add/ : RETRY_LIMIT in 2m 35s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/88/660088/1/openstack-check/tripleo-build-containers-fedora-28-stein/b800e8c/ : SUCCESS in 30m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5db460941dbfb72c07d3c9586148d50164f383d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 14:44:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/88/660088/1/check/kolla-build-centos-binary/16acef5/ : SUCCESS in 24m 48s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/88/660088/1/check/kolla-build-centos-source/ef938ce/ : SUCCESS in 42m 54s\n- kolla-ansible-centos-source http://logs.openstack.org/88/660088/1/check/kolla-ansible-centos-source/585ee78/ : SUCCESS in 59m 58s (non-voting)\n- kolla-ansible-centos-binary http://logs.openstack.org/88/660088/1/check/kolla-ansible-centos-binary/4d84d0e/ : SUCCESS in 1h 12m 33s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/88/660088/1/check/kolla-build-oraclelinux-binary/d7f8b94/ : SUCCESS in 45m 45s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/88/660088/1/check/kolla-build-oraclelinux-source/9bcdd93/ : SUCCESS in 47m 55s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/88/660088/1/check/kolla-ansible-oraclelinux-source/45578f5/ : SUCCESS in 1h 10m 57s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/88/660088/1/check/kolla-ansible-oraclelinux-binary/8eb8736/ : SUCCESS in 1h 03m 09s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/88/660088/1/check/openstack-tox-pep8/0ec9dd3/ : SUCCESS in 4m 16s\n- openstack-tox-py27 http://logs.openstack.org/88/660088/1/check/openstack-tox-py27/b52f7af/ : SUCCESS in 4m 12s\n- openstack-tox-py35 http://logs.openstack.org/88/660088/1/check/openstack-tox-py35/364fdb9/ : SUCCESS in 4m 03s\n- build-openstack-sphinx-docs http://logs.openstack.org/88/660088/1/check/build-openstack-sphinx-docs/60f9e44/html/ : SUCCESS in 4m 18s\n- kolla-build-ubuntu-binary http://logs.openstack.org/88/660088/1/check/kolla-build-ubuntu-binary/c9ef643/ : SUCCESS in 21m 45s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/88/660088/1/check/kolla-build-ubuntu-source/1ee5aad/ : SUCCESS in 28m 00s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/88/660088/1/check/kolla-ansible-ubuntu-source/6a66d75/ : SUCCESS in 57m 48s (non-voting)\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/88/660088/1/check/kolla-ansible-ubuntu-binary/a596ab3/ : SUCCESS in 1h 00m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2d61a99fcbf65fcf1017f33a1f8e6a8ca2545efb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-05-22 09:23:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7b0d231813106db9cac6282b80ffdcae476b83e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 09:23:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1aab53a1b847a6797e75006c57e2773fcbdabf13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 10:11:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"50e0cfd79d51408ae3d37cde7eaf547e6892f0ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 10:11:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/88/660088/1/gate/kolla-build-centos-source/74eb759/ : SUCCESS in 42m 26s\n- kolla-build-oraclelinux-source http://logs.openstack.org/88/660088/1/gate/kolla-build-oraclelinux-source/b0840b8/ : SUCCESS in 42m 48s\n- openstack-tox-pep8 http://logs.openstack.org/88/660088/1/gate/openstack-tox-pep8/c83ae90/ : SUCCESS in 3m 52s\n- openstack-tox-py27 http://logs.openstack.org/88/660088/1/gate/openstack-tox-py27/939a683/ : SUCCESS in 2m 58s\n- openstack-tox-py35 http://logs.openstack.org/88/660088/1/gate/openstack-tox-py35/06b8e54/ : SUCCESS in 3m 36s\n- build-openstack-sphinx-docs http://logs.openstack.org/88/660088/1/gate/build-openstack-sphinx-docs/e441c9d/html/ : SUCCESS in 3m 36s\n- kolla-build-ubuntu-source http://logs.openstack.org/88/660088/1/gate/kolla-build-ubuntu-source/d102fbb/ : SUCCESS in 29m 07s","accounts_in_message":[],"_revision_number":1},{"id":"b50476950d34e1dcaa5b6838d11c7446e9ab6abf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-22 10:14:11.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/pike as commit 518dc6a2783585ecafcc8926d5f8eb9915a7af07","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8dd4c953fe996c47138706f7cf74c9f2e354d899","revisions":{"8dd4c953fe996c47138706f7cf74c9f2e354d899":{"kind":"REWORK","_number":1,"created":"2019-05-20 10:47:09.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/88/660088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/660088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/660088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/660088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/660088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/660088/1"}}},"commit":{"parents":[{"commit":"8dbb962494750fe9c164113616e4864d9d84ac3e","subject":"Install udevadm in ironic-conductor image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8dbb962494750fe9c164113616e4864d9d84ac3e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-20 10:44:52.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-20 10:44:52.000000000","tz":60},"subject":"Install ceph-mds in ceph-base on ubuntu/debian","message":"Install ceph-mds in ceph-base on ubuntu/debian\n\nStable branch only\n\nbackport: pike\n\nIn the queens and earlier ceph images on ubuntu, ceph-mds is not\ninstalled. At that point, ceph-mds was a separate package from ceph.\nFrom the ceph release used in Rocky onwards, ceph-mds is packaged with\nceph.\n\nChange-Id: If691820ea3dea3e8758819be42db60732680aad9\nCloses-Bug: #1828774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8dd4c953fe996c47138706f7cf74c9f2e354d899"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8dd4c953fe996c47138706f7cf74c9f2e354d899"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
