)]}'
{"id":"openstack%2Fkolla~660160","triplet_id":"openstack%2Fkolla~stable%2Fqueens~I11b76317af047af4d3d35d9979c036347b5321b9","project":"openstack/kolla","branch":"stable/queens","topic":"bump-queens","hashtags":[],"change_id":"I11b76317af047af4d3d35d9979c036347b5321b9","subject":"Bump service versions for Queens","status":"MERGED","created":"2019-05-20 16:22:54.000000000","updated":"2019-05-28 13:00:20.000000000","submitted":"2019-05-28 13:00:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660160-1559048420965-bcd83196","meta_rev_id":"2514c3eb877327cb9b64abdf0d5ce0778389c163","_number":660160,"virtual_id_number":660160,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-05-28 13:00:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-21 09:03:32.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":2,"date":"2019-05-28 11:58:30.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":2,"date":"2019-05-24 20:42:44.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":1,"date":"2019-05-28 11:58:30.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 09:03:32.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-24 20:42:44.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-05-28 11:58:30.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-05-28 13:00:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"935eaf8e398623a03d3e8d3a708f813e3eb2d793","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-20 16:22:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de8c4163ac3539bcf291dc32a2a328df84d2fc8a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-20 16:57:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/60/660160/1/openstack-check/tripleo-build-containers-fedora-28-master/2acc14f/ : RETRY_LIMIT in 4m 46s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/60/660160/1/openstack-check/tripleo-build-containers-fedora-28-stein/0d1b4c9/ : SUCCESS in 32m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"028932bbd4709ca189d827bb8f862c5c958aa549","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 17:42:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/60/660160/1/check/kolla-build-centos-binary/c22e852/ : SUCCESS in 30m 46s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/60/660160/1/check/kolla-build-centos-source/3947275/ : FAILURE in 44m 32s\n- kolla-ansible-centos-source http://logs.openstack.org/60/660160/1/check/kolla-ansible-centos-source/c4d95d9/ : SUCCESS in 1h 04m 11s (non-voting)\n- kolla-ansible-centos-binary http://logs.openstack.org/60/660160/1/check/kolla-ansible-centos-binary/145c227/ : SUCCESS in 56m 18s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/60/660160/1/check/kolla-build-oraclelinux-binary/362c2fa/ : SUCCESS in 41m 17s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/60/660160/1/check/kolla-build-oraclelinux-source/0ad8e98/ : FAILURE in 36m 31s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/60/660160/1/check/kolla-ansible-oraclelinux-source/7dec72e/ : SUCCESS in 1h 07m 24s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/60/660160/1/check/kolla-ansible-oraclelinux-binary/12f8bc9/ : FAILURE in 57m 25s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/60/660160/1/check/openstack-tox-pep8/6373b14/ : SUCCESS in 3m 31s\n- openstack-tox-py27 http://logs.openstack.org/60/660160/1/check/openstack-tox-py27/20cb995/ : SUCCESS in 3m 02s\n- openstack-tox-py35 http://logs.openstack.org/60/660160/1/check/openstack-tox-py35/c300873/ : SUCCESS in 3m 45s\n- build-openstack-sphinx-docs http://logs.openstack.org/60/660160/1/check/build-openstack-sphinx-docs/1c7ef4c/html/ : SUCCESS in 5m 58s\n- kolla-build-ubuntu-binary http://logs.openstack.org/60/660160/1/check/kolla-build-ubuntu-binary/c67aabe/ : SUCCESS in 27m 47s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/60/660160/1/check/kolla-build-ubuntu-source/c0d2215/ : FAILURE in 30m 11s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/60/660160/1/check/kolla-ansible-ubuntu-source/c37af44/ : SUCCESS in 52m 01s (non-voting)\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/60/660160/1/check/kolla-ansible-ubuntu-binary/d29c175/ : SUCCESS in 1h 04m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ef6043c53ab20c846fd2dd959484bc591c3cd537","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-20 18:11:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7d5282298dba152ee5cab7bc456b05a088fda24d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-20 18:50:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/60/660160/2/openstack-check/tripleo-build-containers-fedora-28-master/4645483/ : RETRY_LIMIT in 3m 30s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/60/660160/2/openstack-check/tripleo-build-containers-fedora-28-stein/bc4831d/ : SUCCESS in 37m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"29ec58618bf8fab1bdfdebf979fbb7ad24f28abe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 19:28:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/60/660160/2/check/kolla-build-centos-binary/0cf4e84/ : SUCCESS in 37m 17s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/60/660160/2/check/kolla-build-centos-source/d8dcaec/ : FAILURE in 42m 19s\n- kolla-ansible-centos-source http://logs.openstack.org/60/660160/2/check/kolla-ansible-centos-source/3acbc10/ : SUCCESS in 57m 30s (non-voting)\n- kolla-ansible-centos-binary http://logs.openstack.org/60/660160/2/check/kolla-ansible-centos-binary/e544647/ : SUCCESS in 56m 52s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/60/660160/2/check/kolla-build-oraclelinux-binary/69bf7c3/ : SUCCESS in 40m 03s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/60/660160/2/check/kolla-build-oraclelinux-source/6528707/ : FAILURE in 41m 49s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/60/660160/2/check/kolla-ansible-oraclelinux-source/e6b3c4b/ : SUCCESS in 1h 10m 04s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/60/660160/2/check/kolla-ansible-oraclelinux-binary/a8a4179/ : SUCCESS in 1h 03m 09s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/60/660160/2/check/openstack-tox-pep8/dd2a1f7/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/60/660160/2/check/openstack-tox-py27/592ad89/ : SUCCESS in 3m 05s\n- openstack-tox-py35 http://logs.openstack.org/60/660160/2/check/openstack-tox-py35/4512a85/ : SUCCESS in 3m 18s\n- build-openstack-sphinx-docs http://logs.openstack.org/60/660160/2/check/build-openstack-sphinx-docs/edba473/html/ : SUCCESS in 3m 25s\n- kolla-build-ubuntu-binary http://logs.openstack.org/60/660160/2/check/kolla-build-ubuntu-binary/f2fe29c/ : SUCCESS in 25m 14s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/60/660160/2/check/kolla-build-ubuntu-source/b4154b3/ : FAILURE in 26m 11s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/60/660160/2/check/kolla-ansible-ubuntu-source/33adcd9/ : SUCCESS in 52m 59s (non-voting)\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/60/660160/2/check/kolla-ansible-ubuntu-binary/a5c026f/ : SUCCESS in 1h 00m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1019db81f7bf02055409efcb49ea5b135bcc6dcf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-21 08:29:01.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"44c504a1c7213f0250c290c1044947ae5bfb382b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-21 09:03:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/60/660160/3/openstack-check/tripleo-build-containers-fedora-28-master/b9ef630/ : RETRY_LIMIT in 2m 36s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/60/660160/3/openstack-check/tripleo-build-containers-fedora-28-stein/68f7246/ : SUCCESS in 31m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"98fdee0c3dc1cfa30ade3436b227d05b511cbf77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 10:28:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/60/660160/3/check/kolla-build-centos-binary/04845c6/ : SUCCESS in 32m 49s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/60/660160/3/check/kolla-build-centos-source/85849e9/ : SUCCESS in 31m 37s\n- kolla-ansible-centos-source http://logs.openstack.org/60/660160/3/check/kolla-ansible-centos-source/0388ea9/ : SUCCESS in 1h 11m 16s (non-voting)\n- kolla-ansible-centos-binary http://logs.openstack.org/60/660160/3/check/kolla-ansible-centos-binary/2bc3614/ : SUCCESS in 1h 01m 37s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/60/660160/3/check/kolla-build-oraclelinux-binary/526584d/ : SUCCESS in 46m 00s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/60/660160/3/check/kolla-build-oraclelinux-source/98e2fc2/ : SUCCESS in 36m 55s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/60/660160/3/check/kolla-ansible-oraclelinux-source/544e6db/ : SUCCESS in 1h 05m 13s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/60/660160/3/check/kolla-ansible-oraclelinux-binary/034e322/ : SUCCESS in 1h 05m 48s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/60/660160/3/check/openstack-tox-pep8/8fafd63/ : SUCCESS in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/60/660160/3/check/openstack-tox-py27/ea8c81c/ : SUCCESS in 3m 43s\n- openstack-tox-py35 http://logs.openstack.org/60/660160/3/check/openstack-tox-py35/e128086/ : SUCCESS in 3m 36s\n- build-openstack-sphinx-docs http://logs.openstack.org/60/660160/3/check/build-openstack-sphinx-docs/194ade8/html/ : SUCCESS in 3m 43s\n- kolla-build-ubuntu-binary http://logs.openstack.org/60/660160/3/check/kolla-build-ubuntu-binary/1bc40a8/ : SUCCESS in 23m 45s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/60/660160/3/check/kolla-build-ubuntu-source/3cc8725/ : SUCCESS in 28m 31s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/60/660160/3/check/kolla-ansible-ubuntu-source/a924f8c/ : SUCCESS in 53m 17s (non-voting)\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/60/660160/3/check/kolla-ansible-ubuntu-binary/a8988d2/ : SUCCESS in 1h 01m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1e69b78657cb855623ac82dfa13be0b62feb7ad6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-24 20:42:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d635e8eccb4ba495379d182458ae8fc641d1104b","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-05-28 11:58:30.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f4eec1bcfcffea2a0456d8646f327711054d9c55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 11:58:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5a1a9bd05774bc06cbde3b6aa7fa2a52a85736fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 13:00:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d4964bae116e160a09ae6291fac6ae04afc16b95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 13:00:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/60/660160/3/gate/kolla-build-centos-source/41a0466/ : SUCCESS in 45m 10s\n- kolla-build-oraclelinux-source http://logs.openstack.org/60/660160/3/gate/kolla-build-oraclelinux-source/3702816/ : SUCCESS in 45m 22s\n- openstack-tox-pep8 http://logs.openstack.org/60/660160/3/gate/openstack-tox-pep8/c20cd6f/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/60/660160/3/gate/openstack-tox-py27/d3bc284/ : SUCCESS in 3m 31s\n- openstack-tox-py35 http://logs.openstack.org/60/660160/3/gate/openstack-tox-py35/dcd5b50/ : SUCCESS in 3m 37s\n- build-openstack-sphinx-docs http://logs.openstack.org/60/660160/3/gate/build-openstack-sphinx-docs/7ee05cb/html/ : SUCCESS in 3m 30s\n- kolla-build-ubuntu-source http://logs.openstack.org/60/660160/3/gate/kolla-build-ubuntu-source/a3fa80b/ : SUCCESS in 30m 19s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ed2860eb599d31b04a0d07973836871d901f8ab5","revisions":{"7d4b86fd1749db85d670adad982b7041d93b94e2":{"kind":"REWORK","_number":1,"created":"2019-05-20 16:22:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/60/660160/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/660160/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/660160/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/660160/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/660160/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/660160/1"}}},"commit":{"parents":[{"commit":"8f45874930d7eb2f525c146b1f421ea6f58e24eb","subject":"Merge \"Add sqlite package to manila-share\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f45874930d7eb2f525c146b1f421ea6f58e24eb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-20 16:22:34.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-20 16:22:34.000000000","tz":60},"subject":"Bump service versions for Queens","message":"Bump service versions for Queens\n\nChange-Id: I11b76317af047af4d3d35d9979c036347b5321b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7d4b86fd1749db85d670adad982b7041d93b94e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7d4b86fd1749db85d670adad982b7041d93b94e2"}]},"branch":"refs/heads/stable/queens"},"b5b1546ba2c67b8216165072cf922ec1c2a196ef":{"kind":"REWORK","_number":2,"created":"2019-05-20 18:11:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/60/660160/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/660160/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/660160/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/660160/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/660160/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/660160/2"}}},"commit":{"parents":[{"commit":"8f45874930d7eb2f525c146b1f421ea6f58e24eb","subject":"Merge \"Add sqlite package to manila-share\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f45874930d7eb2f525c146b1f421ea6f58e24eb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-20 16:22:34.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-20 18:11:17.000000000","tz":60},"subject":"Bump service versions for Queens","message":"Bump service versions for Queens\n\nChange-Id: I11b76317af047af4d3d35d9979c036347b5321b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5b1546ba2c67b8216165072cf922ec1c2a196ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5b1546ba2c67b8216165072cf922ec1c2a196ef"}]},"branch":"refs/heads/stable/queens"},"ed2860eb599d31b04a0d07973836871d901f8ab5":{"kind":"REWORK","_number":3,"created":"2019-05-21 08:29:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/60/660160/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/660160/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/660160/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/660160/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/660160/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/660160/3"}}},"commit":{"parents":[{"commit":"8f45874930d7eb2f525c146b1f421ea6f58e24eb","subject":"Merge \"Add sqlite package to manila-share\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f45874930d7eb2f525c146b1f421ea6f58e24eb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-20 16:22:34.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-21 08:28:19.000000000","tz":0},"subject":"Bump service versions for Queens","message":"Bump service versions for Queens\n\nChange-Id: I11b76317af047af4d3d35d9979c036347b5321b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed2860eb599d31b04a0d07973836871d901f8ab5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed2860eb599d31b04a0d07973836871d901f8ab5"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
