)]}'
{"id":"openstack%2Fkolla~660623","triplet_id":"openstack%2Fkolla~stable%2Fpike~If691820ea3dea3e8758819be42db60732680aad9","project":"openstack/kolla","branch":"stable/pike","topic":"bug/1828774-stable/pike","hashtags":[],"change_id":"If691820ea3dea3e8758819be42db60732680aad9","subject":"Install ceph-mds in ceph-base on ubuntu/debian","status":"MERGED","created":"2019-05-22 10:14:11.000000000","updated":"2019-05-24 22:28:50.000000000","submitted":"2019-05-24 22:28:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660623-1558736930335-e2f62435","meta_rev_id":"f8b29128af5b7280477a1200f59f03d924f0f7af","_number":660623,"virtual_id_number":660623,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-05-24 22:28:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-22 11:03:04.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-05-22 11:59:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-05-24 20:42:12.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-05-24 20:42:12.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 11:03:04.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-22 11:59:31.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-05-24 20:42:12.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-05-24 22:28:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2488ab5625091bd49d39569a29eac0ec0f1c1e98","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-22 10:14:11.000000000","message":"Patch Set 1: Cherry Picked from branch stable/queens.","accounts_in_message":[],"_revision_number":1},{"id":"07f9a1927a6b26e02850cc482fdd5817e9647507","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-22 11:03:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/23/660623/1/openstack-check/tripleo-build-containers-fedora-28-master/5ccd532/ : RETRY_LIMIT in 3m 24s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/23/660623/1/openstack-check/tripleo-build-containers-fedora-28-stein/422b904/ : SUCCESS in 40m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b85a2f7122b3d2779c48021eca0d8886162512f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 11:36:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/23/660623/1/check/openstack-tox-pep8/ef564f4/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/23/660623/1/check/openstack-tox-py27/714e661/ : SUCCESS in 3m 25s\n- openstack-tox-py35 http://logs.openstack.org/23/660623/1/check/openstack-tox-py35/6148300/ : SUCCESS in 3m 15s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/660623/1/check/build-openstack-sphinx-docs/680c44b/html/ : SUCCESS in 9m 12s\n- kolla-build-centos-binary http://logs.openstack.org/23/660623/1/check/kolla-build-centos-binary/64f4662/ : SUCCESS in 33m 18s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/23/660623/1/check/kolla-build-centos-source/e655dde/ : SUCCESS in 49m 06s\n- kolla-build-ubuntu-binary http://logs.openstack.org/23/660623/1/check/kolla-build-ubuntu-binary/b39f831/ : SUCCESS in 27m 48s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/23/660623/1/check/kolla-build-ubuntu-source/7f3e882/ : SUCCESS in 28m 49s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/23/660623/1/check/kolla-build-oraclelinux-binary/a69eed2/ : SUCCESS in 48m 34s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/23/660623/1/check/kolla-build-oraclelinux-source/5d452a1/ : SUCCESS in 38m 26s\n- kolla-ansible-centos-source http://logs.openstack.org/23/660623/1/check/kolla-ansible-centos-source/4d0259b/ : SUCCESS in 1h 04m 56s (non-voting)\n- kolla-ansible-centos-binary http://logs.openstack.org/23/660623/1/check/kolla-ansible-centos-binary/5267398/ : SUCCESS in 56m 33s (non-voting)\n- kolla-ansible-ubuntu-source http://logs.openstack.org/23/660623/1/check/kolla-ansible-ubuntu-source/f9299fa/ : SUCCESS in 53m 09s (non-voting)\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/23/660623/1/check/kolla-ansible-ubuntu-binary/832ad47/ : SUCCESS in 53m 03s (non-voting)\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/23/660623/1/check/kolla-ansible-oraclelinux-source/c2df1b6/ : SUCCESS in 1h 12m 20s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/23/660623/1/check/kolla-ansible-oraclelinux-binary/8e913da/ : SUCCESS in 1h 10m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0c598b301f60da1716830e7412c7409cab31efdc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-05-22 11:59:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8c7032149374133907a9e240245c1fdaac947607","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-24 20:42:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1c47792d4d8068e4c3a7020bc56164776f4e950c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 20:42:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"512de56dc5704a3f41e3d2184654ec8433bcf0da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 22:28:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/23/660623/1/gate/openstack-tox-pep8/dba28d3/ : SUCCESS in 3m 16s\n- openstack-tox-py27 http://logs.openstack.org/23/660623/1/gate/openstack-tox-py27/1870719/ : SUCCESS in 4m 32s\n- openstack-tox-py35 http://logs.openstack.org/23/660623/1/gate/openstack-tox-py35/46751fa/ : SUCCESS in 2m 51s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/660623/1/gate/build-openstack-sphinx-docs/f2cd4d2/html/ : SUCCESS in 3m 21s\n- kolla-build-centos-source http://logs.openstack.org/23/660623/1/gate/kolla-build-centos-source/03dcd53/ : SUCCESS in 43m 46s\n- kolla-build-ubuntu-source http://logs.openstack.org/23/660623/1/gate/kolla-build-ubuntu-source/dd8ad08/ : SUCCESS in 30m 56s\n- kolla-build-oraclelinux-source http://logs.openstack.org/23/660623/1/gate/kolla-build-oraclelinux-source/1a95a7b/ : SUCCESS in 45m 03s","accounts_in_message":[],"_revision_number":1},{"id":"05811c5c1323287e4bc3fed1bf6b64b39e3cc178","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 22:28:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"518dc6a2783585ecafcc8926d5f8eb9915a7af07","revisions":{"518dc6a2783585ecafcc8926d5f8eb9915a7af07":{"kind":"REWORK","_number":1,"created":"2019-05-22 10:14:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/23/660623/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/660623/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/660623/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/660623/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/660623/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/660623/1"}}},"commit":{"parents":[{"commit":"9458153859538e2da1000fdaeb0ae6d64fffabb1","subject":"Add sqlite package to manila-share","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9458153859538e2da1000fdaeb0ae6d64fffabb1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-20 10:44:52.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-22 10:14:11.000000000","tz":0},"subject":"Install ceph-mds in ceph-base on ubuntu/debian","message":"Install ceph-mds in ceph-base on ubuntu/debian\n\nStable branch only\n\nbackport: pike\n\nIn the queens and earlier ceph images on ubuntu, ceph-mds is not\ninstalled. At that point, ceph-mds was a separate package from ceph.\nFrom the ceph release used in Rocky onwards, ceph-mds is packaged with\nceph.\n\nChange-Id: If691820ea3dea3e8758819be42db60732680aad9\nCloses-Bug: #1828774\n(cherry picked from commit 8dd4c953fe996c47138706f7cf74c9f2e354d899)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/518dc6a2783585ecafcc8926d5f8eb9915a7af07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/518dc6a2783585ecafcc8926d5f8eb9915a7af07"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
