)]}'
{"id":"openstack%2Fkolla~660864","triplet_id":"openstack%2Fkolla~stable%2Frocky~Idbfc6741d371adcb06ed97d578d75d3bce862743","project":"openstack/kolla","branch":"stable/rocky","topic":"bug/1824783-stable/rocky","hashtags":[],"change_id":"Idbfc6741d371adcb06ed97d578d75d3bce862743","subject":"Fix multiple mon deployment failures","status":"MERGED","created":"2019-05-23 01:25:31.000000000","updated":"2019-05-24 22:28:49.000000000","submitted":"2019-05-24 22:28:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660864-1558736929263-d19f6385","meta_rev_id":"60d733f3348aedf7b3545b4e6640604b66d0ee36","_number":660864,"virtual_id_number":660864,"owner":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-05-24 22:28:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-23 01:55:53.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-05-24 20:42:01.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-05-24 20:42:01.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-23 01:55:53.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-24 20:42:01.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-05-24 22:28:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1798620131614f4d66aed94e039b753ed8c07cbc","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-05-23 01:25:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9f5a893cb4856816f3b2fb80c40dfaf1abfccb20","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-23 01:55:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/64/660864/1/openstack-check/tripleo-build-containers-fedora-28-master/087b8fe/ : RETRY_LIMIT in 2m 40s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/64/660864/1/openstack-check/tripleo-build-containers-fedora-28-stein/1393220/ : SUCCESS in 28m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b37d69bee13071dd1df0383dbb0b48adab5ddd49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 03:36:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/64/660864/1/check/kolla-build-centos-binary/9242062/ : SUCCESS in 37m 01s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/64/660864/1/check/kolla-build-centos-source/39856a1/ : SUCCESS in 43m 01s\n- kolla-ansible-centos-source http://logs.openstack.org/64/660864/1/check/kolla-ansible-centos-source/f3d2389/ : SUCCESS in 1h 07m 08s\n- kolla-ansible-centos-binary http://logs.openstack.org/64/660864/1/check/kolla-ansible-centos-binary/8d1b716/ : SUCCESS in 1h 10m 01s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/64/660864/1/check/kolla-build-oraclelinux-binary/1d6847e/ : SUCCESS in 54m 39s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/64/660864/1/check/kolla-build-oraclelinux-source/89e5395/ : SUCCESS in 53m 52s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/64/660864/1/check/kolla-ansible-oraclelinux-source/461d703/ : SUCCESS in 1h 13m 49s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/64/660864/1/check/kolla-ansible-oraclelinux-binary/69e61e9/ : SUCCESS in 1h 05m 08s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/64/660864/1/check/openstack-tox-lower-constraints/1e38fc0/ : SUCCESS in 3m 05s\n- openstack-tox-pep8 http://logs.openstack.org/64/660864/1/check/openstack-tox-pep8/db55cf0/ : SUCCESS in 3m 35s\n- openstack-tox-py27 http://logs.openstack.org/64/660864/1/check/openstack-tox-py27/92d8928/ : SUCCESS in 3m 23s\n- openstack-tox-py35 http://logs.openstack.org/64/660864/1/check/openstack-tox-py35/1d33219/ : SUCCESS in 3m 35s\n- build-openstack-sphinx-docs http://logs.openstack.org/64/660864/1/check/build-openstack-sphinx-docs/e0ba254/html/ : SUCCESS in 3m 09s\n- kolla-build-ubuntu-binary http://logs.openstack.org/64/660864/1/check/kolla-build-ubuntu-binary/2aaf640/ : SUCCESS in 1h 32m 52s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/64/660864/1/check/kolla-build-ubuntu-source/90b8f97/ : SUCCESS in 32m 35s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/64/660864/1/check/kolla-ansible-ubuntu-source/e434e14/ : SUCCESS in 59m 20s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/64/660864/1/check/kolla-ansible-ubuntu-binary/35f0633/ : TIMED_OUT in 2h 01m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d28507d019f8be5b04db36da7175d6f188800dcb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-24 20:42:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"322fe870cfcf1464b96e4cb879ca8cf9562f989c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 20:42:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6dce9c04114341aa3a287bd3b32a0b8803f3ee8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 22:28:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/64/660864/1/gate/kolla-build-centos-source/6f21a2f/ : SUCCESS in 47m 21s\n- kolla-build-oraclelinux-source http://logs.openstack.org/64/660864/1/gate/kolla-build-oraclelinux-source/0196623/ : SUCCESS in 1h 01m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/64/660864/1/gate/openstack-tox-lower-constraints/6fb9b3f/ : SUCCESS in 3m 26s\n- openstack-tox-pep8 http://logs.openstack.org/64/660864/1/gate/openstack-tox-pep8/66c31dc/ : SUCCESS in 4m 47s\n- openstack-tox-py27 http://logs.openstack.org/64/660864/1/gate/openstack-tox-py27/1a18c73/ : SUCCESS in 3m 32s\n- openstack-tox-py35 http://logs.openstack.org/64/660864/1/gate/openstack-tox-py35/83bc159/ : SUCCESS in 3m 28s\n- build-openstack-sphinx-docs http://logs.openstack.org/64/660864/1/gate/build-openstack-sphinx-docs/e46f1d9/html/ : SUCCESS in 3m 29s\n- kolla-build-ubuntu-source http://logs.openstack.org/64/660864/1/gate/kolla-build-ubuntu-source/461d040/ : SUCCESS in 33m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8a850252273a043464ac4ddb017ccd5c159c605a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 22:28:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"63d63ab97c87d4470be1cd72575869c32e1cb9c9","revisions":{"63d63ab97c87d4470be1cd72575869c32e1cb9c9":{"kind":"REWORK","_number":1,"created":"2019-05-23 01:25:31.000000000","uploader":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"ref":"refs/changes/64/660864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/660864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/660864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/660864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/660864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/660864/1"}}},"commit":{"parents":[{"commit":"e1af4b6a3bfcb81e683a2525096c41bcebfe3d37","subject":"Merge \"Use latest release of Monasca Log API\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1af4b6a3bfcb81e683a2525096c41bcebfe3d37"}]}],"author":{"name":"wangwei","email":"wangwei@navercorp.com","date":"2019-04-15 09:36:40.000000000","tz":540},"committer":{"name":"wangwei","email":"wang.wei@navercorp.com","date":"2019-05-23 01:25:31.000000000","tz":0},"subject":"Fix multiple mon deployment failures","message":"Fix multiple mon deployment failures\n\nWhen 4 or 5 mons are deployed, all mons will always be in the election\nstate, and the ceph cluster cannot be accessed.\n\nAfter comparing with ceph-deploy, I found that when the cluster was\ninitialized, it did not use monmap. When adding a new mon node, it\nneeds to use monmap. According to the above situation, I fixed the\nproblem.\n\nChange-Id: Idbfc6741d371adcb06ed97d578d75d3bce862743\nCloses-bug: #1824783\n(cherry picked from commit 285814353af29395932e15d78c6a154796978a21)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63d63ab97c87d4470be1cd72575869c32e1cb9c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63d63ab97c87d4470be1cd72575869c32e1cb9c9"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
