)]}'
{"id":"openstack%2Fkolla~660953","triplet_id":"openstack%2Fkolla~master~I515f9764ca93a41a871e520696193a9298856d08","project":"openstack/kolla","branch":"master","topic":"bug/1830147","hashtags":[],"change_id":"I515f9764ca93a41a871e520696193a9298856d08","subject":"Use td-agent-gem to install fluentd monasca plugin on ubuntu","status":"MERGED","created":"2019-05-23 09:28:03.000000000","updated":"2019-05-31 13:17:41.000000000","submitted":"2019-05-31 13:17:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660953-1559308661255-3d7c396e","meta_rev_id":"215080608228a96a9dd86fd08b03d08178611802","_number":660953,"virtual_id_number":660953,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17109,"name":"Chiawei Xie","email":"jeffxie@iii.org.tw","username":"dommgifer"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":26981,"name":"Marek Svensson","email":"marek@marex.st","username":"ivveh"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-05-31 13:17:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-29 15:24:06.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":17109,"name":"Chiawei Xie","email":"jeffxie@iii.org.tw","username":"dommgifer"},"all":[{"value":1,"date":"2019-05-30 02:08:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17109,"name":"Chiawei Xie","email":"jeffxie@iii.org.tw","username":"dommgifer"},{"value":2,"date":"2019-05-31 08:19:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":26981,"name":"Marek Svensson","email":"marek@marex.st","username":"ivveh"},{"value":2,"date":"2019-05-31 11:43:52.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":17109,"name":"Chiawei Xie","email":"jeffxie@iii.org.tw","username":"dommgifer"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":26981,"name":"Marek Svensson","email":"marek@marex.st","username":"ivveh"},{"value":1,"date":"2019-05-31 11:43:52.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17109,"name":"Chiawei Xie","email":"jeffxie@iii.org.tw","username":"dommgifer"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":26981,"name":"Marek Svensson","email":"marek@marex.st","username":"ivveh"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17109,"name":"Chiawei Xie","email":"jeffxie@iii.org.tw","username":"dommgifer"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":26981,"name":"Marek Svensson","email":"marek@marex.st","username":"ivveh"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-05-29 15:03:28.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17109,"name":"Chiawei Xie","email":"jeffxie@iii.org.tw","username":"dommgifer"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":26981,"name":"Marek Svensson","email":"marek@marex.st","username":"ivveh"},{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-25 14:40:15.000000000","updated_by":{"_account_id":26981,"name":"Marek Svensson","email":"marek@marex.st","username":"ivveh"},"reviewer":{"_account_id":26981,"name":"Marek Svensson","email":"marek@marex.st","username":"ivveh"},"state":"REVIEWER"},{"updated":"2019-05-26 05:42:27.000000000","updated_by":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"reviewer":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"state":"REVIEWER"},{"updated":"2019-05-29 09:24:11.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-05-29 15:24:06.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-30 02:08:56.000000000","updated_by":{"_account_id":17109,"name":"Chiawei Xie","email":"jeffxie@iii.org.tw","username":"dommgifer"},"reviewer":{"_account_id":17109,"name":"Chiawei Xie","email":"jeffxie@iii.org.tw","username":"dommgifer"},"state":"REVIEWER"},{"updated":"2019-05-31 08:19:29.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-05-31 11:43:52.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-05-31 13:17:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6fd24f1a3e9e1b0c8f2cc7c58832ccc7b6d4c25e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-23 09:28:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a475b36408db705875a5cbf24f4fe4d15d830e3","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-05-23 09:31:08.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks for fixing","accounts_in_message":[],"_revision_number":1},{"id":"caa81649c2b7a32d43f7f6b92cab8b8664ac014b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-23 10:05:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/53/660953/1/openstack-check/tripleo-build-containers-fedora-28-master/bdd6cc7/ : SUCCESS in 30m 21s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/53/660953/1/openstack-check/tripleo-build-containers-fedora-28-stein/59e15bf/ : SUCCESS in 30m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82b96e536085123bd4b106c0795321c08c4b8442","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 10:59:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/53/660953/1/check/kolla-build-centos-binary/8b3292f/ : SUCCESS in 28m 20s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/53/660953/1/check/kolla-build-centos-source/35fcc2c/ : SUCCESS in 55m 50s\n- kolla-ansible-centos-source http://logs.openstack.org/53/660953/1/check/kolla-ansible-centos-source/36dab7c/ : SUCCESS in 1h 18m 30s\n- kolla-ansible-centos-binary http://logs.openstack.org/53/660953/1/check/kolla-ansible-centos-binary/646e4e2/ : SUCCESS in 1h 03m 59s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/53/660953/1/check/kolla-ansible-centos-source-upgrade/b81b07a/ : SUCCESS in 1h 17m 49s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/53/660953/1/check/tripleo-build-containers-centos-7/32e9a43/ : SUCCESS in 41m 59s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/53/660953/1/check/kolla-build-oraclelinux-binary/f468976/ : SUCCESS in 50m 44s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/53/660953/1/check/kolla-build-oraclelinux-source/bbcc415/ : SUCCESS in 1h 04m 22s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/53/660953/1/check/kolla-ansible-oraclelinux-source/4608656/ : SUCCESS in 1h 23m 25s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/53/660953/1/check/kolla-ansible-oraclelinux-binary/938ba57/ : SUCCESS in 1h 16m 37s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/53/660953/1/check/openstack-tox-lower-constraints/dc2e5d5/ : SUCCESS in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/53/660953/1/check/openstack-tox-pep8/475ab2d/ : SUCCESS in 3m 09s\n- openstack-tox-py27 http://logs.openstack.org/53/660953/1/check/openstack-tox-py27/2046dfd/ : SUCCESS in 3m 24s\n- openstack-tox-py36 http://logs.openstack.org/53/660953/1/check/openstack-tox-py36/6d2c266/ : SUCCESS in 2m 52s\n- openstack-tox-py37 http://logs.openstack.org/53/660953/1/check/openstack-tox-py37/268fd30/ : SUCCESS in 3m 37s\n- openstack-tox-docs http://logs.openstack.org/53/660953/1/check/openstack-tox-docs/ac24c6e/html/ : SUCCESS in 3m 03s\n- kolla-build-ubuntu-binary http://logs.openstack.org/53/660953/1/check/kolla-build-ubuntu-binary/469ace2/ : SUCCESS in 30m 20s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/53/660953/1/check/kolla-build-ubuntu-source/e6e8eea/ : SUCCESS in 33m 38s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/53/660953/1/check/kolla-ansible-ubuntu-source/1eda164/ : SUCCESS in 49m 31s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/53/660953/1/check/kolla-ansible-ubuntu-binary/9e479b0/ : SUCCESS in 48m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/53/660953/1/check/kolla-ansible-ubuntu-source-upgrade/9204293/ : SUCCESS in 59m 21s","accounts_in_message":[],"_revision_number":1},{"id":"71efbbad0cc43106f6602dd20288185cabf0c02a","author":{"_account_id":26981,"name":"Marek Svensson","email":"marek@marex.st","username":"ivveh"},"date":"2019-05-25 14:40:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3981718e511709259393f7f3988e04d7a017634f","author":{"_account_id":26981,"name":"Marek Svensson","email":"marek@marex.st","username":"ivveh"},"date":"2019-05-25 14:43:04.000000000","message":"Patch Set 1:\n\nActually just replacing the final gem to td-agent-gem did it for me.","accounts_in_message":[],"_revision_number":1},{"id":"b7a134d46dfd11b4812e43f94017a1919dde5f0a","author":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"date":"2019-05-26 05:42:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7a3935105e4b76143372872392ddc24b7c2706bf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-28 17:48:22.000000000","message":"Patch Set 1:\n\n\u003e Actually just replacing the final gem to td-agent-gem did it for\n \u003e me.\n\nYeah that\u0027s essentially what this does.","accounts_in_message":[],"_revision_number":1},{"id":"ceedbc45db017709c641b3e851d0a3a1f441c45a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-29 09:22:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"00756157ed4d27970574c75ff6fc5a55e7f888fb","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-05-29 09:24:11.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c331d1f40f7eff43334d177f20a01bf9fcad7d95","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-29 10:01:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/53/660953/2/openstack-check/tripleo-build-containers-fedora-28-master/0a816d3/ : SUCCESS in 30m 49s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/53/660953/2/openstack-check/tripleo-build-containers-fedora-28-stein/2155782/ : SUCCESS in 33m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a03d9582c5098b72fa0a7f7aa3459598dfcb35c2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-29 10:56:04.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"97b48ea6ab21203bae114b5c0dfdc3f43bf65c49","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-29 11:00:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"744718a8d7ce9f46c453e93d70f3f6580b345a28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 11:12:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/53/660953/2/check/kolla-build-centos-binary/b328df7/ : SUCCESS in 41m 17s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/53/660953/2/check/kolla-build-centos-source/bce249f/ : SUCCESS in 48m 23s\n- kolla-ansible-centos-source http://logs.openstack.org/53/660953/2/check/kolla-ansible-centos-source/cff4389/ : SUCCESS in 1h 19m 08s\n- kolla-ansible-centos-binary http://logs.openstack.org/53/660953/2/check/kolla-ansible-centos-binary/bc345ff/ : SUCCESS in 1h 10m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/53/660953/2/check/kolla-ansible-centos-source-upgrade/c63bfe8/ : SUCCESS in 1h 15m 25s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/53/660953/2/check/tripleo-build-containers-centos-7/6a400d3/ : SUCCESS in 45m 48s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/53/660953/2/check/kolla-build-oraclelinux-binary/f19e611/ : SUCCESS in 51m 34s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/53/660953/2/check/kolla-build-oraclelinux-source/4e611fb/ : SUCCESS in 1h 00m 39s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/53/660953/2/check/kolla-ansible-oraclelinux-source/71d3e7c/ : SUCCESS in 1h 20m 19s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/53/660953/2/check/kolla-ansible-oraclelinux-binary/62bca97/ : SUCCESS in 1h 08m 57s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/53/660953/2/check/openstack-tox-lower-constraints/bfb001c/ : SUCCESS in 3m 47s\n- openstack-tox-pep8 http://logs.openstack.org/53/660953/2/check/openstack-tox-pep8/585891e/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/53/660953/2/check/openstack-tox-py27/33e11b4/ : SUCCESS in 3m 15s\n- openstack-tox-py36 http://logs.openstack.org/53/660953/2/check/openstack-tox-py36/3adb6d9/ : SUCCESS in 3m 09s\n- openstack-tox-py37 http://logs.openstack.org/53/660953/2/check/openstack-tox-py37/fb19920/ : SUCCESS in 3m 07s\n- openstack-tox-docs http://logs.openstack.org/53/660953/2/check/openstack-tox-docs/a72dec3/html/ : SUCCESS in 3m 02s\n- kolla-build-ubuntu-binary http://logs.openstack.org/53/660953/2/check/kolla-build-ubuntu-binary/5b6d6ea/ : FAILURE in 28m 11s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/53/660953/2/check/kolla-build-ubuntu-source/35455d8/ : FAILURE in 29m 42s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/53/660953/2/check/kolla-ansible-ubuntu-source/bbeebf2/ : FAILURE in 21m 59s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/53/660953/2/check/kolla-ansible-ubuntu-binary/85411bb/ : FAILURE in 19m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/53/660953/2/check/kolla-ansible-ubuntu-source-upgrade/d76bc48/ : FAILURE in 23m 27s","accounts_in_message":[],"_revision_number":2},{"id":"3db8a73d2fd224178817292579bf56aec3728ac8","author":{"_account_id":17109,"name":"Chiawei Xie","email":"jeffxie@iii.org.tw","username":"dommgifer"},"date":"2019-05-29 11:20:11.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"36159cb09372c0041deec1c5f3a17e41ce021d2b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-29 14:41:56.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e4a0eabfe37c4754708a28adf4d0e79b85f477eb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-29 15:03:28.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"45927ec24afba9bfd6b6b952aa122d80cc1b6657","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-29 15:24:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/53/660953/3/openstack-check/tripleo-build-containers-fedora-28-master/e8ed6ce/ : SUCCESS in 31m 27s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/53/660953/3/openstack-check/tripleo-build-containers-fedora-28-stein/d8647ba/ : SUCCESS in 35m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"de57d3c9b68a6dafce9bcc6a1a41174e1cf63797","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 16:12:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/53/660953/3/check/kolla-build-centos-binary/a300a61/ : SUCCESS in 46m 26s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/53/660953/3/check/kolla-build-centos-source/60b4723/ : SUCCESS in 40m 31s\n- kolla-ansible-centos-source http://logs.openstack.org/53/660953/3/check/kolla-ansible-centos-source/85da4e5/ : SUCCESS in 1h 06m 04s\n- kolla-ansible-centos-binary http://logs.openstack.org/53/660953/3/check/kolla-ansible-centos-binary/76563ed/ : SUCCESS in 1h 04m 36s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/53/660953/3/check/kolla-ansible-centos-source-upgrade/3c3b0a3/ : SUCCESS in 1h 15m 33s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/53/660953/3/check/tripleo-build-containers-centos-7/9f9ea85/ : SUCCESS in 41m 35s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/53/660953/3/check/kolla-build-oraclelinux-binary/035d30f/ : SUCCESS in 1h 01m 32s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/53/660953/3/check/kolla-build-oraclelinux-source/62096ba/ : SUCCESS in 47m 09s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/53/660953/3/check/kolla-ansible-oraclelinux-source/d693d9f/ : SUCCESS in 1h 23m 09s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/53/660953/3/check/kolla-ansible-oraclelinux-binary/e22672e/ : SUCCESS in 1h 09m 41s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/53/660953/3/check/openstack-tox-lower-constraints/eac4b94/ : SUCCESS in 3m 42s\n- openstack-tox-pep8 http://logs.openstack.org/53/660953/3/check/openstack-tox-pep8/cb09bad/ : SUCCESS in 2m 43s\n- openstack-tox-py27 http://logs.openstack.org/53/660953/3/check/openstack-tox-py27/53edb03/ : SUCCESS in 2m 41s\n- openstack-tox-py36 http://logs.openstack.org/53/660953/3/check/openstack-tox-py36/6b2bc1b/ : SUCCESS in 2m 52s\n- openstack-tox-py37 http://logs.openstack.org/53/660953/3/check/openstack-tox-py37/3625ae5/ : SUCCESS in 4m 50s\n- openstack-tox-docs http://logs.openstack.org/53/660953/3/check/openstack-tox-docs/34a221c/html/ : SUCCESS in 2m 40s\n- kolla-build-ubuntu-binary http://logs.openstack.org/53/660953/3/check/kolla-build-ubuntu-binary/7af343e/ : SUCCESS in 34m 22s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/53/660953/3/check/kolla-build-ubuntu-source/bc2b188/ : SUCCESS in 39m 45s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/53/660953/3/check/kolla-ansible-ubuntu-source/6fd8173/ : SUCCESS in 51m 10s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/53/660953/3/check/kolla-ansible-ubuntu-binary/7869fff/ : SUCCESS in 46m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/53/660953/3/check/kolla-ansible-ubuntu-source-upgrade/233f2c9/ : SUCCESS in 1h 01m 04s","accounts_in_message":[],"_revision_number":3},{"id":"f5bde152d9d23e45fe9e1dc72524fa2fd42cae36","author":{"_account_id":17109,"name":"Chiawei Xie","email":"jeffxie@iii.org.tw","username":"dommgifer"},"date":"2019-05-30 02:08:56.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"68e415d827badec5cafb16aacc9502f025752ef2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-05-31 08:19:29.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a088b7edb785321eb8df7a0c4a8e1c34a4b372df","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-31 11:43:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"55f88b1813f05eb88aff1873e2ef116490d9e2c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 11:44:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"77b0a0d8108a9298072fde3f310bf80fa286aa3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 13:17:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/53/660953/3/gate/kolla-build-centos-source/1f717ad/ : SUCCESS in 1h 01m 46s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/53/660953/3/gate/tripleo-build-containers-centos-7/7cf518d/ : SUCCESS in 35m 17s\n- kolla-build-oraclelinux-source http://logs.openstack.org/53/660953/3/gate/kolla-build-oraclelinux-source/338a450/ : SUCCESS in 56m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/660953/3/gate/openstack-tox-lower-constraints/90c8d44/ : SUCCESS in 4m 10s\n- openstack-tox-pep8 http://logs.openstack.org/53/660953/3/gate/openstack-tox-pep8/0baf91a/ : SUCCESS in 3m 10s\n- openstack-tox-py27 http://logs.openstack.org/53/660953/3/gate/openstack-tox-py27/49207cd/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/53/660953/3/gate/openstack-tox-py36/2375703/ : SUCCESS in 3m 42s\n- openstack-tox-py37 http://logs.openstack.org/53/660953/3/gate/openstack-tox-py37/9faa25e/ : SUCCESS in 3m 52s\n- openstack-tox-docs http://logs.openstack.org/53/660953/3/gate/openstack-tox-docs/46a4012/html/ : SUCCESS in 2m 38s\n- kolla-build-ubuntu-source http://logs.openstack.org/53/660953/3/gate/kolla-build-ubuntu-source/4468d69/ : SUCCESS in 33m 25s","accounts_in_message":[],"_revision_number":3},{"id":"69ceea8c503c2b1aa4d630534c51d56d9a066c69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 13:17:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c585ca5055c7a92fdbe5826030d3529e5c397540","revisions":{"a2ffa05a02dc96cad0d996a420cfa77bd3c3ac0f":{"kind":"REWORK","_number":1,"created":"2019-05-23 09:28:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/53/660953/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/660953/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/660953/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/660953/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/660953/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/660953/1"}}},"commit":{"parents":[{"commit":"14f7501441924a8e533fc9b75ce16719de3de945","subject":"Merge \"Fix multiple mon deployment failures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14f7501441924a8e533fc9b75ce16719de3de945"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-23 09:25:28.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-23 09:25:28.000000000","tz":60},"subject":"Use td-agent-gem to install fluentd monasca plugin on ubuntu","message":"Use td-agent-gem to install fluentd monasca plugin on ubuntu\n\nIf monasca is enabled on ubuntu, we see the following error in fluentd:\n\n[error]: config error file\u003d\"/etc/td-agent/td-agent.conf\"\nerror_class\u003dFluent::ConfigError error\u003d\"Unknown output plugin \u0027monasca\u0027.\nRun \u0027gem search -rd fluent-plugin\u0027 to find plugins\"\n\nThis is because we use \u0027fluent-gem install\u0027 to install the gem, which\nworks on CentOS but on Ubuntu it is necessary to use \u0027td-agent-gem\ninstall\u0027.\n\nChange-Id: I515f9764ca93a41a871e520696193a9298856d08\nCloses-Bug: #1830147\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2ffa05a02dc96cad0d996a420cfa77bd3c3ac0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2ffa05a02dc96cad0d996a420cfa77bd3c3ac0f"}]},"branch":"refs/heads/master"},"18baefacd989f2a951812979d9991d82ac792f5b":{"kind":"REWORK","_number":2,"created":"2019-05-29 09:22:12.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/53/660953/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/660953/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/660953/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/660953/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/660953/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/660953/2"}}},"commit":{"parents":[{"commit":"3e9cf08ac59c6b38b1542ac12e19ff44af92eb12","subject":"Merge \"Install fluentd gem on debian/ubuntu for non x86_64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e9cf08ac59c6b38b1542ac12e19ff44af92eb12"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-23 09:25:28.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-29 09:22:01.000000000","tz":60},"subject":"Use td-agent-gem to install fluentd monasca plugin on ubuntu","message":"Use td-agent-gem to install fluentd monasca plugin on ubuntu\n\nIf monasca is enabled on ubuntu, we see the following error in fluentd:\n\n[error]: config error file\u003d\"/etc/td-agent/td-agent.conf\"\nerror_class\u003dFluent::ConfigError error\u003d\"Unknown output plugin \u0027monasca\u0027.\nRun \u0027gem search -rd fluent-plugin\u0027 to find plugins\"\n\nThis is because we use \u0027fluent-gem install\u0027 to install the gem, which\nworks on CentOS but on Ubuntu it is necessary to use \u0027td-agent-gem\ninstall\u0027.\n\nChange-Id: I515f9764ca93a41a871e520696193a9298856d08\nCloses-Bug: #1830147\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18baefacd989f2a951812979d9991d82ac792f5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18baefacd989f2a951812979d9991d82ac792f5b"}]},"branch":"refs/heads/master"},"c585ca5055c7a92fdbe5826030d3529e5c397540":{"kind":"REWORK","_number":3,"created":"2019-05-29 14:41:56.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/53/660953/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/660953/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/660953/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/660953/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/660953/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/660953/3"}}},"commit":{"parents":[{"commit":"3e9cf08ac59c6b38b1542ac12e19ff44af92eb12","subject":"Merge \"Install fluentd gem on debian/ubuntu for non x86_64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e9cf08ac59c6b38b1542ac12e19ff44af92eb12"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-23 09:25:28.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-29 14:41:52.000000000","tz":0},"subject":"Use td-agent-gem to install fluentd monasca plugin on ubuntu","message":"Use td-agent-gem to install fluentd monasca plugin on ubuntu\n\nIf monasca is enabled on ubuntu, we see the following error in fluentd:\n\n[error]: config error file\u003d\"/etc/td-agent/td-agent.conf\"\nerror_class\u003dFluent::ConfigError error\u003d\"Unknown output plugin \u0027monasca\u0027.\nRun \u0027gem search -rd fluent-plugin\u0027 to find plugins\"\n\nThis is because we use \u0027fluent-gem install\u0027 to install the gem, which\nworks on CentOS but on Ubuntu it is necessary to use \u0027td-agent-gem\ninstall\u0027.\n\nChange-Id: I515f9764ca93a41a871e520696193a9298856d08\nCloses-Bug: #1830147\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c585ca5055c7a92fdbe5826030d3529e5c397540"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c585ca5055c7a92fdbe5826030d3529e5c397540"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
