)]}'
{"id":"openstack%2Fkolla~660954","triplet_id":"openstack%2Fkolla~stable%2Fstein~I0a5fc2d47ccb4f64dbf8fca7092240cd40777b28","project":"openstack/kolla","branch":"stable/stein","topic":"feature/run_ceph_gates_on_ceph_change-stable/stein","hashtags":[],"change_id":"I0a5fc2d47ccb4f64dbf8fca7092240cd40777b28","subject":"Add CI trigger for ceph jobs","status":"MERGED","created":"2019-05-23 09:29:39.000000000","updated":"2019-05-31 09:34:34.000000000","submitted":"2019-05-31 09:34:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660954-1559295274446-229039fc","meta_rev_id":"bf305936b81a7c173202a85afb55109a5f4c530b","_number":660954,"virtual_id_number":660954,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-05-31 09:34:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},"all":[{"value":1,"date":"2019-05-26 09:33:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},{"value":2,"date":"2019-05-31 08:22:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-05-23 09:30:09.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},{"value":1,"date":"2019-05-31 08:22:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-26 09:33:46.000000000","updated_by":{"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},"reviewer":{"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},"state":"REVIEWER"},{"updated":"2019-05-31 08:22:01.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-05-31 09:34:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cd83c348a368fb58a38ccc64c2510e4a4748b024","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-23 09:29:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"332563bd7ebc64fed62a623609723c6d4a9c74b3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-23 09:30:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e18dcd2ebee358f7635851ba9753498c99165f74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 10:55:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/54/660954/1/check/kolla-build-centos-binary/dd90188/ : FAILURE in 39m 26s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/54/660954/1/check/kolla-build-centos-source/e942afe/ : SUCCESS in 46m 37s\n- kolla-ansible-centos-source http://logs.openstack.org/54/660954/1/check/kolla-ansible-centos-source/c5e336b/ : SUCCESS in 1h 19m 23s\n- kolla-ansible-centos-binary http://logs.openstack.org/54/660954/1/check/kolla-ansible-centos-binary/08001b9/ : FAILURE in 33m 01s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/54/660954/1/check/kolla-ansible-centos-source-upgrade/673cade/ : SUCCESS in 1h 14m 43s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/54/660954/1/check/tripleo-build-containers-centos-7/11fb980/ : SUCCESS in 41m 05s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/54/660954/1/check/kolla-build-oraclelinux-binary/ca9aeca/ : FAILURE in 51m 37s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/54/660954/1/check/kolla-build-oraclelinux-source/3ee35f9/ : SUCCESS in 1h 15m 19s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/54/660954/1/check/kolla-ansible-oraclelinux-source/1fad85d/ : SUCCESS in 1h 13m 41s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/54/660954/1/check/kolla-ansible-oraclelinux-binary/64c9454/ : FAILURE in 31m 24s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/54/660954/1/check/openstack-tox-lower-constraints/a0c9f2a/ : SUCCESS in 4m 56s\n- openstack-tox-pep8 http://logs.openstack.org/54/660954/1/check/openstack-tox-pep8/168f87d/ : SUCCESS in 2m 43s\n- openstack-tox-py27 http://logs.openstack.org/54/660954/1/check/openstack-tox-py27/555f302/ : SUCCESS in 2m 55s\n- openstack-tox-py36 http://logs.openstack.org/54/660954/1/check/openstack-tox-py36/adbb21f/ : SUCCESS in 3m 03s\n- openstack-tox-py37 http://logs.openstack.org/54/660954/1/check/openstack-tox-py37/58793e3/ : SUCCESS in 3m 15s\n- openstack-tox-docs http://logs.openstack.org/54/660954/1/check/openstack-tox-docs/fc5bf16/html/ : SUCCESS in 3m 00s\n- kolla-build-ubuntu-binary http://logs.openstack.org/54/660954/1/check/kolla-build-ubuntu-binary/809798b/ : SUCCESS in 26m 52s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/54/660954/1/check/kolla-build-ubuntu-source/abffb48/ : SUCCESS in 30m 46s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/54/660954/1/check/kolla-ansible-ubuntu-source/6010998/ : SUCCESS in 53m 30s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/54/660954/1/check/kolla-ansible-ubuntu-binary/5b0889f/ : SUCCESS in 51m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/54/660954/1/check/kolla-ansible-ubuntu-source-upgrade/a087346/ : SUCCESS in 1h 13m 15s","accounts_in_message":[],"_revision_number":1},{"id":"097e476153128cdfe3c0649974ce84b9e4a763de","author":{"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},"date":"2019-05-26 09:33:46.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"774fd7a2586aa2af45825b33429ec90dc8079dcc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-26 11:20:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/54/660954/1/check/kolla-build-centos-binary/5e81dd8/ : FAILURE in 39m 56s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/54/660954/1/check/kolla-build-centos-source/e2ec5b2/ : SUCCESS in 38m 35s\n- kolla-ansible-centos-source http://logs.openstack.org/54/660954/1/check/kolla-ansible-centos-source/e552b44/ : SUCCESS in 1h 05m 15s\n- kolla-ansible-centos-binary http://logs.openstack.org/54/660954/1/check/kolla-ansible-centos-binary/497758f/ : FAILURE in 30m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/54/660954/1/check/kolla-ansible-centos-source-upgrade/62a5640/ : SUCCESS in 1h 32m 17s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/54/660954/1/check/tripleo-build-containers-centos-7/92e73c2/ : SUCCESS in 38m 48s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/54/660954/1/check/kolla-build-oraclelinux-binary/1dab22a/ : FAILURE in 58m 11s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/54/660954/1/check/kolla-build-oraclelinux-source/52b6784/ : SUCCESS in 46m 55s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/54/660954/1/check/kolla-ansible-oraclelinux-source/bbb8185/ : SUCCESS in 1h 06m 21s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/54/660954/1/check/kolla-ansible-oraclelinux-binary/5cfd050/ : FAILURE in 40m 21s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/54/660954/1/check/openstack-tox-lower-constraints/c28fe04/ : SUCCESS in 4m 03s\n- openstack-tox-pep8 http://logs.openstack.org/54/660954/1/check/openstack-tox-pep8/4f20c7e/ : SUCCESS in 3m 33s\n- openstack-tox-py27 http://logs.openstack.org/54/660954/1/check/openstack-tox-py27/9dd4cc8/ : SUCCESS in 3m 03s\n- openstack-tox-py36 http://logs.openstack.org/54/660954/1/check/openstack-tox-py36/082f8c4/ : SUCCESS in 2m 56s\n- openstack-tox-py37 http://logs.openstack.org/54/660954/1/check/openstack-tox-py37/1d3e82a/ : SUCCESS in 3m 54s\n- openstack-tox-docs http://logs.openstack.org/54/660954/1/check/openstack-tox-docs/5569685/html/ : SUCCESS in 2m 55s\n- kolla-build-ubuntu-binary http://logs.openstack.org/54/660954/1/check/kolla-build-ubuntu-binary/66b0378/ : SUCCESS in 37m 03s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/54/660954/1/check/kolla-build-ubuntu-source/35746a6/ : SUCCESS in 36m 18s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/54/660954/1/check/kolla-ansible-ubuntu-source/31b7ade/ : SUCCESS in 52m 58s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/54/660954/1/check/kolla-ansible-ubuntu-binary/8194d97/ : SUCCESS in 46m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/54/660954/1/check/kolla-ansible-ubuntu-source-upgrade/7df81e3/ : SUCCESS in 55m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e70284924f0ae1a450592fc93884cc392e2aedf0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-05-31 08:22:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bba8c49da4e0dd908c92ae14ac330b1e950752cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 08:22:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"915d966cd35e911e8241c7a911d0a51126d43534","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 09:34:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c45ffc3646a5400583fc5cfa5e0736ef43f4e0a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 09:34:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/54/660954/1/gate/kolla-build-centos-source/9c7954e/ : SUCCESS in 50m 50s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/54/660954/1/gate/tripleo-build-containers-centos-7/423b658/ : SUCCESS in 47m 34s\n- kolla-build-oraclelinux-source http://logs.openstack.org/54/660954/1/gate/kolla-build-oraclelinux-source/13a0324/ : SUCCESS in 49m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/54/660954/1/gate/openstack-tox-lower-constraints/4884f36/ : SUCCESS in 3m 45s\n- openstack-tox-pep8 http://logs.openstack.org/54/660954/1/gate/openstack-tox-pep8/e4a4c32/ : SUCCESS in 3m 09s\n- openstack-tox-py27 http://logs.openstack.org/54/660954/1/gate/openstack-tox-py27/f07be8b/ : SUCCESS in 3m 11s\n- openstack-tox-py36 http://logs.openstack.org/54/660954/1/gate/openstack-tox-py36/819b3d9/ : SUCCESS in 3m 11s\n- openstack-tox-py37 http://logs.openstack.org/54/660954/1/gate/openstack-tox-py37/e93d616/ : SUCCESS in 3m 09s\n- openstack-tox-docs http://logs.openstack.org/54/660954/1/gate/openstack-tox-docs/89231b0/html/ : SUCCESS in 2m 52s\n- kolla-build-ubuntu-source http://logs.openstack.org/54/660954/1/gate/kolla-build-ubuntu-source/97d8184/ : SUCCESS in 29m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb4ad16744889464d44d3a16e2d1a279deb33842","revisions":{"eb4ad16744889464d44d3a16e2d1a279deb33842":{"kind":"REWORK","_number":1,"created":"2019-05-23 09:29:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/54/660954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/660954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/660954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/660954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/660954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/660954/1"}}},"commit":{"parents":[{"commit":"728d49e14068a6977c6e6cb601278dcf69b2e9ff","subject":"Merge \"Add sqlite package to manila-share\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/728d49e14068a6977c6e6cb601278dcf69b2e9ff"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-04-03 12:06:22.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-23 09:29:39.000000000","tz":0},"subject":"Add CI trigger for ceph jobs","message":"Add CI trigger for ceph jobs\n\nAdd a trigger for running CI ceph jobs when docker/ceph/.* and\ndocker/cinder/.* files are being changed.\n\nChange-Id: I0a5fc2d47ccb4f64dbf8fca7092240cd40777b28\n(cherry picked from commit 14f1507d2c2002c05cb30196b6650f6a8b23206a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb4ad16744889464d44d3a16e2d1a279deb33842"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb4ad16744889464d44d3a16e2d1a279deb33842"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
