)]}'
{"id":"openstack%2Fkolla~660955","triplet_id":"openstack%2Fkolla~stable%2Frocky~I0a5fc2d47ccb4f64dbf8fca7092240cd40777b28","project":"openstack/kolla","branch":"stable/rocky","topic":"feature/run_ceph_gates_on_ceph_change-stable/rocky","hashtags":[],"change_id":"I0a5fc2d47ccb4f64dbf8fca7092240cd40777b28","subject":"Add CI trigger for ceph jobs","status":"MERGED","created":"2019-05-23 09:33:50.000000000","updated":"2019-05-28 20:02:32.000000000","submitted":"2019-05-28 20:02:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660955-1559073752935-08ba7da0","meta_rev_id":"8c22fc3d30e327539c074dc3d4c93d2461db84cc","_number":660955,"virtual_id_number":660955,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-05-28 20:02:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-05-24 20:41:27.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-05-23 14:08:34.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-05-24 20:41:27.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-23 09:35:45.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-05-24 20:41:27.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-05-28 20:02:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b62e0c0767550a35c16ce1057d30dbe46c86fe52","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-23 09:33:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33b5fa47c6d5e0eaea844c576cc7d077b62b313a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-05-23 09:35:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ab4929f6564a397b91287587059e82abd4728dfb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-23 11:06:54.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5e493c8b84d6f1961f6d7222ae88633748ab7b2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 13:10:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/55/660955/2/check/kolla-build-centos-binary/c24ff76/ : SUCCESS in 34m 10s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/55/660955/2/check/kolla-build-centos-source/a1c59b4/ : SUCCESS in 46m 48s\n- kolla-ansible-centos-source http://logs.openstack.org/55/660955/2/check/kolla-ansible-centos-source/041c5f0/ : SUCCESS in 1h 16m 30s\n- kolla-ansible-centos-binary http://logs.openstack.org/55/660955/2/check/kolla-ansible-centos-binary/3b1902c/ : SUCCESS in 1h 12m 51s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/55/660955/2/check/kolla-build-oraclelinux-binary/c1018d2/ : SUCCESS in 44m 37s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/55/660955/2/check/kolla-build-oraclelinux-source/dbf829f/ : SUCCESS in 52m 11s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/55/660955/2/check/kolla-ansible-oraclelinux-source/269a102/ : SUCCESS in 1h 26m 37s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/55/660955/2/check/kolla-ansible-oraclelinux-binary/8cc799d/ : SUCCESS in 1h 13m 44s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/55/660955/2/check/openstack-tox-lower-constraints/d4cffc3/ : SUCCESS in 3m 27s\n- openstack-tox-pep8 http://logs.openstack.org/55/660955/2/check/openstack-tox-pep8/1a052f4/ : SUCCESS in 3m 31s\n- openstack-tox-py27 http://logs.openstack.org/55/660955/2/check/openstack-tox-py27/6ee15b4/ : SUCCESS in 3m 27s\n- openstack-tox-py35 http://logs.openstack.org/55/660955/2/check/openstack-tox-py35/4e9d574/ : SUCCESS in 3m 16s\n- build-openstack-sphinx-docs http://logs.openstack.org/55/660955/2/check/build-openstack-sphinx-docs/c52ab4a/html/ : SUCCESS in 3m 16s\n- kolla-build-ubuntu-binary http://logs.openstack.org/55/660955/2/check/kolla-build-ubuntu-binary/567415b/ : SUCCESS in 1h 15m 35s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/55/660955/2/check/kolla-build-ubuntu-source/4e89929/ : SUCCESS in 35m 33s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/55/660955/2/check/kolla-ansible-ubuntu-source/db9e782/ : SUCCESS in 1h 02m 27s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/55/660955/2/check/kolla-ansible-ubuntu-binary/06ffc3c/ : SUCCESS in 1h 57m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0d91a59af3a7373d0f6175936a2b484a51408711","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-23 14:08:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"535a73dda8a0587121fbb03c7cf9e464c665602b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-24 20:41:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8599123d474e410ea8890ea2777a94157ce9e0b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 20:41:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"28e376305af646354347d1fe824cbce9b20fca1c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 21:40:23.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source http://logs.openstack.org/55/660955/2/gate/kolla-build-centos-source/d04a4c3/ : SUCCESS in 38m 23s\n- kolla-build-oraclelinux-source http://logs.openstack.org/55/660955/2/gate/kolla-build-oraclelinux-source/d0c2403/ : SUCCESS in 50m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/660955/2/gate/openstack-tox-lower-constraints/def29ce/ : SUCCESS in 3m 19s\n- openstack-tox-pep8 http://logs.openstack.org/55/660955/2/gate/openstack-tox-pep8/3b31ce7/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/55/660955/2/gate/openstack-tox-py27/c02d742/ : SUCCESS in 5m 03s\n- openstack-tox-py35 http://logs.openstack.org/55/660955/2/gate/openstack-tox-py35/2079c0a/ : SUCCESS in 3m 22s\n- build-openstack-sphinx-docs http://logs.openstack.org/55/660955/2/gate/build-openstack-sphinx-docs/5a2884b/html/ : SUCCESS in 3m 59s\n- kolla-build-ubuntu-source finger://ze04.openstack.org/391a9c4a5e2446b29c853e9ab714306c : RETRY_LIMIT in 39m 41s","accounts_in_message":[],"_revision_number":2},{"id":"8b2f2dcdde6e7ddb360a157bbfe7bcb45440b86a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-28 16:45:09.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a52e6cae0684a7bf4962d90da3232228b1915450","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 18:56:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/55/660955/2/check/kolla-build-centos-binary/5132a05/ : SUCCESS in 35m 32s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/55/660955/2/check/kolla-build-centos-source/cfb8e86/ : SUCCESS in 40m 44s\n- kolla-ansible-centos-source http://logs.openstack.org/55/660955/2/check/kolla-ansible-centos-source/10a927d/ : SUCCESS in 1h 25m 43s\n- kolla-ansible-centos-binary http://logs.openstack.org/55/660955/2/check/kolla-ansible-centos-binary/d68385a/ : SUCCESS in 1h 06m 16s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/55/660955/2/check/kolla-build-oraclelinux-binary/7522a1f/ : FAILURE in 3m 37s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/55/660955/2/check/kolla-build-oraclelinux-source/10ecd7c/ : SUCCESS in 42m 05s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/55/660955/2/check/kolla-ansible-oraclelinux-source/7e80d5a/ : SUCCESS in 1h 12m 37s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/55/660955/2/check/kolla-ansible-oraclelinux-binary/86d7b03/ : SUCCESS in 1h 03m 51s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/55/660955/2/check/openstack-tox-lower-constraints/8b484da/ : SUCCESS in 3m 19s\n- openstack-tox-pep8 http://logs.openstack.org/55/660955/2/check/openstack-tox-pep8/70e724c/ : SUCCESS in 4m 15s\n- openstack-tox-py27 http://logs.openstack.org/55/660955/2/check/openstack-tox-py27/4d876ac/ : SUCCESS in 3m 29s\n- openstack-tox-py35 http://logs.openstack.org/55/660955/2/check/openstack-tox-py35/36ca736/ : SUCCESS in 2m 50s\n- build-openstack-sphinx-docs http://logs.openstack.org/55/660955/2/check/build-openstack-sphinx-docs/be80991/html/ : SUCCESS in 3m 37s\n- kolla-build-ubuntu-binary http://logs.openstack.org/55/660955/2/check/kolla-build-ubuntu-binary/43a7483/ : SUCCESS in 1h 27m 00s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/55/660955/2/check/kolla-build-ubuntu-source/421d801/ : SUCCESS in 29m 49s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/55/660955/2/check/kolla-ansible-ubuntu-source/724aa5e/ : SUCCESS in 47m 09s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/55/660955/2/check/kolla-ansible-ubuntu-binary/f9f4b17/ : SUCCESS in 1h 54m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"29b4f1203b1fee1869d824c13b656471c9cac887","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 18:56:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"74230196072677bc5bb0f5fcb179823ea39d7c76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 20:02:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/55/660955/2/gate/kolla-build-centos-source/323550a/ : SUCCESS in 58m 40s\n- kolla-build-oraclelinux-source http://logs.openstack.org/55/660955/2/gate/kolla-build-oraclelinux-source/ae01eb2/ : SUCCESS in 49m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/660955/2/gate/openstack-tox-lower-constraints/95fff1b/ : SUCCESS in 3m 10s\n- openstack-tox-pep8 http://logs.openstack.org/55/660955/2/gate/openstack-tox-pep8/2c46780/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/55/660955/2/gate/openstack-tox-py27/470a934/ : SUCCESS in 2m 47s\n- openstack-tox-py35 http://logs.openstack.org/55/660955/2/gate/openstack-tox-py35/ffed0fb/ : SUCCESS in 2m 58s\n- build-openstack-sphinx-docs http://logs.openstack.org/55/660955/2/gate/build-openstack-sphinx-docs/b21b39f/html/ : SUCCESS in 3m 42s\n- kolla-build-ubuntu-source http://logs.openstack.org/55/660955/2/gate/kolla-build-ubuntu-source/db4c2cb/ : SUCCESS in 44m 30s","accounts_in_message":[],"_revision_number":2},{"id":"7e913e3e1b9e22d713b16d959003fa004d97a74a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 20:02:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e4929da14d971a408f0193abcaedb3880459f244","revisions":{"825bf4015999e588ab21912740466dddfa9c9fb7":{"kind":"REWORK","_number":1,"created":"2019-05-23 09:33:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/55/660955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/660955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/660955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/660955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/660955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/660955/1"}}},"commit":{"parents":[{"commit":"e1af4b6a3bfcb81e683a2525096c41bcebfe3d37","subject":"Merge \"Use latest release of Monasca Log API\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1af4b6a3bfcb81e683a2525096c41bcebfe3d37"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-04-03 12:06:22.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-23 09:33:39.000000000","tz":60},"subject":"Add CI trigger for ceph jobs","message":"Add CI trigger for ceph jobs\n\nAdd a trigger for running CI ceph jobs when docker/ceph/.* and\ndocker/cinder/.* files are being changed.\n\nChange-Id: I0a5fc2d47ccb4f64dbf8fca7092240cd40777b28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/825bf4015999e588ab21912740466dddfa9c9fb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/825bf4015999e588ab21912740466dddfa9c9fb7"}]},"branch":"refs/heads/stable/rocky"},"e4929da14d971a408f0193abcaedb3880459f244":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-23 11:06:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/55/660955/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/660955/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/660955/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/660955/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/660955/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/660955/2"}}},"commit":{"parents":[{"commit":"e1af4b6a3bfcb81e683a2525096c41bcebfe3d37","subject":"Merge \"Use latest release of Monasca Log API\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1af4b6a3bfcb81e683a2525096c41bcebfe3d37"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-04-03 12:06:22.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-23 11:05:57.000000000","tz":60},"subject":"Add CI trigger for ceph jobs","message":"Add CI trigger for ceph jobs\n\nAdd a trigger for running CI ceph jobs when docker/ceph/.* and\ndocker/cinder/.* files are being changed.\n\nChange-Id: I0a5fc2d47ccb4f64dbf8fca7092240cd40777b28\n(cherry picked from commit 14f1507d2c2002c05cb30196b6650f6a8b23206a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4929da14d971a408f0193abcaedb3880459f244"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4929da14d971a408f0193abcaedb3880459f244"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
