)]}'
{"id":"openstack%2Fkolla~660988","triplet_id":"openstack%2Fkolla~master~Ib3953b49247bb36ac8ce25578c05c6bcb2a1905a","project":"openstack/kolla","branch":"master","topic":"bug/1830184","hashtags":[],"change_id":"Ib3953b49247bb36ac8ce25578c05c6bcb2a1905a","subject":"Bump Monasca Fluentd plugin version","status":"MERGED","created":"2019-05-23 12:57:40.000000000","updated":"2019-05-24 22:28:51.000000000","submitted":"2019-05-24 22:28:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660988-1558736931557-04c46964","meta_rev_id":"7c4e2ac5cb3b42e8531407185caddc5b9d03bba2","_number":660988,"virtual_id_number":660988,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-05-24 22:28:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-23 13:32:07.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-05-24 20:44:44.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-05-23 15:47:07.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-05-24 20:44:44.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-23 13:32:07.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-23 15:47:07.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-05-24 20:44:44.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-05-24 22:28:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ad96056f1f58062faec47fdc2785160878d02891","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-05-23 12:57:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"720974ab4cf64b11b6f9915773500a31d1b0aa8d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-23 13:32:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/88/660988/1/openstack-check/tripleo-build-containers-fedora-28-master/8c87984/ : SUCCESS in 29m 48s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/88/660988/1/openstack-check/tripleo-build-containers-fedora-28-stein/096fb57/ : SUCCESS in 32m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a8ba6f65089971f9b79fb842f79a373ae2314aa","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-05-23 14:48:03.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 594b0645307432ba007657a135dd8fa34d84c7eb","accounts_in_message":[],"_revision_number":1},{"id":"492b6bc1cbc55dc2920bc240eac932249809ddaa","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-05-23 14:49:28.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit c7e0f0db6eb7782501329be401dce05a34c0a1ed","accounts_in_message":[],"_revision_number":1},{"id":"962cc32f0597101958eabcd7325f314e00dfcf4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 14:51:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/88/660988/1/check/kolla-build-centos-binary/0effc45/ : SUCCESS in 46m 20s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/88/660988/1/check/kolla-build-centos-source/045ad07/ : SUCCESS in 49m 27s\n- kolla-ansible-centos-source http://logs.openstack.org/88/660988/1/check/kolla-ansible-centos-source/565c242/ : SUCCESS in 1h 25m 27s\n- kolla-ansible-centos-binary http://logs.openstack.org/88/660988/1/check/kolla-ansible-centos-binary/cae167c/ : SUCCESS in 59m 47s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/88/660988/1/check/kolla-ansible-centos-source-upgrade/676e514/ : SUCCESS in 1h 18m 02s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/88/660988/1/check/tripleo-build-containers-centos-7/2394c2c/ : SUCCESS in 40m 04s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/88/660988/1/check/kolla-build-oraclelinux-binary/c9aaa52/ : SUCCESS in 1h 13m 12s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/88/660988/1/check/kolla-build-oraclelinux-source/72b643b/ : SUCCESS in 1h 13m 41s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/88/660988/1/check/kolla-ansible-oraclelinux-source/9a965c8/ : SUCCESS in 1h 29m 37s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/88/660988/1/check/kolla-ansible-oraclelinux-binary/511954f/ : SUCCESS in 1h 21m 34s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/88/660988/1/check/openstack-tox-lower-constraints/290a8e5/ : SUCCESS in 3m 59s\n- openstack-tox-pep8 http://logs.openstack.org/88/660988/1/check/openstack-tox-pep8/7fd9c11/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/88/660988/1/check/openstack-tox-py27/5d44230/ : SUCCESS in 3m 05s\n- openstack-tox-py36 http://logs.openstack.org/88/660988/1/check/openstack-tox-py36/99718a7/ : SUCCESS in 3m 52s\n- openstack-tox-py37 http://logs.openstack.org/88/660988/1/check/openstack-tox-py37/5a2bea0/ : SUCCESS in 3m 36s\n- openstack-tox-docs http://logs.openstack.org/88/660988/1/check/openstack-tox-docs/d60281c/html/ : SUCCESS in 3m 10s\n- kolla-build-ubuntu-binary http://logs.openstack.org/88/660988/1/check/kolla-build-ubuntu-binary/14cf388/ : SUCCESS in 58m 28s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/88/660988/1/check/kolla-build-ubuntu-source/0f12596/ : SUCCESS in 32m 55s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/88/660988/1/check/kolla-ansible-ubuntu-source/307732f/ : SUCCESS in 53m 37s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/88/660988/1/check/kolla-ansible-ubuntu-binary/ba67eaf/ : SUCCESS in 50m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/88/660988/1/check/kolla-ansible-ubuntu-source-upgrade/0ea2719/ : SUCCESS in 1h 04m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b85c753adad7abc9e632534a6762be59949d3991","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-23 15:47:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b8315f98552bc0dace95aa5b4c0c2ed5f7923a63","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-24 20:44:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"859d6b44c9b19928045e3363b9272adbe3670971","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 20:47:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"442b7ba7ce803e7b56b36315b95cf99d7b81eb99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 22:28:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fa2b888d0c8ed5447059bfbae10a92f2a020e6bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 22:28:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/88/660988/1/gate/kolla-build-centos-source/489dc66/ : SUCCESS in 50m 14s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/88/660988/1/gate/tripleo-build-containers-centos-7/0ef8d10/ : SUCCESS in 40m 56s\n- kolla-build-oraclelinux-source http://logs.openstack.org/88/660988/1/gate/kolla-build-oraclelinux-source/10c6c9e/ : SUCCESS in 51m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/660988/1/gate/openstack-tox-lower-constraints/2a4ed9f/ : SUCCESS in 4m 00s\n- openstack-tox-pep8 http://logs.openstack.org/88/660988/1/gate/openstack-tox-pep8/071bf55/ : SUCCESS in 3m 02s\n- openstack-tox-py27 http://logs.openstack.org/88/660988/1/gate/openstack-tox-py27/c8af8ff/ : SUCCESS in 3m 04s\n- openstack-tox-py36 http://logs.openstack.org/88/660988/1/gate/openstack-tox-py36/9f31f9b/ : SUCCESS in 3m 37s\n- openstack-tox-py37 http://logs.openstack.org/88/660988/1/gate/openstack-tox-py37/9867cf9/ : SUCCESS in 3m 13s\n- openstack-tox-docs http://logs.openstack.org/88/660988/1/gate/openstack-tox-docs/88610da/html/ : SUCCESS in 2m 31s\n- kolla-build-ubuntu-source http://logs.openstack.org/88/660988/1/gate/kolla-build-ubuntu-source/39a6cf0/ : SUCCESS in 28m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"529d2b54ed6f04df45732fccfdb9e2f0214d65c1","revisions":{"529d2b54ed6f04df45732fccfdb9e2f0214d65c1":{"kind":"REWORK","_number":1,"created":"2019-05-23 12:57:40.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/88/660988/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/660988/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/660988/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/660988/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/660988/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/660988/1"}}},"commit":{"parents":[{"commit":"14f7501441924a8e533fc9b75ce16719de3de945","subject":"Merge \"Fix multiple mon deployment failures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14f7501441924a8e533fc9b75ce16719de3de945"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-05-23 10:43:17.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-05-23 12:57:26.000000000","tz":60},"subject":"Bump Monasca Fluentd plugin version","message":"Bump Monasca Fluentd plugin version\n\nBump version to include support for setting log message field\nname.\n\nChange-Id: Ib3953b49247bb36ac8ce25578c05c6bcb2a1905a\nPartial-Bug: #1830184\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/529d2b54ed6f04df45732fccfdb9e2f0214d65c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/529d2b54ed6f04df45732fccfdb9e2f0214d65c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
