)]}'
{"id":"openstack%2Fkolla~661016","triplet_id":"openstack%2Fkolla~stable%2Frocky~Ib3953b49247bb36ac8ce25578c05c6bcb2a1905a","project":"openstack/kolla","branch":"stable/rocky","topic":"bug/1830184-stable/rocky","hashtags":[],"change_id":"Ib3953b49247bb36ac8ce25578c05c6bcb2a1905a","subject":"Bump Monasca Fluentd plugin version","status":"MERGED","created":"2019-05-23 14:48:03.000000000","updated":"2019-05-28 18:44:26.000000000","submitted":"2019-05-28 18:44:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661016-1559069066501-7cc59f37","meta_rev_id":"79f4c9bf662d8c41766ffdef3f3e13dbce51dd40","_number":661016,"virtual_id_number":661016,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-05-28 18:44:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-23 16:02:32.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-05-24 20:41:09.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-05-28 17:47:08.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-05-28 17:47:08.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-23 16:02:32.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-24 20:41:09.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-05-28 17:47:08.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-05-28 18:44:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3f42f161ce2fe568b469a4fea9a42c2dbeacdf06","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-05-23 14:48:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"39b63a29b9c081a9bc9d6d2fd0b04e958571ee5d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-23 16:02:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/16/661016/1/openstack-check/tripleo-build-containers-fedora-28-master/d75096b/ : RETRY_LIMIT in 2m 57s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/16/661016/1/openstack-check/tripleo-build-containers-fedora-28-stein/573b48f/ : SUCCESS in 30m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"33337131ab418eec18f027cf786670148686dd7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 16:58:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/16/661016/1/check/kolla-build-centos-binary/0de2b18/ : SUCCESS in 40m 59s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/16/661016/1/check/kolla-build-centos-source/61aa055/ : SUCCESS in 48m 40s\n- kolla-ansible-centos-source http://logs.openstack.org/16/661016/1/check/kolla-ansible-centos-source/7a71c2b/ : SUCCESS in 1h 11m 44s\n- kolla-ansible-centos-binary http://logs.openstack.org/16/661016/1/check/kolla-ansible-centos-binary/59d7671/ : SUCCESS in 1h 10m 41s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/16/661016/1/check/kolla-build-oraclelinux-binary/d74d54e/ : SUCCESS in 50m 58s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/16/661016/1/check/kolla-build-oraclelinux-source/eaadd8e/ : SUCCESS in 51m 25s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/16/661016/1/check/kolla-ansible-oraclelinux-source/a2753ea/ : SUCCESS in 1h 22m 50s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/16/661016/1/check/kolla-ansible-oraclelinux-binary/b6d1ddb/ : SUCCESS in 1h 06m 29s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/16/661016/1/check/openstack-tox-lower-constraints/4c0c4e5/ : SUCCESS in 3m 47s\n- openstack-tox-pep8 http://logs.openstack.org/16/661016/1/check/openstack-tox-pep8/c3d0c6e/ : SUCCESS in 3m 32s\n- openstack-tox-py27 http://logs.openstack.org/16/661016/1/check/openstack-tox-py27/f312506/ : SUCCESS in 3m 20s\n- openstack-tox-py35 http://logs.openstack.org/16/661016/1/check/openstack-tox-py35/207a60b/ : SUCCESS in 3m 43s\n- build-openstack-sphinx-docs http://logs.openstack.org/16/661016/1/check/build-openstack-sphinx-docs/27d3cff/html/ : SUCCESS in 3m 20s\n- kolla-build-ubuntu-binary http://logs.openstack.org/16/661016/1/check/kolla-build-ubuntu-binary/859e8be/ : SUCCESS in 1h 42m 48s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/16/661016/1/check/kolla-build-ubuntu-source/538e093/ : SUCCESS in 30m 51s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/16/661016/1/check/kolla-ansible-ubuntu-source/d273834/ : SUCCESS in 1h 05m 20s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/16/661016/1/check/kolla-ansible-ubuntu-binary/5cd873d/ : TIMED_OUT in 2h 01m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"56478cb95bb58f173c2d9300ca852de6728eba83","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-24 20:41:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"78049cd014c34ad6b796bbd6e4a3365d29b0c9e4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-28 17:47:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d8699878cc24cb866645ca40b7dad9981cf48015","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 17:47:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2ff2f2abd9913a6a1fbbac6f70616416002d5bb0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 18:44:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/16/661016/1/gate/kolla-build-centos-source/5a5f515/ : SUCCESS in 43m 12s\n- kolla-build-oraclelinux-source http://logs.openstack.org/16/661016/1/gate/kolla-build-oraclelinux-source/d06be61/ : SUCCESS in 51m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/661016/1/gate/openstack-tox-lower-constraints/b68ebd2/ : SUCCESS in 3m 33s\n- openstack-tox-pep8 http://logs.openstack.org/16/661016/1/gate/openstack-tox-pep8/d091c5f/ : SUCCESS in 3m 38s\n- openstack-tox-py27 http://logs.openstack.org/16/661016/1/gate/openstack-tox-py27/404d0dc/ : SUCCESS in 3m 41s\n- openstack-tox-py35 http://logs.openstack.org/16/661016/1/gate/openstack-tox-py35/71ca848/ : SUCCESS in 3m 54s\n- build-openstack-sphinx-docs http://logs.openstack.org/16/661016/1/gate/build-openstack-sphinx-docs/a21b5d1/html/ : SUCCESS in 3m 45s\n- kolla-build-ubuntu-source http://logs.openstack.org/16/661016/1/gate/kolla-build-ubuntu-source/0d29e2a/ : SUCCESS in 35m 08s","accounts_in_message":[],"_revision_number":1},{"id":"fed61098be8533e6d54baf13f829f5cfe1555035","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 18:44:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"594b0645307432ba007657a135dd8fa34d84c7eb","revisions":{"594b0645307432ba007657a135dd8fa34d84c7eb":{"kind":"REWORK","_number":1,"created":"2019-05-23 14:48:03.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/16/661016/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/661016/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/661016/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/661016/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/661016/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/661016/1"}}},"commit":{"parents":[{"commit":"e1af4b6a3bfcb81e683a2525096c41bcebfe3d37","subject":"Merge \"Use latest release of Monasca Log API\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1af4b6a3bfcb81e683a2525096c41bcebfe3d37"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-05-23 10:43:17.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-05-23 14:48:03.000000000","tz":0},"subject":"Bump Monasca Fluentd plugin version","message":"Bump Monasca Fluentd plugin version\n\nBump version to include support for setting log message field\nname.\n\nChange-Id: Ib3953b49247bb36ac8ce25578c05c6bcb2a1905a\nPartial-Bug: #1830184\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/594b0645307432ba007657a135dd8fa34d84c7eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/594b0645307432ba007657a135dd8fa34d84c7eb"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
