)]}'
{"id":"openstack%2Fkolla~661248","triplet_id":"openstack%2Fkolla~master~I9c087164f32cb8a69d2155bc68b6f62233ad58c6","project":"openstack/kolla","branch":"master","topic":"install-arping-on-rhel8","hashtags":[],"change_id":"I9c087164f32cb8a69d2155bc68b6f62233ad58c6","subject":"Install iputils package in neutron containers","status":"MERGED","created":"2019-05-24 10:14:50.000000000","updated":"2019-05-24 20:48:15.000000000","submitted":"2019-05-24 20:17:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661248-1558729043298-19ad1844","meta_rev_id":"ea9f96e479429be98947b59fd35d2df44052e0f2","_number":661248,"virtual_id_number":661248,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2019-05-24 20:17:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-24 10:50:45.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-05-24 19:15:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-05-24 13:28:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-05-24 10:46:12.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-05-24 18:01:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-05-24 19:15:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-24 10:46:12.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-05-24 10:50:45.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-24 13:28:40.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-05-24 18:01:40.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-05-24 19:15:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-05-24 20:17:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"97e522fb02c5b2cb482b07dbc1f3aee46dd10ee4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-24 10:14:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe3b01d5a8b0cf36ff28a1450bfdeb2d58c0e262","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-24 10:46:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"830ca69b3b8ed14a8db53d2c457e06b7393c05b6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-24 10:50:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/48/661248/1/openstack-check/tripleo-build-containers-fedora-28-master/74aae88/ : SUCCESS in 31m 09s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/48/661248/1/openstack-check/tripleo-build-containers-fedora-28-stein/ce7cd0c/ : SUCCESS in 29m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2af0792ae2ff409d82c4ceb2046f4e0828a9fc3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 11:41:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/48/661248/1/check/kolla-build-centos-binary/afece2c/ : SUCCESS in 42m 47s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/48/661248/1/check/kolla-build-centos-source/bb30a79/ : SUCCESS in 45m 43s\n- kolla-ansible-centos-source http://logs.openstack.org/48/661248/1/check/kolla-ansible-centos-source/3542044/ : SUCCESS in 1h 04m 00s\n- kolla-ansible-centos-binary http://logs.openstack.org/48/661248/1/check/kolla-ansible-centos-binary/97d8b50/ : FAILURE in 1h 22m 26s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/48/661248/1/check/kolla-ansible-centos-source-upgrade/a97b12d/ : SUCCESS in 1h 16m 46s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/48/661248/1/check/tripleo-build-containers-centos-7/dcec23a/ : SUCCESS in 39m 04s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/48/661248/1/check/kolla-build-oraclelinux-binary/5219724/ : SUCCESS in 42m 15s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/48/661248/1/check/kolla-build-oraclelinux-source/429c812/ : SUCCESS in 57m 33s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/48/661248/1/check/kolla-ansible-oraclelinux-source/e456573/ : SUCCESS in 1h 17m 24s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/48/661248/1/check/kolla-ansible-oraclelinux-binary/f15cee5/ : SUCCESS in 1h 09m 46s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/48/661248/1/check/openstack-tox-lower-constraints/9a3c4a3/ : SUCCESS in 3m 40s\n- openstack-tox-pep8 http://logs.openstack.org/48/661248/1/check/openstack-tox-pep8/42cf0dd/ : SUCCESS in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/48/661248/1/check/openstack-tox-py27/e136e78/ : SUCCESS in 2m 34s\n- openstack-tox-py36 http://logs.openstack.org/48/661248/1/check/openstack-tox-py36/7ecaf41/ : SUCCESS in 2m 56s\n- openstack-tox-py37 http://logs.openstack.org/48/661248/1/check/openstack-tox-py37/03ae49d/ : SUCCESS in 2m 59s\n- openstack-tox-docs http://logs.openstack.org/48/661248/1/check/openstack-tox-docs/3f258b4/html/ : SUCCESS in 2m 18s\n- kolla-build-ubuntu-binary http://logs.openstack.org/48/661248/1/check/kolla-build-ubuntu-binary/57707d9/ : SUCCESS in 27m 37s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/48/661248/1/check/kolla-build-ubuntu-source/15bd5bc/ : SUCCESS in 36m 01s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/48/661248/1/check/kolla-ansible-ubuntu-source/2793d10/ : SUCCESS in 50m 50s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/48/661248/1/check/kolla-ansible-ubuntu-binary/7df00f2/ : SUCCESS in 44m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/48/661248/1/check/kolla-ansible-ubuntu-source-upgrade/6875cda/ : SUCCESS in 59m 53s","accounts_in_message":[],"_revision_number":1},{"id":"2ce42448800156c58b5000b1a18515b98ad574d0","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-24 13:28:40.000000000","message":"Patch Set 1: Code-Review+1\n\nAnd dependency is fine for 7 too (same package), looks good to me","accounts_in_message":[],"_revision_number":1},{"id":"d0e1da24a5849d42b261c9f52c676e544818af0f","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-05-24 18:01:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b2a48a2fb0e9f1836ada3496e5155801333da24e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-05-24 19:15:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"36cd737691844c38417b991ae74e09b18019d13d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 19:15:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5db01febe732cce47f8b7d0527b0fc5ed6cd09f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 20:17:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/48/661248/1/gate/kolla-build-centos-source/223fac0/ : SUCCESS in 51m 26s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/48/661248/1/gate/tripleo-build-containers-centos-7/16bab83/ : SUCCESS in 44m 02s\n- kolla-build-oraclelinux-source http://logs.openstack.org/48/661248/1/gate/kolla-build-oraclelinux-source/0cd09af/ : SUCCESS in 42m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/48/661248/1/gate/openstack-tox-lower-constraints/8ff7ed6/ : SUCCESS in 3m 28s\n- openstack-tox-pep8 http://logs.openstack.org/48/661248/1/gate/openstack-tox-pep8/a39d299/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/48/661248/1/gate/openstack-tox-py27/ebfbf3c/ : SUCCESS in 3m 17s\n- openstack-tox-py36 http://logs.openstack.org/48/661248/1/gate/openstack-tox-py36/b2ed28e/ : SUCCESS in 4m 18s\n- openstack-tox-py37 http://logs.openstack.org/48/661248/1/gate/openstack-tox-py37/9735038/ : SUCCESS in 3m 27s\n- openstack-tox-docs http://logs.openstack.org/48/661248/1/gate/openstack-tox-docs/76de7cc/html/ : SUCCESS in 2m 52s\n- kolla-build-ubuntu-source http://logs.openstack.org/48/661248/1/gate/kolla-build-ubuntu-source/d76b26f/ : SUCCESS in 38m 58s","accounts_in_message":[],"_revision_number":1},{"id":"240acd32f5740c02d77788364e6a312ef7d8d5af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 20:17:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c4df5ea9ded0e154ba638d035a1a6c3991cfb46c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-24 20:48:15.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit ceffa53f9f89a1c0e07867c5dcb971d28f7d9dc5","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68b998f27efb252a73b0ee4e496d4a3ce32206e2","revisions":{"68b998f27efb252a73b0ee4e496d4a3ce32206e2":{"kind":"REWORK","_number":1,"created":"2019-05-24 10:14:50.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/48/661248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/661248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/661248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/661248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/661248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/661248/1"}}},"commit":{"parents":[{"commit":"14f7501441924a8e533fc9b75ce16719de3de945","subject":"Merge \"Fix multiple mon deployment failures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14f7501441924a8e533fc9b75ce16719de3de945"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-24 10:12:28.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-24 10:14:46.000000000","tz":120},"subject":"Install iputils package in neutron containers","message":"Install iputils package in neutron containers\n\nThis package provides e.g. arping tool which is used by Neutron\nservices.\nOn e.g. RHEL7 it was probably installed as some dependency to other\npackage but on RHEL8 it\u0027s missing which cause some errors e.g.\nin neutron_l3_agent container.\n\nChange-Id: I9c087164f32cb8a69d2155bc68b6f62233ad58c6\nRelated-Bug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1713321\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68b998f27efb252a73b0ee4e496d4a3ce32206e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68b998f27efb252a73b0ee4e496d4a3ce32206e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
