)]}'
{"id":"openstack%2Fkolla~661369","triplet_id":"openstack%2Fkolla~stable%2Frocky~I7babdb794bec3b08047101e2537f030e2217cedb","project":"openstack/kolla","branch":"stable/rocky","topic":"bug/1830420","hashtags":[],"change_id":"I7babdb794bec3b08047101e2537f030e2217cedb","subject":"Remove HTTP configuration shipped in openstack-nova-placement-api RPM","status":"MERGED","created":"2019-05-24 18:10:39.000000000","updated":"2019-05-31 09:34:33.000000000","submitted":"2019-05-31 09:34:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661369-1559295273215-bed2ba93","meta_rev_id":"8b410a028708543668b46ceea72f0fa37c6d07b0","_number":661369,"virtual_id_number":661369,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":25257,"name":"limeng","email":"limeng@awcloud.com","username":"Linmeng"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-05-31 09:34:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-28 18:27:08.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"all":[{"value":1,"date":"2019-05-30 01:29:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":2,"date":"2019-05-31 08:20:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-05-30 01:36:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25257,"name":"limeng","email":"limeng@awcloud.com","username":"Linmeng"},{"value":2,"date":"2019-05-28 17:50:37.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-30 03:14:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":1,"date":"2019-05-31 08:20:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":25257,"name":"limeng","email":"limeng@awcloud.com","username":"Linmeng"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":25257,"name":"limeng","email":"limeng@awcloud.com","username":"Linmeng"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":25257,"name":"limeng","email":"limeng@awcloud.com","username":"Linmeng"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25257,"name":"limeng","email":"limeng@awcloud.com","username":"Linmeng"},{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-28 17:50:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-05-28 18:27:08.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-30 01:29:41.000000000","updated_by":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"reviewer":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"state":"REVIEWER"},{"updated":"2019-05-30 01:36:56.000000000","updated_by":{"_account_id":25257,"name":"limeng","email":"limeng@awcloud.com","username":"Linmeng"},"reviewer":{"_account_id":25257,"name":"limeng","email":"limeng@awcloud.com","username":"Linmeng"},"state":"REVIEWER"},{"updated":"2019-05-30 03:14:54.000000000","updated_by":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"reviewer":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"state":"REVIEWER"},{"updated":"2019-05-31 08:20:55.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-05-31 09:34:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b514ac8e10175d4663e5895a5b1eb33acf9f1482","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-24 18:10:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c25b3d5976a7aede424fdee61e7a3186541c0c8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-24 18:11:56.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"234d8d055bbd3e89fe4d24674f344886ee0c15d6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-24 18:48:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/69/661369/2/openstack-check/tripleo-build-containers-fedora-28-master/7478395/ : RETRY_LIMIT in 2m 49s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/69/661369/2/openstack-check/tripleo-build-containers-fedora-28-stein/ab67ed5/ : SUCCESS in 32m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e7226f78a35c361125ce37f80934ab528f66f6f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 20:03:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/69/661369/2/check/kolla-build-centos-binary/acce4fc/ : SUCCESS in 35m 44s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/661369/2/check/kolla-build-centos-source/8cfd2e6/ : SUCCESS in 48m 35s\n- kolla-ansible-centos-source http://logs.openstack.org/69/661369/2/check/kolla-ansible-centos-source/e6e1671/ : FAILURE in 5m 38s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/661369/2/check/kolla-ansible-centos-binary/f1921d9/ : SUCCESS in 1h 19m 17s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/661369/2/check/kolla-build-oraclelinux-binary/76bee52/ : FAILURE in 5m 12s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/661369/2/check/kolla-build-oraclelinux-source/b4fa61a/ : SUCCESS in 54m 06s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/661369/2/check/kolla-ansible-oraclelinux-source/2517cec/ : SUCCESS in 1h 19m 20s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/661369/2/check/kolla-ansible-oraclelinux-binary/d5f5b55/ : SUCCESS in 1h 10m 02s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/69/661369/2/check/openstack-tox-lower-constraints/ad9e1f7/ : SUCCESS in 4m 19s\n- openstack-tox-pep8 http://logs.openstack.org/69/661369/2/check/openstack-tox-pep8/10fe90b/ : SUCCESS in 3m 41s\n- openstack-tox-py27 http://logs.openstack.org/69/661369/2/check/openstack-tox-py27/00e2477/ : SUCCESS in 2m 49s\n- openstack-tox-py35 http://logs.openstack.org/69/661369/2/check/openstack-tox-py35/9ce918a/ : SUCCESS in 3m 15s\n- build-openstack-sphinx-docs http://logs.openstack.org/69/661369/2/check/build-openstack-sphinx-docs/5316d10/html/ : SUCCESS in 4m 08s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/661369/2/check/kolla-build-ubuntu-binary/b0b786a/ : SUCCESS in 1h 23m 17s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/661369/2/check/kolla-build-ubuntu-source/7e6f843/ : SUCCESS in 33m 27s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/661369/2/check/kolla-ansible-ubuntu-source/dcde1ca/ : SUCCESS in 59m 04s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/661369/2/check/kolla-ansible-ubuntu-binary/8589bab/ : SUCCESS in 1h 40m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"85d7a3cea08c2dcc8b25c15e24373115cf45fc36","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-28 17:50:37.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9299c3b8577ea9ace9ec2c4e280734ffc0fcfe27","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-28 18:27:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/69/661369/2/openstack-check/tripleo-build-containers-fedora-28-master/9618416/ : RETRY_LIMIT in 3m 06s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/69/661369/2/openstack-check/tripleo-build-containers-fedora-28-stein/76aaef6/ : SUCCESS in 34m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cef73900ad91216ad1d28315ebf8ba9197d55b76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 19:49:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/69/661369/2/check/kolla-build-centos-binary/f45808f/ : SUCCESS in 31m 47s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/69/661369/2/check/kolla-build-centos-source/6fb2166/ : SUCCESS in 46m 53s\n- kolla-ansible-centos-source http://logs.openstack.org/69/661369/2/check/kolla-ansible-centos-source/898b4c2/ : SUCCESS in 1h 07m 12s\n- kolla-ansible-centos-binary http://logs.openstack.org/69/661369/2/check/kolla-ansible-centos-binary/2c18429/ : SUCCESS in 1h 03m 15s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/69/661369/2/check/kolla-build-oraclelinux-binary/423de4a/ : SUCCESS in 46m 07s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/661369/2/check/kolla-build-oraclelinux-source/db21631/ : SUCCESS in 45m 47s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/69/661369/2/check/kolla-ansible-oraclelinux-source/b99ed54/ : SUCCESS in 1h 11m 13s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/69/661369/2/check/kolla-ansible-oraclelinux-binary/f30ff29/ : SUCCESS in 1h 15m 19s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/69/661369/2/check/openstack-tox-lower-constraints/37ed269/ : SUCCESS in 3m 35s\n- openstack-tox-pep8 http://logs.openstack.org/69/661369/2/check/openstack-tox-pep8/2eae0c1/ : SUCCESS in 4m 04s\n- openstack-tox-py27 http://logs.openstack.org/69/661369/2/check/openstack-tox-py27/51bc9a0/ : SUCCESS in 3m 44s\n- openstack-tox-py35 http://logs.openstack.org/69/661369/2/check/openstack-tox-py35/b9592db/ : SUCCESS in 3m 30s\n- build-openstack-sphinx-docs http://logs.openstack.org/69/661369/2/check/build-openstack-sphinx-docs/48ba4e3/html/ : SUCCESS in 3m 47s\n- kolla-build-ubuntu-binary http://logs.openstack.org/69/661369/2/check/kolla-build-ubuntu-binary/8e1554f/ : SUCCESS in 1h 29m 25s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/69/661369/2/check/kolla-build-ubuntu-source/fec9b7d/ : SUCCESS in 35m 44s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/69/661369/2/check/kolla-ansible-ubuntu-source/27d246b/ : SUCCESS in 1h 06m 51s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/69/661369/2/check/kolla-ansible-ubuntu-binary/4d823b3/ : SUCCESS in 1h 51m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"77d60e4598c751cf7a7beda4edcf76c1076386cb","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-05-30 01:29:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f674a333b0a3fa7a8a4e49bc5f04b3b626daf531","author":{"_account_id":25257,"name":"limeng","email":"limeng@awcloud.com","username":"Linmeng"},"date":"2019-05-30 01:36:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d9263730bc96575de989e7a8cd8786d003531cdf","author":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"date":"2019-05-30 03:14:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3fb88c5b1e5567383c7385014fe446e10ad969bd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-05-31 08:20:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2822c703826d75866063db858143ec0bea95f3fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 08:21:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dc8a8bbdbb09a8a04cd590c1e62dc3afd955cc77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 09:34:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/69/661369/2/gate/kolla-build-centos-source/f591ad1/ : SUCCESS in 42m 48s\n- kolla-build-oraclelinux-source http://logs.openstack.org/69/661369/2/gate/kolla-build-oraclelinux-source/2a6b1c4/ : SUCCESS in 43m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/661369/2/gate/openstack-tox-lower-constraints/1a41df4/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/69/661369/2/gate/openstack-tox-pep8/06e4954/ : SUCCESS in 4m 09s\n- openstack-tox-py27 http://logs.openstack.org/69/661369/2/gate/openstack-tox-py27/87ba9b4/ : SUCCESS in 3m 59s\n- openstack-tox-py35 http://logs.openstack.org/69/661369/2/gate/openstack-tox-py35/6650ac5/ : SUCCESS in 4m 26s\n- build-openstack-sphinx-docs http://logs.openstack.org/69/661369/2/gate/build-openstack-sphinx-docs/9d6431c/html/ : SUCCESS in 4m 09s\n- kolla-build-ubuntu-source http://logs.openstack.org/69/661369/2/gate/kolla-build-ubuntu-source/5d328fe/ : SUCCESS in 30m 53s","accounts_in_message":[],"_revision_number":2},{"id":"3af4916d74889df3ee86c08c7f9e4a502fecac22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 09:34:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8e9cc856bfbdcb97e04df4c950aaeaf485cd5cc4","revisions":{"fe2615830b7ba70e02ea290c4038c0ed6378ed88":{"kind":"REWORK","_number":1,"created":"2019-05-24 18:10:39.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/69/661369/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/661369/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/661369/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/661369/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/661369/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/661369/1"}}},"commit":{"parents":[{"commit":"e1af4b6a3bfcb81e683a2525096c41bcebfe3d37","subject":"Merge \"Use latest release of Monasca Log API\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1af4b6a3bfcb81e683a2525096c41bcebfe3d37"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-24 18:03:51.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-24 18:03:51.000000000","tz":60},"subject":"Remove HTTP configuration shipped in openstack-nova-placement-api RPM","message":"Remove HTTP configuration shipped in openstack-nova-placement-api RPM\n\nThe RDO openstack-nova-placement-api RPM ships with an HTTP\nconfiguration deploying placement on port 8778, while Kolla-Ansible\ncreates another file to configure it to listen on port 8780. If\n`placement_api_port` is changed to use port 8778, the placement_api\ncontainer will fail to start because the port is already in use.\n\nThis commit removes the configuration file shipped with the RPM package\nto avoid this conflict.\n\nChange-Id: I7babdb794bec3b08047101e2537f030e2217cedb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe2615830b7ba70e02ea290c4038c0ed6378ed88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe2615830b7ba70e02ea290c4038c0ed6378ed88"}]},"branch":"refs/heads/stable/rocky"},"8e9cc856bfbdcb97e04df4c950aaeaf485cd5cc4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-24 18:11:56.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/69/661369/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/661369/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/661369/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/661369/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/661369/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/661369/2"}}},"commit":{"parents":[{"commit":"e1af4b6a3bfcb81e683a2525096c41bcebfe3d37","subject":"Merge \"Use latest release of Monasca Log API\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1af4b6a3bfcb81e683a2525096c41bcebfe3d37"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-24 18:11:42.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-24 18:11:42.000000000","tz":60},"subject":"Remove HTTP configuration shipped in openstack-nova-placement-api RPM","message":"Remove HTTP configuration shipped in openstack-nova-placement-api RPM\n\nThe RDO openstack-nova-placement-api RPM ships with an HTTP\nconfiguration deploying placement on port 8778, while Kolla-Ansible\ncreates another file to configure it to listen on port 8780. If\n`placement_api_port` is changed to use port 8778, the placement_api\ncontainer will fail to start because the port is already in use.\n\nThis commit removes the configuration file shipped with the RPM package\nto avoid this conflict.\n\nChange-Id: I7babdb794bec3b08047101e2537f030e2217cedb\nCloses-Bug: #1830420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e9cc856bfbdcb97e04df4c950aaeaf485cd5cc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e9cc856bfbdcb97e04df4c950aaeaf485cd5cc4"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
