)]}'
{"id":"openstack%2Fkolla~661395","triplet_id":"openstack%2Fkolla~stable%2Fstein~I9c087164f32cb8a69d2155bc68b6f62233ad58c6","project":"openstack/kolla","branch":"stable/stein","topic":"install-arping-on-rhel8-stable/stein","hashtags":[],"change_id":"I9c087164f32cb8a69d2155bc68b6f62233ad58c6","subject":"Install iputils package in neutron containers","status":"MERGED","created":"2019-05-24 20:48:15.000000000","updated":"2019-06-13 14:07:58.000000000","submitted":"2019-05-28 23:37:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661395-1559086642047-ea30ef60","meta_rev_id":"9718bb0b7424511078e14ca7fc76f4728da04891","_number":661395,"virtual_id_number":661395,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2019-05-28 22:00:03.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2019-05-27 14:26:23.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2019-05-28 23:37:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},{"value":0,"date":"2019-05-28 10:45:25.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"recommended":{"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},"all":[{"value":1,"date":"2019-05-27 14:23:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":1,"date":"2019-05-28 05:39:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},{"value":2,"date":"2019-05-28 10:08:52.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":2,"date":"2019-05-28 07:20:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-05-28 22:27:08.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-05-28 01:36:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-28 05:52:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-28 14:14:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-05-28 22:27:08.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-28 14:14:42.000000000","_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":1,"date":"2019-05-28 07:20:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-27 14:23:34.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"REVIEWER"},{"updated":"2019-05-27 14:26:23.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-05-28 01:36:00.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-05-28 05:39:12.000000000","updated_by":{"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},"reviewer":{"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},"state":"REVIEWER"},{"updated":"2019-05-28 05:52:13.000000000","updated_by":{"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},"reviewer":{"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},"state":"REVIEWER"},{"updated":"2019-05-28 07:20:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-05-28 10:08:52.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-05-28 10:45:25.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-28 14:14:42.000000000","updated_by":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"reviewer":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"state":"REVIEWER"},{"updated":"2019-05-28 22:27:08.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-05-28 23:37:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"964b8242d579efbf2b58d99af4131c58dbb0fe03","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-24 20:48:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e56aea224197730170fcb8873e4c5180510deeb9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-24 21:22:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/95/661395/1/openstack-check/tripleo-build-containers-fedora-28-master/9306f0c/ : SUCCESS in 32m 40s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/95/661395/1/openstack-check/tripleo-build-containers-fedora-28-stein/4ea16c4/ : SUCCESS in 32m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af94953d3435ae22eee53bf0d3387524bcfb413f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 22:18:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/95/661395/1/check/kolla-build-centos-binary/9b3090a/ : FAILURE in 39m 24s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/95/661395/1/check/kolla-build-centos-source/65aa012/ : SUCCESS in 40m 13s\n- kolla-ansible-centos-source http://logs.openstack.org/95/661395/1/check/kolla-ansible-centos-source/0c16987/ : SUCCESS in 1h 12m 15s\n- kolla-ansible-centos-binary http://logs.openstack.org/95/661395/1/check/kolla-ansible-centos-binary/6242774/ : FAILURE in 31m 51s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/95/661395/1/check/kolla-ansible-centos-source-upgrade/1a27c16/ : SUCCESS in 1h 11m 21s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/95/661395/1/check/tripleo-build-containers-centos-7/83ec553/ : SUCCESS in 39m 43s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/95/661395/1/check/kolla-build-oraclelinux-binary/b46c505/ : FAILURE in 1h 07m 42s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/95/661395/1/check/kolla-build-oraclelinux-source/c29d0cf/ : SUCCESS in 44m 44s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/95/661395/1/check/kolla-ansible-oraclelinux-source/1db5613/ : SUCCESS in 1h 21m 20s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/95/661395/1/check/kolla-ansible-oraclelinux-binary/1afda73/ : FAILURE in 38m 46s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/95/661395/1/check/openstack-tox-lower-constraints/1ad42b2/ : SUCCESS in 3m 54s\n- openstack-tox-pep8 http://logs.openstack.org/95/661395/1/check/openstack-tox-pep8/0e73510/ : SUCCESS in 3m 02s\n- openstack-tox-py27 http://logs.openstack.org/95/661395/1/check/openstack-tox-py27/39717e9/ : SUCCESS in 3m 08s\n- openstack-tox-py36 http://logs.openstack.org/95/661395/1/check/openstack-tox-py36/08cb1dd/ : SUCCESS in 3m 17s\n- openstack-tox-py37 http://logs.openstack.org/95/661395/1/check/openstack-tox-py37/4851cd5/ : SUCCESS in 4m 41s\n- openstack-tox-docs http://logs.openstack.org/95/661395/1/check/openstack-tox-docs/d7a5c95/html/ : SUCCESS in 3m 18s\n- kolla-build-ubuntu-binary http://logs.openstack.org/95/661395/1/check/kolla-build-ubuntu-binary/4286880/ : SUCCESS in 1h 28m 10s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/95/661395/1/check/kolla-build-ubuntu-source/4794c1e/ : SUCCESS in 29m 02s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/95/661395/1/check/kolla-ansible-ubuntu-source/cedf617/ : SUCCESS in 1h 10m 41s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/95/661395/1/check/kolla-ansible-ubuntu-binary/f72dcd3/ : SUCCESS in 45m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/95/661395/1/check/kolla-ansible-ubuntu-source-upgrade/e94598a/ : SUCCESS in 54m 41s","accounts_in_message":[],"_revision_number":1},{"id":"23e71390c62ba02d6ecb9797b1f23ea3d19abe0d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-25 06:33:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2b0efa3b86ca6eb7dce3fabc99f54aec1b8dad14","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-25 07:13:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/95/661395/2/openstack-check/tripleo-build-containers-fedora-28-master/f831716/ : SUCCESS in 31m 55s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/95/661395/2/openstack-check/tripleo-build-containers-fedora-28-stein/2fca9dc/ : SUCCESS in 33m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cd3e815ca0164dbcedcb18eb18b5ea5bf0ffa541","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 08:02:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/95/661395/2/check/kolla-build-centos-binary/912d59b/ : FAILURE in 38m 27s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/95/661395/2/check/kolla-build-centos-source/60e0ba9/ : SUCCESS in 47m 39s\n- kolla-ansible-centos-source http://logs.openstack.org/95/661395/2/check/kolla-ansible-centos-source/0e3cdb8/ : SUCCESS in 1h 07m 33s\n- kolla-ansible-centos-binary http://logs.openstack.org/95/661395/2/check/kolla-ansible-centos-binary/5d41177/ : FAILURE in 25m 35s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/95/661395/2/check/kolla-ansible-centos-source-upgrade/5984649/ : SUCCESS in 1h 14m 10s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/95/661395/2/check/tripleo-build-containers-centos-7/17f05e4/ : SUCCESS in 41m 53s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/95/661395/2/check/kolla-build-oraclelinux-binary/a744463/ : FAILURE in 58m 49s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/95/661395/2/check/kolla-build-oraclelinux-source/0d7d15a/ : SUCCESS in 50m 29s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/95/661395/2/check/kolla-ansible-oraclelinux-source/a4b5376/ : SUCCESS in 1h 20m 32s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/95/661395/2/check/kolla-ansible-oraclelinux-binary/19efe03/ : FAILURE in 32m 30s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/95/661395/2/check/openstack-tox-lower-constraints/ee4ab2a/ : SUCCESS in 3m 42s\n- openstack-tox-pep8 http://logs.openstack.org/95/661395/2/check/openstack-tox-pep8/2deeea9/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/95/661395/2/check/openstack-tox-py27/59cbef4/ : SUCCESS in 2m 37s\n- openstack-tox-py36 http://logs.openstack.org/95/661395/2/check/openstack-tox-py36/753382b/ : SUCCESS in 3m 52s\n- openstack-tox-py37 http://logs.openstack.org/95/661395/2/check/openstack-tox-py37/22e6843/ : SUCCESS in 3m 19s\n- openstack-tox-docs http://logs.openstack.org/95/661395/2/check/openstack-tox-docs/b76712c/html/ : SUCCESS in 2m 42s\n- kolla-build-ubuntu-binary http://logs.openstack.org/95/661395/2/check/kolla-build-ubuntu-binary/35c3094/ : SUCCESS in 27m 53s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/95/661395/2/check/kolla-build-ubuntu-source/cabb0fd/ : SUCCESS in 28m 28s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/95/661395/2/check/kolla-ansible-ubuntu-source/129276f/ : SUCCESS in 49m 28s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/95/661395/2/check/kolla-ansible-ubuntu-binary/930baee/ : SUCCESS in 46m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/95/661395/2/check/kolla-ansible-ubuntu-source-upgrade/e10f5cf/ : SUCCESS in 55m 12s","accounts_in_message":[],"_revision_number":2},{"id":"40614fb99adb4c7b053c99cb07d2824e976df247","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2019-05-27 14:22:32.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"edb711d9d9643731a7e8454bb7f51b259727b749","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2019-05-27 14:23:34.000000000","message":"Patch Set 2: Code-Review+1\n\nok, I see now this is a backport, please consider followup cleanup.","accounts_in_message":[],"_revision_number":2},{"id":"40ea54bf30eee3783aa769b403113f32bdffd837","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-27 14:26:23.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cad5df2dcbf8fa671a50bf1cbeb8eed6b8c75727","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2019-05-27 15:06:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c3de6b0e301e35fb9b64527af66c8d80f6424143","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-05-28 01:36:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"83ab1bb056da0a84b9f7bdae45001593fbda2d9e","author":{"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},"date":"2019-05-28 05:39:12.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f1e7764157d656450d2de17540bb3f82ecbd484e","author":{"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},"date":"2019-05-28 05:52:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fbbaed225bcf8a4e87a95d9c9850416decc06998","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-05-28 07:20:10.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1\n\ncentos-binary failures are related to wrong package name (unrelated to this change)","accounts_in_message":[],"_revision_number":2},{"id":"84c637137f2cf858b25d4caf55ae82e172f87eef","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-05-28 10:08:52.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"861935f53463c504684d3fc14a6ea76be395477b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-28 10:45:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/95/661395/2/openstack-check/tripleo-build-containers-fedora-28-master/14cf532/ : SUCCESS in 33m 23s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/95/661395/2/openstack-check/tripleo-build-containers-fedora-28-stein/74c4295/ : SUCCESS in 33m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"276576693d024411200ed8650d5f276c76f0286b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 11:36:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/95/661395/2/check/kolla-build-centos-binary/710469d/ : FAILURE in 25m 59s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/95/661395/2/check/kolla-build-centos-source/1cdad75/ : SUCCESS in 47m 32s\n- kolla-ansible-centos-source http://logs.openstack.org/95/661395/2/check/kolla-ansible-centos-source/4422551/ : SUCCESS in 1h 12m 22s\n- kolla-ansible-centos-binary http://logs.openstack.org/95/661395/2/check/kolla-ansible-centos-binary/ec168b3/ : FAILURE in 28m 31s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/95/661395/2/check/kolla-ansible-centos-source-upgrade/b430360/ : SUCCESS in 1h 16m 56s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/95/661395/2/check/tripleo-build-containers-centos-7/40060cc/ : SUCCESS in 50m 08s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/95/661395/2/check/kolla-build-oraclelinux-binary/56470c5/ : FAILURE in 55m 05s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/95/661395/2/check/kolla-build-oraclelinux-source/5fb6c25/ : SUCCESS in 37m 36s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/95/661395/2/check/kolla-ansible-oraclelinux-source/7e2aeaf/ : SUCCESS in 1h 19m 25s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/95/661395/2/check/kolla-ansible-oraclelinux-binary/c5e2ae6/ : FAILURE in 29m 45s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/95/661395/2/check/openstack-tox-lower-constraints/881ca2f/ : SUCCESS in 4m 04s\n- openstack-tox-pep8 http://logs.openstack.org/95/661395/2/check/openstack-tox-pep8/c1fe59f/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/95/661395/2/check/openstack-tox-py27/d91692d/ : SUCCESS in 3m 07s\n- openstack-tox-py36 http://logs.openstack.org/95/661395/2/check/openstack-tox-py36/85557a4/ : SUCCESS in 2m 59s\n- openstack-tox-py37 http://logs.openstack.org/95/661395/2/check/openstack-tox-py37/5abbe80/ : SUCCESS in 3m 06s\n- openstack-tox-docs http://logs.openstack.org/95/661395/2/check/openstack-tox-docs/52a0c1f/html/ : SUCCESS in 3m 01s\n- kolla-build-ubuntu-binary http://logs.openstack.org/95/661395/2/check/kolla-build-ubuntu-binary/b3cc786/ : SUCCESS in 33m 27s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/95/661395/2/check/kolla-build-ubuntu-source/15e97bc/ : SUCCESS in 30m 17s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/95/661395/2/check/kolla-ansible-ubuntu-source/1bcc4e2/ : SUCCESS in 53m 17s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/95/661395/2/check/kolla-ansible-ubuntu-binary/b64cd44/ : SUCCESS in 47m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/95/661395/2/check/kolla-ansible-ubuntu-source-upgrade/6bff1d4/ : SUCCESS in 57m 21s","accounts_in_message":[],"_revision_number":2},{"id":"573198151ffdf0711982afec73abeedce5ffe8a9","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2019-05-28 14:14:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nCI failures unrelated are unrelated to the change","accounts_in_message":[],"_revision_number":2},{"id":"6db328458cd8216b237bc5c9060dedf54b66b6a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 14:14:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"daa7e414f0180df6569c417390cd5065296c070e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 15:28:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/95/661395/2/gate/kolla-build-centos-source/bff405b/ : SUCCESS in 45m 55s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/95/661395/2/gate/tripleo-build-containers-centos-7/cede46e/ : SUCCESS in 45m 30s\n- kolla-build-oraclelinux-source http://logs.openstack.org/95/661395/2/gate/kolla-build-oraclelinux-source/a9296dc/ : SUCCESS in 1h 00m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/95/661395/2/gate/openstack-tox-lower-constraints/63925a5/ : SUCCESS in 3m 54s\n- openstack-tox-pep8 http://logs.openstack.org/95/661395/2/gate/openstack-tox-pep8/42f6709/ : SUCCESS in 3m 03s\n- openstack-tox-py27 http://logs.openstack.org/95/661395/2/gate/openstack-tox-py27/e12d243/ : SUCCESS in 3m 02s\n- openstack-tox-py36 http://logs.openstack.org/95/661395/2/gate/openstack-tox-py36/72356d1/ : SUCCESS in 3m 17s\n- openstack-tox-py37 http://logs.openstack.org/95/661395/2/gate/openstack-tox-py37/f3fda06/ : SUCCESS in 3m 40s\n- openstack-tox-docs http://logs.openstack.org/95/661395/2/gate/openstack-tox-docs/664b99f/html/ : SUCCESS in 2m 36s\n- kolla-build-ubuntu-source http://logs.openstack.org/95/661395/2/gate/kolla-build-ubuntu-source/a73ecef/ : SUCCESS in 30m 22s","accounts_in_message":[],"_revision_number":2},{"id":"4983f0cde0c3450557c081f2ad255b69d78c2780","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-28 22:00:03.000000000","message":"Patch Set 2:\n\nVerified +2 but not merged, is something missing here?","accounts_in_message":[],"_revision_number":2},{"id":"bdb094add8f5f6f7a288cc931b2ed1be10fda10e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-28 22:27:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c32951061ce88f732f3ca9c42493f3b4761b86af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 22:27:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4e2aee8ece3ccdda8abaaf85009a0b03e56c0b4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 23:37:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/95/661395/2/gate/kolla-build-centos-source/f325506/ : SUCCESS in 45m 23s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/95/661395/2/gate/tripleo-build-containers-centos-7/e217d7a/ : SUCCESS in 35m 50s\n- kolla-build-oraclelinux-source http://logs.openstack.org/95/661395/2/gate/kolla-build-oraclelinux-source/e8ad0b6/ : SUCCESS in 56m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/95/661395/2/gate/openstack-tox-lower-constraints/99ef55a/ : SUCCESS in 6m 01s\n- openstack-tox-pep8 http://logs.openstack.org/95/661395/2/gate/openstack-tox-pep8/35bba96/ : SUCCESS in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/95/661395/2/gate/openstack-tox-py27/4dce2ab/ : SUCCESS in 5m 02s\n- openstack-tox-py36 http://logs.openstack.org/95/661395/2/gate/openstack-tox-py36/baabbc2/ : SUCCESS in 4m 35s\n- openstack-tox-py37 http://logs.openstack.org/95/661395/2/gate/openstack-tox-py37/f33daf3/ : SUCCESS in 3m 33s\n- openstack-tox-docs http://logs.openstack.org/95/661395/2/gate/openstack-tox-docs/208e620/html/ : SUCCESS in 2m 50s\n- kolla-build-ubuntu-source http://logs.openstack.org/95/661395/2/gate/kolla-build-ubuntu-source/ff153bc/ : SUCCESS in 35m 58s","accounts_in_message":[],"_revision_number":2},{"id":"3f4dfe0db0d7ae93e34e93e64a1ab4f241274aff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 23:37:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"354466f74fb31f2d1f1f39db8b40220c1e6eaa01","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-06-13 14:07:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ad4435618010ebfe2e2c6b5ee5b16c04a0abab04","revisions":{"ceffa53f9f89a1c0e07867c5dcb971d28f7d9dc5":{"kind":"REWORK","_number":1,"created":"2019-05-24 20:48:15.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/95/661395/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/661395/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/661395/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/661395/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/661395/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/661395/1"}}},"commit":{"parents":[{"commit":"728d49e14068a6977c6e6cb601278dcf69b2e9ff","subject":"Merge \"Add sqlite package to manila-share\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/728d49e14068a6977c6e6cb601278dcf69b2e9ff"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-24 10:12:28.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-05-24 20:48:15.000000000","tz":0},"subject":"Install iputils package in neutron containers","message":"Install iputils package in neutron containers\n\nThis package provides e.g. arping tool which is used by Neutron\nservices.\nOn e.g. RHEL7 it was probably installed as some dependency to other\npackage but on RHEL8 it\u0027s missing which cause some errors e.g.\nin neutron_l3_agent container.\n\nChange-Id: I9c087164f32cb8a69d2155bc68b6f62233ad58c6\nRelated-Bug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1713321\n(cherry picked from commit 68b998f27efb252a73b0ee4e496d4a3ce32206e2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ceffa53f9f89a1c0e07867c5dcb971d28f7d9dc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ceffa53f9f89a1c0e07867c5dcb971d28f7d9dc5"}]},"branch":"refs/heads/stable/stein"},"ad4435618010ebfe2e2c6b5ee5b16c04a0abab04":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-25 06:33:14.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/661395/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/661395/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/661395/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/661395/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/661395/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/661395/2"}}},"commit":{"parents":[{"commit":"403b40610eead3e99d8da011709b18959a5fa028","subject":"Merge \"Improve doc of kolla-build \u0027--docker-dir\u0027 option\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/403b40610eead3e99d8da011709b18959a5fa028"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-24 10:12:28.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-25 06:33:14.000000000","tz":0},"subject":"Install iputils package in neutron containers","message":"Install iputils package in neutron containers\n\nThis package provides e.g. arping tool which is used by Neutron\nservices.\nOn e.g. RHEL7 it was probably installed as some dependency to other\npackage but on RHEL8 it\u0027s missing which cause some errors e.g.\nin neutron_l3_agent container.\n\nChange-Id: I9c087164f32cb8a69d2155bc68b6f62233ad58c6\nRelated-Bug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1713321\n(cherry picked from commit 68b998f27efb252a73b0ee4e496d4a3ce32206e2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad4435618010ebfe2e2c6b5ee5b16c04a0abab04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad4435618010ebfe2e2c6b5ee5b16c04a0abab04"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
