)]}'
{"id":"openstack%2Fkolla~661686","triplet_id":"openstack%2Fkolla~stable%2Fstein~Ifff5c60f5690098619bf5a43f25ae3895a96e0d2","project":"openstack/kolla","branch":"stable/stein","topic":"split-placement-stable/stein","hashtags":[],"change_id":"Ifff5c60f5690098619bf5a43f25ae3895a96e0d2","subject":"Remove nova-placement-api","status":"ABANDONED","created":"2019-05-28 07:27:46.000000000","updated":"2019-06-06 06:37:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1005e503a96fa4f8c19fb07c735aa6546768ddf9","_number":661686,"virtual_id_number":661686,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-05-28 08:50:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-06-06 06:30:10.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-06-06 06:31:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":2,"date":"2019-06-06 05:45:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},{"value":1,"date":"2019-05-28 07:54:51.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"date":"2019-05-28 07:27:46.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"date":"2019-06-06 05:45:53.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-28 07:27:46.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2019-05-28 08:50:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-06 05:45:53.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-06 05:45:59.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-06 06:30:10.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-06 06:31:26.000000000","updated_by":{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},"reviewer":{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},"state":"REVIEWER"}],"messages":[{"id":"76e9dcf8a3a280917797ac1a76e1fed890d9009a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-05-28 07:27:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e8d3fadb639234ef29392daba16a0d16f3abc272","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-05-28 07:30:10.000000000","message":"Patch Set 1:\n\nBackporting this change, because binary builds fail - RDO lacks nova-placement rpms.","accounts_in_message":[],"_revision_number":1},{"id":"146aa0cea20ab93e65d7d90cd5a2bf3dc1ac2e99","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-05-28 07:54:51.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"c7f151be1b85609b999c055b8438a4eeb8a82d93","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-28 08:02:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/86/661686/1/openstack-check/tripleo-build-containers-fedora-28-master/f3104b5/ : SUCCESS in 31m 05s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/86/661686/1/openstack-check/tripleo-build-containers-fedora-28-stein/cfd5e71/ : SUCCESS in 31m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a6efc6e7375f949aa69a1dc4a0239e79763bf2eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 08:50:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/86/661686/1/check/kolla-build-centos-binary/df6cb69/ : SUCCESS in 43m 11s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/86/661686/1/check/kolla-build-centos-source/33380e5/ : SUCCESS in 49m 42s\n- kolla-ansible-centos-source http://logs.openstack.org/86/661686/1/check/kolla-ansible-centos-source/9e3b75e/ : SUCCESS in 1h 12m 09s\n- kolla-ansible-centos-binary http://logs.openstack.org/86/661686/1/check/kolla-ansible-centos-binary/e3498bf/ : SUCCESS in 1h 00m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/86/661686/1/check/kolla-ansible-centos-source-upgrade/d204f5b/ : SUCCESS in 1h 11m 21s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/86/661686/1/check/tripleo-build-containers-centos-7/0e93fce/ : SUCCESS in 43m 34s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/86/661686/1/check/kolla-build-oraclelinux-binary/8167b87/ : SUCCESS in 59m 21s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/86/661686/1/check/kolla-build-oraclelinux-source/79d4ec2/ : SUCCESS in 49m 46s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/86/661686/1/check/kolla-ansible-oraclelinux-source/1b39e61/ : SUCCESS in 1h 16m 30s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/86/661686/1/check/kolla-ansible-oraclelinux-binary/3f361e5/ : SUCCESS in 1h 06m 47s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/86/661686/1/check/openstack-tox-lower-constraints/6e9ac7b/ : SUCCESS in 3m 28s\n- openstack-tox-pep8 http://logs.openstack.org/86/661686/1/check/openstack-tox-pep8/166cb5a/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/86/661686/1/check/openstack-tox-py27/eb6cdd5/ : SUCCESS in 3m 03s\n- openstack-tox-py36 http://logs.openstack.org/86/661686/1/check/openstack-tox-py36/f09eb02/ : SUCCESS in 2m 45s\n- openstack-tox-py37 http://logs.openstack.org/86/661686/1/check/openstack-tox-py37/e9b8bb9/ : SUCCESS in 3m 13s\n- openstack-tox-docs http://logs.openstack.org/86/661686/1/check/openstack-tox-docs/5e2b0ba/html/ : SUCCESS in 2m 12s\n- kolla-build-ubuntu-binary http://logs.openstack.org/86/661686/1/check/kolla-build-ubuntu-binary/5944570/ : SUCCESS in 28m 24s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/86/661686/1/check/kolla-build-ubuntu-source/bef5ac8/ : SUCCESS in 29m 25s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/86/661686/1/check/kolla-ansible-ubuntu-source/e97de43/ : SUCCESS in 51m 39s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/86/661686/1/check/kolla-ansible-ubuntu-binary/ccce7c4/ : SUCCESS in 47m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/86/661686/1/check/kolla-ansible-ubuntu-source-upgrade/addb31d/ : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":1},{"id":"19c2964c910179dfe6e4351333dab79e796616e3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-05-28 09:01:19.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"1d79eca0769783fb7b44f0cc138cfba72647d2f8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-06 05:45:35.000000000","message":"Restored\n\nWe need this to release stein","accounts_in_message":[],"_revision_number":1},{"id":"c1244b1797253d05077d1b422f331fa5957e5bbf","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-06 05:45:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1959d9062d1c499e2251093cce39fb35e4cabb7c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 06:30:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/86/661686/1/openstack-check/tripleo-build-containers-fedora-28-master/41c0c82/ : SUCCESS in 43m 29s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/86/661686/1/openstack-check/tripleo-build-containers-fedora-28-stein/cfeed92/ : SUCCESS in 38m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"230a3c29e33269c8ac47bb1e0b65480e63dbb2cd","author":{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},"date":"2019-06-06 06:31:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1aea6446fdd28a067845dafa1f50ef945558dde6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-06 06:37:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0809ee31ae42327ddcfd5401d1fe4ce0f41986ce","revisions":{"0809ee31ae42327ddcfd5401d1fe4ce0f41986ce":{"kind":"REWORK","_number":1,"created":"2019-05-28 07:27:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/86/661686/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/661686/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/661686/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/661686/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/661686/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/661686/1"}}},"commit":{"parents":[{"commit":"403b40610eead3e99d8da011709b18959a5fa028","subject":"Merge \"Improve doc of kolla-build \u0027--docker-dir\u0027 option\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/403b40610eead3e99d8da011709b18959a5fa028"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2019-04-29 08:34:54.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-05-28 07:27:46.000000000","tz":0},"subject":"Remove nova-placement-api","message":"Remove nova-placement-api\n\nNow that the required changes to extract Placement from Nova have landed\nin kolla-ansible we can remove the nova-placement-api service from\nkolla.\n\nChange-Id: Ifff5c60f5690098619bf5a43f25ae3895a96e0d2\n(cherry picked from commit 755e65c3f9e38bbfdb22238fde33b92fb971ff2c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0809ee31ae42327ddcfd5401d1fe4ce0f41986ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0809ee31ae42327ddcfd5401d1fe4ce0f41986ce"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
